)]}'
{"id":"openstack%2Fneutron~934401","triplet_id":"openstack%2Fneutron~stable%2F2023.1~Idc6ad29bcac23c2397e32f290addfd1877b8b3e0","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2086205","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-11-12 12:22:01.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2024-11-11 08:34:52.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Idc6ad29bcac23c2397e32f290addfd1877b8b3e0","subject":"[OVN] Create the SG rules revision number registers","status":"MERGED","created":"2024-11-08 07:01:47.000000000","updated":"2024-11-12 12:23:38.000000000","submitted":"2024-11-12 12:22:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934401-bug/2086205","meta_rev_id":"f6039e841a96bd389d35785ed1edac02b5c248ca","_number":934401,"virtual_id_number":934401,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-12 12:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2024-11-11 08:34:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2024-11-12 10:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-11-11 09:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2024-11-12 10:37:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-08 08:10:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-08 08:38:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-08 08:47:11.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2024-11-08 21:01:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-11-11 09:09:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-11-12 10:37:19.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"8478254d0f44989f72aa20a909ab4adc8a474629","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-08 07:01:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a39f08375c4af6c130969c381650b2116d3d1688","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 08:10:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b9634a41f7043cea810e80933652893\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79cd23cb1ffd477f952779e7ec74a8db : SUCCESS in 25m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e93d01e54f6476cb988ac66e6de689b : SUCCESS in 37m 12s","accounts_in_message":[],"_revision_number":1},{"id":"bc07273e90aee50cb2189660fb045e988482d0c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 08:38:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53b6a036af3c48949afa5c50161f8a6e\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/bab4c58efea0474891b1d87cb08eae07 : SUCCESS in 1h 25m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af938a6594d44f2fa664f1661ffd7984 : SUCCESS in 29m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4c4a909ec864a91be1bf1caa539f3e2 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7a0a3108f564aad97e87da3b78d7e09 : SUCCESS in 22m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbb505c61ae848d6bb68d049aa2346c7 : SUCCESS in 25m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b94894c6a14f6bafb2a350fa67157b : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f99adbc9d9524a37a49201d3abc1ef6f : FAILURE in 35m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c708ff8005d8417195cc5f3a81c8868a : SUCCESS in 53m 51s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1513db23617c4378aa86ee57190c0e4c : SUCCESS in 46m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/849993b8f93444c8a4b0e4c88e5838b5 : SUCCESS in 1h 23m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d65adb7523a449ab87911b35e8020499 : SUCCESS in 1h 01m 57s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4892577c2eaa4d32a7f961e31ff490a5 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2ed81cfe9ea280c39eb9d08414b72ff20b8104ee","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-08 08:47:11.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/748c4e780216409290cb2847160ea775\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/deaee46efcd046c6b5acb10435bed273 : SUCCESS in 1h 44m 12s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c75de19557a74007a47aeb32617c5dbd : FAILURE in 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"954b16f7883e73c318d4fa622b00075dd3886950","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-08 09:15:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f8b6bcd65c3d9706986059f8745745cc5398954d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 10:51:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae894fe3e8d24f9abb1df71d9020736f\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/a682b0d096a046929ca07cebef278a31 : SUCCESS in 1h 35m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04ca461c2b7849d882dbd8945d516e2c : SUCCESS in 35m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd43e808118649448c81b6c68a4f1ddc : SUCCESS in 12m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86e456cd90a24f1fb9f821d81f990882 : SUCCESS in 18m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abb18b3687d741fe9796b8221edb106d : SUCCESS in 20m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a112a1c10848c0ae833b435e338c49 : SUCCESS in 11m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/26152bf08d8e4f80b16ba027d3b49f91 : FAILURE in 32m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0f4432098513431eaf2e50d244d7d6aa : SUCCESS in 41m 35s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7bfddccaa6b7470a858710d9cb5bcd74 : SUCCESS in 1h 04m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8d7dc92bb01c4e019c4881a7090cc663 : SUCCESS in 1h 19m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ddf0424ea7a441adb4820331891d2820 : SUCCESS in 1h 00m 53s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/67385efe3d554e2f809698fc9004bde5 : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":2},{"id":"36784d695a282b3a14485eb664d26c9c6a0f724c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-08 10:54:46.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/7cf20b627c1743db933dd540af075e7f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/dab391e0f9b541f0bdfd9543cbddec8f : SUCCESS in 1h 38m 22s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b89298ea11714464ae76a82247003679 : FAILURE in 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"381eb61c32e6dcb41e696102ff0e701790cb1525","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 12:16:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12d0b6e711b2453fab9bae1e287133f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd72e63d79b54826af591a29d81c7d00 : SUCCESS in 44m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11e47e6e63a04e0c82c2dae93d4c26d1 : SUCCESS in 25m 11s","accounts_in_message":[],"_revision_number":2},{"id":"f57b2bca05234bbccd60399fe2ba91373d149838","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-11 06:55:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"06780ce55044c2b61b56f774eb1c75453382eab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 07:32:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bf6ea82e1f34c109a7fd6470c6ca78c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc19767a4881476799e05eb18e4e3d64 : SUCCESS in 25m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d20beda154b4eea96f0f1f4c153722e : SUCCESS in 35m 11s","accounts_in_message":[],"_revision_number":3},{"id":"6d48bb60fd8d8aedda06c6010037900b15764f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 08:23:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd0f3768f5444548cec616d241336f2\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/99ffd63a0248476db18550f9a03e3ffe : SUCCESS in 1h 27m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b9ba3bc3bef41ef84a1011a7c081d99 : SUCCESS in 25m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c772302f13410c936b0d7f2ae30f8a : SUCCESS in 13m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ff597aa4b964f95b2d3b1217d7fecf0 : SUCCESS in 20m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de7b7cdb9b424b4d9f44c7020bd537ba : SUCCESS in 24m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/968eba71ecfe45c78464bab897d29373 : SUCCESS in 8m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad7483169af24a20a0608ff1a602ed07 : SUCCESS in 38m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/216e4838e1f74d9a877b14142f682654 : SUCCESS in 54m 25s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/67a89d03edaa438b836218786834c513 : SUCCESS in 48m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c5edcdb7e1a34138acf5b8c54537421a : SUCCESS in 1h 06m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/93dc302961784afe9e95d81eea9e5124 : SUCCESS in 53m 58s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/24a539a98b714ee396bba15ce2c2f105 : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":3},{"id":"b39a2a6bce86c13142fdb8fa9487a4cb97cef055","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-11 08:34:52.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/a6017c7b487e44579fad4638e082973c\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9301ed9477a84f50a30ffcbe49799ba9 : SUCCESS in 1h 38m 06s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/129250f2dd4c4485ac41ec178a3aa9cd : FAILURE in 36m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a4ad8b4470801b120f3aa1004e9e961ac4f7cdf","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-11 09:09:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1cb9abf2ffb52eb19ec07340bcdf47f9c17fb14e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-11-12 10:37:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"19e3796eba465e134887a009c3a6a0994cd7de0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 10:37:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1b8e4430a47bffca9390ad6a497ef21432a8ee52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:22:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5908c7ca5bc548378cf6ee87b8f1a9e2\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/95f4e40a614141c88e41be9b263462f6 : SUCCESS in 1h 36m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6d35d832fc49f9ac42cbcaca3511a9 : SUCCESS in 13m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f5e6462e1ca46709c9d240816adca21 : SUCCESS in 19m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0732ce7411a9461e9020b09e814861bb : SUCCESS in 22m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a665886cad4e0f9b48a8b2e6b9670c : SUCCESS in 8m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffbdd277ff574b1795e975ff8de1abdc : SUCCESS in 56m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/34af4d2ccb364dc7a0bca545dde53c82 : SUCCESS in 56m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e9bc8ce57e4d4d72a3f56ef4064d3ee8 : SUCCESS in 1h 16m 03s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d8744afe65284a6b895c83d180ca783d : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":3},{"id":"6dba3afdb394a18a8121fc17b7ecf776618019cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:22:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f6039e841a96bd389d35785ed1edac02b5c248ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:23:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3abaacb11927488abd3e140722f00a45\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8fdde8694741c9ac330085d633e073 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1a3e059497216f8060767e0847f5fb53a1c13709","revisions":{"495f8e452da43b3db6e34074f71b1b94ee6eb9fd":{"kind":"REWORK","_number":1,"created":"2024-11-08 07:01:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/934401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/934401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/934401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/934401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/934401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/934401/1"}}},"commit":{"parents":[{"commit":"161a6b5d32072ad4bcd6794f8bdacbdcc9241806","subject":"Merge \"[OVN] Fix the revision number retrieval method\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/161a6b5d32072ad4bcd6794f8bdacbdcc9241806"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-31 23:33:58.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-06 09:59:10.000000000","tz":0},"subject":"[OVN] Create the SG rules revision number registers","message":"[OVN] Create the SG rules revision number registers\n\nDuring a security group creation, the default security group rules are\nalso added. This patch is creating the security group rules revision\nnumber registers and bumping them to their first revision.\n\nCloses-Bug: #2086205\nChange-Id: Idc6ad29bcac23c2397e32f290addfd1877b8b3e0\n(cherry picked from commit e0ee8bd7726a24747ee5028cb31f9b62cfcfcc29)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/495f8e452da43b3db6e34074f71b1b94ee6eb9fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/495f8e452da43b3db6e34074f71b1b94ee6eb9fd"}]},"branch":"refs/heads/stable/2023.1"},"44f798531b03b07cbd3fda4a849e59534bd0b647":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-08 09:15:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/934401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/934401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/934401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/934401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/934401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/934401/2"}}},"commit":{"parents":[{"commit":"04b56f44ef1613b5d77fc76010804ac2afe7cf55","subject":"Add logging details of the http response in the unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04b56f44ef1613b5d77fc76010804ac2afe7cf55"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-31 23:33:58.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-11-08 09:15:12.000000000","tz":0},"subject":"[OVN] Create the SG rules revision number registers","message":"[OVN] Create the SG rules revision number registers\n\nDuring a security group creation, the default security group rules are\nalso added. This patch is creating the security group rules revision\nnumber registers and bumping them to their first revision.\n\nCloses-Bug: #2086205\nChange-Id: Idc6ad29bcac23c2397e32f290addfd1877b8b3e0\n(cherry picked from commit e0ee8bd7726a24747ee5028cb31f9b62cfcfcc29)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44f798531b03b07cbd3fda4a849e59534bd0b647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44f798531b03b07cbd3fda4a849e59534bd0b647"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"},"1a3e059497216f8060767e0847f5fb53a1c13709":{"kind":"REWORK","_number":3,"created":"2024-11-11 06:55:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/934401/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/934401/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/934401/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/934401/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/934401/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/934401/3"}}},"commit":{"parents":[{"commit":"04b56f44ef1613b5d77fc76010804ac2afe7cf55","subject":"Add logging details of the http response in the unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04b56f44ef1613b5d77fc76010804ac2afe7cf55"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-31 23:33:58.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-08 16:38:52.000000000","tz":0},"subject":"[OVN] Create the SG rules revision number registers","message":"[OVN] Create the SG rules revision number registers\n\nDuring a security group creation, the default security group rules are\nalso added. This patch is creating the security group rules revision\nnumber registers and bumping them to their first revision.\n\nConflicts:\n  neutron/tests/unit/db/test_db_base_plugin_v2.py\n\nCloses-Bug: #2086205\nChange-Id: Idc6ad29bcac23c2397e32f290addfd1877b8b3e0\n(cherry picked from commit e0ee8bd7726a24747ee5028cb31f9b62cfcfcc29)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3e059497216f8060767e0847f5fb53a1c13709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a3e059497216f8060767e0847f5fb53a1c13709"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
