)]}'
{"id":"openstack%2Fneutron~934652","triplet_id":"openstack%2Fneutron~master~I96152f86595eefa375067ade5a303ffb35c80348","project":"openstack/neutron","branch":"master","topic":"bug/2087824","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-11-19 07:18:13.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-11-19 13:32:38.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-11-15 11:42:41.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-11-19 13:32:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96152f86595eefa375067ade5a303ffb35c80348","subject":"[OVN] Do not update a deleted Logical_Switch_Port","status":"MERGED","created":"2024-11-11 13:41:00.000000000","updated":"2024-11-19 13:34:06.000000000","submitted":"2024-11-19 13:32:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934652-bug/2087824","meta_rev_id":"01af2bd9045a94e7452060711d2dbba37b2f2562","_number":934652,"virtual_id_number":934652,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-11-12 09:03:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-19 13:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-15 11:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-11-12 21:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-15 11:42:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 14:06:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-11 14:14:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-11 15:59:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 14:50:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-11-12 21:46:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-11-14 08:16:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-11-14 08:16:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-11-14 08:16:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"b93939d76c541416064314614a8736ce287c4fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-11 13:41:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80f510ad39fbe3d7cd0c17e2e1854ef59fc33ff9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-11 13:42:01.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"762ba35a4a2bf1e038dfacfa86f2d30dc90c5706","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 14:06:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c1160bfb5f84b92ad98407294ad7c1d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe6acfdad4884f29a9bcd45bc808f246 : SUCCESS in 22m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4c48b1ceee841769267617544fd0b9d : SUCCESS in 18m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0398447b50b21bb22e6b6ca8c34dc5cc38d1b7ea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-11-11 14:14:00.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2024/Arista_Tempest_Testrefs/changes/52/934652/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"209df35023fa7df56072fbef36b185540e8cae01","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-11-11 14:47:04.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2024/Arista_Tempest_Testrefs/changes/52/934652/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"2663219f0149e4b7405bb9f0d109a2e709b27e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 15:59:16.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/438141517513426d8b6b3decc6852944\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/afbfa507d34b4f568325bf59ce3883b3 : SUCCESS in 2h 14m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ee4145467894c4489326afdcd82d747 : SUCCESS in 42m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0146178c8d74b41b37ed8a7df19a821 : FAILURE in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60d08de321c6427ea76f20bb96a2bf5b : SUCCESS in 21m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63ee4ae228fc421198489fb2f077061d : SUCCESS in 23m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4d44150a5bb43fda04ddc7b1cde4bac : SUCCESS in 9m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8cf9e828bebd4ba8b0a0e99e1bef8c2a : SUCCESS in 39m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/94cee0a217474ee3b944bd615ec3e61c : SUCCESS in 44m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f318e15b128e47c3bda29e1f47ec4e5f : SUCCESS in 1h 11m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0b68957c1d8e419489519846b3da6710 : SUCCESS in 1h 11m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4d5dd3d523834e5ead8bca05bfa4199b : SUCCESS in 59m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/75b14718f2494116bccc984dcb35bd89 : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":2},{"id":"162a152f1031be05fe21a2d6f421b94c44f23be9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-12 08:21:21.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0e17e306dc3c77453f74996b345506d812ba1c0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 08:49:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/901c560410ce467c9bd0fa86a191b047\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a12eba45998a48dda06dab2b5fd1ad3e : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fca4e1c95aa74131930fc8c7b5d66cc2 : SUCCESS in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"70de04c4faa65ee3fcbbb60646421368418e3a79","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-11-12 09:03:38.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2024/Arista_Tempest_Testrefs/changes/52/934652/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"8757dff062375fb2d828acc456076d143a0c097f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 10:16:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/294419bbe9aa443f938a504adb36b026\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5ac978ee998944d9bdbb82902448ac95 : SUCCESS in 1h 53m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54c87171549d426ab78b2e1b6bc92e87 : SUCCESS in 42m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/441dafc5ff464c02927c5d2358908ed5 : SUCCESS in 8m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1eafd06886a4d28a3cfa124e1cbbb38 : SUCCESS in 18m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bbb8b58bb4c345c7af37ba7ea58d507d : SUCCESS in 16m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4c1b410b93d40a69a3d450af3afecb6 : SUCCESS in 6m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e59055c2b93446399a3883e29f0fcce0 : SUCCESS in 53m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfb0c55c86fd44348a0b80e8ae1a41f4 : SUCCESS in 51m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/124be04de8d54b1ea2c7bf3c6f62ef6a : SUCCESS in 41m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1f02a2b9923046bb8485650d8efb5f00 : SUCCESS in 1h 13m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d8b64507ca0b404a902ebe7fe7a53301 : SUCCESS in 1h 09m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/90047ebc19af422786d45410e08ba936 : SUCCESS in 1h 13m 12s","accounts_in_message":[],"_revision_number":3},{"id":"fa03ae32428b36a887e3f99fd95acebc2068386e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-12 21:46:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"636d5dd377381085c875131e2bc6e5a5046a3b07","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 00:21:41.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5415ee9cdff84deb9abd58e294caad4d\n\n- neutron-tempest-plugin-api-ovn-eventlet https://zuul.opendev.org/t/openstack/build/0d9b2487cd9a43d8a9919aa8fa61a5ae : SUCCESS in 1h 44m 34s (non-voting)\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b22d3b36a971418d9a5d622600394c31 : SUCCESS in 2h 16m 14s\n- neutron-functional-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e2f6f2eed3de4679a6b50d43001a690e : SUCCESS in 57m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d088bd3af0c844dd97d87d1e0c1cb2a7 : SUCCESS in 1h 22m 49s\n- neutron-ovn-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/d769192d13214641b56b7b3b384f4f23 : SUCCESS in 1h 33m 47s\n- openstack-tox-py312-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a27c66389a9d4df78f4694f50ee97ffc : SUCCESS in 17m 34s\n- openstack-tox-py312-with-oslo-master https://zuul.opendev.org/t/openstack/build/6c65a48c4c0047b7a16ca6bbfc4c69ab : SUCCESS in 22m 36s\n- openstack-tox-py312-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/f310ecebe6b240e8a66bbbd2004e2564 : SUCCESS in 19m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4d766e76c0704881a9bc24ab89a55a2e : SUCCESS in 44m 52s\n- neutron-functional-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/55d080222ae749378f29b64b2e1ea123 : SUCCESS in 46m 47s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/1b7fb3348eab4613b6f66e1bd169b398 : SUCCESS in 45m 12s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/edbe71f1a4074903a6bdf8b029f7eec8 : SUCCESS in 44m 31s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2bf9a304ced44ea3b1f13f10600b6e3a : SUCCESS in 2h 14m 40s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/00ff53ee859a43ca85cf738707a6f09d : SUCCESS in 1h 04m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/c7c0249875e14b5fbdabb3d51e565519 : SUCCESS in 1h 08m 31s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/77a84f339a0240e6b7033d7d889d9e2a : SUCCESS in 1h 18m 10s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e8c78bb9d0484fa0ac33749ead01ff54 : SUCCESS in 1h 27m 20s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/0f3a2d0200594790b5531fcb6c861f1e : SUCCESS in 1h 19m 41s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/0a648c9d6af54750b8d8e445df3991c5 : SUCCESS in 1h 06m 23s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/46eaffe00fbb48f9a1c421be47768f6f : SUCCESS in 49m 50s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/a788694870a64dcda10594130df4c2d1 : SUCCESS in 2h 08m 46s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/596a6e4db97b4e5da0e2d49fd0349039 : SUCCESS in 2h 31m 17s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/7caa22c81a3c4f1ea398b8201c8aa89b : SUCCESS in 1h 32m 27s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ec8cd91845c1416caf30f3d18bddfd21 : SUCCESS in 1h 08m 57s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/083bae115984463dbe0476b2ffbae834 : FAILURE in 1h 11m 16s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/376a2b49ed1f42f1aa948dc9a2c587b3 : SUCCESS in 55m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e8643988e0364204ac0918041ac06b33 : SUCCESS in 55m 33s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7bb30dc3a76340d2aa62e0a990401bd8 : SUCCESS in 1h 28m 32s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/ec7a3aca778e4121b456136f6ed160a9 : SUCCESS in 29m 15s\n- openstack-tox-py39-with-oslo-master https://zuul.opendev.org/t/openstack/build/bff57480871e4c499939ebd86ac42f85 : SUCCESS in 24m 48s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8001db29080446c5877160a4f97d65d0 : SUCCESS in 24m 28s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/cde9aa28bcbb46429ffa32c2d1054f50 : SUCCESS in 41m 34s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/e500cdc0f6a8484aab2d6c1ba6ce7832 : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":3},{"id":"ba321f6faef7f0f0369f073f89230a468e44cf78","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-11-15 11:42:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c476734533668310ea74d64949977a65d0c1a4af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 11:43:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cd78e9ee47df05d88550227885b11e2ccb7214aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 13:12:48.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0326135ebeb4a26898a31a3de8aa9f5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9e322d7682154041a84bfe9b68a179e9 : FAILURE in 31m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad0cf5dfc3c34297acf62b72eb06d73c : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad05f6a1fdbd4ff3bd32d9582f63fb79 : SUCCESS in 20m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2fc66a7798384843a608d14f456a965d : SUCCESS in 20m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7eaa085a6114f16bf2f80848ffa462b : SUCCESS in 6m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/22bd251146804adfb55346af31c3928f : FAILURE in 49m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/25f3031653064ab49bae5c52fd4ab6f6 : SUCCESS in 1h 11m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6da38385af764894b401c1d4471f59dd : SUCCESS in 1h 03m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0ffbb36c80b040ae9ec04312a0601a94 : SUCCESS in 1h 23m 49s","accounts_in_message":[],"_revision_number":3},{"id":"bf7f07b28cf2a45fe99740184a4be58e45ef59dc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-19 07:18:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"75ed4805f8898b9ac3278a988009904c5c7df86b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 07:48:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a6549d33f648179bc0dfb5008896c7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bca74709a2f4ae89362ac249ddce36e : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64ce533d30364510871cdac85a95b01f : SUCCESS in 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"52bc74eb14b48b8c18839ee8141db2cdec732a0f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-11-19 09:26:34.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2024/Arista_Tempest_Testrefs/changes/52/934652/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"889d10bfc7b0f37548f9bdf2ece33523ce3f82ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 09:47:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73ecc937e4674ccb9aebf96c4aa5d58f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/04abb14c417f42de9edc9b4a751b8fca : SUCCESS in 2h 27m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43563351d69547b496a8f4590e935d95 : SUCCESS in 30m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f45b9ba9a9c84f3b9378c062c8564cbe : SUCCESS in 9m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbe5651877854c77a0f68823ea56c36f : SUCCESS in 18m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/768ed0484209425d8b0b89c2d5a2ff55 : SUCCESS in 18m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad472dc7f02940508a5a69931a978638 : SUCCESS in 10m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bd5f6666624c45ef8f54a9ab9ab9fa99 : SUCCESS in 1h 09m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/68f17ba069e54d108e41950270b9320e : SUCCESS in 51m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/298907d0bc16427ca83ed296915f1c2a : SUCCESS in 1h 05m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/479a0e3c5be84a03ad68f28f968c1e56 : SUCCESS in 1h 10m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2bdcad56df07490ca389654ad56dbda4 : SUCCESS in 1h 10m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c86c6ffde897467493fea4fd4401fe05 : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":3},{"id":"74bbff0299c89c85ba613a05ea457975f98da3fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 09:47:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"588a7799055f629f133ee929dad480c772c4f43c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:32:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6ab958f15414af986c26780bb30a4ad\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d95d930872004ded8860530fa12fd4f0 : SUCCESS in 2h 45m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b3d203002142f1ac3359c7b57039db : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c473dea7765140d79d1cec2e8ebf2cc5 : SUCCESS in 19m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/024a7a6f41c742c8834b58ca76ce9cf6 : SUCCESS in 19m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d1e6eeb74a463a9e0aa034f3397a3c : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b68848e92324c88aa49bd2085b0ba9c : SUCCESS in 55m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a18d49531b3e45b6956adcdbd562c9ce : SUCCESS in 1h 00m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9c19b59c5ad943ac88126f90ff7fe944 : SUCCESS in 58m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/74efe0c3cebc4f42b2bc484f4e4dd9de : SUCCESS in 1h 23m 05s","accounts_in_message":[],"_revision_number":3},{"id":"0b3828fb39cb99d347ff956924292c2f71e97198","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:32:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"01af2bd9045a94e7452060711d2dbba37b2f2562","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:34:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c202abc325ff4dbda7fe39d9f9ea5413\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a3c8c27410471b85f63f5913a35053 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5828c43c904d938644144bd204a262bd36c3f516","revisions":{"bb21e04d26417228f7142789caf2fe7e5c91aad3":{"kind":"REWORK","_number":1,"created":"2024-11-11 13:41:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/934652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/934652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/934652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/934652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/934652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/934652/1"}}},"commit":{"parents":[{"commit":"28b2e819e3196cf46cf2a644f18439f7c0169cc7","subject":"Merge \"Fix dependencies for mypy tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28b2e819e3196cf46cf2a644f18439f7c0169cc7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-08 22:32:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-08 23:24:19.000000000","tz":0},"subject":"[OVN] Do not update a deleted Logical_Switch_Port","message":"[OVN] Do not update a deleted Logical_Switch_Port\n\nThe method``OVNClient.update_port`` can be called from\n``update_port_postcommit``; at this point, other concurrent call\ncan have deleted the Logical_Switch_Port. This check prevents from\nfailing the API call that was issued first.\n\nCloses-Bug: #2087824\nChange-Id: I96152f86595eefa375067ade5a303ffb35c80348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb21e04d26417228f7142789caf2fe7e5c91aad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb21e04d26417228f7142789caf2fe7e5c91aad3"}]},"branch":"refs/heads/master"},"ee9cce649c65a8cbc57ba63200d2027aeeea0e31":{"kind":"NO_CHANGE","_number":2,"created":"2024-11-11 13:42:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/934652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/934652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/934652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/934652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/934652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/934652/2"}}},"commit":{"parents":[{"commit":"28b2e819e3196cf46cf2a644f18439f7c0169cc7","subject":"Merge \"Fix dependencies for mypy tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28b2e819e3196cf46cf2a644f18439f7c0169cc7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-08 22:32:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-08 23:25:35.000000000","tz":0},"subject":"[OVN] Do not update a deleted Logical_Switch_Port","message":"[OVN] Do not update a deleted Logical_Switch_Port\n\nThe method``OVNClient.update_port`` can be called from\n``update_port_postcommit``; at this point, other concurrent call\ncan have deleted the Logical_Switch_Port. This check prevents from\nfailing the API call that was issued first.\n\nCloses-Bug: #2087824\nChange-Id: I96152f86595eefa375067ade5a303ffb35c80348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee9cce649c65a8cbc57ba63200d2027aeeea0e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee9cce649c65a8cbc57ba63200d2027aeeea0e31"}]},"branch":"refs/heads/master"},"5828c43c904d938644144bd204a262bd36c3f516":{"kind":"REWORK","_number":3,"created":"2024-11-12 08:21:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/934652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/934652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/934652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/934652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/934652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/934652/3"}}},"commit":{"parents":[{"commit":"28b2e819e3196cf46cf2a644f18439f7c0169cc7","subject":"Merge \"Fix dependencies for mypy tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28b2e819e3196cf46cf2a644f18439f7c0169cc7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-08 22:32:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-11-12 08:21:20.000000000","tz":0},"subject":"[OVN] Do not update a deleted Logical_Switch_Port","message":"[OVN] Do not update a deleted Logical_Switch_Port\n\nThe method``OVNClient.update_port`` can be called from\n``update_port_postcommit``; at this point, other concurrent call\ncan have deleted the Logical_Switch_Port. This check prevents from\nfailing the API call that was issued first.\n\nCloses-Bug: #2087824\nChange-Id: I96152f86595eefa375067ade5a303ffb35c80348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5828c43c904d938644144bd204a262bd36c3f516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5828c43c904d938644144bd204a262bd36c3f516"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
