)]}'
{"id":"openstack%2Fneutron~935039","triplet_id":"openstack%2Fneutron~stable%2F2024.1~Ifb5ab94ca68a9ab84407b54ac632164860b7a3a8","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2086787","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-11-27 10:04:31.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-11-27 14:53:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifb5ab94ca68a9ab84407b54ac632164860b7a3a8","subject":"[stable-only] Change the agent AZ query by agent type","status":"MERGED","created":"2024-11-14 07:04:26.000000000","updated":"2024-11-27 14:54:34.000000000","submitted":"2024-11-27 14:53:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935039-bug/2086787","meta_rev_id":"cf116feae48217b0b0b8304d9d3be67caaa71b48","_number":935039,"virtual_id_number":935039,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-27 14:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-11-26 15:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-11-26 15:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-11-26 15:54:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 07:35:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-14 09:20:10.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-11-14 09:25:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-14 16:14:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8535d81520bb730068cacded23158e61c6fc46b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-14 07:04:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79568d108d50fade3e5e0b6902d1ab3949d84298","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 07:35:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/335494bba2ca478e8d2dd3becf3c6b8d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6c44fc7950343d8aa5e4ae22b0ea926 : SUCCESS in 23m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c560fbc973884a869476e3018461b7b9 : SUCCESS in 29m 01s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d004493fcc604a49af4d2505858259dd : SUCCESS in 23m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59bd874432b36dd9246c797358ed84721e45226a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-14 09:20:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b9fe0a8d421252bdebddf350e88bdd2b66bcfb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 09:25:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/939b08e54db74d92a09c997a74c79a43\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/7f089969751e4ea59dede0dfa4471234 : SUCCESS in 1h 43m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/2f2203a9889c4c9faf1a20492d544e5d : SUCCESS in 1h 41m 01s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/a5266fd6d2134bb299a6fb3dc9e6b1a9 : SUCCESS in 2h 20m 11s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/a21727468f74429ab52475ce110af0c6 : SUCCESS in 38m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a51fcada1ad4f62b49a42693e79b7a2 : SUCCESS in 40m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e95221e589ed4620af995862a3ac32d8 : SUCCESS in 10m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec9dafe8e67d43dd8244f7f443c0ec33 : SUCCESS in 25m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/166f2c0d0900439b9f02fb889265dc83 : SUCCESS in 25m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa3dcfc763c4f71a13a0152d180e6c5 : SUCCESS in 10m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5c0700fa4c014ab8b5fc73e997f21756 : SUCCESS in 34m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4df2b7922da545d9b84697ea6c546680 : SUCCESS in 1h 01m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cae26c59765744d3b0e8fd8b0ebe6065 : FAILURE in 53m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/37f060d134124b0eb195dca2ab0902de : SUCCESS in 1h 30m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c87cca02a8374f729093d6b73d826e31 : SUCCESS in 51m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/88b91d5ff5a64a65b9307dca9ac289ac : SUCCESS in 1h 03m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/54457d0951584c28b93c281eee659572 : SUCCESS in 1h 18m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6638e9612679492cbb5d40834ef0c351 : SUCCESS in 1h 15m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a47c08cf393444bea60e7693c6745fcd : FAILURE in 55m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/220c7dc44b074b8c9fc2caf94a860e3c : SUCCESS in 1h 00m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/748e2c7a7f89421eb3679a1f7cc75754 : SUCCESS in 1h 29m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1aa1128d29004fc9bc09fb523af8f3fe : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d34168f9b19879bda636346b1c927737625f2879","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-14 11:03:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b1b10f82d16a0b7882596f9edeb263a083c6f4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:39:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e79f8c88fd14e7394b937ac28839dac\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3270828d99b45048d96e1e87acb3a2f : SUCCESS in 25m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c70bbcd5196426792f945853caa3b06 : SUCCESS in 30m 58s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a5467851c8c41e48d6f8a491875eda0 : SUCCESS in 21m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4259c54a049c1b3d51dbf5dba9556e46b3e98421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:53:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5064ce7fc24472b8af9ffa64230ef79\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/3f6d0bb310c74a949aeea5a1c40e3b01 : SUCCESS in 1h 05m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/982b9e124f0a4a8282a7c9ec3230f1a0 : SUCCESS in 1h 14m 06s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/af7f89c720dd4105ae6f4a821bea68b4 : SUCCESS in 1h 46m 32s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/bf8d839294334260b0c2012e5e9f69d1 : SUCCESS in 29m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1b440af62af409cb0428a4a724e90c9 : SUCCESS in 42m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f2f1195ed04b4d96da184743adf66f : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c85cdc0e90246bd949cafdecfe5ebc0 : SUCCESS in 22m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b41e89eede12403bb9e255e7ed1d9d8c : SUCCESS in 18m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8d453ff0b340d088e0a80932e9c21c : SUCCESS in 12m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/334a8a5cf3144c26ba67ca1cde8c364b : SUCCESS in 32m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f11bb2e8e4ee42f89345a7fde86ef91f : SUCCESS in 1h 13m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d0efdf6af924c0fad79133ad3a2164d : SUCCESS in 54m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5ee0056cb934a0cbb379fa06c78d97a : SUCCESS in 1h 37m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a21c1feed1064d6e83501d8e4716a4c0 : SUCCESS in 46m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/408d47b24e424215ba9465047cb7f9ca : SUCCESS in 50m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9169a5330577439abcfee029365bd8a1 : SUCCESS in 1h 33m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/85133fa7432b48e1846578a3cae5213c : SUCCESS in 1h 17m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1095c7c35eeb493fac818d8690113533 : SUCCESS in 54m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b3e8d00340dd4e5782b932d767079a6b : SUCCESS in 1h 07m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f0aee8038c81432eaaec34d1617aacbb : SUCCESS in 1h 28m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/42a20a58780c403583ed23f020867ec5 : SUCCESS in 1h 16m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5e884f3be3d3c95014de27e74d93f0eb5f3c84cc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-14 16:14:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6078af2687b61da25fb56d30f41750d7ca81eef2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 16:15:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f6018bcb867a3b709ad44908488e79c803388bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 18:50:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e58dca9f2704aa78d44d84ef40abbbc\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/bd6d23fc1c854228a6462e6c8cd386a5 : SUCCESS in 2h 16m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d66214b0bca2498bafd0dc6f1a66d259 : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/029843d9cb8640e4ab13c6267e85256c : SUCCESS in 23m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e20b90f7eef49fc8e1c12815f9ad40b : SUCCESS in 18m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43cb51bd96e0471f9c3b111f8a34b7f3 : SUCCESS in 9m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eab2a05bb91e4078b020b61d5f91810c : SUCCESS in 56m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/565aaade74fd4f678afb0e1512bc71a7 : SUCCESS in 1h 31m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/73f2abbf7a4d40839c7e6c6cb659ec35 : FAILURE in 25m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab60e44033a541d596890045abb62dd1 : FAILURE in 36m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/57f61a4976ac4193ad040f6aa0bc8acc : SUCCESS in 44m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/703be4dc33ed4a59a7284fab832dc6b0 : SUCCESS in 55m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/05f8931fd1264ffea80d82a5af6bbd11 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"08ad4c07cc677de63ad851ed30032c4816f8a4f8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-15 10:24:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f21708eb4477576206ed98e73b9ad1a0a104d4b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 10:53:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf704b1fc03479e84d7b0a5ee8da67b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb8852d75d2140a2956c8662494b3542 : SUCCESS in 25m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/31c718a54b3a4ff2a94839e8ae39c114 : SUCCESS in 26m 08s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5924f129b4c2466ea239cd0dd56d9a11 : SUCCESS in 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ee556f5177c369e2d99b0f390f4a976595d2078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 12:32:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8f4013101ed4d2c8190376b14879dfb\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/0fa9e3880f4842e5a3c17bc3aca01138 : FAILURE in 24m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/6f897c139ca04d4780bda6249514d3c4 : FAILURE in 26m 25s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/0b7140de2950450da57ed049b8ae6c45 : FAILURE in 25m 24s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/eb0ce17ecf304684bb0560e5223cea63 : FAILURE in 37m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b82780be4d2d4975a2789175b2c1a7c1 : SUCCESS in 32m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d844f8560e99432ba0387cfdfe6b705d : SUCCESS in 9m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f1b8bfde1bb441b85ab8cfb93f241af : SUCCESS in 26m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c76bc1181619429dbf15a417c25364d3 : SUCCESS in 24m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cacbb8d4e998459f91e5cfadd3d31f12 : SUCCESS in 6m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e7c8469a05b443c4a41a0a3dd200341a : SUCCESS in 1h 01m 13s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7d51848abd40461bb3bd2244c792cbba : SUCCESS in 1h 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db967fd5b49e4100ba53c999cb480861 : SUCCESS in 44m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e964a4b7355840d097fc695686887a66 : SUCCESS in 2h 06m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/660e0988f33d4b178f1fbaeeeb8e3047 : SUCCESS in 1h 05m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/97cfee0d2ca745279b02a08f5b9a70c8 : SUCCESS in 1h 04m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fc34555762fb414397616f7214d8a86a : SUCCESS in 1h 29m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5afe87fe71ea458ea06b56bd0057c8a6 : SUCCESS in 1h 09m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d8e94c9754074e0d8bf5c9dca0128846 : SUCCESS in 53m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d8937397b9ca4445bb8640c57c715255 : SUCCESS in 1h 15m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f46ae0eb45c146e980cd77b06845c3e6 : SUCCESS in 1h 31m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/04cb5847ec6142829785aa4788d98666 : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"43c2ce1527f87d3d4fa874a21f2f3fd7a61c981c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-19 10:35:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a0a7b568ebf400d8f0f6eb504b4af3a31e56d38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 11:28:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d91e11ee8f02458f9b14e4bf9c7f8ecf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a9f8081caab4af082812f6a60fafd1d : SUCCESS in 41m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5b30f356e684b1e85dff00b5d619566 : SUCCESS in 49m 02s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/74e7e420fe044f19aa706c5d283ae7e4 : SUCCESS in 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fc3bc0ae54478db74b8fdf8bfd38b4b2e002e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 12:29:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0b8984a556d4d938fc5a82ad08422f4\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/ad86711057a846fc932c6f84bde1115e : SUCCESS in 1h 06m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/0f8b0cc10c5449e68bab9dd160cbd579 : SUCCESS in 1h 24m 52s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/d9515c9f40d04745af20ae18962f7cfe : SUCCESS in 1h 37m 33s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/b562cc4a5d664097ac048ef16ababf95 : SUCCESS in 30m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a27e867f478244d1ac0858295fd38231 : SUCCESS in 40m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec1111635ca74a9f876baaf30b50e4b6 : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97818eda99094ad8b8b365713d01805d : SUCCESS in 28m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78197dc7c3764390ac345bba8d03f223 : SUCCESS in 24m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054f5096f120403897ea4a5ad3459916 : SUCCESS in 11m 22s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f00d3b1dcf1a4673bc6e8454af874797 : FAILURE in 16m 45s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d6d872e53c7348318d302b18b3ddb3f1 : FAILURE in 25m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/998833715698408689a63106d0150f99 : SUCCESS in 54m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/39c727de7ac84f1d8be9d376fd573f73 : SUCCESS in 1h 47m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ba267b1535f4c319fd7f119aec65753 : SUCCESS in 1h 01m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4b79c19206434788a7394aad743ca391 : SUCCESS in 1h 05m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3c1c9b84b7aa41acb3978b3fbc0b4fc6 : SUCCESS in 1h 36m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/45722e3e568d419386a2cce75a351ba8 : SUCCESS in 1h 18m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5fd69cd6c9c347b0a27556744ce242e4 : SUCCESS in 49m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/df4e494ca6b14c85ba55cfcec77bbe2c : SUCCESS in 1h 08m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5c262e0f77274ca49db885e82561914a : SUCCESS in 1h 19m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4aeb5aa710af4d8a90faba785367be00 : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e5a604bbad22138f60ac263007d2e7d5b2508e24","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-20 10:28:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2237aba39be083fe7ae4063cf95c8cff592779a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 11:02:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be3157f20cbd4d1e883e199ff24e12c6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b55b2c04d664780b92f7be1b8327371 : SUCCESS in 26m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0bf90d82475b48df86214107023e7201 : SUCCESS in 28m 51s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad815df5cc8c4def8d230a478282372b : SUCCESS in 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"988622c9acf28f973945a247b1dfcf7fbdf99447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:39:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/931db726b74a4524afc2ca0d60f441a6\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/34441e55742e40ce9b8835c9f72b8d0c : SUCCESS in 1h 24m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/ac4b96025a69438f8fdaf29280d9222f : SUCCESS in 2h 05m 59s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/7331956304bf4b5baa7a5c7aad4a07a2 : SUCCESS in 2h 07m 59s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/7d921381b2ad4736a742df025b3012c3 : SUCCESS in 28m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/117282e3d6254eeaa95e6a2550d9f94e : SUCCESS in 37m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a5097be0f114aa89a5216018e338d14 : SUCCESS in 9m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71b161d4bf204760804fed188ad44202 : SUCCESS in 27m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f710693e4fd34e908dd1d713a2201070 : SUCCESS in 22m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea381fdedd2e4218928c63876c5dd7e1 : SUCCESS in 10m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/54b48fd15d154579be536efb16ff7ae0 : FAILURE in 23m 31s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/71bbd5fd865b44cf968a5bdfdc3ab684 : FAILURE in 17m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a96bfe9341845eb8257b002ed91715d : FAILURE in 41m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/145fff1d91634b1abd0525b6b180ae23 : SUCCESS in 1h 36m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a117207774a2496eb76bc0d359e00daa : SUCCESS in 58m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e061fe29e2a94fcf9867e7d8b847bf7e : SUCCESS in 48m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3fff2234bfb946b18166a16d6e6d77af : SUCCESS in 1h 26m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4df170d4258c47209daab94932dfbc91 : SUCCESS in 1h 25m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/83486b6a40084755a52fc8159ac45ce1 : SUCCESS in 52m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/130f748860044a2c912b07f74c01e848 : SUCCESS in 1h 10m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/066b6a1d906a44b9a830aaa0c7256266 : SUCCESS in 1h 35m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9bbe7061484148b8883d37dd6c65af6b : SUCCESS in 1h 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2e53b48908a0e7a723ee831453f8241b53a592ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-23 13:20:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"90e56a544f8add516da745b13815e0151eaa02b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 14:12:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d49310a20bcc47539406f173c9ee0840\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ced0b9d3e93f4dc8b761a3643ab2259a : SUCCESS in 48m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e420eb0424b64dee8702af81d8020dc0 : SUCCESS in 49m 04s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d50a76c2aa2e49dfb0ff949749009622 : SUCCESS in 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"25cbfb9dfb06376a32329e87bed1154bbb61b3ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 15:13:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c953dc5441c45bb98ee41a98a121922\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/8c0586831ebc463d9a24b9b17e8d25f7 : SUCCESS in 1h 15m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/8cb935999f8941cb888c90e7cec0f509 : SUCCESS in 1h 41m 57s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/6fcc9177c2f8449c873cadf11ac81d25 : SUCCESS in 1h 44m 34s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/70e2d3cc6289470c8237dd5819fe4c12 : SUCCESS in 42m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18da1c86e5e24e39b09b4b8458a1ffa2 : SUCCESS in 37m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d7732079d9b45998adae368014a6b09 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dbcc7b9d441433c85ee41d471f57732 : SUCCESS in 27m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c77f0a52d81144fa8f5ffb402b4c3033 : SUCCESS in 23m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a006b5d79e2a4ef9baaf17dcb3fde862 : SUCCESS in 11m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0372d7db19f34165ad76d429997b53b5 : FAILURE in 25m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/faa8c762886f4752b91ce6e4b81bc87e : FAILURE in 33m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1bd1f15b9ea84c31a39380be5b19bca4 : SUCCESS in 41m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad67d159baab4e3aabac892bd662d9a2 : SUCCESS in 1h 18m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca7ef74a66d34cde994c641eea6f6725 : SUCCESS in 51m 42s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/07fd5782329445458038b5e326824b83 : SUCCESS in 47m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/da0bd12587fb44dcbfbd61970ab3d018 : SUCCESS in 1h 27m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb61f9d057dc4234a83db915beb723e4 : SUCCESS in 1h 27m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f5bd0b11de6845238ae6ed9584a46c9e : SUCCESS in 38m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fdeb5990520844a383ba300881a1a0ca : SUCCESS in 1h 10m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f9dd13a46d854a0fa49be3339c97a963 : SUCCESS in 1h 03m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/75ac5b71c913413fb059d9824cfd9aec : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d79fcd6791517c71f9bdb2672df5d69c50bac7d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-26 15:45:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cd9e56986827fa00fe18706d5214a786849588d7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-26 15:54:31.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0c072a510be7dbd5054e164d1f1c3a9040c42bf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:52:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6409b01cbce48d3bd614b337b379c2f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f652a7fc5ff740999693cae5a9987daf : SUCCESS in 37m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9e6399778b64fc481999ca20a14bae2 : SUCCESS in 29m 46s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15fd159debf148c48084b15160b63a87 : SUCCESS in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e4a2a11ea76e4256a01e185127847f2522b578a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 17:28:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18510e5b331941e18b29d840dd255e36\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/0c4391b39ec84329adf4659780995d9d : SUCCESS in 1h 30m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/35c93e94f7844103a9847a91c19c8f60 : SUCCESS in 1h 25m 20s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/b38920b308a6446fb427ec232a4c8481 : SUCCESS in 1h 25m 50s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/ad14d60bce854c349efc61b7e6dca6c0 : SUCCESS in 29m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26a2819841964d45833f0aabf10f6ce9 : SUCCESS in 37m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457ad89346094239845ae1ba7c7b0e10 : SUCCESS in 9m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56440fbad9144f31a73c852d69025f72 : SUCCESS in 24m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8934a25129f4eee9de4611fc1cf2cbe : SUCCESS in 23m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0fb7391c0a4551af769083a7801551 : SUCCESS in 11m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7512e19e9b6495fb7d353110dea6423 : FAILURE in 47m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/161e9187894e4a09932a0fcf70da6827 : SUCCESS in 1h 36m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1819b8a971924419a7806ce47d8be828 : SUCCESS in 55m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b736bfe9905b4cbe898e4e8d9a2f2cf6 : SUCCESS in 1h 02m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/14e28dfeac0344e69a222b2b4d7a93f5 : SUCCESS in 1h 32m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5ed17326e10f4bd8ae753e54da8305fd : SUCCESS in 1h 19m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6681e615c2dd45b296b2d9a1e4fce898 : SUCCESS in 1h 01m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7afca421aff84055b0387c1461378123 : SUCCESS in 1h 29m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/78dd9e5fb29145efb9bec87e128c7fdd : SUCCESS in 1h 30m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9404435dccdf4b16b54d7baccead588b : SUCCESS in 1h 13m 23s","accounts_in_message":[],"_revision_number":3},{"id":"740b3a9446a49c1d6e6da31fc0a70a194bbaa1bb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-27 10:04:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7fd8ae38f23974ab5b9138ac9691e827f6deacc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 10:50:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/704b7022c79c45acb9098c304623d182\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b415ac6c28c9413aa577563542a8ca33 : SUCCESS in 43m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/462a4edc79e2449184194939f0393bc2 : SUCCESS in 38m 09s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/36ba4dbb2f8a45f6aa682d7a1c45f9d0 : SUCCESS in 30m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c90894814cf273290c13ddf3f02b2a659eb77869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:12:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48ed97fa2d11432f86a8652f15c27d1c\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/4ff3dc28bb7343d58ec527f64b526906 : SUCCESS in 1h 51m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/5e8309066f2f473dbef099e8da474102 : SUCCESS in 1h 28m 03s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/f68b9f83cc7b4fd499379df9c96871d7 : SUCCESS in 2h 05m 04s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/b616c53c960a4f42a9c46cfa5c681530 : SUCCESS in 45m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b7c1e1895fd48dc9efcea3bb1742ecf : SUCCESS in 35m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b42efb78609043069bb349ab8a0d65fc : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5d9278e920d4f1a910d79b08f516570 : SUCCESS in 26m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f4f5bb8c1f9424eb96b3310fb5270fc : SUCCESS in 23m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3642453c33c34521b8537400c4a553d6 : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/373d2dea64d143608a030dfa1f7019ea : SUCCESS in 54m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/50d38b714e2247f08b866664b55fb15c : SUCCESS in 1h 41m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dca1ef5f4d194ab89da0f94df6ad70c4 : SUCCESS in 1h 02m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3158c79a67d04a5b8560b91a6092615e : SUCCESS in 56m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b087888a8417451db1a4acfe34522cbf : SUCCESS in 1h 26m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/47364b1163144ce1af38c81b2d40fa92 : SUCCESS in 47m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5da529bb56cc46f0a16a3f079e2fd51c : SUCCESS in 1h 00m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e5db8abb164b4b469abe6aafd2491c97 : SUCCESS in 1h 24m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c759ad1c415b4a6b9967f923bb1ef2d8 : SUCCESS in 1h 33m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9f4e6393f2d34113b5119bc1683f1d6a : SUCCESS in 1h 16m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ddb58a617c676a23237829200e492124b00c21f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:13:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"389dfd4a177f5fcc54cbaf3adf0be0f861b63c47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 14:53:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f164653fea4721a08ab554576a56b4\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/77249bd918c949769d5943c05e2f3f36 : SUCCESS in 1h 44m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd9a24f164f641b9afdb9e3b3dee13bd : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e41bb5fcd944bba876b6dd01cc8677f : SUCCESS in 27m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29fe42df84e24536812a7e339f78748c : SUCCESS in 28m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e22e10547940bdac1d5162234ac660 : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/008d5c87b1a3422386921b6cca45d155 : SUCCESS in 54m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/acc41171bf08447e929f417b30fbaa7f : SUCCESS in 2h 18m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4f0c9147668c4c20bc4366c9214df65e : SUCCESS in 48m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/21fee52a557347b3bd2f48a6b13389ba : SUCCESS in 49m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5ea8674b33554fa7b5d6de03e8d5d64c : SUCCESS in 36m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3e9dc27c147f45b5a781b8e974c3d672 : SUCCESS in 47m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d8247ad85528449f86b84616a9af5984 : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8e9b09bddff142bcec1913ac53bfe24166a57da8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 14:53:13.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"cf116feae48217b0b0b8304d9d3be67caaa71b48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 14:54:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0455281a364a44578709c7c40a9beb70\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2640c279bb4f4ea78c6365e83a887b8f : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3273f11115af8b4f10619b6b1161c7f7a69144de","revisions":{"bbdb3c3cd95cd03532eeb2f8d34f5acd9db4b8d2":{"kind":"REWORK","_number":1,"created":"2024-11-14 07:04:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/935039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/935039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/935039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/935039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/935039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/935039/1"}}},"commit":{"parents":[{"commit":"bcf248e6eaea6b8dddc8f0af45054649ca8d0627","subject":"[Fullstack] Use only one keepalived-state-change thread in L3 agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcf248e6eaea6b8dddc8f0af45054649ca8d0627"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-12 10:00:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-13 19:58:17.000000000","tz":0},"subject":"[stable-only] Change the agent AZ query by agent type","message":"[stable-only] Change the agent AZ query by agent type\n\nIn PostgreSQL, it is needed to add the field used in the query\nfield (\"select\") clause in the \"group_by\" one. This patch changes\nthe scope of the query to only select the filtered fields (availability\nzones and agent type).\n\nBecause the PostgreSQL support has been dropped in master branch\n(Epoxy, 2025.1), this patch is only for stable branches.\n\nCloses-Bug: #2086787\nChange-Id: Ifb5ab94ca68a9ab84407b54ac632164860b7a3a8\n(cherry picked from commit ebafc58e692c1533137be2fcc8a8027e262f84ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbdb3c3cd95cd03532eeb2f8d34f5acd9db4b8d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbdb3c3cd95cd03532eeb2f8d34f5acd9db4b8d2"}]},"branch":"refs/heads/stable/2024.1"},"0348800d3096694980b054c4f39435846157307b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-23 13:20:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/935039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/935039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/935039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/935039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/935039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/935039/2"}}},"commit":{"parents":[{"commit":"18a43641fa9349e8571384d511874d92eb24b81e","subject":"[stable-only] Change reference of 2023.1 branch in skip jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18a43641fa9349e8571384d511874d92eb24b81e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-12 10:00:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-11-23 13:20:58.000000000","tz":0},"subject":"[stable-only] Change the agent AZ query by agent type","message":"[stable-only] Change the agent AZ query by agent type\n\nIn PostgreSQL, it is needed to add the field used in the query\nfield (\"select\") clause in the \"group_by\" one. This patch changes\nthe scope of the query to only select the filtered fields (availability\nzones and agent type).\n\nBecause the PostgreSQL support has been dropped in master branch\n(Epoxy, 2025.1), this patch is only for stable branches.\n\nCloses-Bug: #2086787\nChange-Id: Ifb5ab94ca68a9ab84407b54ac632164860b7a3a8\n(cherry picked from commit ebafc58e692c1533137be2fcc8a8027e262f84ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0348800d3096694980b054c4f39435846157307b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0348800d3096694980b054c4f39435846157307b"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"},"3273f11115af8b4f10619b6b1161c7f7a69144de":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-26 15:45:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/935039/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/935039/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/935039/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/935039/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/935039/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/935039/3"}}},"commit":{"parents":[{"commit":"098b94906a4a48f2739d317a5f2c24b6f35c651f","subject":"[stable-only] Drop skip-level jobs in the CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/098b94906a4a48f2739d317a5f2c24b6f35c651f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-12 10:00:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-11-26 15:45:37.000000000","tz":0},"subject":"[stable-only] Change the agent AZ query by agent type","message":"[stable-only] Change the agent AZ query by agent type\n\nIn PostgreSQL, it is needed to add the field used in the query\nfield (\"select\") clause in the \"group_by\" one. This patch changes\nthe scope of the query to only select the filtered fields (availability\nzones and agent type).\n\nBecause the PostgreSQL support has been dropped in master branch\n(Epoxy, 2025.1), this patch is only for stable branches.\n\nCloses-Bug: #2086787\nChange-Id: Ifb5ab94ca68a9ab84407b54ac632164860b7a3a8\n(cherry picked from commit ebafc58e692c1533137be2fcc8a8027e262f84ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3273f11115af8b4f10619b6b1161c7f7a69144de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3273f11115af8b4f10619b6b1161c7f7a69144de"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
