)]}'
{"id":"openstack%2Fneutron~935652","triplet_id":"openstack%2Fneutron~master~I56cefac5f44487717e298846419cac8a51658798","project":"openstack/neutron","branch":"master","topic":"bug/2084782","hashtags":[],"change_id":"I56cefac5f44487717e298846419cac8a51658798","subject":"WIP \u003d\u003d [OVN] Make provider network traffic centralized always","status":"ABANDONED","created":"2024-11-19 13:24:12.000000000","updated":"2025-01-02 13:40:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"3497029511ac094eb0869069f23df01d0989cbd7","_number":935652,"virtual_id_number":935652,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-11-27 13:13:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-27 14:26:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-11-19 13:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-19 15:20:21.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-19 15:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-19 18:14:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-11-27 08:08:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf45f13d3004f64c187139bee40b221d8e8bf261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-19 13:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0a0c2a40328ddf17a123dff2026e2eed3d57142","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-19 13:24:21.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"2b8ed3ffa16a4d05e2d190399a6db99e728dc878","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:57:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d247882810b410181a728e0f65f96fb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d6f1b2dd7184764b4a475af6800e024 : SUCCESS in 25m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37cbfcd0bf5d4bf6b9b357a57ba94b77 : SUCCESS in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12e63350e34944057f738a6e405f8cf5771e651a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-11-19 15:20:21.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2024/Arista_Tempest_Testrefs/changes/52/935652/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"aeb6e0d2ac4d96b886bf4957b9376a5e45160a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 15:55:20.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/346eba0cd1a04009ab0a84b85c368389\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6714e475a8a04590bd9a97ef6e3e74d1 : SUCCESS in 2h 29m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baab3f4973c042cf819daa721addcb9b : SUCCESS in 27m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a561f8c64a2d4053845f9b80945fe17a : FAILURE in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20d0cb07a92b4039b750d9790780e528 : SUCCESS in 22m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/772fb12a4b584168b2a7d2f934d9cb11 : SUCCESS in 17m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848ed01eef824e31924cd0b728b2390d : SUCCESS in 10m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/240d4f000a3849658f2ffbc7697f3a24 : SUCCESS in 1h 16m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fb4304603389437f9e2aba5add818d48 : FAILURE in 55m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/50d41d7b987f4e2994bd31759030575e : SUCCESS in 1h 29m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1a9401e2cee84cdeb1d6cbf5cf55c395 : SUCCESS in 1h 27m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8cf55c5db2ac48518e741da4aa6741d3 : SUCCESS in 59m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e64f49f915344001b30aee2b207e6537 : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c18f4f537fd5f148cbab8340b3e2174350e53d12","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-27 07:36:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cf761e253509bc3b3fb3eccf597b23522cbeff2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 08:05:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f7ed7d936e54467a8d5ab04aa0a2150\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/708682dc88c9408fadd8b7df325e9bd4 : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/837e0adb5a07435ea12627e104d10aba : SUCCESS in 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad915295f306d0332b369fdb1dec56ea87404ca6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-11-27 08:08:40.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2024/Arista_Tempest_Testrefs/changes/52/935652/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"03cf401395e64d2af7d666ce73d2f648b0df3da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 09:27:24.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c3442e67d054bc0b8e45b81cfaab910\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/223988e008424cc2b34e8fb23b92d135 : SUCCESS in 1h 49m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58d089ae048a4bc585b8730fcb353ec8 : SUCCESS in 30m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff5ad4402df4a96820770d63d2eb76d : FAILURE in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79ea4a926b244c8aa1f7027434216665 : SUCCESS in 22m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a26472f7f31e4ba495947f33d0665128 : SUCCESS in 20m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f896742fad465da746ef071811f8af : SUCCESS in 5m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a9e6e61af6a84c0b89bb9a7e7f9edfa1 : SUCCESS in 50m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dc0e5c262c484e62ba4edb7d013f7eda : FAILURE in 37m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fd6094ab7db4b7c9b69adcf0741cf71 : FAILURE in 48m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e7d70aeae9374a5b83dc8ab618768994 : SUCCESS in 1h 04m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/05f95e9ac5aa4425b1a319d60b887ad1 : SUCCESS in 33m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/856d4493123d45cdb015d774d4b54203 : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":2},{"id":"70de78c5b8ba8b2c382e92b8268bcac7c8d63a54","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-27 11:43:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0a5fee70863a2e945b7f88e0d4b1597dc264132f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:44:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ca11862ee3d494bb37c858b10a084b2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70d99067b6d4469dab035b21d314ea12 : SUCCESS in 33m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/361cbcb5abd341629720f3cdbb1d8e8e : SUCCESS in 25m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"453fa6fbc794843be80ad804649ffa4a4bee50ec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-11-27 13:13:30.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2024/Arista_Tempest_Testrefs/changes/52/935652/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"0c38db20312cda3dc139353574ecd3f0b4d5334e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 14:26:02.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc508b2fbbd64099a0915c552c33cfc2\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7a4cfce4abe24bb1a73c80cb7c0b1293 : SUCCESS in 2h 40m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0e3c27a718b4ffe8714bb5e0bba9edc : SUCCESS in 30m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/224c5b388fa04c4d928bb4112c5e4a7a : FAILURE in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64a7f0f6c5dd430c9476225764a60ed9 : SUCCESS in 27m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d5b02ca36e947c59c036ca88612a677 : SUCCESS in 22m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6968d31989c44122a646414663eeb9b5 : SUCCESS in 10m 09s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d0b06e868d7a4ee4848dc009f6dfeb87 : SUCCESS in 1h 04m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/30d914849a6249d7989618f5892dd4f9 : SUCCESS in 51m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/59edfeb213e64d56afe5f09d78426a06 : SUCCESS in 48m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/151b8fe2ae664b919814b0920e72bf92 : SUCCESS in 1h 06m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd6375f6860643adb46e26bed8090f7c : SUCCESS in 1h 03m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2f10fb81efe54689b6a67171f3cee3e3 : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":3},{"id":"3497029511ac094eb0869069f23df01d0989cbd7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-02 13:40:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"194549b72dda9b069cc68e19fdce9fc5e99a2499","revisions":{"da1ccaa2af2ec11114db960025503b3de6259e0d":{"kind":"REWORK","_number":1,"created":"2024-11-19 13:24:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/935652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/935652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/935652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/935652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/935652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/935652/1"}}},"commit":{"parents":[{"commit":"3441c96fc4e7cc5c543cd3ecb7db4fc5c5f4e238","subject":"Merge \"functional: Allow to override OS_FAIL_ON_MISSING_DEPS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3441c96fc4e7cc5c543cd3ecb7db4fc5c5f4e238"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-19 13:23:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-19 13:23:57.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] Make provider network traffic centralized always","message":"WIP \u003d\u003d [OVN] Make provider network traffic centralized always\n\nLP 2084782\n\nTODO: everything\nChange-Id: I56cefac5f44487717e298846419cac8a51658798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da1ccaa2af2ec11114db960025503b3de6259e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da1ccaa2af2ec11114db960025503b3de6259e0d"}]},"branch":"refs/heads/master"},"61aabd3b3a6a31178b962acfe40d475ee0d79232":{"kind":"REWORK","_number":2,"created":"2024-11-27 07:36:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/935652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/935652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/935652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/935652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/935652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/935652/2"}}},"commit":{"parents":[{"commit":"54dba256eaee43821d9b74e2d4c8b46abe19c308","subject":"[FT] ``Queue.external_ids:queue-num`` should be a number","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54dba256eaee43821d9b74e2d4c8b46abe19c308"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-19 13:23:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-27 07:35:46.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] Make provider network traffic centralized always","message":"WIP \u003d\u003d [OVN] Make provider network traffic centralized always\n\nLP 2084782\n\nTODO: everything\nChange-Id: I56cefac5f44487717e298846419cac8a51658798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61aabd3b3a6a31178b962acfe40d475ee0d79232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61aabd3b3a6a31178b962acfe40d475ee0d79232"}]},"branch":"refs/heads/master"},"194549b72dda9b069cc68e19fdce9fc5e99a2499":{"kind":"REWORK","_number":3,"created":"2024-11-27 11:43:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/935652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/935652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/935652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/935652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/935652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/935652/3"}}},"commit":{"parents":[{"commit":"a452224dfa869790237288269b2049052c701178","subject":"Merge \"rpc/dhcp: avoid get_network_info to return segments if not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a452224dfa869790237288269b2049052c701178"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-19 13:23:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-27 11:43:15.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] Make provider network traffic centralized always","message":"WIP \u003d\u003d [OVN] Make provider network traffic centralized always\n\nLP 2084782\n\nTODO: everything\nChange-Id: I56cefac5f44487717e298846419cac8a51658798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/194549b72dda9b069cc68e19fdce9fc5e99a2499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/194549b72dda9b069cc68e19fdce9fc5e99a2499"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
