)]}'
{"id":"openstack%2Fneutron~936813","triplet_id":"openstack%2Fneutron~master~I208ba38bff9191cabcc1325fec516d0b0179c97c","project":"openstack/neutron","branch":"master","topic":"bug/2083570","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-12-18 10:09:59.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-12-16 04:50:26.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}},"hashtags":[],"change_id":"I208ba38bff9191cabcc1325fec516d0b0179c97c","subject":"Fast exit when initially creating tunnel allocations","status":"MERGED","created":"2024-12-02 09:30:31.000000000","updated":"2024-12-18 15:22:21.000000000","submitted":"2024-12-18 15:20:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936813-bug/2083570","meta_rev_id":"ec3d242fdcdfcc02ea4a6506a10912bf8920f1f4","_number":936813,"virtual_id_number":936813,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-12-11 16:04:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-18 15:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-16 04:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-12-13 15:29:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-12-13 11:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-12-14 23:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-12-13 15:29:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-12-16 21:19:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 10:02:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-02 10:46:32.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 13:09:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 15:39:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-12-04 14:30:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-12-09 06:31:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2024-12-13 11:54:50.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-12-13 15:29:37.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-12-16 04:50:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"044ebb94faadd68cfd00bf5a53e17c997a1a15f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-02 09:30:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8a74795a76354d25fef776d81fc942fe97d3624","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 10:02:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e917014a56744d77b44c3179a6166149\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce754c0d63564f0380c35c53633a7a27 : SUCCESS in 29m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f96022b8dd914bc59dccf0dd95212276 : SUCCESS in 24m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1a4040cff2b67f0b6ae5c80ba3304407b3011f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-02 10:30:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"df8d50a8e43c167cd3675ed0d7b29abc255a1da7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-02 10:46:32.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c9ea0b20f3670adbb7030de55a6f650d70232f5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 11:03:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df461457b45e4d0c89def6a52e185cc6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e62b657eaec045cda407359c7ffe07f8 : SUCCESS in 24m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57ff773008414e1ab568784167a0c18a : SUCCESS in 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aed6bcd03742707733fc9afdd687f2152e695b1d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-02 11:50:53.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"cb5c194bf2a22a164da384dbc72a42e453fe6f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 13:09:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/041b0ae6183d47e693c452a6b811486d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f2b0b4a0821c4b35b6ce20f33ec81aaf : SUCCESS in 2h 08m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c92cc6966c8d4322be413876bd114e2e : SUCCESS in 2h 13m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/111fe95fb98c4d438c8ffb793150a43d : SUCCESS in 2h 10m 49s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2b87dd3161a4b9487542bb25bc0e8cc : SUCCESS in 1h 44m 33s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04f568ce09d2422ebe069e73484fee0c : SUCCESS in 2h 05m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/491278d7b8f24008bf23c8f173e57fc9 : SUCCESS in 1h 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130e3c7144e1488d989d82ffb09c0885 : FAILURE in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84367fdae0dc48978b098b3bc2e40b23 : SUCCESS in 22m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d526de8ceb24baeafb849f9c027f292 : SUCCESS in 25m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830f1a2591174a27bde7085d29c0840e : SUCCESS in 11m 32s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/958952301b204c27811c62dd38f44bf0 : SUCCESS in 59m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/495f101a3de04b62a22ab357e249844f : SUCCESS in 38m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f8ed3c9ed6894d31b5c16e71551fef8b : SUCCESS in 51m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a5688c14392b411fa9e51d19b4d0cb68 : SUCCESS in 2h 36m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/429a032d8c074cefb04d6254686a76ee : SUCCESS in 1h 07m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb984f21d38e4c509771060a6d9ad552 : SUCCESS in 1h 17m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e641924081d1495ba5b552e7afaa0883 : SUCCESS in 1h 39m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3e1c53fbc6b849d5a87b0bedc816e57d : POST_FAILURE in 56m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4381c2d14de743c2aed0c985f31e28ff : SUCCESS in 1h 12m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5cee41ae0d33429c960d288af3b27811 : SUCCESS in 1h 12m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3e11657f19b44581904fb0622fb8768c : SUCCESS in 1h 18m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cdf364487da44d48b3bce74ba1a78d35 : SUCCESS in 1h 12m 44s\n\nWarning:\n  Comments left for invalid file neutron/common/wsgi_utils.py","accounts_in_message":[],"_revision_number":2},{"id":"665b7296cdec4f40fa5b74a9597aa4ce149fedf3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-02 13:19:13.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5310b070104b7d83ee914c6ffbc77506b61bafd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-03 07:29:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"27054bedff1c5ceed9862c7d95e1414c680eb074","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-03 07:50:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a118d40614778af183a382233a09c85086f3b4aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-03 07:52:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69d33eed4d4544cb61da521fb5753e7e3214038c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 08:26:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e15667ac1afb4af6abdadcca22e38f4d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01033cbe9f9d48f88eaabe2794cd2e15 : SUCCESS in 23m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9e30a2754914760aa52586ad632ef6c : SUCCESS in 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"27b61f13982c8fef642687c2759da2d94206a6a3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-03 09:48:13.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7fef8bcf2c75c939bc379b1125a7d0e794179246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 10:19:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18ecf91ce2f84409a9a639bc4b4b9afc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7c11f6e24f34ef0b42bcd998f4cc088 : SUCCESS in 2h 18m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36a7dca6f8314cb7a454d37d31ecec8e : SUCCESS in 2h 26m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8ca0f16fa6ef458cbc46ada42780aecb : SUCCESS in 2h 10m 21s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8ad5330a9ec469987373e0285974381 : SUCCESS in 2h 12m 18s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c23b3b9869af45de8d8e52b9cfac7ba7 : FAILURE in 2h 09m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d10a21a9c3fc4c59ba5f18e5a3b0f15c : SUCCESS in 51m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/041dbf8f1d9848499aa5acc2b930d536 : SUCCESS in 11m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db6f686983ce44daa72d32d67afffd08 : SUCCESS in 21m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea849f7205da4b74aa37b5992b883d74 : SUCCESS in 23m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4897980dcc724d4ab9e002796434499f : SUCCESS in 10m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/76e8a1fbb9704f728c41263f578c65c7 : SUCCESS in 1h 08m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f0e22f08db7546818e46ace4f7ed7610 : SUCCESS in 1h 04m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6ed922f370554a89ad2f48f1cf6d2620 : FAILURE in 1h 04m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f55cdc2eeb9848a3ac5629f2492cae61 : SUCCESS in 2h 17m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdcdb92603084fd995ee33b6a71d5ba2 : SUCCESS in 1h 01m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d56cf3e8b09420c9a0bd542783f6b84 : SUCCESS in 1h 15m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e134ab5934f047bea1b7f489a5bcc3ef : SUCCESS in 1h 22m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3e664d816132434a99064ee5f52fbe16 : SUCCESS in 58m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ba261a4d541a42dfb566bb412202971e : SUCCESS in 56m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a0b46e34d6bf4d80b75be48ad92f1601 : SUCCESS in 1h 00m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f250234b272d4fcb8947cf8296df1293 : SUCCESS in 1h 04m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2edda53851d34829ae92ddb118447aed : SUCCESS in 1h 33m 20s","accounts_in_message":[],"_revision_number":4},{"id":"b2388d5bd7f625390615f0b6d5bd36138f56940a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-03 10:19:58.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"710d5788ca8f6062d334272ea60f42ac81299f94","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-04 07:32:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"271c1d0f300cd80c6df6ee7fee67f376080a74f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 08:11:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed3ab8983b85489b813b6207a7f631a0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97f145f67cc548d59131f4333eef4a66 : SUCCESS in 35m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c8d4603e5f94b44ab34831fcf320abb : SUCCESS in 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c6daeb54260a0bdca0a87dee00b44dd883908c35","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-04 08:48:22.000000000","message":"Patch Set 4:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"9ac19f1d71949db5ba238c3338bbb874bb8f21fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 10:12:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd41d9556598495bb6f2d3df7a34b07d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/941989539a174d56a6b18384cfc0ae2a : SUCCESS in 1h 32m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36b272812f00457c9aa9e0aa8f850a34 : SUCCESS in 2h 31m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5866f092e7a7446f807f1bdaeb164d49 : SUCCESS in 2h 09m 02s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e187ab4565504ca48c17054ad9cb9926 : SUCCESS in 1h 52m 47s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60eff1853c2b4f0bba0b3a0f9b588e22 : SUCCESS in 2h 13m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de11f1d0e91c48e8ada90737906dd79b : SUCCESS in 33m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0974cbe011db40f4ba830ce1882b200d : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1490cca25c04870aec7d6aa882da85d : SUCCESS in 22m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9077a9a97f644212a0e960e53e2cbf43 : SUCCESS in 16m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f54833f8a649d78abf1f0c70869c9d : SUCCESS in 10m 37s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8291720349384c0cbf86b8a9b4c26d29 : SUCCESS in 48m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/dea3121a535847399ab82355768b0fc7 : SUCCESS in 1h 02m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6c3dece779de496a9a0fd0e08cbfe24d : SUCCESS in 55m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/236c5242469049108d4c038e2ec44802 : SUCCESS in 1h 23m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/78630a70a18547d2bde1b49ee0b6ad51 : SUCCESS in 58m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfba76e52a484b8b93ed994501ac6b87 : SUCCESS in 1h 12m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cafc2e0f5839405db3db2aa66b04e5d2 : SUCCESS in 1h 35m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/53570b4e1c744d3fbe05b91ef45eaf85 : SUCCESS in 1h 21m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1411232fed2b4dda9aaddae398d78f71 : SUCCESS in 1h 06m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/277e04903c3e4dc5a28ca2fea18ea54a : SUCCESS in 53m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb8ef77c9ffc4d0aa77781a0760279a0 : SUCCESS in 1h 11m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1e189c38158c427ab223d6bd09e9d6ce : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":4},{"id":"128ef56591b85a550916f0c87007477aeaac5c4b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-04 11:03:33.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"c627cc2ab823409c9fa1739486c8f141c044e60d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-04 14:30:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2ed3c81e653d556f58516e48b698e7a9b0dff24a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-05 08:26:54.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a2f501daa467c81715a934eac8b3890aa203be34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 09:06:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c1c385db76246c4ab5a7487af006b7a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e06c921843ac4d92b45140bdc59211a6 : SUCCESS in 37m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/33fa17e325f74f7bad26dda9eeb2d928 : SUCCESS in 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fa819537d673317baf7425f59fef1251554a1246","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-05 09:31:19.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"c51797e76cbf5404b3daa053d830637f352f138f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 10:30:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddc699060dc640c08abbe536113bc49d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0dd277e84aa24da1896445f9f2c2845b : SUCCESS in 2h 01m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ed38601c2e4465080d043889f455824 : SUCCESS in 1h 34m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/942c72beaa3e44c1858df03b82c6ca25 : SUCCESS in 1h 51m 44s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecc7e473da9643f2b0ba4750376f7dbc : SUCCESS in 1h 54m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/050933cfcb7444febf8799492e306680 : SUCCESS in 1h 53m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d750b2c8cf9148fa9358f85ef6059fe1 : SUCCESS in 40m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbd43d1e6be44893bb52b614f7ba2e98 : SUCCESS in 11m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffb21ffb6123444fa873ab2fbaa7a9f2 : SUCCESS in 18m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/926acb3af0ed49abaf71acb285a40cdd : SUCCESS in 22m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05aecdc668c840e588ca70ff6bef19f1 : SUCCESS in 11m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8c780636406347b49424191c268cbddf : SUCCESS in 55m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/807d6829fb384e6697dea945d72bc7a8 : SUCCESS in 39m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5edbb4c1874b4c839467f012819f9589 : RETRY_LIMIT in 4m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8a62f1130fef4213a3fb2a996b034a0f : RETRY_LIMIT in 8m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/402c43e2474942ea96dff2dd82604160 : SUCCESS in 58m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6c0e27d6f104563b2adc015f0fce44a : SUCCESS in 1h 20m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1ab87bbdcb04430aa8a83513d54c48e4 : SUCCESS in 1h 06m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a51a001712e14c1591ef3c3c29799019 : SUCCESS in 1h 25m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7ea9e8877c4a447e95a32290980c43ba : SUCCESS in 1h 05m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c46ee4659426483c848122d309c3f13a : SUCCESS in 55m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3088f5b819dc4d48b7fe24e9abcfcd61 : SUCCESS in 1h 13m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a7083ca9d9e4488eb31a30197b4ec9c5 : SUCCESS in 1h 25m 09s","accounts_in_message":[],"_revision_number":5},{"id":"1da6b74e5ae93c9fd9edeb51247771ae7a20fe64","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-05 11:24:33.000000000","message":"Patch Set 5: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cc90978dd4664bc488865a5f42b129fa3decd296","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 12:00:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5c58007b86a4bc5b1a375ab03c07e8a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b025e8c6946457dbf18bc8b99cd5f87 : SUCCESS in 30m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43d2f9c165ce43f4bb5d0072737bc8d1 : SUCCESS in 33m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a79ba3c40b98079b86ddfb1c69c0a4e59f7c5d97","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-05 13:16:22.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"e9f654cb873ca450b9c12e0aa3e562e9c7f3b389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 13:50:51.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/790294f9b3c04f7ab59ddc1f125c9c35\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/edbcd2f3072440f6b545cb7243c8cfe0 : SUCCESS in 1h 45m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/64ef7d2961ec402abd4d00fc93492e90 : SUCCESS in 2h 24m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/51964f9721ec482c99a2c06d735cafca : SUCCESS in 2h 18m 18s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf2db850d14a4f439858f0c370922d58 : SUCCESS in 1h 25m 38s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e1ab7f6123a4e029ca768a155330e75 : SUCCESS in 1h 42m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83bbdeafcf15490b9570db18c457f70f : SUCCESS in 45m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f1e37ec998743e2b75b45d17e375edb : SUCCESS in 12m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3e088a200e042338b159aef4f2be029 : SUCCESS in 21m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe2583437cbf4f6da7d41a83966f3dc8 : SUCCESS in 17m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e827cfa165a451fa1433c159cd5714f : SUCCESS in 8m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/eb57d953f2c94856879f623d39511576 : SUCCESS in 1h 00m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/31ca867473bc4d1093fe8419cb4ea66f : SUCCESS in 1h 03m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8ea8c79c769746d399b551e61f000d42 : RETRY_LIMIT in 15m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a5ea8965e2ca42d190bb648a9bdb7b0a : RETRY_LIMIT in 16m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c819401001f43f1ab02fc678c9ab3cb : SUCCESS in 51m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/54332da0c962456bbca10a85bc58a1f3 : SUCCESS in 1h 17m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aad6ed6c97c041aca15e814162a224dc : SUCCESS in 1h 27m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8809251f58bb4d2fa09bb1924275ca4c : SUCCESS in 1h 17m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3171f8e8ea2e4dc0b6df6c78ca00a569 : SUCCESS in 53m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ffebbcf4b18f42338595e7f0dcf2dc98 : SUCCESS in 52m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74f1c14f72fa4b1289044e22097b7c44 : SUCCESS in 59m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/332c041d848f49c6937ee88a379b12e1 : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":5},{"id":"851f3d9b3cd2254ffa7471417d8ac45ef4e66a30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-06 12:32:48.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c83b2e2d590f5e0fdbec15bd4be97e27a5ffab86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 13:34:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfe6681ebca14a77b60b19033bbe9ca7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa9352c33c5a4b429268706fceb7f267 : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/87addbe5846240598db44ba27b6c39cb : SUCCESS in 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6b48b608ec1344d2be2d0b1db0cb0629c674318a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 15:14:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dcb89eff15a4de28ba02474c3ebd10b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bcfffb3ea2e541e68238e5cc7b279442 : SUCCESS in 1h 44m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d0aab72e452b4a26bfc16425d5b00f84 : SUCCESS in 2h 08m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/19480d50e121416f96dae3000c4d7412 : SUCCESS in 2h 08m 03s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/694c780dccd54c7c9b37158eaefb63de : SUCCESS in 1h 26m 15s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54096e3c65bb4ac3b493d0d632e44ea1 : SUCCESS in 1h 55m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44b05ffbb8684fc6b1e0d5c3c5d4dc67 : SUCCESS in 52m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79f289dff8043eaaacd4987984fb17d : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a798d162cb14ee1a13f61b27c632bf6 : SUCCESS in 30m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c161d1d42622432e9185f5d5cb24ab2b : SUCCESS in 23m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85143c0baf2146829c336ba7437d7497 : SUCCESS in 11m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a4774f1eeafd4ca385c15bdfc82716f3 : SUCCESS in 1h 14m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/da7a8e1c8e754c8cbbd4a6a0ce9bc1a2 : SUCCESS in 1h 04m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/427835ba67bf4db2b00549e7b6afaeb4 : SUCCESS in 1h 04m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4c7baa1df25d4818a9edf20a7b60a0ca : FAILURE in 1h 55m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e709c917291448daacb4a3148c7ca743 : SUCCESS in 1h 05m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb920bb767d9482481f6565cf2b51eb6 : SUCCESS in 1h 22m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9dedf046d59a4936a7a21337769f9978 : SUCCESS in 1h 19m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5cb18f39004847fb9dc1e6de9f053357 : SUCCESS in 1h 28m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/322c0229ba484e7c922480ccbdb10755 : SUCCESS in 1h 04m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1de91e9558c04da0872d6c0d6d2ec2fc : SUCCESS in 1h 09m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd6fbdc401394e3380b806bd216ca7b3 : SUCCESS in 1h 22m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/33f2d4e3877b42368f969fdb6fafddbe : SUCCESS in 1h 16m 54s","accounts_in_message":[],"_revision_number":6},{"id":"d54d2145f6143416765752628a38a32f08959940","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-06 15:37:53.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"4625d24dc45fe16314bc4067c13f1c11bac4e0f0","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-09 06:31:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5a1fa8f81138dd1cefd0cebef9ff45527b83c948","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-09 06:56:12.000000000","message":"Patch Set 6: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"510bf4163073ef6eb85816384e88ef2c6fc20712","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-09 07:20:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d0aa9a9fb287d8389ab72ad1d5e372076b82d5a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-09 07:20:55.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"77ccd2b9ef200820f0b2c2d0d2121408db0fb6e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 07:21:05.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936428 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"f21f185ff1850efcdedd132d248db070e5d7f22f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 08:25:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c057ef864f9b4f42bfd135a42c2bbf84\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04e89f108edb4610ae14c66773b44047 : SUCCESS in 38m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c47dd2d9d93f44018ebbecef2b5bfc2b : SUCCESS in 24m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1514067ae80152aa049eccffa23c40eb8ef95676","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-09 08:30:06.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"0ae4d6ed19facf485eede7e4eca3ef8ee1fddcbe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-09 09:32:31.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"900c51ce1a893f40477a3f1d00616002fec43939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 09:58:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/068db42bfca7407ca8a579c3a51b7fd6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c1b54053be274f9f9648830b8a029e95 : SUCCESS in 1h 58m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/261095aef53b496da912c1df14eaab3b : SUCCESS in 2h 11m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8bb50194196344f1ab09b91bee8dc878 : SUCCESS in 2h 07m 31s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ae4375ddaf34f14b0c164e299f0d092 : SUCCESS in 1h 44m 43s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a3c8377ec834e7da2b1edd496dc75ed : SUCCESS in 2h 34m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fa9e9ef04a44a2c9972ad9022b387ef : SUCCESS in 44m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af7f03d80fc0419f8912c209826b91a9 : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b79fae40da34405cb966cd79d5f155cf : SUCCESS in 20m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/825d7f836c9c459dacd0bd4814b39cb9 : SUCCESS in 18m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd6dc3775364e93b27949d8d45b1e2c : SUCCESS in 9m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/080dc2604c984eb2a5a88183316126fb : SUCCESS in 46m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/af381dc330ea4d70b039f3f2c159fbae : SUCCESS in 1h 03m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d79edec7ebb0438f8e0fc921e98ac65b : SUCCESS in 40m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/64e9ce859c30483ab183a8ff3371d782 : SUCCESS in 2h 25m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/80f2bd7d61574671b2f36887203a00f8 : SUCCESS in 52m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/83265b2af2954a71ad39fe9815efab9d : SUCCESS in 1h 15m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e0911ebff5054489a3e99005110a3b40 : SUCCESS in 1h 39m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/05e866e77eb949fd98931d0ac2c6ffc9 : SUCCESS in 1h 28m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7ecda63654814a5aac2c7f425304f920 : SUCCESS in 1h 04m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2f1781f7b1954245bf8ffa013b7df1ea : SUCCESS in 1h 15m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5eea910cc08d43079b2e17109c73f7d8 : SUCCESS in 1h 09m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b26945bf29e14064a179e0384e658b18 : SUCCESS in 1h 29m 35s","accounts_in_message":[],"_revision_number":7},{"id":"36db4c2c2cbdac9d4b6abb7bfb5bf5a28f530372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-11 08:30:18.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"7cd1c38820c96351f7c8b226f3f3f8ab96f28079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-11 08:32:43.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"6d92bd27378431cb6191d96f8f836b3fcb75228b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-11 08:33:14.000000000","message":"Patch Set 9: Review-Priority+2","accounts_in_message":[],"_revision_number":9},{"id":"23a9aea36fda46725572f634c84253c8aa4459cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 09:06:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8cac6b0801a4a36a2b39d8a87a68b7a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b4d8ea54972493692f40eacb6af3b63 : SUCCESS in 25m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c45dea2ce2747a6b46af8606d6211eb : SUCCESS in 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f0e01ac84e32013e827ad817de35ed635fa65000","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-11 09:33:55.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"2ec88577799f0559d38269e3a3f2a8df04b863db","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-11 10:06:05.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"ef39d9f17e3e425bdbf9c115e06e2cb85c404cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 11:31:57.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b65e434900834ee488806a47c3a3cdbc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d4eff3ed2c6f4dbc82f54d1955190d3a : SUCCESS in 2h 31m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/41e0be66244e45e3beddd50a140bfb75 : SUCCESS in 2h 15m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b0581d3e61764fa6af159b76e914f36f : SUCCESS in 1h 34m 53s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bc7d6b596c44a75a059bbc4ffb4068e : SUCCESS in 1h 26m 34s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bed9a6c102444bb4a308d7868bf81f8e : SUCCESS in 2h 58m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0785f323f1374ffe9485bbfb690f974c : SUCCESS in 53m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe453fd6dc474dee94ab9a53d9ae2f67 : SUCCESS in 10m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1232c99d5cc34857a8a7e18b139f064a : SUCCESS in 22m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62bcc62ee8454f919f10c41f75570d26 : SUCCESS in 18m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/408c23d688204db793b8289b3cf9c917 : SUCCESS in 10m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2fe4db5bcc9d43bb89d963d6149ea9de : SUCCESS in 47m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0d23aaa02de34f158a31dce05e3f1f43 : SUCCESS in 39m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/75b9e3e2682749859bef07bbf6c06c71 : SUCCESS in 42m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/637cc562a09c47949a2de1748c4e5bb1 : FAILURE in 1h 24m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/10e9a3960b5848c6b7ab869571761142 : SUCCESS in 1h 00m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9be4407792b4ee792b2995ef2af2428 : SUCCESS in 1h 17m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/53116e2864db44e98f3026eb064981da : SUCCESS in 1h 40m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2fa842bb1a5943f297374c4374e7db63 : SUCCESS in 1h 29m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4098ad5ab0f849e0a1c36f6f94eb6ba2 : SUCCESS in 1h 14m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6d410a2b3e4f406988f3be242fb19d2c : SUCCESS in 1h 04m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c0a9b0840af049d68712a1fdef6d1a12 : SUCCESS in 1h 18m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/29fc53910d5d406990eff9e65abb1ee2 : SUCCESS in 1h 18m 51s","accounts_in_message":[],"_revision_number":9},{"id":"43c18922bca3983a2bbebc304f3ec6b1e4dbf678","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-11 13:38:22.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"a6c86b2ce4dc0b80761853fcbcbca87e29174459","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 14:28:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5afe55986ea0456e835fd0291320dbc3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e41909daf89649c1868c60239a333255 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b67f780d673747d091bf06e1bc8cf512 : SUCCESS in 26m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"525393093423015b8c96bd2da1094e6955877e07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 15:42:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cfa7b489366485d8c112eff3447a4b3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f1491efcd7324dfcb693d10668f792c0 : SUCCESS in 1h 49m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2dde025c84ca450c91851332c53dc04d : SUCCESS in 1h 47m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d95077b53ff8483b839749efa4a00a24 : SUCCESS in 1h 51m 12s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e9b99b23efc4d22ba41b2a523af6b00 : SUCCESS in 1h 34m 53s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6833704c7a254a90bddd6a6040bee991 : SUCCESS in 1h 38m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ade1d901172349b68e69e3b0f3e635f9 : SUCCESS in 1h 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68e5e32af994f9c945793dcc48bc04d : SUCCESS in 10m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d148f8c2d212491ea15ab6d8faaca66f : SUCCESS in 22m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a14243ceb04648a98e5f39f8dcfd015b : SUCCESS in 20m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7de8ebb21e14e419b9bb5b2ee610bfc : SUCCESS in 9m 45s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6b0c94be4b7f46e48638057492c152e1 : SUCCESS in 34m 32s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4463c5eadfff41f8a001214720c74a13 : SUCCESS in 1h 11m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f2fc5d35b13545babf931211a0046f0f : SUCCESS in 1h 16m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9500428e13334b81a05d2f5f49d897f1 : SUCCESS in 1h 55m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/92de24f5dd4a485993b1804b6916c3c9 : SUCCESS in 1h 07m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/01badca4c7484b87a71796cb0671095b : SUCCESS in 1h 02m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0a07dd6676184078b3b6c96d9b0ec86a : SUCCESS in 1h 43m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d3105bf868e4b68892962ce2001cb42 : SUCCESS in 1h 37m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0cd2d1eca6414f8f8fdbe00b0fac19ce : SUCCESS in 1h 32m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/55592a058dc04fdeba7cd5feffc3d54e : SUCCESS in 1h 31m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c5af4d0f9204b918342b9712e319963 : SUCCESS in 1h 23m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/65543a1a0c0b4f318bc8b90d42a3ceef : SUCCESS in 1h 34m 46s","accounts_in_message":[],"_revision_number":10},{"id":"9df81e418936ccad2983e2d11cab558085ae03fd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-11 16:04:57.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"ac9b1c088caf00d36a021875353f5bba62c05a7e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-13 11:54:50.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"3753f18bab8b37c5d8e8b3949f53da1f42af40b5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-13 15:29:37.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a82bc1ccb9a929135ed2efa091cb007e13c59273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 18:26:58.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"b1d3c03db18e03e7935903245cab7ede7cc8de64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 20:57:32.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/486e07bb4d71475188b5b26515fea422\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f88b379e38a64672b42b6add1037d3ab : FAILURE in 27m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6b470a3c08a34ccfb22ecaa2123c7ce7 : SUCCESS in 2h 28m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/57af463655e549b7b97814108a657505 : FAILURE in 28m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac6cf6d4e8948ed95ce43c0e045743a : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/781851719d4a4e9d8965f0dc9dc26cee : SUCCESS in 20m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a6ee401cdf947b09508ff1c4fee36de : SUCCESS in 21m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d012d3c0b1ac498f8080a678644b257f : SUCCESS in 9m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/22379678885944fd92464eda255a0053 : SUCCESS in 48m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b37bcb76873f4b53a89a5909602dae07 : SUCCESS in 2h 06m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ecb0d799c330420fa4712e725275cc90 : SUCCESS in 1h 27m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ed11a50ef0a48f1b11bf21ec2c5952d : SUCCESS in 55m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/49f60e34aa9b4064b8b649e6a50fa000 : SUCCESS in 1h 03m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bccc58ce24e6434db0ce68f93e5d3bad : SUCCESS in 1h 00m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a0491c9c2acd4cb4b187722796c25e64 : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":10},{"id":"ccee338bdf1470405f75ca023047c829946bdc81","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-13 21:26:05.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"c0c4c8fe007c781be840ab1e2eedf77491b0c397","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 21:56:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ccd22286cd4148bff6266236d7308e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/daa24ad45fe2411e8c4a2bf50bb22a79 : SUCCESS in 28m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0802f9fb4477403cb6aaa51d47f487aa : SUCCESS in 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"222b1a6c386a0a88410290aa4ddf4cbfe0e4f2a2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-13 21:58:56.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"957317ed16e8a184ad884a39ff31b39861aafba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 23:39:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f42379356a6b487287f7ac8533698211\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/91bca5d1d3364658bb27f29f25460288 : FAILURE in 20m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a9032a42b13b4060a0ac4b0e9d127167 : SUCCESS in 2h 04m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aa9edce7b1c04f3b9a5b734ad637a56b : SUCCESS in 1h 49m 58s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9d88c538ba644c1bd4b5d0c30d0d844 : SUCCESS in 1h 34m 19s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/796c4e057f8241bb968609a0b6f1de6d : FAILURE in 22m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e07e63f0c5a49c2ae1ac7f8efb23e1d : SUCCESS in 28m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20fc88d1f36042da84547700733eb017 : SUCCESS in 8m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed45d9b5d36d4f028c9d4ada1d81c6db : SUCCESS in 18m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2769bbd533f645459c5fbe1846838bd9 : SUCCESS in 15m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7fc5aa5c86c49299492d8cfa5c59e6f : SUCCESS in 10m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e1b9a9fc156742878eca1e11b704dbd6 : SUCCESS in 38m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e5dee65ce87a420b8123a381659f5ef4 : SUCCESS in 1h 18m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0c58bf2e045f478eae0a16d5b2434da3 : SUCCESS in 57m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a5ab840e494c4faabc1ad72c144ddee9 : SUCCESS in 2h 06m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/edd71ac2c5cf41c7b4f3c6db36b3b974 : SUCCESS in 1h 02m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/79f8c40ff99549e1ad3bae95e0c7669f : SUCCESS in 1h 15m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3f9e1f8817e142ef8a0ab222e378fa96 : SUCCESS in 1h 26m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0c04764e7fd342c583c5fba20c53a312 : SUCCESS in 53m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d2936bc52ea64f12a667afdcb82910f6 : SUCCESS in 43m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3fec6a90c14e4a8caddda8190c1c7f97 : SUCCESS in 33m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/85c86843e7f74333b4a26e9668607551 : FAILURE in 56m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1490b304d1554a86902627cb9934d01c : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":10},{"id":"7d5d9cde6a8990ba3870ea207be6c3b5385766d2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-13 23:56:09.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"913c7f48fd9a7b9d8671dad36778f833a9b37759","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 00:23:14.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5ce74ec96a3406080a1ab4ed673b943\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/247a7550bb6746f7b663ec4750d9464a : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b3255a63301b4cbeac0ee1aaaab75ac9 : SUCCESS in 25m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4337504e0d16bed109b7120caac9be23a0c3fa85","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-14 00:28:20.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"61a220d8ffb3b0fa1ff2386f0b5498eab395df92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 02:24:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ed846d3cfc44eb7b9217869e59a2ed6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3ab143c70a80404fab5758e008e2e052 : SUCCESS in 1h 55m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5220484a35974009b0e0ffb57f61c3b9 : SUCCESS in 2h 26m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/78e3727583174978ad9554c5bb335040 : SUCCESS in 1h 59m 18s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6aba4b624b4c4440a12c73d93447339a : SUCCESS in 1h 30m 05s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9a7dff4015e4e26a9a967f170db26a0 : SUCCESS in 2h 18m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c47c6a59429c47e3a7011bb026f0c547 : SUCCESS in 30m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a25754ecafb94d57b0e4bbb9ee4cf66c : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2f09ff93a2e494195681eb4ca701840 : SUCCESS in 18m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8127c5b7d3d4375a60e288c0c5a9ca3 : SUCCESS in 16m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ce4d13db014ec6a10983e838892210 : SUCCESS in 9m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/966012e917324f7fab523f4d6d88ae8e : SUCCESS in 54m 24s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/33d332dd909843a7bf11d1c8f01ee16a : SUCCESS in 1h 06m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d8862a09a0a4434a8ea02b935da82a9b : SUCCESS in 46m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ad9c15ebe7e94bc0afde28cc4b3fbb1a : SUCCESS in 1h 18m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/495bf2033bb54488bb79a9c256b2c2f9 : SUCCESS in 44m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f94bba7bd75e4be3904d2bb42624e2c3 : SUCCESS in 1h 15m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0a0da6dbe48e48fd9d51601f81906bfd : SUCCESS in 1h 33m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b9312bb590964111b640789e08989f88 : SUCCESS in 55m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a93718048f2047fe8212dbe070bb43a3 : SUCCESS in 1h 16m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8b8743f75e6648debfbdd3752cfaba89 : SUCCESS in 54m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef1016d49a7d4522a4e6af962ea72558 : SUCCESS in 49m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1567d0fc1986439e9add1c33096f8765 : SUCCESS in 1h 22m 19s","accounts_in_message":[],"_revision_number":10},{"id":"ffe7f2d3eaed172fcb1443adee57b625f9535ee1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 02:25:10.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"09ea47a8027ea48acfb577071480621a02243100","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 04:35:59.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27769d17cdbf4ac5994ed8806dfa56f2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bc8d731165b44370878e0acc79c85e83 : SUCCESS in 1h 54m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/00e7e070d5a04413897d95779246b525 : FAILURE in 23m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9fd05e1cfa2a4339b58dd12ad4b0da4c : FAILURE in 26m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52dbfb5d13b24e77b86f265e9d643cc5 : SUCCESS in 21m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a9d6ed9c58944d5822b2e997cfbbe05 : SUCCESS in 23m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/934eb04e262e4790babc5dc9c467b209 : SUCCESS in 20m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a95e13cd2194c708f2a5a05e186e8e9 : SUCCESS in 12m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/94cd46afc48c4284a48c0446a43f412f : SUCCESS in 1h 14m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f477130ccbc94d16954a9a31709502fb : SUCCESS in 1h 59m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4d81af2160db4c8086db7ba8201540d4 : SUCCESS in 1h 31m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0e58017efb149c3aaf091a0ca973637 : SUCCESS in 1h 11m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5ae12ae0b5d4a689e58257ae11c205e : SUCCESS in 1h 22m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5b620c17984e4f0da14e1b02d13e09d8 : SUCCESS in 51m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb40bc9b6fe0455da1dd1fceb887fb82 : SUCCESS in 1h 45m 58s","accounts_in_message":[],"_revision_number":10},{"id":"3998b747ac5e6ecad900796c80e455d0b0eccd3a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-14 04:56:17.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"450fd9141cfad47f8569bb16a422b6b77bf49abc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-14 05:28:09.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"ec37683c4a5e551c4c8d5e352a189a2d8116e877","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 05:30:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa3938c8f1a64a3493c1f0773b102c44\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/086c7ef65b1e465d8cb3329523f189cf : SUCCESS in 29m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/24b3ddd5be454f22ba5230893e2c3ea4 : SUCCESS in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f657265b86d889dd4058c501b5764f435aa68857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 07:29:59.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c044e21a5da49478c6f511c6d73b0ed\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/af3347d74b6a419091109c686a87ff37 : SUCCESS in 1h 27m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5540455a83284afb8a1e421225c670ee : SUCCESS in 1h 40m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/354522bd4e4c4ac9a812b6d07e2d8c97 : SUCCESS in 1h 32m 14s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdde83358d55433f90583c66d124be57 : SUCCESS in 1h 14m 31s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/806d9d097f8d4164bbb5d204017e720f : SUCCESS in 2h 15m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/817f5885c6ea434fbccd82473e353cca : SUCCESS in 30m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23632f30f1844f45b2a6d01845ce07df : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a205aa0f31ac45db9f35ed0ac1344da7 : SUCCESS in 17m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cd075c4d6d94c258e179e8f25d8b08f : SUCCESS in 16m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868656252d1547ccb82004920f715783 : SUCCESS in 5m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/00569c7ef0d442f9b90ece183802b5dd : SUCCESS in 32m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aac0eb26eb7149a3b3fab46626c0e28c : SUCCESS in 39m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/44918051d59c4bc99701b828439719b0 : SUCCESS in 35m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ef42e9e44a464277a9053dd39ebce86c : FAILURE in 1h 45m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6c8ec53ea444622b45ddb919be6e384 : SUCCESS in 1h 03m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f77ce16ae8e447eb0db0e493be4eb29 : SUCCESS in 45m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b8a1c0163f09411baf5df21263c89cd1 : FAILURE in 1h 05m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4b1f79598fae438e803eab33ebacc7ae : SUCCESS in 1h 08m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3cb8e4a25cee45eaba41700c8ef5773b : SUCCESS in 46m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/595352e69ded4dd786ddab47a75bdc46 : SUCCESS in 1h 19m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/22c6f48759f0446e80c67386e16882f0 : SUCCESS in 36m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9d256a238fcc42dd8f31f4eeadbf14cc : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":10},{"id":"fbe1ac98004a2691009241cc4f35e5f6b56c083e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-14 07:56:22.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"f93c11fe0a97c8ec2a74cecaf480ec045dedacb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 08:19:29.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/095b963301674a87b22b9d3d71c378b6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ea169289cff4094be45e30d34b56f64 : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07bc20224211465ca023b418e009501c : SUCCESS in 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ef49a356ae74f865fea599097df697990cfd4da9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-14 08:28:07.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"93d8b7248f488cb884300e2e81f88163fcb33b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 09:49:29.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3006f9f3013e444b85c5ebb0982da21e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c35b518ed1dd452daad417d020f3ed3c : SUCCESS in 1h 08m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe7d303bc6b249cb9370d4c8d47b8ba6 : SUCCESS in 1h 48m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0a9aa36ee27f4d0b97691b63597a3c47 : SUCCESS in 1h 25m 09s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6359a8d4af9c40c2a7a1f554a11eeb1e : SUCCESS in 1h 13m 25s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ef0eac19f604155b41c07bfc4b62535 : SUCCESS in 1h 50m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57300f0a01e54b88bf6b8edd67a4b5ac : SUCCESS in 30m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0215132ff1264878b276075246d41dd3 : SUCCESS in 8m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfa5ce7428234a5ca09a1d78930ff098 : SUCCESS in 22m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b049124203284612bbef093944c87e92 : SUCCESS in 17m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1900a0554d64408c983eccc057594ee5 : SUCCESS in 8m 44s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/91aa74035e3e49518609ab7ddbfdcb5a : SUCCESS in 53m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2578b4fbc61a42c798c05669f82ee895 : SUCCESS in 1h 06m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ed410361c72240f283be3648daff5308 : SUCCESS in 50m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4fc5cd326a5e4363af7868da1c27d714 : FAILURE in 1h 51m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/54c90d9dbabb4238b625a4ed5cbb51b3 : SUCCESS in 55m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/77c642a5035e4c60b3b912832a4c5646 : SUCCESS in 1h 12m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/06f2675056974a869f754a506e17ddb2 : SUCCESS in 1h 31m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/820eeb71e52c4104ac7ecabd6a0821f6 : SUCCESS in 1h 23m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fb25063c8c154981b143d3381b8529cc : SUCCESS in 1h 08m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a7b344246d884c21878a88764a8ae3f7 : SUCCESS in 59m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1f090f93693f40eb917da45e8d0fabf3 : SUCCESS in 1h 09m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a5eb5a903cf74e06a5623255d0f3fa4a : SUCCESS in 1h 31m 07s","accounts_in_message":[],"_revision_number":10},{"id":"46d8373c742c721e286093974322beb70b229d15","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-14 09:56:25.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"ca77bb394d12f1fc09854f3ba500e58ee788768e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 10:18:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bfa9b34b7964057a53560131efa8109\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d670ebb2fe54f7d9e94ca304cbf3697 : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9586152a125342069985494d93d94a09 : SUCCESS in 20m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"86ee295fa4a64f0c26f745098a292c069482cd5f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-14 10:27:56.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"6ef8a81d2d5400d36952085eea96b149b40da4bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 12:06:09.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91d81ccc40e042a98ac9218de75a54a6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e4a94b7342c4c299c5a62cedf9b4e34 : SUCCESS in 1h 36m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9072e5d1194c44eb9b297a6fabe33be9 : SUCCESS in 1h 21m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e7d6f64e87d04fe9a2564d847a580053 : SUCCESS in 1h 41m 45s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9fcf694f04c4c219fa7ad8f828162c3 : SUCCESS in 1h 37m 31s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20a146e212b04853ab8c1840c591c5b0 : FAILURE in 1h 50m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/544f7ea05e5c41ccbf5013d0fa4cec03 : SUCCESS in 50m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b99ffd8202f466a9288b6d302d90c40 : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5966e41f569a491e9738b7acbcf375b9 : SUCCESS in 16m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52fe7143a89c40afa0f2d03a3e7a518f : SUCCESS in 16m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4515abdf55e54be58548d81b3d121f5d : SUCCESS in 9m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/564ca759de5a4728a78fe16a82e3893d : SUCCESS in 37m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d33642b7cba74509b3d0e1b465a9c40b : SUCCESS in 42m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b550b23a5ac14e1f88c1bf6b9b8bde8e : SUCCESS in 53m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7f7046cf9aeb441689dc3567c7c4cbf5 : SUCCESS in 2h 04m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/300edaaa034a4bdda6e6bfc07e0c307f : SUCCESS in 1h 04m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e66ad187b2d41e08d73a5a362577228 : SUCCESS in 1h 05m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1ad1891b235e474faf88139b40e71158 : SUCCESS in 1h 45m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8aa77d4926a344ad9cfe5e2b457dc877 : SUCCESS in 1h 17m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7b5a26edfca544a6990522c555409bcd : SUCCESS in 1h 11m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/82007c492d204ed2a29a3fd3b6fed7c4 : SUCCESS in 1h 06m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8228d7e690ec427982364aebad41b300 : SUCCESS in 1h 12m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d1a2b13a1514f53982adfcd35566204 : SUCCESS in 1h 09m 22s","accounts_in_message":[],"_revision_number":10},{"id":"e30a9f893cb7713948d367eefe5c95498aabcf0a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-14 12:26:31.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"639757ff3ca5990d4af88250f2630ca72edc860a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 12:48:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edac2ff39ef74866a0b8dfd5c7f6b825\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d258d7e9f6024892bc515b7d67bd957a : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cb3665cf0c0742d2bff87b068e5550ab : SUCCESS in 16m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"abdda38a52b3954ea7c48585453da879573f917a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-14 12:57:24.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"e66f0f8650e9b35ba4fd46100fe5369358648fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 14:59:43.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/339b69c1e48048bcbd9df10454ecc88a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/df5c89667e9a491f8d7553971198123c : SUCCESS in 1h 40m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/df8c777056a44288b9319dbe78aaf803 : SUCCESS in 1h 49m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bfe26c3ad2f14660b9727109863d2afc : SUCCESS in 1h 54m 02s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22de3f776a144d47ae1739af297aef58 : SUCCESS in 1h 29m 48s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63145ab4f17647f49fd1757a23fe5a9c : SUCCESS in 2h 31m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9ad8865b9354e65bb1bcb3e8f80d151 : SUCCESS in 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a55c1d2a5a43489901a70b01a5e995 : SUCCESS in 8m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0ef03fa0815443e84bf87bd96f55f0b : SUCCESS in 18m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30f4df5c27604f40831c5c7be5c1009c : SUCCESS in 16m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df32909dee4649fb96d0806606c5b9d9 : SUCCESS in 8m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f57a9cb163ca405fb3cbe485cc84ebe1 : SUCCESS in 38m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/59c0ea19638a4377a45042daa1d6632f : SUCCESS in 1h 07m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a370182f9fe647d090d4719135a234a0 : SUCCESS in 37m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/460671c107ba420d8f88b5e86856c35a : FAILURE in 1h 47m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c59e6c7502bb44459101eeefbbc59481 : SUCCESS in 59m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1bb3c000da04ed68aa5f134c50dbe47 : SUCCESS in 1h 13m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6fefb66465c4413bba30eacda7746b30 : SUCCESS in 1h 48m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/981549082be54500bbc10eced97f43d2 : SUCCESS in 1h 44m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c5e2d4f46f954a06a3880bc53f281624 : SUCCESS in 1h 09m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/04415bf87a4f4ed687a3d5ddfe8755d7 : SUCCESS in 1h 07m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3b26ec9a7b1947879bc2f611373fb5c8 : SUCCESS in 1h 12m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a20fab3823e34176a6c92c99cf665b40 : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":10},{"id":"2357adba00381cb2ae626c6a2a7268b3d136f245","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-14 15:26:36.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"6ffb209914071592f807a1e153193d13453f36be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 15:50:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d227ff329c64c83bd70f0488c0878eb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd895451d6684e4ea101458352166728 : SUCCESS in 22m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c0883996ee5463abb2bcd1ee8985d8a : SUCCESS in 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dfe698c6925851dd9a94b015c612648390170e19","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-14 15:58:13.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"b7d72d23efa2a708b8b1f0ec6f227a111a3040d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 18:12:23.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/635a004c14294e0ba1718673de004506\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/042bfa5e85134331a17398eee01d5b5e : SUCCESS in 1h 14m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1f71c96f9355456d8549863873f4385a : SUCCESS in 1h 19m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c051469c2e5a423eae7b4a393cde69fb : SUCCESS in 1h 29m 14s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cef6737263948e2a9bd5980af920ee7 : SUCCESS in 1h 49m 34s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfbe706a72154c6ca5aa2bbe5a4ac036 : SUCCESS in 2h 34m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feb0b5e8ebc94ad7a30edff22048906c : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/917a585aab1d4c348419ae0c49f20166 : SUCCESS in 8m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/089bbeea6ca844a2aef3494d87fe5598 : SUCCESS in 19m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66ae6b9a31ff4efa8ec0a97aa86c1b04 : SUCCESS in 19m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e7a1cc18ec4139895456831c5ef888 : SUCCESS in 8m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6f4203a05ce3444f839cb999471fc244 : SUCCESS in 45m 32s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8f1795be94d54fd1b6639d1907b97bf7 : SUCCESS in 51m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a57dd4e2e8c24303a7265d8b943ab5b4 : SUCCESS in 40m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8c0583a4f5e54821abe72b3b5f60181c : FAILURE in 1h 36m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d1a40ee4abe4617bd6ff777040ad200 : SUCCESS in 45m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/09bbebdc46ed4168826f82c11f63724a : SUCCESS in 52m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/50aca300abf143b7b1fdce8d9ceefcf2 : SUCCESS in 1h 20m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/693f4852542d4354ab45b2d80a79eb10 : SUCCESS in 1h 00m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fc219995e2bb4c3fb8e50bdfbe420b3d : SUCCESS in 1h 08m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/724d2347ce4b4cc48c7aca0d647dfeb1 : SUCCESS in 46m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b86ab28d13014f5da0f961643abf159e : SUCCESS in 51m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/10d24b48e1b446aca33bb3cbbc3554bf : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":10},{"id":"09dafafdd1a2d2952309c96ea38baef47adcff77","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-14 18:26:41.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"9bf9413e5e09b686670c651cb867accae3d7dce8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 18:52:31.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d920aecde74d699850bb9b641e4a73\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/028937d3510640e789a17d3869320a00 : SUCCESS in 23m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57a04c8c065044eca79f0b6898c331c2 : SUCCESS in 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ea3dff5cd658ab5a86cfe94de4689effc3123a6d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-14 18:58:19.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"f610eab1a7d0194e8bd7c837bfac6d6a8d5e9efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 20:29:09.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec87d44b1dd44db493df39446305b10a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8e059576532c43b1b134d44ccd97ca44 : FAILURE in 19m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ef12619f40f94551b0632e23545bac82 : SUCCESS in 1h 32m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd8c6597d9dc4b18b92468712f265ba5 : SUCCESS in 1h 56m 55s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f487f8d86e4c49daa1f7f6bf233cb566 : SUCCESS in 1h 37m 07s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aad6ecaadad43ad80f68d1e84478365 : SUCCESS in 2h 01m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c87afce1b89d4795b2f1051ab9de3358 : SUCCESS in 34m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498eb8c9c4b6463c9b7d6503b9aaf9a2 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a9516c5098549d4966fe3d1db07caae : SUCCESS in 22m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/102b390c880a44eeb15e471fcab074df : SUCCESS in 15m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e35f6101de742d0af57e4317c753a28 : SUCCESS in 8m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1ab747868c9d4c168dacf7f3ef7f2879 : SUCCESS in 40m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4d5ca3ae5868478a8c41eb153320d8d9 : SUCCESS in 49m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f2d5f426030d481c93e65e91daec35e6 : FAILURE in 34m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9bc58d7c87cf41f0b9fcb4ea47102dc2 : SUCCESS in 1h 54m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0ba1aaadd334555b0460e75baa7a326 : SUCCESS in 43m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/590cc4c448c24082a4049d24208271dd : SUCCESS in 55m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b644969405804014a1fc43450b249fba : SUCCESS in 1h 22m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ea5fd15fef7b42328abc0655f8e7c013 : SUCCESS in 1h 05m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/657745ec25da4c25a73695f6be0a5535 : SUCCESS in 51m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/46938c4b2dba4745bd5330cc42c33418 : SUCCESS in 40m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8f5ea11df0f248b8933d959198e4fee6 : SUCCESS in 45m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/678c4a0b5a5e4ec99426ac372976157d : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":10},{"id":"fb995f7147e253001227c506530b84f96014baf7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-14 20:35:13.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"ec97aba71edc6d9be6eab7af9226f7b155aaa020","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 20:57:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd997bd75c648e38327a4a64649f0ba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ea0b59374f4499084acf79d815577e4 : SUCCESS in 20m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8480d19cc3ba4ffb9cab22d3120ba42f : SUCCESS in 17m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7a796b85885dc6e661dc2336927f4c09c523d1e5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-14 21:36:29.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"f8f49ebe825865fa03899188219172a07990742e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 22:52:49.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e4df47ccd1a452bb163930577ef2a40\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/766d957cc00f47ec94f0deeeef52ebd4 : FAILURE in 20m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b3232017bd404e42ac08e28082fcf4ed : FAILURE in 20m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ccaa58657f454774a0f616488e761470 : SUCCESS in 1h 49m 57s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d79ee1714dce457ab8c29e95b0a0a36a : SUCCESS in 1h 24m 11s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b28974013dba424e84aed82d12b11095 : SUCCESS in 2h 16m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59f840dce2d749f38c7f52d2d25ebf19 : SUCCESS in 39m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6053662e2394c5eaa0d2218ed65cea5 : SUCCESS in 13m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a68e08549094c448c83a8b9fbc9fcc9 : SUCCESS in 21m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff278cb1ad6e4e06a41ad1ad2862df34 : SUCCESS in 22m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78a712d87286443db2234d0a254fd6c5 : SUCCESS in 11m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2fa0328300594ceea1f318eb422720df : SUCCESS in 52m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f587f2cccdc94f78ac2040d4d68bba9a : SUCCESS in 56m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c63d0ec3c2f44ee595fe5eb71a4c0f85 : SUCCESS in 44m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ad4f3690c0644a2daa550ed2ab7bc036 : SUCCESS in 2h 00m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5518facb12f43469baca7982ac2d537 : SUCCESS in 41m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/78f9629181e242888854cf37324a396a : SUCCESS in 58m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c65a63fd7d8a421b88deb048e2f56f9d : SUCCESS in 1h 05m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4dbc886bae174aedbc86fee0366f2a57 : SUCCESS in 1h 33m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/af968269e6494640b645b48ffa0eac34 : SUCCESS in 45m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8b8e182961514d1babdc0b59f38fe4cb : SUCCESS in 41m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56687c59142646528de188d42527e1f5 : SUCCESS in 1h 00m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/002b600afbcb4240aa98744b36f895cb : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":10},{"id":"10ac4b3003c099ea1fd7a613a22a4604233fb48c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-14 23:03:03.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"01c03909d932545f5d75cbf4292dea17188f4ca4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-14 23:05:17.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"255aca5c2010a651e88248fbe314bd8f9cca02b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-14 23:32:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f98d004d45f44dec822b79e6993b1439\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b09f31206054322a354c53e10a6e3b9 : SUCCESS in 26m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9b26a596259049c58263d122333e2552 : SUCCESS in 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"03243e3fc3c4b1652411b7220eae7fdb6b5a03b5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-14 23:34:14.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"2350e5bfa040ac6b1f82303efef8a9610a5c9b47","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 00:05:22.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"88c7dbbeec272b848eaa58278f1797b7cea7284e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 01:45:33.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f441a692b4be40d289f8a6f5f6acb770\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cc7098eb06b44b3fb5268fb3ad7289f7 : SUCCESS in 2h 03m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a3613b054de4770a4881894f6827733 : SUCCESS in 1h 55m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b6ac1deb87494c7685a40c757e622926 : SUCCESS in 2h 40m 51s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8988651a763b4cc0b63dff23a8aa19c4 : SUCCESS in 1h 58m 54s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b377c3c20b044c21a15cffd26fc801be : SUCCESS in 2h 10m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d9b4e461cee4f4ba5933df97034094e : SUCCESS in 34m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3372916d20db4e3699561f3c3eeb10c6 : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1007550483f4088b73ab2740a56addf : SUCCESS in 24m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2cf56e5996e4af1b93b44d2a89dcc8c : SUCCESS in 16m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d795c541fe7445aaafad08e16bcecf5 : SUCCESS in 8m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f5aa2bd9a0a44b42a6a2f1ca289081d8 : SUCCESS in 54m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3d304256b20a4fe09c8d0efb994be501 : SUCCESS in 59m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bcefbfc92d2e47128a7965fcc9a916a2 : FAILURE in 48m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4392d472ef854745a98e8eff0f94d7f4 : SUCCESS in 1h 18m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f68986274b6643f5a51a818aeee3a2dc : SUCCESS in 1h 04m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fadf0c92fce6491a9bdd52d6506d884b : SUCCESS in 1h 11m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d5b060e057d04d60b7434b92e8de740a : SUCCESS in 1h 50m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/74609c26ab89440f93b6a30390373385 : SUCCESS in 1h 17m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4b1d18cb8c6646e3a35ba6d079f20f97 : SUCCESS in 1h 02m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e60e652ecf21475390f186ae85104545 : SUCCESS in 1h 11m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/766e6695e4544b03ad11976112b4d22c : SUCCESS in 1h 24m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c89b679828de4c2984fe6fa443efab27 : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":10},{"id":"a06a15c2082b440536f0fac27e49813591dade7f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-15 02:05:23.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"4a8106a22a4f73eb08520560ed1ccc0e4fb67c47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 02:51:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/defdcd33956c4c32abfe86c89726b12d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44208fa8f3fb4599aeab768e6b92a77d : SUCCESS in 33m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a7812d61a38d445da3800cb67eda263a : SUCCESS in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"901bf17816c294a59bf58c2cfaebade9414cb280","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 03:12:51.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"724a5f081cbc279af82eb2475e875f9288966915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 04:27:47.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69fd22c2a57a4f51a4a2725f6671f56a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/970e26ae03c24c1fb2cee0a41fd76849 : SUCCESS in 1h 50m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bfae5f57be804764b1ee1b0e846088c8 : SUCCESS in 1h 39m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7c776c1f07994ca59c117ff7d72017d3 : SUCCESS in 1h 50m 48s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8129603a2b648e09380d6338915e37f : SUCCESS in 1h 26m 53s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33bcf66081ed49dc8a4719021d93f45d : SUCCESS in 1h 53m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6289c16f0e934cc1b771900ad59befbb : SUCCESS in 36m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de67c0afcee4a41b10444e929141b81 : SUCCESS in 20m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/011f65e5bd0e456ab46732040a1d2fc6 : SUCCESS in 22m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5705e1eac1d641a593f6211caa6a0994 : SUCCESS in 26m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa8ea9d2438400793327380183c76d5 : SUCCESS in 9m 12s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1e716eaa53c541259df7ef7a79f088d2 : SUCCESS in 1h 23m 04s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ed96134c204a4b998f5005cca7c2d101 : SUCCESS in 1h 04m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c90091cb3ddb45fc851e3b1c88f3b15b : SUCCESS in 47m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3dcdfbfa8e4d426c9fca989b34f1b4a6 : SUCCESS in 1h 58m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/11859eaa17e74c76a1804c3111729c1f : SUCCESS in 55m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e0834bb84d94d8bbd9757e7127901d2 : SUCCESS in 1h 18m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7a330ea4619d48fea9e005bac554f573 : SUCCESS in 1h 27m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9cfb64acd5564c12a594a0b7e7ce4b4a : SUCCESS in 1h 29m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eadc317d1e4948c8a3e23a3cbabb155d : SUCCESS in 1h 04m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ef62a0186ee646308c2ec1787e8b5065 : SUCCESS in 1h 07m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e5e80875f2e34b609e8c019d740243b6 : SUCCESS in 1h 32m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/31aae56d50bd44e6993d70b18653364e : SUCCESS in 1h 30m 58s","accounts_in_message":[],"_revision_number":10},{"id":"3d9cc8613ffca02c518dcd4364f8a1a990ff1698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 04:28:18.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"f59053c5ab61a5375787784587545f3376a0843f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 06:03:27.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff1c2dc4576043b88e69e18ad0d9ed64\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fc39d415da48473a898bb3bd5652b657 : SUCCESS in 1h 28m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c4951ae44e8948889a301dcbbd9f18f6 : SUCCESS in 1h 34m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e22b6c3f5a394ef48cdc9e66972c6d18 : SUCCESS in 1h 29m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec278f878d748efa53bc6e49f0990ce : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/590549ec283246e2b4a6b838b9b7464f : SUCCESS in 16m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de9884d5a15c48b9a1cf332ff001b31e : SUCCESS in 15m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5aad3ca9d1843b2a82220b8ccd00426 : SUCCESS in 8m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c28e398db6b48f29ec29a3d794014fc : FAILURE in 50m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/93d294dd249945438d9e7039d7ec9d27 : SUCCESS in 1h 21m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7e778ea9a00f41bc95835d5c9b450cfc : SUCCESS in 1h 21m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ce353981b1e4511b2c7bc9581682fed : SUCCESS in 49m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8b7787b152a43439897d949fe210df1 : SUCCESS in 41m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/addc9830ea824125aa37151e9b9f8fa7 : SUCCESS in 52m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a66e62dea77142dfb8e9c48ba04162d9 : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":10},{"id":"426142ef789ee476725308c4d8c08dd5f8990949","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-15 06:05:31.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"26889f411d696ea9dd95ca0d8df2ec869966417a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 06:32:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa5aadc507cb455fb8d32a91adf7aa55\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a24e2f79990e43069bf005d80adfae61 : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/25100c8c7de94dc281a12e01fdccc045 : SUCCESS in 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"341a058940f88eff08241ef345a03056cffc9526","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 06:37:32.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"49b7e1d27bb2fc55615a5e5a267303f0d3db31c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 07:59:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2b6f4be6f3848ffb843d25fbd76bea0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bf0e45c0121849d287f0da224691177f : SUCCESS in 1h 53m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/252c848bb9ec487b96e8dcc0c64ae2c2 : SUCCESS in 1h 28m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/335c4d7982254055ac155bb6933ceb48 : FAILURE in 19m 12s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/709b702c3d954413be64ee659a9346cd : FAILURE in 16m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0beee045087c4ad6952d85d6541eb033 : SUCCESS in 1h 52m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a6fda24912c4eb08d737ffa6f9f97b7 : SUCCESS in 28m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3165a6d2c11b49e59d48318aaf4878b6 : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06faacbc3c0b4914b7bdf4b01d3cc482 : SUCCESS in 24m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4aa0ad4400a047129d620e254d67d2c3 : SUCCESS in 15m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e55d2642b2cb4214a9d09286e3fe20eb : SUCCESS in 9m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1747d73357864655a343ef8672761763 : SUCCESS in 36m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/21a49a7d5abd4c338190770344fa89c2 : SUCCESS in 46m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ed502f85cae3405da87f3836b27417af : SUCCESS in 35m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ac08a7ea5b1341ea8d2a7c1fd542b23e : SUCCESS in 1h 18m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ae791beb71243e4b16445ada55ee9d6 : SUCCESS in 45m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f70a6951ae44e44a9ba2d0739371496 : SUCCESS in 1h 20m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ab12da27d77c4594be91ee5f25e7d6b7 : SUCCESS in 1h 18m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e232380fe0943a4906fe0cce20b244c : SUCCESS in 1h 02m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f019ca7f2a874f81b087a2de292eaed3 : SUCCESS in 1h 16m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/98cfe20e7f5f4a6fbcb8521298d68145 : SUCCESS in 32m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b1679b4056ae4ad08f09da330da5a0bb : SUCCESS in 52m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c1a279ed8b8849a6bb45b0eeae1d0a63 : SUCCESS in 1h 14m 25s","accounts_in_message":[],"_revision_number":10},{"id":"52f89a915716870261713e076fc1309c0072cc5d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-15 08:05:35.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"611fb0367dbd22d5eac76ea1061f87278bc7d6b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 08:28:50.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9252f5b29786467d8a13de7d05c18233\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/626657a1b18a4f66a4228c2c18aa1586 : SUCCESS in 21m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b67c11f707ef4eb987eb33b5379b56b0 : SUCCESS in 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f89c763cfa9fa0095beef6b3546f54141ea66598","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 08:37:57.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"768a472680c78ccadb4fee4c9d2a51d4111ab678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 10:03:18.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a97630ccf2f94b0d9cff97a804521b3f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ef901e0d728241b7bcd2bc627c6ac0a9 : SUCCESS in 1h 09m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/68aeff55136b429baa30eba9bcc0a6d1 : SUCCESS in 1h 43m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/36bb5f6d55dd4a4ba4e538e54fb2d702 : SUCCESS in 1h 20m 33s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7365eab949964c2488ecefd80d8e9ac1 : FAILURE in 22m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acab47e2ee8f4bb78bb9cc47b0748d44 : FAILURE in 19m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4c8f5d7fac349afbb4868a5083ea16f : SUCCESS in 32m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd47dcad57bc4f09a459238e2f6d05e2 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1362e45f63e8496abacfae4fd1177c4f : SUCCESS in 17m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45f6931c5f0245eaaf4fb80a9d7187bd : SUCCESS in 15m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7d1fdf91e6485fb848fc35088cf9db : SUCCESS in 11m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/260672402d534013a00f4ea1ea3ee26f : SUCCESS in 41m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f653a833bf83424985e49cecdc3523c9 : SUCCESS in 1h 06m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd8c7b70980a4c34b113e63b58576537 : SUCCESS in 41m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/21765bc30681417bbea524f882373970 : SUCCESS in 1h 51m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b42d4274b3f44d2288e80d6b6ca8e7ac : SUCCESS in 51m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c8c940dcfb242c9b7f7b946b9f038cc : SUCCESS in 58m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7f13c3c60c5a450b9c9ecd7e6dc2fa3d : SUCCESS in 1h 51m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5064f444b2da47dfb6e4ccd604ef2c65 : SUCCESS in 1h 12m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3477b87c1c784ea38dcd646ccc5a57c8 : SUCCESS in 1h 05m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fe66898774414e82918e001367221dd6 : SUCCESS in 1h 13m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7338b82168d64605bda678ab53739887 : SUCCESS in 53m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f4165228ffa42a3b83a971f53475fd8 : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":10},{"id":"be71d24435db54730e0601176894ec1b4c414dc6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-15 10:05:39.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"7ee1f8fcac5faf84822c1bafb4d2f8b3f32835ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 10:26:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecef02321ab8467f8237258866ac6ed0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f582232e50e42c5a8d930e6073d7cc1 : SUCCESS in 19m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e157a9ef1484258a51c803525b48c59 : SUCCESS in 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4e4b3a05362278982e247b25d6cf7510e1e0f49b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 10:37:25.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"94a99341c58ebf4d1bd0570af3578be1f822fcea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 12:29:11.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14b9031f93b4486a8bbbe9fcf7dddb7d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d7735bea737047c58dc467b83266a3d7 : FAILURE in 20m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/383de08692b34879a0d4346cc7164ce2 : FAILURE in 20m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ad188752f574472da2af2c3d3e0271b4 : FAILURE in 20m 51s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a9eac53b4544718ab1551b4a314f5a0 : SUCCESS in 1h 50m 07s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2669202ae0794a129a3dce8f1c1840e6 : SUCCESS in 2h 22m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e3736141c9548ffb6f8e7a508758641 : SUCCESS in 52m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f1a1df850b4f088e2ce43ae4b0f31c : SUCCESS in 11m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f56a47177214ed586051386f9076af8 : SUCCESS in 16m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/309d18a3b7454593bb6fd8edd5a87625 : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b702cbcb4d0456bacb6a1060f0b9fc0 : SUCCESS in 8m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/028f68420dd3407b9fae1d942c7f3dd0 : SUCCESS in 43m 52s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/48d72a0f3c5a48eba6bda2f92d9064c0 : SUCCESS in 49m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6f9dcfec3974434c95489acc2c7e66e6 : SUCCESS in 40m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8abf5c4ffa7c4e669461a0716771ae2c : FAILURE in 1h 14m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c8ac640cf054ea7aa283318389d2614 : SUCCESS in 40m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2dfb39bdcff40e4b4367140a0302dd5 : SUCCESS in 53m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ad834fd0d95426f90a93fb5dfa6d017 : SUCCESS in 1h 18m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f6456ffb65934ee39e506ff390e58ad7 : SUCCESS in 1h 09m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3ebba54cf8344e0ba454c901d4d628e3 : SUCCESS in 48m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2e6da0fb70e94bbca6dd552882e7a383 : SUCCESS in 47m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f3a5fa4262c24097b7f35599650ff866 : SUCCESS in 45m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3a02e67a6cac4b5d9c50359091d77e9e : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":10},{"id":"6e32e23ee6860d1a783ee4229a00c01ba6e69866","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-15 12:35:44.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"5fa3e6599ef7b2b8f88cf6fda0400587f41b271f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 13:06:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64fd9e049ad74c6dacfd17381f94aa1e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b135fd064a6842b896b6ba59fcb6e21c : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/052e2ba84f714f399c776f8f8c38ebef : SUCCESS in 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"31e698f6d00fe304e9e3db9ff4e50862d07bd672","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 13:07:25.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"cf0a0c50f1ed910129d49c1aa4e6c32b19ad3e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 14:48:04.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5064185e8895480384c9780196087309\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b02280be5aa34060b9361a2b77dcc3ed : SUCCESS in 1h 12m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0d31fd037c054c2194706616519ead02 : SUCCESS in 2h 11m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/74845c97fa4d43db836193984cc3f7ca : SUCCESS in 1h 52m 27s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b935b6b9b82747bdb7c2b4cf764f209d : SUCCESS in 1h 26m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1e05543221b4fc69796d0275360ef2b : FAILURE in 17m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1205516f1964c199d6e621853db6477 : SUCCESS in 55m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef14e08bc647434c8b331b56195a565c : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/492b290e9c014012941674bb5503c4a8 : SUCCESS in 16m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2ca68a6b7ff46f7a46ba749339b2e3c : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c82277dd2c4813895b05fb97f6091c : SUCCESS in 7m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5cddb7196bbe42169d7b231c1c7196e6 : SUCCESS in 56m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f4750e6ed70b4b85b575dd3d5e5fe3d5 : SUCCESS in 46m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/909c6f64cadd48fdae0038aa7c0ffb08 : FAILURE in 52m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b93d7b62572d42a49c3f13c720d759b5 : FAILURE in 1h 44m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2740e87195d543c2b59a91d320d2b72b : SUCCESS in 41m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/23eaff7c8382457fb912388e0a920f48 : SUCCESS in 50m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/56c9ceba5d3141fcb46823bfac198cd1 : SUCCESS in 1h 20m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e339340313f44dc95870103657e9765 : SUCCESS in 1h 10m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d1de08b05f53434cba1f50d6e23220a6 : SUCCESS in 1h 02m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a839f20a7cf8426b8c2bb5188fb8bbef : SUCCESS in 44m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0fb3a06f4a12480eb39a03bd7454ca05 : SUCCESS in 49m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f44b7878d8943f6b3ccc730ba7af0f5 : SUCCESS in 1h 25m 39s","accounts_in_message":[],"_revision_number":10},{"id":"0f7f87c903612e9da7c76f455acaf21bb8add87f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-15 15:05:48.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"fe2af849fd51c94dfd7d9ffe944143b3b51fa68d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 15:27:53.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7cd6c7845ff4c8f83cb73c02d79c3bd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ce62c6d5caf423d808dd98193649cac : SUCCESS in 20m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b29bad56d2bd4e85994b19a47228e855 : SUCCESS in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"63f3b177848900a9c4b0dd8677233f2c5df61932","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 15:37:46.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"7061cdde09530a777817787df2c8cf2889bbc4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 16:58:58.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad55dafeb06d461692f812cc163f7cc1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/85ef3d2cd33c4a0e97495446ea2df2e9 : SUCCESS in 1h 38m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a75c36d6c0794ddeae641996e5b29920 : SUCCESS in 1h 38m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7b989d70c75542ec883afdcd9e6be45d : SUCCESS in 1h 28m 17s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b661c93acba4fb5a5dd332bf703b379 : SUCCESS in 1h 17m 37s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cd6b6295e0b476bb6bea33f3154b64d : FAILURE in 18m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63e4d463073740b39ea85fc00836863c : SUCCESS in 30m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e8bce1362e421089ef220b49402ff7 : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2988e8a63104bf986be6715e80642e3 : SUCCESS in 16m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7cb01af08144508ad037eb4023fd377 : SUCCESS in 14m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2d05258b3e4e6a907e0501e22d2128 : SUCCESS in 9m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ea10d2b0c7bc49d2ba17fdff09c2b223 : SUCCESS in 41m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/22398ef31a4c4375923dc2e9435b2ee3 : SUCCESS in 47m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3618906019ea4fff94b2a96b1e0722c5 : SUCCESS in 37m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9bf8ac43aebe448394b55d533129f1f3 : SUCCESS in 1h 27m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4957c562a4184fb3ab4e9c01e74678b3 : SUCCESS in 39m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f219bfd275634c098ea780e5cb761154 : SUCCESS in 51m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/42354dd9a45e442da910bf572f03df86 : SUCCESS in 1h 40m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/efc490d5e8384b32b7968917b3acbb4d : SUCCESS in 1h 11m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0c3299d3cecd483cad6ba98ceda81546 : SUCCESS in 1h 09m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/50119e43d150482dae0aa092d242bb8b : SUCCESS in 32m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1b153d65144e422fa27042b7297f8061 : SUCCESS in 51m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3b3ff314ec214e45ab57a45771d583bf : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":10},{"id":"c4af1e3b571ba1efffccddd4964d5ac3954579d9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-15 17:05:52.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"60e8f9172623ee1edad488e653f163c288322e72","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 17:37:38.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"cc1d651570c031ec25387deb88771b3a7ccbafca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 17:45:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af69490066bf4c67829115e09e3872b0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1bf041ac8e04402b59df799f644f4a3 : SUCCESS in 33m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/08ff1f1f937a42c0b532f44b83f81cbb : SUCCESS in 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"77e4b457cd82beab3b821eab00a9ee0c0f940e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 19:00:01.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2198b0cca0bf45c795f77639ac1ce7fe\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3d23a5257a18456babcd0ea308b18a0f : SUCCESS in 1h 09m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cb18fa2d6a9342c58387379ef767d63c : SUCCESS in 1h 31m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca178075aeb24cb190e024f0c81899ff : SUCCESS in 1h 41m 30s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddb1f53630e24b18b04f79bc30ffb52f : SUCCESS in 1h 15m 57s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/712640a712284e3db4d81a600463a4da : SUCCESS in 1h 44m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71266c441d89487287fb2d4c1842f8e8 : SUCCESS in 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bba6744b03648c0a33ba47d2dcade51 : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/940818c59bee4c2ca8a8d5363b02ad90 : SUCCESS in 18m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6899213bae34d0990b2a7b6fb9faede : SUCCESS in 15m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a16bd83be2e64bb784a9074f5eda1a06 : SUCCESS in 10m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bd2095f7a5904399acadd0bc0a0ea7f6 : SUCCESS in 40m 09s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9ad22d0d121a434ca1adbb54eb1f0985 : SUCCESS in 50m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5fec7d36d57c4bfa836aa5459d7acf26 : SUCCESS in 41m 14s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c122741b2fd9472f93ba5a76f213d2c4 : SUCCESS in 1h 53m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad162170752f4f55a681d79d4303bb93 : SUCCESS in 46m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a270a8dbb4e45a98ca60b0a9260a22f : SUCCESS in 55m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/264bb5a4849e480cbc493ce57fcc6d67 : SUCCESS in 1h 19m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/efdae1ba0c974b3197f7ad78bf53cb7a : SUCCESS in 1h 17m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a9dc9e563851457896d0fda827e5f28b : SUCCESS in 58m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/43bce40d2c0140e1a56bc45f1d8e5f72 : SUCCESS in 46m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0da415dca0e1417ebde2e397ab5799b2 : SUCCESS in 49m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ed8556f91b5f4ca88307751cd238301d : FAILURE in 1h 03m 41s","accounts_in_message":[],"_revision_number":10},{"id":"21f45ddd7358c18f25041dd4a0e602da4d3fa038","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-15 19:05:56.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"70060a5b388917907eb7fd0378e2286a323688eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 19:34:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8313250302e84ef3b540767302098d0c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42fdadb4d610415082c3a93a211f80af : SUCCESS in 21m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5efa7ec21cc84337bfed61eea628f788 : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c56d488580c2b1445a6e95c52c77104fd96fcfc4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 19:38:01.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"cb9d4f93df27e8cbb3025bad84224c8dd0be8505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 21:25:52.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8298076d087f492781a20109e8ca8687\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9eac4998168d48f98292d2a1a7c18f12 : SUCCESS in 1h 47m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ae7387d1e3a4c109bbc51d4c805737e : SUCCESS in 1h 46m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/81fceee642794e87822bd5849d0a5ea8 : SUCCESS in 2h 02m 56s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5acbe4d923bc4593b0d6b703ec1f98fb : SUCCESS in 1h 25m 46s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0853ad7ac51458cb474f438201f8f9a : SUCCESS in 2h 18m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba346035586d40ff8a8ae49c309dda94 : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7060eaa19722460e921842df61194bca : SUCCESS in 8m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07947cba02234ac395a0e7f7f6b10e9f : SUCCESS in 20m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b875b99c2ea4c4490879d3bdd5b0a0b : SUCCESS in 18m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5040f7d10d48dea70fd26f322d77c4 : SUCCESS in 9m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6449fe2dd044478b83716ca5467d77ec : SUCCESS in 45m 09s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3f29e264501b49b48fc91376ca3dc273 : SUCCESS in 53m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/93a6b281823c4f5f850bae3578968f91 : FAILURE in 56m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6bba9b2ddd174e31a7ce1bcdada04ec6 : SUCCESS in 1h 23m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9a8e16028ef45f480c0b5f4a67eef43 : SUCCESS in 40m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ed27e52e7ad49069b0dbc49e31477a6 : SUCCESS in 48m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f56cd54a323645b2b0c0c9a4f37806f8 : SUCCESS in 1h 47m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c746f218a4d9451abda91796f67f0769 : SUCCESS in 1h 36m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d0b3076c10c0470ab8a1e87762f5de57 : SUCCESS in 1h 03m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b7a99fbcd83d467ea5632890ddceac6b : SUCCESS in 1h 06m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a5378f3620cc488f895e51e580319e09 : SUCCESS in 1h 08m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/33c408bc8366408f82fe32b3d632c0ac : SUCCESS in 1h 28m 05s","accounts_in_message":[],"_revision_number":10},{"id":"791f7264b77b86b9c17fd6eed25c459edc383662","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-15 21:36:02.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"f6ef608220d388fd0601c2ccfc45b30cabfc4f7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 21:58:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1809977070584e2dabf81d9475806363\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9e9750310cb41a19469a683719c4eb8 : SUCCESS in 20m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b61224588bc541a088245736e906c324 : SUCCESS in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f711e84a16f595802cfe84dd06a04e6bf933f76f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-15 22:07:22.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"0f996300237cacbaecf602c65e8a8579a37e2577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-15 23:33:28.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6832c400d7d04254836bd82b2b75951a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/276ef9e07f98403e9d871910ff9f8deb : SUCCESS in 1h 08m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/617872b32a70436d909021add92c6d53 : SUCCESS in 1h 40m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/72087fa4ac204f07a67efeb25261268a : SUCCESS in 1h 32m 40s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7afdd1c2bd024f709488aac14b42a332 : SUCCESS in 1h 19m 37s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d5b7258bea045968f1e389a2dd62ee1 : SUCCESS in 1h 29m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7c1280540c74565b64399aff81c95e0 : SUCCESS in 49m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e93015b3154ec38eca248bb09eb179 : SUCCESS in 11m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e69f1d95644c4330b07d46a35b8f874b : SUCCESS in 25m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/864f8cf30c6747be9164afa17654c052 : SUCCESS in 23m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129605ec94984f599dee07d5259a93a4 : SUCCESS in 9m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/655cfc592e954d2488f0895b6214f675 : SUCCESS in 59m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/41315e523ea547a38a74c1c9245abddf : SUCCESS in 38m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f224d8d02dc94c0886625b2d457d9fc3 : SUCCESS in 37m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/12d36a50c5f441a89ce5950a35fd9de8 : FAILURE in 1h 55m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9dd4300827f40f183dcf3c387946e47 : SUCCESS in 49m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f269d548e09409ba45685c3cf5ab7d0 : SUCCESS in 56m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5463301411404decbd2bbbb57de5c1af : SUCCESS in 1h 22m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/42a9aa49dba741e19d3db7c1df82be9f : SUCCESS in 1h 18m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5a5c92451e44e4c8a6ce4ff38f95c74 : SUCCESS in 56m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/20cf127ad8924205ab17417bdb25f5f2 : SUCCESS in 49m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b7d2246d6020408b98760f3715e090d1 : SUCCESS in 1h 01m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f6b3282084844411a25e35aa7aa7a443 : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":10},{"id":"ff27068b32ec23ad2c695ea1520fb80c36a4141a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-16 04:50:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"21d763bb25b4c1797a04d70201a78694495c947b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-16 04:50:34.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"dbaf9b7096833261e7d6dd1864a99607d13fc5a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 05:14:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/513f2f8b95d04f22840bd26ef02fd557\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ecc3b6ff8ae42deacb0417311db5b58 : SUCCESS in 21m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0412aa8a14f948c9ab0d9cc51cd6052c : SUCCESS in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5ff199037cc4ba4cfa2fea31c920211d8686718a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-16 05:56:04.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"49007f92aa70327f57f228c94a689cca7bcbb8f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 07:13:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e51076541333453aa2a916262a0ba151\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8bdd5c93cd9249259adcf3ebd224e23d : SUCCESS in 1h 18m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fa11675619c143639b0940bf5e7b8794 : SUCCESS in 2h 02m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/516a25a040dc464ba6d3f4e9ec0e6e3f : SUCCESS in 2h 20m 54s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e147541e0f14e77bd0f07e54ac7dd6c : SUCCESS in 1h 47m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d07f873739f4ea4889da05312a9d19a : SUCCESS in 1h 53m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6768919226d4919b89b70107f278451 : SUCCESS in 49m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f4949ff595433881185df3568b311a : SUCCESS in 11m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78d529dceb0340a18eab3018f3e1d797 : SUCCESS in 23m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2f82a1ffd5e4ac1a4579ac5f1efcce8 : SUCCESS in 20m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c86381146334bc6b881a0026977815b : SUCCESS in 8m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a12a709bd5d74ba6b0535a71c0d664c0 : SUCCESS in 57m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3724f0df02cc462db5da66822f3091ec : SUCCESS in 1h 00m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bcf615da88274d23a2feabb286718805 : SUCCESS in 35m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a69686d12c374d469b6b7001cb1892dc : SUCCESS in 2h 01m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e08f8a12801d4735a8049051b38d1ff2 : SUCCESS in 43m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fff4314452c94a8594c622f7713f99a0 : SUCCESS in 58m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5a52bab781f04d7aaa9e71302ffec908 : SUCCESS in 1h 40m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0597d2b7926847c687943994fd76f1fb : SUCCESS in 1h 12m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7bb92bc8508e4f9c96ed2318ca3ce075 : SUCCESS in 49m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/58fd93164431478cb14913a56dc9667f : SUCCESS in 50m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/976aba23bbe24beabc66622eb7790b63 : SUCCESS in 56m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7c1149f736a14e6ca8d969da8049ca36 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":10},{"id":"b677a9c31bf9f930f5aa8ded91499c4f90cdb50d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 07:13:44.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ab5c107be736cb5f6d05de11f1b1e40f4e4bac69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 09:16:11.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb259c59c69146ee91b8bbac1b68d4ce\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/22f75773593e4e1a9e90e6f7bf489f21 : SUCCESS in 1h 38m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cb4d0e50d48744c9a4c95c9529060ebf : FAILURE in 1h 51m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bfbaaf38624946b29f3f0f8d0c224240 : SUCCESS in 1h 34m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e47528238874f4b9acc440148cf8798 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b0b20ecfbcf42b9ab3530a9db3b558b : SUCCESS in 23m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76a7a4b5d9904f998baf7b1d538d8322 : SUCCESS in 19m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c34a76b4324256b5a81691d88a25e8 : SUCCESS in 9m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1190bb0336ff4769b31273cfc22528f6 : SUCCESS in 43m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c9bbd99537a74ef08ee0b721deb907ac : FAILURE in 1h 38m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cc2dc0b294cd46b6b626be3aaf7a1873 : SUCCESS in 1h 48m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a438cc4b146c4951b0d88fdceecf01b7 : SUCCESS in 1h 16m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/743945b385de46019277d428f88cb4d3 : SUCCESS in 55m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b561154ff00342c39d1a9f3ea9aa6be8 : SUCCESS in 52m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d8f49800858b475f8cbe6e7c2a80ee30 : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":10},{"id":"0aae94f8df4d182af2a1a659ff1dbec675669bbd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-16 21:19:20.000000000","message":"Patch Set 10: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"68a79f2dc5eb0d150785a1d3bff646ea22072b55","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-16 21:51:22.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"8d97074a9faab5cc77ce4ad9cab156b503699a58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 22:09:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8681290c986469285bddf7c611e7d86\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba02368522e74f50b360e13fdb6c9d96 : SUCCESS in 30m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80b6a12d170740ea84af0da3589ac667 : SUCCESS in 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0c2e25fa7de7040b47e144b2a34aa9d4df944f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 00:38:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf468be339994ff8b505674ed303a305\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a9a61a9bc7584d0fa9163595b1e19a08 : SUCCESS in 1h 28m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ac6458ea1554052a4babb202a729ca7 : SUCCESS in 1h 23m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f6e35796f2f4b6d8081ebc8661b4273 : SUCCESS in 1h 30m 25s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e43805b9054427bbe023c93d436c134 : SUCCESS in 1h 07m 54s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7676e3d6a3241149d58f05279c2ab06 : TIMED_OUT in 3h 06m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/685b78bf51994106863872bb854c5b54 : SUCCESS in 35m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e84cbb681c1474fbbcc160a5dfe9ca6 : SUCCESS in 12m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a6b5bb4a028450192b0ec9e9d841547 : SUCCESS in 23m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38c4d1708dda46a0a71525daa346ba59 : SUCCESS in 21m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15debc753014a25aba6c4980674708b : SUCCESS in 10m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/402c930f0e6c43bb98d88ec3576279bb : SUCCESS in 52m 23s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b8480fe015bb41d896816e80f7935533 : SUCCESS in 54m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f7f6abb387b54a9883c5914afae35cf4 : SUCCESS in 43m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/acf0b3d58e32486f9f816aa826a5b99e : SUCCESS in 1h 44m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a1323f34539477e9cf531c5549ae64b : SUCCESS in 52m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b3c354563fc4c63a2923eae5030cb7f : SUCCESS in 1h 17m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/faca46e1b09d4fd9a1ba650155d5a0a8 : SUCCESS in 1h 29m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/08bce5ae663043aa806e4ab2f4cc4cf5 : SUCCESS in 1h 45m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/97f68c76f07544f58628493104e9a63e : SUCCESS in 53m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ae97e13dc00d4f1b9d1299eb85ffef73 : SUCCESS in 52m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/72358bf59c654e34be210a38b180d76a : SUCCESS in 1h 02m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6347989b629849bba1b1e3c6b4880062 : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":10},{"id":"bc237db259792eef91df4d3aac8dda6cd78c5110","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-17 08:49:18.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"cc4731c28b15f3597caf3c196237c28b40f07f9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 09:36:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87eab454855e4b7b9d44b2703910fb01\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6db4967113ae475fa3ef91907dd1d9f0 : SUCCESS in 28m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8391d6972f794831ad18e4590cadb992 : SUCCESS in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"703ea8b6973b8a3fce6fcdd59ae1aeedc72b0c65","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-17 09:44:30.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"c632bb1a14cee2068926cc085dc232f704e5a528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 10:56:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f25e6975f83543e1b242f067430f5119\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dbe10a530377418c97b464b025c2b8bb : SUCCESS in 1h 53m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/80e6194fbae049f4a4711b6ca841e460 : SUCCESS in 2h 01m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ac860cf36c704dd3a9e4a9cb16b9633d : SUCCESS in 1h 48m 40s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f68d037383848c7af2ec731a0583766 : SUCCESS in 1h 55m 00s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12a8243fee194f0ebd13b56f8e3cf6a8 : SUCCESS in 1h 51m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5683bec414d74376a8eebd36977b20a4 : SUCCESS in 57m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594c4b140c5d4377a1f566c43a74ddee : SUCCESS in 12m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/468f90ef6f7548748e367da9abd4acd3 : SUCCESS in 24m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6355c485dcb94516a107479da6ceeda3 : SUCCESS in 22m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4713fe5d28483ba02447b94eaaf7bb : SUCCESS in 10m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3a1ed89896fa4573befff2d5b60bf10d : SUCCESS in 57m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f1bf5150e90d4d608bc0c2c24fbc76a1 : SUCCESS in 1h 05m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/33746c15d18f41c1855e97233724e33d : SUCCESS in 46m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8e0566a0212e4324a95fdba1c639af97 : SUCCESS in 1h 45m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/976e794e0cbf461d81cfeee28f01ae14 : SUCCESS in 52m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0ba7bec453f4b998911ad2ac54ff9a2 : SUCCESS in 1h 25m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6f9d5de14eee480cb875897e846d9984 : SUCCESS in 1h 49m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/185d513454834c08b3bee7042f6c53a1 : SUCCESS in 1h 38m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6f206e6224bb41e2ab4bd3cdb920e41d : SUCCESS in 59m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2303a5f997ba4426b9f5c735039bf661 : SUCCESS in 1h 02m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7d3fdb7710e74fcfa7a3aab0c81aada9 : SUCCESS in 1h 14m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9e73aabb530d4daca542a8824c658bd9 : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":10},{"id":"b8811117cf8b5f923caaaa57ac73ecc931b8066d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 10:56:59.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"21d03b2fd6d984fac1bebf62c7b7df2880464b39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 13:33:21.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cde5af2952145609a666c632060144e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c4da0da35e874950a11e94e98a1766db : SUCCESS in 1h 26m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/461d4500ad4e41d781bd2db4fea5f3ae : SUCCESS in 1h 57m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7fcd926d0ca2439bbd12023199c7b25c : SUCCESS in 2h 34m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e130a3301dc477ca3852a137523169d : SUCCESS in 9m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddda81ac8b39438f811143423efe08c4 : SUCCESS in 22m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a24e143423745bcbc1cf641b8d61276 : SUCCESS in 21m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a842dba1d1394ff0a7a67e82426ea882 : SUCCESS in 8m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/61295a7632354f76b1ce18d582e58393 : SUCCESS in 44m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/143aece4392443b780aa8847412cf6ff : SUCCESS in 1h 40m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/facd1986e5c3484cbeb8ede7c6d22f51 : FAILURE in 49m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4dce4f13c7854fda959542315155b55e : SUCCESS in 1h 15m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce4c88a7a86345fa95b31e5395eb2427 : FAILURE in 41m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2b15d66c1d394242a74446f03e121bab : SUCCESS in 52m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ad36a4ddadf4332aa90290b713767a8 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":10},{"id":"5f65168661ae681cc00f7c9084d5eb19a8854b06","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-17 13:49:27.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"f3f8486a5c0eaf8ef5675db7a6fc8523285b2dae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 15:06:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50f0849e7e764543b00134d4688128d5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53192892b17843afa3b879394db88f20 : SUCCESS in 30m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b333946f54b744fb96621245341a9985 : SUCCESS in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3af6ce62ddcfbbf6df628069e8dbe85b17ef6c3a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-17 15:34:05.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"29b8c650bdc1773e2c5147db347422efd74d6f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 16:38:22.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbdc051517c84525ae7de2729751fb4e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/958e22c872324ace926105fa9a75ecbc : SUCCESS in 1h 54m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8787fa29dfa84aeb966a152d2234afac : SUCCESS in 2h 45m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/593287713e7a415d96ee9b7421e47932 : SUCCESS in 1h 56m 46s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/027db05c9aad42328fc956f655009205 : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42f12f0b797a40ec8fdd990ce44035f3 : SUCCESS in 1h 49m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c3b505d0e564b26aba463a21c65a85f : SUCCESS in 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b337906cebd1437d85c4d60ea10e792c : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c11bcf6eaed04d06bb9a0ccaa0b12f0f : SUCCESS in 20m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7944eddf686d4bf5b3b44704662932b3 : SUCCESS in 19m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5811dfc588416d8405f88ad4c12955 : SUCCESS in 10m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0e07d129fa0d49e5a0b68303b9f315f9 : FAILURE in 54m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a3e54de9ce2e4460b0798839ed51c7a7 : SUCCESS in 1h 06m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4392bf0d391e444d87d342ef857f148b : SUCCESS in 57m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/aa11cf1b377740f9a58cd630d8e8f37b : SUCCESS in 1h 48m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/97f6a25dd361499c8404f7b673210bd6 : FAILURE in 49m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e0e744a29ab42d79e25862c026c5dad : SUCCESS in 1h 10m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8058015d37e247d7ac05e37c6067c3bf : SUCCESS in 1h 49m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c5574e91df644650afd293e19f52a120 : FAILURE in 1h 16m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/340b37aeff264d9d8552f06ef154be47 : SUCCESS in 1h 07m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/150570c993784adaad3f3efe2885b8c2 : SUCCESS in 1h 01m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/72baa2fec7f142619770f1e87c31ab55 : SUCCESS in 1h 10m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8dada2672a344d3e8c186cb9de1b8e6e : SUCCESS in 1h 10m 27s","accounts_in_message":[],"_revision_number":10},{"id":"35004ee7aeeac69751f87ca091e9c4654de5e4a6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-17 16:49:33.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"1cea4e95781e441e523419725d5e93962b47edb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 19:18:19.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3890f77c71124f1bb4cd2cafe697847e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/22c1ebbb5a394198a03a3bdeab75e3b9 : SUCCESS in 2h 10m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a495aa9db21948f69dad4d320cee7fb5 : SUCCESS in 2h 18m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/74feddd504ee49b19164f98f968a868d : SUCCESS in 2h 26m 20s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2386bf5ad39a4740a285391be00fcd48 : FAILURE in 1h 44m 20s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f6c5a77e4e0443ba24b8d1bfefde416 : SUCCESS in 2h 13m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4995f55e79e48c190f6b5bd27885cb2 : SUCCESS in 35m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad89434aaff34a29905756776798025e : SUCCESS in 11m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f9ef9d88e824416b72b04ca24a8699f : SUCCESS in 17m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80499c85976a4bb08f4134df63c0ca0d : SUCCESS in 20m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/721850f814ed4a0e8e9afbf75220a5d2 : SUCCESS in 10m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/34cc7d745da541fe933ce233ec39f698 : FAILURE in 1h 04m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/70a2514eaf9d42f5ba82a0cba3e5564c : SUCCESS in 58m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9251336e67d4407e9eb00e57d0e7a074 : SUCCESS in 44m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fb071d73c52e40ebb3312f8cfdb1d276 : SUCCESS in 1h 39m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/16bcd7a53ea949299abc7640b9732b1a : FAILURE in 44m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a6913c1669944dca145165c5a64fe58 : SUCCESS in 1h 06m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cdc19c86c09a4e519b475012ce0a04d6 : SUCCESS in 1h 31m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a6290c0ecdb7401aa7fdad35e9647ed8 : FAILURE in 1h 23m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/963b569bd2754229a5192d1e0da0a4e3 : SUCCESS in 53m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a0c9a9fed1b5436786892ae4d837e7b1 : SUCCESS in 32m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/295e3af738754a66b169c3a82a494e7d : SUCCESS in 1h 03m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3bd4af95689f4e8daa19ddbff57ae6f3 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":10},{"id":"ef62664ea7607e6a53541d3f27174a20ee18603e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-17 19:19:37.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"2bc334a20e2e42b0851572e27548cbde0ecef261","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-17 19:43:33.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"6f969beed58b228377804026eb72e06b2ad70874","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 20:48:31.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef05f48db18e4d7db809d1372e743701\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51b3a3b4bb13401ca5fda2b08d547c70 : SUCCESS in 23m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7f5c9ffa0364e0f804425d4dc18581f : SUCCESS in 30m 36s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1052acc8c8358d7d05ea737618e695396718565d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 21:49:32.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef0aaad55b1f46dcac50c11adaf2e603\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d2a7ed09a7934891bd4e9107a465fbdc : SUCCESS in 1h 48m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c2fe193c78d247d9aa22c96b5316606e : SUCCESS in 1h 44m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0ec823b32db64058bc11f51b6bf8d54e : SUCCESS in 2h 06m 20s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c138e48cc7dc414eba1e1d62588fe230 : FAILURE in 1h 25m 18s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b04c0369b744885988022dabcd6517b : SUCCESS in 2h 27m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a3d9db2458b4043bc3853130079216b : SUCCESS in 33m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ddcb90128f34fbda7eddd2f4f42073d : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41872e5537964f159d4a31cdc6be29ba : SUCCESS in 25m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60563621a0e44f0ba0a8b06907df1560 : SUCCESS in 21m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3880176098324e6fbfb0258109feb866 : SUCCESS in 10m 21s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a793c1e6acc442e2b597f53a8c779fed : FAILURE in 59m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e7740ecc3aaa4c76a5d3bbe2bbffb352 : SUCCESS in 1h 09m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc94d456495348158e79c1e5b396e069 : SUCCESS in 1h 02m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f8794ca6194744aab155e32e7f4a1c04 : SUCCESS in 2h 06m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0b53046f1424b8fa5a9ae76ffd00f85 : FAILURE in 55m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8748593815c14cbf92b13f36b074989e : SUCCESS in 1h 20m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/30a3175cc57243efbc14c161958115db : SUCCESS in 1h 46m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c13fa890cdbc43ac88777e5344692d4e : FAILURE in 1h 23m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7e63d9eec4d24a53a952592ee1678d85 : SUCCESS in 58m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/19a55ce1b7f74f1ea8ccfa76a13389b7 : SUCCESS in 1h 04m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5425a621de3426c81f3bb41aaec06f0 : SUCCESS in 1h 02m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/67ea7b53bbaa485fa2fafc4febbf4801 : SUCCESS in 1h 09m 53s","accounts_in_message":[],"_revision_number":10},{"id":"00c85a4807ae5c71299adf8c9033490d7c61b5fa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-17 21:49:42.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"568050d6cbbc51975eb3d6e2461ed7b17825ca99","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-17 22:20:18.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"c6e7d9e270856d316cd8c0b9be9db3f8ca7c5c72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 23:45:44.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f01e8037f08746868eda691ae3c97082\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e1126280bde40268eb4bf46af440d80 : SUCCESS in 29m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4d7257ff69b4785842a5d45e04fdfa5 : SUCCESS in 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e52f23b70a503f410a56fa31e08b65b92bb95ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 00:37:05.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1323113c61d8462892264ed1c1bffb35\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6e3299e81f964f05ba7766b5d936c467 : SUCCESS in 1h 13m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f3d8dacf08e142aeb21f154403d287e9 : SUCCESS in 1h 57m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ec92d739e2a34f04866af207866202fd : SUCCESS in 2h 25m 38s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae2524da4bc4472b8e9b8711bf158954 : FAILURE in 1h 51m 43s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed32b23f29c04508b6830313c9d11eb6 : SUCCESS in 2h 43m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7458f80c7a7e4f60854da7dbce6e8440 : SUCCESS in 58m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/057b867b5f13424f9ff14ddd735fa67a : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf1d8d39020c494989caf11c3fbce9c1 : SUCCESS in 18m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5af29fdbc7145289cffe62f199bc4fc : SUCCESS in 19m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c9f877621145718abe3cc5501a0b39 : SUCCESS in 9m 45s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6a2af39fbb6e4ce8afd6f2437f2575f5 : FAILURE in 52m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/430f408739194013b04b8cb3dc9b0df4 : SUCCESS in 1h 14m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/65488cce479645b5b4d55a5831f3891e : SUCCESS in 56m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7142e53b0535447a90fa77da07fe1fdb : SUCCESS in 1h 54m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcf0fad308274fa6b2b6fabea8bfaa98 : FAILURE in 57m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d63eb954ec554aab92a5cbbcd2cf8807 : SUCCESS in 49m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f215d3147b6a4d3fb23bd1e85023c346 : SUCCESS in 1h 24m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f2fdd5f7e03940a482bcaf1af037f6ba : FAILURE in 1h 22m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fdd381dc41504c9f92db13c2ab204c05 : SUCCESS in 52m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b785ca380f8c4076bf4f0036066e3ae2 : SUCCESS in 49m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/193587b5a6264af5bf56fe71db1048ee : SUCCESS in 55m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/296d43514364406ea120857fcc75a120 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":10},{"id":"aeaf186d7fd0bab9e548923509f85f3f59300eb1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-18 00:49:48.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"7fdce4a7cc0e08d7d02b37fe1530b870ee30f877","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 01:14:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a25a434dda3471390d53d1a89c82365\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a9e27523b3744a79a449ef47e963510 : SUCCESS in 22m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a64cbecb78964b5ebe21de4a3221b138 : SUCCESS in 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"84b142db7ec5f506bb7d401329c08d5033830bd2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-18 01:22:21.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"1e88f0a56e81b38c27fe0cbbf0bc1bb915a7fb5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 03:56:12.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24c54028bc264e7f9b1deddca2e03932\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ef9c6eaa7b0b42d6be038c028d9d217a : SUCCESS in 1h 44m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4c1e3982f6b3441aa3300fedfabfa7fd : SUCCESS in 2h 12m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee67318f6fb24046846d294859ac2ea5 : SUCCESS in 2h 15m 19s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/090775bfdd3c41b8b196395c7afb31f7 : FAILURE in 1h 55m 12s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f5916130e7f417abd8b0ef33ff9f0f5 : SUCCESS in 2h 05m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d77f43f107e49eeae977535d21131d6 : SUCCESS in 29m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5990ffc9e86f4e759001d5cab7f228a4 : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/510945bb15974cf880412ff3768da903 : SUCCESS in 22m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa04968bcb074cc0a4d0daaacf4a7396 : SUCCESS in 28m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1677a88f9f3948949f1d29b02b1d072a : SUCCESS in 9m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0825fe2bb7384090848589a00377fd37 : FAILURE in 51m 45s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7525d80282c744eeb3dc1e2d39d56e36 : SUCCESS in 1h 06m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fa68468505464f80b4df31ec92134552 : SUCCESS in 1h 02m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a8a53971fed945c9b110a025bf7fdd0e : SUCCESS in 1h 58m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4850de879ad408c94d862ecdb2555cd : FAILURE in 55m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/44aa9bc77b854245916e79067b18abd1 : SUCCESS in 1h 24m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/436ddc0b4da242c2be1f15ef6dcba649 : FAILURE in 3h 05m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e5ea17a436814ee883e383d04a23579b : FAILURE in 1h 52m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e9b732872eef4d998ee4bf4280133e43 : SUCCESS in 1h 14m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/11326069a314414d8c8739c20f06099c : SUCCESS in 1h 06m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4e0b26f8a1f4eecb83e6e4fc9478eae : SUCCESS in 1h 04m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bfd1f451d6f4497dbfad20032310a988 : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":10},{"id":"f2046bba5a5104d8ef1334d369da338cf2f27c06","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-18 04:19:53.000000000","message":"Patch Set 10:\n\nrecheck functional","accounts_in_message":[],"_revision_number":10},{"id":"b61350c4cbd3eb6cf29417c9f7319cb371d45f57","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-18 04:24:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5b4160b2bcd256bab5a6e52d07404cb96d39346f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 04:50:26.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/967372cade1f47e59e98122437f351a6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f6655c0e904408fa5c8cf8e341f954b : SUCCESS in 28m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e8ffa0df0975485bbb4125f2b3f4d1c1 : SUCCESS in 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"438f66d19966614d66d2fe5208466d1938dc7c47","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-18 05:03:12.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"4b66d3dd44168d9dcfa974f779eb4c9a0dc22032","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-18 06:32:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4dc5876f103e47dd484b265ae23d78154b8cda8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 06:38:33.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/124215b90ac64483a24cfa905fb9e81e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c1911be87f3346edbee39f06cf4a509e : SUCCESS in 1h 37m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78265a424cd346f983b001c55ccb15b0 : SUCCESS in 1h 57m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9e31e8ee4720440998572bc59d1a4dc1 : SUCCESS in 2h 09m 25s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e140e39bbac481a9e42c2632f69dd3d : FAILURE in 1h 28m 54s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d68fbe2a6a55458f8a34899e5b229e97 : SUCCESS in 2h 09m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/124e16c1521d4f88a8c101d12b5760cb : SUCCESS in 33m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f670da72e440af8fe9ea708c007a86 : SUCCESS in 9m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c25e3b9ca8148229ffc8df4f88ab2a5 : SUCCESS in 20m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4e844a94ad24ba187f19db5714e80c3 : SUCCESS in 16m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5898f16a9fb4deb963f790344358aa2 : SUCCESS in 12m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/061bee57d3744b9795b479fb0f537c81 : FAILURE in 47m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/68288da03e1c4b8585b468964d45d013 : SUCCESS in 51m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2bf89786ff5c4059aad2fdb5fe800d73 : SUCCESS in 43m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/22e0359745314425b93949c751010571 : SUCCESS in 1h 50m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/448581c0698d410981c0fe452e9ef388 : FAILURE in 45m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e5af1c7a1534b67a377f7cc383a0ed9 : SUCCESS in 1h 09m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bca51f23e4654859a51c362d6295c84f : SUCCESS in 1h 25m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/01685929cdcd4b7fa00d9865ef322e1c : FAILURE in 1h 10m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4273e72a3af841a197d7824574fcbbab : SUCCESS in 45m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7ac6ddc4607143f2a998ebcb857ae38f : SUCCESS in 47m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4d334f015da84529b55ad6fad2b2148c : SUCCESS in 53m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f2cda3e194d54e45a39f07220a68c613 : SUCCESS in 1h 04m 52s","accounts_in_message":[],"_revision_number":10},{"id":"ac85943e4b9f48ae8f977ef1dd026f83701fb32d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-18 06:41:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8533142bcf6523c7f96eae9cca119cfdfdb6d22a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 07:06:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e11a45462594ac4a9308771dcaca88f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84993400dde74a6db5827e050e5a6325 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e64df1f3893c474789e07e3018de0d1b : SUCCESS in 20m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c38b49bc047eff9c59788480dea5fa57b3895acd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-18 07:14:08.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"cb795cbe7fcd6b8dd65ebf7eacbff789784626c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 09:07:58.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed6a3eabfa6b483bacc7f8f8ccdec5e7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f0e2e334d4074ea5a9d640e7704f076b : SUCCESS in 1h 11m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be342eb932e3411a8084415bd0101edc : SUCCESS in 1h 25m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c0d347a17fce4fc09ea05624cbebb257 : SUCCESS in 1h 52m 24s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/655e9f0037f64fb7a197f0a7c5e95c01 : SUCCESS in 1h 40m 27s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfe34823c1c84062af8253bb80e3012e : SUCCESS in 2h 24m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bae996b44524d20913ae968d4cf7a4f : SUCCESS in 52m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe065b303e5f438baf471786106eb532 : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e8adc27cfd4403a91c4f397cdcfbcb9 : SUCCESS in 22m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4284075d4d024ccb9af6792300a59231 : SUCCESS in 20m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d54ef246b7d492db512afa0cf61c799 : SUCCESS in 10m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/19e0f9d42de644cfaaa5a47ab0e7f94d : SUCCESS in 44m 23s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/40136403ddd54331a47a602bcebaa34f : SUCCESS in 1h 01m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5b285832b2944593b6954805c68600dc : FAILURE in 1h 04m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/91d1cba0a6c44c638f8f890c2b10f44c : SUCCESS in 2h 25m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/965a6d73061249bf9a5a30a407773ae2 : SUCCESS in 48m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f0e225c5c824e54ba74dfba5aa87d68 : SUCCESS in 1h 06m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6dc1635f5b1e407b8c8851fa46d7e6ab : SUCCESS in 1h 39m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b04efc1ee7aa4ed292fc53b14d14137a : SUCCESS in 1h 12m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/db266b3260c84fca9550a9747f28cfc2 : SUCCESS in 1h 08m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d9276952d29448fabda001b2efb3d9db : SUCCESS in 1h 01m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c85298a0823a4c76bdcfe341366c334a : SUCCESS in 1h 10m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a811327040ac43968a6107ebfca5dbdd : SUCCESS in 1h 12m 14s","accounts_in_message":[],"_revision_number":10},{"id":"7bbe01b90b5fbd4e3baf04c326b58ba8b65a5f34","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-18 10:09:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"50a87a567a6bdb9507923128e7f491604444535c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-18 10:42:27.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/13/936813/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"6c08028ad26257ff071c3ef9001224c1c1faea59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 11:11:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a36519ffab94f638e777506e48d801f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc38a68cb2664b8abec08279491c08d3 : SUCCESS in 30m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b3224eb3344b4f979e38b4406186b6dc : SUCCESS in 29m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5a3018dc0a37fbc0dff79b1708a60c843d703886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 12:21:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c961e58fe54447ec93a0275ffa277f5f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7cbd8b376eaa47b9a595e9eb9f52a3fc : SUCCESS in 1h 25m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/87d9b924c3a74e42840c672e5ea0adbe : SUCCESS in 1h 39m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7ea90c5e54474798bfb9feecc685d044 : SUCCESS in 2h 08m 53s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23680a12190f4d29b2a2dd29a4d53193 : SUCCESS in 1h 25m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2704a2aa62c74901a1828e248bbbdfe4 : SUCCESS in 1h 54m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e64b7cd986764d0a88e1837f82ad89c3 : SUCCESS in 34m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a43782214e54e029b523ff9bb54ad2a : SUCCESS in 11m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f394b96336c4723b876c6dba3ebcc09 : SUCCESS in 21m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a92a4eebe1249309b76936de5c6bf7a : SUCCESS in 21m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40600a107dd040f79c1ec37a2bdef4b3 : SUCCESS in 9m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5555c628925a40e0b6dbf56125703b31 : SUCCESS in 47m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d858613f10844182b9cd1db1128cf4b5 : SUCCESS in 49m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7503499f10214281adc6bfd722cffe1a : SUCCESS in 45m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/da8be90905a34d058e22340ea5633152 : SUCCESS in 1h 34m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3fe54eb8f4e48539ee2462a0dd84dd1 : SUCCESS in 44m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4b23995edd64d11abfb9863f90f7532 : SUCCESS in 58m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/443ccc22ef144cd296a4c324fbe30817 : SUCCESS in 1h 24m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f1cd943158a542a2bbe522d5397805f1 : SUCCESS in 1h 30m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/41df1e76c1b6493382f05015a770baab : SUCCESS in 50m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b5443db4f996481fbe861fb30e7b549d : SUCCESS in 52m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/798148b272ad43f09fd06759ec94414a : SUCCESS in 51m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/43f1c7ec627744c1bbbc347db8e0559e : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":10},{"id":"5d35a9322e534a24d86d8970f314ad3cadebe78f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 12:21:55.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"0e1b1f71c873103d30106ac3b54c9f3ba57b5d45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 15:20:50.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a23f727110346e2a905c9d888e010d1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f00f9c14a38c429c9fa6f2d0d9f977f0 : SUCCESS in 1h 41m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7e5314b44c734b72921a4cd527e9613a : SUCCESS in 2h 21m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4346011fd96947fe8fd1a84fceb55cd7 : SUCCESS in 1h 30m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ded24b675694d9c810157d22021d511 : SUCCESS in 8m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5384f97b6776468286bfda4a5b37b94d : SUCCESS in 23m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dcbf2519230548079b4be6cdd7930bf5 : SUCCESS in 19m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880aadb4285e4d078cb0a2f640bbd53f : SUCCESS in 9m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d4827d1abc0a46ffba1deb87d5e9b9a4 : SUCCESS in 1h 07m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5f059f7172824db482629c580cef7f5e : SUCCESS in 2h 32m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d38c91f4f1c34f2a9c36ca2abe766033 : SUCCESS in 1h 40m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c2f4898b4e64dc9ab1dfae235c79a2a : SUCCESS in 1h 17m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/752a725247ef42e7aad37bf2cce96c1c : SUCCESS in 54m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/576255e66cdb4138918f07be7a8e1491 : SUCCESS in 57m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/941b17360319472195a18c9857557da0 : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":10},{"id":"09513811fabb67078cafce2b29c11f91dd65c631","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 15:20:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"ec3d242fdcdfcc02ea4a6506a10912bf8920f1f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 15:22:21.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2610270ccceb49e29547b956a9dbb3f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f6bca3db484999b12bcabd13021d47 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"dce01d7550642ec478eaeb19deded4c974134ea9","revisions":{"a4982d5767e5ce5dab4713de398ce5ee56bc2289":{"kind":"REWORK","_number":1,"created":"2024-12-02 09:30:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/1"}}},"commit":{"parents":[{"commit":"8e8a9a30885faa220831cb52801ddd5d79ebf177","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e8a9a30885faa220831cb52801ddd5d79ebf177"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:29:57.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4982d5767e5ce5dab4713de398ce5ee56bc2289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4982d5767e5ce5dab4713de398ce5ee56bc2289"}]},"branch":"refs/heads/master"},"55e0610de683ca018905766cd243acd23f394d46":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-02 10:30:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/2"}}},"commit":{"parents":[{"commit":"3ff99931ca7f8fbb5ae5db7084f465c092a80784","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ff99931ca7f8fbb5ae5db7084f465c092a80784"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-12-02 10:30:24.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55e0610de683ca018905766cd243acd23f394d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55e0610de683ca018905766cd243acd23f394d46"}]},"branch":"refs/heads/master","description":"Rebase"},"4c4e5f2c4c35523d6662351960ba87413e0584a9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-03 07:29:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/3"}}},"commit":{"parents":[{"commit":"acf7b145d4a903d9c5bb410e6437bc98772fb354","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/acf7b145d4a903d9c5bb410e6437bc98772fb354"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 15:42:57.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c4e5f2c4c35523d6662351960ba87413e0584a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c4e5f2c4c35523d6662351960ba87413e0584a9"}]},"branch":"refs/heads/master"},"4ef5b3f50a5f217eb18ea1107313ccdd2dbda7dd":{"kind":"REWORK","_number":4,"created":"2024-12-03 07:50:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/4"}}},"commit":{"parents":[{"commit":"acf7b145d4a903d9c5bb410e6437bc98772fb354","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/acf7b145d4a903d9c5bb410e6437bc98772fb354"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-03 07:50:23.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ef5b3f50a5f217eb18ea1107313ccdd2dbda7dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ef5b3f50a5f217eb18ea1107313ccdd2dbda7dd"}]},"branch":"refs/heads/master"},"7f3eac544cb3a78345155cfd2e6ce32d036641d4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-12-05 08:26:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/5"}}},"commit":{"parents":[{"commit":"39c5f2e766c4963e654697cc6407ac3f5c86561c","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39c5f2e766c4963e654697cc6407ac3f5c86561c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-12-05 08:26:54.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f3eac544cb3a78345155cfd2e6ce32d036641d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f3eac544cb3a78345155cfd2e6ce32d036641d4"}]},"branch":"refs/heads/master","description":"Rebase"},"2dda971bbd1908e54819ce26bc3fbc44186b6882":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-12-06 12:32:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/6"}}},"commit":{"parents":[{"commit":"91f94050cc0a15b5cc08df224b1cef15f9c5b38f","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91f94050cc0a15b5cc08df224b1cef15f9c5b38f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-12-06 12:32:48.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2dda971bbd1908e54819ce26bc3fbc44186b6882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2dda971bbd1908e54819ce26bc3fbc44186b6882"}]},"branch":"refs/heads/master","description":"Rebase"},"314982f930543e55d0c67d9b6e0ce5088ffc74a9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-12-09 07:20:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/7"}}},"commit":{"parents":[{"commit":"592984829c64c86c01fa8d4bfd9e907ae0a595bb","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/592984829c64c86c01fa8d4bfd9e907ae0a595bb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-12-09 07:20:55.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/314982f930543e55d0c67d9b6e0ce5088ffc74a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/314982f930543e55d0c67d9b6e0ce5088ffc74a9"}]},"branch":"refs/heads/master","description":"Rebase"},"46453b1effbd77a075676386f1c8529373497e46":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-12-11 08:30:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/8"}}},"commit":{"parents":[{"commit":"b4c7dfbaa8b9672fa0006a3e6c20feff9a3ee0cb","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4c7dfbaa8b9672fa0006a3e6c20feff9a3ee0cb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-12-11 08:30:18.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46453b1effbd77a075676386f1c8529373497e46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46453b1effbd77a075676386f1c8529373497e46"}]},"branch":"refs/heads/master","description":"Rebase"},"0c7f7e260a07bcfa9c477e9a78b984f68c2974df":{"kind":"NO_CHANGE","_number":9,"created":"2024-12-11 08:32:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/9"}}},"commit":{"parents":[{"commit":"b4c7dfbaa8b9672fa0006a3e6c20feff9a3ee0cb","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4c7dfbaa8b9672fa0006a3e6c20feff9a3ee0cb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-11 08:30:34.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c7f7e260a07bcfa9c477e9a78b984f68c2974df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c7f7e260a07bcfa9c477e9a78b984f68c2974df"}]},"branch":"refs/heads/master"},"dce01d7550642ec478eaeb19deded4c974134ea9":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-12-11 13:38:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/936813/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/936813/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/936813/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/936813/10"}}},"commit":{"parents":[{"commit":"865097c6896a87fa5b6d508058db9ff1e6c423d1","subject":"[OVN] Improve initial hash ring setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/865097c6896a87fa5b6d508058db9ff1e6c423d1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 09:23:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-12-11 13:38:22.000000000","tz":0},"subject":"Fast exit when initially creating tunnel allocations","message":"Fast exit when initially creating tunnel allocations\n\nIf the database tunnel allocations match the current configuration, no\nother operation is done. During the initiallization, if several workers\ntry to execute this method, only the first one will update the\nallocations. The next workers will check that the database is correctly\nupdated and will exit this method.\n\nCloses-Bug: #2089940\nRelated-Bug: #2083570\nChange-Id: I208ba38bff9191cabcc1325fec516d0b0179c97c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dce01d7550642ec478eaeb19deded4c974134ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dce01d7550642ec478eaeb19deded4c974134ea9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
