)]}'
{"id":"openstack%2Fneutron~936849","triplet_id":"openstack%2Fneutron~stable%2F2023.2~Id7d0c8e7ba37993b1084519d05e7e2eac095b81b","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2088986","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-12-04 11:02:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7d0c8e7ba37993b1084519d05e7e2eac095b81b","subject":"Fix the tagging policy engine","status":"MERGED","created":"2024-12-02 15:34:43.000000000","updated":"2024-12-04 11:03:47.000000000","submitted":"2024-12-04 11:02:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936849-bug/2088986","meta_rev_id":"159c86ede872ac822627733574a77854acb9ce85","_number":936849,"virtual_id_number":936849,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-04 11:02:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2024-12-04 08:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-12-04 02:30:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2024-12-04 08:59:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 16:40:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-02 18:05:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-04 02:30:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-12-04 08:59:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"88d29bee03e6aad484f24d38ac55cd78bd988b31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-02 15:34:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0733d226e54094f7ca8c5912f2ea2b6196b2a96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 16:40:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/544f2cd2ea3e4bcaa629006c1ee0be08\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af1aa9b476534f0696c02800a273b30c : SUCCESS in 30m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9a0dabe469d49b4b91c350229a8878f : SUCCESS in 45m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3cf367a355d64c18bcacf81600417c64 : SUCCESS in 26m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/29bb681bcdea4587a3e7dbf23425a998 : SUCCESS in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d500a4b5a3d89fab287b5438f52f97abc10face7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 18:05:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/826f1f1261bc4169945fb464d2aeb32f\n\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/457ad8b6134140849938f88da259e55d : SUCCESS in 1h 18m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/cfcdc3d6d6af4bbdb8454eb14730a91b : SUCCESS in 1h 28m 19s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/961b2de834784b829fd94e207f9c7378 : SUCCESS in 2h 23m 42s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/fa1cfa506cad40e2946a904dd6e90aec : SUCCESS in 35m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be9d83a61dd457b863155b69ddf2f8e : SUCCESS in 39m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33da7011146d4c02a5bee17e204aaba5 : SUCCESS in 10m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4640eab4d4947cd881545e441cc961f : SUCCESS in 28m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91776cb9592641dfad7999e8ccb748a5 : SUCCESS in 24m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/712d58613e254bbea45716f2a0c4dc96 : SUCCESS in 27m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6c75bc2918b42f0b24ad7b9ab4acc34 : SUCCESS in 26m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b0e2678dae470e840d392c75e5ec4a : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e9678905e97447eb9cc90a4f457b474 : SUCCESS in 53m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ac6b3b3868a456da5c280b813fece47 : SUCCESS in 1h 51m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/171effa712894c4da3cf6da7133dd940 : SUCCESS in 1h 04m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e92bb0cf73d14e518e9b2c3403a0ba80 : SUCCESS in 1h 50m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ddccaccc2f054fdb9a186abba7e2f3e2 : SUCCESS in 1h 28m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8c6f626eea7649baa605a45a38e7c11c : SUCCESS in 1h 29m 40s","accounts_in_message":[],"_revision_number":1},{"id":"878de97e6b5aa2a1fc6687c45fbf4d319f085262","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-04 02:30:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a3d8f9014c44c3a1f5bffff6cb0336daa17d734","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-04 08:59:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"117419ce59b95e9593b86dc99e5a81abda4a8a8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 09:00:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd52c794bafb80a16c70c56ef70f5808afe2e6f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 11:02:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bafbdf6d75348779249a7d5048532da\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/0e19edc306d74515a36f6c1caad41dfd : SUCCESS in 1h 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/468d65fe048944ada0a011d96330b99f : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ce819741a8f4dcbbb8e35777627155f : SUCCESS in 21m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63607b2db524411483ac91d5892a225c : SUCCESS in 25m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd056ad0693b40718e9a58c391e7bfa1 : SUCCESS in 20m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0ea8a98dca4b87ab5bf98d505e9732 : SUCCESS in 11m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e82eba39bdfd4faa92335f9af18ce185 : SUCCESS in 49m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8fb1790da7ba4719ab61cc505d0749f7 : SUCCESS in 1h 13m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/75ab2e6dd0474c708e471798dd193ac8 : SUCCESS in 45m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6f411b9d03a846b9a85bac5a97fae5fe : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e1e85feeb0135f73b26c2293795f2e48e2e2c6f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 11:02:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"159c86ede872ac822627733574a77854acb9ce85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 11:03:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bfa0a752898472a971a38a12f5a2338\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed532c289e5f428884563ee54531ff37 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e697e1a0eafb1daac504c506b3ad2ed37dda03e7","revisions":{"e697e1a0eafb1daac504c506b3ad2ed37dda03e7":{"kind":"REWORK","_number":1,"created":"2024-12-02 15:34:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/936849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/936849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/936849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/936849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/936849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/936849/1"}}},"commit":{"parents":[{"commit":"cbc01c0637df4200bff756013d694ae13c93e298","subject":"Add policy enforcer for QoS policy \"tags\" service plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbc01c0637df4200bff756013d694ae13c93e298"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-21 11:55:37.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-02 15:34:19.000000000","tz":0},"subject":"Fix the tagging policy engine","message":"Fix the tagging policy engine\n\nThe service tagging policy engine should consider the parent resource or\nthe upper parent resource project ID when checking the policies against\nthe caller project ID.\n\nBefore this patch, as introduced in [1], the target was incorrectly\npopulated with the caller project ID instead of using the resource ID.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/896509/13/neutron/extensions/tagging.py\n\nOSSA-2024-005 CVE-2024-53916\n\nConflitcs:\n    neutron/extensions/tagging.py\n\nCloses-Bug: #2088986\nChange-Id: Id7d0c8e7ba37993b1084519d05e7e2eac095b81b\n(cherry picked from commit fb75d3c4f185bb082f69c121090382d9eb803b94)\n(cherry picked from commit 93e86fa84175b525f5b1dc5df1651a44d60219ba)\n(cherry picked from commit 3510fc4b723a86af0a3fad6933c26cb4f1cae518)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e697e1a0eafb1daac504c506b3ad2ed37dda03e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e697e1a0eafb1daac504c506b3ad2ed37dda03e7"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
