)]}'
{"id":"openstack%2Fneutron~937519","triplet_id":"openstack%2Fneutron~master~I210235901a08e7d4043ff86f29f83a9268724b0f","project":"openstack/neutron","branch":"master","topic":"neutron_api_wsgi_default","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-12-12 07:08:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I210235901a08e7d4043ff86f29f83a9268724b0f","subject":"Test devstack patch","status":"ABANDONED","created":"2024-12-11 07:56:11.000000000","updated":"2024-12-12 07:08:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed5b1a778938dc00d1dc86dc0ac76252b33d0af6","_number":937519,"virtual_id_number":937519,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-12-11 08:27:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-11 10:21:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-11 08:27:47.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-11 08:49:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-11 10:21:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"64c2dc74c2692b93a6ec7656fb8444488c0a0204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-11 07:56:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b39a2a71be5215b4d12de21ccbd6c2e9b35f53f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-11 08:27:47.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/19/937519/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"932e93ef377966be7e306a2a67c5ac9c713d200f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:49:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcbe5018fc9543a3aafa5a9c3422e3bd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5ca93f9d6e74995899fb12aba948e61 : SUCCESS in 33m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eaaa154837c44e1e90ff60566f23e842 : SUCCESS in 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9aa1e473f6ddf3c65daaeb77ee11a1726a2b49b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 10:21:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/391da8eeaddc41fd91c5a349b6e95c38\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/28a13f6120c34f3e8ed325ae20e780c6 : SUCCESS in 1h 46m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8c6b089b78a43bc85311979a79dde2a : SUCCESS in 1h 56m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3688d3e8fbb646f4a0c07f7f18b5ec83 : SUCCESS in 2h 03m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e23d216a8c334399ac25c1600fafb3ce : SUCCESS in 38m 15s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a90990f105bd4b3c98c37913f12ebaed : SUCCESS in 1h 25m 06s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34b39c5311344052b74f539c5fd04741 : SUCCESS in 2h 24m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/291564d8d07b43779634d9d33256a1cd : SUCCESS in 50m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c70f56545cc4d20b6fa69d93092ffc0 : SUCCESS in 11m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/203073c713d843bea961719ea293b1fb : SUCCESS in 22m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c45f94740ac42c5b0dfb313c9d05edc : SUCCESS in 20m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25be58e5bc1e4461b181ff05c442aca9 : SUCCESS in 11m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cc4f00150b5141a596b9e6f0d791c684 : SUCCESS in 32m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e80b3189fcbd41c09a27e996e2dea8c3 : SUCCESS in 58m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/836f54d2c6cb4e288b94b60ac17c4c76 : SUCCESS in 55m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a943c3852f9a4485ba85672596ca7330 : SUCCESS in 1h 55m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddb21f2c58c344f2bcb1a6ea1d3c7a64 : SUCCESS in 1h 02m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/00c54d64eb044a1f951721be02f868ad : SUCCESS in 1h 10m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5ce2f7d7b63d46cf96edceac42cdaf72 : SUCCESS in 1h 47m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c46482de769b4f67ae2a1f47ed4aa3c5 : SUCCESS in 1h 26m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/72c462bab50d4eacb0372ee94e63dd97 : SUCCESS in 1h 08m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bba07d28fe56454aba620ef6e4474049 : SUCCESS in 58m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d25b9ff9f34c478ca240e37e1bc1e90e : FAILURE in 1h 15m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e11a75697dc34706bc11b218eafa7ff1 : SUCCESS in 1h 22m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ed5b1a778938dc00d1dc86dc0ac76252b33d0af6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-12 07:08:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3bdb1e432a7c231119218026ef01229320da74b","revisions":{"f3bdb1e432a7c231119218026ef01229320da74b":{"kind":"REWORK","_number":1,"created":"2024-12-11 07:56:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/19/937519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/937519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/937519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/937519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/937519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/937519/1"}}},"commit":{"parents":[{"commit":"83de306105f9329e24c97c4af6c3886de20e7d70","subject":"Merge \"Add a detailed debug message in case of segment allocation fail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83de306105f9329e24c97c4af6c3886de20e7d70"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-11 07:55:37.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-11 07:55:37.000000000","tz":0},"subject":"Test devstack patch","message":"Test devstack patch\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/937518\nChange-Id: I210235901a08e7d4043ff86f29f83a9268724b0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3bdb1e432a7c231119218026ef01229320da74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3bdb1e432a7c231119218026ef01229320da74b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
