)]}'
{"id":"openstack%2Fneutron~937604","triplet_id":"openstack%2Fneutron~unmaintained%2Fzed~I5ee3d89bfa2210630a3a25bcab68c8f54a5f0efb","project":"openstack/neutron","branch":"unmaintained/zed","topic":"dnm_zed","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-01-02 09:39:51.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I5ee3d89bfa2210630a3a25bcab68c8f54a5f0efb","subject":"DNM: test zed","status":"ABANDONED","created":"2024-12-12 09:35:43.000000000","updated":"2025-01-21 07:58:44.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5092d0bf01ed43765192b50b3c0d36c951061f5d","_number":937604,"virtual_id_number":937604,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-02 11:17:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 10:05:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-12 11:29:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8906af92670849a49d19f00f8f12c7e3459eeea0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-12 09:35:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f62328a44aa1dac8112160dd3b0d640e9f4010a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 10:05:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2567a62ba90c4cef934c0ef9991e21bf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2832262840b142efaab24c30941b6fd3 : SUCCESS in 28m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adc6ca6ed84e4eb08365589c0d9a8de2 : SUCCESS in 23m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0414dbf53acc303067cc5d2b0e2a409a52b9e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 11:29:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a0782d662b84026b90aff3d8b5f94bd\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/f6e875d8d81c424d89294be9ac4db625 : FAILURE in 16m 23s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/f2aedfc3d2d9488dbdd2b2e1244cfda0 : SUCCESS in 1h 22m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/bd639acba4d44098b08ec75a60d54720 : SUCCESS in 1h 31m 54s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/459771a733e9467085ef8869a654926b : SUCCESS in 1h 41m 42s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/f3b02cadf19342498afada68229f877e : SUCCESS in 29m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b28da8abcb704006b1b90e2b3aca1c63 : SUCCESS in 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a1b656a7ed54bc5b40a829a09d0e81d : SUCCESS in 13m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6b698e2f51d4f2a8696031225f492ed : SUCCESS in 23m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3cd7ad5762d45c194bf8a8eba4925ee : SUCCESS in 19m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcce75be4c534818aa0362584e983585 : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb56e7c87b843b2b5240fffc5ba5b9b : SUCCESS in 8m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54fc87d3fc7949da981da07b1f252a40 : SUCCESS in 1h 01m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1675b2d5a389403e958a9597404545f3 : SUCCESS in 1h 21m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e1c3f31bbf5e4d98865a8e60cfacf46d : FAILURE in 17m 58s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bfc0e6177db946f98d9df027031b903f : SUCCESS in 1h 17m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/08cd65f2b4c143958446cea60e408d55 : SUCCESS in 1h 08m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abf4a48acf214899a8c6f864043fd8ee : SUCCESS in 36m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b3bcd51ad73a41be8026d5da7ee3dd21 : FAILURE in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b84f2004c8fe6d501e7f08ef7c39c208baf94f5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-02 09:39:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f08fb7ae3895ebac5375f8e7927b58e300b0ca2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 10:07:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c522148446c641069ddb0de70ee2e54b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a5f2f0a2398454f811e1de8e1985a6c : SUCCESS in 26m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85c951ae8602446183d03d9fbd8c36bb : FAILURE in 21m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b4ec3f17ddb11eef45de0770f8f394d0d61fc9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 11:17:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f4d79c87963453194abee22432dfeee\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/2fa19f73746745399ec7fbcb4f496999 : SUCCESS in 1h 34m 55s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/eba611fdc1084e048f08b1f28b844d04 : SUCCESS in 54m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/8298a87a9efc4e93926f7f5aab9baee3 : SUCCESS in 1h 06m 59s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/e6dd5e308aa54c358a47a6f6bd045899 : SUCCESS in 1h 29m 53s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/52d94f43f55f41e78c5a26969c7b4043 : SUCCESS in 21m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec5faede7e624d45bb751272ad3c655a : SUCCESS in 34m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458071e5e4264df1a6cfc48c1b6fd910 : SUCCESS in 13m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c56261bbbd9444e84c24b4c0b810ea1 : SUCCESS in 23m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/438e9e51a6be4bb693fb08b93f5d1919 : SUCCESS in 23m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81d9b82ff309468289fdf993b29d5e1b : SUCCESS in 25m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0c2084533c4f94a81d5163bc0facc9 : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b71d73219544f85bcc6139b3630d5bd : SUCCESS in 54m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f3bdf9c7ea04f3cbcbcae526894ba61 : SUCCESS in 1h 21m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/44d55ccc26314557bba846d0ff29af90 : FAILURE in 10m 27s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f5696da69f734376ac937993fd3ffc24 : SUCCESS in 1h 17m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/357fbd81676a4fc8a04ba9ea67d1700d : SUCCESS in 1h 08m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b965ee83cd5432289e3cfc7d22657ae : SUCCESS in 1h 01m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bddf68d8247645a7bb144d410b58b670 : FAILURE in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5092d0bf01ed43765192b50b3c0d36c951061f5d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-21 07:58:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9ec12a887d9ce7adef3259f179600c7aa6fb589","revisions":{"c9ec12a887d9ce7adef3259f179600c7aa6fb589":{"kind":"REWORK","_number":1,"created":"2024-12-12 09:35:43.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/04/937604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/937604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/937604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/937604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/937604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/937604/1"}}},"commit":{"parents":[{"commit":"266475dc9b52f6e8f14384ba7c10835be09e06ce","subject":"[CI] Functional: Increase Ulimit to 4096","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/266475dc9b52f6e8f14384ba7c10835be09e06ce"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-12-12 09:35:16.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-12-12 09:35:16.000000000","tz":60},"subject":"DNM: test zed","message":"DNM: test zed\n\nChange-Id: I5ee3d89bfa2210630a3a25bcab68c8f54a5f0efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9ec12a887d9ce7adef3259f179600c7aa6fb589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9ec12a887d9ce7adef3259f179600c7aa6fb589"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
