)]}'
{"id":"openstack%2Fneutron~937765","triplet_id":"openstack%2Fneutron~master~I04ac8336db70cf93da43ed17da11c594b441f5eb","project":"openstack/neutron","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-30 21:52:03.000000000","reason":"Change was submitted"},"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2025-06-30 21:52:03.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-06-30 21:52:03.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-06-30 21:52:03.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-06-18 17:55:28.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I04ac8336db70cf93da43ed17da11c594b441f5eb","subject":"[eventlet-removal] ovs: remove the usage of eventlet in the OVS agent","status":"MERGED","created":"2024-12-16 10:44:03.000000000","updated":"2025-07-10 15:44:23.000000000","submitted":"2025-06-30 21:52:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937765-eventlet-removal","meta_rev_id":"9cbe3df6fb105853d1e2c87e789e4bc2eba2ca59","_number":937765,"virtual_id_number":937765,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-06-27 15:23:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 21:52:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-06-27 12:46:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-06-30 19:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-06-30 19:58:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-06-27 12:46:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-16 11:06:50.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-16 11:30:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-16 13:15:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-16 15:18:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-12-17 15:12:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-01-06 15:08:14.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-01-06 18:03:17.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-09 09:34:12.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-03-04 09:46:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-06-30 19:58:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-10 15:40:07.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"dded78cbb36b34e4771b579c1fa5ef51df6dbb71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-16 10:44:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41456f3840aa5a1bc100ce2e1a13f0a283c1ebc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-16 10:59:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6cec7e2c1b8208b0dfa7e4384a19df30b851e78e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-16 11:06:50.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"adff8faebcbbe0c526c18623b732a736d3e50b5e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-16 11:30:32.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"bf0895d25a1867b0166c7e46af561fbcef78b0f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 11:30:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f558af1297774ed6949a7d0ab6e1610b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49cda6adb385444992c55eeb124cdc46 : SUCCESS in 26m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/08ba83e06fd44c8ea5b8e525b530ab07 : SUCCESS in 21m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db258995e414d5224251bbd025642069d5ea3e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 13:15:00.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78765c02d9a142a89088a3e485414c60\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/01719e724e564340bfdf7d12921d3f93 : SUCCESS in 1m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9847dd4df38549309f3fe1c65346aaa6 : FAILURE in 27m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66315e3606db4b388ba49dc43b3c55b6 : FAILURE in 1h 56m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d551aa7ab6544d79bdb55f9bdac9780 : SUCCESS in 2h 14m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ebbcef6ec62d47048a31d3bb0ee8596c : SUCCESS in 41m 13s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5517109a3d340fc89a2b714aa3c0025 : FAILURE in 55m 18s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/138d89b14dce4ae5b03f04a0ca293d33 : SUCCESS in 1h 56m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65d2b33ed6544bd29b5efa5926db4854 : SUCCESS in 54m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab15340a948b4518aef980155b854250 : FAILURE in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/545fdc56918b4226b80aeeaa31d36ac7 : SUCCESS in 20m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cea389305cf423bb0d06fed553cefbd : SUCCESS in 19m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff4b845829b24dbfa2d927159fb1daa9 : SUCCESS in 10m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/db0e49827f174a6ba19e4524350edd9b : SUCCESS in 1h 04m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3a43cced76b449cd8168612162ec500d : FAILURE in 1h 05m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/343cc02beb9a42f69630471fac0ab6ef : FAILURE in 58m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fe6ef8b7b5da443ba684618dcdc18165 : SUCCESS in 1h 55m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b660de8cecc47c3bb163f4f7f716384 : SUCCESS in 1h 02m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9c79a775f434cf7b9739732cf03216c : SUCCESS in 1h 17m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/30a8cca2804b4155acc62b3d17bff4ca : FAILURE in 32m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1af723a0d20044e7a1cc007545108c32 : FAILURE in 59m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cf4b71e01d944273b19cc3f57ef5d284 : SUCCESS in 1h 09m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fe09fb255e6840b68157b7fe653d77fd : FAILURE in 27m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bb09630faffc4814b7b1e0bce65de025 : SUCCESS in 1h 10m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/33026d7f92dd4dd2afb4e9fed0fbf7b0 : SUCCESS in 1h 31m 37s","accounts_in_message":[],"_revision_number":2},{"id":"327bfc14fd38ee20e6fb704768e5dd64c415000d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-16 16:23:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"778a2183f66748fd900a1a82767247431affec48","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-16 16:46:58.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7febf134ef7c91f728800c9649c3c9e1dd977802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 18:57:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9be245521a874fc78d6943e70ec7968e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cc873b1b1f984b5baf60748644ccfe9f : SUCCESS in 2m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7a6a8bf331c441f0892507007b7083ed : FAILURE in 25m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8501338cc364feca9784e9de4f2e444 : FAILURE in 2h 13m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b951549ae13e4323ad665315caaa0fc3 : SUCCESS in 1h 26m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da98a80ba8c54fe5915dbb25bcadf298 : SUCCESS in 40m 09s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/050ce6a4cf264cadb2dc011ad31616f9 : FAILURE in 20m 10s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaa4708ee94a458789cc301d9add7475 : SUCCESS in 2h 32m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16bf601dfa134f0ebcb4b50d4e11d2fa : SUCCESS in 54m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fd985eb5861445aa336527eddaaf24b : SUCCESS in 11m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/889b1a6488c5468bb23f57126a786ed7 : SUCCESS in 21m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e46f68a809a84861ae11f7b6de803cec : SUCCESS in 20m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06cc31d39e844ae3b0c8f8ee56b390cf : SUCCESS in 9m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c6544f0ccff84a57a6f59ef46643fe08 : SUCCESS in 1h 01m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/849d490dfbd8458d861fac4e57cfe6c0 : SUCCESS in 1h 09m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3ebf84ae5a3d4bd5b1cfc6b9ef0903d3 : FAILURE in 59m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b836f9e7647a410ebf0aa4152f739e30 : SUCCESS in 1h 56m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/deeb852d3ef74e929163925454baa869 : SUCCESS in 1h 02m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/14ca9451b16c472bbcac7cd0e8debaf3 : SUCCESS in 1h 17m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e1f09fa609e4431a3cf2abc84171a4d : FAILURE in 30m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5b5636b276a4d61b8427c0eee9d12cf : FAILURE in 29m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6c46bef74ea1453787ef57917a422100 : SUCCESS in 1h 10m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/86ac1dd6a78648b4941c90c56f0fd0bd : FAILURE in 27m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4a07979f733040dd8b11067aa5dd8be4 : SUCCESS in 1h 11m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d2b214918f584691907346974703e4ea : SUCCESS in 1h 15m 55s","accounts_in_message":[],"_revision_number":3},{"id":"f58ca10286e7d628a9ee17b284069fd3999f6c9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 19:05:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/056af341c1f243af8a5144135e240220\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2f5b1901b124fc8aae9d2e225cbb1bd : SUCCESS in 28m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f352afee09f24c0f9f58ba811801d03e : SUCCESS in 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ea9cbbb3a18659cbd671c1bdecde37cafe09cf2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-17 08:15:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a515ee23a2a2ad48009694025de2a0d63c648654","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-17 08:39:47.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"036127e51356bd769fda576aab57eabbbb491c6d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-17 15:12:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"248c92afa01e2475f667d812840218c331a972f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-17 19:58:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"43f6ee3e8ee98f37ee22d7c7a43222d261ff0c52","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-17 21:18:50.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"5581b49d97de57e14fc16d38bd09a71b9b3b2cdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 22:04:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c8ec8e3eb6c4809a3f5bcc297b884d6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/84575e170f634433a293da5e2ed757c6 : SUCCESS in 2m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c56ef32528aa48759a69388e7bb9c1fe : FAILURE in 49m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a6c560538f88467097ff9548c09b00ea : FAILURE in 2h 05m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/78c31d9a381a4d918fbf77202b60c225 : SUCCESS in 1h 59m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/349bc927f64a42ce89baf18f432b20a1 : SUCCESS in 42m 19s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01635d52fb214a87b971f7550e7ce08d : FAILURE in 37m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d03142080d64b47a12aad98eb4e994b : SUCCESS in 2h 02m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2f2dcdc37a04d868ecab30c26c1a045 : SUCCESS in 50m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e0992bc714044b3a33518856eaa1f7d : FAILURE in 10m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1aba84a64974de4a2c1a66b3970cd03 : SUCCESS in 19m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f048e2b47d0842b0b208140a506241bd : SUCCESS in 18m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/935b5d37e8c14cdfa9c21b9e2bd6790a : SUCCESS in 10m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/49e5a1aef5dd4632bd946d5505a82bc2 : FAILURE in 43m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/721ff32d9817457aa1d2a8d95938b95b : SUCCESS in 47m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3d74409310c54f32b3a12dc7de9125ea : FAILURE in 1h 01m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c1c6dae8a17a4d55a34672593c3794d6 : SUCCESS in 1h 50m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc796881d0bf4f34a494d997f1da94ec : FAILURE in 34m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/670073f3b3214c4b9e8122923eb04e57 : SUCCESS in 1h 16m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f0581a8af16147cab83040d95838899e : FAILURE in 50m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c5f87b56fa27482987d9120df1152f7b : FAILURE in 16m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9af27063b30243be919fccbec57d58ec : SUCCESS in 1h 07m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4e9e132f9376430ab195f8d391ff87b4 : FAILURE in 20m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b36cab27859e4c0bb276be8487c44741 : SUCCESS in 48m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/43eb46264c7b4fee86865da4dfe5a947 : SUCCESS in 1h 17m 43s\n\nWarning:\n  Comments left for invalid file neutron/common/eventlet_utils.py","accounts_in_message":[],"_revision_number":5},{"id":"88e4f5e943cabfcea89e2bce02f516c00a08c6fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 22:21:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf6d927e169f49e1b9039652fa320481\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db31b570a70742b6b367be327a301cec : SUCCESS in 31m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49cd7c937999490496de1d2d2e525f66 : SUCCESS in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"580a25c8942bb1d02e288f98efd64514f561cf49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-18 09:40:08.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"552d2e99da2c01e05325396e86961b4081d3f995","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-18 09:41:00.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3a1b33489807bd289a4767cd6f218a170dec3aaa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-18 10:02:51.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"1a770357f2663e20338de9a8c88b3f34b276f864","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 10:59:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04f20745c5a44bfeaa56209baea0881a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a76daed7b8b448c5b697324092a7942b : SUCCESS in 34m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/938555cd1a394213a96505fc849590f5 : SUCCESS in 39m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2ae41df227ca50765a2cde1f0e9604371a2aee9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 12:06:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f6c2e40e1644336b0ddbf67f4d56853\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9e4feadc0c9a4d52b7724d74f95d9986 : SUCCESS in 2m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d910a4bbcf8e452ebcd9c4e5e64d961e : FAILURE in 26m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ffb86940d48647308af1a813e9240596 : SUCCESS in 2h 05m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6ae8e57cf184eacad33dd5ee3bc8628 : SUCCESS in 2h 06m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/047c7884dd694b56ab88335bc1156f19 : SUCCESS in 35m 29s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/279ef9159fa3401d80dc77effcfba726 : FAILURE in 30m 01s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ee5eb5704694b69af02f6d373d856c1 : SUCCESS in 2h 03m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb40d48f366443908318232833c2fb6a : SUCCESS in 33m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba0f9d4229464ca2b68ca6769282bee7 : FAILURE in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55f1aafae9c949769805602502d6ab8e : SUCCESS in 17m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9ec7adbfecd4533959d56a94fa79af0 : SUCCESS in 19m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1890a4225a824e17a692bd084e3f32af : SUCCESS in 8m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/615cb63617904882aa9ee6c384100885 : SUCCESS in 49m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9e3e2a2039e44238beb70f1826ca1568 : SUCCESS in 58m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e73664677c9d41b3995c8c1866e7d445 : SUCCESS in 43m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e783e1fc911a45ffa8992f43fdab4c8b : SUCCESS in 2h 19m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc29d9f7398144ecb65d633b47651317 : SUCCESS in 40m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b82933835204dedb2d602c0900916d3 : SUCCESS in 1h 01m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c05b3c69c85b4c768a7ad5d16ba0a157 : FAILURE in 22m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e9d606242171441195e10505a7502063 : FAILURE in 27m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dc1ebffd80a8427f957bb2e8ff7802d3 : SUCCESS in 56m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1e8b498b368440c7a98cd172e4347a85 : FAILURE in 19m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dcc0c7b3f8f04d7c9c25c39ebcde9e92 : SUCCESS in 57m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fc7569dc083b4d25aa7848b93b1713b7 : SUCCESS in 1h 20m 48s\n\nWarning:\n  Comments left for invalid file neutron/common/eventlet_utils.py","accounts_in_message":[],"_revision_number":6},{"id":"44e24616eca282865532323053e1c9fee866b084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-18 15:02:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f54edd2c618e496086614646f7ac84d891f726ec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-18 17:04:54.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"bc6382169a1c843b260ee4381c5bfaf9dc20fe6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 17:18:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39d5b0c37a9546c69c6be2066a23f426\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/47ecaf4015fe4c08b3f02262c218f01a : SUCCESS in 2m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0925b41804db4eb9a2244fa08dff36cd : FAILURE in 29m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/85104a2392cc421b8cbaf0d3823de752 : SUCCESS in 1h 55m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0eb32851de8d48abb933a9a2040da9bf : SUCCESS in 1h 51m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ccc5683d46340d788340138ca642482 : SUCCESS in 37m 34s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e066854248c4f3ca78a6ce449f5ff96 : FAILURE in 25m 57s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26871c6bacac42929e68ff17d9063b5b : SUCCESS in 1h 42m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea94a4b38e0a44a88b69bb55b0d082ae : SUCCESS in 49m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28f7793b32940acbb4e1e3e829146f2 : SUCCESS in 10m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5e99dd4772846c1aba0cd5c38c5e928 : SUCCESS in 26m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0319a86d98af415fa9e3804d53e7da9f : SUCCESS in 21m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5386309caef44d5945eae2dd807d427 : SUCCESS in 11m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7abc714326c746668440041f22978d7e : SUCCESS in 54m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2bd0cf1d4d384f738e88ea0e42407dc9 : SUCCESS in 1h 06m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6b5f07f6b8974343ade9974cfa03b931 : SUCCESS in 46m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/da93846bc75a4516846ebaeadd13d609 : SUCCESS in 2h 01m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3e32b51c33a4ba19de59a4b5690cb32 : SUCCESS in 1h 07m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/20f14fd63c594292b0117e6226c953da : SUCCESS in 1h 08m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e555b8e0e78146c0bfc47759b29e0140 : FAILURE in 18m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4c424542107f48bc85a04acdeb7fe8c4 : FAILURE in 33m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/be139e78d37e4b37973adb91eb0ad3b5 : SUCCESS in 1h 11m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ca939efef0d944258c50f4129f217723 : FAILURE in 16m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2ad74d6d93e84ca789b6937273e11e0c : SUCCESS in 1h 14m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ef91262b0cf9433aa685ef6cd8d4c81d : SUCCESS in 1h 08m 50s","accounts_in_message":[],"_revision_number":7},{"id":"8b4c0d98c7544ce9e3875b8f613181b572edb3c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 19:21:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5e9e88e61b40e3a8aaf417b68e7e6f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e80872218da344bdb02370d0d32ae8a8 : SUCCESS in 24m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e111646fbb8f428786f7937afb066793 : SUCCESS in 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d5331d8e6106e466181aa62120c3b37ad39ead66","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-19 07:48:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"401306ac62acf03a0eb1b07833be5a5649746744","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-19 09:10:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"75f51d60ed92c58d92a4377c89f767f35469f5e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-23 10:14:17.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2bf33ce70f9edbcf670591224b0e8bfc300601e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 10:49:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99c289e527014aff9052c9dc35d0b6a4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc2e4a6a87f04dcc8a2de22fde6183c8 : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc039b3adcf546b99748ad11632cf9d2 : SUCCESS in 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d3f2639dfa8464cde179785c1a8a76b610a68fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 12:02:10.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c69adcfa8bef43ecb60e6fd9fcfdaa56\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/91906d51837e4f4eb8734b07ea0302ef : SUCCESS in 2m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/052507a8fa214eca9e345a6ccbaf22b0 : FAILURE in 28m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c6825963c5d4a03b7d7bef7736b636b : SUCCESS in 1h 46m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f5285ebe289f4677a9a8b4f8e07cab32 : SUCCESS in 1h 45m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8dfdb90f9a3146d88784f94e397582fb : SUCCESS in 37m 27s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2414fd82b23642bab5d4d567aa830502 : FAILURE in 25m 20s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddaa569aed9e459d9ff2f22e3193323e : SUCCESS in 1h 40m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00dc503ac7044cebb9b75a0c135a16fc : SUCCESS in 56m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeaaea1a43c246e4af3d66ca15cf155b : SUCCESS in 11m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f652d356ef35488ba6321e1c5f444f82 : SUCCESS in 18m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df49aec38490418a8d3504c57e7a14cd : SUCCESS in 17m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9bee3525e9490b9041215f4c860aee : SUCCESS in 10m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2e8cd3c5ae484b149f67adcaf31fce8f : SUCCESS in 59m 18s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b5e1b9dbed044a3fa21c0f0f67acf52a : SUCCESS in 41m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5fd3bd810b1043929213dc595fe1b0bf : FAILURE in 41m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e611f1e0e09a44a59bb33f30a3071c01 : SUCCESS in 1h 38m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/78ff98caaf70431fa38656805b0876eb : SUCCESS in 59m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4945a2133c5541afbf2b6fe618259a2d : SUCCESS in 59m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/84ca3c94740d4347b937b5657e8a8bbc : FAILURE in 30m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb1a0915abca481fa74a4959c6ce222f : FAILURE in 30m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a6d6cdb8e78a43138980cbb0355aa83f : SUCCESS in 1h 01m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3188e6c3d9224594806621a78289e65a : FAILURE in 36m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a12dfd3bdcf948bf9a8b58e11084fbff : SUCCESS in 55m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dc3fc50314994688aadfcafb1be15a9b : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":8},{"id":"5a30425563da7cf1a8f0e1b511b5ad167ba4fbaa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-23 12:42:07.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"b79bedb16a25b4136bdfa2c30aa397996849840e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-31 09:09:05.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"39a901f7d815262051a4893c145180df8ad76c73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 09:41:21.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49480c099f4a4ea3911f05dd089c05c2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb33ca1ce2c343318de276c86aadee0b : SUCCESS in 30m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5538e4d18dec4a719d56690684a0545f : SUCCESS in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b2c8b3ca682c4299a0c130df60122a0365d7b7fe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-31 10:21:47.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"c644590f59a72a47251d375f12da46ed54ea6675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 11:18:56.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb81b81d23304d66a9963222b68bfd1a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c101d321036b47f3bd2e072066575c24 : SUCCESS in 1m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0334588a2c8148d497d48c84b7604615 : FAILURE in 35m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd37ebe882d842c8843210adb1512c7c : FAILURE in 1h 57m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/730d2d3c9f62456cbaa436071bcada46 : SUCCESS in 2h 08m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a50591d439d341af81eeb282a6b02c81 : SUCCESS in 48m 25s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1a1310dadb5470d96723a6203d18da0 : FAILURE in 22m 46s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a605a3820484836b862e3e5c2b9db28 : SUCCESS in 1h 39m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5c57c60e180418b8ffe237c862abddc : SUCCESS in 46m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e08f6fefb884f65ae87165a64fcb263 : SUCCESS in 11m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dfeb648843646668e8248a08dc3f7ca : SUCCESS in 22m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba23069671f4406294e026bea72a2b4a : SUCCESS in 15m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96512cfa118946d4906df6857e7c3204 : SUCCESS in 10m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4fb63aff01ec4590a3601c6fcfa9bc19 : SUCCESS in 1h 01m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4b1c27b8a7c147a19fb8102aadd22b77 : SUCCESS in 57m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/03e7e1c8d9384456b1ff510ed5889e42 : SUCCESS in 1h 00m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2cb29dbaf1924a7382b472b68923dba7 : SUCCESS in 1h 51m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d989b7271c8c45668d5ca0f04c02779f : SUCCESS in 51m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7dd99fb6bae48cfbd505822530bc047 : SUCCESS in 1h 06m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ea76b6d0ab540ca866d9b811c343c9a : FAILURE in 46m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3c65fc92578e406087d9feae17a3dd03 : FAILURE in 20m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/44016b92830c44d592fda174980f3527 : SUCCESS in 55m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e6652c7e7f8c4af5a31ca8bc09a98c93 : FAILURE in 15m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/804f4f189c5f44c99d20eef3ffccd261 : SUCCESS in 53m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/95cccb98b9da4c558d333284960c7ae5 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":9},{"id":"f2af8e4a6d2a62c04a0d2f289b232d54e7027add","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-12-31 13:59:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"78012668ea99de9ba44f09b798c9120ba19e85e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 14:19:10.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/975bebbfc7154ad4be7cf96169bf42a4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4cb2a8e444b46ed8ccb8e6bab608f84 : SUCCESS in 18m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d933deb0b25646fb92837d6e9488a03f : SUCCESS in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0dee69d13455af6c640141ff8888a5c6ae5dc8eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-31 14:51:36.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/65/937765/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"df1c3688e9db537c1a1db39833c79bffa34f0c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 16:03:02.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f29aff06a8bc4d60b4956ad43bfc48e1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f91b6aeb480f4fc5b3a612e57a60ca95 : SUCCESS in 2m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/85cec6d0b4794056a5caa17541dc6426 : FAILURE in 31m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/af06fbdf3ef44c589d7842bdf3045e07 : SUCCESS in 1h 46m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/df2c118f246949aba872190f7de32676 : SUCCESS in 1h 53m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/addad02b1b4644c79aea6a9fd25b63c8 : SUCCESS in 37m 16s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7968dec08bc457193afb030fd11d2f6 : FAILURE in 24m 41s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11ff1b29a25b43cdb601acff9fa90ba8 : SUCCESS in 2h 02m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fe45393577f42c8a4199b9e00039f49 : SUCCESS in 33m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba274d2f891347679845fd9dd32e56fe : SUCCESS in 11m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a2966b00d8049ffb41ae94c1b639846 : SUCCESS in 23m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72ab61b89b4b49bea01c3424eacfbc90 : SUCCESS in 19m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2787dbfa6984772bce21e6d2a7e87db : SUCCESS in 11m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c277aef7015f462b98107e98f45d873b : SUCCESS in 32m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ed4a515c722e4769a5116bbf13705c82 : SUCCESS in 1h 04m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/319d7ddb58fa473ab76c6cc41603f805 : SUCCESS in 52m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4f0281e9a836485488929a3956a3c9c0 : SUCCESS in 1h 56m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc026dd57d7e4c6394124809abf4b57e : SUCCESS in 1h 01m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7a5e69d8a24493195ee42d7f8e36809 : SUCCESS in 46m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/46ef13ca06de4ba3a55cc85f1ad16ccd : FAILURE in 42m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9897fc5b694c4f978798a9e0c347d036 : FAILURE in 28m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a999861f2b934c90920b7a98757f48af : SUCCESS in 58m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/893ca9a1b79142a4bf7604bb7001bc65 : FAILURE in 1h 02m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eaa74b2266584cacbb6d0ba6ca45a92c : SUCCESS in 1h 20m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/16e440c7a2cd46dea59e806d385153e9 : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":9},{"id":"7b4346abd0eff090256f84e3b7c4f3f0cf0a0001","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-01 15:29:38.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"94ab021d70b7a6865d991be65ee66bbd5eb0c2af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-01 15:52:04.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"c7f256807228b15aa25bb699117501221c1f5894","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-01 16:02:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/016f13ea3d7f4f62ba14218aeec286de\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3d0cf13160f4998b628698621c91b0b : SUCCESS in 31m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45c8ce132d1948c48055a46b097a82cb : SUCCESS in 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ed341d700406cfb24d7afd7a948ebd94639acc7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-01 17:47:55.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51a378973293477fa91afef3e63fee7d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b73cecdd4f7a4f14a4c9003251f0f341 : SUCCESS in 2m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/448e7cbc527046d080fbfb3f6a2f3bac : FAILURE in 21m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/396bb1ee5c66403aa65130393eea1413 : FAILURE in 1h 51m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f4b8f3a395a94ea0bc52c8d207e079a7 : SUCCESS in 1h 30m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ed9cf8b1178f46c5842b7b5f21107943 : SUCCESS in 26m 03s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efd22c1c5f8f43ac98da11df97aa7787 : FAILURE in 24m 08s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8abf982422694d59bd33163f06423e41 : SUCCESS in 2h 17m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ed11b578d13403bb86fe97f9845d11f : SUCCESS in 49m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b93dc1a701247eeb3ff0117b6812233 : SUCCESS in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10106ab66833435797b73a3c2a61c8b3 : SUCCESS in 16m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d3e4488bac3449a9920db327d2c3a4e : SUCCESS in 15m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc19d60d2b554f11ab434633f95f6525 : SUCCESS in 10m 56s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/905450c8f5804715848fddd3e84365c8 : SUCCESS in 1h 00m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9fc165aeed114e9fabcabb3228c71599 : SUCCESS in 47m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/62146222db9e4b65ac892a05447a6f9c : SUCCESS in 46m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/42923b6fce3e4004954604bfbaefffd8 : SUCCESS in 1h 29m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9ce6c8ca10845c5ab0f324c4ba8d977 : SUCCESS in 44m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/39212c66730a4806a3186c2691cbcac2 : SUCCESS in 54m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/017508a6f04444c6b79f5bf51a19867c : FAILURE in 41m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99941b957f2a4e4ca7f6c83d262c9f14 : FAILURE in 39m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2ccb1fb59a654194ad256f324ff805a7 : SUCCESS in 41m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1bc5902a8caf48449f032c7cecf41ddb : FAILURE in 47m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ba16d1aab05b469b95dc619a3c69e150 : SUCCESS in 47m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/80e3537a087049c4a6199c653433c3c3 : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":9},{"id":"eb3189efba6d67abead28c5551158e5d55d2e968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-06 14:52:07.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"60214cf313041367b4b1c550c054fa6fe8bd1059","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-06 15:08:14.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7dfa5161cb76639440cc637bbb7e7a82a7142a76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 15:25:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2a879be096c4e718ae53d62751d0309\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83c9f44c748140d3bf56e3b0c00298bf : SUCCESS in 30m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3848785aaf36474a95f97dc828388f44 : SUCCESS in 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1f0658f13900e277d7807c98dd68071930352554","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-06 16:16:59.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"2ff9070a6790052a3d7aa8cbc504ee60bbab3141","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-06 16:29:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c359c2388f801c67b8115c8dca420ea00ae22b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 17:13:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4d8ba1399fd47cba2df5049212b7f60\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2bbb1d227be14f909b002fb3c03aea35 : SUCCESS in 2m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e4ceb381f10e49c0b7d8ed5d048c35d3 : FAILURE in 27m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a02e1ec9abdb430fbd95de1e599a1611 : SUCCESS in 1h 43m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d37c7e3f7ef34c08b7f8bce40fe283bc : SUCCESS in 2h 18m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3818d980e9a04f588d0fd73dcd0b1027 : SUCCESS in 33m 09s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2524943b4eec42549a0d555dd8e10591 : FAILURE in 23m 37s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13a3bbcec9fe4c6aade155c07f701fad : SUCCESS in 2h 09m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46c79a7c883c45e78b2891c0df874ec0 : SUCCESS in 45m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e952555ea34c4abbb95e44401d281851 : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b46d969197904caf9e590cc77789766d : SUCCESS in 23m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1c797645a0b4d38a557e4019fe6e479 : SUCCESS in 18m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66bf3928a8a74e79a32933764359c359 : SUCCESS in 11m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/43236b86466342a7966773632f1e3a67 : SUCCESS in 35m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/02b08a4c3ad241eaa3bb22ffa380cff8 : SUCCESS in 1h 19m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a2d089c687fd4aa9b8ea7c711c26b3d5 : SUCCESS in 1h 00m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/851220bff568439bb6c59f967985e745 : SUCCESS in 1h 41m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84ec41cdb9a847038c7738fbd2177d18 : SUCCESS in 54m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5b63c0d4e6b497ba22ef622399eb5bf : SUCCESS in 1h 08m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e1ed80043c254a9980961e682c9f7c66 : FAILURE in 18m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9a15639300484460af97220966cc022c : FAILURE in 19m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9a75573b8da0434eb18f4ffbc46df040 : SUCCESS in 57m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0b9e790e68c04353a39a21b17e14faf2 : FAILURE in 1h 00m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1dd69c9c74a3400fa1d017615c459a16 : SUCCESS in 1h 06m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bea37ec4b8414321bc7adf417000e068 : SUCCESS in 1h 09m 15s","accounts_in_message":[],"_revision_number":10},{"id":"2bd1a4dabeb42f4327d08c3e4662d1fd994f2ffa","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-06 17:27:31.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b5358ad9474e13810f89ef004fe433e8bfd7b8da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-06 17:32:08.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"6b1772547ac23846a0b508e22553e3fcc7594bf4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-06 18:03:17.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"c3bac2a6d81bf55b56c6fe793c874149f5fea979","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 18:10:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed10082e8128475889f0a27c534e656e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b480a489ab7e4657bb01329bcc0560a0 : SUCCESS in 36m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d5e3cc7f5f54375acce88cb741bef26 : SUCCESS in 24m 10s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"44f10dc75c858f34b55cd5f6a21a338ead9cb21b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 20:13:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8ea6612a9fe405da9c58a26463ab5ad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/48dfff295d514ce2a623eb8d601e26a4 : SUCCESS in 2m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/beba1b6efe144b529b55cbb2e4676f30 : FAILURE in 58m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e87f85a0c6d946a99487091aa302448b : FAILURE in 59m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e5f6a7953c2640d295473a44043187f0 : SUCCESS in 2h 08m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de985fc572864d52ac5fd7760631493a : SUCCESS in 37m 13s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e46402fd5c741c3ab8592fce5287d13 : FAILURE in 1h 18m 07s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6de404354daf49b0be5dd50ca11791fe : SUCCESS in 2h 39m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74ce17dca6be43f28a9f9e725ca398d4 : SUCCESS in 37m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b5ebb6a61f4b5c8806c2f0cdc20b1d : SUCCESS in 13m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1c4b12deed24fc09aefdf53898344b2 : SUCCESS in 17m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eeb01f8fc5d6479cb34b97ca022bc7a0 : SUCCESS in 22m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ddde8a4ba045b39f1bd425dd626c3f : SUCCESS in 10m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fcacc156ec4a4e0da8334e58b1d2d9b0 : FAILURE in 30m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4527d1b75bfe426faf6c89ec1b114f83 : SUCCESS in 1h 10m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d136c1beee42496196c227f10e7a7c0e : FAILURE in 45m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/dd46afe8841947cf8710afdf0223ca06 : SUCCESS in 2h 17m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84897887e0e3458d98368694b921cab8 : FAILURE in 32m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9270ebfbe17c4a5eac2d580c17c51cbf : SUCCESS in 1h 20m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8bba1f463120417ebd3984c46def58ca : FAILURE in 53m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cea708fb5bfd4076b4668483f3caed21 : FAILURE in 37m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/699ef66c3c7644e29ea9802b7929e5fa : SUCCESS in 1h 04m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/18a5b2810109426090b5f04536f91e0c : FAILURE in 43m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8b2b12243d2d4941bf1958b52fb2f60e : SUCCESS in 52m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d2560a0378894f8ea1bd81ee2f71828a : FAILURE in 55m 13s","accounts_in_message":[],"_revision_number":11},{"id":"2f037d14b746013b21ad8f16dd872f4d4f80d208","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-07 09:39:31.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"42234220b5e4e15ff651fc474d7f1fdfe74460fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 10:12:20.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63b89df1343b4c159557b66bfe08c169\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82be1e201d8c454cb4f25a09a25e4ae8 : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6426d03e05e84f24ba06afe25e83e2dc : SUCCESS in 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"811b0fe1acf40fa8a7298a25604b94df1ed1a983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 11:33:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6efacd08a6c428c86075412bc1f0a00\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c94ec3ba99a74ccd9a5592ad2b4f6aa3 : SUCCESS in 1m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/41d4f90ea5aa4919be5b0265920c8e4a : FAILURE in 47m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a2269830d714f93bb70b678ddcee683 : FAILURE in 50m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7c0a6967aa8c4e2abd151bc91f2175ee : FAILURE in 1h 47m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb57eaa8b59a44de88e2da17090c6028 : SUCCESS in 44m 40s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92bdd1bf58d64cb9b6b795c4adc7b53b : FAILURE in 1h 05m 19s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/009fdcbc60fc47b2b35073b783cd246d : SUCCESS in 1h 51m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a7c0e22121e4c06a2030a29ae4837a7 : SUCCESS in 52m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d18edf75da4c49a84705fa022a99ad : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39354e46d7564d7aa3ea7d66428454f3 : SUCCESS in 20m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2440ff819ace406da0f1e3fca678907b : SUCCESS in 18m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d764e1f32c314273ab7fe811bc5fc8c5 : SUCCESS in 11m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2f7d7f573e684d75a9668bf38c9bd195 : FAILURE in 52m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5c6d6dc45414451988bca72c385602b7 : SUCCESS in 1h 07m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0c1735194eff4204b0ace916aee91f7c : FAILURE in 44m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/740ca357c9a64e99a81285f1ca97855e : SUCCESS in 1h 50m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e8fe243ee0b4458ac4604b407665d5a : FAILURE in 51m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5df5f936b9b3415f9ba56e381d3b8e19 : SUCCESS in 1h 13m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a98cda261fbc42888d2773c31b9bfa22 : FAILURE in 35m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7ef99c9cabdc4d7d871e5afc256d3eb3 : FAILURE in 1h 04m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f843b19d66294e05b07ff6fb6ad1fdd2 : SUCCESS in 1h 05m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/52d6665ab6b44b43bc8fc98ea4758a45 : FAILURE in 46m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0d66c2c5d4f94c3e9da4910eeeb44c48 : SUCCESS in 1h 19m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4123acb67f4e438dbaa2edd122e8c77e : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":12},{"id":"40d3c766981e8f532f036d98d3125167b5944dbe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-07 12:17:55.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"77494017709ec6b8f518cb76d6cba11452d0d12a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-07 13:56:38.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"65e6fa59f45742182c2039cba774b4158a3ce33d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-07 14:03:05.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"03843a8c928e0b04640f3adcfcbab383d6f83ec8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 14:37:23.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03b0d6f17f6e4e29b682d1e6104d8243\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e72622b62ddf458fa0b69d17a0f075b8 : SUCCESS in 28m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c79695891a3c4e62ad4c7d904102b718 : SUCCESS in 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c60c518e17bc691d3b57b7a1dce29250ff673fd7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-07 15:31:53.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"7d71653c1d70a8896a48f3e3c344eb45b37e4e7e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-07 16:36:22.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"9c58a90d557441bb87da313e2dd02de3f6036177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 16:42:26.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c17892d96f347908cba6d9f65cd931d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7a9b33fe58554f73a7ca6a2e52523425 : SUCCESS in 1m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b41e763050e14d25b96681ea531252ab : FAILURE in 1h 07m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd19023ebf6240a78ccd92cb5bbb6ae4 : FAILURE in 1h 13m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ce187d298bb94b63be3433f91c95d466 : SUCCESS in 2h 35m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/86088c55e5244b589a51d5e9fa0fa075 : SUCCESS in 46m 20s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe8a96d7e69a4e78b5885a92bb0e66a8 : FAILURE in 42m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de3b521ba53b4ecea71466dfd43c1280 : FAILURE in 2h 09m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfbd0f5f025f413a99ae09cc5c4d22ee : SUCCESS in 32m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/291568003ab74c9fac1771c500c8df72 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0382002d9f6407496017caf499a13be : SUCCESS in 20m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f0736f81e534ec3976f844a02cb1c64 : SUCCESS in 21m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a66f8fb8fd4e509ffb5276251a8e16 : SUCCESS in 11m 45s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fab6847e548149c4a78438ba6dc56b8b : FAILURE in 46m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/699c828ee3d74ac583b5231385de6459 : SUCCESS in 1h 02m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f5683f212e7e43e690597a272a6a41f7 : FAILURE in 55m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7fb26bfbfe584c539f7e856223169adc : SUCCESS in 1h 59m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84d86a3cc4394d9f9145e1426ec617fa : FAILURE in 59m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a317b00b5ffd4b6ca72ff3d04f082c34 : SUCCESS in 1h 00m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5790ef8569a742109bf2d20f596e1444 : FAILURE in 1h 03m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eecbe5ba0f1047868285109e320824dc : FAILURE in 37m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/171c54828b76467491d495d897a636f8 : SUCCESS in 1h 04m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8f43a35c0c1940fcb5133ad322d6e5ae : FAILURE in 43m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c498dc9fab34e9b97b81e9fa41a1570 : SUCCESS in 50m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7c5f581ac6fd46bbaf5f1812f5cf7ccc : SUCCESS in 1h 21m 07s","accounts_in_message":[],"_revision_number":14},{"id":"9a5c8e347aa7e38793b58decfb7cef08e38b1530","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-08 09:46:48.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"390fa7f1a44034f4f5782fbbd36ca74de5802dba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 10:36:33.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0f1c093530f4ae6a54871de3899acd9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1aa7be3599a4395892b251e91e84172 : SUCCESS in 47m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/464a9cf1352c49b9822cdc35c5c5f96d : SUCCESS in 29m 18s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"bd87513800e28a9a1466ddf5af12c0bc774a8402","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-08 11:24:25.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"294cf78b611b8b19e4bfe9ba1437df43ed0b0efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 12:00:43.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dd4d01bc4ec4f2a96472f6dfea74239\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fbc15583e1564ebea7ec493db7adc93d : SUCCESS in 1m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f948b635c0ef4ca8a4b9f6483df53d2a : FAILURE in 1h 07m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3a7b03e8d6d84e428f2a9bd9f18c09d1 : FAILURE in 1h 14m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9c8eec8fa35944d598284ff54818ff59 : SUCCESS in 2h 11m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/14f32b4271c448e985b4c5d4aab82153 : SUCCESS in 45m 37s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfe68bb3df74434ca5514f6c2f2ff742 : FAILURE in 1h 01m 54s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8249d309328d4cdfa052ace86efa47e9 : SUCCESS in 2h 03m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d228c8336f24748a17b0f708d5d7257 : SUCCESS in 1h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3758a163b7744b859c66207983f5f701 : SUCCESS in 11m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f1b14cf87ec44ea9b26e861caea6cd4 : SUCCESS in 23m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c88f40363dd4cabb15d3071b0d1ff24 : SUCCESS in 18m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656ee76f29e440409390f15d27f4e62c : SUCCESS in 11m 37s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/13918152b00c43fc853ed1ba0dea5675 : FAILURE in 52m 00s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/83ddb62b0a55433794474c01b3da1eff : SUCCESS in 59m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4d03bd30caf04a9a98cdd1d9d746df36 : FAILURE in 54m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/26496ff753e849bca5862b1c944e3fa0 : SUCCESS in 2h 09m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d894efa7e2ab4c7f84a91e162d85ba12 : FAILURE in 58m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/10961d0a8c4b4676a3cb5129fe9ffc99 : SUCCESS in 1h 16m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6f29e8138cd046d5a63196d1dcd8b1cc : FAILURE in 1h 04m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7624b95fc6d54ecba9f739de591215d3 : FAILURE in 58m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3b737755bc154119b5fe799066a76e30 : SUCCESS in 55m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/19f49c08c7b940bba6607bc1f7d1749d : FAILURE in 58m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c5316f27addb41c2a94eff8ca5401969 : SUCCESS in 1h 11m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fffff9c6333d47e6b97bb19659e4a8f8 : SUCCESS in 1h 31m 10s","accounts_in_message":[],"_revision_number":15},{"id":"2d778dbe74310a56868b4c1a8b61588ecf0e9958","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-08 16:52:34.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"94135451b1e53536308c412044ea13be1c48b3d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 17:32:07.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51e59fba5ca944d48e1903a1aca88c0a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/077183f8dae945a181f773554aa50b66 : SUCCESS in 26m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3946dbd27cf4034b308b1fd7b068193 : SUCCESS in 36m 58s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ab2d565df8755d77394e3615563237bf7485cd26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 19:05:39.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e889155c4bd748bcb8ca1ba2dd7d165c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/77d5a5325ed34d06a840674c9eacca18 : SUCCESS in 2m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4f86ce8c9b804692bd051e97b002489c : FAILURE in 1h 11m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cd4e2986c277468bbacf8f9fc01df7ad : FAILURE in 1h 11m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/88f0edbee9f64474a16acb3eedc21050 : SUCCESS in 1h 53m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6ea7c03243b47098be501a6a442041e : SUCCESS in 39m 51s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fbc6d365d434cfa91f15108ffca8125 : FAILURE in 1h 11m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a0f1a18896841aba0be2ef2d0adcd9c : SUCCESS in 1h 49m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8949cafbd33451bbb01aefbf59516f9 : SUCCESS in 40m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c60837f23a4f32b17892e6d80e0f4c : SUCCESS in 10m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d237dabff18a48fb9d43cb17facccc20 : SUCCESS in 19m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5d4d1a6e0e54431afa030c8206ee952 : SUCCESS in 23m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff546a6f2ea04185a651bb561c51e36f : SUCCESS in 12m 13s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3960f27377bc412e82da71942b621826 : FAILURE in 59m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9357776388a34d1d8aaa6e246747dece : SUCCESS in 1h 17m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e14d19f535c4912b8a559aed41c2021 : FAILURE in 54m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/60212309edfe40e697803abdae3f301a : SUCCESS in 2h 06m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/452a45209c0c4c1c94e8cfca430f5537 : FAILURE in 1h 05m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b711633c22a446e9f8ab81bec9d363d : SUCCESS in 1h 27m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/285a05a148454548b915312bec4a3b20 : FAILURE in 1h 05m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f6dc92b4f77c4521b7b437301397390e : FAILURE in 1h 04m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dd6605f9434840dc8e24839861e396be : SUCCESS in 1h 02m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/696ce0cb8c34450dbb710a422fa906e4 : FAILURE in 58m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e2ee52621b04e91b3f7b6c91c4f0389 : SUCCESS in 1h 07m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5b1d0caa2dd2485cae8d55f3ffa2c4d7 : SUCCESS in 1h 20m 08s","accounts_in_message":[],"_revision_number":16},{"id":"61bdbe96693e301ba3a2d218fb65399c42a32f6f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-08 19:41:29.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"d991a7582da829a307856cf5178a625ed7f65922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-09 09:33:52.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"ab2b9a2c57a109a5472228f1004bb352ad8afba1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-01-09 09:34:12.000000000","message":"Patch Set 17: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 31, column 31","accounts_in_message":[],"_revision_number":17},{"id":"ca5ffb60322a2c0d70e156eaef2e3c99fddc3d10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 09:34:29.000000000","message":"Patch Set 17: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 31, column 31","accounts_in_message":[],"_revision_number":17},{"id":"81aff70e4db0647d33401e399bfafdf9c203c612","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 09:34:37.000000000","message":"Patch Set 17:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 31, column 31","accounts_in_message":[],"_revision_number":17},{"id":"99735d53aa522995a5e686eb9f810cdb267a7054","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-09 10:28:06.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"8d54fd8ae4fe431e55f06b6b452e179e4b1070af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-09 12:35:20.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"60383777d6c7f27b5d01230ed48813d2ed358204","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 13:21:52.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7bb772714674beab504366dc54c2908\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9aeb68d6f184917a9135b5727209648 : SUCCESS in 29m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3dcdcd2b76d94083a83837410e5d4efa : SUCCESS in 22m 16s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0af5b81df617b6b3fac4dabdc6d1b866973b6590","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-09 14:18:43.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"c85a1306cb56ae31fc9f8db7725cbe39fc6ece58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 14:46:29.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd13ff54204f4a6ba21ce309146b5bfc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5911ec679d04c758e6132b5115adc73 : SUCCESS in 2m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/28078a2f959944e58106d3af3824cf82 : FAILURE in 48m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2f27dfebd605493c81d9dc88c1b99fe7 : FAILURE in 50m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/99893f95b77743a09379df9e6b1dd9a6 : SUCCESS in 2h 06m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/507cb89de5ba4f3ab191bf8884d4c17b : SUCCESS in 35m 23s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a73565d9a90f415e928901fa1ce77ce8 : FAILURE in 37m 52s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4f61a468d7d4c1e94c05b23db168d51 : SUCCESS in 2h 08m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c4522ca80824f30b7a3557e43e5ce05 : SUCCESS in 34m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44626e81f45240b9a28e08482460a345 : SUCCESS in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dee5f1e22c61428d873edaf764398420 : SUCCESS in 17m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/42fd9b7e3e9040e5a636b30a91addc91 : SUCCESS in 17m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f06844b83bf461aa748ba232d454680 : SUCCESS in 11m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e574fc86ef0440d69950abcfc4d1df2f : FAILURE in 29m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/97deda57fcfc4a8cb49d9ad8fdde3160 : SUCCESS in 37m 47s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/7b2bb00d7ee04340bf664c528eae6287 : FAILURE in 15m 10s","accounts_in_message":[],"_revision_number":18},{"id":"571491525e71161d2466bf26116ac6425525ed9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-10 10:28:29.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"7ed8a0bf6d07be0aac8ccd1cc06ac71cf93e0a61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 11:18:24.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3df74a0d9282450a9c3a1cde06a5374c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60d5805de12648cd9741b7a8f8486b50 : SUCCESS in 26m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cf153ea11ce74d78985043d6fe503a9c : SUCCESS in 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"6b884339c9d0f8f3b511e0eb375c5b9a71088fd6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-10 11:31:36.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"2af4de2fffa291aa994bc0e44df185d64e381148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 12:59:14.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/055583c30cef464296b03c5b69668912\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/738d46e29d4048e2bed1e53b0eb48376 : SUCCESS in 3m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d2237ea731cd4f679e9438a68c19fee3 : FAILURE in 52m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/053dffc647eb4d52aa0cdbfe12b753f7 : FAILURE in 52m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5636ec69fc184b6dad5c071f83a18043 : SUCCESS in 2h 26m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9cec68293f2f4d5aad5014cf3cbbdef4 : SUCCESS in 34m 40s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dbe6fca21284e31af98881230368991 : FAILURE in 47m 07s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c93cb00941ad4bdfb972d4253b7cbb53 : SUCCESS in 1h 55m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ab6bb1183d742cb91fc8e2880c1857a : SUCCESS in 48m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2037f46c500441cc967cfc20c333b52a : SUCCESS in 11m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e1823d954bf4aac88a21cb7340b3bf2 : SUCCESS in 19m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3954695f2544157acc96cc244513823 : SUCCESS in 18m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3a7a8ed1b3c4460be2df437fde83a72 : SUCCESS in 12m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/afcfcfba7120462a9a791f6a40d1ea1b : FAILURE in 54m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9fc53b19990f486b878c7cfadfcba098 : SUCCESS in 1h 12m 51s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/99de184433a140b695aff6be00d2f770 : FAILURE in 39m 18s","accounts_in_message":[],"_revision_number":19},{"id":"fd7b522825ee763aaf5d734932d462a32b200498","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-10 16:12:59.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"a87ca5a5fb0d255a70070cdf70f51d410703dd49","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-10 16:55:55.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"4b50323db50fa1c31ca082f7c4c5eafae0d75564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 18:32:28.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d1c7f13febf4f3ba3c6b8b3d43d185f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f822ac8ce1f5402189bc4802f664edcc : SUCCESS in 2m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a4d07a19438e4a7a9b214225e6319359 : FAILURE in 50m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29e9112e0e6344fb8dd3e6f241b57ec0 : FAILURE in 50m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/186827245b724d17b6ffec94e8590213 : SUCCESS in 1h 46m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6506566b24ce453ba30a5d6892c43118 : SUCCESS in 33m 08s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2979c64c31b347a7a2cb7ce9805e49c9 : FAILURE in 1h 12m 42s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2628287aad8b4afeba59641cd8970e78 : SUCCESS in 2h 15m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a51bb007fb254f4db3de7f84fc098d4e : SUCCESS in 53m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5aeea22356b44b4bcff696980c35b3a : SUCCESS in 12m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61e65093258a473a8586d3c2793b6c0a : SUCCESS in 21m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf7cce65f1f54bdd8c11c455b05f3daa : SUCCESS in 24m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a18e42b0f0164c6db42cdcd9579a568a : SUCCESS in 10m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/03ee6678764d40e8a13e617e3aa5ff30 : FAILURE in 54m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/25422317a7b0460ebb8ed3d871a5a29c : SUCCESS in 1h 02m 56s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/454a1e34434c4492afd48ae78cdf7917 : FAILURE in 24m 41s","accounts_in_message":[],"_revision_number":20},{"id":"071c5204338f42cbad604023ae6e2f8873044de5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 18:42:17.000000000","message":"Patch Set 20:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/938337 is needed.","accounts_in_message":[],"_revision_number":20},{"id":"7c7a18f6cba244093fe584119029df6bbe1a007d","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-13 09:11:27.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"5888396dd8fbe8414c86e3f3b5c38840dceba756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-13 09:29:10.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"71ca14d3067948fa1d18c479329e49f5adf93e87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 09:50:02.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb1e464ea97a4fdda09a8da289c35fe1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b61cb5d20592407f9f6ecbf07ceff400 : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/74f4229b1d114d35a312e4f72b0c8a78 : SUCCESS in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"2b569b5633d38f7e6e278ebc3c7d98a5aaf1adc9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-13 10:30:18.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"1b0b0563e98afb8c4ee40dcdf4a8d843f24de5d2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-13 11:01:51.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"1caa39c2e47d8e818e2a53f5b0ff3684267fc99d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-13 11:56:28.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4099218855c9bb72af757d028c9f197a752a8d19","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-13 11:57:08.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"52db066abb489b32d07521da7e6e2bd587db2787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 11:57:15.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/028cdb5a0972411db3a5f4c57ce811e1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec822677749841b88ac8c80e21ebc0e7 : SUCCESS in 3m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6abe00710469490fa46a9e1965770e87 : FAILURE in 1h 07m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eac51eaf23c74a2d97ac9dc76d1480b4 : FAILURE in 1h 03m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/833a3eff83774046b89a7157f93f8d1e : FAILURE in 2h 25m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/91c5eafdd8fd4777836f08af13520366 : SUCCESS in 32m 29s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2593b1c098f04565847938498f26cdfb : FAILURE in 43m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cce1873e6d24fdaa607d8bf564d53a8 : SUCCESS in 1h 35m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/445606985108423ca4c89dc8c76ef45a : SUCCESS in 54m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bb521ba4fb243839b07fb3152c96c99 : SUCCESS in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b742cc82f5b48e582c4c4d3d75a4cdf : SUCCESS in 20m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2f035c81b4b4ffab8e68cb07100a521 : SUCCESS in 20m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a743eddbc747fb9a323b629a2e03b2 : SUCCESS in 6m 12s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6057572b1b0d43a59e73fecc8bfdbe0e : FAILURE in 1h 01m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/461765911010465cacad05e70d09ef59 : SUCCESS in 58m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/61c53e581f8e4df1bc977c2fbc948121 : FAILURE in 1h 01m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/196fd8fc93824ee48482ce4b6a158071 : SUCCESS in 2h 10m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4e6a69b58c6468181088700327230dd : FAILURE in 1h 09m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed58bff5bfcc4223ae8b79bce55d2a9b : SUCCESS in 1h 33m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/28e291819fb54320bdd443bd55750b5e : FAILURE in 1h 02m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/08d38104fc81444aa46498e30e76c3dd : FAILURE in 1h 02m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9367aea8dc0a41a5a023a7a31b80e927 : SUCCESS in 1h 29m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fcac8e015f184335b9c207d239f8db0f : FAILURE in 56m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eed82da60f5a471f805e0d08e702ad93 : SUCCESS in 1h 35m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/875ba72b585047ff8ad861c09935b628 : SUCCESS in 1h 28m 56s","accounts_in_message":[],"_revision_number":21},{"id":"2c5c750940391badac7771f2122073c5380b310f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-15 08:09:58.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"4737d704bb5a37a16962370713bd7ace9ad2da4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-15 08:13:19.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":23},{"id":"065b93511a9dbac6cc4cae1b266a4406283054cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 08:43:42.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f05ad5341d364e199be000874b666fad\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e04abf22290745bdac1d07f915513ab5 : SUCCESS in 24m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e5558f64cd448978569958f9a98142b : SUCCESS in 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"d0438826a5a46623e6b510cb3b57f907ddfe1b34","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-15 09:25:51.000000000","message":"Patch Set 22:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"a35b76bc392c33f8a9128b477609d9b4084b8864","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-15 10:28:22.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"2c41331602e30237b759432b618daca8a9012d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 10:48:05.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d75025b3e7f541d58dda1d2e408fd4b8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ec251a66b0b4677abf518259f665c07 : SUCCESS in 2m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8573c370f62f4df68f70e196e94a9ce3 : FAILURE in 49m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e3b892427b8c4567b7d3bd27fd3b52a9 : FAILURE in 52m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f86fdc63f4c24703ba1dd2aa79b8f0df : SUCCESS in 1h 44m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/37699c98c2c24f679f4a9d693811c97a : SUCCESS in 35m 02s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6e91e4e5bdb44a9b784904109e7644f : FAILURE in 49m 38s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c388df93b2224d2c8742f69fc79aaf42 : SUCCESS in 1h 43m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe3ea1e8e4ce4475bf79017bbcc17f13 : SUCCESS in 51m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6600a835ee8741628af12db73bffa243 : SUCCESS in 11m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1df7120f1111463e96995894a928f99d : SUCCESS in 23m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/346b7e70a1cc41d8a9ec8661e5827887 : SUCCESS in 21m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0966285f7b0e4d9eb3e5fd87922f173a : SUCCESS in 10m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/09b270edcf694c3198fed51d69ba5a4b : FAILURE in 56m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d48754853eca47b5a21d6e7ccc74c747 : SUCCESS in 40m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc157da0aca448fe9b5f48b7a5a5c907 : FAILURE in 54m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4f8593e9dfc34118a4decd1e4164a7dc : SUCCESS in 1h 52m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dae1f84dba048a8b24a11b23bfa21ab : FAILURE in 40m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4749c66a00c4f45ab8473ea5f924f8c : SUCCESS in 1h 12m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1f403ed12c064d23aa3898b220ec7cc0 : FAILURE in 56m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ba003826f4c145d6bf067dfc8346f793 : FAILURE in 37m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/10d68951ceb94aaaa6253692b179d89a : SUCCESS in 1h 14m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7b454694ec394d0aa2f8c822ffc20336 : FAILURE in 53m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0277e0b91d9e4609a4dee270fcba8b44 : TIMED_OUT in 2h 33m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5cb22d7d8e814b1080939ef83e177829 : SUCCESS in 1h 26m 34s","accounts_in_message":[],"_revision_number":23},{"id":"100a085fbd5573e81b3ec9d339b0cdd33a814a75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-15 11:36:48.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"fadd97e07df04075654c0fec13d017a5c41ceace","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-15 13:46:53.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"da2bc31183c73d74e04506a1d8384f63417d4382","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-15 13:53:18.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"320f8833e7578f9ada68e6b9967c36028ea5b73f","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-15 13:54:41.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"ec907444bc657e495aca02e465795ce152360b24","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-15 13:55:06.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"acdadd1f0e8c1650482ce80dbb1251b41b963a10","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-15 17:07:51.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"44d344e509f712532456960bd43a88d5aada2bae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-16 12:27:37.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"69ef4f3f3c592322da1e371baf84323cb6138a43","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-16 14:39:19.000000000","message":"Patch Set 26: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"528e6631ee66c22a3b272191cc73aaa14b6a936e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 14:54:31.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de22f36ab6d4416abee25d89c5999fbf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fa1a0b66fcd944b4943a93ae1312d33d : SUCCESS in 2m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0bd70e0cbc8a4c389ade6b5b2e244b6d : FAILURE in 54m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8dc5a98b50c421dbbf87f046c967a35 : FAILURE in 57m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/924aabede51a4e209d32261e9bfd1b81 : SUCCESS in 2h 25m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/135a463e40f44096a59706112f622b5d : SUCCESS in 35m 46s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03d4369dc6bc4bc5b64160957e598646 : FAILURE in 50m 03s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d667a814c52419c939d3ccb9938c40d : SUCCESS in 2h 15m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd5a511f64194a3dabf3107beea0fbfd : SUCCESS in 49m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9b9a2e2cb7e4bc6870dc12725cae5e6 : FAILURE in 9m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c93a3e54a684f229e1f7a049f26edba : SUCCESS in 21m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c767479d88d3490a9ff91a087b2b6e0b : SUCCESS in 20m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25714b6323164196b1a22ec60b736b96 : SUCCESS in 10m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/343cb5f30eb348b3b1e73679bafedf60 : FAILURE in 56m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a9edb5ff498540acbdf7070152d917a3 : SUCCESS in 1h 07m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6df537e91e7449fab2cc4c399d0c9844 : TIMED_OUT in 2h 12m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/886460984240463991f2ed1dec9edc63 : FAILURE in 1h 22m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b74e06b914b642798b59043d5b4d5f96 : FAILURE in 32m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d7a1f3cb6e343b39ea508f89e1d2cc3 : SUCCESS in 1h 23m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c861fff3d6740268ffb895f711a5934 : FAILURE in 59m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a6666bcc12a24e1b822e4d5e65a9cd0a : FAILURE in 1h 01m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/190a0b369be54104bb94c11f57f7a4ef : SUCCESS in 1h 13m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2acc034957024767b017f83e11bad155 : FAILURE in 35m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/30124ccb0c3645cf9b97efd1b958523a : SUCCESS in 1h 11m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c85d34f98df4f5c979e84b5b84eb2e3 : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":26},{"id":"609eb453b1d8185e9c80cd6ccb0a7722885cb40f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:38:19.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9635c85f0e04ad885df86756dfcdda1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e1ae8220c7a45b1b8982a5683bd5f24 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f6c7eaefed064fb392004e4b6a3fe464 : SUCCESS in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"62159fcbc51540a9e34db3ed3a24cf977fe924a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-17 08:08:21.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"d130969f20b61881e6a65404759ca5190b344309","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 08:57:08.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14dc527c391a4bd2bdf1d49de1d27ab1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9528d24428e84c3eba839e03fdc81590 : SUCCESS in 36m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9bcb31093bc043ee9b6086231f71bce8 : SUCCESS in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"46fa1e92457956e97fbbda5b53c0b73fa2317ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 10:33:59.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc4d3e0489ff45caacb2880b5013652e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/065debadbf504715b4029fcff3137d85 : SUCCESS in 2m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b8bf1138dd704ebbb8a99d197c603f60 : FAILURE in 1h 11m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0e9155c306564bd7935d179ac2b592f7 : FAILURE in 55m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ddb5215417654165b180a4f7eefb3228 : SUCCESS in 1h 52m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e3381650d62b4359ae5f1b82d6bd0dd5 : SUCCESS in 35m 10s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c593b78c7f4c4cb4972be451408dab89 : FAILURE in 44m 19s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0608c0296e9469b8f0aae809ff7c14f : SUCCESS in 2h 11m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe5ceaf69aec447192b3717387598183 : SUCCESS in 50m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12b4a85f54a4db090e410e8d1cf7f35 : FAILURE in 12m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f7f53a95bd849faa0aca6733c0f8c1a : SUCCESS in 19m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/441847cb14e04ff98e287b51fd76d85f : SUCCESS in 18m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/366788ccd0a74a6eb453181421703885 : SUCCESS in 10m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/02b04c5a11d24b47b4e4636577d6f82a : FAILURE in 53m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b148523e4f5c4661a953f1dc724132c9 : SUCCESS in 1h 13m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f0766a7a0e944ab98269445974513dde : TIMED_OUT in 2h 13m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e239e12d04ac41889ccfe41258ecc9c8 : FAILURE in 2h 02m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f70a5f33b2bc46039e6ee46227c97ceb : FAILURE in 41m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa1ec9268fec47fd93a0b0de7d28fb67 : SUCCESS in 1h 11m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/54a022bd4c9f4b8188596206878e7ea1 : FAILURE in 56m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3807ddc2563c47fa8b650f21bc475920 : FAILURE in 56m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/57ab2d08a6264bc8a69860d49a9e8d41 : SUCCESS in 1h 11m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c5752b934e384cb4a743553473c61ff7 : FAILURE in 53m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ee4ae5f87b3c4785a408f264b163e987 : SUCCESS in 1h 03m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/86c526b7bfbd404eb5db4a2410318ce8 : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":27},{"id":"d72126dba3c807c3c5b3b425f8ed6ab1bda4ef7b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-17 13:15:10.000000000","message":"Patch Set 27: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/27 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":27},{"id":"8e31604df21d68d608635e345767da651f39ee70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-20 12:06:50.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"0d391aebebdd686a5d9ec86ce423208f31ad95e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 13:09:54.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a197fa22a82848f1a38b74e926f86449\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25a4a21d8111417688ea74a69dc7e138 : FAILURE in 30m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/699548193485418782b673f57438fe84 : FAILURE in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"a4391b01faacaba8414058afcede6accf40bdaf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 14:04:45.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48d43e52bcc343a9bf56590b6b1a4ad5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/257e0776860548629bbffd5576bef851 : SUCCESS in 2m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f3ec6b20039b4c0f806bdedaa7c3fe32 : FAILURE in 1h 10m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/624d999d00ff41528e193697f6b3a708 : FAILURE in 1h 14m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dca037c7a28d419fa4500b4318572094 : FAILURE in 20m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de3272539d4f4f06be9db5e5c9b91728 : SUCCESS in 33m 06s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84a94c790bd6479ebd9d3ce59451b933 : FAILURE in 1h 12m 10s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/898de08dd1424945aa1930a6f5e64745 : SUCCESS in 1h 42m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bce4eee517534308831d151cf4176025 : FAILURE in 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b520ee9e1334e23b43846771b168f43 : FAILURE in 11m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e91cd90e0f84cc49750f43070d77247 : FAILURE in 20m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4eacc6fc191b4b7cb5a5210e5c580ae4 : FAILURE in 16m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1812faffac4b15a1438071f6334182 : SUCCESS in 10m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7113884a0639437ca0f84b7c2f0e392d : FAILURE in 50m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/287fafc2a55840e9894eab815f86f0e2 : SUCCESS in 54m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/741f07649e0f432488e6beb3bc73bdf8 : FAILURE in 1h 33m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a8dc6bae722c42b3ba332234fbd3cfcd : FAILURE in 1h 51m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/21eb0e335bbf4d92898d9d39e33b4e73 : FAILURE in 52m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/97fefe1ec70546a18350d43f3c7773f6 : SUCCESS in 1h 10m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c6b91b3d258b4d21b8793fce9be8bd84 : FAILURE in 1h 03m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7b3f44e9f47d4e6ba88060dd7cbe4d9c : FAILURE in 1h 02m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1c21be5c625f434abaf7b43887f65ce7 : SUCCESS in 1h 20m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/663f59797e7c4e20863cb358e20e25fd : FAILURE in 1h 06m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ccc48c9b74f94a9cbe969616431dc234 : SUCCESS in 1h 10m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3914924a49db4b6d82e45cf8ad3a92d1 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":28},{"id":"237ae11d403237fcfdad6c67ded5b934fd12e5de","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-20 14:23:03.000000000","message":"Patch Set 28: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/28 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":28},{"id":"0c4a90701b7e08a34ba7413ad79e2db7d0f5ebb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-20 16:09:15.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"d2aaf3b3623be8d728656da0510dd98f56c8ea68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 18:27:27.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/155a4772dec7404a853bbdc47d4457a9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef81ebbcab2343ae8d49bd5fe2f37f18 : SUCCESS in 35m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/14eec44bc389428fa0d947fc9b78924e : SUCCESS in 39m 48s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"4ef76f2bbd9a04f8ddbbc15db9469283d1b26fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 19:24:40.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/112e5b6784b34c9baeb15eb60b5091de\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/50b60b6219f8469f9a5ae4ad4e991cff : SUCCESS in 2m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1bc50232c3c048ebafad99171834f272 : FAILURE in 1h 05m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9191f019defa43ca8b829b3ac50be552 : FAILURE in 50m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/87de31438e524de6ba18e37486d63e3a : SUCCESS in 2h 11m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e189728df59b44ecaa5e5f1718abf4e8 : SUCCESS in 35m 34s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0e51f4f05d94071ad072e803deddaef : FAILURE in 50m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b13cac963ab4711b654c6b32b842e41 : SUCCESS in 1h 44m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3269c4f40c304d388330253b54cec2e2 : SUCCESS in 46m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21d55a6a7bec42698cf4bb2963c48218 : FAILURE in 11m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/febe9f1ebb984d9a972424e6f493d6b7 : SUCCESS in 20m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/184546971d074af99469879549863665 : SUCCESS in 21m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8563ecb1407346d598d4c87825c9848d : SUCCESS in 10m 56s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/65388e57648d475694a38e29cd8b17ab : FAILURE in 51m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2c86322d670343d59002d46b753ab6f5 : SUCCESS in 56m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/17bc128f1d6642d2b502e2a46159d7ea : TIMED_OUT in 2h 12m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4644c5cb34ad407382dc74d08938adaf : SUCCESS in 2h 14m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c406663a591d4b00b7a81943bb1f1792 : FAILURE in 53m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0091bb0b51fd47c7bdbc19df1e4e344f : SUCCESS in 1h 20m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8312d4af0c7c40818203f1180d0e430d : FAILURE in 49m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6102fcff317140cba594aa882df43945 : FAILURE in 1h 08m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/35244be098ee4926bf02ea4f063abeb9 : SUCCESS in 1h 00m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2dbb96e474074c79b1b8c8b3e6baf4c9 : FAILURE in 1h 04m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c10487c33b28456f8af954c1c1b04888 : SUCCESS in 1h 15m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cd1b5ee602d24dfd9f07424bbb4baa4a : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":29},{"id":"80e9f539c33c4748e637594d78c1d1865a60b4a0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-20 20:27:03.000000000","message":"Patch Set 29: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/29 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":29},{"id":"6a2105688ed7d233ab27623ef793add6a1cd5ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-21 07:40:39.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"2904427803e1e0f52df390b86164145d22d3351a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-21 09:18:19.000000000","message":"Patch Set 30: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/30 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":30},{"id":"45dc322fba2fc03e04ca9766a15bb464cde8c2c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 09:50:11.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daaf682a285f4bea9d117586d85518b4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/daa354bf50134974ac623a3e98911941 : SUCCESS in 26m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b3b1d4694f5f45419121ff1b596d4bd4 : SUCCESS in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"65435df0b99ddc83bc8a1c51bbf07c18da903ed1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 10:00:57.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b313b72524549958bf6e5d8847e285e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d0c59a08cf584fc68600299e53675ffb : SUCCESS in 1m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6c9c27f95c33491f87325f8376cf5346 : FAILURE in 53m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0efe439f363940ac8e7f60bb007d5ff4 : FAILURE in 58m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/87a002517c9a4f9d913e71d80943ec28 : SUCCESS in 1h 58m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf8a368c58db42cb8380c41dd7b266f8 : SUCCESS in 35m 21s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb3e45ceeb024c9795afafa6038fb272 : FAILURE in 47m 13s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e52b760719c9407497bcdd5d0ba6852a : SUCCESS in 1h 44m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60447482518c4a80829aef32ee7e9f3b : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84853b8d85904e8ebda60e6f27907da1 : FAILURE in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/211fda96760648d8bcaa29ca4cc066de : SUCCESS in 21m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ed827dffa5a4f0eb73da452e84fe304 : SUCCESS in 16m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb21b591df740a2aa19459eec4cce1f : SUCCESS in 9m 45s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/34f6f1399ebe42eb9b555af657493926 : FAILURE in 52m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a2425545b94449cebc4ce285e462ce84 : SUCCESS in 41m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a380e13da0074931aa178ae92e030a57 : TIMED_OUT in 2h 13m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f5243ae645534b808de4d1abd4995a92 : FAILURE in 1h 54m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/71f11332e6a54758aada20d029dcbfe1 : FAILURE in 45m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b80b51722c414fcca2d69dcbe6f81c77 : SUCCESS in 1h 19m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/68899946184d4cd999e39fbd38ff4522 : FAILURE in 55m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/571f3920cd3c4b218268baae68c337e9 : FAILURE in 1h 06m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a931daac5d7a4166abd291ded56a2931 : SUCCESS in 53m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/97358b77dfc94d948b30d6a4cdbf0ea7 : FAILURE in 1h 11m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/12a93cbc9082420e87d6e8652acfc1b3 : SUCCESS in 1h 11m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/267d94b5667b4cc08e4d6f9d8bd1f57b : SUCCESS in 1h 30m 26s","accounts_in_message":[],"_revision_number":30},{"id":"ef8d30b7ac7126bf70bd085cf8dab0eb313e136b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-21 14:11:17.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"54c2ae02416d306052f96b09a6869f020945cc71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 16:19:23.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab7fea3d077e4499939fbf93554461bf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e583ed4e183a490cbf7dcfa0618e7fbb : SUCCESS in 2m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2ff9949f2ead4732843598066f1d356e : FAILURE in 59m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f36cb3fb87984b3bab5cdf46701d9952 : FAILURE in 1h 05m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d313fedb2fbe43e0a1340450a1cc3029 : SUCCESS in 2h 06m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aba4c2693e8c44f087f2ffbb2b77ebbc : SUCCESS in 47m 57s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6a94f3d78aa4427afba4fe4a8ebdb1a : FAILURE in 57m 57s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dff2871a164425693bd62467f5d9d38 : SUCCESS in 2h 02m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64200b7a9bf44858803f58717889b396 : SUCCESS in 43m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b64358fba44e2bb6debf656a297fe0 : FAILURE in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c66b663bdcc2491a80a905fbee56ebd4 : SUCCESS in 19m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54feef4e44a24751a701f4aadd9db385 : SUCCESS in 19m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad23bd0695344d2930ae2c856fd7ed8 : SUCCESS in 10m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1f24001183f84243bad823a64296854b : FAILURE in 45m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9c17adee1e0142febbd304a4915dcc86 : SUCCESS in 1h 00m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/362bdfbe60114beabecd7d1c34c62038 : FAILURE in 15m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/34aa6c4d88654855a2a4c26e8af10bf2 : SUCCESS in 1h 45m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/87e15b59905d4f338763ee91d7346e1d : FAILURE in 1h 06m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a0d0256654f463aada50bc4ad108941 : SUCCESS in 1h 03m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9b2a7324dbf24470b06bd5429bf3dcc7 : FAILURE in 1h 00m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3208c2d395fa4d038f220c9a4a6eaea5 : FAILURE in 1h 06m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d0108e21869d453682d1399e08997f40 : SUCCESS in 57m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/608eb136d0864cc696339bdc8a76049c : FAILURE in 1h 02m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7e7f5d382e54bcaacc6198baa8b30c6 : SUCCESS in 1h 09m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e0d064f61ccf465ea7124e797e6b9763 : SUCCESS in 1h 17m 28s\n\nWarning:\n  Comments left for invalid file neutron/tests/functional/agent/linux/test_async_process.py\n  Comments left for invalid file neutron/tests/unit/common/test_utils.py","accounts_in_message":[],"_revision_number":31},{"id":"deb145bca61592056460b351e6d4b75d4bd2d24e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-21 17:10:52.000000000","message":"Patch Set 31: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/31 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":31},{"id":"82d14c439987e32eaa4060da9dcdc24c109695a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 18:07:25.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f91c2b55744c39b431734a62e7e0c2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a951bca459154a36ac35366291c48587 : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eac8a928f7254da5b3a96e4a6802e848 : SUCCESS in 41m 28s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"223a7a436cba23fa0f7f892ffd963eed859cc245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-22 07:44:35.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"74ac04e585cc675a968214571e2625b605021448","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-22 09:19:55.000000000","message":"Patch Set 32: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/32 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":32},{"id":"8aac083d44e1b85903683f6334453fac1a9f1aa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 09:23:16.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f42f34d0f1a4671b398ff7ed62460f1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1293dc5eac44da492f94e9280674079 : SUCCESS in 25m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32701332000c42a4b31ee5f3d7848975 : SUCCESS in 24m 34s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"a4d282c23c987096f4fc5887c31701690c278989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 10:42:35.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15a2b8ba016f422485d0e2aa711615aa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2d478a1b375e400a97a6e717c8bb9fd5 : FAILURE in 2m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b727e2691f70402a8b25d93a55700676 : FAILURE in 1h 15m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/435755ed1b5246cea8d9d62d203410a3 : FAILURE in 58m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c589484de89d4fbab9c8fb2c1a08e252 : SUCCESS in 2h 11m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7b4216ac60b34f1ab7b1dd1450daf85a : SUCCESS in 45m 38s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/952bef4fce90430d95d5fc92868f4a81 : FAILURE in 45m 22s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2c2a6269e0c45bbaa748bdeee33b8eb : SUCCESS in 2h 12m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52ba349f2cdd4662a0ba968e60867096 : SUCCESS in 34m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634e2195deed489aa3cd7434dacb88df : FAILURE in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b23b09bd7dfc47219645d5a26a913bdc : SUCCESS in 25m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76ff6a89bd73492cb4635a803b479e57 : SUCCESS in 18m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6967f31099f413b8aa66908b392ccbf : SUCCESS in 11m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5058fddadc424fd98a934bbd511d3b4b : FAILURE in 55m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/152f29d880de4875a7314ca781402fc3 : SUCCESS in 1h 05m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/690a9805aa9846249cb3c9b7f25ed541 : TIMED_OUT in 2h 12m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/69f100c14aec4b459242c142791964ff : FAILURE in 2h 45m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/70c910e6ae7641ce8e74ec4251d13790 : FAILURE in 58m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d09bc90843624ba883932de533d3ee7f : SUCCESS in 1h 24m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fd1ef3cd32af4d52a2757e44e800c439 : FAILURE in 1h 00m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/09fc3f8c15bc4275b774753784ecdc69 : FAILURE in 1h 07m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3226bf639c8f4002a5b15ae526ead1ea : SUCCESS in 1h 13m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1cd188ae8a114b319136760ed805e202 : FAILURE in 1h 10m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/76b5bfc4734b4209bce69911286f2e5b : SUCCESS in 1h 19m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b652e018cd514930ad0d090ca34a43c7 : SUCCESS in 1h 12m 30s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/common/test_utils.py","accounts_in_message":[],"_revision_number":32},{"id":"880310a6a598b47024c6ca276a3a71a56b4d054c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-22 12:50:52.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"b3b2d0a9bd9300acd84283ec8cca3d4242c6bf10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 13:37:52.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c328129069d49498a5304718c4d4f17\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88002cbbfbc242d49373034ab1c22810 : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f839ca81a526402fb9d6f06b01bcf9bd : SUCCESS in 21m 47s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"856947623d0bb51d570e84c5d44584f861b82b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 15:05:23.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab8d8c04f2804b7287bc4b8b09920901\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cf3ae61578404692a48636f4cb351c00 : FAILURE in 2m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3bf973a8022e40ba863b54b964703b2e : FAILURE in 48m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b90a66d514334e05b4594191c32d5297 : FAILURE in 52m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bf22df14b313452492ae6d33415dacfe : SUCCESS in 2h 03m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/233e7625c6f34c75a52efaff8d65f9ae : SUCCESS in 40m 07s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/972afb4a8d6b447cbb81a0e4f0b8c1cd : FAILURE in 50m 23s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/873be4c65db34a16883bcd1ffe4998ed : SUCCESS in 1h 58m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3aeea1cdbe4346c1877eb2feac9b907e : SUCCESS in 52m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23628060e4f94a7f9633ad4905e79e3a : FAILURE in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe01ba4ed98a47229ba0897ab47e8ad4 : SUCCESS in 21m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e3bb441df904ce880854796ccd7a682 : SUCCESS in 18m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d2573fdacb48fca3230ddf78e12b03 : SUCCESS in 10m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0e597d99f9c04c6bba88744b76b0c74d : FAILURE in 52m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/84e285e812764b849ba3eadb4685936d : FAILURE in 56m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c3258deaf36b4db5afe6491c6577b651 : FAILURE in 1h 29m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2fd43a01412c487080c7c6e3151697c0 : FAILURE in 2h 07m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4509843898bb41ba8c05c6f82e4091d2 : FAILURE in 53m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/88c1aad7e4b74af8bac329f96a031b01 : SUCCESS in 1h 05m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fadd355c41b749d0bad8bd94c35d401e : FAILURE in 1h 07m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aad1eba3c8ee4136963c1d84ff9301c3 : FAILURE in 57m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/130b9ef6c2f4461796792ec1b44b3c65 : SUCCESS in 1h 02m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/90873749b00f4c42837275606574edd4 : FAILURE in 56m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c77d238fdb74f21ace2de89bb7c6ff4 : SUCCESS in 1h 14m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dec8546302e14618afe6d01e666442ae : SUCCESS in 1h 16m 47s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/common/test_utils.py","accounts_in_message":[],"_revision_number":33},{"id":"5f233026031d66e1d9223783d062164b218cadca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-22 15:37:03.000000000","message":"Patch Set 33: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/33 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":33},{"id":"aceefed3581a1a7ca9adb0cea7f11a70b9b3ad36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-23 15:42:04.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"53a48db483c6af085866fbd54990b85f65458608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 18:02:42.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cdd48bd6a364264881b88b48c4b1708\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9fea36d6b3f94d12be33da9123049fef : FAILURE in 1m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/df007e16d2a841ea91e4e54cc8a009b4 : FAILURE in 1h 00m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a8ad316d09034530b078b432dc865870 : FAILURE in 56m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/59e94ea4a94b42eba3b262e8d8d4c76b : SUCCESS in 2h 13m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e0667685e13742b3b72d7d9ab4f774cb : SUCCESS in 37m 06s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f61043ab28b34b6b867f2ad70c432dc8 : FAILURE in 51m 38s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7613339a7c04bea8da89a5e34d3ee29 : SUCCESS in 1h 55m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76e9c7aab6cc41a8ba704f232a7fc75a : SUCCESS in 47m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1d7e17fa0c40faa5ec50cf8c74e7a3 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8612924168ba4ec3952c9f60842ad97f : SUCCESS in 23m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/875b14c7263c490f9906b13816214f1c : SUCCESS in 24m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7883c6d3f93f4238af891d5ffb61f5af : SUCCESS in 9m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e49c31fd75184425bec3bdff13c253ec : FAILURE in 51m 09s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0ee88cf3690f417fbf5754fdec93d38f : SUCCESS in 1h 07m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/10979ab889de4fff9f1f90e8630e4ff4 : TIMED_OUT in 2h 12m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0c74020cd7e3455eb064db3b8d56fbed : FAILURE in 1h 33m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8deac37c52f4fa593a3febae36e3056 : FAILURE in 54m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b667e7bd79b426dbd74e5723d42cb2c : SUCCESS in 1h 14m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/851630b505254268bb62661e2dc43170 : FAILURE in 51m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/289fd1c7f29647ef81b075de52616c64 : FAILURE in 1h 04m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a355195a76524e038e36b297f6a55e2d : SUCCESS in 58m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f8fd760632814ac4a603250ec2729e02 : FAILURE in 1h 01m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9a2c28f7c30d4a2a98a0e02211ef9875 : SUCCESS in 1h 11m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aecd2dcafb25411ca9000b840a0323a7 : SUCCESS in 1h 18m 11s","accounts_in_message":[],"_revision_number":34},{"id":"34ef45a9bf14b4c4e666e46e772196c674870dbd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-23 21:23:35.000000000","message":"Patch Set 34: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/34 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":34},{"id":"1c29f43d36ec7c35ca1b297acf1ad4c73f21f82b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 23:02:36.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb52fb43ff5f4e8b9f678804225a426a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7970f364d10041f1af282debf260dafa : SUCCESS in 37m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91e549fb65bc48f08d5c49039a1a5519 : SUCCESS in 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"6e7b18ec7962f420f00d68d90efca65545b40af2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-24 08:20:19.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"9b845fd9a16fab0443c39ccc3d7d2a0d4e12ab63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 10:15:25.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b479fe2145cc4fbdbab6e0aaa87479e6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9871cc8d356d43bb9b7445a1de891de7 : SUCCESS in 34m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/33b81620ec6c433abc624a3c2524bc5a : SUCCESS in 20m 06s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"d2f5ff155a419093b7b7b4c25ae8d1fe7089817c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-24 10:55:30.000000000","message":"Patch Set 35: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/35 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":35},{"id":"2b6c331d373d4ae9506597d261aad6ac575b93fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 11:28:21.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ad2ecb5cc934fd9a4d125bea707bef6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/982e0caee00547df8de6d6560780c765 : FAILURE in 3m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/682632d19e774f78a80232e3f53391c0 : FAILURE in 1h 06m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/163587a26f2a4f56aebde7ff8417c52e : FAILURE in 1h 31m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca612bbc054e43f99a0aeb4a37207d4c : SUCCESS in 2h 38m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fd794ec0cb514f6ca508f04109fdba9d : SUCCESS in 41m 03s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6631be047c29490f988292f5fc2a9ee0 : FAILURE in 47m 58s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26a92e08447f44a7852da57b5b4e16fb : SUCCESS in 2h 41m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/663f4c49a4d24c58baf789c2c8668bd4 : SUCCESS in 58m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7962a571474e489094f2b2328e2f93b7 : SUCCESS in 11m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b0b3f01f1db420caea25d6f41fa02aa : SUCCESS in 24m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05c4993ecde24b228b583fb98a9eb8b7 : SUCCESS in 29m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c65173dfbed41358eca758ae3055c54 : SUCCESS in 10m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/022ef4962b184b02abf3e62678dd1cc1 : FAILURE in 1h 58m 24s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a0fe9be258764b1c9a1def4ff24a4bb8 : SUCCESS in 1h 09m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d349621568024b1f9fe68d6f10da3432 : TIMED_OUT in 2h 12m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5c40200671f7443cad8c47975d8e1423 : SUCCESS in 1h 55m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/08c9e7fb278942b69ec836b6d091959f : FAILURE in 2h 04m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7669b954b18a46508f2f880ee89d11e0 : SUCCESS in 1h 22m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bd7a8db355cf4af1a2edb2737af77cf7 : FAILURE in 2h 12m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3f4c2ce25ecd4ea58cb08b6a7b18da36 : FAILURE in 59m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/02efd46d54fd4c2fb6b160319183d3f0 : SUCCESS in 1h 41m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a33a026950c74347b5d9d5fbd2a06d4a : FAILURE in 1h 01m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c1f9c9598eb14b74801a4f394ab85072 : SUCCESS in 1h 40m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5b52f418f47e42cb8a0bee61b857f108 : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":35},{"id":"d1ada6eb6587bc149c047a36b33870f8a57474b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-24 15:07:33.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"e75e974ed323a744a3d0e56ae06b5f131969d729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 17:51:59.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9446cf9cc44343eab9cbcb3dcc2c0f23\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd47dfb9bf37405cb27b0b9bbb8560e1 : SUCCESS in 3m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4e438da5dc6d4b2b8c888674992365e0 : FAILURE in 54m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b9043595f9b84732b434b084d8e9a17e : FAILURE in 53m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a929565c256644268e0547ec0d5640e4 : SUCCESS in 1h 53m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d4e020917874b6ba0411d6a64bf2671 : SUCCESS in 38m 21s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29f1f765312549afaaf7c59063caad13 : FAILURE in 54m 05s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d404334e5af41ffa601cfa6f3dd2106 : SUCCESS in 2h 42m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ae6fbc6b28046729430973fe7e88cb6 : FAILURE in 53m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096dde5f6804406eabdef11a685f32ff : SUCCESS in 13m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8835b6aebf564beeaef63300c13730ee : FAILURE in 21m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cfa9a0d152444780a2a7911a7bf7387e : FAILURE in 20m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e8e621605849f2966d2ca71dc3ab63 : SUCCESS in 11m 21s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/43dd749b032d4ca387d6a3351588f939 : FAILURE in 57m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/708083425c57464cbbd22f713cc0caed : SUCCESS in 1h 07m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/152f6932b0314f7ea707d9443dae213c : TIMED_OUT in 2h 12m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9a949a1275814bacbc5c3b3dc0b4cc28 : SUCCESS in 1h 46m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/99886181e3b545d384d59da030147e68 : FAILURE in 48m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d411defcbd8f48f587d3ee71856063a1 : SUCCESS in 48m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/deda670867274daf8920ce664e44c155 : FAILURE in 54m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e240eff02072419f80927d1f3ffaa7a5 : FAILURE in 50m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/95b441bfaafb48a88c9afae4506b1f02 : SUCCESS in 58m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d7442d802fd74b32a5233e0796554e26 : FAILURE in 57m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d82fa6aa176449d5a0d85ee206d46cb2 : SUCCESS in 1h 21m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1a42af62738949bba9ad844425de7c77 : SUCCESS in 1h 15m 20s","accounts_in_message":[],"_revision_number":36},{"id":"fb49a0675c1b1c588dabd7d23a278f916dd1b1f3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-24 18:06:01.000000000","message":"Patch Set 36: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/36 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":36},{"id":"f2e6dc83db0d3f38a675f41322f2045ab47b54b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-24 18:16:54.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"76b661834c47ec13771dd900cd58c6e3af1e5533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 20:26:25.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64fe028b1ba5488983cd45b089c95349\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/158fde318323484f84cf09ce30d5f203 : SUCCESS in 1m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/26ccc0cb71874b7f952dcf5ff0ae31e1 : FAILURE in 1h 05m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd108ca290464975a4e134398cf813db : FAILURE in 1h 12m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d76cbf9053694f2a80c2452ae4418c0b : SUCCESS in 2h 07m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f644da3cde7440899e3cb059fd66825 : FAILURE in 22m 19s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59cf706742e04f298beaba549899ea58 : FAILURE in 45m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebddedeee7e04e5b86b85464a8ec8a9a : SUCCESS in 1h 58m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/966c78744edb4e908dd90307b9d15204 : SUCCESS in 35m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d475cac447434a8f5652ed69bea4b4 : FAILURE in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7940e124f2b548f68b994c93a5414311 : SUCCESS in 29m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31ebeb0870744a239f894382c818d1a8 : SUCCESS in 19m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a799d7632a784d0db73d66b11b9134fa : SUCCESS in 11m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f2318e4eb92e4d338205dce1e05a4a37 : FAILURE in 1h 22m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a49a83c4778849ddb8b1fb9706121259 : SUCCESS in 52m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c68d7fd867e24e2bab9863e55dcd3dd2 : FAILURE in 1h 19m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fdb07415e0684490a9fbb5def2c3fab4 : SUCCESS in 1h 52m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/942196dbcb1d4118a879ecc1d6da1ac9 : FAILURE in 1h 24m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3870d46a5954316bbadaac4706370b1 : SUCCESS in 1h 14m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6f796bf28f4242c98dfc227a2053aede : FAILURE in 50m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/40e0efc248d846b78a2e2ba44f8bb687 : FAILURE in 54m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/35dc2f48f697492bb343c3a39a1eddf0 : SUCCESS in 55m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ee480f7d188f4f8a9b97de5c0af7be6d : FAILURE in 48m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3dfec2a69aad4b82a249b5c46501b419 : SUCCESS in 58m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d09efeeabd94725a3fe100ed9d183bd : SUCCESS in 1h 07m 24s\n\nWarning:\n  Comments left for invalid file neutron/common/utils.py","accounts_in_message":[],"_revision_number":37},{"id":"f7a53029ccc451c471e0e69edda97f450b267fff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 22:40:32.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dabab8bbb2954ff387ce13ff126145a1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4400152ed964d72803d581e665beb6c : SUCCESS in 30m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3fa958d776004b0da875888fc2252fb9 : SUCCESS in 24m 06s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"b1098191bfc5cee9358eef096cfb4b22cb855024","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-25 01:17:44.000000000","message":"Patch Set 37: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/37 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":37},{"id":"63d86c7e001c907ff3a8ad6e2e0e76499aa09b8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-26 21:08:18.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"4981e86c039fb04a865f901d6a3bf57ac18da898","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 21:43:06.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/767508b370ef4a019cf22100de86859b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c821bf5c25ba48feb190c9cc84ec75fd : FAILURE in 32m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/03750b1818734ccb924428c7ee1f12ca : FAILURE in 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"01907c976b27c8cce224ae50ad3c0eb508de8804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 23:29:51.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8125568c7fba4d1286b9e73bc7568df0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f59553876e6d4df8a678f2814977d0c8 : SUCCESS in 2m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cf53718c0c83494ab7669590b23d0408 : FAILURE in 51m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a9a57964f9b94466bfb185bf766cf5a6 : FAILURE in 41m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/30a06aa3af8d47ae8a1d67f07585bef0 : SUCCESS in 2h 08m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0233d20bec574f118403208e7d6b809c : SUCCESS in 35m 44s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee74145a85ca4c7abb80120aa12cc39d : FAILURE in 48m 54s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdc2ee2ee46c4308ab52759c49f45843 : FAILURE in 2h 17m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d1cc64ecce24a70b3f09082367b6fcd : FAILURE in 46m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e24b7dbb3c94ae2b5ff45cf29a43bbd : SUCCESS in 14m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67d178dc92b8448b95310da670d2fa05 : FAILURE in 19m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf02c163b3834021b8ec8ba6936c2971 : FAILURE in 17m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9764a21fd4a744378e778c6b5238672e : SUCCESS in 10m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0060b3ef08cc458b93055f0d00648899 : FAILURE in 55m 15s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c3daa660aebe4cc1ba6a44cb9b9ad60d : SUCCESS in 1h 04m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/771ec15a831b4b79b79a9459fc92e9bd : FAILURE in 1h 10m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/10f2bfc5cb6245d7b4787ba22438f584 : SUCCESS in 1h 50m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f90b21a1ee44865b401578261c05fb5 : FAILURE in 31m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e16ce67c437b4db2ba000e4e0ccd5fc5 : SUCCESS in 1h 04m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/21373fa2567b4a38bd95eb57c94ec168 : FAILURE in 49m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/726cbbb48b0f49f6b1b9252b84e972fc : FAILURE in 46m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fa799765d6c64c489b4889e76a9c7fde : SUCCESS in 58m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/455af81567dd4a52a2c5620976b5cd54 : FAILURE in 1h 23m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f2084626df84388b483a293a0a19fb7 : SUCCESS in 1h 21m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0b12e849379c4a79bf5d844d0ebcefa5 : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":38},{"id":"fdbd88e9fe9ce37aaabcd926f16767608e6fe8dc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-26 23:45:10.000000000","message":"Patch Set 38: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/38 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":38},{"id":"93587a9106eaa51db796bfda30e131ffdc5ae967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-27 07:56:29.000000000","message":"Uploaded patch set 39: Patch Set 38 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"501fec53abbd510b67667f27e1bd70111d8f7a5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 08:37:09.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfd27913713143d7bafcc219e1e923d6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2651a9a9fd6a4b24b0c55a1d10e7b992 : SUCCESS in 38m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/669d3cd9ee1f4c0eb923984537b7d295 : SUCCESS in 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"f3f79dbf7ffb729b804cf7f159564c3388c0eb6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-27 09:17:57.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":40},{"id":"b7f49da24254650b619ac3bb2f2d37e0544fb416","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 09:56:08.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8369b37649d4c7a86a1fe97a9872941\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1689ff8bf1514564956adafd95f12a4c : FAILURE in 25m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f23f06453c24bc597a005832827de05 : FAILURE in 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"ed51515b4bf0f3727671aa2b508d9345f9a093e0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-27 10:29:53.000000000","message":"Patch Set 39: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/39 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":39},{"id":"14723816e7d77af3efb2e6eeeed8f51a0a40f2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 11:42:03.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee893ac74c59480284896d53270810f9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/52bfb785712844b59d8edd6c2525e80c : SUCCESS in 2m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/13fdd3b3c0e043369da882ed209931f7 : FAILURE in 54m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f07b7738392d4d099e37cd10dd9a4035 : FAILURE in 52m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3657dc93c3ac48d99dd346985ec251d5 : SUCCESS in 1h 53m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7cb7c9eafc03491da300d091e89c0c64 : SUCCESS in 36m 19s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79dbd64395de4f87816971bb77ed178a : FAILURE in 50m 32s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c66e947da4f455c926f319e29644fb4 : SUCCESS in 2h 06m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2f9adff4336456484da7e8cc3c5981e : FAILURE in 54m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c06c512e54b4ac584ab641cfdaf31c8 : FAILURE in 9m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b47f39cd78444deb994c2f3afd160af : FAILURE in 23m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b069b80ecd6d488d99702cddf644b97b : FAILURE in 19m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc511a616d44d0ead723e4098f3c2f0 : SUCCESS in 21m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fea99ea911f049199c6ed2a2df39f319 : FAILURE in 43m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c36f93a286814d108297559368960d0c : SUCCESS in 1h 16m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2fc4fa73d16046f68c0d905e27d63446 : FAILURE in 1h 21m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c4489b7b77164cdb9866647c6d9d30e9 : SUCCESS in 2h 07m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/611c82657b574fd6b31ed305fd74e71a : FAILURE in 50m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/46a9d30efcc24392ad96b45259507eaa : SUCCESS in 1h 12m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9aa96c560d66443286fcd376285bc7ac : FAILURE in 1h 11m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99d84b03d1b44c0c99a87e7dc09f1d2f : FAILURE in 1h 04m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ee9be995a311406189938992c1e3ee0f : SUCCESS in 56m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/22beb35648e34d78af0fefb6d6d80a1b : FAILURE in 57m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3688dbad65934df1be04313cbc2c443e : SUCCESS in 1h 34m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/42b1975af7904bf8b8289d12278e9046 : SUCCESS in 1h 15m 49s\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/ovs_oskenapp.py","accounts_in_message":[],"_revision_number":40},{"id":"f0c701c86cd32cf470137353a6f30c2bda298a03","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-27 12:11:01.000000000","message":"Patch Set 40: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/40 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":40},{"id":"1e1e67c2439da30f01078a8eed94674b4153d23c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-28 19:27:02.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":41},{"id":"9cd20dab6f988cac852123aaae6b0eea8aaba793","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 20:07:27.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c65aa6259d624cd0a89e40ba2eeb69b9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c97b03bb639346b889cb61af9be4e7ad : FAILURE in 22m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/048ca65348584d24b7a5de4ad2a3b056 : FAILURE in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"6900797a2f90aab71e96db7bd5aeadac107f1f0b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-28 20:30:33.000000000","message":"Patch Set 41: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/41 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":41},{"id":"c3e28eb0690d67e109ca8debd55eb72ec6907417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 21:56:20.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6c6076031074e3297f348a2d8d3f8b3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/59c91dcdf492415c89b2975432a703bd : SUCCESS in 3m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/40f41fd605a748ceab297a8a83d32722 : FAILURE in 50m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/461c8e21dd7447bf81a25f553a0ace80 : FAILURE in 50m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/85c5ce5d22594f5c935a56883cd6606a : SUCCESS in 2h 23m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/847eed0b2d3a4feebad0228acc9eefcc : SUCCESS in 37m 32s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a331befa13ca4d16a466ee751b31e23b : FAILURE in 39m 57s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55d2f65b697648aba017b9a685098874 : SUCCESS in 1h 57m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88830c1146754812b5a6846bae5d2c05 : FAILURE in 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f6818069c34c16a3e957a564d86771 : SUCCESS in 12m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65fd867b35e1423ea1ae8ecc1aab78b2 : FAILURE in 23m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bbc72415703e4b0cb9a0a53a5c0bac69 : FAILURE in 15m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/674b858e086245ef8b3e2a3b903fe1db : SUCCESS in 10m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b7b405376658454ead6bea090e59dfb4 : FAILURE in 1h 00m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f6a5016cc25648d490bb2198fb90ed68 : SUCCESS in 1h 10m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c109928491e54dc195c5143bf09ee0ee : FAILURE in 1h 24m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7ffc6a124d3841889e669ed9ad11d434 : SUCCESS in 1h 49m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ae70a1fc4b243ed8140d59650708509 : FAILURE in 51m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/07231fa1288e4bae908b445c23161b65 : SUCCESS in 1h 21m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a8853035962e4d62939efc50d106d876 : FAILURE in 1h 08m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/80543bd5d7cf4fee9353518ea9f29d92 : FAILURE in 1h 01m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8f711c10e9d8416489e58bdc29ec2def : SUCCESS in 1h 14m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4fc4d287c81f4376aefa1bf90283492b : FAILURE in 59m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ccf8945bec8442af82fdb49238622f00 : SUCCESS in 1h 18m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9519096c19da4daca47e1a0451ef83d9 : SUCCESS in 1h 31m 47s","accounts_in_message":[],"_revision_number":41},{"id":"185c6246a819cede0e9d3890e9977737b5b65113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-29 17:09:25.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"6711a69a908f23a5b3179441fbcaa28896edf29b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 17:47:30.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6508da829f9546ffa48d759fcd930082\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61144e9cf05f4a90ae1c6be7f4934f5e : FAILURE in 25m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5261b9f98fdc4658bb2c4b30d209e07c : FAILURE in 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"e3054607798165331480ec0ca34971ef375c44fe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-29 17:51:12.000000000","message":"Patch Set 42:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/42 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":42},{"id":"00415ce6e7200edcb4bc091188e7af07de6a4b63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 19:55:01.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85603a97ae4e4726ad34acbbc69cff1c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d7c9d8dc8664049ac9a78a6e6e38606 : SUCCESS in 2m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c4eb396cde634e2b914b884fb08ebb17 : FAILURE in 51m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1292309de68144ad82f5aa9cc5452549 : FAILURE in 51m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2c3302414984e579cc2e9dc457628d1 : SUCCESS in 2h 42m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/40adfa8e96ee4110a0897d573bedfabd : SUCCESS in 35m 58s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32cdbabec33d404cbeeee47ed145d997 : FAILURE in 48m 15s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/186ce36c0d464857bdd3fa9d348766b1 : SUCCESS in 1h 50m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/505085d6fe3e4e2bbc1bb4d6bc97a7ba : FAILURE in 50m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08152dedc7704a6e86c37b25cae633de : SUCCESS in 18m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44360647d535463da46dea5dc30892e9 : FAILURE in 24m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c302c3f893064e1696455c64c450fcba : FAILURE in 24m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d1c6a6550a44c71a5792e7121b60478 : SUCCESS in 20m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c6f7e8881d594f53b7de1eadb1e475eb : FAILURE in 1h 35m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4287c7ffdc844659aba926dcfb3e0871 : SUCCESS in 1h 05m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5e48540a279f43faa5a36607fa6bd510 : FAILURE in 1h 24m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f241aa4ee21e4dc7aed3272bea4838b3 : SUCCESS in 1h 56m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f236e382484e446596d9c75919b3bbe8 : FAILURE in 30m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd59cae3e3054da9999a02815b03c3df : SUCCESS in 1h 11m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c6e0fbc0a2174f6d930ceed377f7e6f2 : FAILURE in 1h 00m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6016be8eb9f249a0b3075da93f89d0fc : FAILURE in 47m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0586c648d0ed4f3b98bed8dff1c4a291 : SUCCESS in 1h 08m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0abb5fcf4c784f3c905e2dc88b9d3fcb : FAILURE in 1h 33m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/59a7877fb31f437fa1085fa7d571082b : SUCCESS in 1h 15m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4531e626b253406980b4a8897d9c34b5 : SUCCESS in 1h 27m 19s","accounts_in_message":[],"_revision_number":42},{"id":"2038cd41b2866b29ebe04e116abb9ac203791899","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-30 10:50:52.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":43},{"id":"d9d52892f573563b08af2b0acfff5ca3cbb0016b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-30 11:31:18.000000000","message":"Patch Set 43:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/43 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":43},{"id":"783531692ad31894f80252a01fb63aef04fee433","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 12:24:43.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c81588b9b664448c95d42c6e61606209\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eeab6f9bfc9945009170dd48d7ddf67c : FAILURE in 26m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0ed3f3470c7843b59d09ebd26b53c1d1 : FAILURE in 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"f88fde50ccf4271b0cccb5f72a5c099fd99604e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 13:11:58.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/053acee04b1149c99efb528b495da3e2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4c54fe7f97804da3b414c18468327991 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f0aab80f94df4bb799c5dfb8f60d4cf4 : FAILURE in 57m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e5cf5f66e783431281e6f5d33d9565c2 : FAILURE in 59m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/edcf5e94f35043179f7ab5bf224011f9 : SUCCESS in 2h 01m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70d310ac5bf74c28bdc587bbabfb5e9a : SUCCESS in 39m 32s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c468f6be550a4e9cb1ecdaa1ca12c2ef : FAILURE in 51m 25s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/152018c48b2a4b7c881357b9b3bdfc01 : SUCCESS in 2h 08m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c34fc50b51c4f2cbbdb3f3822dc3ee8 : FAILURE in 33m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed6856464974fcda1ccefbda6b5db80 : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44adad6dde2742e1b30edc80eeb80bba : FAILURE in 30m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d0d9bad78fb40d399030898930761d3 : FAILURE in 19m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9089d3bb52db4ca4b334fd430b1a19f0 : SUCCESS in 10m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/51a8eaacda2544b4924f3aa21bc82bcd : FAILURE in 56m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2e6c89b0acc9422a87394a82b2679a53 : SUCCESS in 1h 14m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/015d51b16d574881a9e6082cc2fe932e : TIMED_OUT in 2h 13m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9b917989cf3e4ab690035aebb0f7ef79 : SUCCESS in 1h 41m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec325a3f86be486b847355f738b4192f : FAILURE in 56m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9dbd311e7d234e728b241114eed6daac : SUCCESS in 1h 12m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2779623f20c0464782d13632af537d91 : FAILURE in 1h 02m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/101c170761234580bed2cb567e3cb50b : FAILURE in 1h 04m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/26fb84481ad24f14816a4eb55e0e05f3 : SUCCESS in 1h 10m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e3a2d267595c4575baf411a723142f6e : FAILURE in 1h 27m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f9dbc22489d4d708f207afdd2aa4212 : SUCCESS in 1h 27m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/81620b15afff4de8bca35e6c4f73543e : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":43},{"id":"b66167a6b20a0a725ebe0f1ac2b0d3f7a0bb1ebb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-30 15:44:06.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"b3c59f7c5ea885afbe9fd70df8660cebbe3ad5ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 16:21:32.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b0fbe91272436c8f15c5b81d14aded\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e789042b390243f3ad3d37b7201a7a75 : FAILURE in 35m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b20a62ef4387470db776512e371dc5c1 : FAILURE in 22m 35s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"d75f839f735163a07091cd7e4ec97e08f1c3685d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-30 16:25:24.000000000","message":"Patch Set 44:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/44 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":44},{"id":"92a852c1a1d1648cb5c3f3983a1894fb1ae700d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 18:33:02.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b50a1438a60147128f5f52373a96c555\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ef6af61e1f944b959cb03872e72caa4a : SUCCESS in 1m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3f78e470883e4de98c0b28dd616a3aa7 : FAILURE in 1h 15m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1aec3c88ef0e4079b6dd323bae1eae96 : FAILURE in 1h 14m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d1459a1db2794d67946005731ed03fd6 : SUCCESS in 2h 41m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd12c93b7c8746de9e8fb4d6d9d80cb7 : SUCCESS in 50m 29s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30f8aba42c7b4413a64980c2c78eaee2 : FAILURE in 1h 06m 48s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec9b2c1c3eb441e1a3e23df8e63deb6a : FAILURE in 2h 15m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fb50e0b7df5444ba9d0118fca4f60f4 : FAILURE in 32m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319a8a1acc59473abd2373f511d85896 : SUCCESS in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c401b9af64fd4b6e9e9239121078aa7b : FAILURE in 21m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0b5186763d6495fb289d8cfbb213212 : FAILURE in 21m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40878169ba5743018c027bfac2c69264 : SUCCESS in 9m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/27f53867314c403cb459524f856e2760 : FAILURE in 53m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/010f9916e3e14098b696ba9851d134f4 : SUCCESS in 1h 23m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ce1e57ff23fe4a96892807affecd4815 : FAILURE in 22m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f90d8fad64e94f5da044818667643fc8 : SUCCESS in 2h 02m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b14048ddee8c47b3b512ec99309e8c9d : FAILURE in 43m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/96417c7f052d416b808dd75399348a5c : FAILURE in 1h 49m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/be934eface474210bdfce006adb2349f : FAILURE in 55m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eb4cc744470e4f6b8272df1815d236ac : FAILURE in 1h 02m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d8ab3c9b3eef422899b7c8d7294a2035 : SUCCESS in 1h 57m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/58bed87634e44b67a0a00872ebe256cb : FAILURE in 1h 26m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe9a570f8a834127b674c1222f225f71 : SUCCESS in 1h 28m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d5d40f85c1e34e399f38ce46f63df00b : SUCCESS in 1h 23m 55s","accounts_in_message":[],"_revision_number":44},{"id":"339bafab84937971d1865509814d3cc8d0c2c065","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-31 08:26:26.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"cdb6177e47989df55bce76f7ecaf217a5a0e286b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-31 09:07:05.000000000","message":"Patch Set 45:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/45 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":45},{"id":"c2adaf370d50bf01f9896cebc23e72d627a2e38c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 10:20:40.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/971a481276904ddb90761e99aae1e2ca\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de87ede28a3c4583b9024b8b3ff3a333 : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f76d4fdf56f4ff797fa542bda122726 : SUCCESS in 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"4458ca468069d3c701ea205b4f6dc74c99f8908c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 11:48:20.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bc9eb82b951408a98f865032bdc2e4d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/01bea01bbdc94ff3bdb753b3c7c96bcc : SUCCESS in 3m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/599cc51ba1634b30b6afde80ca6b9078 : FAILURE in 47m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c4ade1fdb0841a4af5dd8e9f091c598 : FAILURE in 1h 14m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e12b8708aff74bae8f25577ab836811e : SUCCESS in 1h 55m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/606ed02fd20b428f91fb9b67c46057ed : SUCCESS in 42m 42s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43938afab1e5479eaf65370306c7a313 : FAILURE in 43m 02s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a64aca511d9432b99f2072dcd6d92fe : SUCCESS in 1h 39m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b7e31e3e90143f0b030ac426eb50347 : SUCCESS in 33m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee289bed39854c35a0f4eba8d224338d : SUCCESS in 10m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d751e9d4c9b4400d811db24cd025a6ef : SUCCESS in 21m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6ccd39180a8460abba6746c897406f7 : SUCCESS in 20m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0faf39070aa24405a1245e24034a7abc : SUCCESS in 10m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/18132f1402f64d5b83db432e2e5d2c73 : FAILURE in 55m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f425aeb65edc4f27af335a3c160474c7 : SUCCESS in 1h 08m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/676a8573ba7f42d9b865794a7ea1f487 : FAILURE in 31m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fcfdb2da7d4544bfb786cedfc5b6a077 : SUCCESS in 2h 38m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d15b93826b1444abba1d5a98ca84e26c : FAILURE in 1h 01m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/39868ab19795477f8e28ab5beac83ca8 : SUCCESS in 1h 22m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e3778a2dba4f4b81b86d66be8143b9fd : FAILURE in 2h 21m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/32cf7388fc4d4f98b8aa4dee7e5bfa48 : FAILURE in 1h 03m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bc7d5fa7540d4c1ea1985ff876d4bc51 : SUCCESS in 2h 01m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dbbb23f7002948c98f8b86f09d98b348 : FAILURE in 1h 56m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0560ec5f40b4a0093185f99c0c0c16a : SUCCESS in 1h 17m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c1e8d31a241546b9901045b7618cf70d : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":45},{"id":"d03cfc68e36ffe2da3b6387ed0ef3624ef3053a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-31 13:02:12.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":46},{"id":"6ebeccbabfe463f23f4291eab9e5adca68bcf169","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-31 13:44:37.000000000","message":"Patch Set 46:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/46 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":46},{"id":"b4ef7f3a90f21de9038637403db124eefa09513c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 14:21:31.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/948c7e9f3b894d6ca5820a47d0378847\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/987222a0d7d3490488d1907461277567 : SUCCESS in 28m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/332a11b6713a4b79b4fac69718819bc0 : SUCCESS in 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"2ce6b26bd7c6d1b345003a69fa75e80838d06dc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:03:34.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/caca845183b04ffcbccd01f9dbb67fa3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6cc0f133e92e4788841e9b158ed64925 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c727e0d552ee404fb6346c13b265da95 : FAILURE in 50m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dacf5dbb66374d55aa7e3cbda78bb58d : FAILURE in 54m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1b29d6b442464da4b70fcc284aba74a9 : SUCCESS in 1h 49m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c01e9e4d071b492cb5b64fd034c83415 : SUCCESS in 34m 53s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cbad519b15f469e803002aa864663c4 : FAILURE in 45m 10s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11b7985810314d3eb1675a47d4dc339c : SUCCESS in 1h 42m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/223e7032c2d8493a9d7d5cad9bac4b78 : SUCCESS in 57m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c3987e6e15646e2999e157490bb458a : SUCCESS in 27m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be24f253679447b6b656e52d93854abe : SUCCESS in 40m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd3e8a82e2064db4894fbdaf47a702ed : SUCCESS in 19m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b5090f3becf490a900eec4e15cc853e : SUCCESS in 12m 15s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a96c26ca4ab24838b47f536c67fb7f19 : FAILURE in 47m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/26f9568bc73242d1a88a0720fa488233 : SUCCESS in 1h 08m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1e0928682aa34b73b1dba82b8725304a : FAILURE in 53m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1f22c940dd9941608744cae18fa434be : SUCCESS in 2h 55m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a280135bdf694f0986b7150960cceacf : FAILURE in 1h 17m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/095c665d24c14ef4b47f96b627854796 : SUCCESS in 1h 11m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cb74716217c84cdcb9e734cf0e563447 : FAILURE in 1h 35m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a89becbd6485463bbf8fae3e0e821908 : FAILURE in 1h 03m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5d4ab26f604841c2b63d4b9c5ed0bc05 : SUCCESS in 1h 14m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/eb29169da775464285bf0d705d907860 : FAILURE in 1h 48m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c65e01a46275417fa7dfef2cdd86aad3 : SUCCESS in 1h 53m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0716af244f234c738532cdc3a3f57e8d : TIMED_OUT in 2h 04m 37s","accounts_in_message":[],"_revision_number":46},{"id":"60a408be41e96a054761179f6e3c49f48531059b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-01-31 16:13:37.000000000","message":"Patch Set 47: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":47},{"id":"cfef08498ff05908f5064010117049c55566bc16","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-31 16:35:57.000000000","message":"Patch Set 47:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/65/937765/47 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":47},{"id":"87fac965120fdf7a60cf57fcc577efcac034a2d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:45:53.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5f72b3679145a896c0ed33cff20560\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92bbff0b917d4516b8e8b5e8ee332463 : SUCCESS in 26m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4bbcdc9991642959c9045d5c8cb41e7 : SUCCESS in 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"c043afe09116d5cbf79015a9743afc1e64343e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 18:46:45.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/926bf334f1e64bef9e481a34806b9e8f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/00bd7b2e2df34ae7bc3e5848837c7cb8 : SUCCESS in 2m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cbb5136cc7154e378abe6ccaa3b58d5b : FAILURE in 51m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/792fe14b849e4df49262ddee9a31b25f : FAILURE in 1h 11m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b6b3a67761d646e5a4d50ff73dec13be : SUCCESS in 2h 13m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00650f0cc8b447188eadf0fc2785ec15 : SUCCESS in 46m 51s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67a465cebb8140b39847792c33ce42bb : FAILURE in 1h 03m 38s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/356e527161ea4d779a1b3521c8b39a72 : SUCCESS in 2h 04m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bcfcb0ae55449729223fd3fcf5570cf : SUCCESS in 34m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bc33662aefe4ee597f0e108e1252f98 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbd198e4ffd2405fbaf9b90cd0813169 : SUCCESS in 24m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05468d6dd00647169b16aa8b5828523d : SUCCESS in 25m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae02597fcdee4ebd81c7fc7f5e0b6ac0 : SUCCESS in 9m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/db0edd5998d7446e9766be468d4397d9 : FAILURE in 51m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/12ea956aaa5e4dd580b526b0a254a174 : SUCCESS in 48m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2238a944750940048725deeb56a404e3 : FAILURE in 35m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f5d65a1b5db44ae6a97353b5980d66c8 : SUCCESS in 2h 31m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/04f1264703dd4b278bc31bec1c79bf3d : FAILURE in 38m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/37f24c8bca2b41958e400cbe877e7ef3 : SUCCESS in 1h 19m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b5ef4f30f95a411fa3217838900b2adf : FAILURE in 48m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2cb280e081fd4b44823ea86eb3a2dea7 : FAILURE in 1h 00m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/760836fc39bf496fba9ebaf6197eea9c : SUCCESS in 1h 06m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c4f052bf13142b185361a99e70e8bb0 : FAILURE in 1h 04m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2ede4e061c1244c2a25453aa80617e84 : SUCCESS in 1h 10m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/360ecd7f3b814d37a638f7327041977e : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":47},{"id":"831d25da5fb31b3572be2c9c8cefa07d78f6356f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-04 10:07:07.000000000","message":"Uploaded patch set 48.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":48},{"id":"3b7bade72b63632d9d1eeb20ab5fe419ed95cc18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-04 10:12:28.000000000","message":"Uploaded patch set 49: Patch Set 48 was rebased.","accounts_in_message":[],"_revision_number":49},{"id":"0758f5a76c1ded3b82344cd33b07dc4d437b613d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 10:55:00.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5806f12eceb942cfbe39653304ecc2ef\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1559fa1e0ad4cc0ba136269db72cf2f : SUCCESS in 40m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/85bc939439764b1b904c8eca189b6be8 : SUCCESS in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"f61f71abde93609afc2d6e478a72059ec1fa5cb8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-04 11:43:56.000000000","message":"Patch Set 48:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/48 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":48},{"id":"341ddb3671b0e2e6369749fc42183d6e0ef36438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 12:43:53.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba2d9fd45087458aa83795cc9f07bef3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/60233c1103e84e909087b2bd7f49276f : SUCCESS in 2m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d8d50cb251ff49faaa52709b994b7454 : FAILURE in 1h 09m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5b66c446d8d84a55b41c2e958e16f648 : FAILURE in 1h 30m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de24bb64a854439abec6c9b5cd093986 : SUCCESS in 1h 53m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dcd231a8ba394c699a8d2a182fe73f41 : SUCCESS in 32m 06s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/312a26e84f6b47169a49ab475181ff14 : FAILURE in 59m 08s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69accfccfe7b4bd4880c7b31a836e2c9 : FAILURE in 2h 28m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31bf3f9d0f7249f7a4b598bf8c9f2013 : SUCCESS in 49m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d2efd8074a4e8b85e54f0854ff77cc : SUCCESS in 13m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3159a92d060749c5b47f77141cfcb331 : SUCCESS in 22m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73aeb8384cbe4f48a9e2469c71cc5f82 : SUCCESS in 21m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4903a968c448f7aaf642feb5537d67 : SUCCESS in 9m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a34a82a3a4b14e87b8d8cc071d527bf0 : FAILURE in 57m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f0f4c4ea9df64f0cb59f161f5943cc0d : SUCCESS in 1h 07m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/80422941121746058a394596972e3a30 : FAILURE in 26m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/57ee38e33f75430aaacc365ed3cb32de : FAILURE in 1h 27m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bffd106a80da4688842b639d0582d99a : FAILURE in 1h 52m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/76c053aa735a4373ae15b8db557228fb : SUCCESS in 2h 04m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1a9525502ad94a35866e7cc891cbec7a : FAILURE in 1h 54m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/67050f50ae0b4c12b59b9992f9d84ae5 : FAILURE in 1h 50m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9f8b2ca8bd7e46e5befba657f48458b0 : SUCCESS in 1h 24m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/af44d4f490a942dfb2af4a0ef6c54b36 : FAILURE in 1h 21m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ae330ba1975c4b91ae62310ad9da6cb7 : SUCCESS in 1h 19m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f6a0d97ec906405088be161858212706 : SUCCESS in 1h 50m 53s","accounts_in_message":[],"_revision_number":49},{"id":"a43dcb63635aa755db0abc07495f67de69ac388b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-04 13:50:55.000000000","message":"Patch Set 49:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/49 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":49},{"id":"a0d07c7a06e727f53638c3cd1d112145c1722ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-05 09:21:08.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":50},{"id":"866b6046f445e0c1094f1b3ba7a36f40bfac6483","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:02:15.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8f394542d2348afb92f3d6d5b312df8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33913111602649438ee58ce4f0aa0d0d : SUCCESS in 38m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9061327e4d384fc8975cecc3ff43ba2c : SUCCESS in 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"9f150a9685a628563ddd7b072bb4f640a5942b9b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-05 11:07:32.000000000","message":"Patch Set 50:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/50 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":50},{"id":"b3d380c8b6022fec6b0ba7efa4f5cd20efdaf7af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 11:26:02.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f46440ce1bb541be96e659715b8ea403\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/45922cb0c4b544c7a48123f2179b7ac8 : SUCCESS in 2m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/60b965dce47545478b1d840fe0cd8ac2 : FAILURE in 52m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ab8081ce3644703b37a0284a13de84f : FAILURE in 54m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/49dd4d33efdb4589ab7fa576af8bde91 : SUCCESS in 1h 57m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/583a70d480314b30a5f901f199c2ea02 : SUCCESS in 34m 15s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b358b713c6244ff2ada2ac37137612d8 : FAILURE in 45m 09s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/960a467ce04d4fc29764b6a532cbfbab : SUCCESS in 1h 52m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf8c55e2fe0d4cfb9d3e079d985eafaa : SUCCESS in 48m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90bae66a8a0e49b3b51bc1a51ad7b48b : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c74a896481f4a078253d64fd27c7d6f : SUCCESS in 25m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/123aa74784e049008d51e590d3dfcc6a : SUCCESS in 21m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55439638739d4e72b906e4506def5fa6 : SUCCESS in 10m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/28c04e9da2d8496f8060b4d3c01a3ade : FAILURE in 53m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/99164e2bec8f4f33a49c018a820d7ddf : SUCCESS in 1h 07m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/294941d3f7e9401d80e779fd95e2e2ea : FAILURE in 32m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/87b5100b10ac4e54aa2c300eff8af4b2 : FAILURE in 1h 49m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c62d202b6de447a2932bcb439613f804 : FAILURE in 50m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b0852e157364b8eb39961c5d012abb7 : SUCCESS in 1h 37m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e9796155147e4284b37503b4d4afc696 : FAILURE in 59m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7233d57a3f27407aa35600c22a0b8468 : FAILURE in 52m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6e9b095359e7456d902d0a9312a1461e : SUCCESS in 1h 09m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ca2af16228344f1497038308c3a12d15 : FAILURE in 54m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/526ef4b8ae4f4c0380497bd9100c0edb : SUCCESS in 1h 32m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/05af5212f3aa4dd88cd09eb8052e9ee1 : SUCCESS in 1h 20m 52s","accounts_in_message":[],"_revision_number":50},{"id":"97a0cc0aafc422e222810d431003334299798444","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-06 10:23:02.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":51},{"id":"963564f9e123e18e4568a1b212b4263dc091d6b3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-06 11:04:57.000000000","message":"Patch Set 51:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/51 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":51},{"id":"89366d6a74ee8a77e45b4216b4e44dbf789d0ed3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 11:33:38.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d0a239c8764dd59d96105b1386c6b6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74df707c29bd44ecb1dc8cadd1fcab91 : SUCCESS in 42m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7dd637d2a58540b980a46fc0cd31cc5e : SUCCESS in 25m 06s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"7bfb53d16dca0108b3f3fc767eba64394718a8ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 12:27:54.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c2b21f0e7134c558f7fb0d972ef28e7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b75d6a7a79c4acbb39186342e2a5645 : SUCCESS in 6m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/afd8327e4aab454a8ca3631f567ade5f : FAILURE in 1h 04m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e355e8033f44548bcc4c401c7844596 : FAILURE in 1h 11m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e5421362340c4e68a76cc0389d605b26 : SUCCESS in 2h 02m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/910f886db04a45e2b6eb8089d64ce1e0 : SUCCESS in 35m 44s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e887f8480974241b25174630d2f5e5c : FAILURE in 46m 44s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4c106f65fb54dcf9b200519d5b53c33 : SUCCESS in 1h 47m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08a7fb4b41b84fafa1b3d52f0f23a208 : SUCCESS in 46m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6001d70575d74e16bb60c1189b414153 : SUCCESS in 11m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c42806d04a774471b5921e621da252e4 : SUCCESS in 21m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4319f2b583a4240b4f18e1984c63c29 : SUCCESS in 20m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51feda011d14ace8065453ff9fccee2 : SUCCESS in 9m 30s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cac355399f1d48748bc3aa9143e8ee53 : FAILURE in 36m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/42047ccbc79e444fbe10b614f32311fa : SUCCESS in 1h 07m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ed1f16fb996a4eb78a0e1ca79772f3f2 : FAILURE in 28m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a5aef5af144d4672b2840c681ebf413f : SUCCESS in 1h 27m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4eae5ba93f964135a15e9a3ebc1d7914 : FAILURE in 54m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5b66014abcc48588b1f24fc733cb1fa : SUCCESS in 1h 34m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c4e41b007a544304aa0c1cef3e201e5e : FAILURE in 49m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2b852cb12fa94389bd4e538ef7bc1692 : FAILURE in 1h 01m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dab8c588d4da46d5aac1687f4442a5f0 : FAILURE in 58m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/21c1f7aafc5c44e9b1dfb0a2165afbc8 : FAILURE in 38m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/509b0abee8a545d8b0f658d8765d561b : SUCCESS in 1h 14m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d32677b50d2d42fea6c77169c9e62b98 : SUCCESS in 1h 06m 37s","accounts_in_message":[],"_revision_number":51},{"id":"d257d62bad33d92f4467e359b0fb38832675c1c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-06 14:34:17.000000000","message":"Uploaded patch set 52.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":52},{"id":"dbccb758c6f6b3fbc1eb58f06cb5c72708596dc4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-02-06 14:34:41.000000000","message":"Patch Set 52: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  while scanning a plain scalar\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 20, column 11\n  found a tab character that violates indentation\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 21, column 1","accounts_in_message":[],"_revision_number":52},{"id":"e102638255be4a90484ec88a7b6acf27a65e46e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 14:34:56.000000000","message":"Patch Set 52:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  while scanning a plain scalar\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 20, column 11\n  found a tab character that violates indentation\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 21, column 1","accounts_in_message":[],"_revision_number":52},{"id":"ca090e337f0efd92b5409b2db5a8232bbcfcfcd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 14:35:00.000000000","message":"Patch Set 52: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  while scanning a plain scalar\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 20, column 11\n  found a tab character that violates indentation\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 21, column 1","accounts_in_message":[],"_revision_number":52},{"id":"75d24ccdaf12015064059b1323c646969763beb2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-06 14:55:07.000000000","message":"Patch Set 52:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/52 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":52},{"id":"641c93f1b90a71385be3077497e15ebdf8728560","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-06 16:13:19.000000000","message":"Uploaded patch set 53.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":53},{"id":"193b14a072aeb5629a1ffb4e439846d1439c5279","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-06 17:59:29.000000000","message":"Patch Set 53:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/53 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":53},{"id":"40cbc2511e617972bf0fdad8f309c5a5f563e528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 18:19:42.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c62b85dd8c14c4984e03be1c280a7df\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/29ab6eee7a3744b08019b7aa268f1727 : SUCCESS in 2m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7df9086d4a66488b9389e86ba6505b3b : FAILURE in 1h 20m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1bf0378104434c13b3e9ffd4dd1819ba : FAILURE in 52m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9963d06f3bb840cab8f583ec75572745 : SUCCESS in 1h 57m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82fc4e36929847debcc4381162a0b105 : SUCCESS in 36m 53s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04ed89a735d7424fa13e0668dbcad70d : FAILURE in 56m 17s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c1c013cb2844481bf1cd1a2d6b66910 : SUCCESS in 2h 01m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba3dce4b518c4981a6fb8c5247482089 : SUCCESS in 53m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e6da40b7e24192acd04875e829e495 : SUCCESS in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fb48d7273aa4dfe84a63ec060723a07 : SUCCESS in 20m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7f81ecfbe7a4b87b1796f6d6e8b91ed : SUCCESS in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ac46dd466649eaac3c59d5f88070b6 : SUCCESS in 10m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/19f2bf9b240d4a76bfdedc4d871a47e6 : FAILURE in 54m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a3e0d64243d84dc58df9287f602f492f : SUCCESS in 1h 17m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/47b6be4a5e7f470fa6fc54165efa6fa9 : FAILURE in 37m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/28c63a19cbf44514aec64a4ebca6ad45 : SUCCESS in 1h 59m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/89456fb411ef448c8161f7006621a5d5 : FAILURE in 1h 01m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/795bc2bfef9341de933dbbebe7114469 : SUCCESS in 1h 19m 18s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/0538d30960b8465e9cf7b0bb2f05d60f : FAILURE in 1h 15m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aa3ed83d2f364090ae9200e91b62ff1a : FAILURE in 1h 00m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c12f1363a9e84dbaa6c06ea4903fae0a : FAILURE in 1h 00m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/30a645c32a574415b6fc973c66c60fe5 : SUCCESS in 1h 07m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/056e9e23909e4f029309550a00a8e753 : FAILURE in 1h 01m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/504c402e6ac54aeab4a6b3796562fa0a : SUCCESS in 1h 14m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bcc7c3ee7ef549229279d56cbaa9dc2d : SUCCESS in 1h 28m 14s","accounts_in_message":[],"_revision_number":53},{"id":"44d089e5fc1e05f2002a7d64f1afc61f121fe097","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 18:45:54.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/152b62c369074519912405d25e8faa33\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e98eee213d9f437598fbd4f69f6fd846 : SUCCESS in 47m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4af67de592e64bf2840c48bc35d324b0 : SUCCESS in 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"16adbe497e22ddf7351ea94aaceb85da28fb8283","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-07 12:24:18.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"18719e4dfe00569df8fe69f4a381a4a5d8d16bd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 12:57:59.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc7a526d85a4218bc039984d022bfd3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec98a2db47714640a1474a3704a1051e : SUCCESS in 26m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/314481b7e67a4b66ad75ffa11459b870 : SUCCESS in 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"2e6cc4873dd90d5c6dc68f01b5b652774f43a7c9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-07 13:26:37.000000000","message":"Patch Set 53:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/53 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":53},{"id":"336d02e65e9090267fa906042f4179f7b421968e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 15:27:16.000000000","message":"Patch Set 53:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cda5c7ce6c4439ba0ccc2c6882e4c3f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/371de2e333c34100b5e85bc1968f722d : SUCCESS in 2m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9afac66e6b95413783db8cdf5d22b195 : FAILURE in 50m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe7144ecf7fd4e2d8ccb1edc611ddf1a : FAILURE in 50m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c38c26c3d4e047c7a5cd4decdd644cf5 : SUCCESS in 3h 00m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2baed4c5f60a4611a8c2f5ec66bb8ce7 : SUCCESS in 33m 48s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8b57ae7c7044ca694932f906bd2a8b4 : FAILURE in 46m 58s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/328d0aecbd904bcebcf937b2f475b9de : SUCCESS in 1h 32m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2eea22a42ec432bb16b6835de980963 : SUCCESS in 37m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f233748097ad4c7194ee11dd34847910 : SUCCESS in 14m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edffa9b82e924b2f9d0170840e987c9b : SUCCESS in 21m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1944119829394324910303fa2dcc912a : SUCCESS in 19m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d982324e252640c887e8e29c0553c443 : SUCCESS in 9m 55s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ebbdc00e12a445a5bbfbef0ac1c103cc : FAILURE in 30m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c3e0f4e7dc324236839583e5e97930b1 : SUCCESS in 1h 15m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0d36e80b2e4440ea926025aa97213a43 : FAILURE in 41m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/95cb57f3eff6481c855c4046d9005ab8 : SUCCESS in 2h 33m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/31513506e90f4f9bbd2a8e0061b50009 : FAILURE in 57m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eed1e0b070ba419895244bcad1641894 : SUCCESS in 1h 21m 05s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/951cd292c91c455b90679b72b3939ef2 : FAILURE in 1h 03m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5d6d5a8a47fd414f9d7d53ec795db750 : FAILURE in 52m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/027a849675e3447a9dfff3f1f1a20727 : FAILURE in 50m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6da786d5a3e54f62bd4fcbd7ad605c03 : SUCCESS in 1h 09m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a33f617cb9b14c078a0ffd2b49d02b01 : FAILURE in 1h 09m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f563142694cc408287f3dfd4e2dd4f69 : SUCCESS in 1h 20m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7a35cd16056f45f8b1970ece5e60a58c : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":53},{"id":"21e37ef98d623b4b14d11a148bed736cee514ff6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-07 16:03:51.000000000","message":"Uploaded patch set 54.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":54},{"id":"5f7f8fc6776e975c6d4929b300b12453503bda6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 16:55:39.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f5f5ec155143dd8c7dd50dcf2bf9e7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44808a7ae515449ba08e7484b65992f4 : SUCCESS in 26m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/419fb3b2d5ec4c49a7fb8b8a6e08c416 : SUCCESS in 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"7602fa1b302bb4c9aba0ae2ae4ce557308380193","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-07 18:19:18.000000000","message":"Patch Set 54:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/54 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":54},{"id":"cf9e711a4ae66cb1f2ceb610e498c1cdff4cbc8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 19:16:02.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e284ad0b5aed40a186729f306ba600c2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/24f5bba4ed274a739702a03934ece419 : SUCCESS in 3m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/43c7b28d60134564804c3726ed81ec9e : FAILURE in 30m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36661926ecb64126afd60d7a7b4ab624 : FAILURE in 30m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b9144593be8f4c4da0cac0ede6fefffa : SUCCESS in 2h 08m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3bd906bc7b4743529998c46e299c1cc3 : SUCCESS in 36m 32s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94488d3348ff48ccb4bc59f9edfd64f7 : FAILURE in 26m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b0550c42f4b4ba8abcf6e04be919cb0 : SUCCESS in 2h 09m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/651d8ae69eee4029a6c7631bc4e5c549 : SUCCESS in 53m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fbd9f62d15d4a9aab429bbca1b5c0f4 : SUCCESS in 18m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b22444b4042a422cbb79db123edd457c : SUCCESS in 27m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64f91607da974b4aabfc8dc8224708ee : SUCCESS in 26m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4af5f7d2bb44a8e94a0933f65fcb30c : SUCCESS in 20m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/46ee9d6dbf7c41d1bf7b2a5a2cdc2424 : FAILURE in 1h 19m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d46475037bdd47f29b069e7a92e52639 : SUCCESS in 1h 33m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/791d124bfc62497ca8d9e1c507a72102 : FAILURE in 52m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a2d5c7f5c0384d61b3324bab9cbdb38c : TIMED_OUT in 3h 02m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/71597c3e526f4d38b4cf1aefea6084f6 : FAILURE in 1h 02m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5aafc0cb09db43b5b8ffb28326c2eb25 : SUCCESS in 1h 26m 43s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/34eb3e7b60c24ee6aaebca2dca4312d0 : FAILURE in 28m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/93f25169fd16401f8d35ac051dae61e3 : FAILURE in 29m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/965c43eed12946968294eee0fd0a06e2 : FAILURE in 23m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ece5102b8fca4ec887da43f93e1f9afb : SUCCESS in 57m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f6470fa1bc844415a3b04e47281a4e8f : FAILURE in 17m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/509e96a0ef7345eba55b8efb77eab2cd : SUCCESS in 1h 18m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5407f2e703d34dfe98cd4f32da153fb7 : SUCCESS in 1h 40m 58s","accounts_in_message":[],"_revision_number":54},{"id":"fbbf585602f556b05f27c824aa7eb20f88e0ed2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-10 09:17:56.000000000","message":"Uploaded patch set 55: New patch set was added with same tree, parent tree, and commit message as Patch Set 54.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":55},{"id":"7636017e6494772d35222658df5ec167a1a65296","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 09:58:05.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/755bc75838d4433a98a632495609f4d2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c70c4144d6649b3a0ccaffc3ca70170 : SUCCESS in 27m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff5a5d26ce154a80ba92f88ee3c2c076 : SUCCESS in 37m 20s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"159f9fd8c7217d7a34c98b9e56af4be9273e896a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-10 10:00:52.000000000","message":"Patch Set 55:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/55 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":55},{"id":"bcc4b968f7f10dc291d9926eede67c152c94593b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 11:57:46.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45aee88c016c4015a50008944c6e0577\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/42b13443ab0f4e7bafa103f891f65f1c : SUCCESS in 2m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/04685d49ee6c4ee29e51a733ae40768c : FAILURE in 28m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98a16e00c3f34c92bc950821dc86ae31 : FAILURE in 28m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/784984bd16e94a0ebc92a0d89e2d6f1d : SUCCESS in 1h 57m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c01e811904d46bd8deeb99dbe4d6cba : SUCCESS in 34m 53s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7461d25865a14525ba171a9cd1e8e026 : FAILURE in 22m 55s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf32cae3eadb45188967c1528f61ffbd : SUCCESS in 2h 06m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc3d9d5f3cc9453d8c4ec969cb10a65e : SUCCESS in 52m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8654f3b9068f49269aaddea23739a57a : SUCCESS in 10m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7b1aa36af52496b90997b0ac9fab007 : SUCCESS in 19m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7985720472f48d4a2cd4e6cfee80682 : SUCCESS in 19m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac8cfed53024af2845de0d674e49fec : SUCCESS in 8m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e70b1152c2e54f64b520f7602455ae0f : FAILURE in 54m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e6f06f34675847fba218d7d09c5bfed8 : SUCCESS in 58m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f57b9d7341de457687ba139d6b382e12 : FAILURE in 41m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/faf01f8cc00f46738e39389eb1a5d94e : FAILURE in 2h 28m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/70ae8ae859e64f7a80f472534d1095b9 : FAILURE in 51m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c709d42db2924359824831910506988d : SUCCESS in 59m 14s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/6542ebd7c8cf48dcaf5227d675157d15 : FAILURE in 20m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/501368f6ef9f48ec988a2c684366f643 : FAILURE in 38m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f7963921785c4a8e953aee25d0bbad36 : FAILURE in 17m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a3931b4c536d436cb26895aeb9981225 : SUCCESS in 1h 02m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/331b4ab6ad2b4dd9b22cf0f26621ee40 : FAILURE in 29m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/577013e864194ee0869e9a8e8b41700b : SUCCESS in 1h 25m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/be34ba97da984fbabfebb544214d2027 : SUCCESS in 1h 20m 32s","accounts_in_message":[],"_revision_number":55},{"id":"3d90a8114dd27e85420dcf4d393686b8f80c5817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-10 13:00:27.000000000","message":"Uploaded patch set 56: Patch Set 55 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":56},{"id":"be706a82ab00ea2632d53b1a63e4a84f89fad1e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 13:37:47.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aab6ca255c474afd997f9a517c52f8c1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5f9713c578d47a7b5f6cf5f1813c947 : SUCCESS in 34m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/70b4ff16b5d940d6ab9f08adabcd445b : SUCCESS in 34m 10s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"a53e448e8a62864086e901ca317c5680dc87bde7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-10 14:12:15.000000000","message":"Patch Set 56:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/56 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":56},{"id":"4fb7111e408acc03c2cb495e228c3082680dca78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 16:16:43.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aae90bfe11304b67a7b83a14b3ec452a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a899fd4402a4e9598b78009e077a493 : SUCCESS in 2m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c5c984f84eca470e975928d3e466aa6f : FAILURE in 31m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e839686128149d290b28c0a68a5dea3 : FAILURE in 32m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5cecbdee7c3647a5b80fc49cfd91423a : SUCCESS in 3h 02m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/091119d779c34e15b996999915e8643d : SUCCESS in 33m 56s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2ec2d85911c4858b55f40b67d31fe57 : FAILURE in 32m 32s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82edc961974147dc803ffe8a88a1bf5e : SUCCESS in 1h 58m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c44d02372704fb480138df7adda701f : SUCCESS in 51m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e0472be7fe64bf6837ccb96f325a170 : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aab3a17dec640c9b34e8a35d2b7d0d7 : SUCCESS in 21m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9e9d139081f40e3b40dfcaf517fd889 : SUCCESS in 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62730a3fc78f4de88dd2d7e870536846 : SUCCESS in 10m 59s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/294a834de70649b38e7677a3cf62ba75 : FAILURE in 49m 18s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fec27620b3d7481ebb1c5fbe79aa9aa5 : SUCCESS in 1h 08m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2f3f0460d4234b9b8556f02c80be6f76 : SUCCESS in 1h 07m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/97e0a0ae351a4ec28e7ded07f3c07de8 : FAILURE in 3h 00m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4bfc84fe2ae481dad6d2bb3204bac3d : FAILURE in 1h 08m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdc74745444a492ab6dc7e1d186a3e9c : SUCCESS in 1h 18m 42s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/9fb18565d36347529f7d38e127d70c25 : FAILURE in 29m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fc637ce938814f0a9bf45266db1777be : FAILURE in 34m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1818420850144f1d96cd5c0fd901e045 : FAILURE in 30m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d2093fdc86d740739b0458332acb52a7 : SUCCESS in 1h 07m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ba22a288a5bd49a59319985f96e92146 : FAILURE in 33m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b3b49cbf457444aa2460082a785e4be : FAILURE in 1h 23m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/335ac6dd130c44e2b916bb0c38839d1d : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":56},{"id":"e61ef14e33d589b93b4e833638c619d923de788a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-10 19:30:17.000000000","message":"Uploaded patch set 57.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":57},{"id":"7ef217785d5e32e0ed7be2a7cad980ae657deb9f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-10 20:09:53.000000000","message":"Patch Set 57:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/57 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":57},{"id":"78f251911419b2b34b6d05b559a80d066e6e3fce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 21:25:16.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e07501bdd83f4db0aa1707142ff1d569\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6d6acfcca2d42d7bcf49597cbbf0242 : SUCCESS in 32m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2473f4b584943e1867a7b6778d963f9 : SUCCESS in 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"2fc9cd21b63f14059c2cf7d58b8a499bbc8e34ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 21:38:03.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8d584896efd42a383bd158eacfffc6e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1ad9f0855ba04adaa54e3931640d9305 : SUCCESS in 3m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/54f447599ef64324be2828a1f7de3480 : FAILURE in 56m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1bddf5be5f004a15a2440422d4013a26 : FAILURE in 54m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53d3f254e17a46d1b8a4149e945f8073 : SUCCESS in 2h 04m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/75a11ac57aff4d42b928c8f6bc76e3da : SUCCESS in 38m 29s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c1c8924b03046fe9c46088db7fb2fce : FAILURE in 54m 23s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9190d54975844ae79a7b9bbc93c6db3a : SUCCESS in 1h 47m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/990d2f1b928845a0a7690e4ba0230155 : SUCCESS in 31m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2589b1a404ac483db5aa9f979be84383 : SUCCESS in 20m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d34c17b6c3c042ba9a9e762eb17d3c6e : SUCCESS in 21m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7392be018e524a709f2624ffbebabf9f : SUCCESS in 17m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9baa614482664d1dbc3eae003777caa9 : SUCCESS in 10m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9fc5bc81dd6c42bcadbefb1876370215 : FAILURE in 1h 00m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/52d292c22cac4dd28c655c8a011c176c : SUCCESS in 1h 16m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8e41c8c71c174e3bbd2c0c996f3808ca : SUCCESS in 49m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3f3b8856e13746a0a4160868e8c251e6 : SUCCESS in 1h 38m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e880ba1816949808624833b68749b87 : FAILURE in 42m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c66ef310350b4746a2bf877aebf1ba93 : SUCCESS in 58m 12s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/d4dd61f8297f4c38af0fe7e58a4eadd8 : FAILURE in 53m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a20e8e1076c9498386422c65a1402d6a : FAILURE in 1h 04m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/709813a3ec70499eacbe9e534672ae7b : FAILURE in 58m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f7ec1ee42fab484c9fba9f75e00f7def : SUCCESS in 1h 01m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/08284af093c74367af1243a664b894c7 : FAILURE in 1h 01m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/539942acbe8a4dfaa7d60f919782a4ec : SUCCESS in 1h 06m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/475ffe3ee8bc48e2bdc44c3957406637 : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":57},{"id":"abf1bf2169c7f61bcc498a93236429a73f6540dd","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-11 07:25:02.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"434fdf9a951a2d19ce3ee7870f26423f76f98485","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 07:58:02.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5970cba803a845179cb448a21f9bd6b4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/259f7a76d28e405ca265081fd3b07049 : SUCCESS in 30m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b65acee369ce423db06edabbd38ba605 : SUCCESS in 26m 56s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"bde804353f426f8d80d1720508b4baf0e01dde27","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-11 07:59:14.000000000","message":"Patch Set 57:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/57 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":57},{"id":"c98aacfb26cb32ab42c72efd9efd118d2b2f5c90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 09:41:09.000000000","message":"Patch Set 57:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7751f90b9de547bdab56c08d3cdf1123\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bba870e5857748d7ac7f86c2bb19f587 : SUCCESS in 3m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2e3aa53214d5440487475911e3f2856e : FAILURE in 1h 06m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a79138e0cad41e5977a0d54a5638a5c : FAILURE in 53m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4b525a08492841e4982e0ce0814768b3 : SUCCESS in 2h 14m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/60a86d4ea74b49aea41181c1d3758ecd : SUCCESS in 29m 39s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d9d5a7b2f534afcb7a42ec0f15d60fd : FAILURE in 1h 03m 47s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/097eac3e4c4044d4b097c857a1a0f90b : SUCCESS in 1h 38m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0a31f482c8d458099a34dcf3cf85ddb : SUCCESS in 1h 00m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728fe6802fe84e748af4ace17ca90c40 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0435d99bbf63483b9231879f8f067c4a : SUCCESS in 19m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa67d0e1d6424865b22c5c30791a3e8b : SUCCESS in 17m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b31a9135a73e4596ad05fdf83fdb42e1 : SUCCESS in 9m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/99ed1c026d4344beaea661077e8b8cbf : FAILURE in 31m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d0ab6a7f741d47f5bb4351e21c599374 : SUCCESS in 1h 10m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/047b4ebb9e91417c95e17e6c45de3606 : SUCCESS in 1h 01m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d0473374f4fb43458fd60b59690c7194 : SUCCESS in 1h 43m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2f67200d98c43398a29b9b05c92fdee : FAILURE in 39m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7402881851d54c14b0ff753f36d2ce98 : SUCCESS in 1h 12m 13s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/d61a9d723150406ba39bc12c67b7d2b6 : FAILURE in 45m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/12c8480ddac44770af37d7de910aa998 : FAILURE in 1h 06m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5e8de0cc78bc454196255d90bebbe8bb : FAILURE in 51m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/37c85b9e0b574bf299d477889ec9a1c9 : SUCCESS in 59m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1b1f2cd878344c42808cf71601c8535b : FAILURE in 1h 10m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b33477fba7246cfb5213cdf96e8f438 : SUCCESS in 45m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/25176baf51c84230b34cc3221d03d21c : SUCCESS in 1h 17m 55s","accounts_in_message":[],"_revision_number":57},{"id":"aca54041bc5627795abbe6b359e294fa7014ed20","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-11 10:38:35.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"4bdb5e475ed8bcc0c7dec986e743c64481a2b659","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:16:28.000000000","message":"Patch Set 57:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/938337 is needed.","accounts_in_message":[],"_revision_number":57},{"id":"e76e52ea872bf309df224d5213692cd4a96bacbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:16:36.000000000","message":"Patch Set 57:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/938337 is needed.","accounts_in_message":[],"_revision_number":57},{"id":"135ef2e4d30bb7cb2752a79cf04f0d8bc91471d9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-11 11:19:35.000000000","message":"Patch Set 57:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/57 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":57},{"id":"f8505231d40249b75ffa3c6619d51124dc8bebe3","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-11 13:39:26.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"ca6b374471f1c5df29ece1f963eb40704d51913a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-11 14:00:36.000000000","message":"Patch Set 57:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/57 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":57},{"id":"147e0a2df85624050c3f699870fc11a6edc97b7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 15:31:30.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/540ee9cd53a44423914a2e3755f32340\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4076ebb530b4ac78c251df44b7a640c : SUCCESS in 38m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/462c154f1d0a40f5993ee8bfb14cf5cd : SUCCESS in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"ce0ae987df30e43af8c7bced58956053eb9d197c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 15:49:22.000000000","message":"Patch Set 57:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e79877a66f4848a4adbf1126eca8c947\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1feec8dd0acf418da4eb01d7943d57c2 : SUCCESS in 3m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e29510197676405a93cf7cc4dfd3b106 : FAILURE in 56m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/df8b319559bc415ca3df6a48af19c69f : FAILURE in 54m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e01c8b7b18ab4da38c107b02c43e8d45 : SUCCESS in 1h 43m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57118be287c54d7eb9aeabf867f56601 : SUCCESS in 28m 02s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30a0e16879304bed80c143e6a0a1ad2e : FAILURE in 39m 54s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04f569fca1674f37a62150c9d611b768 : SUCCESS in 1h 42m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/434283080f99441e9c0e98b38bafbb65 : SUCCESS in 51m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e681c879912143b4ad012b6a3765f2cb : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87839d2bc29a4240867868e5df03f287 : SUCCESS in 18m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6181b45d236147f588b229ec9a92e1bc : SUCCESS in 18m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae618fff345841f18dc98adb49bb281c : SUCCESS in 11m 00s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e579f21e89d64d208739849cca98f013 : FAILURE in 59m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c67c7596820d4c90b614d5e24a95c925 : SUCCESS in 1h 07m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/df1a6ae7566a4644b477dcdb1cb408c0 : SUCCESS in 1h 29m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d6005ae44869447198d6c998930dd899 : SUCCESS in 2h 03m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/229a2c4639224e3bbc202cde7e8ad545 : FAILURE in 58m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6da657c8f64f40e7b313a2b2c8a32fbe : SUCCESS in 1h 11m 25s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/87e4ca9d237f421993b5b50e7b346f22 : FAILURE in 1h 15m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f10e0cdf90c943cca8a9e04ba0be2fbf : FAILURE in 1h 08m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e9ed84eaa354a3d9b33d1b33eacfa60 : FAILURE in 1h 06m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/883b5df99c5e41f9af5eb191584f9986 : SUCCESS in 1h 11m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fb697668ec6843e180ca9a71cb87e9bc : FAILURE in 1h 18m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e381ea38aebe40e1af36f761124bcc35 : SUCCESS in 1h 30m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f6c01e08650844968f9ae389bc865fbe : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":57},{"id":"e0b0ea1c8e6cb708ef11ab3a03e8f146328eb8f6","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-11 16:19:59.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"6844402cb7ddf683f352c132af8e0e1e0827d442","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-11 16:41:04.000000000","message":"Patch Set 57:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/57 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":57},{"id":"ef51f9bdaefaab201114e555c2ed07b2540162f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 16:53:51.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/addb4fe4fdcc4b48be4f58fc54876cc3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65f6b1450b2447088854cff3ada21762 : SUCCESS in 31m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f6d837221baa470c9f0ad92f39972cd1 : SUCCESS in 30m 37s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"5fc642ac12d3b5311043e6fa1ad6b65fdd915ed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 19:24:25.000000000","message":"Patch Set 57:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0f9dde2d3254f08ad9982dfb5fcc1bc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/724c4ed9762442aa9499e3d1893d3050 : SUCCESS in 2m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/982383f31b524955a845ad4c21426926 : FAILURE in 1h 07m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/746942991c2b4781be320875d2988c37 : FAILURE in 1h 09m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4d313611cbf74b8fb6d8be914ac2b581 : SUCCESS in 2h 56m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f3676325a80d4ded8087f2e3ec1d7123 : SUCCESS in 43m 23s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a26a8cd632c9470687d8c647f062fb69 : FAILURE in 1h 01m 09s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4f9a2e715f34bc3b99f3230876c19a3 : SUCCESS in 2h 15m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d7ff50ddca847358529a5e5eaab08b7 : SUCCESS in 57m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd80f66f8d2946d9b7dff12f69fd1c65 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a2d412b23bb4852bc7897a38698017e : SUCCESS in 24m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4c2b33150f743dcac57ddaade16d639 : SUCCESS in 17m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bdef242d76e49a7bb459cfde7673996 : SUCCESS in 10m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b94500b802074363a8c68ee5a1e2b0b3 : FAILURE in 29m 41s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/afedb715f0ef44a3a1901ed73fbee4da : SUCCESS in 1h 13m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3dc0962a83ce4f74bed328c78866c365 : SUCCESS in 47m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b2f6e87b05614e1183fc006af5e03507 : SUCCESS in 2h 23m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9dccef1199674150b4c3c9a54a5a0bb5 : FAILURE in 58m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7170512337784bca86cd631f0a456ef1 : SUCCESS in 46m 43s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/8b35d5e78b1c4af8ab7ba6c457da39f8 : FAILURE in 48m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/74625adee7294cb5895dab7de62c12a9 : FAILURE in 43m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ff69485cae754e6db94604721c078baa : FAILURE in 51m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ef4914833bfb410d8dd35f80b9e32f84 : SUCCESS in 53m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6feea2665fe645cd8df7463549aa9e27 : FAILURE in 1h 05m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/216acc26612843248618ebd8001a47ff : SUCCESS in 1h 22m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9e979c1cd4e84cee8be3008e9f7ed344 : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":57},{"id":"56f86c940af0803ff1cce5d28791d0f3ed3a2577","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-12 09:17:04.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"c4bc704fe8e169a1cc47eca202682af901d9c8b0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-12 09:38:34.000000000","message":"Patch Set 57:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/57 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":57},{"id":"ee4b82592e41431095361ddbc5144e710ac1a478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 09:56:33.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c632457b2c42c28bb394b1acdf957f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ad3f10f8b084be6897b52a7a3d9a0be : SUCCESS in 25m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44fefb34790142cab09be78017befe64 : SUCCESS in 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"e46177b7e04239c1e570c7929fb0ed1efb396315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 11:42:43.000000000","message":"Patch Set 57:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f135c5e5b2374b11ad79fba295c47b2d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a3f9d58d8a2a47b380e58079bb7e6319 : SUCCESS in 3m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/41ac599acb194e809c2f44b131d22008 : FAILURE in 50m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fbc44b5081e647a09d7ce3bd1759627e : FAILURE in 54m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52a65c16cd9748f28c05b71b3cd011a7 : SUCCESS in 2h 11m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fde2162c392f49c6b7ba1a76affa7869 : SUCCESS in 43m 02s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5adf4efa3e5f4892af41f48cdee5bbf9 : FAILURE in 45m 01s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e16146a2664b4a5bbad810a024af97ea : SUCCESS in 2h 19m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9084ccadf204aec92e721f727597f60 : SUCCESS in 58m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7708caa71f3d4871a009893f5853553d : SUCCESS in 12m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adaa61d294694e3d9fda78c5b514d2a0 : SUCCESS in 23m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d467071411b4cc98312189e943912c5 : SUCCESS in 17m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6085dc73dae24db7b89212738f73ef21 : SUCCESS in 10m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1754dd3c1f364f56927fe1925e4380ba : FAILURE in 55m 40s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2f284cbcdd7741ea9d0fedef69990e69 : SUCCESS in 1h 08m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/42f129b638104aceaa73fd62115ca114 : SUCCESS in 1h 13m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8792a73a755f41d4a30f1449d42b483e : SUCCESS in 1h 40m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d8d903e840e4f24a6b4c0a969b5e3b2 : FAILURE in 58m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/925445439a674e4ea80d792438fa6bfe : SUCCESS in 1h 05m 51s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/7250ad5865c741e4b83ba0ec20681d79 : FAILURE in 58m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3d901f43ce654f91ae68285882ac0c42 : FAILURE in 50m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5ddaa974eb424b67818afccb07857ad0 : FAILURE in 51m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c19a316f5aa74ce6b448c61f05c23736 : SUCCESS in 1h 10m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/45e3af884366445b8908bbdf419b5ade : FAILURE in 1h 08m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1665b7bd6d2e40ce9963bebe86cf558a : SUCCESS in 1h 21m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bd5827d94d1940f2b5d8dbfc7754177c : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":57},{"id":"963186dcb93f165caaa0b0051eef4dd8960c06e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-12 14:14:28.000000000","message":"Uploaded patch set 58: Patch Set 57 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":58},{"id":"8df18b890d8c307092fc04a6b24f84208eefcc18","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-12 14:35:30.000000000","message":"Patch Set 58:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/58 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":58},{"id":"9a70bc8701d9cc7587f73cfd15529e6d48b6df8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 14:51:14.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68ca45812a764a4eba39cc032b994409\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2809a4b98ea24f6cae3b961e76febbe1 : SUCCESS in 34m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/11e227074e954b26b036c14d717032fc : SUCCESS in 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"c15154ea4e6b9de383a07ae53aa3a996c1b81044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 16:28:10.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f49bf4dc489f433ba4c2981fb1e3695a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed27f4771749431dbcfc5580d6222bdd : SUCCESS in 3m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/07f8002c247f4bdb88721a966ac9dcae : FAILURE in 56m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/25253f93d4e5425ba4784bd1e1a4b565 : FAILURE in 1h 00m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4445e3fa3dd7434cadf82d9279f130b8 : SUCCESS in 1h 56m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e0f2c558d7d46c68095def54fd248f0 : SUCCESS in 37m 19s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75d1503b5e8240bdb51122ba1f5aeae5 : FAILURE in 52m 34s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ddbacdf4b6f442ab6a144b562b03721 : SUCCESS in 2h 12m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81a316ec08a74e4d8504828a097b0c40 : SUCCESS in 57m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bcd8b54603240b5b4d08079fccb6af2 : FAILURE in 8m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05339b99caf34f3386cba68218adabed : SUCCESS in 22m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3076fcd0282a4038bcc0c15d6f0bacde : SUCCESS in 17m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7f1df00a4be418e8ffd8e05a87ccbaf : SUCCESS in 9m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/10b67aae265041d98750b4e4aa840512 : FAILURE in 45m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4512708095844c63b6e1ef589a178c2f : SUCCESS in 1h 04m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/11f81d37d5e743048ab018b2159bd88b : FAILURE in 38m 53s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/43b6c420606f4dabb924465e5553bc17 : FAILURE in 1h 46m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4a44cef2a4b45bbb8afb9b7af06f8b2 : FAILURE in 54m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/335290abbd514a8ba6dcef01e590396a : SUCCESS in 1h 08m 58s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/b1ab32d8c8ef4efabbdab99aad45ae9f : SUCCESS in 1h 29m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e7b0e5b9c2f47ef930bf2566d866c08 : FAILURE in 1h 01m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d263fbac28f746c8934d35aed3b61c1a : FAILURE in 1h 02m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9c468e9593984272a7af362d48790af2 : SUCCESS in 59m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ce991f448d5c40248c8fc0aba0787b43 : FAILURE in 1h 06m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/96dfc16f7e4949f0910fad7acc68c989 : SUCCESS in 1h 37m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e1c162f620d04fd9b83f5e20ff5bd14b : SUCCESS in 1h 33m 12s\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py","accounts_in_message":[],"_revision_number":58},{"id":"1ce71c7027c8cd628336970dcd20f71004fc8176","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-12 17:48:18.000000000","message":"Uploaded patch set 59: Patch Set 58 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":59},{"id":"efe62fde9d186665bc712e076f54c3abb0b2b28d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 18:26:47.000000000","message":"Patch Set 59:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5d7d4d386104cfa9f945506d8485d95\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac44222e2b3545a1aa337a22e39d73bd : SUCCESS in 30m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b58abc0f3bc34f309cef65860894821f : SUCCESS in 29m 54s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"0ef75ee86ecb7d4bf2b2141ddbce1a18e6d94df1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-12 18:30:30.000000000","message":"Patch Set 59:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/59 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":59},{"id":"44de52a7a94c8d1899cc2f2f03833d16035628c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 20:12:09.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87c94725876949818cb5c6a4dd27e21d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cc5e2bdfcedf4e7faf682c09bbcc1eb2 : SUCCESS in 3m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/83c3e05b67ec43b4892193fe1b4b2b9e : FAILURE in 1h 14m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08d942007cd2443b97021c6826836af0 : FAILURE in 1h 15m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c434ec226e2d4e5eb56fb177523b7886 : SUCCESS in 2h 17m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e583775aec4c4e67bbb7f6c09c75c7ae : SUCCESS in 34m 55s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef3379eae1f245f0b42586c0a06d45f6 : FAILURE in 44m 32s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee530b1c5f074e148c48176c2a53eaf5 : SUCCESS in 2h 19m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/187911cb85a64f7c9ae23493e654c32e : SUCCESS in 31m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b87e1b9cfc94f4ab0100442c5ebfcf2 : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/655edb54f85d4c1c8c0a0d4dedd190a1 : SUCCESS in 24m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/47d093bd939c41c7bc0876d7e0fb8678 : SUCCESS in 19m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c384972e3f014af78af9b0179e5b0bf6 : SUCCESS in 12m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/35a03b132d2c4bfdb7d44277350b8f1f : FAILURE in 41m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/db3429e8aaec4de5bf39e1285b33427e : SUCCESS in 56m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0630533740df44cbac0b92f341dd6f03 : SUCCESS in 50m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/075aea8caf00444b9a2634c875546d9a : SUCCESS in 2h 13m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b14f8cea206345f1a411b8b06d6ef8ce : FAILURE in 55m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/177848e98c3c44d69fcd517060cd90ba : SUCCESS in 1h 27m 20s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/5203d34078bc4e29af4f0b190f0c2041 : SUCCESS in 1h 19m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3e5ed846406e44c4994c248bf0bd109d : FAILURE in 58m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fdcad15413674fb6b11440f373a40738 : FAILURE in 59m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/64b9082f54e44442ab7aef7d005c30ab : SUCCESS in 1h 11m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f60aeaef4c0943429de711376c6d02dc : FAILURE in 1h 00m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/de30e8e56fe341fba1b9dbc64707c8fd : SUCCESS in 1h 11m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0dbdc25d7efd481495c9a56b9ad39c53 : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":59},{"id":"a8d1677dc3d9644aeb0cae4f7910da0fb422ba6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-15 16:40:44.000000000","message":"Uploaded patch set 60.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":60},{"id":"68022b57dc990df9ca429199730bb5dda547c70c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:41:28.000000000","message":"Patch Set 60:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d71e8906d1949c4b2ad033465f2fb1d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/166d8134cd184abda68ccaca0d0c402a : NODE_FAILURE Node(set) request 300-0026348620 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a05782aa0cd4ce48dd3bc5f7f7648d6 : NODE_FAILURE Node(set) request 300-0026348621 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"badd99e9fa727fe23fd7c9a79ac5d501eb9e2453","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-15 17:22:10.000000000","message":"Patch Set 60:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/60 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":60},{"id":"f0dcd3e50eb4b92b8cf584731b85112e5d439d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 18:33:17.000000000","message":"Patch Set 60: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35bd2b2035934eef83ae995aa438a6e1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/da81cb79a3e34d57a2025a209f64419e : SUCCESS in 2m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1e0d379247424bda9d0c0d2941e57ddd : FAILURE in 50m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/99baaeea730645d180ce839a66c41cfc : FAILURE in 59m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6184785700e847f2830c86377d158c3a : SUCCESS in 1h 32m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8cec892a50f45f6828a74a9ba6bc534 : FAILURE in 13m 17s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54d2d50fe3234bc2b0c1065a2b8a3e51 : FAILURE in 49m 18s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3890d4dc612545d7a61006753199908a : SUCCESS in 1h 38m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/068c96f3dde24053a8487f4348d44bf0 : SUCCESS in 48m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6691a7a95a24b4b9e140ad03054441c : SUCCESS in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9ca7d4a30824505a223a876b0065fbe : SUCCESS in 22m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cd12e5d508d4c5ca4f0c82954fa5e88 : SUCCESS in 20m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09c799a2acb84986805b61de96ef02c8 : SUCCESS in 9m 44s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/49cc2fb1c3e04bb88dd8ca69a01b8605 : FAILURE in 35m 18s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5413b46f23584c84a588aab9249292f5 : SUCCESS in 45m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5419b08b233d4aa0869cbb7ea4ee8b2d : SUCCESS in 57m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3c7cf1e980c34b81b2e73c7b1e4e0964 : SUCCESS in 1h 21m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed6c268141b34b98abf61ea660c6ad2d : FAILURE in 51m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a899c9cc59334b01909ded79ae4a7592 : SUCCESS in 56m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b0b42e0ac4a545bf8d095c3bcf813eb6 : FAILURE in 37m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5484824698e74d7c99e5188d45d8cedb : FAILURE in 38m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/837de10166584cddb556ec555ff4432b : SUCCESS in 1h 09m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7aab4fc3391745ada31f5a18ac0068ab : FAILURE in 1h 03m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c0a94e8955cc4ebeba6b143f784286c2 : SUCCESS in 1h 13m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c09bba4217d242dfa81283f41bc5d53f : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":60},{"id":"2e323b0caa57ff9ede9942bcc637435222ab7ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-16 11:25:52.000000000","message":"Uploaded patch set 61: Patch Set 60 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":61},{"id":"1885b35f46bc66bf19adf12c63afb0c4202c5098","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 11:26:35.000000000","message":"Patch Set 61:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe73c3bfebd64bbb941182b15db42cea\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3936865aac0542a7bd8cce336f393b6b : NODE_FAILURE Node(set) request 300-0026351768 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4526f2e130124f29b5fe5238d0415a21 : NODE_FAILURE Node(set) request 300-0026351769 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"a6f17ee4a3463f31fd5156001466cbe472c9707a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-16 12:07:17.000000000","message":"Patch Set 61:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/61 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":61},{"id":"ed7cbab1fbc06b5d263d49126d40f5e12749925e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 13:33:18.000000000","message":"Patch Set 61: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bd0ab537b8349608281f7d06c2dde8a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c2216d62c6a54ec0891b10fcb9d69d75 : SUCCESS in 2m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/88dd96524f11488aa97ff57b61fbf26d : FAILURE in 45m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e844aa851754e7a91dc5aa94784e412 : FAILURE in 1h 06m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d14b866b519415b877352507f42b257 : SUCCESS in 1h 30m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b1c3262a4a5425db8fac3f5aec2b154 : FAILURE in 23m 00s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/190448bd99734b659d0c2eb5d7cea8ae : FAILURE in 46m 03s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/794bc61c47c444eb8623bf77b61f2751 : SUCCESS in 1h 46m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5d259e6f6ef4b84afb90b758cf21c74 : SUCCESS in 49m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c26dfad23b4840e79fde959d132a914e : FAILURE in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d234b3e48e3f467cacd91276c8fbe43c : SUCCESS in 24m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e714398970314c84ac686f30c309f48d : SUCCESS in 22m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d317416902bc40abb3fba3bd54a33cc7 : SUCCESS in 9m 30s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a3e8170492c2491d81e4112e71158b53 : FAILURE in 48m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/23fd1aeb4bb04f0ca6c086436fff2418 : SUCCESS in 50m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f89bd06a96e6444e8891d1ef71559468 : SUCCESS in 1h 03m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d3da314d6fc04b2787e348d2b8230a77 : SUCCESS in 2h 00m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e3740c37dcc4a8ca57af05e4340e34c : FAILURE in 43m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/db0e3924666543a3a9ae8675072bf7de : SUCCESS in 1h 11m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ac8ec82d1242451d8534e1fd2b875718 : FAILURE in 44m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2ec092c048bc419186146a03369ead7d : FAILURE in 57m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/53925747763049d5a3f83ea2b1fcc5f2 : SUCCESS in 47m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d081cae415694c7a8d9c54f36e1fb495 : FAILURE in 1h 00m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7897400b4e3a4f1b9e7d6797eda69902 : SUCCESS in 1h 18m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/800fcd4e301a4d75808a58a2028040c4 : SUCCESS in 1h 23m 17s\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/main.py","accounts_in_message":[],"_revision_number":61},{"id":"7e63f6e62dc1a05cdd1084df2ed846212ac0ad56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-17 10:49:24.000000000","message":"Uploaded patch set 62: Patch Set 61 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":62},{"id":"dd883e19aff6cc872ef0c143d857dc08b16f367b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 10:50:31.000000000","message":"Patch Set 62:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a11df318b78f498d9fe41f79ed86b5d2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37673f217f714f8bbb470a339d5f4e12 : NODE_FAILURE Node(set) request 300-0026356089 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1998704bab264882927f2e7c71bf1ae6 : NODE_FAILURE Node(set) request 300-0026356090 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"3fe8331db7ec39a167a4a835b48215f59ceffb72","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-17 11:10:29.000000000","message":"Patch Set 62:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/62 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":62},{"id":"dcdec5ef411f2c501b2eb78dc1868c75bd308cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 13:15:47.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d46dc02e4054c779c45be7842929022\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1c680c7887f946ae9cc38d4d60a7213b : SUCCESS in 2m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d531f486c8f7471986946cd01e9f9ab8 : FAILURE in 50m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8ae1b0e14de847efb070c1fe45078156 : FAILURE in 1h 15m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3f2de2cdac594744bfb4189b67b85cd7 : SUCCESS in 2h 03m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5df4bb31665245a9a9b9853b322f5498 : FAILURE in 21m 30s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c53a3c72d574b24b8e3cae8f8411603 : FAILURE in 41m 27s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47e342aeda984d5e9742ef7d859a0dd4 : SUCCESS in 2h 09m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79e8627e11a049b4b702cb8a895b5408 : SUCCESS in 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5d5f872d9b4690ae8fafbd67a8c2fc : SUCCESS in 10m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21709bc9b13d42189b2bd8002a0e3198 : SUCCESS in 19m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d61ef157f2c846ffa5c2fb6d9f081c45 : SUCCESS in 21m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1087b161ba704f8e8d50738b9a07a6b3 : SUCCESS in 8m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7a7af2f6efd845f5be317df4e8091aed : FAILURE in 52m 13s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bdce21497cc24cfebf5c44b04ea4dad2 : SUCCESS in 1h 14m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7e46535e94cb4d6db4028a7a802b7267 : SUCCESS in 1h 12m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0474e8effab24d689891f5c21ea401cf : SUCCESS in 1h 57m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ed4f650b555445584558898fd850e33 : FAILURE in 1h 00m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/31ade2ca2d504bb296db28d7b05d16d5 : SUCCESS in 1h 07m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9921ab8a38494abebbd96da120030a30 : FAILURE in 1h 07m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f20e1f43d6454d31a3f9286be0eaccdc : FAILURE in 54m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/83bdd2feb599464dba0d2903ff6c6b3b : SUCCESS in 1h 21m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/77d4eec033b8416580e9a43bb0a8cb7b : FAILURE in 1h 07m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/413d3628a6a243bc89f204222b0a5fa5 : SUCCESS in 1h 07m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/077df50aa5104571b412406957b7428c : SUCCESS in 1h 27m 39s","accounts_in_message":[],"_revision_number":62},{"id":"8a72068b93cd30aeebfc9151e2c556e4a5f1f707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-17 14:34:49.000000000","message":"Uploaded patch set 63.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":63},{"id":"8fb224657b8c8c7364e1eef698cee291319a903c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 14:37:07.000000000","message":"Patch Set 63:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a0f792cef334e12bdb4e17d507a16fc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9f8679c84db41408701f1c6720c04ee : NODE_FAILURE Node(set) request 300-0026358455 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/089f3b14b20245e2825035d5b810f764 : NODE_FAILURE Node(set) request 300-0026358456 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"7f3f1b3024d5881b6e81c2f5d4211615bd53ff94","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-17 14:44:21.000000000","message":"Patch Set 63:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/63 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":63},{"id":"ddd771650d37c38bc2d1f0016526248f1fba25d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 16:05:18.000000000","message":"Patch Set 63: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/444656e22618444d82b73d84843fc9db\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d6d745734fd54092b057f85b226399b6 : SUCCESS in 3m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a93350cc429b4917b8da13812aa5fd9b : FAILURE in 10m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e94c311952654433a39b98ab8f14e340 : FAILURE in 11m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/26703c29d98e42ffb705f925526b4779 : FAILURE in 12m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5473f7e82791437fbf75025451b5dd06 : FAILURE in 12m 02s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1db7c1b005446ee9295dd8aad35ace3 : FAILURE in 8m 33s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/536d543331134581946e63982fb6c2cd : FAILURE in 9m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3b4b91e4beb46a4a4c8b3c613b86d61 : FAILURE in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0585c60e62449e9ab4d1171caf749d : FAILURE in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61d6aa6017a841cf84d2d0724452d920 : FAILURE in 8m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/712042417e384117a36ca1e5d10871c6 : FAILURE in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b4cdd526294ecd9ebd68e369360436 : FAILURE in 19m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/55f644261e4549e3a27671ad9f7b6fc5 : SUCCESS in 6m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/80cf56c2104b4a98a751b61444c5d0c0 : FAILURE in 49m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/280d1469cdb34a8593ba2e1c9988d5fd : FAILURE in 52m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/57c46041ebac48e98903943e8f2de261 : FAILURE in 37m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3c2b388afd1a452c914e228fb432f75a : FAILURE in 35m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c582d17df5184088b2a6f061bce8353e : FAILURE in 41m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/861928e8b86c40c4a47963ef732c57f7 : FAILURE in 59m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a500ebfe59a64d73a41a75c6e26f573d : FAILURE in 40m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ec12377041684775a56f812ad663a864 : FAILURE in 23m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c966dc7de8ed4e5da5bb050db1e11912 : FAILURE in 39m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/70c63d0bde2c474dafdb0c4c3114a164 : FAILURE in 10m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/96dc00944ab44f828997144541326824 : FAILURE in 11m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1dd778c0e0e14297add35bef686a7d9c : FAILURE in 38m 23s","accounts_in_message":[],"_revision_number":63},{"id":"dad612934e6afaccc34d73eab37ade6a1b9874e0","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-18 15:18:26.000000000","message":"Patch Set 63:\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"b1846b5825f39f9844351c2524ce7136c90be056","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-18 15:28:08.000000000","message":"Patch Set 63:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/63 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":63},{"id":"2240be4cc1c233c684dbe0861c6a2678d439a84b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 15:29:38.000000000","message":"Patch Set 63:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70624121ceb4752b7b1cc3619cc72d4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8e206e37f944441adf98a1195681028 : FAILURE in 8m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4d265eed31346b289aa5adfd5b41cda : FAILURE in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"8d017b00b7f2431e9515abf5b0a41a208f9a5e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 16:19:45.000000000","message":"Patch Set 63:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f872581981c4d0d914dee3b68fc4b02\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5414ec4c219046e189a7498489bfff02 : SUCCESS in 3m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ed6779bcc17f470cb418c1a6487f81ff : FAILURE in 11m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1759eb299fb54426840df1625450daf5 : FAILURE in 12m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4e4d15b6f2b248f1a0d27200f7bb4eaf : FAILURE in 9m 57s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38c0ac390b254d4887ee27cca4fdac92 : FAILURE in 11m 21s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/106a2c5f12094a31aa06b1c69af091ab : FAILURE in 8m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33ac34d5e20a40acb0895d0499e990ca : FAILURE in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb6734b8768d4c91ad62f92afa1a5557 : FAILURE in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54f9198601e04afc820a1be06c9bdabb : FAILURE in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b33bd2f675c247bf99a3edd7ac57d4b9 : FAILURE in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a05e0433cae4187b8c7aae4e884d7a4 : FAILURE in 3m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d253234afa1401cb624c33a7e119f73 : SUCCESS in 3m 56s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/650c97c18642462bb1e6ff94c5630c9f : FAILURE in 40m 05s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4c4842473253474b94a3b99d0656496a : FAILURE in 55m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2752f9d8c42d4e0dba3d173effc8f8ae : FAILURE in 24m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8574e1600fd8413aa1178c94e03bdd53 : FAILURE in 17m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2905dbe6f22e4dcb8db44d5849e363c2 : FAILURE in 52m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/56b581b672694200a3325b704d8e3a90 : FAILURE in 54m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3a343b08c1b74351a067e727e01cba1c : FAILURE in 12m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d00a1d7e4064597ad4d1855806f4e95 : FAILURE in 13m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2554b41ce6ce4eea9eb1b4cffb649b02 : FAILURE in 12m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6e22c5fe19584f6681c2424bb96ee9bf : FAILURE in 18m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3bd1185c5e824da1966fb55dbb381f4b : FAILURE in 9m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e96027de1e7e4ef199b55fbd57fde92b : FAILURE in 11m 47s","accounts_in_message":[],"_revision_number":63},{"id":"0c5ebaf5e7894c1a451abfa2f7e1d1cd185af713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-18 16:55:39.000000000","message":"Uploaded patch set 64: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":64},{"id":"36ff64c8be0b2024e88a079357b83732a3f88b1a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-18 17:05:46.000000000","message":"Patch Set 64:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/64 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":64},{"id":"ce5968b6840a6738c40eba7a520808d81f693455","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 17:27:57.000000000","message":"Patch Set 64:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71da8ea4d9b24b7d8c234c91b0936985\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12f19f83011b445c8b258189177e492f : SUCCESS in 24m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a5c3362a7ea418f8dd462de0ffb6c04 : SUCCESS in 29m 23s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"7b27899812a6bf2fae9aac3acf857c1b63ec3679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 19:21:44.000000000","message":"Patch Set 64: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ff36e160bda4e658e5d60dd31de62f2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04b141dbc65a46089885db16dc2573fe : SUCCESS in 2m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6dac9a0221944e4e8f8ac5b8b91e79fc : FAILURE in 1h 04m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6bb7e71397eb4475ba0f7b1fbb599c5f : SUCCESS in 2h 23m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dcb680cc562442a7978aa915c2d651ea : SUCCESS in 37m 34s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4dd0a9e8d4ce4300995bebc517b6f13e : FAILURE in 1h 00m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30888c157e9141ee8f4aac3e97eac866 : SUCCESS in 1h 30m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ade373b79d424345a7e75c52e02f5a64 : SUCCESS in 59m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a44cdea0422468999bfdc048aca37e1 : SUCCESS in 12m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f6b85ffe3a5405fa901e615a10de7e6 : SUCCESS in 20m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be524774f48e4e33a41bbfeaaf1e4ff9 : SUCCESS in 21m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1098619daa4e1b855094b127adf31f : SUCCESS in 9m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/10e8af0b92f248c08e56c7720dfe78f9 : SUCCESS in 2m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d959c4df0d6c4a699077e7f6c7d3bd79 : FAILURE in 42m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4e024f4f37ce46ab9f32e8e820fb4940 : SUCCESS in 1h 04m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0a60d95b16f04b0db00b3babfb470e16 : SUCCESS in 1h 11m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b0cefc209f29482dbf60048d016802ad : SUCCESS in 2h 04m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/593982a87c4c4851921aa4fbeab19b52 : FAILURE in 50m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/edf45354b1984611abd0201d39ef52bf : SUCCESS in 1h 01m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad679ddf964441bdb5c72431e398ad23 : FAILURE in 54m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a1159214a6f14e4a9e7e3e0801884499 : FAILURE in 1h 04m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5cd15fde26b04745acee8a9cfaf81474 : SUCCESS in 1h 17m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/00280285b9c94cf6b25aaa80921bdb34 : FAILURE in 1h 41m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f98a9c18126d4d41bdde9d9a07d0a60c : SUCCESS in 1h 12m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/daba488190304871b53b23208b2cf18f : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":64},{"id":"6a01def135c264c36498ad2ca35edfc9035c28c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-18 20:10:15.000000000","message":"Patch Set 65: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":65},{"id":"a0526cb1c207382d04ac436ffe1f578344b32a95","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-18 20:10:29.000000000","message":"Patch Set 65:\n\n(1 comment)","accounts_in_message":[],"_revision_number":65},{"id":"cb124019a612bcc931693985163776031b194217","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 20:39:49.000000000","message":"Patch Set 65:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5ac4ba7c83c452fa4cedeb72ff06ad0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5a74b9f4f5348ecb6cdceb4cba0a807 : SUCCESS in 26m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10450ddce3b846efb46f0a95825f7bcb : SUCCESS in 26m 31s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"2631b5c7d898fe08b4b2a7541af146d663504caa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-18 20:44:38.000000000","message":"Patch Set 65: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/65 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":65},{"id":"1e51ce9540a00c14d9b51e214c17f28a61a32e19","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-18 21:23:04.000000000","message":"Patch Set 65:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/65 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":65},{"id":"eb874469b5b0573b22e9f2628ab285e995371723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 22:17:17.000000000","message":"Patch Set 65: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1e7989c5f6644d8b5d4a4ebf9df36b6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/17b373d395744142a5d69384d1dcb6e4 : SUCCESS in 3m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1953230a56fb48399780d9fe61339655 : FAILURE in 1h 04m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f23d4cb1016d4a03b4319cd696a1a42c : SUCCESS in 1h 59m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2c68128d84a747d9ada453d21ad30d41 : FAILURE in 34m 13s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f906b983a0cb4c248723f035c7701c6e : FAILURE in 46m 32s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7166c32ce1604e45b8adce5d2b0734e1 : SUCCESS in 1h 49m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2059326dd7d843e19b6fc1ceb3586113 : SUCCESS in 56m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0b21de767ef44299c381f2b14054088 : SUCCESS in 13m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbab9dcac6e34f81830794b5cd286849 : SUCCESS in 23m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71d8d50db4154495aa31cdac7880aa34 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72855763b1b9478a8aeb72b612b38644 : SUCCESS in 7m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5541b2811ff14f298a8f8a2352112fde : SUCCESS in 2m 35s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b9fdbbdc023d4fb88502a62915ebf6ab : FAILURE in 42m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/905206e3c1984504ad64fc6cbfea954b : SUCCESS in 1h 11m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7a48c4c1499d43d0bb9e6f14a2133e2c : FAILURE in 48m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/653af8b23ad248e0b3af7b589cf409aa : SUCCESS in 2h 00m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/95339ecd112d484eaf39eee860c7958d : FAILURE in 1h 02m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4bd0f330ea7342858ba78874a5702462 : SUCCESS in 1h 15m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cdf52825bfb749d6b1cc15cbbd9091f9 : FAILURE in 1h 02m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/22ddb9e993b84b03ab0f2fa2fd40ed01 : FAILURE in 1h 00m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ee43b88a3c464e618a0fa803cec395ec : SUCCESS in 1h 07m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/78324b85c0e849f2a0dcebcb2f19a443 : FAILURE in 52m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/473b671651644f3d9109c3f8990c5da8 : SUCCESS in 1h 04m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7a5dedea0d064915a3f110b304002df2 : SUCCESS in 1h 19m 32s","accounts_in_message":[],"_revision_number":65},{"id":"c57a9a9d5670d090b2b8b07479dde14310717170","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-19 08:03:21.000000000","message":"Patch Set 65:\n\n(1 comment)","accounts_in_message":[],"_revision_number":65},{"id":"ba1ce63e8c5a3d9ce903d72112c443b6b2075ac3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 08:35:54.000000000","message":"Patch Set 65:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ebc4eed934a47ac85fd1344e7a98bf6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6ac7c9390084c0498cbd7015b761455 : SUCCESS in 29m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/393b3fa766384d9fad38d4f3993c70de : SUCCESS in 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"7c8a8fd5bdc8be90d2296866a20e09444979ae6d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-19 08:37:20.000000000","message":"Patch Set 65:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/65 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":65},{"id":"bd67b0280c2bf9f84f2749c4379ad8d1918a6377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 10:43:43.000000000","message":"Patch Set 65:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bf8c98eed3e45d386c43285222e3362\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2b2f771b3d844083a4e53b9a2b0120db : SUCCESS in 3m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6d8d50d6043b4b2092d59de6a5d6d114 : FAILURE in 50m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/91e12194d2c845f69356577aefc90cfa : SUCCESS in 1h 40m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1365b3fb2be842f5b90859757d24b1e9 : SUCCESS in 42m 55s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88ab5bb6738d478c84ce962a83be3a5d : FAILURE in 1h 05m 52s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7742af2778ab449093d464e2cdd08a3a : SUCCESS in 2h 08m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a259daca39944c0d942cbd22046752c9 : SUCCESS in 49m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/751b329a9ea645efb082f436c15511b1 : SUCCESS in 12m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/546d74b942d447d095060a07bc355f0f : SUCCESS in 20m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12bb742c177d45d18d12c4bc47b9e848 : SUCCESS in 21m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6dd2c94f57489885044795709bf3d2 : SUCCESS in 10m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4708b80ccd84c5e8faedc5da9cfc2c5 : SUCCESS in 3m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/44a7fe6a7e4f4695a5940d2bbdd5f6d5 : FAILURE in 45m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/023a0d57fc844492817150189a82c098 : SUCCESS in 42m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/03dc8b6490d240b0a3040a4bbe79251b : SUCCESS in 1h 10m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bfbcbb075c384dbabfb4422783dcb667 : SUCCESS in 2h 04m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/954f69983e7644a0a63b1abd6be3dd89 : FAILURE in 1h 03m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7e1daa100794af09752f67d396c0e4c : SUCCESS in 1h 15m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d2ebc5c380c64b6b944bb6d2318e276f : FAILURE in 37m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/04afda92523f49f49fff9a96648a01d0 : FAILURE in 1h 01m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b535000d605b4e61a30c1ade2e109cb5 : SUCCESS in 1h 16m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/235f4832fe364c0ea76ff5ea7b2fcddb : FAILURE in 1h 11m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6e1acee573ba40ee9059ef280ee2bcf4 : SUCCESS in 1h 26m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c7403becc20d43eba014be0a486dff92 : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":65},{"id":"b0563af379cc4901df6d57344c3956de2a46a29b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-19 11:54:44.000000000","message":"Uploaded patch set 66.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":66},{"id":"24faa0fbf39787f6c6486dff6157111fb032c790","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-19 13:41:06.000000000","message":"Patch Set 66: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/66 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":66},{"id":"36be5d28303b6e9af46b832396d1dbdc80a5b9ed","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-19 21:16:07.000000000","message":"Patch Set 66:\n\n(1 comment)","accounts_in_message":[],"_revision_number":66},{"id":"64cb49975120aae852837b8be22ad917a8e7c481","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-19 21:51:31.000000000","message":"Patch Set 66:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/66 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":66},{"id":"5a52d44f2db77695da721c36d2045f8d91e08450","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-20 08:01:00.000000000","message":"Patch Set 66:\n\n(1 comment)","accounts_in_message":[],"_revision_number":66},{"id":"9c09be3f7ef1207c961f68468d65abc4e4b1fb5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-20 08:03:45.000000000","message":"Uploaded patch set 67: Patch Set 66 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":67},{"id":"397bc1da3eecb8fccca3ec7a76256a1840d5d4f7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-20 08:34:47.000000000","message":"Patch Set 66: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/66 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":66},{"id":"4943dfc0d34512aa72dcbede86bc99fd4edad9c0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-20 13:04:34.000000000","message":"Patch Set 67: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/65/937765/67 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":67},{"id":"40aa59f29753996c83a757e7fcbe5c642a232ba4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-03 18:15:40.000000000","message":"Uploaded patch set 68: Patch Set 67 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":68},{"id":"e0492cc869917c9fcd61a4a7dc9ef4c10105d621","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-03-03 18:16:09.000000000","message":"Patch Set 68: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 942125,1","accounts_in_message":[],"_revision_number":68},{"id":"28c861f7296ec6e8995d2b302a5dc7a2d271cab9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-03 21:50:43.000000000","message":"Patch Set 68: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/65/937765/68 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":68},{"id":"7a7bac31e46187a588376dc8429aa81cbbf48601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-04 08:32:37.000000000","message":"Uploaded patch set 69: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":69},{"id":"7a750e3eedb36c655b399d1bb69a94b781c5fe56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 09:06:23.000000000","message":"Patch Set 69:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0300fb0b999641a2940e24e4c69ae7ab\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/523c99f6d4074e8fb93b37b84be56e9c : SUCCESS in 26m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef8f19599d4248368a5074f85d731b8e : SUCCESS in 31m 25s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"08c4407f95ea64ea8d4de18ad65e807c9e868c42","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-04 09:11:22.000000000","message":"Patch Set 69: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/65/937765/69 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":69},{"id":"64799d63ea6f3620c734426cdbbc9c1ad34ff986","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-04 09:46:25.000000000","message":"Patch Set 69: Code-Review+2","accounts_in_message":[],"_revision_number":69},{"id":"6a9e3d9dbd4ef90f1aca01e9b80d10032f4fc90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 10:55:10.000000000","message":"Patch Set 69: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f265a813a21449a9ab5f399d90a7d1f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ad7a4016a73a46f38f96eba645663d29 : SUCCESS in 2m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c9c1ce9c366045d6b0843caa12f3d641 : FAILURE in 1h 01m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/49cf3c2e2c0d4e5da3426ef7f3b24821 : SUCCESS in 2h 20m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f95a15a29c4340a1b9e0ff903e5b6b8e : SUCCESS in 41m 43s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f17b43aba96483e8d8efbed736cebde : FAILURE in 54m 05s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7153028b1a344d9880ee6906680801ca : SUCCESS in 2h 00m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f520c8646f54427a051427ec549e79a : SUCCESS in 53m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9d8a1d13e214105beb18b735de032a7 : SUCCESS in 11m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d67acdb13f140cdb02cc9bf2233b497 : SUCCESS in 21m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce4f8b27ef8f4d319ac527567387d124 : SUCCESS in 20m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5fec3f5bda1417a92ea3a572c5734b9 : SUCCESS in 9m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/da056f27f8f8471b93d10933f5217086 : FAILURE in 50m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3a7c29d15b334bf39f18705e0f0b67d2 : SUCCESS in 1h 21m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1d690caf80fb4dfbb7994d9193bd4dff : SUCCESS in 1h 18m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/abcb33ae347644d9a15e7be34ef903cd : SUCCESS in 2h 11m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5e59b9bd5b5401ba8c52a4582397004 : FAILURE in 56m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b146683a36504dfcb628eb6ab3a5550f : SUCCESS in 1h 02m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/07a9083e04044fdfb47f269541e28ce8 : FAILURE in 53m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/184a87a2313e454cb70a76fcca61971f : FAILURE in 39m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/95ebf0f6e1524021b67ddac0e0fc4f68 : SUCCESS in 1h 05m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/55b3e1bafca746d09b9fec6dd0a53c5a : FAILURE in 58m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8840ada4a36f4db6aa5bcc2ae02108d1 : SUCCESS in 1h 13m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/881a0e381e774bf1b6e53f81f882c114 : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":69},{"id":"bd4f19890ba590835d1849ef6b9e1453da655e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-05 08:22:22.000000000","message":"Uploaded patch set 70: Patch Set 69 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":70},{"id":"a306c00accbfe52c73cd22d6b7250b46b777569d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 08:54:00.000000000","message":"Patch Set 70:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b64c968c7b5492faae6703962ef0684\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db448ac92543460a8e2f46aff4e3554f : SUCCESS in 20m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9804c01dd9974fe2ba3ee2bf2cfb2aee : SUCCESS in 29m 29s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"ab82b4b9b5f39c490b33bc54275e9672a6c442b0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-05 08:57:05.000000000","message":"Patch Set 70: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/65/937765/70 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":70},{"id":"0d25c34000a39323ea3d269964a0f1d3610b78de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 10:45:33.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3e014adafa74b12937e2a06782d25b2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/167c547603ee4a439d8f96d9fe563b5b : SUCCESS in 2m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/100496e8f28242a99325c37603cf27ce : FAILURE in 1h 05m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7093288314424f83bf42108a6854b37b : SUCCESS in 1h 58m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e7955669e6a44ec8aa3993c31ee45d80 : SUCCESS in 44m 12s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee64f609b9b74e77a581d1edeb4c1b88 : FAILURE in 49m 43s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/caa59d4bda5945d5a3e16c2a4dcb1b08 : FAILURE in 2h 21m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9996a15f0b974c3a8492ab655c0b9ad9 : FAILURE in 50m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4dd72a96b5463db5a9b000bef56d00 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e316bb16f0bb486b802282c697d60670 : SUCCESS in 21m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8419e6130a94509a2dd14dbba6792a6 : SUCCESS in 20m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b5e251dce145029fb3315f16ea380c : SUCCESS in 9m 35s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/571789101cf0425db92282b4a3cf0f83 : FAILURE in 38m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3c8cda0d1c5343ff83f01efa4fb5b994 : SUCCESS in 1h 08m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3f8863bd1ad44e7fab3993d56fb64ef1 : SUCCESS in 1h 16m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b9c8be1623384055a3766aea731a54b4 : SUCCESS in 2h 05m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f034bb4273e4dbdb6c5b27af7b4f52f : FAILURE in 54m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0bb407d2f874bd2bcfb89cd12b1550c : SUCCESS in 1h 14m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/341f14fa005c4f95bece92a6f4a35b4d : FAILURE in 57m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c5aa623018a34b8ea27aaa885660608b : FAILURE in 58m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9cf9ce6224b44327b5c7480dfeee1582 : SUCCESS in 1h 27m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d39af7bddb54414d8f497281e1678284 : FAILURE in 42m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0bbc5cb5480145ceaf4d7aa7b4a1a00e : SUCCESS in 1h 15m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/774c2c80f403465f8902d9cc83d16d52 : SUCCESS in 1h 51m 41s","accounts_in_message":[],"_revision_number":70},{"id":"9e5d24f8c82f200a7a9211cf88bc67bf57935d7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-06 07:14:46.000000000","message":"Uploaded patch set 71: Patch Set 70 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":71},{"id":"8281007aaf51d80cc91be2a08c255b247118a652","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 07:46:33.000000000","message":"Patch Set 71:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/182575fc5cb54b088b4c854c6a270eec\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ab3338b3d494db784eaed8363cded0c : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/39222706fd7e441681364809f5ce5c91 : SUCCESS in 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"cc8a9b86767d02b4a7e47635345a150c5dd25b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 09:48:54.000000000","message":"Patch Set 71: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a252c4a6dbe34f70acdf5ed6bcd38240\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1779178561c64f8a85044f7510f08e20 : SUCCESS in 2m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1b0861379d5b402ca8ea180155b1a2ad : FAILURE in 59m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/906f786a2b004cfcaf82a19a5fa86206 : SUCCESS in 2h 30m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fed89b13c40b488ba5b3c33c1f8a68fe : SUCCESS in 45m 38s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ce9cc1add6549f4b0dae7c4be88891f : FAILURE in 1h 06m 21s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef63c482387942729a85a2426356f91c : SUCCESS in 2h 26m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21caeae0b0e1452cb6523eeb087efb34 : SUCCESS in 42m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c31aa6a1f3e4e8bac3e2d4bb585e2c5 : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d59e580e2c394607ad4d57817dd6df95 : SUCCESS in 24m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa6c2dd57ef149458532d6ebd4bf8887 : SUCCESS in 18m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9121ce1ac834585b9c54f7842acfb57 : SUCCESS in 8m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/682261e0c01741448ca90c4fc319c35f : FAILURE in 57m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/02a2ad585f7a411eb6e66d3ffc89b0cb : SUCCESS in 1h 07m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7b82b30d71724542a324b47f2cff36b7 : SUCCESS in 1h 16m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/64f35c6624ed41fab29b774bd1e1f524 : SUCCESS in 2h 20m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae0f961b82b4427faac954b9ae0cf302 : FAILURE in 57m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6714a4b2d30a4b5a9f16d72c27f45018 : SUCCESS in 1h 26m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dbbc9d6ab27a40edabc5f2b30157287d : FAILURE in 1h 12m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e14648f3a848486591f9fa4f26362691 : FAILURE in 1h 08m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cc50eaca84f74342bdd8da61f3199897 : SUCCESS in 1h 09m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/738a6372a5534798a77928d254b77de6 : FAILURE in 1h 01m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5c9d8fa3a57542a28ba9e402be9d7e13 : SUCCESS in 1h 16m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/681befed82d544b685009b2e139ca12a : SUCCESS in 1h 28m 19s","accounts_in_message":[],"_revision_number":71},{"id":"18cf161f031ac5fbecaea9154fb7633511621af7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-06 10:20:15.000000000","message":"Patch Set 71:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/65/937765/71 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":71},{"id":"c85e46d0a2b4ab6644ea383832f39dd132b9f93f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-27 09:23:28.000000000","message":"Uploaded patch set 72: Patch Set 71 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":72},{"id":"33739963d4917be75ed1d60681d78165da690a58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 10:33:13.000000000","message":"Patch Set 72:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73785fa053934310866d2ec0f1d13b47\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/693d84a49a5240a0b70d50cedf292371 : SUCCESS in 33m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/daecb43c18824f98bd0d665dee6c7060 : SUCCESS in 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"56635d918e2256a633e87a0af3655e839d9a8451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 11:34:44.000000000","message":"Patch Set 72: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2a51a90f7144d768924f9b38b8e4b69\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/805e1ec0a021413bbe12200b41bbfca9 : SUCCESS in 1m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/15006dc556a44f528b8a6979cf4a0722 : FAILURE in 53m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bdf299430e0b4e47a953b78a2e2e6eb6 : SUCCESS in 2h 03m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11b0c537a1d94740994d1ba8183b2815 : SUCCESS in 42m 44s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f916ac8f9ed47bebde0aa40d8aaada2 : FAILURE in 50m 27s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e414a0d262104d9ea857a7c8dae058d4 : SUCCESS in 1h 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb48538ea6f74047806b61a9bed16b88 : SUCCESS in 30m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d2a3987862456cbbe50edd21878d73 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c77ee40fa14a4e2ba62923a4856ffd18 : SUCCESS in 45m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71f0c7a769c741d89f77c93980a4d9d5 : SUCCESS in 28m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/089d3ab279ab4b1fbd4bfbeeb1162253 : SUCCESS in 12m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fdf776a9dcae4c2894766e886e39c14b : SUCCESS in 1h 25m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/df5b2641ca6f4107a3d5d88698893ff4 : SUCCESS in 2h 00m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f23db1312af246858498bc9c0bf67454 : FAILURE in 42m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e4fb3d3441a4c04a1c11b2b1a6ffcb3 : SUCCESS in 58m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/66b0b09c37c84a4aa06307b240444429 : FAILURE in 48m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1be1d29bdb664b9b9a6c6f8c24368862 : FAILURE in 50m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0ab4176aa5414186af35293579ac10d4 : SUCCESS in 46m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/217b825ac1b64999832f2686becdfe47 : FAILURE in 27m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b009ba14116040218f3735e7f572d7ef : SUCCESS in 33m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/43b2fbbfcfe74f879eb8418c1fbf53c8 : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":72},{"id":"7a63f036411e91fa3c10b6e847ac5cf17c7dd168","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-27 12:13:28.000000000","message":"Patch Set 72: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/65/937765/72 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":72},{"id":"5738a7bcca362b781d9eca9f8468c2f16fa9683e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-04-15 09:49:25.000000000","message":"Uploaded patch set 73: Patch Set 72 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":73},{"id":"50b5f2fd7f74d28ee148159a9b0744e4fc8076f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 11:20:31.000000000","message":"Patch Set 73:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99acf764889c4a9c8d58a651f9fafaab\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49b741a501e64c85b14ad32c24a3586a : SUCCESS in 24m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/524f596e7397455f801e4438ee4eda0b : SUCCESS in 29m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/02cbf2ef1ae24d59858adab6fe298fe5 : SUCCESS in 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"ad6841436f7debdf768930c8540701b7243dcb90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 12:02:11.000000000","message":"Patch Set 73: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88b0def588074575b1d9093610e55cb7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6839ce9cda2845ab9e8c056b0ae32aca : SUCCESS in 2m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9973e47a0c274d39afa3802f310719bc : FAILURE in 30m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8da5d884a8fc4869a70a3ba5c248d951 : SUCCESS in 2h 04m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a47544f38c7347e4a63aeee4dba7d470 : SUCCESS in 32m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75d2c55dd72047748e6bc0c9bbbfe407 : SUCCESS in 1h 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183d6905cf70400998320c15729545ed : SUCCESS in 12m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8acf367e28c14229a48fe5159b12bb4b : SUCCESS in 18m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/676fa70145084d72bc3026d885b88d14 : SUCCESS in 14m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/928917304d6541809b14a75fdf77412e : SUCCESS in 23m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/652a18c5d5bf4cf695f59fe85255185c : SUCCESS in 12m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4805567c972f49b3a7a8bc95a11dc271 : SUCCESS in 1h 19m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0a573f55a48943a6845d9e1e55173f13 : SUCCESS in 1h 47m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/22db03fc69174f1dbbbb5080c888ff41 : FAILURE in 50m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ea9a9261f66425a94fcbac1ca4f6f06 : SUCCESS in 1h 07m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/056632ad558842d896de670b2cc05d96 : FAILURE in 51m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/11549e4ff8034706896df1f17e172a57 : FAILURE in 54m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/854745a0b04f4d3394ce123245da3924 : SUCCESS in 55m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/760f0ad7336c4d2d9d2a141353057e46 : FAILURE in 25m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf0a320a32054a1b9a0ef4b1a39dfd90 : SUCCESS in 1h 03m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ed6e31ca47a45c5bded54f191ef0a54 : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":73},{"id":"b2572ae2e59bc0ced776cb28a65ba133a5875470","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-15 14:35:32.000000000","message":"Patch Set 73: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/65/937765/73 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":73},{"id":"95e24688e4ffcafd5e5491e91953c02f9fc38cfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-04-23 15:30:09.000000000","message":"Uploaded patch set 74: Patch Set 73 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":74},{"id":"4738053884c017fda08c78db98a62440baa50c31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 16:31:13.000000000","message":"Patch Set 74:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f8ededbb09499ca77ae8d3f42e5db6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/64e6767c7e1945009953fb57c9c7ec46 : SUCCESS in 52m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d1b5dd8353e43fea11e975b7df244fd : SUCCESS in 48m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b226dad9d7b4a56b45f8e3d2dbc296a : SUCCESS in 42m 59s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"be7d40ce58dceda7018e233f982968b8ea0e6995","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-23 17:48:20.000000000","message":"Patch Set 74: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/65/937765/74 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":74},{"id":"cb1cbf551706a039a6d8e6384e719c0d5a8d666b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 18:05:37.000000000","message":"Patch Set 74: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c8fcbb871184ec18e2012d5955d5af2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd8404a302474e12bcddd9f76b0c26c3 : SUCCESS in 2m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c52b99ec307b4dde89a7a5e2508bddeb : FAILURE in 57m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ecaece4983d84a6d85173f7763f2bd62 : SUCCESS in 2h 30m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/adaacc17f4ce421d80b4804e0dfbc5b5 : SUCCESS in 44m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e65060c022b843d9aea240277c69ec5a : SUCCESS in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23492a62b0fd44679b6742fad1028ca3 : SUCCESS in 13m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/657eb298134c47758fb2b46e5adf7d5a : SUCCESS in 22m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c2f6dfc7c074c5f90f684f75140e5cc : SUCCESS in 20m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1bddd09870fe4a7da6c8cb0f5ddc7df5 : SUCCESS in 24m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c4501aea844835b5b7474f741c612b : SUCCESS in 11m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1df6cc5ce00f4391b11c69b01cb0748a : SUCCESS in 1h 33m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c591dc3b940d4d5b922fe2f188d8d783 : SUCCESS in 1h 55m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/32ad07866a824e9292d85eb1a3c374cc : FAILURE in 46m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc4b97f1a9684c1aad0d60070da19b15 : SUCCESS in 51m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ecff366fd3964199b2a6f0cd6df24c49 : FAILURE in 49m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0cdcdd309e8f4944bf614d47ceba08f8 : FAILURE in 51m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e093f448fb814ae2b02e3de9335aaf92 : SUCCESS in 56m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0799c1ff65784f209cb5f32f3fc6760a : FAILURE in 56m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/75197507e369454f9c91a71b5faef7ab : SUCCESS in 1h 23m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8c36b26cfc154c79acb2d480dcdf8acd : SUCCESS in 1h 08m 26s","accounts_in_message":[],"_revision_number":74},{"id":"09259ab0d21ed67e8f0a6546d340ebed5e844594","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-06-02 13:04:56.000000000","message":"Uploaded patch set 75: Patch Set 74 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":75},{"id":"28a12078d51f7590e12070934e1d6afdaedd993e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 14:10:33.000000000","message":"Patch Set 75:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72392cd6723d46baa2983825173f4637\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad13e5c5386c44d39f6809faf11f75d9 : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80025d6bd63f469dbfba23877319dfb0 : SUCCESS in 23m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed2ea30c668c45cab4b4e467c90ec4aa : SUCCESS in 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"14e6316be1dedf408e976f00aa84a4aa7403c9f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 15:36:58.000000000","message":"Patch Set 75: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea032a168315440fb90e5df1472cf608\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4c5cc3110bb4fa68ab4e990d43b8dec : SUCCESS in 4m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1225a66f550048be865c46930265868e : FAILURE in 49m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52de7e6a60394ba5bedacee6cbf845c9 : SUCCESS in 2h 28m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a3dcd7945bbb40cca645f570ac8a0e76 : SUCCESS in 37m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/669ee486ef7d470c92a4cbc71eb294c7 : SUCCESS in 40m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4608011a80f4e4586d489cc5a8931d3 : FAILURE in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85832f45b1bc4eb78c29d8fe2e46d78f : SUCCESS in 21m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/364ad415c44747479b197061cee07c70 : SUCCESS in 24m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8920638ea36047c8a05419ed9e531c61 : SUCCESS in 27m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c6941cedf045fa90c4d25d960bb2d5 : SUCCESS in 13m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/432be541cbd944cdb2a2f48f4aa8d003 : SUCCESS in 1h 15m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/13df7479b3a64deb970c23ae151415ff : SUCCESS in 1h 57m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7018b55cfd394f089cfc4191cd592c6b : FAILURE in 49m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae948d6299d14c1b8d0117d90380a85f : SUCCESS in 1h 03m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5a3bcbe894f04b6ba98331e00e9b6ba3 : FAILURE in 53m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/435f852ce8e84a7fafb4c2f450fcef00 : FAILURE in 48m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fafd5ccf3e424cf492be1d6ee7c71390 : SUCCESS in 59m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/44e699cd2f7b40a2a7076949c12ee86a : FAILURE in 45m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ec517a40f66a45f3a4ff8e97fc5913a3 : SUCCESS in 1h 10m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c2ea01108f784d0f866f22e59fc35003 : SUCCESS in 1h 27m 04s\n\nWarning:\n  Comments left for invalid file neutron/services/trunk/drivers/openvswitch/agent/ovsdb_handler.py","accounts_in_message":[],"_revision_number":75},{"id":"ba55608fa1f44d0c0ebc5d4f7be8a67d5c2cd9f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-06-04 08:17:57.000000000","message":"Uploaded patch set 76: Patch Set 75 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":76},{"id":"72a1820f9b4fab9d70c92f81c0c542bc8e36ac96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 10:25:12.000000000","message":"Patch Set 76:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52488ddcb0244709a7566558aca4759d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f692bb141bd4256b1d3502eda462eca : SUCCESS in 1h 08m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a032038b09d146438dea36d267282e4f : TIMED_OUT in 1h 21m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/53c4a941aaa94ae4ac8125ceed3906a0 : SUCCESS in 1h 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"66b5838b86757762af13834903d97bca307c8f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 10:59:50.000000000","message":"Patch Set 76: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cda4cfba334449ab1b54c5a646b8fcf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1947342bc6344b1b9ee5e71b88c8fcce : SUCCESS in 2m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/08637891ad9a47528e3dc605f83bd3fa : FAILURE in 1h 15m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/628f9bf48d1445cfbd08da0e3fa59491 : SUCCESS in 2h 11m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c95b6ab2590246cdad118157934194d7 : SUCCESS in 39m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee53db1981c04d2192a874b5562fc80c : SUCCESS in 44m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4600aac15a4f42abb987014b0b52753c : SUCCESS in 10m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5636a404270c49228143f311270546ce : SUCCESS in 23m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85018d6d6a6d4cd7b0480d6cfdaa0872 : SUCCESS in 18m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edda3e3a1baf40d58ae9790ad2efc6c5 : SUCCESS in 23m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390c23f896724046ab3f09b74a848f75 : SUCCESS in 11m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bbe60675b12042fea820caf6704315e4 : SUCCESS in 1h 21m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bc4975ee8232427a9f47a977c313c58f : SUCCESS in 2h 03m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e304dba929f412ba87bc77058c1621d : FAILURE in 49m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a656d80a4e5444194ca5b0bb96721a8 : SUCCESS in 1h 06m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/900b7cef27c549e9b088a75c8db6cd22 : FAILURE in 50m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c477182345c84ee2b6a84bc3b851ea5b : FAILURE in 52m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/97e86302a15e4fcc8aa5d8a6144b743b : SUCCESS in 1h 07m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dbb23ea8db0b43b9be63b03179fcad14 : FAILURE in 56m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1b5e659be11744938ab6d9f984708be7 : SUCCESS in 1h 17m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f592a5c678c5443f899d404f7b5975fe : SUCCESS in 1h 22m 19s","accounts_in_message":[],"_revision_number":76},{"id":"edd67666b4aac3e19ec80b06e8cffb3b7a929b71","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-07 18:40:36.000000000","message":"Patch Set 75: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/65/937765/75 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":75},{"id":"8b542b6efc66ab21d7f09221187dcefe562ec862","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-08 02:47:17.000000000","message":"Patch Set 76: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/65/937765/76 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":76},{"id":"1b8443882d751621e34002eccf42c0011e6214ee","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-16 06:48:35.000000000","message":"Patch Set 76: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":76},{"id":"a92fef97f5d00ac7268a7b9727a2c8da69b0b061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-16 09:08:07.000000000","message":"Uploaded patch set 77.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":77},{"id":"1e1afba0eaa08ec98654f97095c45b95235d80d0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-16 09:47:15.000000000","message":"Patch Set 77: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/65/937765/77 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":77},{"id":"ae5dfaae1109ebaf6683d5b57e3e25db1d1a5ac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 09:53:57.000000000","message":"Patch Set 77:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d1d774315374a0b8938de29fc304c29\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/603a47f4f12a4d1597a0a15862c62427 : SUCCESS in 20m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6eff5c69e119478bb37e45a1c467835b : SUCCESS in 27m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f045f28df0804eb48614aeeee8c26217 : SUCCESS in 44m 05s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"30dd78e78553400e872f31aa20bf924a4a9639af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 11:35:53.000000000","message":"Patch Set 77: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f02f9290cc14b6ca52c1bdff63e785e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2383d4933d7d410d9385fa55f984eefa : SUCCESS in 2m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b4375476244b4708ad824a28d042f1fc : SUCCESS in 1h 21m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d1ccf3c9a3324770b28a1e24e9d05822 : SUCCESS in 2h 23m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f6a954990fb54b328f57927ec1229c76 : SUCCESS in 45m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8141e46abe20481886487b857f453f29 : SUCCESS in 28m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b838fb5b1bdb485fbd54e39a2bd7a715 : FAILURE in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94e5ecf9c60543b5b17067c3e9b05a60 : SUCCESS in 17m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/529c049125ae4806b1e50433b2feea89 : SUCCESS in 16m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c445491554042a5ac461abbb10654d9 : SUCCESS in 30m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd84474cc2345e19b11f00cfc6077c6 : SUCCESS in 13m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/008e0ef2a8914ba0b85d25c0c5916516 : SUCCESS in 1h 43m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b4031eeca68d46438edf24bcc9cd1d09 : FAILURE in 2h 27m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4cec40e84c3a458ea3faed071036f2e4 : SUCCESS in 49m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/695984641c4143b087f1117685533c50 : SUCCESS in 1h 07m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/807746d8a62a44699fadf32e5277a530 : SUCCESS in 1h 44m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/130f9f9b8c914f9a8cee3109c8385128 : SUCCESS in 1h 28m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/73497391e86a440095a77fbf6e4d9a8b : SUCCESS in 59m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bd799119dc614e5883d108c405c2353b : SUCCESS in 51m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e649a412b6842e58cabad763db2993c : SUCCESS in 45m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7055c166b1a945328bb1983c8054aeaa : SUCCESS in 1h 15m 05s","accounts_in_message":[],"_revision_number":77},{"id":"9112483a2925457b769303bc735a1acdc775d2bd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-18 17:55:28.000000000","message":"Patch Set 77:\n\n(1 comment)","accounts_in_message":[],"_revision_number":77},{"id":"fb1df48f74fe64f7cf2d93b593254f3a2731c4a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-19 06:57:02.000000000","message":"Patch Set 78: Patch Set 77 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":78},{"id":"6b6d95a368b4cdc4b577f7f0676f51fe9a9e7f9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 07:42:57.000000000","message":"Patch Set 78:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/467d74505b4e40d7ac284ff16ad98ab0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38f76008364b4dc897849c0e364a6eaa : SUCCESS in 26m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aded8e28f7f742b8ad8767d12f4ba746 : SUCCESS in 23m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5874bc8d11e44bc2b8a3191fcedadcb9 : SUCCESS in 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"a992d486549d29a6821f0b2b9cb8bd584f53e2bf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-19 09:24:33.000000000","message":"Patch Set 78: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/65/937765/78 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":78},{"id":"e0ac0b3249af8281e4ffedca8053520f3ec98a6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 09:32:02.000000000","message":"Patch Set 78: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/922b9b67cb8440c6acb8627c5efd9563\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b16f78d17be64a7da0018e15add0e2d3 : SUCCESS in 2m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/64bae3eae0f84c88a309cb632313dddb : SUCCESS in 1h 43m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1c07afea8f334db4b2742f972ad10da3 : SUCCESS in 2h 32m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c6ce5b7d59d54af6bd2505e9511837ac : SUCCESS in 47m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/015c2a61b5334a74b61c35f8a061c23d : SUCCESS in 46m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee379d549e03447289c3c45a7113faf7 : FAILURE in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dc5977594004030bf4a368b3d319fa2 : SUCCESS in 22m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3a9abf7360643de80786e2744c371e0 : SUCCESS in 18m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/961d0ba08a7b4461a0e6e3680acc3fb8 : SUCCESS in 25m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277c7d20bf144168bb1eadea4c835632 : SUCCESS in 11m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3278126d4a934486b987de8d52ed8956 : SUCCESS in 1h 30m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/44e66ee4ff49428c965f9ec8759699f6 : FAILURE in 2h 15m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3302e33a4da248e2a53446385df3153a : SUCCESS in 1h 01m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3a3553086fa48c9b3101647731fded7 : SUCCESS in 58m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9f18f05038cc497fab05f3bae64108c0 : SUCCESS in 1h 49m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6bb9bfa1781148edaee38a0acdbf3adb : SUCCESS in 1h 26m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0e4db09d417848f7bb0400b377dc5e7b : SUCCESS in 58m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/83c0951285f04bd99be888622394f262 : SUCCESS in 1h 07m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e675cff8bc074cf4b722822871294660 : SUCCESS in 1h 15m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6fbd95fd629c454aaf8b2230388f45b6 : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":78},{"id":"1edbc610d58376ae8102bcfa08a56f607e15ce34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-25 11:21:00.000000000","message":"Patch Set 79: Patch Set 78 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":79},{"id":"7c5d22385b41faf7ea1f02744375f0e4c7fe9fb0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-25 13:31:19.000000000","message":"Patch Set 79: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/65/937765/79 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":79},{"id":"1357e0d7f5776520cdcca13a072a955f5eca3e81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 13:37:10.000000000","message":"Patch Set 79: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a0f020de950432ca05caf9fcf96e4bd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/10c64f35e0114eff8683820b837861ef : SUCCESS in 3m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7e908f5cd4d64d4095aabbd6fe0983df : FAILURE in 8m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b88067a07b76488f8fea29b8cffe5353 : SUCCESS in 2h 15m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/473d305d0bbd4a2593c06e144e07ef5f : SUCCESS in 44m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/856efaaaf84341fd868047ef521ba27a : SUCCESS in 46m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38510bff5034c4ca35a86e6e4ce5c04 : FAILURE in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a1f8c8c8b674a0d8ef84c47025605c6 : SUCCESS in 17m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eebff8a899bc4d9da82d6fa7439183af : SUCCESS in 21m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fce0c714f6e4237be2726693985b3a7 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e118a81cc564b71be3ec99906631bf1 : SUCCESS in 13m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8f069d0ec47e47db82c4b400e9524d37 : SUCCESS in 1h 22m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/208b10a894b540529a49319ead73c497 : SUCCESS in 58m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/89a0d9294d5c4278b4b15b4423d082a6 : SUCCESS in 1h 05m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cb11383fa71d475883cb2bfcf3695c20 : SUCCESS in 1h 51m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b7b68bb326c0437aa757b8fadc37074e : SUCCESS in 1h 31m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cd8a6e43e8584007b53e7f254a243032 : SUCCESS in 52m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/daf6a9ce80f246c5823acae1fd03d018 : FAILURE in 8m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43e2c290461247bf9ae214d52ee1b7e4 : SUCCESS in 1h 25m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6f6b19ce2ccc4575aaba23f8b97a2015 : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":79},{"id":"7deaa03f1a566ad7a7a28555b7af617d2f407286","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 14:30:19.000000000","message":"Patch Set 79:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba06356e06d44f518d7b1e85284c89b5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4cc58c7b3f7f4d378c78dbf8eec7dd93 : SUCCESS in 43m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1bcb016605244de38f75203a8e341403 : SUCCESS in 37m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7003517632148d1823f7609f21a6156 : SUCCESS in 52m 46s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"f40ab1af6a8c4dcf111313cbf7fbd3ccfb53bd74","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-26 11:24:30.000000000","message":"Patch Set 79:\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"5b51bbbec68a5cf1c72c9e57e087ca803c352f9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 12:57:17.000000000","message":"Patch Set 79:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a057927346f4bb88fe74cd7e8f39f83\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e7d8b9aae0844b88c433029d17426cc : SUCCESS in 36m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/77481152f15f46dab0ccd6df49d48d0d : SUCCESS in 51m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/66f58bf98dee4851b48242ca96a2e682 : SUCCESS in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"0ac946ce5788e513e6ee2870bd9992b5687bef5e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-26 13:06:20.000000000","message":"Patch Set 79:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/65/937765/79 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":79},{"id":"f89cf64b9bcbc78c3d5d19c1c2fdca69b51469f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 13:19:19.000000000","message":"Patch Set 79:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/000eedcb40ed4b0ebcc7d8c283f60611\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0ab701d1e69443c7bc6a81acfd9fd80f : SUCCESS in 2m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0158abdd38a74c77a281051c224fb3aa : SUCCESS in 1h 16m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cca81b78af6743bbb3194ca45ca0de91 : SUCCESS in 1h 51m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1a2e780d3fe417280660f296221fd60 : SUCCESS in 33m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c64e0bf7c686465487d9bb16dd10d505 : SUCCESS in 45m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5019a8accbc4cfcb078dd7053baf171 : FAILURE in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0ba77f6830c4afa9e74219ae29cbd7c : SUCCESS in 18m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a37195d464d84358bd2475b11b4a4ce6 : SUCCESS in 21m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abf0c2f3a411445aab0d3024c2f133c7 : SUCCESS in 26m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5379231c3a884dfa9a489331ab6df4ea : SUCCESS in 14m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3654362ed0c745948a4a987bd715b4ef : SUCCESS in 1h 41m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e94cd6592a2d400f8d5de90e25510e3a : SUCCESS in 38m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd3f55072499499ab320f230ec2b72c1 : SUCCESS in 1h 02m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6bf9c339d86f49b2a26c97906f8de39f : SUCCESS in 1h 41m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/25d7d45a6cd34e3aaf87b5892ee1dc92 : SUCCESS in 56m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b5248bacc7e34feb99634c71b3dc6b48 : SUCCESS in 1h 11m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ab4e87de561444c593d4fef044d6218e : FAILURE in 48m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2a51a4b26f564ec290d89c0018dda8ec : SUCCESS in 1h 19m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b64481a8d59d43298a402230f7e06c2d : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":79},{"id":"17886a34eb9d5d2b652bfec3dd23e6d1e9d48979","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-27 09:26:02.000000000","message":"Patch Set 79:\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"f996d7fdfce6648f3775e32522f55076a9f7dd1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-27 09:31:30.000000000","message":"Uploaded patch set 80.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":80},{"id":"8d7f53d295406a8414eae19251aa77d72fe7c4bf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-27 09:31:40.000000000","message":"Patch Set 80: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":80},{"id":"8bc8797f3d40f3af62bb31e7799f3bede574a579","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:23:21.000000000","message":"Patch Set 80:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc98b47f2635462fb2c979d54a45c292\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4fb72c080eb84ae3b3c138ad179ecba4 : SUCCESS in 22m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aec8b7296e794eea9aa309d7f18ddffc : SUCCESS in 37m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/699d34a97e3c4d6286b7bfcee6bb99a9 : SUCCESS in 42m 03s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"ce84cd8a98823558590c24181948035731639f2e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-27 11:04:49.000000000","message":"Patch Set 80: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/65/937765/80 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":80},{"id":"abc8386aca23109d2a7b035abc4300fb5ea18381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 11:43:33.000000000","message":"Patch Set 80: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b759c505ff774427a8d0b4923c3f7e8b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a9f0f42d85b3438281713ee2698c5c61 : SUCCESS in 2m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7c8d97fd8a524a5094c8331724519d0c : SUCCESS in 1h 25m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a7b55f39a5da4e5989eb9915e6d305d4 : SUCCESS in 1h 38m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b2bf5b2a4da04471bc2b77096054ea58 : SUCCESS in 24m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d822bc31380040f0896d355fa3a13a02 : SUCCESS in 45m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d59e89adc864e17a871827e412ffe8c : FAILURE in 9m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b87c129a39d341e0ac6c8ad970b7df00 : SUCCESS in 19m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9037a81cde10416a8edf1361b0e397f8 : SUCCESS in 20m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1b7c92915da4f25b0a7ea0a0f915489 : SUCCESS in 24m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e8afb7e0e3546f48800156a988a1b4f : SUCCESS in 13m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/100412879b194df2980b545b98aad971 : SUCCESS in 2h 09m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/55b10fc1af9442c084d5d217e2f6255d : SUCCESS in 57m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/69849d10a35348d6ae03af70b66ec490 : SUCCESS in 40m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1fc3a0904af54b3fb8e392c725cad7d9 : SUCCESS in 1h 53m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5bdbd0466921438684ff21c849523756 : SUCCESS in 1h 37m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/50fe6adcd1dc4dd28ead696ec73b01f2 : SUCCESS in 1h 08m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8a203e3c09f24086a84798ccbcb5469b : SUCCESS in 34m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/39845ed3dc8f4e4a87a8076a1e520f75 : SUCCESS in 1h 12m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a68c260530b34c08837d05d84e9cc4e0 : SUCCESS in 1h 22m 20s\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/ovs_oskenapp.py","accounts_in_message":[],"_revision_number":80},{"id":"bc0f11ef6837d4ed4fceeb876ebec7d4a9777b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-27 12:46:16.000000000","message":"Patch Set 81: Patch Set 80 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":81},{"id":"e8603e58e3e4af5a785bce84f7b79e291a96c5d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 13:35:25.000000000","message":"Patch Set 81:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a344881375504e95b5a3737e60c51d32\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f613dabdfd1f4a6585134fac43594d87 : SUCCESS in 39m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/520e454e1ac44fa4a56f5024207c6d1d : SUCCESS in 36m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cf4e3f6779a84613b579b0b402f222fa : SUCCESS in 45m 57s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"537d5ea2a90e57b717a6d5e30bdf7862083d6360","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-27 15:23:52.000000000","message":"Patch Set 81: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/65/937765/81 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":81},{"id":"8a741168f7527534810ca638265e5245ceede8fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 15:31:17.000000000","message":"Patch Set 81: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eac1de569a90494388dc182bbd76dd83\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/38adbdc8db984c1398fa075828d6adf9 : SUCCESS in 2m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/89c7d1c703424974acbef32cd750648c : SUCCESS in 1h 51m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b37e3f1cc4014390afd17fb27ae79048 : SUCCESS in 2h 41m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/981f8e2d3f8746e7aa6367aa8ccd3eb3 : SUCCESS in 41m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44c5c32aa6424cdba57455faea5bf136 : SUCCESS in 38m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3885c13573472c87e6b0cad21a772b : SUCCESS in 11m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3932e51d277d4fc58185e4e9952b45e2 : SUCCESS in 21m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c94d296961d4dbeb1f04dd8101a066e : SUCCESS in 19m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4952ca90016540aa8c314901b186681f : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944ca9887e4c435e83785334253ea5b4 : SUCCESS in 14m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6d39ed7aa0b54fe7b16be7994082429f : SUCCESS in 2h 03m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6de2804d128847d28cf3e4eda9870b0d : SUCCESS in 49m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8f2fb1621294d8baf6582cc319bbc84 : SUCCESS in 1h 05m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ee89224b6bb046fbb8f25a6d5c636b7b : SUCCESS in 1h 54m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e948698c58947a39099d262440a9dd4 : SUCCESS in 1h 22m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7727b70bbec049d5841cadb55d16358d : SUCCESS in 1h 03m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d02de59dfd6d434488973f590ac494f0 : SUCCESS in 1h 02m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2e0dc5245be04222980e3c2db5cbdf7c : SUCCESS in 41m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ce68526f9a50427db29358a8a7043fcb : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":81},{"id":"90397a80d152974e2e574be20d3624e8eb204e34","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-30 19:58:33.000000000","message":"Patch Set 81: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":81},{"id":"144f8ed26ef83c3541f26bbe335abed24935d6a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 19:58:59.000000000","message":"Patch Set 81: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":81},{"id":"a1c0139ae170c51368d620c3994378e5df71f0a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 21:52:02.000000000","message":"Patch Set 81: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d938e2f0bcd548099d6fca8d4fd56dbe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/885414e44c4b488b9f5ad5a1cbe6edd6 : SUCCESS in 2m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/47e2ba53f33144579a4121fa75705e66 : SUCCESS in 1h 26m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/68944618a6dd4ffb9b6023dace5978d7 : SUCCESS in 1h 43m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6744f0b8fe0b4060a5fcecb6dbd6a2a6 : SUCCESS in 9m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e69d9368c274da0b95121556a10397f : SUCCESS in 22m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83aca62a9df743b3bfb20b22f510c044 : SUCCESS in 21m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bec2f46b44040e383f282cc8891d1c3 : SUCCESS in 11m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/306a1232818d482a83bceaf6e68deb6d : SUCCESS in 1h 52m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/025c377f83f446d983e36bfc92e891ed : SUCCESS in 1h 01m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3d8b372877a4b279c4b050eab695f55 : SUCCESS in 1h 13m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f49acee573e4bfcb0c7fc47041d57be : SUCCESS in 52m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b9eaf84ddb5f4591bccc1952ebe8c35d : SUCCESS in 57m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/88c23515850f4858952e509f717e937d : SUCCESS in 1h 23m 54s","accounts_in_message":[],"_revision_number":81},{"id":"9901e5a92e2fa1f0b428ad3851ab09c2ff2bbbf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 21:52:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":81},{"id":"55d8a016400365367ac0587a85456529006f9134","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 21:53:08.000000000","message":"Patch Set 81:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df93a7c279814d08b93946bc59715161\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d87bd9840474b95a9f58a5a165e06d9 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":81},{"id":"821183c5d91f090e76450a6d9c776594076b64b4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-10 15:40:07.000000000","message":"Patch Set 81:\n\n(1 comment)","accounts_in_message":[],"_revision_number":81},{"id":"9cbe3df6fb105853d1e2c87e789e4bc2eba2ca59","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-10 15:44:23.000000000","message":"Patch Set 81:\n\n(1 comment)","accounts_in_message":[],"_revision_number":81}],"current_revision_number":81,"current_revision":"ad927935ab5868486824c02c478664194e474631","revisions":{"6b996075be9355933225c8b32a45590ae4ff622d":{"kind":"REWORK","_number":1,"created":"2024-12-16 10:44:03.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/1"}}},"commit":{"parents":[{"commit":"79067358f4245d1e72ac4a9cfeb9f04189f77eb7","subject":"Merge \"[OVN] Improve initial hash ring setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79067358f4245d1e72ac4a9cfeb9f04189f77eb7"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:42:45.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b996075be9355933225c8b32a45590ae4ff622d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b996075be9355933225c8b32a45590ae4ff622d"}]},"branch":"refs/heads/master"},"7169e61f0c5f8af1cc4bf8dcf6e8838006afe3eb":{"kind":"REWORK","_number":2,"created":"2024-12-16 10:59:04.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/2"}}},"commit":{"parents":[{"commit":"79067358f4245d1e72ac4a9cfeb9f04189f77eb7","subject":"Merge \"[OVN] Improve initial hash ring setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79067358f4245d1e72ac4a9cfeb9f04189f77eb7"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:58:56.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7169e61f0c5f8af1cc4bf8dcf6e8838006afe3eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7169e61f0c5f8af1cc4bf8dcf6e8838006afe3eb"}]},"branch":"refs/heads/master"},"8b8bbe930be722833e4f36ff4516087b6aa94cb7":{"kind":"REWORK","_number":3,"created":"2024-12-16 16:23:31.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/3"}}},"commit":{"parents":[{"commit":"79067358f4245d1e72ac4a9cfeb9f04189f77eb7","subject":"Merge \"[OVN] Improve initial hash ring setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79067358f4245d1e72ac4a9cfeb9f04189f77eb7"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 16:23:21.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b8bbe930be722833e4f36ff4516087b6aa94cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b8bbe930be722833e4f36ff4516087b6aa94cb7"}]},"branch":"refs/heads/master"},"1d9468640670d7e2ac9f08a864f60da5a4a39a5d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-12-17 08:15:47.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/4"}}},"commit":{"parents":[{"commit":"c4f73542b7559c3cae1cf9f5e0e2fca8bd7c2d83","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4f73542b7559c3cae1cf9f5e0e2fca8bd7c2d83"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-17 08:15:17.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d9468640670d7e2ac9f08a864f60da5a4a39a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d9468640670d7e2ac9f08a864f60da5a4a39a5d"}]},"branch":"refs/heads/master"},"2fa5a08274ef5738bd00393e857c92facb6ac2be":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-12-17 19:58:16.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/5"}}},"commit":{"parents":[{"commit":"78ff851e714499910e9d9de8053361ecdcecc9fc","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78ff851e714499910e9d9de8053361ecdcecc9fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-17 19:58:04.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2fa5a08274ef5738bd00393e857c92facb6ac2be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2fa5a08274ef5738bd00393e857c92facb6ac2be"}]},"branch":"refs/heads/master"},"f6531d2ecaf93f104db56e7f76718a7b6182c525":{"kind":"REWORK","_number":6,"created":"2024-12-18 09:40:08.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/6"}}},"commit":{"parents":[{"commit":"78ff851e714499910e9d9de8053361ecdcecc9fc","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78ff851e714499910e9d9de8053361ecdcecc9fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-18 09:35:03.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6531d2ecaf93f104db56e7f76718a7b6182c525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6531d2ecaf93f104db56e7f76718a7b6182c525"}]},"branch":"refs/heads/master"},"a5a13f2e10c230766f11e1d7cb8252139af61116":{"kind":"REWORK","_number":7,"created":"2024-12-18 15:02:51.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/7"}}},"commit":{"parents":[{"commit":"c2e89b089175010bd63590a583212754f83b61d8","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e89b089175010bd63590a583212754f83b61d8"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-18 15:02:34.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5a13f2e10c230766f11e1d7cb8252139af61116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5a13f2e10c230766f11e1d7cb8252139af61116"}]},"branch":"refs/heads/master"},"5bbe9e0d76aa141d529d28d8abe996f9969242ec":{"kind":"REWORK","_number":8,"created":"2024-12-23 10:14:17.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/8"}}},"commit":{"parents":[{"commit":"dc1bcb08f7efda36bf0aa2a6194bfaa8bed37efb","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc1bcb08f7efda36bf0aa2a6194bfaa8bed37efb"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-23 10:14:04.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bbe9e0d76aa141d529d28d8abe996f9969242ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bbe9e0d76aa141d529d28d8abe996f9969242ec"}]},"branch":"refs/heads/master"},"8f0796602723117db38007ab141e539972dc378d":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2024-12-31 09:09:05.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/9"}}},"commit":{"parents":[{"commit":"a23c94e7d593831e999d576e1ff59f161f0baec5","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a23c94e7d593831e999d576e1ff59f161f0baec5"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-31 09:02:55.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f0796602723117db38007ab141e539972dc378d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f0796602723117db38007ab141e539972dc378d"}]},"branch":"refs/heads/master"},"a0b05fab8d771e84129c797487a0e0c9e6cfd531":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-01-06 14:52:07.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/10"}}},"commit":{"parents":[{"commit":"90523bf705dbabdfa6d8a09dcece57ac024ad59d","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90523bf705dbabdfa6d8a09dcece57ac024ad59d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-06 14:51:07.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0b05fab8d771e84129c797487a0e0c9e6cfd531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0b05fab8d771e84129c797487a0e0c9e6cfd531"}]},"branch":"refs/heads/master"},"1c40853d14240c0b609e252f1e001bffcefd0f37":{"kind":"REWORK","_number":11,"created":"2025-01-06 17:32:08.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/11"}}},"commit":{"parents":[{"commit":"90523bf705dbabdfa6d8a09dcece57ac024ad59d","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90523bf705dbabdfa6d8a09dcece57ac024ad59d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-06 17:31:55.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c40853d14240c0b609e252f1e001bffcefd0f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c40853d14240c0b609e252f1e001bffcefd0f37"}]},"branch":"refs/heads/master"},"d017e94c90307fcf8a46f5ab1e06daa29d454991":{"kind":"REWORK","_number":12,"created":"2025-01-07 09:39:31.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/12"}}},"commit":{"parents":[{"commit":"1a1b1a41cae7cce25fbd94ecbc4987075c6ddd58","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a1b1a41cae7cce25fbd94ecbc4987075c6ddd58"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-07 09:39:20.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d017e94c90307fcf8a46f5ab1e06daa29d454991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d017e94c90307fcf8a46f5ab1e06daa29d454991"}]},"branch":"refs/heads/master"},"73bf9428aad19c983ec115ecba20d2d67db213c5":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-01-07 13:56:38.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/13"}}},"commit":{"parents":[{"commit":"e2a4a87fa3e8deef9b53bc64b94c080ccf64befa","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2a4a87fa3e8deef9b53bc64b94c080ccf64befa"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-07 13:56:28.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73bf9428aad19c983ec115ecba20d2d67db213c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73bf9428aad19c983ec115ecba20d2d67db213c5"}]},"branch":"refs/heads/master"},"b246414c7e3600587a38b19d2673c17b9784f5ab":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-01-07 14:03:05.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/14"}}},"commit":{"parents":[{"commit":"daabd5d6616b77fe89b6be1a62a8910f89ec754e","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daabd5d6616b77fe89b6be1a62a8910f89ec754e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-07 14:02:55.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b246414c7e3600587a38b19d2673c17b9784f5ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b246414c7e3600587a38b19d2673c17b9784f5ab"}]},"branch":"refs/heads/master"},"16fd0b7f97fa32bd549814d572f9c2c144d6ec38":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-01-08 09:46:48.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/15"}}},"commit":{"parents":[{"commit":"5eb7560274751dfaaedb45cfd5430f1b65dd72d8","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5eb7560274751dfaaedb45cfd5430f1b65dd72d8"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-08 09:37:24.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16fd0b7f97fa32bd549814d572f9c2c144d6ec38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16fd0b7f97fa32bd549814d572f9c2c144d6ec38"}]},"branch":"refs/heads/master"},"a7c13d3ab4da261e39ef6a2933007df2d2f397da":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-01-08 16:52:34.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/16"}}},"commit":{"parents":[{"commit":"e892d2fa24652bf2c8c542fa0b165e744753cf36","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e892d2fa24652bf2c8c542fa0b165e744753cf36"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-08 16:52:26.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7c13d3ab4da261e39ef6a2933007df2d2f397da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7c13d3ab4da261e39ef6a2933007df2d2f397da"}]},"branch":"refs/heads/master"},"4da83d212949f360d0ac590755bbc997f813309b":{"kind":"REWORK","_number":17,"created":"2025-01-09 09:33:52.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/17"}}},"commit":{"parents":[{"commit":"e892d2fa24652bf2c8c542fa0b165e744753cf36","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e892d2fa24652bf2c8c542fa0b165e744753cf36"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-09 09:33:34.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4da83d212949f360d0ac590755bbc997f813309b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4da83d212949f360d0ac590755bbc997f813309b"}]},"branch":"refs/heads/master"},"4336b264e6c29f61bbec672b108596718981bba4":{"kind":"REWORK","_number":18,"created":"2025-01-09 12:35:20.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/18"}}},"commit":{"parents":[{"commit":"ded8f88083dd6d9bcbc186fa72cbc5dd2d020bbf","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ded8f88083dd6d9bcbc186fa72cbc5dd2d020bbf"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-09 12:35:10.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4336b264e6c29f61bbec672b108596718981bba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4336b264e6c29f61bbec672b108596718981bba4"}]},"branch":"refs/heads/master"},"d1087b3e8dc3b2c07a465d8779ae67846f418679":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-01-10 10:28:29.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/19"}}},"commit":{"parents":[{"commit":"e420ab619c425fb9e368f669d52e697da68e0737","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e420ab619c425fb9e368f669d52e697da68e0737"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-10 10:28:21.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1087b3e8dc3b2c07a465d8779ae67846f418679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1087b3e8dc3b2c07a465d8779ae67846f418679"}]},"branch":"refs/heads/master"},"a782c194f22c35ed59f00d10250e8e1a1bd2af1e":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2025-01-10 16:12:59.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/20"}}},"commit":{"parents":[{"commit":"b8227b358512102b524045156998d9c2973e6eaf","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8227b358512102b524045156998d9c2973e6eaf"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-10 16:12:51.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a782c194f22c35ed59f00d10250e8e1a1bd2af1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a782c194f22c35ed59f00d10250e8e1a1bd2af1e"}]},"branch":"refs/heads/master"},"f36311b388ad25cc0b432b01d5caab3ab3448dd0":{"kind":"REWORK","_number":21,"created":"2025-01-13 09:29:10.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/21"}}},"commit":{"parents":[{"commit":"b8227b358512102b524045156998d9c2973e6eaf","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8227b358512102b524045156998d9c2973e6eaf"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-13 09:22:46.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f36311b388ad25cc0b432b01d5caab3ab3448dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f36311b388ad25cc0b432b01d5caab3ab3448dd0"}]},"branch":"refs/heads/master"},"24e8eede7f00d1018ca705f5e8414618e131e4e1":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2025-01-15 08:09:58.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/22"}}},"commit":{"parents":[{"commit":"e3e5a8eeea6411a699fa638ddabe77f271a995b3","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3e5a8eeea6411a699fa638ddabe77f271a995b3"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-15 08:09:34.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24e8eede7f00d1018ca705f5e8414618e131e4e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24e8eede7f00d1018ca705f5e8414618e131e4e1"}]},"branch":"refs/heads/master"},"35f9ca663ce108168e17a667872a19d4a5451ba2":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2025-01-15 08:13:19.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/23"}}},"commit":{"parents":[{"commit":"3d104047a818fbb415bb5bd71a1580a70ee48cda","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d104047a818fbb415bb5bd71a1580a70ee48cda"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-15 08:10:47.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35f9ca663ce108168e17a667872a19d4a5451ba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35f9ca663ce108168e17a667872a19d4a5451ba2"}]},"branch":"refs/heads/master"},"d5a55561edf12d73485568f4b8bef3f3965fd4e5":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2025-01-15 11:36:48.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/24"}}},"commit":{"parents":[{"commit":"a02f4ce52a29c239d5ee2cc0539f1f8f87f3a65b","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a02f4ce52a29c239d5ee2cc0539f1f8f87f3a65b"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-15 11:36:40.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5a55561edf12d73485568f4b8bef3f3965fd4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5a55561edf12d73485568f4b8bef3f3965fd4e5"}]},"branch":"refs/heads/master"},"b22494152dd70433ec3adbe9b5b85e36d429ba9a":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2025-01-15 13:53:18.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/25"}}},"commit":{"parents":[{"commit":"9839b515a4298e3d97a9ada8fc121f7cbd853043","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9839b515a4298e3d97a9ada8fc121f7cbd853043"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-15 13:52:40.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b22494152dd70433ec3adbe9b5b85e36d429ba9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b22494152dd70433ec3adbe9b5b85e36d429ba9a"}]},"branch":"refs/heads/master"},"8af590373a316cfdcaa913cf9ebef7e307d0ec4f":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-01-16 12:27:37.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/26"}}},"commit":{"parents":[{"commit":"25cf07b38bbbc4a83d05ee6d10e47aae64b2b52c","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25cf07b38bbbc4a83d05ee6d10e47aae64b2b52c"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-16 12:27:28.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8af590373a316cfdcaa913cf9ebef7e307d0ec4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8af590373a316cfdcaa913cf9ebef7e307d0ec4f"}]},"branch":"refs/heads/master"},"dbef40ef0a74d5b9bd731cf4235b2c2b6a95b9f5":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-01-17 08:08:21.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/27"}}},"commit":{"parents":[{"commit":"8da8b8b7a45388f9986475fae2b751fbc3b630f2","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8da8b8b7a45388f9986475fae2b751fbc3b630f2"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-17 08:08:11.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbef40ef0a74d5b9bd731cf4235b2c2b6a95b9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbef40ef0a74d5b9bd731cf4235b2c2b6a95b9f5"}]},"branch":"refs/heads/master"},"cee960c1284390a78a49f870b9e7f878d879d41a":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2025-01-20 12:06:50.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/28"}}},"commit":{"parents":[{"commit":"5012ca33b48e605421025b32a8a01b6e2b05aca0","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5012ca33b48e605421025b32a8a01b6e2b05aca0"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-20 12:06:41.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cee960c1284390a78a49f870b9e7f878d879d41a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cee960c1284390a78a49f870b9e7f878d879d41a"}]},"branch":"refs/heads/master"},"4c23441f46b45e8420e1eb1a94fb4f469c44a0c7":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2025-01-20 16:09:15.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/29"}}},"commit":{"parents":[{"commit":"d56fe22e3ae5a73b61b11c5112e9c28e178a7fa0","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d56fe22e3ae5a73b61b11c5112e9c28e178a7fa0"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-20 16:09:02.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c23441f46b45e8420e1eb1a94fb4f469c44a0c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c23441f46b45e8420e1eb1a94fb4f469c44a0c7"}]},"branch":"refs/heads/master"},"f33eacc64a673bd3e81c9c8c80e67b34079e0696":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2025-01-21 07:40:39.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/30"}}},"commit":{"parents":[{"commit":"3500dec2746bb36fbe8fca2e726f89f15d6007cf","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3500dec2746bb36fbe8fca2e726f89f15d6007cf"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-21 07:40:22.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f33eacc64a673bd3e81c9c8c80e67b34079e0696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f33eacc64a673bd3e81c9c8c80e67b34079e0696"}]},"branch":"refs/heads/master"},"0217fa4ca6fb2badf6fe9f2f319e7ee7e6ecc81f":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2025-01-21 14:11:17.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/31"}}},"commit":{"parents":[{"commit":"bf637f36c59f31669dce20118488b7ed2f4bbe26","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf637f36c59f31669dce20118488b7ed2f4bbe26"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-21 14:11:05.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0217fa4ca6fb2badf6fe9f2f319e7ee7e6ecc81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0217fa4ca6fb2badf6fe9f2f319e7ee7e6ecc81f"}]},"branch":"refs/heads/master"},"56b3a2a6efebc13176bcdd95bc573c81db06a887":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2025-01-22 07:44:35.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/32"}}},"commit":{"parents":[{"commit":"6265e80bea6eb93fc4544405fa7bcdd75e41bc0e","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6265e80bea6eb93fc4544405fa7bcdd75e41bc0e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-22 07:44:00.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56b3a2a6efebc13176bcdd95bc573c81db06a887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56b3a2a6efebc13176bcdd95bc573c81db06a887"}]},"branch":"refs/heads/master"},"bda4d154f4fb630e00ee0e885708482cdb3b1409":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2025-01-22 12:50:52.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/33"}}},"commit":{"parents":[{"commit":"1df6823e8847aa07c9f47917f8c075bd496a3408","subject":"common: fix wait_until_true to support native thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1df6823e8847aa07c9f47917f8c075bd496a3408"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-22 12:50:43.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bda4d154f4fb630e00ee0e885708482cdb3b1409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bda4d154f4fb630e00ee0e885708482cdb3b1409"}]},"branch":"refs/heads/master"},"9735850c3dcfa21499e9e3e1e1481d1baeb16b95":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2025-01-23 15:42:04.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/34"}}},"commit":{"parents":[{"commit":"49730fe8c757ac47110b932eb28654d44421700a","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49730fe8c757ac47110b932eb28654d44421700a"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-23 15:41:55.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9735850c3dcfa21499e9e3e1e1481d1baeb16b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9735850c3dcfa21499e9e3e1e1481d1baeb16b95"}]},"branch":"refs/heads/master"},"4fa9ed875ee916164ad55d19fcc162b628dd1e24":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2025-01-24 08:20:19.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/35"}}},"commit":{"parents":[{"commit":"9bbcc81e16f110bbbd5bd90214ceeb511dff12fa","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9bbcc81e16f110bbbd5bd90214ceeb511dff12fa"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-24 08:20:10.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4fa9ed875ee916164ad55d19fcc162b628dd1e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4fa9ed875ee916164ad55d19fcc162b628dd1e24"}]},"branch":"refs/heads/master"},"b2fed8d20808d27d65ded072c2fccb91731e1a84":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2025-01-24 15:07:33.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/36"}}},"commit":{"parents":[{"commit":"b873d3f9af8ddbf8a44c97d326a84263e2f7d48b","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b873d3f9af8ddbf8a44c97d326a84263e2f7d48b"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-24 15:07:24.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2fed8d20808d27d65ded072c2fccb91731e1a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2fed8d20808d27d65ded072c2fccb91731e1a84"}]},"branch":"refs/heads/master"},"39798c186888b5d7388329d7e7e6e51f1fabd41d":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2025-01-24 18:16:54.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/37"}}},"commit":{"parents":[{"commit":"2d8ace37565090486483a1ff38dcb8b9ed54124e","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d8ace37565090486483a1ff38dcb8b9ed54124e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-24 18:16:44.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39798c186888b5d7388329d7e7e6e51f1fabd41d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39798c186888b5d7388329d7e7e6e51f1fabd41d"}]},"branch":"refs/heads/master"},"d43cc177d777ab9e4186418d7aa7d9d763bce52d":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2025-01-26 21:08:18.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/38"}}},"commit":{"parents":[{"commit":"e95bd191af74221a5acc43932e511029d9ed3d05","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e95bd191af74221a5acc43932e511029d9ed3d05"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-26 21:07:56.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d43cc177d777ab9e4186418d7aa7d9d763bce52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d43cc177d777ab9e4186418d7aa7d9d763bce52d"}]},"branch":"refs/heads/master"},"0c369db8b800587c404a29f74308ded79adb38cb":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2025-01-27 07:56:29.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/39"}}},"commit":{"parents":[{"commit":"3d59699b189a27aa21d620578d473887f8a191c0","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d59699b189a27aa21d620578d473887f8a191c0"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-27 07:11:35.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c369db8b800587c404a29f74308ded79adb38cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c369db8b800587c404a29f74308ded79adb38cb"}]},"branch":"refs/heads/master"},"6f95a53a2164e43af30e35f64c04a3f0d21cf0e3":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2025-01-27 09:17:57.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/40"}}},"commit":{"parents":[{"commit":"7b836cf9bcc00be741fa1166bdcd2cc084489dab","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b836cf9bcc00be741fa1166bdcd2cc084489dab"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-27 09:17:47.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f95a53a2164e43af30e35f64c04a3f0d21cf0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f95a53a2164e43af30e35f64c04a3f0d21cf0e3"}]},"branch":"refs/heads/master"},"238753d9ca3366f2538a80ef8b99fbec850688a4":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2025-01-28 19:27:02.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/41"}}},"commit":{"parents":[{"commit":"6fc84cb9649f113831b5e8f4e4d339114794173c","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fc84cb9649f113831b5e8f4e4d339114794173c"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-28 19:22:23.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/238753d9ca3366f2538a80ef8b99fbec850688a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/238753d9ca3366f2538a80ef8b99fbec850688a4"}]},"branch":"refs/heads/master"},"19033b1544d3eeff185707d80d081c600bcc867b":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2025-01-29 17:09:25.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/42"}}},"commit":{"parents":[{"commit":"236c75bd0a78313d99469764e6f7497f80867d53","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/236c75bd0a78313d99469764e6f7497f80867d53"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-29 17:09:13.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19033b1544d3eeff185707d80d081c600bcc867b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19033b1544d3eeff185707d80d081c600bcc867b"}]},"branch":"refs/heads/master"},"4980aba759bee2a9c76d8082b76f40c95c29859d":{"kind":"REWORK","_number":43,"created":"2025-01-30 10:50:52.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/43"}}},"commit":{"parents":[{"commit":"875da51e365b60367db948d637dfd6d811ded10f","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/875da51e365b60367db948d637dfd6d811ded10f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-30 10:50:38.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4980aba759bee2a9c76d8082b76f40c95c29859d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4980aba759bee2a9c76d8082b76f40c95c29859d"}]},"branch":"refs/heads/master"},"858cc373cba0577c4bf67ee51901f2ed2c68dc5f":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2025-01-30 15:44:06.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/44"}}},"commit":{"parents":[{"commit":"610d63c52348c13fe3e1bbc13659f79ed829bc07","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/610d63c52348c13fe3e1bbc13659f79ed829bc07"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-30 15:43:58.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/858cc373cba0577c4bf67ee51901f2ed2c68dc5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/858cc373cba0577c4bf67ee51901f2ed2c68dc5f"}]},"branch":"refs/heads/master"},"74b6168d6ece0da2c813492d60d46546b7d8119b":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2025-01-31 08:26:26.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/45"}}},"commit":{"parents":[{"commit":"dd7a0a9b2cc261a0fa394035da46579d51f2d0a7","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd7a0a9b2cc261a0fa394035da46579d51f2d0a7"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-31 08:26:12.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74b6168d6ece0da2c813492d60d46546b7d8119b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74b6168d6ece0da2c813492d60d46546b7d8119b"}]},"branch":"refs/heads/master"},"04f409f9af8d949ed157d71935da484abf15d6e6":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2025-01-31 13:02:12.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/46"}}},"commit":{"parents":[{"commit":"6acc874459f75700cbe0d9ccaf7d1a6cc792be0e","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6acc874459f75700cbe0d9ccaf7d1a6cc792be0e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-31 10:54:37.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04f409f9af8d949ed157d71935da484abf15d6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04f409f9af8d949ed157d71935da484abf15d6e6"}]},"branch":"refs/heads/master"},"1216aa1b611bcf2222982c6841ba766edb5e5a28":{"kind":"NO_CODE_CHANGE","_number":47,"created":"2025-01-31 16:13:37.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/47"}}},"commit":{"parents":[{"commit":"6acc874459f75700cbe0d9ccaf7d1a6cc792be0e","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6acc874459f75700cbe0d9ccaf7d1a6cc792be0e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-01-31 16:13:37.000000000","tz":0},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1216aa1b611bcf2222982c6841ba766edb5e5a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1216aa1b611bcf2222982c6841ba766edb5e5a28"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3d7101947c2f8cd0e988ed06398afa2a8410f162":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":48,"created":"2025-02-04 10:07:07.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/48"}}},"commit":{"parents":[{"commit":"ee107fa2308c6f37523c7b60442f863686589455","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee107fa2308c6f37523c7b60442f863686589455"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-04 10:06:48.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d7101947c2f8cd0e988ed06398afa2a8410f162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d7101947c2f8cd0e988ed06398afa2a8410f162"}]},"branch":"refs/heads/master"},"f30b026c11a768cc88eee4e8213cb336143b1e87":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2025-02-04 10:12:28.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/49"}}},"commit":{"parents":[{"commit":"2d7b67db97814ff27a1f1cc0527eb9ed50baa0a3","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d7b67db97814ff27a1f1cc0527eb9ed50baa0a3"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-04 10:12:18.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f30b026c11a768cc88eee4e8213cb336143b1e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f30b026c11a768cc88eee4e8213cb336143b1e87"}]},"branch":"refs/heads/master"},"e94c6751216de86165efd8a2b79b8f25277e9168":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2025-02-05 09:21:08.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/50"}}},"commit":{"parents":[{"commit":"3e6a3186493867133074e8a6311458a8d5fdb844","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e6a3186493867133074e8a6311458a8d5fdb844"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-05 09:20:58.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e94c6751216de86165efd8a2b79b8f25277e9168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e94c6751216de86165efd8a2b79b8f25277e9168"}]},"branch":"refs/heads/master"},"f7ebed8e4790f47e2c2501fc68e4ddd10985c1e9":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2025-02-06 10:23:02.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/51"}}},"commit":{"parents":[{"commit":"6e03adfe5458583ea78a0613169d9a1b80f15cc8","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e03adfe5458583ea78a0613169d9a1b80f15cc8"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-06 10:22:46.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7ebed8e4790f47e2c2501fc68e4ddd10985c1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7ebed8e4790f47e2c2501fc68e4ddd10985c1e9"}]},"branch":"refs/heads/master"},"1762f805c6297dea781e8628f5bb3c6a0d0bbd5d":{"kind":"REWORK","_number":52,"created":"2025-02-06 14:34:17.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/52"}}},"commit":{"parents":[{"commit":"6e03adfe5458583ea78a0613169d9a1b80f15cc8","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e03adfe5458583ea78a0613169d9a1b80f15cc8"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-06 14:34:02.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1762f805c6297dea781e8628f5bb3c6a0d0bbd5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1762f805c6297dea781e8628f5bb3c6a0d0bbd5d"}]},"branch":"refs/heads/master"},"7ffaf327a92b5022846cdb3c6c583726d9ff361a":{"kind":"REWORK","_number":53,"created":"2025-02-06 16:13:19.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/53"}}},"commit":{"parents":[{"commit":"517a766dc97615c11ccd8904634ded1f30ba4177","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/517a766dc97615c11ccd8904634ded1f30ba4177"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-12-16 10:41:10.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-06 16:12:55.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ffaf327a92b5022846cdb3c6c583726d9ff361a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ffaf327a92b5022846cdb3c6c583726d9ff361a"}]},"branch":"refs/heads/master"},"0f0e7284b5f575b767f75b48f4f0cd4de1de4810":{"kind":"REWORK","_number":54,"created":"2025-02-07 16:03:51.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/54"}}},"commit":{"parents":[{"commit":"f1bcc0e6bd3b66deacb025ca04f7340cd6b8f1b9","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1bcc0e6bd3b66deacb025ca04f7340cd6b8f1b9"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 16:03:38.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f0e7284b5f575b767f75b48f4f0cd4de1de4810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f0e7284b5f575b767f75b48f4f0cd4de1de4810"}]},"branch":"refs/heads/master"},"9242a1399952c12d434af9b3b2daf067bde66642":{"kind":"NO_CHANGE","_number":55,"created":"2025-02-10 09:17:56.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/55"}}},"commit":{"parents":[{"commit":"137799ec1ee7bf41847c2c5f91b24354bf035067","subject":"ofswitch: remove usage of eventlet timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/137799ec1ee7bf41847c2c5f91b24354bf035067"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-10 09:17:45.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9242a1399952c12d434af9b3b2daf067bde66642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9242a1399952c12d434af9b3b2daf067bde66642"}]},"branch":"refs/heads/master"},"b73c0c68ec1d3dafd60ced18240922b67e568ae2":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2025-02-10 13:00:27.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/56"}}},"commit":{"parents":[{"commit":"764f865e25fbd4d4572c942705e53fa2a915286e","subject":"ofswitch: remove usage of eventlet timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/764f865e25fbd4d4572c942705e53fa2a915286e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-10 13:00:17.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b73c0c68ec1d3dafd60ced18240922b67e568ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b73c0c68ec1d3dafd60ced18240922b67e568ae2"}]},"branch":"refs/heads/master"},"9544ef548516c9b73abf9aa8a42278e6bfb8722b":{"kind":"REWORK","_number":57,"created":"2025-02-10 19:30:17.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/57"}}},"commit":{"parents":[{"commit":"764f865e25fbd4d4572c942705e53fa2a915286e","subject":"ofswitch: remove usage of eventlet timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/764f865e25fbd4d4572c942705e53fa2a915286e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-10 19:30:06.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9544ef548516c9b73abf9aa8a42278e6bfb8722b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9544ef548516c9b73abf9aa8a42278e6bfb8722b"}]},"branch":"refs/heads/master"},"83920e0cfc516eabae7685aaeac2aaf4550d796f":{"kind":"TRIVIAL_REBASE","_number":58,"created":"2025-02-12 14:14:28.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/58"}}},"commit":{"parents":[{"commit":"d380df7c3ea9c7fe488494b1f8742caa43f2d96d","subject":"ovs: remove usage of loopingcall.FixedIntervalLoopingCall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d380df7c3ea9c7fe488494b1f8742caa43f2d96d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-12 14:14:15.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83920e0cfc516eabae7685aaeac2aaf4550d796f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83920e0cfc516eabae7685aaeac2aaf4550d796f"}]},"branch":"refs/heads/master"},"d4096237b14f376eae25de1e3835c8466bd3e7cc":{"kind":"TRIVIAL_REBASE","_number":59,"created":"2025-02-12 17:48:18.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/59"}}},"commit":{"parents":[{"commit":"d128eaf862e1bc9efbc5ea29a1312d691d6b23aa","subject":"ovs: remove usage of loopingcall.FixedIntervalLoopingCall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d128eaf862e1bc9efbc5ea29a1312d691d6b23aa"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-12 17:48:05.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/938337\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4096237b14f376eae25de1e3835c8466bd3e7cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4096237b14f376eae25de1e3835c8466bd3e7cc"}]},"branch":"refs/heads/master"},"bc4a1dd2f2f92b65efb9cb297fdafda601c2e7fe":{"kind":"REWORK","_number":60,"created":"2025-02-15 16:40:44.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/60"}}},"commit":{"parents":[{"commit":"4e82e88e0ff7c598eb383c19dc708518ed343a2f","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e82e88e0ff7c598eb383c19dc708518ed343a2f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-15 16:40:36.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc4a1dd2f2f92b65efb9cb297fdafda601c2e7fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc4a1dd2f2f92b65efb9cb297fdafda601c2e7fe"}]},"branch":"refs/heads/master"},"25d91ecef07f8e5799e293b763f0cf5f0ae10ed8":{"kind":"TRIVIAL_REBASE","_number":61,"created":"2025-02-16 11:25:52.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/61"}}},"commit":{"parents":[{"commit":"5b0ec58c76aee42fe58b85a5a1cf6ad8eadeb5e1","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b0ec58c76aee42fe58b85a5a1cf6ad8eadeb5e1"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-16 11:25:35.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25d91ecef07f8e5799e293b763f0cf5f0ae10ed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25d91ecef07f8e5799e293b763f0cf5f0ae10ed8"}]},"branch":"refs/heads/master"},"542318a9adaa18af52f6fbb20f740855085f8ffb":{"kind":"TRIVIAL_REBASE","_number":62,"created":"2025-02-17 10:49:24.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/62"}}},"commit":{"parents":[{"commit":"ec4f2b1c134ed6a8003ad4b88f642c651cfd5bac","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec4f2b1c134ed6a8003ad4b88f642c651cfd5bac"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-17 10:16:17.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/542318a9adaa18af52f6fbb20f740855085f8ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/542318a9adaa18af52f6fbb20f740855085f8ffb"}]},"branch":"refs/heads/master"},"a215df8cb0ad9d8ceca3b705257c603984af830f":{"kind":"REWORK","_number":63,"created":"2025-02-17 14:34:49.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/63"}}},"commit":{"parents":[{"commit":"ec4f2b1c134ed6a8003ad4b88f642c651cfd5bac","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec4f2b1c134ed6a8003ad4b88f642c651cfd5bac"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-17 14:33:02.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a215df8cb0ad9d8ceca3b705257c603984af830f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a215df8cb0ad9d8ceca3b705257c603984af830f"}]},"branch":"refs/heads/master"},"653c6c5e46b7152cecdaa4161ed1d1d838b26a4b":{"kind":"NO_CODE_CHANGE","_number":64,"created":"2025-02-18 16:55:39.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/64"}}},"commit":{"parents":[{"commit":"ec4f2b1c134ed6a8003ad4b88f642c651cfd5bac","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec4f2b1c134ed6a8003ad4b88f642c651cfd5bac"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-18 16:54:22.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/942125\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/653c6c5e46b7152cecdaa4161ed1d1d838b26a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/653c6c5e46b7152cecdaa4161ed1d1d838b26a4b"}]},"branch":"refs/heads/master"},"a699fc8d6ee10757d35497b107ad978704fa0672":{"kind":"NO_CODE_CHANGE","_number":65,"created":"2025-02-18 20:10:15.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/65"}}},"commit":{"parents":[{"commit":"ec4f2b1c134ed6a8003ad4b88f642c651cfd5bac","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec4f2b1c134ed6a8003ad4b88f642c651cfd5bac"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-18 20:10:15.000000000","tz":0},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a699fc8d6ee10757d35497b107ad978704fa0672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a699fc8d6ee10757d35497b107ad978704fa0672"}]},"branch":"refs/heads/master","description":"Edit commit message"},"dc88a86a273d920260f106e81568f5561e7b5614":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":66,"created":"2025-02-19 11:54:44.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/66"}}},"commit":{"parents":[{"commit":"c395e1cfd8c6f1ba0d3bb192398258fc1445e8f4","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c395e1cfd8c6f1ba0d3bb192398258fc1445e8f4"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-19 11:54:35.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/942125\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc88a86a273d920260f106e81568f5561e7b5614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc88a86a273d920260f106e81568f5561e7b5614"}]},"branch":"refs/heads/master"},"1ce8e39b4df9c9adf7dfcf4364e34d58e373ebea":{"kind":"TRIVIAL_REBASE","_number":67,"created":"2025-02-20 08:03:45.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/67"}}},"commit":{"parents":[{"commit":"4c79ad20ed8761de7624e65ccf2ba9d93aa85b90","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c79ad20ed8761de7624e65ccf2ba9d93aa85b90"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-20 08:03:30.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/942125\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ce8e39b4df9c9adf7dfcf4364e34d58e373ebea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ce8e39b4df9c9adf7dfcf4364e34d58e373ebea"}]},"branch":"refs/heads/master"},"030288606ae3777ab4ab58462fece7a454d2bf58":{"kind":"TRIVIAL_REBASE","_number":68,"created":"2025-03-03 18:15:40.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/68"}}},"commit":{"parents":[{"commit":"a23f93b41c51481d33fc8b51e82fae328085bd20","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a23f93b41c51481d33fc8b51e82fae328085bd20"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-03-03 18:15:28.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/942125\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/030288606ae3777ab4ab58462fece7a454d2bf58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/030288606ae3777ab4ab58462fece7a454d2bf58"}]},"branch":"refs/heads/master"},"a2722055c208ff05fd036c2c8c59377f8e93a0c6":{"kind":"NO_CODE_CHANGE","_number":69,"created":"2025-03-04 08:32:37.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/69"}}},"commit":{"parents":[{"commit":"a23f93b41c51481d33fc8b51e82fae328085bd20","subject":"dnm: enable native for os-ken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a23f93b41c51481d33fc8b51e82fae328085bd20"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-03-04 08:31:55.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2722055c208ff05fd036c2c8c59377f8e93a0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2722055c208ff05fd036c2c8c59377f8e93a0c6"}]},"branch":"refs/heads/master"},"705a688b57640203c4df3778343124653fd1cb71":{"kind":"TRIVIAL_REBASE","_number":70,"created":"2025-03-05 08:22:22.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/70"}}},"commit":{"parents":[{"commit":"98ad91efdb63bf77d854c019ae250d5bac0ce26f","subject":"ovs: remove usage of loopingcall.FixedIntervalLoopingCall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98ad91efdb63bf77d854c019ae250d5bac0ce26f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-03-05 08:22:12.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/705a688b57640203c4df3778343124653fd1cb71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/705a688b57640203c4df3778343124653fd1cb71"}]},"branch":"refs/heads/master"},"8c491d9fd8e28594cd0fb518f782e53e899c99ad":{"kind":"TRIVIAL_REBASE","_number":71,"created":"2025-03-06 07:14:46.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/71"}}},"commit":{"parents":[{"commit":"c3fd51b64ea663545defcf2ba4440781e62d9dff","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3fd51b64ea663545defcf2ba4440781e62d9dff"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-03-06 07:14:37.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c491d9fd8e28594cd0fb518f782e53e899c99ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c491d9fd8e28594cd0fb518f782e53e899c99ad"}]},"branch":"refs/heads/master"},"800843faa6faa341d422c3adef07356b1c7b8a07":{"kind":"TRIVIAL_REBASE","_number":72,"created":"2025-03-27 09:23:28.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/72"}}},"commit":{"parents":[{"commit":"45e6533c19906da58520000f746a8c967142980f","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45e6533c19906da58520000f746a8c967142980f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-03-27 09:23:18.000000000","tz":60},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/800843faa6faa341d422c3adef07356b1c7b8a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/800843faa6faa341d422c3adef07356b1c7b8a07"}]},"branch":"refs/heads/master"},"2d58319f8dd7595a4a425c4f9e658e39a21dd939":{"kind":"TRIVIAL_REBASE","_number":73,"created":"2025-04-15 09:49:25.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/73"}}},"commit":{"parents":[{"commit":"59ab5d209d653d0b40b0d9aea906a5b94e182398","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59ab5d209d653d0b40b0d9aea906a5b94e182398"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-04-15 09:49:14.000000000","tz":120},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d58319f8dd7595a4a425c4f9e658e39a21dd939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d58319f8dd7595a4a425c4f9e658e39a21dd939"}]},"branch":"refs/heads/master"},"6f8d516dc47fbb110fa8f0d1663fabb86a69f942":{"kind":"TRIVIAL_REBASE","_number":74,"created":"2025-04-23 15:30:09.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/74"}}},"commit":{"parents":[{"commit":"26e69f6c07dac021658804208cfc7093b6601165","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26e69f6c07dac021658804208cfc7093b6601165"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-04-23 15:30:00.000000000","tz":120},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f8d516dc47fbb110fa8f0d1663fabb86a69f942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f8d516dc47fbb110fa8f0d1663fabb86a69f942"}]},"branch":"refs/heads/master"},"afb7932333146a03744541d9385cde0df2a79b4d":{"kind":"TRIVIAL_REBASE","_number":75,"created":"2025-06-02 13:04:56.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/75"}}},"commit":{"parents":[{"commit":"ce9e99b0e5db46a78d5b784b233429b4f96d34ce","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce9e99b0e5db46a78d5b784b233429b4f96d34ce"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-06-02 13:04:48.000000000","tz":120},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afb7932333146a03744541d9385cde0df2a79b4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afb7932333146a03744541d9385cde0df2a79b4d"}]},"branch":"refs/heads/master"},"2c9683a5d1f24fa8d02f2e3d3edf042aa367345b":{"kind":"TRIVIAL_REBASE","_number":76,"created":"2025-06-04 08:17:57.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/65/937765/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/76"}}},"commit":{"parents":[{"commit":"f70a2fde6b6a0d341e114dbca4bb46affa63be70","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f70a2fde6b6a0d341e114dbca4bb46affa63be70"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-06-04 08:17:44.000000000","tz":120},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c9683a5d1f24fa8d02f2e3d3edf042aa367345b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c9683a5d1f24fa8d02f2e3d3edf042aa367345b"}]},"branch":"refs/heads/master"},"77641e2bdad01d5c0a5aabae571a049d19c48a45":{"kind":"REWORK","_number":77,"created":"2025-06-16 09:08:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/937765/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/77"}}},"commit":{"parents":[{"commit":"f70a2fde6b6a0d341e114dbca4bb46affa63be70","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f70a2fde6b6a0d341e114dbca4bb46affa63be70"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-16 09:06:20.000000000","tz":0},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nNOTES:\n* The \"OSKEN_HUB_TYPE\" is set to \"native\". Once the eventlet removal is\n  finished and the default backend is \"native\", this line will be\n  removed.\n* The ``oslo.service`` backend is set to \"threading\".\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77641e2bdad01d5c0a5aabae571a049d19c48a45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77641e2bdad01d5c0a5aabae571a049d19c48a45"}]},"branch":"refs/heads/master"},"6a5d168c0e3279bd1bf66c8af81c235cd8bfa1aa":{"kind":"TRIVIAL_REBASE","_number":78,"created":"2025-06-19 06:57:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/937765/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/78"}}},"commit":{"parents":[{"commit":"b7e939f65639311bce80b0c96eff06b0b8f5b23e","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7e939f65639311bce80b0c96eff06b0b8f5b23e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-06-19 06:57:02.000000000","tz":0},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nNOTES:\n* The \"OSKEN_HUB_TYPE\" is set to \"native\". Once the eventlet removal is\n  finished and the default backend is \"native\", this line will be\n  removed.\n* The ``oslo.service`` backend is set to \"threading\".\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a5d168c0e3279bd1bf66c8af81c235cd8bfa1aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a5d168c0e3279bd1bf66c8af81c235cd8bfa1aa"}]},"branch":"refs/heads/master","description":"Rebase"},"3f1bb9545dad73ee918bc2c5dbc5585735df3cb4":{"kind":"TRIVIAL_REBASE","_number":79,"created":"2025-06-25 11:21:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/937765/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/79"}}},"commit":{"parents":[{"commit":"c12d408e17d563ecc197245563acffbe747a7bed","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12d408e17d563ecc197245563acffbe747a7bed"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-06-25 11:21:00.000000000","tz":0},"subject":"ovs: remove the usage of eventlet in the OVS agent","message":"ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nNOTES:\n* The \"OSKEN_HUB_TYPE\" is set to \"native\". Once the eventlet removal is\n  finished and the default backend is \"native\", this line will be\n  removed.\n* The ``oslo.service`` backend is set to \"threading\".\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f1bb9545dad73ee918bc2c5dbc5585735df3cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f1bb9545dad73ee918bc2c5dbc5585735df3cb4"}]},"branch":"refs/heads/master","description":"Rebase"},"b5ab1ff36408dcec782c0d2f2e10014e586c38f8":{"kind":"REWORK","_number":80,"created":"2025-06-27 09:31:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/937765/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/80"}}},"commit":{"parents":[{"commit":"8b5b530e4ce36d353efe75959062a925c0d0f477","subject":"[eventlet-removal] ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b5b530e4ce36d353efe75959062a925c0d0f477"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-27 09:28:35.000000000","tz":0},"subject":"[eventlet-removal] ovs: remove the usage of eventlet in the OVS agent","message":"[eventlet-removal] ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nNOTES:\n* The \"OSKEN_HUB_TYPE\" is set to \"native\". Once the eventlet removal is\n  finished and the default backend is \"native\", this line will be\n  removed.\n* The ``oslo.service`` backend is set to \"threading\".\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5ab1ff36408dcec782c0d2f2e10014e586c38f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5ab1ff36408dcec782c0d2f2e10014e586c38f8"}]},"branch":"refs/heads/master"},"ad927935ab5868486824c02c478664194e474631":{"kind":"TRIVIAL_REBASE","_number":81,"created":"2025-06-27 12:46:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/937765/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/937765/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/937765/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/937765/81"}}},"commit":{"parents":[{"commit":"de72132d5112fcdb113b0ce7c7208fc8bd2a1810","subject":"[eventlet-removal] ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de72132d5112fcdb113b0ce7c7208fc8bd2a1810"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:47.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-06-27 12:46:16.000000000","tz":0},"subject":"[eventlet-removal] ovs: remove the usage of eventlet in the OVS agent","message":"[eventlet-removal] ovs: remove the usage of eventlet in the OVS agent\n\nThis commits remove any usage of greenthreads for OVS Agent.\n\nNOTES:\n* The \"OSKEN_HUB_TYPE\" is set to \"native\". Once the eventlet removal is\n  finished and the default backend is \"native\", this line will be\n  removed.\n* The ``oslo.service`` backend is set to \"threading\".\n\nCloses-bug: #2087939\nChange-Id: I04ac8336db70cf93da43ed17da11c594b441f5eb\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad927935ab5868486824c02c478664194e474631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad927935ab5868486824c02c478664194e474631"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
