)]}'
{"id":"openstack%2Fneutron~938404","triplet_id":"openstack%2Fneutron~master~I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63","project":"openstack/neutron","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-11 09:37:43.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-06-11 09:37:43.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-06-11 09:37:43.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-06-11 09:37:43.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-06-10 16:40:21.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63","subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","status":"MERGED","created":"2025-01-03 15:01:03.000000000","updated":"2025-06-11 09:38:57.000000000","submitted":"2025-06-11 09:37:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"938404-eventlet-removal","meta_rev_id":"f7018fe9ee30434b06b2f8b5a553e2a7a59c13b6","_number":938404,"virtual_id_number":938404,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-06-08 09:42:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-11 09:37:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-06-11 07:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-06-10 16:40:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-06-11 07:22:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-03 15:34:03.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-03 17:35:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-03 19:18:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 16:01:11.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-01-12 17:54:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-05-14 08:34:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2025-05-15 18:41:10.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2025-06-10 16:40:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-06-11 07:22:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"81fe7c757c810b20998ef1e0f3738030a49c9858","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-03 15:01:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41a5cc27c1460cdaa81342c143874e82665d2f33","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-03 15:34:03.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/04/938404/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c00730ad52c84105e8dcf9b85ac8282c8296de10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-03 16:14:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"32cf0ce6bf67549cbffa74593a396b363df9b00a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 17:35:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f005d43b01eb4047bda9f46595c4dea9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9830298ae5834d12a15c5a368ebac806 : SUCCESS in 29m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/478ba5cafdf246ce9c67bcf2ac88fc30 : SUCCESS in 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96fceda5306b54a23ebe61a8c22b56e386016d77","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-03 18:38:49.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/04/938404/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5e0d9ac4e77308d65ba8cef7a7f55214c5f580b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 19:18:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96439a4ec38d4f7eb74fcef1f04f727d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6d99989d9e34491bb583ac048dcfda4a : TIMED_OUT in 2h 49m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/30c4f293380645cfa0f57c91640b5f21 : TIMED_OUT in 2h 49m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8f3e9b67ca0747afb738a17ee54a55ad : SUCCESS in 1h 42m 06s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e01ea07d1d6c4cadb05ed591b6167cc7 : TIMED_OUT in 2h 49m 17s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c344842714647d9925c9af54124da6f : SUCCESS in 2h 19m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa544e6ee7b4464f88dbeb645c3d55d8 : SUCCESS in 32m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73604f93b6d64b568742b3a03a87efb8 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7779f576d7245658f5a4ba769f22c6d : SUCCESS in 21m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a76ad1fb9dce4ad0aab88b2289c87e32 : SUCCESS in 17m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb8ebfcb3f7402bbb7839909a0ab03e : SUCCESS in 9m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/57a392fd2eb040b691e6bf3256179764 : SUCCESS in 1h 07m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/499530adf225410c908bb0a53673609e : SUCCESS in 1h 00m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7c6b8eb619224ed784b2ac3108fb700c : FAILURE in 36m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/eb6c4e5b3d1f476597d819cccf6f8804 : FAILURE in 1h 26m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8d2c397a6e24f408621549c9baf0e3f : SUCCESS in 1h 00m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/36c53e0d8ad847fb96e53179afb63af6 : SUCCESS in 1h 17m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/67858d252e3d47ac877816c625375ced : SUCCESS in 1h 35m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/57bc417fedcf403fb02cb40adedcea84 : SUCCESS in 1h 28m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b19095c235af49e6a88a892b409df31f : SUCCESS in 2h 50m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7bd925236e094fa48feb19510bb56fac : SUCCESS in 1h 10m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7816ce6be45c4e09907e345f6ea8e4fa : SUCCESS in 1h 13m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/225f3d66ed194ebd8520fdff48435e82 : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":2},{"id":"17e0a0760bcc74bda634625241b4c70823acb577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-07 16:00:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"53bf3b4b7b59661007f036ff2118d493d77f821e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-01-07 16:01:11.000000000","message":"Patch Set 3: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 938393,4","accounts_in_message":[],"_revision_number":3},{"id":"6d42e99300c8a77d85fb2a125ffb3f21e75de772","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-07 16:58:49.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/04/938404/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"146cf411e68897841441333004f4ff8bb472b59d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-08 08:38:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"7a8063af1f61ea34bb0fe0a45d271a25e1bd41ec","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-01-08 08:39:12.000000000","message":"Patch Set 4: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 938393,4","accounts_in_message":[],"_revision_number":4},{"id":"607a7d71c2b0979f3f6efe5fd89c1823eceb0ef1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-08 09:01:03.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/04/938404/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"100056ee3aca3ff2b5582ffd0121725358276746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-08 09:36:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"16fc5fc4162a1a4a05016607e327997f6d931f03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 10:07:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f0176c90a0b4f74a508084d1c79bb1d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c31b3c5acf04700bc3927b1b6c964c6 : FAILURE in 28m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bea30b098e5d46e9a0d484115ba9eaf3 : FAILURE in 26m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2123ff052c8c85d8c8ef0b417e69260e438f9dcb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-08 10:17:58.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/04/938404/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"eef51f1939b7e798494b316d75a4f3d52724d352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 12:29:02.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62bebdc071fb45beb686ee9f8c8ff8aa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fc33a75dbe3b483590e535f46e2bcabd : SUCCESS in 1m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1b5cd8b37c8745a3b5cecabc10f4c280 : TIMED_OUT in 2h 49m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb1f0d031f094ce2b0e20d50a3fa1c3c : TIMED_OUT in 2h 49m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/85963bfd59fa4316b7750ab003e9ca8f : SUCCESS in 1h 49m 41s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64d554f38bc841e3ad7ab57dd8c89698 : TIMED_OUT in 2h 49m 15s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05f7a4ad0fc444f5b11d7760d884e922 : FAILURE in 2h 09m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8ad3084f46f4558b664995b2d4c04f7 : FAILURE in 31m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd652009a7cb451fb41fe49851831967 : FAILURE in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e8236605f0b4602a698ea72fb7f8a80 : FAILURE in 21m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a09aca189ae4070abc6c5741a4c852a : FAILURE in 19m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc317b6b61b48fe9abd7573d5b92430 : SUCCESS in 8m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/14bed3dbad814dba9b99c02301c683dd : SUCCESS in 1h 08m 52s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/04cf21299a7f48f38130ce6ebf6fefe2 : SUCCESS in 45m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6eacb5c8048349dcbc34c6df975dafa7 : FAILURE in 32m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a765ee27426f48d9afdd9144d846a84c : FAILURE in 2h 32m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ffe11bc1ae64be0afa4c2f5f09bff31 : SUCCESS in 1h 05m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3ea9936dc7a4b96a6ff4362ddf09760 : SUCCESS in 53m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1ec84f11219f420cada0f9f7cd19ede2 : FAILURE in 1h 06m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e1129ed82ab42daab7826e3f5c34d06 : SUCCESS in 1h 24m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/024eb556163a48d9bd76e639318b4e2d : SUCCESS in 1h 03m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0c60c91d9b134795b3a6c2aeabb8f878 : SUCCESS in 1h 07m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5de469fd95dd48f59509823a307f5e7a : SUCCESS in 1h 15m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/edba76ed48a341d19cdf4ba31f893e39 : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":5},{"id":"5afec5fb7aec033bef0cb59d79aecbfe4b6cf4d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-26 12:47:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1a2fef610e1b08a6f6e0f4d8dfd7082627c17355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-26 12:48:17.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"373e7292368f6663ebc99c588b6090f68b980881","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-26 14:09:36.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"1adc0cb0bebbd4d9208c6d89f998bdb1cbfea128","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 14:24:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d601db5ddbfe4569acdf670a4738c752\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/478977174a754613a6afe2f768a8b5d3 : TIMED_OUT in 1h 20m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91b0beb3199a40ef97eccef56a892ed2 : TIMED_OUT in 1h 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d5659e8f21275a7ecf3f5a6c3bebd0af036aef3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-26 16:04:12.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"80306df4c6a5ef5398eb18ff98f38dacb0703605","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-26 16:47:57.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/04/938404/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"49433c453900768057e8bbc996afd2511a0a964b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 18:26:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64436b564c0b427a874aa8f57472192b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fdc9bd7d1cc045f4a7b696014b1daac9 : TIMED_OUT in 1h 20m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4cbecbf7081b4ef481b371f6a6836760 : TIMED_OUT in 1h 20m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"99f7d336380808507084c0b81d18376f0e37c735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 20:08:52.000000000","message":"Patch Set 8: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90c1ac23811547d6b4d1a1882ea5581d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e9368d5dad04a689cc6e86cf66eb30c : SUCCESS in 2m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/24e7fcc2f7234c5daad677c00fffcc11 : TIMED_OUT in 2h 49m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6434940c6e5c4254a0c75c03c2f76ab3 : SUCCESS in 2h 17m 48s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9744e91cee14bcfbacfa41861c1c4b1 : TIMED_OUT in 2h 48m 54s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0d6e795aacc4572991434bd99836179 : SUCCESS in 1h 49m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4894e1bc18e437597430176c3384969 : TIMED_OUT in 1h 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75e59e6f3f164556b509a0b003e15789 : FAILURE in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01c069cb1d1c438d9c1ef108a7815399 : TIMED_OUT in 1h 00m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ae2224e82eb49a3b41584f719dfe1f3 : TIMED_OUT in 1h 00m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/881686f867b243f0a60a366ca8083aff : SUCCESS in 12m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fb128003660b490ca1820d6b848a8e0f : TIMED_OUT in 2h 13m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c8a7d980ceab451089f82d4f1bcd038e : FAILURE in 2h 04m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4aefa26da68143dab10026d92020b1c4 : SUCCESS in 50m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f69dfb9a0a2341aa98db8c30f2a45c04 : SUCCESS in 43m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7b8c063d354343218e2d3df6719d6f26 : FAILURE in 1h 19m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4ebff8c186e24a769a3ef63426804236 : FAILURE in 1h 17m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/693d5c977008432aa126bc14a4213bfc : SUCCESS in 1h 06m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/780a1a1634374fffb3f0102a784e1354 : SUCCESS in 1h 22m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7845526d2bca488dafeb757b325e78cc : SUCCESS in 1h 48m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1fbf3a1d05d0419ea5d592cfb268519a : SUCCESS in 1h 21m 52s\n\nWarning:\n  Comments left for invalid file neutron/common/utils.py","accounts_in_message":[],"_revision_number":8},{"id":"5cf5a09202366fa15639eba0eb26f05f24c883a3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-26 20:10:24.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/04/938404/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"9ee3c46d4261ba57c95a40868fec8595dae30bdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-16 09:06:00.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"48f38e98f6b40177959fd31aed90688b79e4ddfb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-16 10:17:32.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/04/938404/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"08608b600365c81e8ab647f756d0e7135f8456d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 11:01:02.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3a2f561c14146a0b91517ef19d224d3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/05333fef9f334f92a7babd0c3050142c : TIMED_OUT in 1h 20m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d57833abdbc4d28baafd971f6c70561 : TIMED_OUT in 1h 20m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b6590deaec71452a8b96e160b1315669 : TIMED_OUT in 1h 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5f9034765c34a7e2fb4c4a7b619184b3ac50e2a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 11:57:25.000000000","message":"Patch Set 9: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e767ee4244e5402cbf71c79c84c10280\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f9d8e5808104a5ca4dda1982b41473a : SUCCESS in 2m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7cd5acf190ae495d963f87b81143bf75 : TIMED_OUT in 2h 49m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3ccde2b184cf4f099f3521e283633fdc : SUCCESS in 1h 54m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cba34cf0df1041799d7d62ce9b14d15e : TIMED_OUT in 1h 21m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57dd9bed61d44fe19bfc7b1176e6aa47 : FAILURE in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f9a6297b21249dfa4c4d02258e9d370 : TIMED_OUT in 1h 01m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3eed46a0281b4c17a550e2ed6a5ade09 : TIMED_OUT in 1h 00m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c0d68fefc4442a1910c0d796ccf805a : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6524328425354d1fac2dc5bf45fd90aa : SUCCESS in 17m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/763573d01bf749238fa39a511308dea1 : TIMED_OUT in 2h 13m 53s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/be5772c64c1047bca07c4b145592a0c1 : FAILURE in 2h 00m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f6907be3e0c409f9d0eaf9823e3ed78 : SUCCESS in 1h 00m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8faf69c371f24c85a8b7ecdafafda247 : SUCCESS in 54m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/86d2db918cc24508b06aa6f9395f17cf : FAILURE in 1h 03m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ef310ae6b6264f95906c43de70b46763 : SUCCESS in 1h 17m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f1eaccc482454a389eb16b480d0ab915 : SUCCESS in 46m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8a081cfef1ca450abb612c347acddef4 : SUCCESS in 44m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1739f6652add40bba65eab9418795a00 : SUCCESS in 1h 13m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5b0dbee6e1254ddcac813aa73f7b739c : SUCCESS in 1h 25m 07s\n\nWarning:\n  Comments left for invalid file neutron/agent/dhcp/agent.py","accounts_in_message":[],"_revision_number":9},{"id":"da51889cc6515b31b854e97bc441b80ad0436a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-21 09:11:34.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"3678e54b50d249af1efbd7fa833ea5c85297952b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 09:32:30.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fabe509560425faeff55efd7ec69b5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/044dba75a5c149faa6f5333fa11b9a2a : FAILURE in 18m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4df94a57aec4b88bc921d5f3aa46a5d : FAILURE in 17m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/590e210bf43f411c97c2d97d3d76b660 : FAILURE in 19m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"79037dd412f350a162a5063296696e0ca5c7a4b8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-21 09:45:30.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/04/938404/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"ffd630dcb744d29bb9aa571a24ae16b0f1f3f45a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 12:02:22.000000000","message":"Patch Set 10: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb28606f2a074cd6bed26036a3798fbf\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a257549b49f64f228b90de0b8a393183 : TIMED_OUT in 2h 49m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/284da3e48fec4ac5aff3ea8cb91aeb36 : SUCCESS in 2h 39m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22e95889118045639703c1af982b297e : FAILURE in 26m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32d388334a6043a7ae19e11d7f445451 : FAILURE in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43d941d6441948e2bd51f32b3c2d6c38 : FAILURE in 15m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1b5c6c8d2e643918e288da19433ee4f : FAILURE in 19m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/074b093981b84fb2ad6d1cbb5a4bceda : FAILURE in 27m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08db8105a0841fb824f133629873394 : SUCCESS in 11m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1820fb89aa9c46508449dc2dac199b78 : FAILURE in 1h 41m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/14eaec514dea47fbab616c2c81f63b28 : FAILURE in 1h 42m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6cceabfca3554385ab97454a838c5cc9 : SUCCESS in 56m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/12fd0fe67e9c4c2e95086ffeb2b853e3 : SUCCESS in 1h 02m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/797933818219426f9261ab168777f353 : SUCCESS in 1h 33m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/96860a964d4349eca948fd496ace458f : SUCCESS in 1h 35m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8ebad1987ab042cebe6d1972b4ea1546 : SUCCESS in 54m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dd00ed812e6843c0a30c64ff2afcb0aa : SUCCESS in 58m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c76b74de5f34c51a6aa0e64e8a1ffa8 : SUCCESS in 1h 13m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5e6d93c2edec420c97d42d33edbc6a2b : SUCCESS in 1h 12m 28s\n\nWarning:\n  Comments left for invalid file neutron/agent/dhcp/agent.py","accounts_in_message":[],"_revision_number":10},{"id":"59441640bd5682c70e6072b1e9a07df6dc150935","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-08 08:32:41.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"be4829b1727bc2313a4bdc3d95d3f19ccac12f40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 09:03:14.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bc8a5eccd0344668bde54341d7a5aaa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10f6a3e8a2ea4833a441c8df8f578c96 : FAILURE in 23m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9725b94da150450685788674806d42f9 : FAILURE in 28m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ba295f13e7fc4fb6885a2df8c7a1a402 : FAILURE in 19m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7cb5c31a510be03e43bd3ba34b198ea9b2cf711e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-08 09:18:45.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5c4eec3085a383bdc2c3e0b24788e0690599aede","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-08 09:34:05.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"9d8f878db036b9384a91dd0372f72094f2fbe83b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 09:52:08.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e163425a2e44aaf8183cdcc64b5bca1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/023b4df5af7a4a28957054d61a4eb646 : FAILURE in 18m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e342355debfe4258a5e1745475e351d0 : FAILURE in 31m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/011e6be706af451cb56b87f8418ee43b : FAILURE in 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fd1ffb5ffdb9bf2be0ec8b05e0360539066d8e58","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-08 10:47:56.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"d028d96b94efeb96b958bf93712d2f28fd8dede9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 12:08:43.000000000","message":"Patch Set 12: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/215b55289a434a038f65baa944971148\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e295932623e8480e968cbee2836467f5 : TIMED_OUT in 2h 49m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/91eab3dd7f8b4b1f9ee96dc6cfca06c7 : SUCCESS in 1h 33m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b37f0230b1e64cd2bd311666561901b6 : FAILURE in 38m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55989e26b6d741e98864f497b003dad7 : FAILURE in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27e17d8b80f34a2f85e1ede2ab3141de : FAILURE in 14m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/442cf326ddfd4bb5948864dcb64b4973 : FAILURE in 15m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b5f3121dd5349a49176a3cb01e91d66 : FAILURE in 26m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/872501aeefaf4828a99c465aef915067 : SUCCESS in 5m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/601e9aa04e4c483da13ca2d416c3191f : FAILURE in 36m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/75a00830aec545f68abfd849819e1cfd : FAILURE in 2h 24m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/04b08492d38f4b8e8e7af5b8dcfe0d3f : SUCCESS in 47m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9706c55214f402fa6a0080a8962bbc6 : SUCCESS in 53m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f2cd7dc04c3f42ec8c6e30bb8517c98c : FAILURE in 1h 17m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ed69c05316a14b009c6f28194444c723 : FAILURE in 54m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8c420acc30f049cbbda5ba7ec934a7ab : SUCCESS in 55m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e2bf2c073369451bb663b1e43ccc9f57 : SUCCESS in 53m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f53ea1d873ad4a22abedf518b1398f61 : SUCCESS in 1h 15m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c73c089866f4421ad9263a23ba1cd15 : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":12},{"id":"fcc3308a2a43cd3740c5abccd0bbcf9d3e376cfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-12 05:37:27.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"43715bfbed6167d4ad9e04305c2ddbd86bffda50","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-12 06:12:20.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"1cf1fe2c22b6cca5a0096d27f85a728acaba9c1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 07:00:17.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e40f66cd8624645aaa4b59421dfe897\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a358cc02589e4cf8a93db6b236172901 : TIMED_OUT in 1h 20m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9bcea00f9fab47c9a7fb03af895e056d : TIMED_OUT in 1h 20m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e7a600916f91464c9c810594a4602792 : TIMED_OUT in 1h 20m 47s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7e3ca932f14ae44f6ffa7827380d6734c4665071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 08:33:20.000000000","message":"Patch Set 13: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4221f5753dbf4a739375d401181bc154\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2428dd07ac814a268446b7f8721ec297 : TIMED_OUT in 2h 49m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0b1a6bc1e9d140aa952ccbd72f43e5d9 : SUCCESS in 1h 57m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afc77327014549599ca722fd00e41a43 : TIMED_OUT in 1h 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5aeea6d24f749078b6502f73eefe7c1 : FAILURE in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bff34f1454c44bda3101cb7b5b007d3 : TIMED_OUT in 1h 00m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a15fe741f2a3437d9da777b12538e4c7 : TIMED_OUT in 1h 00m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b877a2f901a4f808b3907b4925553eb : TIMED_OUT in 1h 00m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa4374df039440896a237aba4a7efab : SUCCESS in 9m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1b8c360de1724daeb343b6fc714ef102 : TIMED_OUT in 2h 12m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/16bbd57f953b4ad59ec8afe86fcee3d2 : FAILURE in 2h 15m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9ff9ddddaca40519337545ea0fd1bab : SUCCESS in 41m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f11634b17dd4a69badf3669608cb1ae : SUCCESS in 1h 01m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2d15770cd07740e2bf2b370b89b07b71 : FAILURE in 1h 12m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a90e1c5a192440cc8ec8551f0d8ded73 : FAILURE in 1h 09m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ed5a3cc13be64c08880aaefe4e3ea1dc : SUCCESS in 1h 05m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e5b1214b0975482e920c45ced3cfcf38 : SUCCESS in 57m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/54b3141e9c0b4efa87590a298a4090e1 : FAILURE in 49m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a023c06f97f84e828ff087c52d0404c4 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":13},{"id":"a942e1195c1f20d4f58e43b7d39de66555e52203","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-12 11:56:36.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"06fdf9fd576fbfa14676a5c4188c7e306151a0eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 10:55:12.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"4dc614e5d0f02aeb6335e2a5204852e0f8387567","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 10:55:23.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"bb98b93cb1925cfb995b28815dfc0ae48705193a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 10:56:51.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"7999c07cb76c2228f0fb1aac76bd480733f5b9e8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 10:57:02.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"30aef94cb89097b704398e0474d45fbd4c6f2f76","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-13 11:30:59.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"3dfac6b8446cb612b27e954faf202587a9b7c6cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 11:45:02.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8aed3c1e9a7844358f28a385ca9825d0\n\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/fdd0baa0aa044e26952cf3c81c164a90 : FAILURE in 47m 11s","accounts_in_message":[],"_revision_number":15},{"id":"a545fb6f7b1505fcb2da913824599595672c7230","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-13 12:11:24.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"a63b9c7a611e99e8b0c25825547c27dcd15a8c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 12:38:59.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"53ad9c9d13a4aec39fcf0ed9d43fb2f5ed0c7dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:14:34.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57f32b3b86a448dfa0a9153c8e0d622a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8230f09f89a748b198c60f69f0516567 : SUCCESS in 3m 05s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/2bc376fb60704dfc9230d96643455c3f : FAILURE in 23m 44s\n- neutron-tempest-plugin-openvswitch-with-oslo-master https://zuul.opendev.org/t/openstack/build/be43cad2023f4f67b1766d071ab503d2 : FAILURE in 33m 02s","accounts_in_message":[],"_revision_number":16},{"id":"ba10f80aa60cc4062dd38034dbbc2390347b4237","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-13 13:54:08.000000000","message":"Patch Set 16:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"28b133dccd4df04b7fda23fe1fb335f157b69490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 16:02:53.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"3533c9cb4991d0cdd07dbc0f2c5cee1fc397ae71","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-13 16:29:01.000000000","message":"Patch Set 17:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"d472d7019bbbb0a18d79ec6c57da98b1e721d2b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 18:52:58.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56aa9a5f20b34e87955e55dc8b620656\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a5def29e337d4af687a500e087e95bab : SUCCESS in 2m 45s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/2d455b7d2142490d8accdbda0451884a : FAILURE in 1h 20m 28s\n- neutron-tempest-plugin-openvswitch-with-oslo-master https://zuul.opendev.org/t/openstack/build/fe247d2c0ce742f0840b09c110ade6d8 : TIMED_OUT in 2h 48m 28s","accounts_in_message":[],"_revision_number":17},{"id":"8c9bf9e85b20186d8a5c2cbdf1f7fa6efdac690a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-05-14 08:34:44.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"97011ce82e67691f34a08e24e481c297db3db6a8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-05-14 08:36:17.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e4e3e3966ef710f76d91807ea7d56115417c59e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-14 09:30:23.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"f43ed3ab371e5a4b54bd61072596e598be18824a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-14 09:32:32.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"00108667447e412e18e39fdbd35aff5bb5e474d9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-14 10:06:38.000000000","message":"Patch Set 17:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"e9177ba603ed4d4e9d3d20560e7afa9abf358ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 12:03:19.000000000","message":"Patch Set 17:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/945720 is needed.","accounts_in_message":[],"_revision_number":17},{"id":"6d0a9e3765f6c60352855d2aac6a539bcbc806a5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-14 12:38:43.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"afad0f7f0a5ed93c442b229b4bb68a2160eebd8e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-14 13:04:34.000000000","message":"Patch Set 17:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"f823105f7a88f2992b2f8f8b30d02856422198ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 14:54:40.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee989c3a6f4b445ba422b03b68ed74c4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/16a8d7e78e124378b71617ab41fdd647 : SUCCESS in 2m 36s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/002370855ead49c5859e8a9eb32b93e9 : SUCCESS in 1h 18m 33s\n- neutron-tempest-plugin-openvswitch-with-oslo-master https://zuul.opendev.org/t/openstack/build/5da57d62ea7e40d1b02321c46b82799b : SUCCESS in 2h 12m 14s","accounts_in_message":[],"_revision_number":17},{"id":"23ecd851f68f95c3d9e757b50673d03a3912978e","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-05-15 18:41:10.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"b590df14c2d24c05404264b37f8cfb7ceea3b4b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-26 11:20:16.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"1171e8b7866c214690b5bc21f769040c857f89db","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-26 12:00:14.000000000","message":"Patch Set 18:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"8bb58793feb5e23d84a9a7c6c2f3ad4cb3aad97e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-26 14:11:50.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"a711f0aa0851faec79628f2d9ea08e4ecc8a3c3c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-26 15:00:12.000000000","message":"Patch Set 19:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/04/938404/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"cde1498a3db26758d2a3965bb6531e13931cb968","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 15:35:17.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77e1b8d2b29045f095d584627ae63a9b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2bc655899afe4ba38ba3fda44bee7179 : TIMED_OUT in 1h 21m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a138806cbce74f41954419c4a4f35643 : TIMED_OUT in 1h 20m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d878db6f5fc4ae98146d081e2fd3a64 : TIMED_OUT in 1h 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"558f87875f2b9a9bed6e9b9ba2625a8b13cadfe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 16:30:13.000000000","message":"Patch Set 19: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/200f14a6e27f4d549aeef2c5c5cefc47\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d20c94ca6012438e8d7b8d8592222190 : SUCCESS in 3m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c6be6973a317461f90dac8f540f65691 : SUCCESS in 1h 35m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d1f042d637704da3b40fd10b8715f0c8 : SUCCESS in 2h 15m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64127df2838141d09f12370fdf36ce06 : TIMED_OUT in 1h 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bab14590c184cbb8556fb7ac3b4c08a : FAILURE in 6m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/545d404010a44b3db8de62debf7b7cda : TIMED_OUT in 1h 00m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25ac4fc379544205bbac8f245cc1d427 : TIMED_OUT in 1h 00m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/073403f1bd4f4d97ae11a935f9bcd7cb : TIMED_OUT in 1h 00m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c1f07d60af743958188be846ef5780d : SUCCESS in 11m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/676f692540e0487f82b79b5dcc62fc4e : FAILURE in 2m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/949c4e07782844d2b67b8f20dc5f8f69 : TIMED_OUT in 2h 13m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/074a61f4199b48e89f0cc8da53725292 : FAILURE in 2h 06m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad433e8c96af4bdeb105fb9f7eaaa563 : SUCCESS in 52m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4f894656ed044d58c3c6990fb04793e : SUCCESS in 1h 02m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7d070e3c27334271b97c3b978b0e7367 : SUCCESS in 1h 34m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d7b18261fad74ea9a70b856791b8cdd8 : SUCCESS in 1h 27m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9c036e518b0e4a869c35d7abf387f92f : SUCCESS in 50m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ca1bb21fdac9495ea2026c2a5c14c970 : SUCCESS in 55m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/114c3242511f4a7589cb2cfc2b096f73 : SUCCESS in 1h 09m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e0ed8ecbf43844e5903eadeebd643cab : SUCCESS in 1h 16m 22s","accounts_in_message":[],"_revision_number":19},{"id":"9db4057cbf08dac36c9100add6349ee6e6e06ff8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-27 05:56:05.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"8d92385dda41ae1ad5d227f85d94d9520006ad92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-27 12:51:03.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"e0c157f90ed771d0468302957ecdc75d046d3481","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-27 13:47:30.000000000","message":"Topic bug/2087943 removed","accounts_in_message":[],"_revision_number":20},{"id":"c5b2c763a16d8da1221c4c20893a6cd3cf77af34","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-27 13:47:40.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":20},{"id":"f1377410b8217910593e27e9cd7e2382ea3cf1f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-27 14:57:29.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"3728c8c1990539733ac6b5c099e9c27273a5062c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 16:11:55.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b65a0c4504394ddda4be42bbf5802d8c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a544b6369e464bc8ad83e1c9042f3005 : SUCCESS in 42m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7030eb36902a4d93854b3af598bcf044 : SUCCESS in 53m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/10e6fa8a15f64d2a9ca42e940d3bd587 : SUCCESS in 44m 02s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"92373d31d5a5f2914d1bc5e16ffcd4c74980a72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 17:05:50.000000000","message":"Patch Set 21: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cafb214d75a742dda72a3d2d84bde9cf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1d2d7b297f3f451a86c782d094713334 : SUCCESS in 2m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bfbee207579e4bcb9cb74c29e19be67e : SUCCESS in 1h 32m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5d6d5f21180f427a9710659fa3cb38d0 : SUCCESS in 1h 49m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03823a76e394449b87ba1042c2b1d762 : SUCCESS in 57m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/604fcece1072412a8bd9e2f2b99ccd45 : FAILURE in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa700cb730154a8d8765d1c7265276e0 : SUCCESS in 50m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7640f6f7958949118cde5d3647ad4d81 : SUCCESS in 40m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6dbe642fc4904244a456d802e5c5f99a : SUCCESS in 54m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4361f13cee74dafa78cfa220c429a35 : SUCCESS in 13m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a051245a574e46338056fef056b0be8b : SUCCESS in 3m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ac78a4b05d6c4240b4f1f9030a30002d : SUCCESS in 1h 12m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/84f3cab6677547649204475818659847 : FAILURE in 2h 04m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd94fa1f9ad745cfae72ecb1d83aee3e : SUCCESS in 49m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/df606cd4ace5429dae3823e30e140a78 : SUCCESS in 1h 11m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3c62649a289b4045b0a6e9c302b45ac6 : SUCCESS in 1h 38m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d12c8887c8684b0ab35184a8367f3d9e : SUCCESS in 1h 30m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ee66b16ac45e46b296c5c23d9261a6c4 : SUCCESS in 1h 11m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0e5f2c75ba994bde995375012ab3cbf0 : SUCCESS in 1h 02m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/397ebf632d5e48e3a80d1d8a203a747c : SUCCESS in 1h 15m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/18e56f342fb54aa9ab30570de2cfa688 : SUCCESS in 1h 15m 55s","accounts_in_message":[],"_revision_number":21},{"id":"243cbc3b9ebea51b715ad9f61a62c2747ce88a6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-28 06:43:33.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"25d354de4a054a66f86bac941dd2b5ee10d16006","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 07:36:48.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f26362c85cea45dfbb29b150ffd27fe6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/762542492467492ab9b09caaa77f078f : SUCCESS in 51m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/76705e12849243b7b8e666c08bb79bec : SUCCESS in 50m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da8e18cfc2aa45c68f5a4d3f6ff4172f : SUCCESS in 43m 16s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"2ca0c210d11b62ccf675ad546d454dcb817e514e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 08:42:44.000000000","message":"Patch Set 22: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e98abfa73054344bc196c9dd46473f0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/790675c6c6fd4dc1a69a43656ab45442 : SUCCESS in 2m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1e79ad9f7f8f4fab996d908e0b4ce142 : SUCCESS in 1h 24m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ea7592b28a7b4358857427f02356c80e : SUCCESS in 1h 57m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f28cacca70e943c4901f62c3849b0184 : SUCCESS in 53m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96f520a6fd5640faa62e744c243a6d20 : FAILURE in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7936d7957e3b4d4ca2162152f09fa852 : SUCCESS in 38m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6514e1d3e5974f0f9eaf7869985bfc22 : SUCCESS in 44m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f4607faf9154f9dbc7d435454a45c3d : SUCCESS in 50m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7141e1840f54c42bbc0e4a7e3665081 : SUCCESS in 9m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/878ba04352264c23995f6682448befe3 : SUCCESS in 2m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/232c5a32037c4e15a2f9c99039ca50d4 : SUCCESS in 1h 08m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/61ba918031a44c4f904adfbf55f046f0 : SUCCESS in 1h 57m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c03e86caefb842d58bd95f21a4f83cea : SUCCESS in 46m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/61ec55a85ad74bf5b4d3e4a3a71a88c1 : SUCCESS in 55m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1eadf1fd51854604afcafb2b45349be3 : SUCCESS in 1h 22m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/280a38c4be28425099a42be1ca6c35b1 : SUCCESS in 1h 24m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/df37c7f8923244d3902f502408e02ac5 : SUCCESS in 1h 01m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2afb39e53820492f923e5836aae44845 : SUCCESS in 48m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4aef4c89d7a44d94b80b57a252e616c0 : FAILURE in 56m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6030b28d5f884abbb2d7b95def88e45d : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":22},{"id":"df285ee883da043908a024166d4a8821ef4afae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-29 10:49:14.000000000","message":"Patch Set 23: Published edit on patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"06b0b3d125326b161582a01a6670ab24ede5dc4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 12:12:39.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7703161949bd43cbabafba18a93063e8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c46af0479f164bfeb8173f112c43837f : SUCCESS in 40m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a623f77332c24255b4520bdcd3313187 : SUCCESS in 42m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da00951da4284bbb8fa3b377b9a0fc86 : TIMED_OUT in 1h 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"04cec6494e70c9593a26bb4eb749a56cf86430c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 13:27:27.000000000","message":"Patch Set 23: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b99189ea85004cb791f875f59d50bfc4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7de7b2a138054d84b6538e19eb9fb5d7 : SUCCESS in 2m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7127f0fa2af24d3981ae76dc2d306d49 : SUCCESS in 1h 33m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/af91337fee3249d2b88204d9f48b2641 : SUCCESS in 2h 35m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/824664de24c14268a478a0940160df91 : SUCCESS in 1h 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b93fa4a62e91483b9911b35c7ee19f53 : FAILURE in 9m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/854159747efc4aaba760faebcb6288c4 : SUCCESS in 41m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1baa73c7cb6a43f4b8339dc7687e6d4d : SUCCESS in 44m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fabc704ec8fa4ed9b93e3db1e1b30999 : SUCCESS in 44m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ac7a033b564b6d9c8ec967f3391a55 : SUCCESS in 11m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/edc276acae5740a38fb68574e14cc5fd : SUCCESS in 2m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/39f04ddcaf5d456cbedd38aff0d36549 : TIMED_OUT in 2h 12m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/474037a5fd964ba7ab2d7f6223d3e0cf : SUCCESS in 1h 40m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6574b65d3b1470c91035df82ac0b551 : SUCCESS in 49m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ffe2b1d9f764dcb8b67f9b5e12ed825 : SUCCESS in 1h 24m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/56c4e8af826a40a1a5b9fa97738bcba7 : SUCCESS in 1h 41m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/94f36153edb7447ba4788affc47e0ca6 : SUCCESS in 1h 19m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/84e194395a254438b0dba282651616e0 : SUCCESS in 1h 04m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8b968e9862e64712966357cc1731bb1a : SUCCESS in 1h 06m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b7a1bf8f86224820b88498c1899d8042 : SUCCESS in 1h 12m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4b8a1e01946d4f58b9e7503fc00203ea : SUCCESS in 1h 37m 06s","accounts_in_message":[],"_revision_number":23},{"id":"ebd0adef347701f6991b1a456aaf7ece7a389fac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-29 15:11:30.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"4609c737715eaaf9ccb98f55a2f6f09d3e3aeb0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 16:06:26.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9810511a3a94c318e41ce5ec6781f70\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec29671d213c485d9cfa0d4d0ae42209 : SUCCESS in 44m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91eb9faf60574745a357d9d2118c2a85 : SUCCESS in 40m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1bae94258c5e47e39c58ff3ac1b7eef2 : SUCCESS in 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"2b3fe8744d6506c6565b9afd08f857ed3285d629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 17:05:52.000000000","message":"Patch Set 24: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c3c1372665c42ed9e7ef3f060944640\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ba9169e465e4666aee721981313d941 : SUCCESS in 2m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/19d1edc50b2447d2bd98a56160351bf8 : SUCCESS in 1h 49m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8368c73e215e4b0f81d1fe45ad9b711d : FAILURE in 54m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26f28716276c4034a286f7e266740c66 : SUCCESS in 54m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33afa683c036474b84ffe3f0d146aa8e : FAILURE in 9m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cac5940077bb4aeb98116201752fe68e : SUCCESS in 50m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a13bf41f2b984f3a848328b3b01fe793 : SUCCESS in 49m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6e9e2fefb1541ce89baffa930c7c117 : SUCCESS in 54m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c4871afc78c4a9c867e4ec4b753a724 : SUCCESS in 13m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f5b2ecb17f714bd0a001d3f480c28f5a : SUCCESS in 2m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/019dd2f2603440f3af185ce360afbd84 : SUCCESS in 1h 24m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cd925ae9bfe44ceea974ff14b6797991 : SUCCESS in 1h 39m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9720c9b67944aeda4f66ae19b49bd11 : SUCCESS in 59m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/27856ae1b9e24c32a168015f35f2c170 : SUCCESS in 1h 02m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e4a0ff92f1a84fd6a436406e3fb7c1c3 : SUCCESS in 1h 49m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d1007e55b36b4b46bac12feb311a5e08 : SUCCESS in 1h 30m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4e00dbbb062c4f63bc71cbf02e395eeb : SUCCESS in 57m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8dcf6ddfdb0c48c395687aed768f054c : SUCCESS in 57m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0772ecb94d2b4fff9daacb0839cd357d : SUCCESS in 1h 13m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3595dfe785e3476d9f9e17edab1941f6 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":24},{"id":"837c0898669b0d722d14c7a831634ef965edff3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-03 06:48:08.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"f7c4d048ece5b17496fa940f73259287b2627148","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 08:41:40.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be5e3fd988b44d97b87cd597e606607f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/79ab7036f6f64eeab4a7dbcd4959125c : SUCCESS in 1h 12m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7004d92634914fdf85095348c85b21f6 : SUCCESS in 45m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/345dd2a4ce4043f3a21b6fc8b708c816 : TIMED_OUT in 1h 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"8f04044a9a5cf7e44f3099ad348731d43181af17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 08:59:43.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a13d7f5448c3411abe7e21f2dc4240c3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3225839bf21546e5a380c7fcab85cab2 : SUCCESS in 2m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/85a3631300ad4b36ab4b37eafc64e256 : SUCCESS in 1h 24m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/00cc6eb569b84f7ca5d936b1402ed31d : SUCCESS in 2h 08m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b1a431770d44338ae8031ac70b62365 : SUCCESS in 1h 02m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6e3876b98364d2fbeaa95e885d558ca : SUCCESS in 8m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1985c561dba24c86aa1fa8b8fc223d5b : SUCCESS in 46m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e986e6a03ee84bc6a82e65a795b7a7f2 : SUCCESS in 41m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f49364d7b8564584a236723aee86840a : SUCCESS in 48m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7173849c31d5413b8c24a876c17fcb67 : SUCCESS in 12m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b55d953a00043e3a65362beec5c68cd : SUCCESS in 2m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aac71f1ece97462ebc2631646497dac2 : SUCCESS in 1h 05m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a8804b5595894dd2a720738d7e31798f : SUCCESS in 1h 31m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e82f88d138942e1b1a16871afce70c9 : SUCCESS in 58m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7611c8555e644a63a17eff2b4d560be0 : SUCCESS in 1h 07m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/20246aadb520401aa975fde7de958d9c : SUCCESS in 1h 45m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6612ef03456c4e578d28bbaee0a336a0 : SUCCESS in 1h 15m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0b9e6fb0bf7445ff8c31cc303b2791fe : SUCCESS in 1h 11m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bed7c0a08658407499b28a43ae7ddfe7 : SUCCESS in 1h 06m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df9d878987e944728a7821b649eacdf8 : SUCCESS in 53m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/369a1241f10a40699a83e547e8971bce : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":25},{"id":"e0bf76abfc2026a5a48e91f4439f6f137d132184","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-05 06:10:33.000000000","message":"Patch Set 25: Review-Priority+1","accounts_in_message":[],"_revision_number":25},{"id":"da2069161a1e960a3c72fefebf329474c8b017bc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-05 18:19:48.000000000","message":"Patch Set 25:\n\n(6 comments)","accounts_in_message":[],"_revision_number":25},{"id":"b88fed23603968ed022aa6e1330438b22ef46da0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-06 11:16:53.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"34cfb764ce15b8b5a992fb51fed61b975a461367","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 12:07:17.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa18fec8fc114a9f9f521ed9a3a10ba4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c8227d652af4273b6de941224bcd447 : SUCCESS in 46m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/02117331ea744c689d4031f7dcba4040 : SUCCESS in 48m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/91bc266de85d4f14bf831302fbc60908 : SUCCESS in 46m 31s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"0d7b82e1f7ea9d819316ed554720d02dde12e9d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 13:42:15.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d25c73632bb54df58818797a11e2952d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e92fb25dc934f0fa3065dadcdf96082 : SUCCESS in 2m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/15341fd1ff944d15ab636e89e67a4c3f : SUCCESS in 1h 48m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/842cbd049bb44d74be3129aa29b92972 : SUCCESS in 2h 24m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c069ae15a1f4887adbb84eafdc86c4e : SUCCESS in 1h 04m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cea72887c6c64bdfa8f98bb900741a8e : SUCCESS in 10m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b24784bce9e547d3a51ff2e641782789 : SUCCESS in 47m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a64ef8de33bc413fa482f0af264f242a : SUCCESS in 46m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d133ee1756624f85be70943ed47118d1 : SUCCESS in 52m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35016a72f1574fa0b91866e53232598b : SUCCESS in 11m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/111a05372b0e4f8c8687ecd7d9bffee3 : SUCCESS in 2m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2acd74aa7cf94db5bb0e24994dcae04a : SUCCESS in 1h 19m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a414c0db7df94360b0a93b8aea43fb80 : SUCCESS in 1h 11m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/10430fe39b87434490d0ae154e99f4b5 : SUCCESS in 45m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/22d22fcfe40346db9de3ce65196fe9b5 : SUCCESS in 1h 05m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c0127de64ebe4142b912ce69ace75d61 : SUCCESS in 2h 05m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c199cef049824aab891e41a2468ff365 : SUCCESS in 1h 12m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/89904cff90f0443ead5afa4f1d85c2c3 : SUCCESS in 1h 28m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5136f5c080ee4c3ab1c5606f80d62dc8 : SUCCESS in 1h 01m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f886556d2e2d4b44841ccd9d54ba6d6d : SUCCESS in 1h 04m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a2e6a07d2b8e4c4cb09ce014d0033ba4 : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":26},{"id":"aa59fd30e9af73162f519e33f5a66a22269335a3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-06 22:46:29.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/04/938404/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"2a7d98f8e43b8104f917447ce5e47b81d89c2541","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-06 23:25:39.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/04/938404/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"26a129a448dd9eba0a7e52d3d867d5b8ab256ef8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-07 01:12:10.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/04/938404/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"ebe20c98e5ba164b16fb6125b2dd702042573ee2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-07 06:49:45.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/04/938404/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"b7971b11d79c3aaad8c400ae60598b0f13b01ede","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-07 09:40:46.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/04/938404/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"ccb42503445ff95eb524a5f1eb05a613f75c81c3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-07 20:25:17.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/04/938404/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"1eb61ae6717e0f896d2ccd8fc57512d05cfaf637","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-08 09:42:13.000000000","message":"Patch Set 26: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/04/938404/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"316c88c57b5785473e498d8cf09e014b87f8da7e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-09 09:16:26.000000000","message":"Patch Set 26:\n\n(5 comments)","accounts_in_message":[],"_revision_number":26},{"id":"ad731ae239e2ed64d87b544f61e5bfe819fa2ce7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-10 16:40:21.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"cf8ea2a3ec1ac7aa5a22f16fdfbf2cb006a5d2bf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-06-11 07:22:56.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"00c3da0bc437a2cc635dab3460f99321297d187d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 07:23:20.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"cc54faf20294ebdda5e990f0497b6dba98d3dabc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 09:37:42.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafa582af3ab4703b737ea9ee4cc55c4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/05c9e017fb2748abbd905460f1110cdc : SUCCESS in 2m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0bd4b6f4b9444d228bd627949814762a : SUCCESS in 1h 32m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/17556dbebbed404b936706dd0d5ac9f3 : SUCCESS in 2h 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ae2f9ca008f4b3580863076061b3220 : SUCCESS in 10m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd0471f8cf84443181b16df45d8ae558 : SUCCESS in 44m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65aac39001884b5b8fec55dea7097c77 : SUCCESS in 59m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57dc7e3e7b81407a82f2369aa5f02e4e : SUCCESS in 11m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2281568c74a4365aeb6ceb4efc3c7db : SUCCESS in 2m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9732d6819849450f8a43bb29849f73b7 : SUCCESS in 1h 06m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d3b1c2912e2345998664946d2c9bf5cb : SUCCESS in 1h 33m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/13dc87e818c34cbfa21ae77ea463904c : SUCCESS in 1h 43m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/949f861a2da84a4ab7ca8e34ab5fb930 : SUCCESS in 31m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/633854cf41984fe49d277ee25d23d3c6 : SUCCESS in 45m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8dbdc77e2eb344ef86f603520632343b : SUCCESS in 56m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/15c7b0c132134a369fe645c7d83a4891 : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":26},{"id":"674325ac1011b2d3d1a597915a127d0b9386a43e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 09:37:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26},{"id":"f7018fe9ee30434b06b2f8b5a553e2a7a59c13b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 09:38:57.000000000","message":"Patch Set 26:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b101cf108f34c08b98e84421ea78767\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a8f38fa09ff4cf8bed3479bc7a657cd : SUCCESS in 56s","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"5292b13bbdd4e00dadbbef72400b737aab3dd0ff","revisions":{"db3e35c371aed5e6be54ecc3f3b9cf612f7f3f1b":{"kind":"REWORK","_number":1,"created":"2025-01-03 15:01:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/1"}}},"commit":{"parents":[{"commit":"7a345845f62ada3a2db862bc0b3ab56eb97ed91f","subject":"[eventlet-deprecation] Use non-eventlet metadata proxy in OVN metadata agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a345845f62ada3a2db862bc0b3ab56eb97ed91f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 15:00:44.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db3e35c371aed5e6be54ecc3f3b9cf612f7f3f1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db3e35c371aed5e6be54ecc3f3b9cf612f7f3f1b"}]},"branch":"refs/heads/master"},"9e910db43f984666b4790525ea73c24df02bd3da":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-01-03 16:14:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/2"}}},"commit":{"parents":[{"commit":"b5b5f6dbbefa68f0a1e719009be8046229e85443","subject":"[eventlet-deprecation] Use non-eventlet metadata proxy in OVN metadata agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5b5f6dbbefa68f0a1e719009be8046229e85443"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-01-03 16:14:36.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e910db43f984666b4790525ea73c24df02bd3da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e910db43f984666b4790525ea73c24df02bd3da"}]},"branch":"refs/heads/master","description":"Rebase"},"2a30124f626894217200ccd61fede8bbf081a8d3":{"kind":"REWORK","_number":3,"created":"2025-01-07 16:00:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/3"}}},"commit":{"parents":[{"commit":"b5b5f6dbbefa68f0a1e719009be8046229e85443","subject":"[eventlet-deprecation] Use non-eventlet metadata proxy in OVN metadata agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5b5f6dbbefa68f0a1e719009be8046229e85443"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-07 15:48:57.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a30124f626894217200ccd61fede8bbf081a8d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a30124f626894217200ccd61fede8bbf081a8d3"}]},"branch":"refs/heads/master"},"ff90f4e685e0331d368ca4d21acbfc1ea09d6c6a":{"kind":"REWORK","_number":4,"created":"2025-01-08 08:38:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/4"}}},"commit":{"parents":[{"commit":"b5b5f6dbbefa68f0a1e719009be8046229e85443","subject":"[eventlet-deprecation] Use non-eventlet metadata proxy in OVN metadata agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5b5f6dbbefa68f0a1e719009be8046229e85443"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-08 08:32:33.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff90f4e685e0331d368ca4d21acbfc1ea09d6c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff90f4e685e0331d368ca4d21acbfc1ea09d6c6a"}]},"branch":"refs/heads/master"},"d299df41ff8f9478eee6dfd94bee006419f5f9d0":{"kind":"REWORK","_number":5,"created":"2025-01-08 09:36:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/5"}}},"commit":{"parents":[{"commit":"29ed88a37e1796f096cbe57059d187b03bde17f8","subject":"[eventlet-deprecation] Replace ``eventlet.spawn_n`` usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29ed88a37e1796f096cbe57059d187b03bde17f8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-08 08:46:27.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d299df41ff8f9478eee6dfd94bee006419f5f9d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d299df41ff8f9478eee6dfd94bee006419f5f9d0"}]},"branch":"refs/heads/master"},"ea44011117e2d7aefe7440658753f3de9410be58":{"kind":"REWORK","_number":6,"created":"2025-03-26 12:47:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/6"}}},"commit":{"parents":[{"commit":"4976cea8fb2a4f26430effb64fa89c16819bf530","subject":"[eventlet-removal] Replace ``eventlet.GreenPool``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4976cea8fb2a4f26430effb64fa89c16819bf530"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-25 21:56:53.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-deprecation] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea44011117e2d7aefe7440658753f3de9410be58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea44011117e2d7aefe7440658753f3de9410be58"}]},"branch":"refs/heads/master"},"c064e18fbf5399e60fbbabdcbd30f71de93b616d":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-03-26 12:48:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/7"}}},"commit":{"parents":[{"commit":"4976cea8fb2a4f26430effb64fa89c16819bf530","subject":"[eventlet-removal] Replace ``eventlet.GreenPool``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4976cea8fb2a4f26430effb64fa89c16819bf530"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-25 21:57:14.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c064e18fbf5399e60fbbabdcbd30f71de93b616d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c064e18fbf5399e60fbbabdcbd30f71de93b616d"}]},"branch":"refs/heads/master"},"358971f5aeef96b88aa951b0b1940dc8d6b6d045":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-03-26 16:04:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/8"}}},"commit":{"parents":[{"commit":"f248075574338bf31f5a2956939007a30782311e","subject":"[eventlet-removal] Replace ``eventlet.GreenPool``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f248075574338bf31f5a2956939007a30782311e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-03-26 16:04:12.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/358971f5aeef96b88aa951b0b1940dc8d6b6d045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/358971f5aeef96b88aa951b0b1940dc8d6b6d045"}]},"branch":"refs/heads/master","description":"Rebase"},"67a5f0fd5a73006025511e51dc7ed26aa05b2075":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-04-16 09:06:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/9"}}},"commit":{"parents":[{"commit":"ed32f6a0d8892bb9f8b3456a0a245b7de0ef859d","subject":"[eventlet-removal] Replace ``eventlet.GreenPool`` in the DHCP agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed32f6a0d8892bb9f8b3456a0a245b7de0ef859d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-16 09:05:19.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67a5f0fd5a73006025511e51dc7ed26aa05b2075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67a5f0fd5a73006025511e51dc7ed26aa05b2075"}]},"branch":"refs/heads/master"},"cc9e685cf862762aea9fa9aa4eaad473c5d8235e":{"kind":"REWORK","_number":10,"created":"2025-04-21 09:11:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/10"}}},"commit":{"parents":[{"commit":"ed32f6a0d8892bb9f8b3456a0a245b7de0ef859d","subject":"[eventlet-removal] Replace ``eventlet.GreenPool`` in the DHCP agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed32f6a0d8892bb9f8b3456a0a245b7de0ef859d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-21 09:11:15.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc9e685cf862762aea9fa9aa4eaad473c5d8235e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc9e685cf862762aea9fa9aa4eaad473c5d8235e"}]},"branch":"refs/heads/master"},"af439014b73d7d612cfdeb97a71fda2d67ca9f11":{"kind":"REWORK","_number":11,"created":"2025-05-08 08:32:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/11"}}},"commit":{"parents":[{"commit":"344baa4d4ba821baa8307c22d4cfe6494b19496c","subject":"add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/344baa4d4ba821baa8307c22d4cfe6494b19496c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-08 08:32:27.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af439014b73d7d612cfdeb97a71fda2d67ca9f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af439014b73d7d612cfdeb97a71fda2d67ca9f11"}]},"branch":"refs/heads/master"},"f14a1af24e73e56f92d316c0e40d8091db449a68":{"kind":"REWORK","_number":12,"created":"2025-05-08 09:18:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/12"}}},"commit":{"parents":[{"commit":"4e91d1969a542ae361872c7fe35a2cfcb33f4e1c","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e91d1969a542ae361872c7fe35a2cfcb33f4e1c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-08 09:18:06.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f14a1af24e73e56f92d316c0e40d8091db449a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f14a1af24e73e56f92d316c0e40d8091db449a68"}]},"branch":"refs/heads/master"},"cb2f602dc1cb5d5b8616bca3a2990f7c3d0bfb97":{"kind":"REWORK","_number":13,"created":"2025-05-12 05:37:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/13"}}},"commit":{"parents":[{"commit":"474b4f6fc9e5c3bed923310157854d9030abdce6","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/474b4f6fc9e5c3bed923310157854d9030abdce6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-12 05:37:13.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb2f602dc1cb5d5b8616bca3a2990f7c3d0bfb97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb2f602dc1cb5d5b8616bca3a2990f7c3d0bfb97"}]},"branch":"refs/heads/master"},"b08f2cb366402ceea8b194482ccb3887f60d3c47":{"kind":"REWORK","_number":14,"created":"2025-05-13 10:55:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/14"}}},"commit":{"parents":[{"commit":"474b4f6fc9e5c3bed923310157854d9030abdce6","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/474b4f6fc9e5c3bed923310157854d9030abdce6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-10 09:56:04.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b08f2cb366402ceea8b194482ccb3887f60d3c47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b08f2cb366402ceea8b194482ccb3887f60d3c47"}]},"branch":"refs/heads/master"},"4399c50e9ca0994be4b239e4133b9bc353aede51":{"kind":"REWORK","_number":15,"created":"2025-05-13 10:56:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/15"}}},"commit":{"parents":[{"commit":"474b4f6fc9e5c3bed923310157854d9030abdce6","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/474b4f6fc9e5c3bed923310157854d9030abdce6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-10 10:00:45.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4399c50e9ca0994be4b239e4133b9bc353aede51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4399c50e9ca0994be4b239e4133b9bc353aede51"}]},"branch":"refs/heads/master"},"d0cbeada3023955b96073dca0ee25df717935cfe":{"kind":"REWORK","_number":16,"created":"2025-05-13 12:38:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/16"}}},"commit":{"parents":[{"commit":"474b4f6fc9e5c3bed923310157854d9030abdce6","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/474b4f6fc9e5c3bed923310157854d9030abdce6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-10 11:42:34.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0cbeada3023955b96073dca0ee25df717935cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0cbeada3023955b96073dca0ee25df717935cfe"}]},"branch":"refs/heads/master"},"336752d2b72ee11b4e9d46cb49e9f592c6ca3272":{"kind":"REWORK","_number":17,"created":"2025-05-13 16:02:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/17"}}},"commit":{"parents":[{"commit":"474b4f6fc9e5c3bed923310157854d9030abdce6","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/474b4f6fc9e5c3bed923310157854d9030abdce6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-05-13 16:02:53.000000000","tz":0},"subject":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server","message":"WIP \u003d\u003d [eventlet-removal] Change the state change WSGI server\n\nTOOD: testing, documentation, description\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nPartial-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/336752d2b72ee11b4e9d46cb49e9f592c6ca3272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/336752d2b72ee11b4e9d46cb49e9f592c6ca3272"}]},"branch":"refs/heads/master"},"35d3fed97076618eb0f1bee74ad600143b7f46ce":{"kind":"REWORK","_number":18,"created":"2025-05-26 11:20:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/18"}}},"commit":{"parents":[{"commit":"474b4f6fc9e5c3bed923310157854d9030abdce6","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/474b4f6fc9e5c3bed923310157854d9030abdce6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-25 20:10:26.000000000","tz":0},"subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","message":"[eventlet-removal] Remove the usage of eventlet in the L3 agent\n\nThis patch removes the usage of eventlet in the L3 agent. It removes\nthe last bits using this library and changes the implementation of the\nWSGI server.\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``. The\n``KeepalivedStateChangeHandler`` class is now inheriting from\n``socketserver.StreamRequestHandler``. This change is\nsimilar to the changes done for the Metadata agents before [1][2]\n\nThis patch bumps the ``oslo.service`` library to 4.2.0, that includes\n[3].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/938393\n[2]https://review.opendev.org/c/openstack/neutron/+/942916\n[3]https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nCloses-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35d3fed97076618eb0f1bee74ad600143b7f46ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35d3fed97076618eb0f1bee74ad600143b7f46ce"}]},"branch":"refs/heads/master"},"46ea0784308db9cc071d458587aab02805ec8a83":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-05-26 14:11:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/19"}}},"commit":{"parents":[{"commit":"99bd0c9331c901e9f1633b58f2e52a8a7aef69d3","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bd0c9331c901e9f1633b58f2e52a8a7aef69d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-05-26 14:11:50.000000000","tz":0},"subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","message":"[eventlet-removal] Remove the usage of eventlet in the L3 agent\n\nThis patch removes the usage of eventlet in the L3 agent. It removes\nthe last bits using this library and changes the implementation of the\nWSGI server.\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``. The\n``KeepalivedStateChangeHandler`` class is now inheriting from\n``socketserver.StreamRequestHandler``. This change is\nsimilar to the changes done for the Metadata agents before [1][2]\n\nThis patch bumps the ``oslo.service`` library to 4.2.0, that includes\n[3].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/938393\n[2]https://review.opendev.org/c/openstack/neutron/+/942916\n[3]https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nCloses-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46ea0784308db9cc071d458587aab02805ec8a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46ea0784308db9cc071d458587aab02805ec8a83"}]},"branch":"refs/heads/master","description":"Rebase"},"349e46f21bb56d110ed802bba8acf390e7253b32":{"kind":"REWORK","_number":20,"created":"2025-05-27 12:51:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/20"}}},"commit":{"parents":[{"commit":"99bd0c9331c901e9f1633b58f2e52a8a7aef69d3","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bd0c9331c901e9f1633b58f2e52a8a7aef69d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-27 12:50:48.000000000","tz":0},"subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","message":"[eventlet-removal] Remove the usage of eventlet in the L3 agent\n\nThis patch removes the usage of eventlet in the L3 agent. It removes\nthe last bits using this library and changes the implementation of the\nWSGI server.\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``. The\n``KeepalivedStateChangeHandler`` class is now inheriting from\n``socketserver.StreamRequestHandler``. This change is\nsimilar to the changes done for the Metadata agents before [1][2]\n\nThis patch bumps the ``oslo.service`` library to 4.2.0, that includes\n[3].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/938393\n[2]https://review.opendev.org/c/openstack/neutron/+/942916\n[3]https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nCloses-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/349e46f21bb56d110ed802bba8acf390e7253b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/349e46f21bb56d110ed802bba8acf390e7253b32"}]},"branch":"refs/heads/master"},"41ba294a2e8fd540ec6d97d7ac491084f2c29fc6":{"kind":"REWORK","_number":21,"created":"2025-05-27 14:57:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/21"}}},"commit":{"parents":[{"commit":"99bd0c9331c901e9f1633b58f2e52a8a7aef69d3","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bd0c9331c901e9f1633b58f2e52a8a7aef69d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-27 14:50:57.000000000","tz":0},"subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","message":"[eventlet-removal] Remove the usage of eventlet in the L3 agent\n\nThis patch removes the usage of eventlet in the L3 agent. It removes\nthe last bits using this library and changes the implementation of the\nWSGI server.\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``. The\n``KeepalivedStateChangeHandler`` class is now inheriting from\n``socketserver.StreamRequestHandler``. This change is\nsimilar to the changes done for the Metadata agents before [1][2]\n\nThis patch bumps the ``oslo.service`` library to 4.2.0, that includes\n[3]. Note that the requirements line is also requesting the dependencies\nfor \"threading\", that install aditional libraries.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/938393\n[2]https://review.opendev.org/c/openstack/neutron/+/942916\n[3]https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nCloses-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41ba294a2e8fd540ec6d97d7ac491084f2c29fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41ba294a2e8fd540ec6d97d7ac491084f2c29fc6"}]},"branch":"refs/heads/master"},"7027f73942af9839666a89f298f37a0bf8da6c9c":{"kind":"REWORK","_number":22,"created":"2025-05-28 06:43:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/22"}}},"commit":{"parents":[{"commit":"99bd0c9331c901e9f1633b58f2e52a8a7aef69d3","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bd0c9331c901e9f1633b58f2e52a8a7aef69d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-28 06:43:18.000000000","tz":0},"subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","message":"[eventlet-removal] Remove the usage of eventlet in the L3 agent\n\nThis patch removes the usage of eventlet in the L3 agent. It removes\nthe last bits using this library and changes the implementation of the\nWSGI server.\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``. The\n``KeepalivedStateChangeHandler`` class is now inheriting from\n``socketserver.StreamRequestHandler``. This change is\nsimilar to the changes done for the Metadata agents before [1][2]\n\nThis patch bumps the ``oslo.service`` library to 4.2.0, that includes\n[3]. Note that the requirements line is also requesting the dependencies\nfor \"threading\", that install aditional libraries.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/938393\n[2]https://review.opendev.org/c/openstack/neutron/+/942916\n[3]https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nCloses-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7027f73942af9839666a89f298f37a0bf8da6c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7027f73942af9839666a89f298f37a0bf8da6c9c"}]},"branch":"refs/heads/master"},"120733e6d70821992eebb8158db3e0ea25d3d4bf":{"kind":"REWORK","_number":23,"created":"2025-05-29 10:49:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/23"}}},"commit":{"parents":[{"commit":"99bd0c9331c901e9f1633b58f2e52a8a7aef69d3","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bd0c9331c901e9f1633b58f2e52a8a7aef69d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-05-29 10:49:14.000000000","tz":0},"subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","message":"[eventlet-removal] Remove the usage of eventlet in the L3 agent\n\nThis patch removes the usage of eventlet in the L3 agent. It removes\nthe last bits using this library and changes the implementation of the\nWSGI server.\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``. The\n``KeepalivedStateChangeHandler`` class is now inheriting from\n``socketserver.StreamRequestHandler``. This change is\nsimilar to the changes done for the Metadata agents before [1][2]\n\nThis patch bumps the ``oslo.service`` library to 4.2.0, that includes\n[3]. Note that the requirements line is also requesting the dependencies\nfor \"threading\", that install aditional libraries.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/938393\n[2]https://review.opendev.org/c/openstack/neutron/+/942916\n[3]https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nCloses-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/120733e6d70821992eebb8158db3e0ea25d3d4bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/120733e6d70821992eebb8158db3e0ea25d3d4bf"}]},"branch":"refs/heads/master"},"b9465258929e663a7cfdb7a64ceb34ac29658e0d":{"kind":"REWORK","_number":24,"created":"2025-05-29 15:11:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/24"}}},"commit":{"parents":[{"commit":"99bd0c9331c901e9f1633b58f2e52a8a7aef69d3","subject":"Move the RPC initialization to the ``Service.start`` phase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bd0c9331c901e9f1633b58f2e52a8a7aef69d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-05-29 15:11:30.000000000","tz":0},"subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","message":"[eventlet-removal] Remove the usage of eventlet in the L3 agent\n\nThis patch removes the usage of eventlet in the L3 agent. It removes\nthe last bits using this library and changes the implementation of the\nWSGI server.\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``. The\n``KeepalivedStateChangeHandler`` class is now inheriting from\n``socketserver.StreamRequestHandler``. This change is\nsimilar to the changes done for the Metadata agents before [1][2]\n\nThis patch bumps the ``oslo.service`` library to 4.2.0, that includes\n[3]. Note that the requirements line is also requesting the dependencies\nfor \"threading\", that install aditional libraries.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/938393\n[2]https://review.opendev.org/c/openstack/neutron/+/942916\n[3]https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nCloses-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9465258929e663a7cfdb7a64ceb34ac29658e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9465258929e663a7cfdb7a64ceb34ac29658e0d"}]},"branch":"refs/heads/master"},"7fbaebbcb406d911700953f0ed2a908f1efd3cef":{"kind":"REWORK","_number":25,"created":"2025-06-03 06:48:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/25"}}},"commit":{"parents":[{"commit":"21a0588ed89b79f65b73d55ef3ba99cce59ff94d","subject":"Merge \"Deprecate `vlan_qinq` and `vlan_transparent` config options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21a0588ed89b79f65b73d55ef3ba99cce59ff94d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-02 06:29:23.000000000","tz":0},"subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","message":"[eventlet-removal] Remove the usage of eventlet in the L3 agent\n\nThis patch removes the usage of eventlet in the L3 agent. It removes\nthe last bits using this library and changes the implementation of the\nWSGI server.\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``. The\n``KeepalivedStateChangeHandler`` class is now inheriting from\n``socketserver.StreamRequestHandler``. This change is\nsimilar to the changes done for the Metadata agents before [1][2]\n\nThis patch bumps the ``oslo.service`` library to 4.2.0, that includes\n[3]. Note that the requirements line is also requesting the dependencies\nfor \"threading\", that install aditional libraries.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/938393\n[2]https://review.opendev.org/c/openstack/neutron/+/942916\n[3]https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nCloses-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fbaebbcb406d911700953f0ed2a908f1efd3cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fbaebbcb406d911700953f0ed2a908f1efd3cef"}]},"branch":"refs/heads/master"},"5292b13bbdd4e00dadbbef72400b737aab3dd0ff":{"kind":"REWORK","_number":26,"created":"2025-06-06 11:16:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/938404/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/938404/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/938404/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/938404/26"}}},"commit":{"parents":[{"commit":"18dff1b31a731e0c81b501ca75fa1a85dd17fa70","subject":"Merge \"[eventlet-removal] DHCP: move RPC init to service.start phase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18dff1b31a731e0c81b501ca75fa1a85dd17fa70"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-03 14:58:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-06 11:16:34.000000000","tz":0},"subject":"[eventlet-removal] Remove the usage of eventlet in the L3 agent","message":"[eventlet-removal] Remove the usage of eventlet in the L3 agent\n\nThis patch removes the usage of eventlet in the L3 agent. It removes\nthe last bits using this library and changes the implementation of the\nWSGI server.\n\nThis patch replaces the WSGI ``neutron-keepalived-state-change``\nserver with an implementation based on\n``socketserver.ThreadingUnixStreamServer``. The\n``KeepalivedStateChangeHandler`` class is now inheriting from\n``socketserver.StreamRequestHandler``. This change is\nsimilar to the changes done for the Metadata agents before [1][2]\n\nThis patch bumps the ``oslo.service`` library to 4.2.0, that includes\n[3]. Note that the requirements line is also requesting the dependencies\nfor \"threading\", that install aditional libraries.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/938393\n[2]https://review.opendev.org/c/openstack/neutron/+/942916\n[3]https://review.opendev.org/c/openstack/oslo.service/+/945720\n\nCloses-Bug: #2087943\nChange-Id: I82f7b4e4c4f165bab114a0ab5ee4948d3ee8ce63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5292b13bbdd4e00dadbbef72400b737aab3dd0ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5292b13bbdd4e00dadbbef72400b737aab3dd0ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
