)]}'
{"id":"openstack%2Fneutron~938696","triplet_id":"openstack%2Fneutron~master~I63eb51a2c60e93e1808e84c7bb9f82f032f461ba","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I63eb51a2c60e93e1808e84c7bb9f82f032f461ba","subject":"[wip] Add test to compare time on process loop","status":"ABANDONED","created":"2025-01-08 19:49:34.000000000","updated":"2025-02-05 21:50:50.000000000","total_comment_count":1,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"c516ec3d5c0273dcb6bd6d266b5d68abd6cf3a27","_number":938696,"virtual_id_number":938696,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-01-08 23:46:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-08 22:22:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-01-08 20:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-08 22:22:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-08 23:46:05.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43c74452e87c1a89887a12d840ffd7531d8d324c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-01-08 19:49:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"286c957cd51c04d0bc81a85ddfd5fd658f673aea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 20:20:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7400a899e50a49d78e881bb758f174f3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83a11fa0cacd4ce8a80332d0dba6d5c4 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/78258030e846413783897a580ae06799 : SUCCESS in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdd5fbb8896a97dbeb3b6423e7d708cec7536424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 22:22:33.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0a5c5271ca5499e91e4224a3e8cbf4d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/790ac4398e2446ab929ae9633d320792 : SUCCESS in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/388f0da2079f49c4bcf511fc5c28bd08 : FAILURE in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7637c98bb1ec43bba0b5389d5fb14304 : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df257bd0f1e149339aed86d9c15da2b8 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f243fd57284c6981c6bc1289d022aa : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21f03e59dee44951aad8a1225086b30e : SUCCESS in 8m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/929359a555da4fa690d2bd1000fde0e8 : SUCCESS in 1h 10m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e11cebb141974433bb1822bb1d1f8df4 : SUCCESS in 2h 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ff71dff5dc08a2b37dd3e3d62c8200e458771978","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-01-08 22:26:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bea6c2a92ccf73123f94e7c7b8da94d2736ea1a8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-08 23:46:05.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/96/938696/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c516ec3d5c0273dcb6bd6d266b5d68abd6cf3a27","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-05 21:50:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25213448fa09218ac682df25de6cc6ec412272b8","revisions":{"25213448fa09218ac682df25de6cc6ec412272b8":{"kind":"REWORK","_number":1,"created":"2025-01-08 19:49:34.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/96/938696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/938696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/938696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/938696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/938696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/938696/1"}}},"commit":{"parents":[{"commit":"c4549307c5553b2a043b78aa631630d3dbd7e5de","subject":"[L3] Reduce to one single thread the event processing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4549307c5553b2a043b78aa631630d3dbd7e5de"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2025-01-08 19:45:14.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2025-01-08 19:49:12.000000000","tz":60},"subject":"[wip] Add test to compare time on process loop","message":"[wip] Add test to compare time on process loop\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\nChange-Id: I63eb51a2c60e93e1808e84c7bb9f82f032f461ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25213448fa09218ac682df25de6cc6ec412272b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25213448fa09218ac682df25de6cc6ec412272b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
