)]}'
{"id":"openstack%2Fneutron~938849","triplet_id":"openstack%2Fneutron~master~If5e06dcf3a323692137d2c7cbb8ca725d9d7d821","project":"openstack/neutron","branch":"master","topic":"bsd","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2025-01-10 12:52:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If5e06dcf3a323692137d2c7cbb8ca725d9d7d821","subject":"Remove is_bsd skips in test suite","status":"MERGED","created":"2025-01-09 22:14:16.000000000","updated":"2025-01-10 12:52:56.000000000","submitted":"2025-01-10 12:52:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938849-bsd","meta_rev_id":"792d20660cbf043a58f215e34adc51e63f0c6b9c","_number":938849,"virtual_id_number":938849,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-01-10 00:07:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 12:52:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-10 08:23:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-01-09 22:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-10 08:23:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-09 22:40:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-09 22:47:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-01-10 00:07:18.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 00:51:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 08:23:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"0e5a458bcbeef10c9d887c49fc33126aea9aa1a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-09 22:14:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c027f4755aa54d3fd867525ab73a74c77670f92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 22:40:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d489f8096be468c9ace1367fd37e6a3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f95c2503cad40a88d2cc26f3bbdc977 : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96b0d816b0dc4e3d92d8368609034f5a : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91ca64ecab30686ff8f7b6416a1aa1abfc73ccf1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-09 22:47:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c08818d6a3b0c9a675c8ce327ab9340d235db88","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-10 00:07:18.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/49/938849/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"65ace5fe1d02668910143bbb29fbd9fb45be3a4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 00:51:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36b982fa59434aa6857140fc29c865c9\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/461216a282a84cfdadb1d9ca5066f2a9 : SUCCESS in 2h 06m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d59b23b99e144864bf37606d029988a5 : SUCCESS in 1h 39m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7dbd8c5a99894dc09b8e35b355af38f9 : SUCCESS in 2h 34m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da4a99fbdf85409eadbcd4626d29135e : SUCCESS in 44m 28s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f9f29a7af0749d0bb58f90b601da711 : SUCCESS in 1h 43m 25s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26f5535eccd44f5bb4fa33a0bd48f232 : SUCCESS in 2h 06m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/120d5bcc92c44cd29352c909fbf1f793 : SUCCESS in 58m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d6fc0e253f04ba5b99237ea9ac7ab84 : SUCCESS in 11m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d7a5b884a1040e9ade5e999b00b5b3c : SUCCESS in 24m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/425368a67c194e6f9d1292bc1f652429 : SUCCESS in 18m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53b355be89b49e9812b4d7dfa654b6b : SUCCESS in 10m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/206e9fcea853404581b81861fd768e83 : SUCCESS in 1h 18m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7a99ff6d86944883ac56438c705a5ffd : SUCCESS in 2h 01m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b136c0cf8822e18bf49f138a74a9d9c9f74e7ac2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-10 08:23:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f7c7fb4b39fff181ab72710f86805eb5ae40e31b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 08:24:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9dd933f015abfed61a5bd301bc3908b00726cd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 12:52:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45924a59991345ba84c344ce4fe3e30d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8cf39c1d514f4b3fba936868a536f8c8 : SUCCESS in 1h 46m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6524735ef5004800b30389b39860127c : SUCCESS in 1h 50m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/96990e5d0277499c86d9414290732f9b : SUCCESS in 2h 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/285234de262046dd967967dfc44d59b5 : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3db22d5ca2e3485a81ca8ef353561381 : SUCCESS in 24m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e10feddbf7a74f21bbb835f3464637db : SUCCESS in 21m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8993c6fce78f4a9f95aec43272e0c3d1 : SUCCESS in 10m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/876b6ce8bc4a497e8b0e01265f13ea79 : SUCCESS in 1h 21m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d66753028b724e02b407bf46b252b8dd : SUCCESS in 2h 00m 43s","accounts_in_message":[],"_revision_number":1},{"id":"792d20660cbf043a58f215e34adc51e63f0c6b9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 12:52:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4437501c2e2d6747043efcf04643e581f1d02a6","revisions":{"c4437501c2e2d6747043efcf04643e581f1d02a6":{"kind":"REWORK","_number":1,"created":"2025-01-09 22:14:16.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/938849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/938849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/938849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/938849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/938849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/938849/1"}}},"commit":{"parents":[{"commit":"da06d1a7d8938715962040095a78a0c5902f8a85","subject":"Merge \"Setup DVR router ARP for AAP with correct MAC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da06d1a7d8938715962040095a78a0c5902f8a85"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-09 22:13:21.000000000","tz":-300},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-09 22:13:21.000000000","tz":-300},"subject":"Remove is_bsd skips in test suite","message":"Remove is_bsd skips in test suite\n\nWe don\u0027t support running neutron - or its tests - on macos. The suite\nfails in multiple other test cases on the platform anyway.\n\nTrivialFix\n\nChange-Id: If5e06dcf3a323692137d2c7cbb8ca725d9d7d821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4437501c2e2d6747043efcf04643e581f1d02a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4437501c2e2d6747043efcf04643e581f1d02a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
