)]}'
{"id":"openstack%2Fneutron~938913","triplet_id":"openstack%2Fneutron~stable%2F2023.2~Ic11992ba3ed91980189efbacdc2a54fba64fcf7c","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2092659-stable/2024.1-stable/2023.2","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-01-24 12:48:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic11992ba3ed91980189efbacdc2a54fba64fcf7c","subject":"Make sure that policy enforcer is initialized before use","status":"MERGED","created":"2025-01-10 15:19:37.000000000","updated":"2025-01-24 12:49:39.000000000","submitted":"2025-01-24 12:48:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938913-bug/2092659-stable/2024.1-stable/2023.2","cherry_pick_of_change":938912,"cherry_pick_of_patch_set":1,"meta_rev_id":"9682cddbd1c1a39481d82839eff604b323c34bb9","_number":938913,"virtual_id_number":938913,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-24 12:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-01-24 08:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-01-22 16:13:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-01-24 08:31:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 17:41:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-22 16:13:58.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2025-01-24 08:31:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"3a5c4bfb6336d74e00272c84b954bf31ad1b8492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-10 15:19:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"0e834eec6a8ef2b2509a786b793feaf0b31e3f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:41:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e9eb8d2b12f440182de960e6cc1b56b\n\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/e332cf27c82a4303996bd8f31982193a : SUCCESS in 1h 12m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/1f23e2880607452d80477b1acb9db9d6 : SUCCESS in 2h 16m 05s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/1db4b1953f1a470ea680dfbded25fea7 : SUCCESS in 1h 49m 37s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/c528255a2d5248789fa5a68fc8e901f7 : SUCCESS in 28m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73868ad16014450b8ba761e21f5997b0 : SUCCESS in 40m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8576e95e7fe64002a3d6313058af9301 : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/175eb54ebeef4ee887063d749e85c2b9 : SUCCESS in 25m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3789f3bd5faf4652a5960f341b2df36c : SUCCESS in 24m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9005f25dd3545389d9e990236a34ff4 : SUCCESS in 27m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2539ca7f3ae6425ab90ce0b3cc375fd9 : SUCCESS in 26m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b06d5cacce4f95903189aa0abcc41f : SUCCESS in 10m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5cf6676684c24672b9adb5b8413204a6 : SUCCESS in 52m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8bc46a12e34488b974c04af6593428e : SUCCESS in 1h 26m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a92803e8904742cfa7c2c21b767dd088 : SUCCESS in 56m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4f4bf6419ed145598c25f1c7f1bb5aa5 : SUCCESS in 1h 10m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ea8471455d1f4241a68f98d2c32ec4fb : SUCCESS in 1h 07m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/81b2556b98584899abe04008aba3e62d : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9d762e7b872549a698074a320301d676b57efc3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 18:00:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff31aec4db614eff8c0a18a58e632e8f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ab60bec0ecc49a891f51e1b0abaa8ef : SUCCESS in 36m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b25e989809f4b6cbdfa032312080372 : SUCCESS in 35m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e34114ec03643919880623370940941 : SUCCESS in 30m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/38493aab721e41ed86949ec40cd8ed96 : SUCCESS in 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"432578d155e74c0025fe60d90da34149a739aa8e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-01-22 16:13:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6477d236352e1c1248199cdc33dccd28a929a95","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-24 08:31:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89a14153aa04040a752ab3cd91204f538b801588","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 08:32:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"335b02a87d6d1b9810443c83d2b3165174c74dbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 12:48:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f1ca8367a864baeb54f8a47e9a8341d\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/1f8d726549564273b30a420d4f534507 : SUCCESS in 1h 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ddfe06ee29461fbc309401f7cbbb13 : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b53746181f3a4314b6019157f009009c : SUCCESS in 25m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6feab86bf180407982ce99e1fe11daf8 : SUCCESS in 23m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9218d737011845128cdd04f345cf6a93 : SUCCESS in 25m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4863cc4c7cb4429bf7c049e6b1fe3db : SUCCESS in 10m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/946de81a3f864b37b8d3c214480aff67 : SUCCESS in 59m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/45c9773e029041ef975f74cf08da2548 : SUCCESS in 1h 47m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/87e8f6eb316f40348116d9c1a03cc864 : SUCCESS in 58m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b33d115f9c524d7da7cee2b65edd18b3 : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7e6190958ed11c84dc100d9c32c255db88a7a581","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 12:48:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9682cddbd1c1a39481d82839eff604b323c34bb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 12:49:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67b22af3543c4e53ae05ebd94c74bf78\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a367f9df784650ad8d50f62a5f24e3 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03bf0d7c388ea2fee378c8a335813eaac8c3247a","revisions":{"03bf0d7c388ea2fee378c8a335813eaac8c3247a":{"kind":"REWORK","_number":1,"created":"2025-01-10 15:19:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/13/938913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/938913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/938913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/938913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/938913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/938913/1"}}},"commit":{"parents":[{"commit":"81e38bc8972c7dbc7f15957aed91c5581fde4079","subject":"Use the correct input for OVN agent deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81e38bc8972c7dbc7f15957aed91c5581fde4079"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-08 10:05:29.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-10 15:19:37.000000000","tz":0},"subject":"Make sure that policy enforcer is initialized before use","message":"Make sure that policy enforcer is initialized before use\n\nIn some cases policy.enforce() or policy.check() functions were called\nwithout calling policy.init() before and that could lead to errors like\nmentioned in the related bug.\nTo avoid that and to make it less error prone in future, this patch\nmoves call of the policy.init() method directly to the enforce() and\ncheck() methods and removes it from all other places. Other modules\nshould not need to carry about policy initialization at all.\n\nConflicts:\n    neutron/extensions/tagging.py\n\nCloses-Bug: #2092659\nChange-Id: Ic11992ba3ed91980189efbacdc2a54fba64fcf7c\n(cherry picked from commit 08afd36e549892e35be7356547fd7b525cae0bb3)\n(cherry picked from commit c94b09f14af56a5dc1e14d9707444060e77f1ea8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/03bf0d7c388ea2fee378c8a335813eaac8c3247a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/03bf0d7c388ea2fee378c8a335813eaac8c3247a"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
