)]}'
{"id":"openstack%2Fneutron~939166","triplet_id":"openstack%2Fneutron~master~I2895eb92be760c4bba7b22345d1802cf7cec1f94","project":"openstack/neutron","branch":"master","topic":"ovn-maint-missing","attention_set":{},"removed_from_attention_set":{"7353":{"account":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"last_update":"2025-04-04 07:42:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2895eb92be760c4bba7b22345d1802cf7cec1f94","subject":"fix: add try/except to allow maintenance to run on unmanaged segments","status":"ABANDONED","created":"2025-01-13 20:47:10.000000000","updated":"2025-04-04 07:42:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4b85502af268485bb0bef87b73789db974667f07","_number":939166,"virtual_id_number":939166,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-01-14 01:54:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-13 23:43:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-13 22:24:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-13 23:43:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 01:54:35.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-04 07:42:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"0556657609fb1ed973119bedcd3732e5b73925a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2025-01-13 20:47:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b85640bd90ae315ad2cddd6d665a3b3a0c89027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 22:24:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1a83e4639fc48148a072fc1bc41d490\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc069bfc935f4bce87367aa90b9b4526 : SUCCESS in 21m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/14faee3af15c4bd6ab83c9904314128d : SUCCESS in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24e92abc392269c84e4d22f9a8c47ac3a5b4c041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 23:43:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1321590203ab40eca2954c27790a2ffe\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d80a561b25194b6892f4400f846dd08e : SUCCESS in 2h 24m 17s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0cf68327f974736be34028169774444 : SUCCESS in 2h 32m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6271b2d3ad8245189a1070fe4058226b : SUCCESS in 49m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c33497550585452c90bff1c0c53e66cb : FAILURE in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18594237f1054735a5739aebbf9ce424 : SUCCESS in 22m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa1f265d8a7545c0b8c20cb6acf65645 : SUCCESS in 16m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb6f2f5b9b94990a5fc02b99d15fc68 : SUCCESS in 9m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/25efc2df35c6447ca2b619806ec8bc68 : SUCCESS in 56m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0a2a151130aa4a24929c5030f3edbba7 : SUCCESS in 1h 12m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0f18688887045598179de09fb180fc4 : SUCCESS in 1h 11m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5c51cc8a0a234e77ac2434131dfd72a8 : SUCCESS in 57m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/83a3c1a54bc143de8050b6655c522086 : SUCCESS in 1h 04m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f1c8ae580604495fbbd6473a7c712bfb : SUCCESS in 1h 24m 33s","accounts_in_message":[],"_revision_number":1},{"id":"72a31d8cee0941d5dfa07660c2e87d6657adc569","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-14 01:54:35.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/66/939166/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4b85502af268485bb0bef87b73789db974667f07","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-04-04 07:42:42.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f8fd77ce7bc0d8df77437da4a558466ca698b78","revisions":{"0f8fd77ce7bc0d8df77437da4a558466ca698b78":{"kind":"REWORK","_number":1,"created":"2025-01-13 20:47:10.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/939166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/939166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/939166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/939166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/939166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/939166/1"}}},"commit":{"parents":[{"commit":"96b951eb218eecbf0efaf3489b2b95fd9e3d2f37","subject":"Merge \"[eventlet-deprecation] Remove eventlet usage in ``ProcessMonitor``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96b951eb218eecbf0efaf3489b2b95fd9e3d2f37"}]}],"author":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2025-01-13 19:50:59.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2025-01-13 20:47:07.000000000","tz":-360},"subject":"fix: add try/except to allow maintenance to run on unmanaged segments","message":"fix: add try/except to allow maintenance to run on unmanaged segments\n\nThis change will allow neutron maintenance operate on OVN segments that may not be\nowned by neutron itself. This change is important when running an OVN cluster outside\nof OpenStack.\n\nChange-Id: I2895eb92be760c4bba7b22345d1802cf7cec1f94\nSigned-off-by: Kevin Carter \u003ckevin.carter@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f8fd77ce7bc0d8df77437da4a558466ca698b78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f8fd77ce7bc0d8df77437da4a558466ca698b78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
