)]}'
{"id":"openstack%2Fneutron~939439","triplet_id":"openstack%2Fneutron~master~Ic57a110046069414358a523a20af18e7d237a229","project":"openstack/neutron","branch":"master","topic":"ft","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-01-17 15:26:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14525\u003e replied on the change","reason_account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}},"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2025-03-20 04:20:43.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2025-01-20 14:57:52.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}}},"hashtags":[],"change_id":"Ic57a110046069414358a523a20af18e7d237a229","subject":"Increase ovs operation timeout for functional tests","status":"NEW","created":"2025-01-16 13:00:02.000000000","updated":"2025-03-20 04:20:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"3c022688773bd61a81012159efab807fcd209893","_number":939439,"virtual_id_number":939439,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-01-18 14:35:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-18 15:54:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-16 13:20:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-16 13:35:08.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2025-01-16 15:15:57.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-17 14:52:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-01-17 19:07:49.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2025-01-20 14:57:52.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REMOVED"},{"updated":"2025-03-20 04:20:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"94f28e3f6097331f58384c293c48a7de52c2e1e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-16 13:00:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc211d0e9eb314d443409b3e669dfa8e9d6fa015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:20:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8580c99d956a4869bfcdabe0a26565a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36794f5346ae45f79cf80f2cac288e0d : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4ecbc9c9ae446fa4f6dd2af3ed0b40 : SUCCESS in 8m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/83e4911e89b34f808cefff49da08c2ce : FAILURE in 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"308bfc53d16975bb402ce3d072a8e6c7ab6354cc","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-16 13:35:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"acccba7f41b383932b2607db54833fb35d44ced3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-16 14:55:03.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"23ae1ebef7c18aee231121d18d1d5b5a5b571c3f","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-16 14:55:12.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"fa5b65b4c3b53a04a41aa6a3a2422610dd5dc650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:14:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7020d42e3069416e829f4d925f95d8c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bfa5410e85d465dbddc5fd4644501c3 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9366ce951f084c41a3860f1a26c618c5 : SUCCESS in 11m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/59b9d3945e6641f08cc26d19e81f461c : FAILURE in 17m 11s","accounts_in_message":[],"_revision_number":2},{"id":"021d52e8418c2a910dc273b21e5c021626797014","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-16 15:15:57.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7482c28b11c63be3cc32c5816fa5c7aaefbcf929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-16 16:13:32.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0da6fb4e4dcf041ba84ab4a3654b7673a601217d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 17:51:27.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbec14bff2d442a58f59213637a91b93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5894dcff669746259bf8654d5fb5b26d : FAILURE in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c41536e97645539bb1df060e15ca46 : SUCCESS in 12m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a9f0058ed2534b7abb50a71e5c28dff7 : FAILURE in 52m 00s","accounts_in_message":[],"_revision_number":3},{"id":"59126003982328fe89255fcf9902aeee4d0a2cdd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-16 19:31:49.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"78015c0fa6bb9cf38bc1067dc765bce13d3f4621","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-16 21:06:09.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"8d11aac03f311447b1e9f0221510cd5c5bed39dd","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-17 06:23:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d0b6f91578443e0543c19a5260f2e45ef99360b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-17 14:47:58.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bb6582b66df12f978faf389e16dee938f58dbbad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-17 14:52:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d36a0faeba6dfff86a613599ef53030d9af306bb","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-17 15:26:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"89c0090adad89b5428271f6917a058e3149bdab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 16:10:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f495951de40b4062b4025a631d3738d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb30601b2592412c8fe2968573e3ce0d : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b80f6210fdb148c89b1421e8e9f3ce6d : SUCCESS in 10m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6f616d24cfeb4a17ad3919f267ea9796 : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":4},{"id":"d5997a27823460fd435f94ac1fe14fe7e47c82de","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-17 17:16:41.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ad9d062557371b5c33547ff1c8bddf9cd2ad0aac","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-17 18:57:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0b696233a5e52b715f014f0ed0529261f8d5b4c9","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-17 19:07:49.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"990511b9a5f9fb4ba8105d789904ca83a3aa1c41","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-17 20:20:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"acf7bbb08efd1f944f9ef46afb49220a4e6f2388","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-18 08:43:51.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1b0cbccb650249242285b04b4e2d7825f9df2dea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-18 09:15:49.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"d09799af69d8e504c9a4b20a5d1fe96c5154bf2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 09:45:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa2bb9f244784f5792d52fd62b0366ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10eb424453204338b2e04fda5a4f14f1 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eddbb779253a4c7cb2f2cc29079c203e : SUCCESS in 10m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b91e7a4a29e84ad8881c1de5dc33f935 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":5},{"id":"1c49922fd3500627b94bea3566424dbdd51e3f5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-18 09:57:03.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"065178cde83d0ac1bd0ce2187048c97e38675f32","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-18 09:57:08.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"5a1774ec555f2a25cd4bd7aae1bfb57b2f9046f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-18 10:28:52.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"7e27f73deed261f9369f87e9309df7ecd629f28b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-18 10:44:09.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ab06e172bcc9a313fef268057129eb3ea312a681","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-18 11:06:57.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"0b9768fe73baf64475e0ffb789be0ddfd5dfde0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 12:12:36.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f52b658a7444d51bc004c244ebb8380\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa05990800bd44929c1e47887dd943a9 : FAILURE in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a77bba49ae2414ba08141887885ae8d : SUCCESS in 6m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f29d3ba26e648aebac86434b3f66f10 : FAILURE in 55m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bc261a17791d42dc8a7309bf70eb6b02 : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":7},{"id":"43d57042cea93f411442518d934227c11bd1dc6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-18 12:28:11.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8d90ce81dc4cc9f5a01dfd1d2a77e3c77327aba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 12:31:58.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/939578 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"4032e65cbe5f260c827e13f7b8cd12bf42cf55be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-18 12:33:42.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"756a05232b82c8d3def1972ebce00f13055a18ea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-18 12:58:05.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"941265cc0545fe3800d2845ffc9f13832e673b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 13:07:21.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/939578 is needed.","accounts_in_message":[],"_revision_number":9},{"id":"fc59669a291a172d3afe82f6947b7becee8eb032","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-18 13:10:56.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"48d1a2a90debed56aae9b02f2798802b75335549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 13:26:47.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/939578 is needed.","accounts_in_message":[],"_revision_number":9},{"id":"6fa6b6372575450a6832bd6703af206ab2558586","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-18 13:29:57.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"d7dfb75de13262bdb4176c3f1e7cc6d130457392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-18 13:33:51.000000000","message":"Patch Set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"31d5cfa37306ef9236063815f86023bb71b2a9ed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-18 14:02:59.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"c0335775ecdd211fd1fdee89e2254b769b17d2ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-18 14:35:41.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/39/939439/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"3def3a9b46131e70b64b28c8e6c47303724356b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 15:54:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c918d545aa9145c4a8f58e1f71ba56b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b49cdb18f2944cfbef3d9f4dbe595d6 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d816669783400481d5900fe7a8aab2 : SUCCESS in 8m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/14b24c156d814735a54d725fb490e2ed : SUCCESS in 1h 11m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/00ecb7c3a60d46f2b2554a2e58590633 : SUCCESS in 2h 19m 47s","accounts_in_message":[],"_revision_number":10},{"id":"28c220e38e24fb5b18d14028110611e93d3bb1c8","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-18 15:58:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c595e40b6873f9b99ed8f3106e0164672ef77cdd","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2025-01-20 14:57:52.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_9656\u003e.","accounts_in_message":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}],"_revision_number":10},{"id":"3c022688773bd61a81012159efab807fcd209893","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-20 04:20:43.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"ebf72cb773a2f217803e1151478e5fafe0f1f845","revisions":{"ccfb39e543c65a50b00ed3583c934acfd62cf480":{"kind":"REWORK","_number":1,"created":"2025-01-16 13:00:02.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/1"}}},"commit":{"parents":[{"commit":"90e29e852eb22b6439d83652f9faaa069ad80dcf","subject":"Merge \"functional: Handle ovsdb monitor returning inserts in different checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90e29e852eb22b6439d83652f9faaa069ad80dcf"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccfb39e543c65a50b00ed3583c934acfd62cf480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccfb39e543c65a50b00ed3583c934acfd62cf480"}]},"branch":"refs/heads/master"},"8b68a26bb9ed568af1565666b5fcb51838936250":{"kind":"REWORK","_number":2,"created":"2025-01-16 14:55:03.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/2"}}},"commit":{"parents":[{"commit":"90e29e852eb22b6439d83652f9faaa069ad80dcf","subject":"Merge \"functional: Handle ovsdb monitor returning inserts in different checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90e29e852eb22b6439d83652f9faaa069ad80dcf"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 14:55:03.000000000","tz":0},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b68a26bb9ed568af1565666b5fcb51838936250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b68a26bb9ed568af1565666b5fcb51838936250"}]},"branch":"refs/heads/master"},"c0ead2fb617f2b59c614bf86bcab9e898153737d":{"kind":"REWORK","_number":3,"created":"2025-01-16 16:13:32.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/3"}}},"commit":{"parents":[{"commit":"90e29e852eb22b6439d83652f9faaa069ad80dcf","subject":"Merge \"functional: Handle ovsdb monitor returning inserts in different checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90e29e852eb22b6439d83652f9faaa069ad80dcf"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 16:13:32.000000000","tz":0},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0ead2fb617f2b59c614bf86bcab9e898153737d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0ead2fb617f2b59c614bf86bcab9e898153737d"}]},"branch":"refs/heads/master"},"cfdc2ff557838c517c3ea1a4a595cd056cd20322":{"kind":"REWORK","_number":4,"created":"2025-01-17 14:47:58.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/4"}}},"commit":{"parents":[{"commit":"90e29e852eb22b6439d83652f9faaa069ad80dcf","subject":"Merge \"functional: Handle ovsdb monitor returning inserts in different checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90e29e852eb22b6439d83652f9faaa069ad80dcf"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-17 14:47:58.000000000","tz":0},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfdc2ff557838c517c3ea1a4a595cd056cd20322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfdc2ff557838c517c3ea1a4a595cd056cd20322"}]},"branch":"refs/heads/master"},"7c60f7c62451c45815ff56ef1099fe22eba75b55":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-01-18 08:43:51.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/5"}}},"commit":{"parents":[{"commit":"90e29e852eb22b6439d83652f9faaa069ad80dcf","subject":"Merge \"functional: Handle ovsdb monitor returning inserts in different checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90e29e852eb22b6439d83652f9faaa069ad80dcf"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-18 08:43:51.000000000","tz":0},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939578\n\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c60f7c62451c45815ff56ef1099fe22eba75b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c60f7c62451c45815ff56ef1099fe22eba75b55"}]},"branch":"refs/heads/master","description":"Edit commit message"},"37a4d1504b67d248ed6d136f94b1a29d96e29b55":{"kind":"REWORK","_number":6,"created":"2025-01-18 09:57:03.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/6"}}},"commit":{"parents":[{"commit":"90e29e852eb22b6439d83652f9faaa069ad80dcf","subject":"Merge \"functional: Handle ovsdb monitor returning inserts in different checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90e29e852eb22b6439d83652f9faaa069ad80dcf"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-18 09:57:03.000000000","tz":0},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939578\n\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37a4d1504b67d248ed6d136f94b1a29d96e29b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37a4d1504b67d248ed6d136f94b1a29d96e29b55"}]},"branch":"refs/heads/master"},"7fd5aa450b9fb3f3d44eefcf59e0b97f4c82aa03":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-01-18 10:44:09.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/7"}}},"commit":{"parents":[{"commit":"90e29e852eb22b6439d83652f9faaa069ad80dcf","subject":"Merge \"functional: Handle ovsdb monitor returning inserts in different checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90e29e852eb22b6439d83652f9faaa069ad80dcf"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-18 10:44:02.000000000","tz":0},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939578\n\nRelated-Bug: #2095224\n\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fd5aa450b9fb3f3d44eefcf59e0b97f4c82aa03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fd5aa450b9fb3f3d44eefcf59e0b97f4c82aa03"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cad00b822ccbcf14208c50ead31fe6fd46a55d8e":{"kind":"REWORK","_number":8,"created":"2025-01-18 12:28:11.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/8"}}},"commit":{"parents":[{"commit":"90e29e852eb22b6439d83652f9faaa069ad80dcf","subject":"Merge \"functional: Handle ovsdb monitor returning inserts in different checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90e29e852eb22b6439d83652f9faaa069ad80dcf"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-18 12:28:11.000000000","tz":0},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939578\n\nRelated-Bug: #2095224\n\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cad00b822ccbcf14208c50ead31fe6fd46a55d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cad00b822ccbcf14208c50ead31fe6fd46a55d8e"}]},"branch":"refs/heads/master"},"c1fab1b3cd43c278e4c63c7354a56e990db8dbb4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-01-18 12:33:42.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/9"}}},"commit":{"parents":[{"commit":"93032d8ab03f86883675655e46b72b8022cd9f95","subject":"Merge \"Add option to configure live migration activation strategy for OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93032d8ab03f86883675655e46b72b8022cd9f95"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-18 12:33:42.000000000","tz":0},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939578\n\nRelated-Bug: #2095224\n\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1fab1b3cd43c278e4c63c7354a56e990db8dbb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1fab1b3cd43c278e4c63c7354a56e990db8dbb4"}]},"branch":"refs/heads/master","description":"Rebase"},"ebf72cb773a2f217803e1151478e5fafe0f1f845":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2025-01-18 13:33:51.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/39/939439/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/939439/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/939439/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/939439/10"}}},"commit":{"parents":[{"commit":"93032d8ab03f86883675655e46b72b8022cd9f95","subject":"Merge \"Add option to configure live migration activation strategy for OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93032d8ab03f86883675655e46b72b8022cd9f95"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-16 12:59:20.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-18 13:33:50.000000000","tz":0},"subject":"Increase ovs operation timeout for functional tests","message":"Increase ovs operation timeout for functional tests\n\nFunctional tests are not stable and failing on slow nodes,\nincrease timeout to 15 seconds.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939578\n\nRelated-Bug: #2095224\nChange-Id: Ic57a110046069414358a523a20af18e7d237a229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebf72cb773a2f217803e1151478e5fafe0f1f845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebf72cb773a2f217803e1151478e5fafe0f1f845"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
