)]}'
{"id":"openstack%2Fneutron~939918","triplet_id":"openstack%2Fneutron~master~I84db3cb1b849bff29f4f67003e8b2847375f85ed","project":"openstack/neutron","branch":"master","topic":"fip-update","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-14 15:04:34.000000000","reason":"Change was abandoned"},"31291":{"account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"last_update":"2025-05-06 06:28:38.000000000","reason":"\u003cGERRIT_ACCOUNT_31291\u003e replied on the change","reason_account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-05-06 00:35:04.000000000","reason":"removed on reply"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-04-14 15:04:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I84db3cb1b849bff29f4f67003e8b2847375f85ed","subject":"Update NAT entry on FIP update","status":"ABANDONED","created":"2025-01-22 22:17:18.000000000","updated":"2026-04-14 15:04:34.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d48fc544fbf7c8d5c723e286a49c827755fef197","_number":939918,"virtual_id_number":939918,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-04-14 16:10:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-14 17:24:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"recommended":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-06 06:28:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":2,"date":"2025-05-06 00:35:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-22 22:57:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-22 23:35:26.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-23 00:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-27 14:47:12.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-03-20 04:07:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-04-08 10:29:59.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"REVIEWER"},{"updated":"2025-05-06 00:35:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"37049d0cd30e3ecd5139f56840e65d170c60a294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-01-22 22:17:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07600214bcb5496bf43178bd6e650c08e761b1db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 22:57:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3efe249d5c9d4d63a564176d5ec55091\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a6de1810c524876b7ec05aac120233d : FAILURE in 26m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eabbbc1c63ff44af88317947f63c68b7 : FAILURE in 30m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"180e60726a70135de42d95b34fd1e0832b4427c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 22:57:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3efe249d5c9d4d63a564176d5ec55091\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a6de1810c524876b7ec05aac120233d : FAILURE in 26m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eabbbc1c63ff44af88317947f63c68b7 : FAILURE in 30m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1facb2b0b536d5516082d7452a8c91c5235697f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-22 23:35:26.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/18/939918/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"30265bd36bb832e0d30ac9c36899685dc30968f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 00:01:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddf08711c2784c2daf21329c28dff54a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b058e66ed0f04a35b5a64c81ad44af0b : SUCCESS in 1h 43m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c8bf445d20945bc9a1554237292f956 : SUCCESS in 1h 40m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20574b82f7034b28b87fbac50def7625 : FAILURE in 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b7ddff44854a998a2ed8658c3b2f81 : SUCCESS in 11m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f883d3d1bc74807b460668984a9feba : FAILURE in 27m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d866d505fc41478785767079f3410c57 : FAILURE in 19m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4423709c94714df1ab033f3b95e9f6d7 : SUCCESS in 9m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1fbd263637be4780bdaf03ea8df9e827 : FAILURE in 1h 07m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/70da52bc80cf4602a08dcc7d12505b36 : FAILURE in 28m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3fe0b77203848268184188a6137f2f7 : SUCCESS in 1h 08m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8ca51e2737df4588beb43b3a3c1b8c9c : FAILURE in 1h 02m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e0b9e39a2b62420f8d44f7590e5b6bfb : SUCCESS in 1h 15m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cfc9f31cac2f416abaeb8e6e5504583c : FAILURE in 1h 18m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fec8ddf1e75d8cbcb724b11b8f7648fc3c67d2fc","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-01-23 00:16:02.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"6a5c5f1e7f6eeedfe5cef44bbb3ac60f67ac4c41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-01-23 21:11:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"18c67b30d671cb557e78b575c45adf32b92a95a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 23:16:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8f574cf92ff4d7bb0cc0aab70cf78e0\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1d6484f7d96e43c1b2bd58b649a1ff1b : FAILURE in 2h 03m 43s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d7d5679e8fb47ff8d17353e0effdbff : FAILURE in 1h 52m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c791676cef9468abbf6f8b2d33e5c16 : SUCCESS in 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c231f5cd7e484b868012c655bf63114f : SUCCESS in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0196dca4a4ff4e20a36e7e2a5c35d591 : SUCCESS in 22m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f13a82acc4f443bf86e859522faa9d9f : SUCCESS in 16m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7639f070b84ea680f20c9381e76c01 : SUCCESS in 9m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/784a03ebaf7f4345aad39e816569ac68 : SUCCESS in 39m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e2968262eb8f43619fbd8cf59d50fe79 : FAILURE in 30m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9d8f59933a84ac688f2c358e29b8655 : SUCCESS in 47m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3d583e47d0bf4791b52756df3d971cc7 : SUCCESS in 1h 18m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8d0d67b8c02e4112b28f0b7d2c6e56b5 : SUCCESS in 1h 13m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a753e251a3bc4e519584d7b77738ff15 : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":2},{"id":"45389a1bf51a34d90b868133cc59fda5ba0cd382","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 00:39:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c03a6d883cda407ebd502e26f938ec90\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90f6047224f24025b3be7b2edc14e90a : SUCCESS in 43m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60570ef3381e4d1e90ad0ba8fdeae5f0 : SUCCESS in 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"52302d296940a232f4894cf8be866876b498fbd6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-24 00:41:29.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/18/939918/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a3191cc261daa05a2ea7c5ea85a0653ebe0f0566","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-01-24 13:10:44.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a7614ea0f97bca422712b837f733eb566feebbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-01-24 21:47:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"de8fbfa3f8a3312b23522e8f96d650bf658ff889","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 23:37:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad70588df30483d8b71baeb73dd8b66\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c92a5b6d12b4c1eaf86adaf509ba42d : SUCCESS in 36m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3ac1fe3875054ca89135f55420a6972a : SUCCESS in 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f82be779e29563d0e79b714d514c6a39d4acd30e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 23:39:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6bce1723e7d4a46b953ad427d285a4e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bd01a9d3159e43dda683137233e3e52f : SUCCESS in 1h 46m 19s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2f6799c5ff84fcf8885d4398c035db2 : SUCCESS in 1h 32m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e52a8cab922c428fb346e7d0d7915b0a : SUCCESS in 32m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fab037718f849daae5e76fa84c899ff : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7bcf289be0b499dae0997f89ea6e291 : SUCCESS in 23m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/225a744ec95e45d2b7e31dbeaf4f6562 : SUCCESS in 21m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38b76084c49d448293c3335082783999 : SUCCESS in 12m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5d14d9e32c3148889f940841941fbe53 : SUCCESS in 49m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/826f1bc35ed14e30b4b4ffb2dbaa3032 : SUCCESS in 1h 43m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec9c8e27229d4540a8666899c5c62b54 : SUCCESS in 1h 15m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8f4a33eca1664729a9b1f611d66a1a43 : SUCCESS in 1h 34m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/742339b282214ec590d260b0067031f4 : SUCCESS in 1h 36m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/32eb90371e584deab70ec159c6c157de : SUCCESS in 1h 45m 13s","accounts_in_message":[],"_revision_number":3},{"id":"efc410d72d039a1f4d81d83565e93bb1f51a4787","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-25 03:33:13.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/18/939918/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f7b9ee6bc5bd77687a7ae31afe41967c63b52bf0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-20 04:07:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f959678067559731f91ba2452165303c22de994","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-04-08 10:29:59.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"806f152a3a170623fed86919d4edfc159a50ca25","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-08 11:41:55.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b7e39f7389f1d1f4739f5d9b8855dd59f0952aa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-04-14 15:21:05.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"517248285590c4c17a28f8ae8a080f586983946b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 15:56:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09359b57c0954cf79dfee31553a29503\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32904000983b4854942169915a10ddf1 : SUCCESS in 33m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3867a3f6f87549a9aa7c4be66f1f1ede : SUCCESS in 18m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/153e26d0d3514551a59d1e6cd41e4037 : SUCCESS in 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0c13cb922bb44c4483ed086def6b8d3a03db67aa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-14 16:10:27.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/18/939918/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f498ac40cf6aa680697d342df08487266621865c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 17:24:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d137f1709a1d41fca0669658f58b6d2e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b9af001bab414989a08f1011bae669e5 : SUCCESS in 2h 01m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/059ee3745e614fd3836ac3355490cdea : SUCCESS in 35m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aebfe204c2244b3fa7f2a7cb7833a8e5 : SUCCESS in 11m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84e6c75c02814884a0b179885649b374 : SUCCESS in 21m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dff77de924d6421f9ff76de897f299f7 : SUCCESS in 14m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe8fa80ac5ba440dabda855b7dc99c5a : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f30ab141064cc1a4bceba133db919c : SUCCESS in 13m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ae35e162c2084009bf7814e11a335187 : SUCCESS in 1h 01m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/619cdac82c684421bb27476b85f43d30 : SUCCESS in 34m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c260dd66d43e408b8274d405432e6b29 : SUCCESS in 1h 01m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef01c686c86b4b7ab1a185eec25843b9 : SUCCESS in 31m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/30d7178720e543ffb9e7f07a3868afc8 : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":4},{"id":"8ce7ddb45f717494d63adbd938d927321a65a8f7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-06 00:35:04.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bb254771cef3be24434ca670f603405485f85319","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-05-06 06:28:38.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d48fc544fbf7c8d5c723e286a49c827755fef197","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-14 15:04:34.000000000","message":"Abandoned\n\nWith core-BGP this not longer creates an issue. Would be a nice-to-have fix to have the behavior consistent","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"12bdad30bf976b8bec05ab9a43742c79ebc3854d","revisions":{"89133525a3ba58eccd0fc6d9049521a83ad8d595":{"kind":"REWORK","_number":1,"created":"2025-01-22 22:17:18.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/18/939918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/939918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/939918/1"}}},"commit":{"parents":[{"commit":"93032d8ab03f86883675655e46b72b8022cd9f95","subject":"Merge \"Add option to configure live migration activation strategy for OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93032d8ab03f86883675655e46b72b8022cd9f95"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-01-17 23:37:16.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-01-22 22:16:41.000000000","tz":0},"subject":"Update NAT entry on FIP update","message":"Update NAT entry on FIP update\n\nPreviously whenever a floating IP was updated with change such as\nchange of the associated port the exisint NAT entry representing the FIP\nin the OVN database was deleted and then created again in the same\ntransaction.\n\nThis behavior generates deleted created ovsdb events and eventual\nconsistancy is not guaranteed meaning that whoever listens to such DB\nchanges cannot tell what happened first.\n\nThis patch simplifes the code so whenever a change to the NAT entry is\nneeded it uses the db_set() command to change only relevant columns in\nthe existing NAT entry.\n\nChange-Id: I84db3cb1b849bff29f4f67003e8b2847375f85ed\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89133525a3ba58eccd0fc6d9049521a83ad8d595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89133525a3ba58eccd0fc6d9049521a83ad8d595"}]},"branch":"refs/heads/master"},"8689f480f5a01602980752bbb65898399716505e":{"kind":"REWORK","_number":2,"created":"2025-01-23 21:11:44.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/18/939918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/939918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/939918/2"}}},"commit":{"parents":[{"commit":"93032d8ab03f86883675655e46b72b8022cd9f95","subject":"Merge \"Add option to configure live migration activation strategy for OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93032d8ab03f86883675655e46b72b8022cd9f95"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-01-17 23:37:16.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-01-23 20:02:37.000000000","tz":0},"subject":"Update NAT entry on FIP update","message":"Update NAT entry on FIP update\n\nPreviously whenever a floating IP was updated with change such as\nchange of the associated port the exisint NAT entry representing the FIP\nin the OVN database was deleted and then created again in the same\ntransaction.\n\nThis behavior generates deleted created ovsdb events and eventual\nconsistancy is not guaranteed meaning that whoever listens to such DB\nchanges cannot tell what happened first.\n\nThis patch simplifes the code so whenever a change to the NAT entry is\nneeded it uses the db_set() command to change only relevant columns in\nthe existing NAT entry.\n\nChange-Id: I84db3cb1b849bff29f4f67003e8b2847375f85ed\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8689f480f5a01602980752bbb65898399716505e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8689f480f5a01602980752bbb65898399716505e"}]},"branch":"refs/heads/master"},"57051ee0d8366e74d6c20eaf2eb021682e591962":{"kind":"REWORK","_number":3,"created":"2025-01-24 21:47:29.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/18/939918/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/939918/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/939918/3"}}},"commit":{"parents":[{"commit":"93032d8ab03f86883675655e46b72b8022cd9f95","subject":"Merge \"Add option to configure live migration activation strategy for OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93032d8ab03f86883675655e46b72b8022cd9f95"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-01-17 23:37:16.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-01-24 17:17:19.000000000","tz":0},"subject":"Update NAT entry on FIP update","message":"Update NAT entry on FIP update\n\nPreviously whenever a floating IP was updated with change such as\nchange of the associated port the exisint NAT entry representing the FIP\nin the OVN database was deleted and then created again in the same\ntransaction.\n\nThis behavior generates deleted created ovsdb events and eventual\nconsistancy is not guaranteed meaning that whoever listens to such DB\nchanges cannot tell what happened first.\n\nThis patch simplifes the code so whenever a change to the NAT entry is\nneeded it uses the db_set() command to change only relevant columns in\nthe existing NAT entry.\n\nChange-Id: I84db3cb1b849bff29f4f67003e8b2847375f85ed\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57051ee0d8366e74d6c20eaf2eb021682e591962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57051ee0d8366e74d6c20eaf2eb021682e591962"}]},"branch":"refs/heads/master"},"12bdad30bf976b8bec05ab9a43742c79ebc3854d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-04-14 15:21:05.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/18/939918/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/939918/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/939918/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/939918/4"}}},"commit":{"parents":[{"commit":"93032d8ab03f86883675655e46b72b8022cd9f95","subject":"Merge \"Add option to configure live migration activation strategy for OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93032d8ab03f86883675655e46b72b8022cd9f95"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-01-17 23:37:16.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-04-14 15:21:05.000000000","tz":0},"subject":"Update NAT entry on FIP update","message":"Update NAT entry on FIP update\n\nPreviously whenever a floating IP was updated with change such as\nchange of the associated port the exisint NAT entry representing the FIP\nin the OVN database was deleted and then created again in the same\ntransaction.\n\nThis behavior generates deleted created ovsdb events and eventual\nconsistancy is not guaranteed meaning that whoever listens to such DB\nchanges cannot tell what happened first.\n\nThis patch simplifes the code so whenever a change to the NAT entry is\nneeded it uses the db_set() command to change only relevant columns in\nthe existing NAT entry.\n\nRelated-bug: #2093679\n\nChange-Id: I84db3cb1b849bff29f4f67003e8b2847375f85ed\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12bdad30bf976b8bec05ab9a43742c79ebc3854d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12bdad30bf976b8bec05ab9a43742c79ebc3854d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
