)]}'
{"id":"openstack%2Fneutron~940123","triplet_id":"openstack%2Fneutron~master~I59b25093df9cf5aa77767492a5b9008bfa11cc07","project":"openstack/neutron","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-01-28 21:13:20.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-01-27 19:58:55.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I59b25093df9cf5aa77767492a5b9008bfa11cc07","subject":"[eventlet-removal][OVN] Require wsgi start-time in the config","status":"MERGED","created":"2025-01-24 12:05:12.000000000","updated":"2025-01-28 21:14:35.000000000","submitted":"2025-01-28 21:13:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940123-eventlet-removal","meta_rev_id":"4b531c8cc22741b0195189378cb3d0bd0880d56c","_number":940123,"virtual_id_number":940123,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-01-27 18:16:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-28 21:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-28 18:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2025-01-27 19:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-28 18:53:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-24 13:33:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-24 14:18:44.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-24 15:30:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-01-27 19:58:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-01-28 18:53:22.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"916acc66b58e4a28a059e6d192c92406fa2ab450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-24 12:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0603efff9ec497f852f6c18e49e6807cec203cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 13:33:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be539bccfd4e40e982681e554d289033\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8061e90b5f9849ed9a3bbd297adc582e : FAILURE in 35m 02s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/653649d54aae4b8eb89a3d3160ce770e : FAILURE in 31m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b86141ed34d047ada2adbf673a821761 : FAILURE in 40m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc52a7ff91c2403093d04232f3491d81 : SUCCESS in 11m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d630e4dc41749ed97b5f195a32728c5 : FAILURE in 17m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84ec037915be433fa3a7e0b295ef2443 : FAILURE in 20m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93dbb0b918b462587b3e67efac3aa85 : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/643e50dbeac94f4995e94ffa5206a5ae : SUCCESS in 15m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/afb711a93a9a4829876d757523c804d5 : FAILURE in 1h 09m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0dcebadfba7a49f58b8013d9ef2c5fd8 : FAILURE in 41m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/232cb1a7c5f34600a5d018a8ad684cba : SUCCESS in 1h 15m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8c8a58bc39fc4b39b38580ffdba925cd : FAILURE in 13m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/42bc26ca6bd54d87ade861242269609e : FAILURE in 17m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e85a90f3b1f1479cbd4aa2ef1bb888fc : FAILURE in 14m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f561da694a890b95e74359ed1acfe10078bd5940","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-24 14:18:44.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/23/940123/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"42dc1d9ffcd49fa44c0179e6d7823d87caef7f61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-24 14:41:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0e6f27673b4843243f47c1baf3b905e6f658372d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-24 16:28:22.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/23/940123/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"cb3e26120512961482b0598ce1ede8ec8316f084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 17:03:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78c9edc045fd4706a0c6c51e65f90bb5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ec0a0cb9739a49efab5f42ac8cc3298d : SUCCESS in 1h 54m 46s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e7c3ea29db94cb5bd68b7edc83514cb : SUCCESS in 1h 59m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65d68dda9d794e85b9d5c23e795b3749 : SUCCESS in 48m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d7fc1bda734880ae3b743f5c8aa553 : SUCCESS in 11m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0820c420ce449369c8baa55ce5e99dd : SUCCESS in 22m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b38099070fb4cf894bc62ac24d0000c : SUCCESS in 19m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd67c45743b9477ca926896bc6009ee5 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f219dcd2166345f5a0b95a12829201f1 : SUCCESS in 10m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a46c2ea27fa843d4a98d2c0182e076b2 : FAILURE in 54m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ac021293f2c44772b95994628c7834eb : TIMED_OUT in 2h 12m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb190b5690ac4cc4af8d11fc74f90cb9 : SUCCESS in 1h 16m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b1ee5117f0ec40f8bdcbb8607ffb5198 : SUCCESS in 59m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f6f8d28e962c4d38befecdb7c86c2933 : SUCCESS in 1h 28m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2cdb68263e684453ab518cf905ab36d2 : FAILURE in 18m 14s","accounts_in_message":[],"_revision_number":2},{"id":"37a10180f0ed571c324c5d0977ca2a0287604757","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 18:40:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bf5f50da2e14c9d830a723b23e3d99c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3f0e143fb59f48759c9c3c4d28f99e97 : SUCCESS in 30m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b0c7ea824e544cba4371a43d551e0b6 : SUCCESS in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7c40eec14a365f4759522752da6d009a75c7f79d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-25 19:47:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c94c40b3e755adeefe1167e703742e15739fa02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 20:21:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c0dfd81ae324781b6c2209790ecfd1e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25bf48fe30124b60aba8246226bc6dd1 : SUCCESS in 25m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/028161a1fcce4ef4b12f9b7b449930e7 : SUCCESS in 31m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1a44d99d1b848b57519a5d3b5e23da3ccdf1884","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-25 20:48:34.000000000","message":"Patch Set 2:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/23/940123/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0b57372c65afcbf9982a0a868077d6f979ad7a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 22:09:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fa3bf9692aa409cade7b3187ae19ba4\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d58a2de445a642d7bb21cdc7635c84b3 : SUCCESS in 2h 07m 35s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d690200a0b674a35bb004383905963b4 : SUCCESS in 1h 46m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfc1db30ec3d420096550e14ac224752 : SUCCESS in 51m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206dd50b491a47bcb342e9e7ae6c75d4 : SUCCESS in 10m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3ae11047f184e25bbfe32d906efb226 : SUCCESS in 17m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c2c4081af1a4126a7dbad0d4b41b28a : SUCCESS in 19m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/539a9bb33b744c10aad228bf4fc68086 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78e8d99f901f400482437a249d8ae653 : SUCCESS in 19m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/37c90795ced84505949a0effb6f11eae : FAILURE in 36m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d456ea7d78044322922d3a9843ab73c4 : TIMED_OUT in 2h 12m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4a931e90fe74ecf9bc1d8bef687437f : SUCCESS in 55m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ef1d48aad9e84091a6a6919f72eb7749 : SUCCESS in 51m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/66b9fbcb1f8e42b09862c885700c6b3f : SUCCESS in 43m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/76e5520b60234217be40232e9a71a610 : FAILURE in 14m 44s","accounts_in_message":[],"_revision_number":2},{"id":"efc0c958bce0ff66e7ba0968d290c3e5f7c1a33a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-27 07:14:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"43f5429765b92db916d4d5d7d608c888b26f4e8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 07:46:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aeb3da044c74e18b776058208e28624\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3464d4a09b174a568a14fa2a9fd474dd : SUCCESS in 30m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/797a283ddd44444d85eadf2c466e4cb2 : SUCCESS in 17m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3039b4452a6dd06b47a2d1fd5530fe7ed3e395a5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-27 07:47:13.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/23/940123/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"653d03cfa65edc8f0c053441fbf774fa1855a621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 09:29:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be34aaf7e39d4d719ab557623fb2c3c3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/29a42d6926504c2ba175ff0b522098c2 : SUCCESS in 1h 42m 49s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/774f57faf2d44fc98794e6375369f64b : SUCCESS in 2h 03m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4d8faf7155342eaada867ede8254ee4 : FAILURE in 50m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98c2464e79464c4d96fb0eea1f438905 : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d77c608fa5994222af90d132d3224619 : SUCCESS in 22m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4cd3937ac8f3467eb0e86f85f0a7684f : SUCCESS in 19m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec9eaf9a35e433b9530888047e1a11a : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93bd14cf2684f69ab8b5d2187ade328 : SUCCESS in 9m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c8f08973f2cd46e082f05a5fed859ce8 : SUCCESS in 1h 04m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2285d291d3c04e2cb5a6d5ab7463e7ba : SUCCESS in 1h 01m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/54f5fbc04eb54526b1e0b9b767340102 : SUCCESS in 1h 10m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/303dffbdf27a47e2b305fa966dc80f1f : SUCCESS in 1h 08m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f070302a3b24b2f87ed6586b707eba6 : SUCCESS in 1h 11m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f37ea3301cbc45298206713f3937168f : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":3},{"id":"dafb0447a2290999e5d5545a25a6105d6dc6bf47","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-27 09:30:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4963299316364d3bee10adae9868698d0131d66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 10:24:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13220abe3d3a4da4a7a02430594f2051\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20b6e8a60bab4adba6d66fd312a07f42 : SUCCESS in 22m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7970df78a317481488c1bdecf712ab2e : SUCCESS in 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c46c62eafbd8d86e1a6a7bb5a29ee5d9310b671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 11:34:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cb22e20d8a5401098ccd79a1adc4b1f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53c05319e67f46afa85180f8544300c0 : SUCCESS in 1h 37m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6eab6f832ca041148146b1545ac14e9b : SUCCESS in 2h 01m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/973e5a3f4dd34e55854a82146269c392 : SUCCESS in 47m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3306bec4b51458c9234ba473d034232 : SUCCESS in 12m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b74cf4399bde4b8b9d72822dc1bd4696 : SUCCESS in 26m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b5ab53cab5446f280240a14feb51cc7 : SUCCESS in 18m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa5bd9147ff4eb5b2d1bc74a99a63fe : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73a70f830f0a4a6c90c42bcffad7aa34 : SUCCESS in 23m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ef80d688ba0543ce83c7c7b1ad24504f : SUCCESS in 1h 04m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b3162128570547409c87a2253e3d9a5a : SUCCESS in 1h 01m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/28db5fadba414b15a4b69dcb176037c2 : SUCCESS in 1h 23m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e368ccea74d340c98f32d441a65e7ad5 : SUCCESS in 1h 15m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb912cd196c943f390efac28d86b3152 : SUCCESS in 1h 04m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d258302d1f84ff5b6a13914fc06347d : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":3},{"id":"163d8683bcb529adf6355575739a6788de53ba57","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-27 13:20:01.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/23/940123/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"fbf7447e60eec56fd22509a27a299ff456771783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-27 13:51:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e348773451887b434d2f85c6472569cd3e996e50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-27 14:10:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"09b567fd0623374cabd8aabb85ba02026519a410","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 14:43:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe57ac1870384b7f8e99dfa01723b16d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9e65054cc92458cae6073aac28a4e6c : SUCCESS in 31m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/38ade72e71464207ab54ec54a245ac51 : SUCCESS in 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"70edc7da5a5dd54a1b98218fa897a6c4b952fb6d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-27 15:30:25.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/23/940123/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"2b21aaf06d5f0e0483a96ab2ec5927eb35d9b0dd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-27 16:03:00.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/23/940123/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"19968f4c1faf2a3df5f6a87339439572709ce555","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-27 16:11:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d37fa59bad09d55ca69e95797a4c17b3e2c7facf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-27 16:18:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"284d35f70be19f8429bcf343f6e8f74d1541c583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-27 16:18:16.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"9df252d0f9dae8bddff3120c2315311eae53f89f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 17:34:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef096f3b53eb498c95817592450b493f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3baaad134245470eae96a5c5103a5559 : SUCCESS in 28m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc7f7961dc894537b67687a3951f980c : SUCCESS in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"829a70b6a2016ededa63ea8ef77691585b2cc48b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-01-27 18:16:20.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2025/Arista_Tempest_Testrefs/changes/23/940123/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"5b4c12cc7f4e751715fc9265073b92ae3d7a102c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 18:34:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e628800dd14753b4139e242442edd8\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c3c48124c4cc4ffd8c4bef24589bb1af : SUCCESS in 2h 13m 37s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6874f9e1eee4937aca79179d16a1167 : SUCCESS in 2h 10m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da96dd36271748bbb4e209556b122efe : SUCCESS in 32m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a3b2ccc734a4644ad8bafac2881d0ab : SUCCESS in 11m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1482fad99de744cc961cb7701b514c75 : SUCCESS in 18m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/233b68bc258844af94a712ff0e49f07d : SUCCESS in 18m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7f11580d78408ab5cc50db78eab2fe : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71dfcbefcb1748b8b870184b5e9e8aff : SUCCESS in 12m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b67e209f4eb543b0bf969e30fdedf767 : SUCCESS in 59m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/68bc80834e994321afbeec3eeab0e751 : SUCCESS in 59m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b645cd28536b4af79c97d085953e0f91 : SUCCESS in 1h 16m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/97200be92a7b4c00bcdd48a390f5b5c2 : SUCCESS in 55m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6cce7072556a439eae65b6a66c712c4c : SUCCESS in 1h 13m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8a7449dd485c4738bbc471ea1541e179 : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":6},{"id":"ffbc20400c72f64ca737e8bd76f5a224d3c5f046","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-27 19:58:55.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"98846aa58476b265a877c237d268af3e40103560","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-01-28 18:53:22.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2392427bf97df96194c8ef37bfc445d32ac332e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 18:54:05.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4339b27d4e8e3a273f139ac8b2abf41732328a79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 21:13:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b34819db7934e1d83a0c58848d766dc\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b664d6359d2c44c1be5b350336b062f2 : SUCCESS in 2h 18m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd7a921676e4858b1f98a675364d8cb : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb5feb1592d24d3c96a79e037a7b63ba : SUCCESS in 18m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5bf861ba90449a4a3da4a3f24b0cf53 : SUCCESS in 19m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/035d1f888bf54a1098b8fdd8f588b647 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8808c7cc61a4160b7c263ff40bac7e0 : SUCCESS in 10m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/279325ee1f0c4d8bb0aabce6464b2c88 : SUCCESS in 58m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/994684254ce04a03a33808c8da1c4a59 : SUCCESS in 1h 15m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/13ffab124e3747089589ce701937e99d : SUCCESS in 48m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aa0a248be0d44568b3f57a0b68ed502f : SUCCESS in 1h 15m 43s","accounts_in_message":[],"_revision_number":6},{"id":"d73d73f15e6c78b3f0fe4bce611dac3987645162","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 21:13:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"4b531c8cc22741b0195189378cb3d0bd0880d56c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 21:14:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52d3299d3a604a7f87bb66e82d9db3df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94332be3556b40d9bc91d7aa5af32418 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4830587f13894244ba745c54f6790cbb : SUCCESS in 53s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b0c02c28929dedb4ed8a460994e71e72282058c5","revisions":{"580571b772873206f213d9cbd1418f3a9286c4f2":{"kind":"REWORK","_number":1,"created":"2025-01-24 12:05:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/940123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/940123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/940123/1"}}},"commit":{"parents":[{"commit":"3ff599700ca5591c244efd87c4cd2ef6dcf10c48","subject":"[eventlet-removal] Remove the usage of eventlet in the Neutron API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ff599700ca5591c244efd87c4cd2ef6dcf10c48"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-24 11:52:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-24 12:04:47.000000000","tz":0},"subject":"[eventlet-removal][OVN] Require wsgi start-time in the config","message":"[eventlet-removal][OVN] Require wsgi start-time in the config\n\nCurrently there is one single method to run the Neutron API that is\nusing uWSGI. This method requires a specific hash ring manager\ninitialization, different to the eventlet server based.\n\nBecause the second has been removed and is no longer available, as\nlong as Neutron continues the effors of removing eventlet from the\nsource code, the second way to initialize the hash ring manager\nis removed.\n\nRelated-Bug: #2083570\nChange-Id: I59b25093df9cf5aa77767492a5b9008bfa11cc07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/580571b772873206f213d9cbd1418f3a9286c4f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/580571b772873206f213d9cbd1418f3a9286c4f2"}]},"branch":"refs/heads/master"},"031ec3dbdd1b42ae03ddc58e4dabbac5f55b4fd1":{"kind":"REWORK","_number":2,"created":"2025-01-24 14:41:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/940123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/940123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/940123/2"}}},"commit":{"parents":[{"commit":"3ff599700ca5591c244efd87c4cd2ef6dcf10c48","subject":"[eventlet-removal] Remove the usage of eventlet in the Neutron API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ff599700ca5591c244efd87c4cd2ef6dcf10c48"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-24 11:52:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-24 14:41:41.000000000","tz":0},"subject":"[eventlet-removal][OVN] Require wsgi start-time in the config","message":"[eventlet-removal][OVN] Require wsgi start-time in the config\n\nCurrently there is one single method to run the Neutron API that is\nusing uWSGI. This method requires a specific hash ring manager\ninitialization, different to the eventlet server based.\n\nBecause the second has been removed and is no longer available, as\nlong as Neutron continues the effors of removing eventlet from the\nsource code, the second way to initialize the hash ring manager\nis removed.\n\nRelated-Bug: #2083570\nChange-Id: I59b25093df9cf5aa77767492a5b9008bfa11cc07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/031ec3dbdd1b42ae03ddc58e4dabbac5f55b4fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/031ec3dbdd1b42ae03ddc58e4dabbac5f55b4fd1"}]},"branch":"refs/heads/master"},"56b89d2628e78c6efdc9f17079439775d69d5116":{"kind":"REWORK","_number":3,"created":"2025-01-27 07:14:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/940123/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/940123/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/940123/3"}}},"commit":{"parents":[{"commit":"3ff599700ca5591c244efd87c4cd2ef6dcf10c48","subject":"[eventlet-removal] Remove the usage of eventlet in the Neutron API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ff599700ca5591c244efd87c4cd2ef6dcf10c48"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-24 11:52:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-27 07:14:05.000000000","tz":0},"subject":"[eventlet-removal][OVN] Require wsgi start-time in the config","message":"[eventlet-removal][OVN] Require wsgi start-time in the config\n\nCurrently there is one single method to run the Neutron API that is\nusing uWSGI. This method requires a specific hash ring manager\ninitialization, different to the eventlet server based.\n\nBecause the second has been removed and is no longer available, as\nlong as Neutron continues the effors of removing eventlet from the\nsource code, the second way to initialize the hash ring manager\nis removed.\n\nRelated-Bug: #2083570\nChange-Id: I59b25093df9cf5aa77767492a5b9008bfa11cc07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56b89d2628e78c6efdc9f17079439775d69d5116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56b89d2628e78c6efdc9f17079439775d69d5116"}]},"branch":"refs/heads/master"},"1178571d37f088eb2a771f7e7846f12f9daf5a9b":{"kind":"REWORK","_number":4,"created":"2025-01-27 13:51:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/940123/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/940123/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/940123/4"}}},"commit":{"parents":[{"commit":"3ff599700ca5591c244efd87c4cd2ef6dcf10c48","subject":"[eventlet-removal] Remove the usage of eventlet in the Neutron API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ff599700ca5591c244efd87c4cd2ef6dcf10c48"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-24 11:52:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-27 13:51:13.000000000","tz":0},"subject":"[eventlet-removal][OVN] Require wsgi start-time in the config","message":"[eventlet-removal][OVN] Require wsgi start-time in the config\n\nCurrently there is one single method to run the Neutron API that is\nusing uWSGI. This method requires a specific hash ring manager\ninitialization, different to the eventlet server based.\n\nBecause the second has been removed and is no longer available, as\nlong as Neutron continues the effors of removing eventlet from the\nsource code, the second way to initialize the hash ring manager\nis removed.\n\nRelated-Bug: #2083570\nChange-Id: I59b25093df9cf5aa77767492a5b9008bfa11cc07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1178571d37f088eb2a771f7e7846f12f9daf5a9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1178571d37f088eb2a771f7e7846f12f9daf5a9b"}]},"branch":"refs/heads/master"},"0761b72f23c37d8aaff6c45929cad6036d481189":{"kind":"REWORK","_number":5,"created":"2025-01-27 14:10:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/940123/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/940123/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/940123/5"}}},"commit":{"parents":[{"commit":"3ff599700ca5591c244efd87c4cd2ef6dcf10c48","subject":"[eventlet-removal] Remove the usage of eventlet in the Neutron API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ff599700ca5591c244efd87c4cd2ef6dcf10c48"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-24 11:52:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-27 14:03:14.000000000","tz":0},"subject":"[eventlet-removal][OVN] Require wsgi start-time in the config","message":"[eventlet-removal][OVN] Require wsgi start-time in the config\n\nCurrently there is one single method to run the Neutron API that is\nusing uWSGI. This method requires a specific hash ring manager\ninitialization, different to the eventlet server based.\n\nBecause the second has been removed and is no longer available, as\nlong as Neutron continues the effors of removing eventlet from the\nsource code, the second way to initialize the hash ring manager\nis removed.\n\nRelated-Bug: #2083570\nChange-Id: I59b25093df9cf5aa77767492a5b9008bfa11cc07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0761b72f23c37d8aaff6c45929cad6036d481189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0761b72f23c37d8aaff6c45929cad6036d481189"}]},"branch":"refs/heads/master"},"b0c02c28929dedb4ed8a460994e71e72282058c5":{"kind":"REWORK","_number":6,"created":"2025-01-27 16:18:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/940123/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/940123/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/940123/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/940123/6"}}},"commit":{"parents":[{"commit":"3ff599700ca5591c244efd87c4cd2ef6dcf10c48","subject":"[eventlet-removal] Remove the usage of eventlet in the Neutron API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ff599700ca5591c244efd87c4cd2ef6dcf10c48"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-01-24 11:52:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-01-27 16:18:15.000000000","tz":0},"subject":"[eventlet-removal][OVN] Require wsgi start-time in the config","message":"[eventlet-removal][OVN] Require wsgi start-time in the config\n\nCurrently there is one single method to run the Neutron API that is\nusing uWSGI. This method requires a specific hash ring manager\ninitialization, different to the eventlet server based.\n\nBecause the second has been removed and is no longer available, as\nlong as Neutron continues the effors of removing eventlet from the\nsource code, the second way to initialize the hash ring manager\nis removed.\n\nRelated-Bug: #2083570\nChange-Id: I59b25093df9cf5aa77767492a5b9008bfa11cc07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0c02c28929dedb4ed8a460994e71e72282058c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0c02c28929dedb4ed8a460994e71e72282058c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
