)]}'
{"id":"openstack%2Fneutron~940366","triplet_id":"openstack%2Fneutron~master~Ia077097c11a1ccc9e03d5e4ceed068790c5e8758","project":"openstack/neutron","branch":"master","topic":"bug/2096912","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-02-04 08:24:39.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-02-05 14:46:40.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-02-04 09:04:43.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"Ia077097c11a1ccc9e03d5e4ceed068790c5e8758","subject":"[Grenade] Explicitly enable correct neutron services in devstack config","status":"MERGED","created":"2025-01-29 11:22:12.000000000","updated":"2025-02-05 14:48:53.000000000","submitted":"2025-02-05 14:46:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"940366-bug/2096912","meta_rev_id":"c7547603d1409936bcd72bd02928c59bccee202b","_number":940366,"virtual_id_number":940366,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-05 14:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-02-04 08:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-02-04 09:04:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-02-04 09:04:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-02-03 15:03:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-29 14:00:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-01-29 14:19:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-30 16:10:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-01-30 19:37:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-02-03 08:37:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-02-04 13:42:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"c53437bb6328568865660995f350e8b676980253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-29 11:22:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff87d4db3462e97c51326a2e85da7ff0e1508b3b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-29 14:00:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f2d16a49dfdc921941970dafca1920cfe901b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 14:19:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cb2ad66ff304cb094215bb2e1279b25\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b6dece13a1204c32ae52daf46d8d6490 : SUCCESS in 1h 48m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/84b4dd7c8b7f4a4c8bfbed9810ec1e6c : SUCCESS in 2h 26m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/64abc23f3330418582176af2439e9952 : SUCCESS in 2h 54m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5de8560d2c644bbaa9cdff7be268f03b : SUCCESS in 49m 19s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b597c93c14e445a5a1fc132368bec189 : SUCCESS in 2h 21m 04s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a60b19fa031f433ea5473ecd24e719da : SUCCESS in 1h 58m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d11cc2f9295447ca42af551b7712355 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925e98a4782444ae8d207da94ff136ca : SUCCESS in 11m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/56f1abf593f6439ebf717d2654c40b62 : FAILURE in 28m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c0739968ae4d457fb3e582fe28dd26cf : FAILURE in 30m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b6410917c714f2883993a1f46eb607c : FAILURE in 35m 08s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e7bb624cd2844f6b9c245de68ced0cc : FAILURE in 32m 41s","accounts_in_message":[],"_revision_number":1},{"id":"11027d1701ad859b2c57176a08430e923ba728d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-30 11:01:50.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f68ad65db16df1e543a3759063f51b78891b8c22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-30 11:02:22.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2d5e754c165236b61da449b75829a567ec4cd10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-30 11:47:13.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9f83086d496122acf897a67fdea9e3b71bf4c21b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-30 13:25:34.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"437cf54a1575d4407e77cd6a316471f09c4bd4c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-30 14:53:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0c95dac41b2d5d804315d52e7f54e4b722b590b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-30 16:10:58.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1b7077c67364ce59e382b58fb6f41e51d1be7f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 16:52:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cc4a332a34d43e59ae5110a20caeea7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/842eddc278704deb95da8d61ce559d39 : SUCCESS in 1h 49m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eeacfb16cb71406585176359ae556624 : SUCCESS in 1h 53m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/75ea139fa3874a288632f04f21c9a537 : SUCCESS in 1h 44m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a33dd3ac9ce4d73bdfeb52f4ba6c6ae : SUCCESS in 34m 37s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1842c8b91fb465eb305e9f48cc6b6e3 : SUCCESS in 1h 36m 58s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3961bc62263546c381d36be168954f3e : SUCCESS in 1h 37m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/160e37aa3af944969022f00b5beae2b4 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e37ffbac4ab45548c5144fe0d4efcc0 : SUCCESS in 9m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/df25abe3f98b4be9bffb0f33f31a7e8a : SUCCESS in 50m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3878301e727e42de9ec85e255185d899 : SUCCESS in 1h 06m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2371c1873a7f4c69a9c6f525a78f239b : SUCCESS in 55m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/adfa364d1e4840aab18271215b3908e9 : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":6},{"id":"c74e78b08eb60910e6d60c6c5db797436a172be5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-01 11:38:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"0cdf0d80b7b1f80128e51769a9604349d1ca3852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 13:39:20.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/940438 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"5d18d1c2e9efd61d72816be8fe00359cd8d55efc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-01 13:39:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"536a0739f8d2a7190f4643f00e47ec903dea2e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 16:44:24.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c31c39d37e6472093a97d3435020f8c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/71870cf97d3044719841a4c2f06f1cdb : SUCCESS in 2h 01m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fa967c335907471da50d2d7885ae3677 : SUCCESS in 2h 29m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4f249fd1a6784e9b9512b269c27305eb : SUCCESS in 2h 38m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5812d9119f18496196d4b2369f167fd9 : SUCCESS in 27m 00s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d00534423e914986a361d706c9f8b082 : SUCCESS in 1h 41m 15s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e83830daa3d45399d929b88ce2a8a18 : TIMED_OUT in 3h 03m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4baf25b87f5b4804993f3f0178b5e073 : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95f9a530e1fc42ee9ce05a4c8df4d456 : SUCCESS in 9m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5c98da22954542669cdbb055f5a24703 : SUCCESS in 1h 00m 32s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/569c82509d174b4aa58ce91a1f082f65 : SUCCESS in 39m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4cbb7e0a21c4ddeb8e500b070bc3454 : FAILURE in 50m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6790e0bab35470b8e49f050d0f00a84 : FAILURE in 57m 49s","accounts_in_message":[],"_revision_number":7},{"id":"2c5616fcf31fd2deb5a6094d6459fca9057a17cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-02 09:36:36.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"5634c1e5eb1a5fb873961a64ebe2fbf979093879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 12:13:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ea6da4eb8b64ba1a7e39ecf6eb8b3da\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fc25da1366c249cd8f950d7e63ed3169 : FAILURE in 1h 14m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/efb965b8c492430286012afd06fd81ff : SUCCESS in 1h 53m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7739b16c604a490cb061079842517650 : SUCCESS in 2h 04m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5261617479f144c6b99bebca7dca5dba : SUCCESS in 33m 01s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09990fe444d1475eae807602b74526d2 : SUCCESS in 1h 23m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00483846ea974a5a86ae80797c01834e : SUCCESS in 2h 27m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c7ed6319c0405db3bb3080452ae7c9 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10bcdc9a094549a1ab29680afcdb98d0 : SUCCESS in 8m 37s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6383d0225e4f41498f285245a2607380 : SUCCESS in 38m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/96281950b4084de7a4c23ebd24a6565c : SUCCESS in 1h 04m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5aba7efde0ff47b7a091d805514bc13d : SUCCESS in 1h 01m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ceda6d10195d45c3ab6435d2bb2de786 : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":8},{"id":"1c68471bea292d202c03a143eaf343ba340fd26d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-02 12:27:22.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e128029d70ff503d24cdd7ecbc04aa9ea001f9f9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-02 12:27:29.000000000","message":"Patch Set 8: Review-Priority+2","accounts_in_message":[],"_revision_number":8},{"id":"39f18d03513d0cc72fce29aa057672bd934e6161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 14:31:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a83789f4567d41bd87dde2f3284fcd57\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/747f52146ffc4544b3c3f795019a792d : SUCCESS in 1h 34m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42def1e5bfee4a2386beec2467981fb6 : SUCCESS in 2h 02m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7de7a53148ba48b98db7ec1e1bb54633 : SUCCESS in 1h 53m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e8e13be8f47d47c29d930026ab04666d : SUCCESS in 33m 45s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f19f02b266d43a1ab03e627b82bbec3 : SUCCESS in 1h 27m 06s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6d40272e0c545f5a04767b5b9258692 : SUCCESS in 1h 52m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8d4feea5c44714b69dc76dd70ef506 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ee3c5b727a4c9f9612a2e48ef0fe24 : SUCCESS in 13m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/12a5520244f34431ad76e48301fe07f6 : SUCCESS in 48m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e8225de00975499bbbaed5000171ee23 : SUCCESS in 51m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9192acd3a7924c1485f363f322fbe759 : SUCCESS in 48m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d23f9c7bd5d4c8e8680a275962eb959 : SUCCESS in 1h 20m 25s","accounts_in_message":[],"_revision_number":8},{"id":"8c5bc279d04cbf93e15bdf0588c485707d9b3e7a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-03 08:37:02.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"266cd19541c29916280d909b1ab463371d2429e3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-03 12:16:17.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8d63665b1926c2fed1e62cdd84242608ee47cbf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-03 15:03:38.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"e744bcea3cf11254f86dcc5f8cbf7ab2d49eec7b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-03 15:03:51.000000000","message":"Patch Set 9: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"13557fa2f47a8bee8285ff38e2b746edc57378ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 17:16:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/020f62de6ed147b8a401b94f3739c97e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/765fc4af1c3148c483624be3b3b62c47 : SUCCESS in 1h 35m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4173020f163e482a90715484e127ece8 : SUCCESS in 1h 45m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/919d974132d148b3a34f6b52503a8179 : SUCCESS in 2h 11m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ca2c3386edaa4e41b5c8fc14fc2af3cb : SUCCESS in 38m 55s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/536bbb17e12142a6ba6bb8d7a6e5f083 : SUCCESS in 1h 35m 10s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a660da155404cecada124bf3251a266 : SUCCESS in 2h 05m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e5e499a0e6a4d33b8f2f038e32a98e1 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4809a56d6d14967851c30083ddf60e5 : SUCCESS in 9m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/34b9cb8ca95a40ac9505e59f7b39f98a : SUCCESS in 1h 14m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aa67e552ab484c4e8a62eb9011d61eef : SUCCESS in 53m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/28aa2ded1da345cca2a5411e3af151f9 : SUCCESS in 38m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2454fded3a048c781605f236d8fbe4b : SUCCESS in 1h 39m 38s","accounts_in_message":[],"_revision_number":9},{"id":"ef15ec2bdbd916c32ddd3a51e6f26f9ebcc919c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-04 08:24:39.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"2c1b596c6e7676fc4917e9200cb628e22f269265","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-04 09:04:43.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"de0477682eeb924870a9bb8ceb3af922c2802e5f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-02-04 13:42:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0bb459e420b4d6bd1eccc966ab1cd53824ef2dc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 12:41:56.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e637d3e4c5dbf5a30da5cee4104816d2fe959063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 14:46:40.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed550c75789140e18a1810fc2f2dde27\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f56f3f23cd574fce9c401b256ddfce7a : SUCCESS in 1h 54m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b32db406ffe44617ac782dcaf8830c63 : SUCCESS in 1h 56m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4ee5b9af96884cd1a468cb7655689f48 : SUCCESS in 1h 59m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050a8b2fdccd4465b9a01357cdf598da : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad6a4d4610f47ad90def08e8bf3c645 : SUCCESS in 9m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3ac31ea946c425e8fd54ff3745dc32e : SUCCESS in 1h 20m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6369f00eb26a492592c5948f5060ce51 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":9},{"id":"ed6b07d8f0eb026e7dd197735e58116680054ffc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 14:46:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"c7547603d1409936bcd72bd02928c59bccee202b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 14:48:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b44047d1dd944d4844a2fcc756f63f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3482e2f6be484ff2843c33bbee83f564 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"0632996b1783881711b0e3c2e31044204d549f9c","revisions":{"5b389b9dcbb43e01d6421286bdc19ad8f5db304c":{"kind":"REWORK","_number":1,"created":"2025-01-29 11:22:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/940366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/940366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/940366/1"}}},"commit":{"parents":[{"commit":"8c479abcbb4e909764ad05f020ea0bb2f7409b52","subject":"Merge \"Remove linuxbridge driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c479abcbb4e909764ad05f020ea0bb2f7409b52"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"subject":"[Grenade] Explicitly enable correct neutron services in devstack config","message":"[Grenade] Explicitly enable correct neutron services in devstack config\n\nIt seems that when neutron is enabed in the Devstack\u0027s services as\n\"enable_service q-svc\" it may be started properly by Devstack but when\nit comes to stopping it, it tries to stop services expected to be\nenabled when it is deployed under mod_wsgi and that fails.\n\nSo this patch enables correct services in the devstack\u0027s configuration\nso that those services can be started properly.\n\nCloses-bug: #2096912\nChange-Id: Ia077097c11a1ccc9e03d5e4ceed068790c5e8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b389b9dcbb43e01d6421286bdc19ad8f5db304c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b389b9dcbb43e01d6421286bdc19ad8f5db304c"}]},"branch":"refs/heads/master"},"9d5545102364fa2a885894309c5dd3b0caff1aec":{"kind":"REWORK","_number":2,"created":"2025-01-30 11:01:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/940366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/940366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/940366/2"}}},"commit":{"parents":[{"commit":"8c479abcbb4e909764ad05f020ea0bb2f7409b52","subject":"Merge \"Remove linuxbridge driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c479abcbb4e909764ad05f020ea0bb2f7409b52"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-30 11:01:50.000000000","tz":0},"subject":"[Grenade] Explicitly enable correct neutron services in devstack config","message":"[Grenade] Explicitly enable correct neutron services in devstack config\n\nIt seems that when neutron is enabed in the Devstack\u0027s services as\n\"enable_service q-svc\" it may be started properly by Devstack but when\nit comes to stopping it, it tries to stop services expected to be\nenabled when it is deployed under mod_wsgi and that fails.\n\nSo this patch enables correct services in the devstack\u0027s configuration\nso that those services can be started properly.\n\nCloses-bug: #2096912\nChange-Id: Ia077097c11a1ccc9e03d5e4ceed068790c5e8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d5545102364fa2a885894309c5dd3b0caff1aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d5545102364fa2a885894309c5dd3b0caff1aec"}]},"branch":"refs/heads/master"},"3082c6596b67dc8a7c8831dceefa0506e5d8168d":{"kind":"REWORK","_number":3,"created":"2025-01-30 11:02:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/940366/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/940366/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/940366/3"}}},"commit":{"parents":[{"commit":"8c479abcbb4e909764ad05f020ea0bb2f7409b52","subject":"Merge \"Remove linuxbridge driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c479abcbb4e909764ad05f020ea0bb2f7409b52"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-30 11:02:21.000000000","tz":0},"subject":"[Grenade] Explicitly enable correct neutron services in devstack config","message":"[Grenade] Explicitly enable correct neutron services in devstack config\n\nIt seems that when neutron is enabed in the Devstack\u0027s services as\n\"enable_service q-svc\" it may be started properly by Devstack but when\nit comes to stopping it, it tries to stop services expected to be\nenabled when it is deployed under mod_wsgi and that fails.\n\nSo this patch enables correct services in the devstack\u0027s configuration\nso that those services can be started properly.\n\nCloses-bug: #2096912\nChange-Id: Ia077097c11a1ccc9e03d5e4ceed068790c5e8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3082c6596b67dc8a7c8831dceefa0506e5d8168d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3082c6596b67dc8a7c8831dceefa0506e5d8168d"}]},"branch":"refs/heads/master"},"346cfb8ea155d0f5ef482e9fb4aa685d676bdeaa":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-01-30 11:47:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/940366/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/940366/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/940366/4"}}},"commit":{"parents":[{"commit":"8c479abcbb4e909764ad05f020ea0bb2f7409b52","subject":"Merge \"Remove linuxbridge driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c479abcbb4e909764ad05f020ea0bb2f7409b52"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-30 11:47:12.000000000","tz":0},"subject":"[Grenade] Explicitly enable correct neutron services in devstack config","message":"[Grenade] Explicitly enable correct neutron services in devstack config\n\nIt seems that when neutron is enabed in the Devstack\u0027s services as\n\"enable_service q-svc\" it may be started properly by Devstack but when\nit comes to stopping it, it tries to stop services expected to be\nenabled when it is deployed under mod_wsgi and that fails.\n\nSo this patch enables correct services in the devstack\u0027s configuration\nso that those services can be started properly.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940438\n\nCloses-bug: #2096912\nChange-Id: Ia077097c11a1ccc9e03d5e4ceed068790c5e8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/346cfb8ea155d0f5ef482e9fb4aa685d676bdeaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/346cfb8ea155d0f5ef482e9fb4aa685d676bdeaa"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1cc70841450ea6018077037fee59b3336f299883":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-01-30 13:25:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/940366/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/940366/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/940366/5"}}},"commit":{"parents":[{"commit":"8c479abcbb4e909764ad05f020ea0bb2f7409b52","subject":"Merge \"Remove linuxbridge driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c479abcbb4e909764ad05f020ea0bb2f7409b52"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-30 13:25:34.000000000","tz":0},"subject":"[Grenade] Explicitly enable correct neutron services in devstack config","message":"[Grenade] Explicitly enable correct neutron services in devstack config\n\nIt seems that when neutron is enabed in the Devstack\u0027s services as\n\"enable_service q-svc\" it may be started properly by Devstack but when\nit comes to stopping it, it tries to stop services expected to be\nenabled when it is deployed under mod_wsgi and that fails.\n\nSo this patch enables correct services in the devstack\u0027s configuration\nso that those services can be started properly.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940438\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940447\n\nCloses-bug: #2096912\nChange-Id: Ia077097c11a1ccc9e03d5e4ceed068790c5e8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1cc70841450ea6018077037fee59b3336f299883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1cc70841450ea6018077037fee59b3336f299883"}]},"branch":"refs/heads/master","description":"Edit commit message"},"22cdcb4248238a52f1ad8e63ebddfef9928847d3":{"kind":"REWORK","_number":6,"created":"2025-01-30 14:53:24.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/940366/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/940366/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/940366/6"}}},"commit":{"parents":[{"commit":"8c479abcbb4e909764ad05f020ea0bb2f7409b52","subject":"Merge \"Remove linuxbridge driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c479abcbb4e909764ad05f020ea0bb2f7409b52"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-30 14:53:04.000000000","tz":60},"subject":"[Grenade] Explicitly enable correct neutron services in devstack config","message":"[Grenade] Explicitly enable correct neutron services in devstack config\n\nIt seems that when neutron is enabed in the Devstack\u0027s services as\n\"enable_service q-svc\" it may be started properly by Devstack but when\nit comes to stopping it, it tries to stop services expected to be\nenabled when it is deployed under mod_wsgi and that fails.\n\nSo this patch enables correct services in the devstack\u0027s configuration\nso that those services can be started properly.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940438\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940447\n\nCloses-bug: #2096912\nChange-Id: Ia077097c11a1ccc9e03d5e4ceed068790c5e8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22cdcb4248238a52f1ad8e63ebddfef9928847d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22cdcb4248238a52f1ad8e63ebddfef9928847d3"}]},"branch":"refs/heads/master"},"a88388697c4ab0f09b49f70c0d6f911c76647649":{"kind":"REWORK","_number":7,"created":"2025-02-01 11:38:51.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/940366/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/940366/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/940366/7"}}},"commit":{"parents":[{"commit":"62f39610edfa43c79015097e7c1c14d6d7cc5a5a","subject":"Merge \"[eventlet-removal] Use non-eventlet metadata proxy in OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62f39610edfa43c79015097e7c1c14d6d7cc5a5a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-02-01 11:38:28.000000000","tz":60},"subject":"[Grenade] Explicitly enable correct neutron services in devstack config","message":"[Grenade] Explicitly enable correct neutron services in devstack config\n\nIt seems that when neutron is enabed in the Devstack\u0027s services as\n\"enable_service q-svc\" it may be started properly by Devstack but when\nit comes to stopping it, it tries to stop services expected to be\nenabled when it is deployed under mod_wsgi and that fails.\n\nSo this patch enables correct services in the devstack\u0027s configuration\nso that those services can be started properly.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940438\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940447\n\nCloses-bug: #2096912\nChange-Id: Ia077097c11a1ccc9e03d5e4ceed068790c5e8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a88388697c4ab0f09b49f70c0d6f911c76647649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a88388697c4ab0f09b49f70c0d6f911c76647649"}]},"branch":"refs/heads/master"},"8905660664a0814c6140d7281dc066a0395fa5d9":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-02-02 09:36:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/940366/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/940366/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/940366/8"}}},"commit":{"parents":[{"commit":"62f39610edfa43c79015097e7c1c14d6d7cc5a5a","subject":"Merge \"[eventlet-removal] Use non-eventlet metadata proxy in OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62f39610edfa43c79015097e7c1c14d6d7cc5a5a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-02-02 09:36:35.000000000","tz":0},"subject":"[Grenade] Explicitly enable correct neutron services in devstack config","message":"[Grenade] Explicitly enable correct neutron services in devstack config\n\nIt seems that when neutron is enabed in the Devstack\u0027s services as\n\"enable_service q-svc\" it may be started properly by Devstack but when\nit comes to stopping it, it tries to stop services expected to be\nenabled when it is deployed under mod_wsgi and that fails.\n\nSo this patch enables correct services in the devstack\u0027s configuration\nso that those services can be started properly.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940438\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940447\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/940579\n\nCloses-bug: #2096912\nChange-Id: Ia077097c11a1ccc9e03d5e4ceed068790c5e8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8905660664a0814c6140d7281dc066a0395fa5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8905660664a0814c6140d7281dc066a0395fa5d9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0632996b1783881711b0e3c2e31044204d549f9c":{"kind":"REWORK","_number":9,"created":"2025-02-03 15:03:38.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/940366/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/940366/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/940366/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/940366/9"}}},"commit":{"parents":[{"commit":"62f39610edfa43c79015097e7c1c14d6d7cc5a5a","subject":"Merge \"[eventlet-removal] Use non-eventlet metadata proxy in OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62f39610edfa43c79015097e7c1c14d6d7cc5a5a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-29 11:18:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-02-03 15:03:38.000000000","tz":0},"subject":"[Grenade] Explicitly enable correct neutron services in devstack config","message":"[Grenade] Explicitly enable correct neutron services in devstack config\n\nIt seems that when neutron is enabed in the Devstack\u0027s services as\n\"enable_service q-svc\" it may be started properly by Devstack but when\nit comes to stopping it, it tries to stop services expected to be\nenabled when it is deployed under mod_wsgi and that fails.\n\nSo this patch enables correct services in the devstack\u0027s configuration\nso that those services can be started properly.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940438\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/940447\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/940579\n\nCloses-bug: #2096912\nChange-Id: Ia077097c11a1ccc9e03d5e4ceed068790c5e8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0632996b1783881711b0e3c2e31044204d549f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0632996b1783881711b0e3c2e31044204d549f9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
