)]}'
{"id":"openstack%2Fneutron~940949","triplet_id":"openstack%2Fneutron~stable%2F2024.2~I0b7e7e8c506a9632aac911c1b4a0d19653dd5239","project":"openstack/neutron","branch":"stable/2024.2","topic":"bug/2095576-stable/2024.2","attention_set":{},"removed_from_attention_set":{"37685":{"account":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"last_update":"2025-02-07 18:31:27.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-02-07 18:31:27.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-02-07 16:17:57.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I0b7e7e8c506a9632aac911c1b4a0d19653dd5239","subject":"OVS: Set log level to INFO for unconfigured ofport cases","status":"MERGED","created":"2025-02-07 10:03:30.000000000","updated":"2025-02-07 18:33:10.000000000","submitted":"2025-02-07 18:31:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940949-bug/2095576-stable/2024.2","cherry_pick_of_change":940474,"cherry_pick_of_patch_set":1,"meta_rev_id":"e9880b29a5b5788796f2c6e5e924a754863b2167","_number":940949,"virtual_id_number":940949,"owner":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-07 18:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-02-07 10:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-02-07 16:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-02-07 16:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-07 10:05:25.000000000","updated_by":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-02-07 10:05:25.000000000","updated_by":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-02-07 10:26:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-02-07 10:42:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-07 12:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52c81ff6be66a9befab3387321d6495065626dba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"date":"2025-02-07 10:03:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e8eb3123670dc1cb996fde749d10cb0f77c79dd0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-07 10:26:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9a1151440f7814f0be26b20d7ea70103f0e7865","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 10:42:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85cf497cfc94a7a803666d030f5bdd1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a7d9f0ece694f41864618fddccfb728 : SUCCESS in 36m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a36d1bea8a14cd2bc22e9af8afa42f4 : SUCCESS in 24m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61183d963c23b6dd8e163b2e0e55ddca33802e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 12:09:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042eecc86a1347fa87646d40b455a2c6\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/9efb88f4f902463f8d44d485ba0a47f5 : SUCCESS in 1h 32m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/1f6cfbf7554f4eed861a3e892bea8002 : SUCCESS in 1h 29m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5739694b1ec44f0e9a2b567afa7f3674 : SUCCESS in 33m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/720511288afc436cbef3e3c18ff47143 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c788bfd5ffdf4158b1e6a025a47a3525 : SUCCESS in 20m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28471f8a61be46998c72f625c877ad35 : SUCCESS in 18m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ee2af1fa68149aca2e46d6cf8c0f321 : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365ff04bcbca42ab946b51353152221c : SUCCESS in 7m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54f352bdcc044ccd82cebbe4094e3fde : SUCCESS in 56m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6f0998269ab49dfb997c720045305df : SUCCESS in 2h 05m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d0e1acbb5e94af2a2ff66e6949bd17c : SUCCESS in 1h 05m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/65afc70199f144fabbfdebae34e92d1f : SUCCESS in 50m 05s","accounts_in_message":[],"_revision_number":1},{"id":"69a0715a5d3d76b929c202efed6b49dd2de8c4ad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-07 16:17:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60cafdea3edb24064b0af184d6bbda0242c53d30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 16:18:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85f22fd7b74e936cfb935bfe18474ed836923db3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:31:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a219223e4a41473ba3ff1985470c1e17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/049040675da741b6acf86e8561f73b9d : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d5e2c4b10a245e788429a7f6f997b7c : SUCCESS in 23m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e488a7b3f80464cba0a717a0abbe298 : SUCCESS in 23m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93a8c410a74492cab03e62f918efa27 : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3833c5a84f0f496f86ab9cbe9c5b766e : SUCCESS in 51m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f6268c0c64b4c75a79e7b56fac50c29 : SUCCESS in 1h 37m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4233de6c1fa049e08e28cf16cae48891 : SUCCESS in 1h 23m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f11d22f85e34d94bacd2b26a06b7f45e948d0c68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:31:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e9880b29a5b5788796f2c6e5e924a754863b2167","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:33:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7133c89c6fff40a1964d87132f6c302b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b36f87d97ac468db8b283cc4ec0f06d : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b396faf5088f805b062aef5e20f2354dad8cd205","revisions":{"b396faf5088f805b062aef5e20f2354dad8cd205":{"kind":"REWORK","_number":1,"created":"2025-02-07 10:03:30.000000000","uploader":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"ref":"refs/changes/49/940949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/940949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/940949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/940949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/940949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/940949/1"}}},"commit":{"parents":[{"commit":"318f2df7f51b85aafcb4e78fb66ebb12bde4f95a","subject":"[Functional tests] Add logging router interfaces in metadata IPv6 tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/318f2df7f51b85aafcb4e78fb66ebb12bde4f95a"}]}],"author":{"name":"Konstantin Eremin","email":"k0dt@k0dt.ru","date":"2025-01-30 19:27:11.000000000","tz":180},"committer":{"name":"Konstantin Eremin","email":"k0dt@k0dt.ru","date":"2025-02-07 10:03:30.000000000","tz":0},"subject":"OVS: Set log level to INFO for unconfigured ofport cases","message":"OVS: Set log level to INFO for unconfigured ofport cases\n\nLog an ERROR if the VIF port has no ofport, which indicates\nthat the port might not be able to transmit traffic. This can\nhappen if the ovs-vswitchd service is down, but the ovsdb-server\nservice is still running. In such cases, the port record might\nbe created in the database, but it won\u0027t be processed by\novs-vswitchd, resulting in an empty list ofport value.\n\nI was unable to reproduce the ofport\u003d[] scenario using the\nOpenStack CLI. However, I managed to achieve this by using the\n«ovs-vsctl add-port» command while the ovs-vswitchd service was\nstopped. Reusing this port in Neutron is impossible. I believe\nthis situation is highly unlikely to occur in practice, but I\ndecided to leave the check as it is, just in case.\n\nWhen the ofport is set to INVALID_OFPORT, it indicates that\nthe port is in a transitional state and has not yet been fully\nconfigured. This is not an error condition but rather a normal\npart of the port\u0027s lifecycle, as it may take some time for the\nport to be initialized and assigned a valid ofport. Operators\nshould be aware of this state for monitoring purposes, but no\nimmediate action is required.\n\nCloses-Bug: #2095576\nChange-Id: I0b7e7e8c506a9632aac911c1b4a0d19653dd5239\n(cherry picked from commit a81aa17471a44545f71a044fb0008dc576be2f03)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b396faf5088f805b062aef5e20f2354dad8cd205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b396faf5088f805b062aef5e20f2354dad8cd205"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
