)]}'
{"id":"openstack%2Fneutron~940982","triplet_id":"openstack%2Fneutron~master~Ide6720bde733d75352feb6ee160d23318c7406e3","project":"openstack/neutron","branch":"master","topic":"bug/2087939","attention_set":{},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2025-06-19 11:33:12.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-06-19 11:33:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide6720bde733d75352feb6ee160d23318c7406e3","subject":"ofswitch: remove usage of eventlet timeout","status":"MERGED","created":"2025-02-07 16:03:51.000000000","updated":"2025-06-19 11:34:32.000000000","submitted":"2025-06-19 11:33:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"940982-bug/2087939","meta_rev_id":"9e6768236acf24534591977358bfec9894af9de4","_number":940982,"virtual_id_number":940982,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-06-19 07:45:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-19 11:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-06-19 06:56:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-19 06:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-06-19 06:56:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-07 16:52:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-02-07 17:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-07 17:58:39.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-07 18:51:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 10:52:47.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-28 13:41:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-05-05 08:17:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-05-12 13:18:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"49ffab275cef590439c2d675ac241dadf22cefbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-07 16:03:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e35d90d0aefbf4f9ce2b581b9bd366a5f3314cf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 17:19:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa683aba5cab49ff86993f9a17019623\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4cd48c5934d64eaa847cf5dadc350270 : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5654138dc21c4bb990b6f9e97772f716 : SUCCESS in 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5525bf040f6157469f08400e2fd9e642fffcca43","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-07 17:58:39.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/82/940982/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"2ab70c605a92699ee328fc282b59caa52ba038fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:51:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c824cd0ee5549e3887d131a8b8b760b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4ba661df30eb4d5585e8bc744fc07af6 : SUCCESS in 1h 27m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4f3ca32aed464ca8ab7ad096695a5fcf : SUCCESS in 1h 40m 40s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4a9426b20c845eaae6e711feb69c995 : SUCCESS in 1h 22m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/892347077dfe45d0b4f6243badb98073 : SUCCESS in 51m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a95615bda6aa4f26921c160a99122852 : SUCCESS in 16m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc485edec58547018225804b38762a23 : SUCCESS in 22m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3265f18bccc474fb818c0b54b3f0a8b : SUCCESS in 23m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9575c7c6c5e240768d0e0638b8cd047d : SUCCESS in 11m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/075d36fa8f394b5fa90f66809f019c79 : SUCCESS in 52m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/33866a4070ea457bafec14c0d3f71aaa : FAILURE in 55m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f27beb9c141348f3bf758bc111635aa8 : SUCCESS in 2h 35m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a31cef0d51d1470194788c64dc670430 : SUCCESS in 1h 01m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/22590e4cd2744a7ba55af069886af89a : SUCCESS in 2h 07m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d9a82c5359a14164bee30036cb62827f : FAILURE in 1h 33m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f84a938950fe4c399fcc1a08106bc6ef : SUCCESS in 1h 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b518b5157b9bb026010d459bf42cd511d13ece12","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-09 14:50:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b048cc40b3b9ce313f4fc0af867d15bc34f8726","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 15:29:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/880b0a26d0b1403fab000a5691ad8bd6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2cb0c70d55fa401a935644d8ab99b6f3 : SUCCESS in 37m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/161f1a14d729465e9a321714858d90cb : SUCCESS in 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b8d508e0f5ea189355b23ab0867b4b65d2f03d5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-09 16:32:20.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/82/940982/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"346f8bc8c88f77fdd49fc8afd07281e33e6cb70c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 17:25:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39b2e88040e54539b7881ad2411402e8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0eee20bbc8724ca09b1e3c97867ac244 : SUCCESS in 1h 12m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6a95c10d4aea49c3be30674a5e3f78e6 : SUCCESS in 2h 26m 58s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b4315bc49a44f06bb0667caf3518b77 : SUCCESS in 1h 45m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/251ec2fde0bd4e71ad9e6e2ecc566ae7 : SUCCESS in 44m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6816722647d64047953866a7d1a13dec : SUCCESS in 11m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c805c7c1d646443890c6d4631b5bac93 : SUCCESS in 18m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0212734e58224c84aa67ea0aa16e0507 : SUCCESS in 17m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df042d646f414efc9aa9f2453a6ccf70 : SUCCESS in 10m 02s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0c9f56469968419292839077d13c4bcb : SUCCESS in 59m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5ef037e7167e47c0995d942e1072c559 : FAILURE in 25m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/89954f1bab224573b31d2bc80986a795 : SUCCESS in 2h 01m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/746ad7b576374701b01762c96d16586f : SUCCESS in 59m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e5bfcf20d916421abad7fd682da5200c : SUCCESS in 1h 29m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3dcbbbaa27db4f82823fd4c3268983bb : FAILURE in 1h 14m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/84d3bbd20d09499c965a8f51230ffe7b : SUCCESS in 1h 05m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7e67cf592175b26acb4242b350eaf77095a3fb08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-10 09:17:56.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3ced48649c76a4f5c5ccb850e6030273aed8176b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-10 09:39:29.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/82/940982/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"bb4a865657809527e700c0f4145b42ea2fb0aaa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 10:00:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc100caa9cd54134874a9c18a386edb6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/553d316906224087804b114fe92292a0 : SUCCESS in 40m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/35db8c7e43da4f588ef20fdda77eaec9 : SUCCESS in 33m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4e8defc76a63b548c1220ff314dc8e6b61730fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 11:48:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b53bdd875ddf4f8baad69ec15888ad08\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e4532d8ecc6f4218ad36667f73111bf8 : SUCCESS in 1h 53m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b573778014264ed1bf979e46ff9d586e : SUCCESS in 1h 56m 14s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9351578092b4986a08a3b5646142dde : SUCCESS in 1h 44m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/531f3204d4da43558e3bb4668043fd06 : SUCCESS in 50m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d29aa61e2849008caca2036c04c240 : SUCCESS in 11m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36900026a9ab4aadb44addf47af6a6e2 : SUCCESS in 24m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/015aa750aabf4b2aba1ddae7084d75dc : SUCCESS in 16m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129d87f12aaa456ba994439680369a10 : SUCCESS in 10m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/76f48a2e9e4f4121aab72ed5d09568f6 : SUCCESS in 54m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/66c3684129ab44f28d32c9f03a6b5c9c : FAILURE in 31m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/43c143b2a17e45239c160d0b78260b52 : SUCCESS in 2h 27m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/889b1df84a6c41aa835441a00b56de5e : SUCCESS in 1h 12m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/588c3eba312742679f2b67087f6b23b1 : SUCCESS in 1h 44m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/20d805e302eb4314877e208415af079d : SUCCESS in 1h 29m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/74abe91d324a4a0eb02bde2d275cacf4 : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":2},{"id":"bd0de7d8f1090b6652af870c2908bbc87ed98bfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-10 13:00:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7a612034e9a23b917b09600ffb2e3027db501c9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 13:38:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e87f9c9a228543d586292906048cf9c3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79d7c79ab1544ce08cbd1794ac288ffa : SUCCESS in 35m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/547c23e081ae4f04b5990ac4ef4b9843 : SUCCESS in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b16af51b4327a9223f6cdb9afa9dfdf8ee19b7b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-10 13:50:54.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/82/940982/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"8c023ae8d30a6417e272f6fe433a84fa540fa82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 15:09:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b80060acc6f49f882a542c95643c701\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a145febf082749fd97710c4f284ef872 : SUCCESS in 1h 57m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6621f0e4f3764870af9122e188ab2c3e : SUCCESS in 2h 07m 29s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5bd6c59a68f452a85bd8ac2c7a31bac : SUCCESS in 2h 05m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63b9d3f0da2f4cdc9a6dd183d3a3c7cd : SUCCESS in 50m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/338bd57af63d4bb09e6a62e9cfe6cab1 : SUCCESS in 11m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb022c71e0064efda3644f69eb25de43 : SUCCESS in 21m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e8f617625ac430caf99c5f167ef2367 : SUCCESS in 19m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6bca5716fb944d890b5050b283a92fb : SUCCESS in 10m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/41eb31ea2b9b4b299abaa28e9158d37b : SUCCESS in 54m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a4a2f9ee84a841ec9ba561e79c6046e4 : SUCCESS in 1h 00m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/58bb4a177c744761b16ac2aa7614ea2d : SUCCESS in 1h 55m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/21c5f6fad8814623aae69bcae5fc229c : SUCCESS in 56m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1268eef46f4d4d798f981d6b3d266962 : SUCCESS in 1h 43m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/55d3990c5e2345328024d962f2c60b42 : SUCCESS in 1h 23m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1a2ac7e1389248f9bc39d6ed1351521d : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":3},{"id":"2e23f5f34eb5d13bbc3511bea7bc00a6b6b95e45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-20 08:03:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"759319f51273c46a4463effece971ce8e15e4203","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 09:56:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2c7d086d76f46d3917051bc3dfc8408\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d9718baa76d4af7a7ff2c87a508b8c6 : SUCCESS in 54m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8272a92e447449d0a284a7673d30811b : SUCCESS in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"08e76155614c50000c4ac734848f1a4cc0fdf4e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:23:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89293d0679144aebb9cc3bca3dbfeb68\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6518f6ff6037462aa43c98703f8cba1a : SUCCESS in 1h 25m 55s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0838584efb584ef8800b912a20e93561 : SUCCESS in 1h 25m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/128ae59068e14907bc4e143dcadef5fd : SUCCESS in 34m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae016aaab574934951b1eeb42a5ff00 : SUCCESS in 14m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30ad922ba2f84b8bbba389d0b9c066c7 : SUCCESS in 31m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69f660f0095742519358d39c7bf6658d : SUCCESS in 18m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/359c0fec27814d0ba675218db1205251 : SUCCESS in 7m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4f8a98c105ee45338f49a434e188164d : SUCCESS in 35m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/da789989f11d41918c450c55680060ff : SUCCESS in 1h 13m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/14ae994371634640a04f53a39917c69f : SUCCESS in 2h 03m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/001dc6a894ab4002b2956b787c25f259 : SUCCESS in 1h 05m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad75fb0529de4047af5e37f6ab88408f : SUCCESS in 1h 44m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ec898baf39e349babeb00f052d17ed0c : SUCCESS in 1h 07m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c69559837a8e4cd89b7e56e7b2d815f8 : SUCCESS in 1h 29m 53s","accounts_in_message":[],"_revision_number":4},{"id":"9528bb3a7c915313868e2f804ce1782190cdb5e4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-20 10:52:47.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/82/940982/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"bae33b0dafc305b8ebd6cc8c3d72177339e5fa42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-03 18:15:40.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"bb2c73a61fa981c3b8367fcaa1ca5d26d0e7565e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 19:06:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9fe17bd5c7840c090a8eaa5fde0c3bd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72d928d76ae54523b73c2d16627d9e4c : SUCCESS in 48m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6aafec35a3941fba52fd07eee2b60e9 : SUCCESS in 45m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cfb1f45e8210a2fd8345ecb1a1b2fced9809806a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-03 19:50:07.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/82/940982/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"adccebd33494b7cb94b166e19436233722e846a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 20:28:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa57859da6ee46c089689120d683d205\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c60689404c99401a9d2c3b5010d341a7 : SUCCESS in 1h 38m 36s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a9f4fed97464196b010afa1ac02ab6d : SUCCESS in 1h 27m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/966d0f55b44f45388da28ca7683d4daa : SUCCESS in 51m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108a05ec791f42e4b9a4e56668972c9b : SUCCESS in 14m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85d8be0b1ef34de1a92bcc4ad3dd4cb0 : SUCCESS in 23m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c716e5e007b74b48903afd2f06d3ec1d : SUCCESS in 20m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b503edc0bdee485784f80095de952ca8 : SUCCESS in 11m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ae02c06f1a724165a9df81d3fc726083 : SUCCESS in 46m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b60e105308e74b878897845afd0ddc2a : SUCCESS in 1h 09m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/45b38b911d5c4008839f3b91e3fa867f : FAILURE in 1h 47m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/950d5bd400094ddfa060f57b2c4a2bbd : SUCCESS in 57m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6e69d2b440c24e1e99a9281db6472795 : SUCCESS in 1h 59m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4a72e63b37744aec88e6f3a2b120df13 : SUCCESS in 1h 44m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cb50f391759b4eacb3c7af96880f1a94 : SUCCESS in 1h 10m 15s","accounts_in_message":[],"_revision_number":5},{"id":"8ca9999f774176bdf3e81c89f478dc56bd9a8af4","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-04 08:36:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f12d79f1fd25367e3769cc472ed7118335cac933","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 09:05:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67d14c9799e448de851351cfee3c7396\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19265aecc3434532bb3ead32c9a26d69 : SUCCESS in 27m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4b3670eee1f9493f8350ce74d83dc7f5 : SUCCESS in 17m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"69b66843878d1e552a79ce4fbb6e4593aa9f4470","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-04 09:54:11.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/82/940982/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"963c369d472074e010a6a21f8058aabf07d4641d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 10:46:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a5e5cb414874752b65ddcfb479b256e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cf8c514f134649829d177152a6f4206a : SUCCESS in 1h 40m 30s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a78cad2d5f174b9785597d9ca5e3f8d2 : SUCCESS in 1h 44m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36bf3f24a02e49058cd9e55c28d8ee30 : SUCCESS in 47m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50768beeb5cd45eca949d805126d3c9d : SUCCESS in 11m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/134dff7d7236474d852fd3986c0778ae : SUCCESS in 22m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/efa8515ec2b34178b8babc252a9404d5 : SUCCESS in 19m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38160d5d537941b8b69768c420a7f441 : SUCCESS in 9m 19s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/31786b8018ff4c71bf9f1eed0c26257a : SUCCESS in 55m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b7001f21f6554a8daaf43aeb08741a8d : SUCCESS in 1h 18m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/556c2bb1128f4a35ae14050c74a428f8 : SUCCESS in 2h 05m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/31172bda090c449a8bb3b362cd82e8b8 : SUCCESS in 1h 04m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b90866e06c00451f825c48de6edf3090 : SUCCESS in 1h 47m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4168ec66bd134987aae7551e555873a8 : SUCCESS in 1h 36m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/867b5e98f9db41ef9f1299d218cd9f54 : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":5},{"id":"21cd9a74d1f172a8a579eb6ebaf1a715f57bccdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-06 07:14:46.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a2f751b2a98da0b0984d8f1db83e9ce4994ca70b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 07:54:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d518e3e3eb044f69058fc4465a92edd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/229da4405135432997039b6a19019e29 : SUCCESS in 37m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4a614373bd9a44c9bba996ee4f73de3a : SUCCESS in 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f53f83b48c9100e5c87f5b641cb44ad58d88db7b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-06 09:27:14.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/82/940982/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"b53bfda836e49e4fe458b1ed5fdcdc4bf88a2fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 09:52:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026a69e4b4d44f22b73971a7904f0aa1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/88bf15fc6b254e44a5eb33099ce10930 : SUCCESS in 2h 35m 29s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03a153dd29744fbcaa34d8f5396fd99c : SUCCESS in 1h 56m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04f1a1cc3c724479b553ae774d3ad395 : SUCCESS in 38m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef141086f507492a8ccb5a3b85b83c59 : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/effc3b61d6834713ac3db34e56a7fc40 : SUCCESS in 20m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38f8ff9b85ce40a18228a7019475adfa : SUCCESS in 23m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d911fad768494da2cd3e4b98fa247a : SUCCESS in 9m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d2651d3d6fb241528c2ba43d081cdc80 : SUCCESS in 53m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d9e42cec53f649209a935436e1700678 : SUCCESS in 1h 19m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b9598894cbd74341897ad4e76a57dc90 : SUCCESS in 1h 59m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7c8f4910a914d93bcc9bccc5ebe21cb : SUCCESS in 1h 03m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/37b2828b7712465bb76c6d126630ffe2 : SUCCESS in 1h 50m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/52e1fb06fc58424cabb5d9178a673c27 : SUCCESS in 1h 40m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/170a9f121010463790628c51cdb656ac : SUCCESS in 1h 12m 46s","accounts_in_message":[],"_revision_number":6},{"id":"97a98689786146a9a223783f4596e518303a7d81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-27 09:23:28.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5e95b8f07599bcec5d58a08456f72139b75e0c83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 10:02:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d84a38d9ea9d445a9cd2d4fd704b07f1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e9b1054179a49e3b256fae3e6a1b46a : SUCCESS in 27m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b7a71e9cae7a44b88a184ccad43f5f84 : SUCCESS in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"79c8a61b27ae3c8c60feee9e0617a2db848c22fc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-27 10:34:01.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/82/940982/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"e9fe5265b33d7111c29a5d28d66eb64767f2c2aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 11:32:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e95e32cc658f4afb93edea1ecd3d7beb\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/81cb786cfcbe4fb79b3b6835ac95e2c7 : SUCCESS in 1h 32m 50s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9c7c6613f304c62b435bc079993a084 : SUCCESS in 1h 24m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6d4636fc82f4164bc09e1b885fe5055 : SUCCESS in 39m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b517936ff0e14948bdc8b5ffd96b0314 : SUCCESS in 11m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ef65f292a2742dfa3aab8059205624c : SUCCESS in 21m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c32bd60da43f40359d727f9ae58860ca : SUCCESS in 18m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7034e3290f564001b6851a94458ecf82 : SUCCESS in 11m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/10473351d61c4744a4d204ba92a31904 : SUCCESS in 1h 16m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/67c87a9fc4fe4a209798302eb875f1dc : SUCCESS in 1h 54m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1ecf9dac3354e8290143b800da8d282 : SUCCESS in 48m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/01c06c49b6d742a4bce8b4d0004a1a86 : SUCCESS in 1h 33m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15d4132918b746cbb5d9fcbf43551137 : SUCCESS in 1h 45m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/80a98657c6044712846f0278c6e7bbed : SUCCESS in 34m 12s","accounts_in_message":[],"_revision_number":7},{"id":"fccb5e337327f43fb793a8f8524fd3605a03ab58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-04-15 09:49:25.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"dba2aedf557104c0002442b6c9b2f10e4dfc24a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 10:47:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d44f504f9794de3a410b860f32e95fa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91f1bd8ff7d04a3ab52e04280a50f1dc : SUCCESS in 22m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/75d5d95028804f1db4576fe9bec1024c : SUCCESS in 17m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3eee31ba2e034022bc75bfb5efa7d610 : SUCCESS in 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"521d8f44926ef19657796287bafb9a86e640a3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 11:37:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d221f1f0600d475a949f0872c5e0f997\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/85c853ae1e584f35a658ffb8c3dcfbc7 : SUCCESS in 1h 03m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04666a9054bd4e88a900d428b0374435 : SUCCESS in 27m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8767a2eb516c4ccd9d54cde1cabf40be : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e30cc1ccc4a047228a1fdc2b714542a1 : SUCCESS in 23m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d4fa667403f40af8a9096691c64fa69 : SUCCESS in 20m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8de49a1ad9b40978cf5b57a194fcbb1 : SUCCESS in 24m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9913f556651e49ccb349e754ea0a6817 : SUCCESS in 12m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7b0e34bd07354312b5ef742ca542b765 : SUCCESS in 1h 19m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f38f35ea986049319a83ab237dc0fc8d : SUCCESS in 1h 42m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5921c286ceb9438bb90e9b99b8565823 : SUCCESS in 56m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e599180abd334819a0191a563896cde9 : SUCCESS in 1h 26m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/14d9025800d34b6f849735113779c5a4 : SUCCESS in 1h 26m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/17299d19468243a0824da83002524cd6 : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":8},{"id":"20a5304b7c358212e2b59542464543ac93bfa578","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-15 14:01:21.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/82/940982/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"21568e0785a85aa796158cceae96ce1c401f4904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-04-23 15:30:09.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"99c68e47eb35e773484ea9b984bf9e1b36c0e97f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 16:16:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59dc71bbf9324626b02dffecafbca3ae\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c8ed57abf950444cbd5213ce86bf3460 : SUCCESS in 37m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0395b906248d458c8e3b2619d6b3a48d : SUCCESS in 35m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/69ca9078b3814a3ea8b604909f9a11a3 : SUCCESS in 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ad4a02bfefc29470fa4ddff14fa9633fb6a118d5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-23 17:14:31.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/82/940982/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"6b5b2ee21b2d53f42ca51a85a89aa1f26312d0d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 17:50:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3861024c63f54c82a8eafa9a68ac3b46\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6912dc47f38d4d5795861117d08d68af : SUCCESS in 1h 12m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc1d6694a83c4d4780bc7c036d51bff4 : SUCCESS in 37m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3cc3eb7847e46f9971747c40766e740 : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a214ec21bf7463db4c6681f1c9b1303 : SUCCESS in 21m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21153c9354924e87ad7582d54f73d4d6 : SUCCESS in 19m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/82763ff31b26433689c067706bd749ea : SUCCESS in 25m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/183ebd60b33b49d995b36e634113ecc6 : SUCCESS in 11m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f97cc29dbf54d78814463ef04ded6b4 : SUCCESS in 1h 32m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1f656329e34e47638d4a2968579ae140 : SUCCESS in 1h 55m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/546053103f1a4e0bbbc1ff227b9f4d51 : SUCCESS in 51m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cebc06aadf5b4044aa1165515cd99108 : SUCCESS in 1h 36m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b0c37f896424a56a0a99aadf4dd24b5 : SUCCESS in 1h 22m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6d3ad37f5bc64d88998020b521fc3bf3 : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":9},{"id":"2ffecad43d3025abf7f5df165e4f1037a9768a86","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-04-28 13:41:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"74d859eafd321e77f430081da5f43bab33e132d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-05 08:17:08.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"48bf93811d3c662abef270dd911b8013f516c6d4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-12 13:18:16.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"f1d5ab8aeda7f663898ef6a535ccedab5b44b24c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-06-02 13:04:56.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"6de897df9692762a335a888f8babdc52ce39ded3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 14:23:18.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4668a20b4038416daef8d112b2675e20\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0bd1a38d796c4f0197b776ea3e4beb71 : SUCCESS in 1h 07m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4754ae881154bc0a9bc0e22ac86b19e : SUCCESS in 1h 05m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f1fafeba5fd40d7a36025d3adf77bd5 : SUCCESS in 1h 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cbbe8b71758d5d1ce1517e22611ae76e568fdc09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 15:17:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/740cc57eba0542f89d24f798beb29355\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5875c06310904034bed59f76c6672ff3 : SUCCESS in 1h 43m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11c1fc186ab64adbb34e03d112daee24 : SUCCESS in 34m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee56cdb869284cc29b98ca08e1ddcf02 : FAILURE in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de13301b00f84ce6ba2dd14d2d227085 : SUCCESS in 22m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0aff40b50d9c47a4863a65b8dcc17a89 : SUCCESS in 21m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a36be35f70664092b7da59078f9e1d40 : SUCCESS in 24m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/193a250b085440b794d1d55bb59a3ac3 : SUCCESS in 12m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c49b9b7f7ad6490e8192fb1582a132b3 : SUCCESS in 1h 44m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8405198734d44f7f8c4e76983c70bc89 : SUCCESS in 1h 37m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d26dddea9ac4be4ae7a3c705760631f : SUCCESS in 54m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/add0cbc50c324b1ea14c499b53c2bc43 : SUCCESS in 1h 36m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/49b21c10494142a18c5d8d48a435c67e : SUCCESS in 1h 19m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/096e6654aec4405da57f44eb46178b29 : FAILURE in 1h 05m 13s\n\nWarning:\n  Comments left for invalid file neutron/services/trunk/drivers/openvswitch/agent/ovsdb_handler.py","accounts_in_message":[],"_revision_number":10},{"id":"f8e9d2b9af2e2c206e537b8eccd4a50cddf283fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-06-04 08:17:57.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"4fec45fb3e8ed22dac4ad544e58130336d9533a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 10:11:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8737da62f4ee4fdeb3ace1571a93a54b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fa6eb04757a645be8dc584e9630129af : SUCCESS in 1h 20m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/faf914912c5a4f6eb0bf6e5094e442ee : SUCCESS in 49m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc54f059f494ec995856600ddfb33e0 : SUCCESS in 15m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cfbfcba82dc4fd3b58298bc02d1a606 : SUCCESS in 22m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/882691a2ffcf441e881825efa0b759a7 : SUCCESS in 30m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/343c779b2f7242f58114a6229ddeb8f6 : SUCCESS in 23m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709e1ea17a1845468a2d3a96b2f2f4e7 : SUCCESS in 14m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5e96bdd77a1f447bb9a58ed18e882a8c : SUCCESS in 1h 20m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8697361440d74d33ae77041b30e83c38 : FAILURE in 1h 41m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/320b853c9be04969a63417a5fc8d07b4 : SUCCESS in 1h 00m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1249841a5e1f410caffec34494a85a6f : SUCCESS in 1h 48m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7f43376ae75d4fe69469289154b2f67b : SUCCESS in 1h 42m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d49b1fea0a964a5a92ff8c17f91326de : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":11},{"id":"7f5e8bd67ed5ef6e0f489bffb0a3943f8c8e1c1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 16:28:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f40af2e38c4d4b74befaff916f4644e7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd134756fd8c448686e1029e229a9263 : SUCCESS in 1h 09m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3ddc58dc9ee0477292f397745b127f88 : TIMED_OUT in 1h 21m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b62830335c3e41c8bf74f393398a5617 : RETRY_LIMIT in 1m 35s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"bce240dfaef695640c793f94041aca5e518a13db","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-07 17:33:41.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/82/940982/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"4aed7a698e4b435a0a77db57300f1ad32af27dc7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-08 03:25:29.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/82/940982/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"6ff7c1c0a14af57096f1a6c136204f447752c8de","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-11 06:31:50.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"4edbfe59a5fe95aa6749791751e431edc98c382a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-11 06:32:01.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"344556fd42d7e3066076f1743f617e23fdd0a176","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-11 07:11:38.000000000","message":"Patch Set 11:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/82/940982/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"6bd6a6f16d4f8f1fc2df4ed11cbb9977fc072996","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 08:02:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c72bbf563da54fbd83b605d066eefb8b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc2942e934464021bdf95798b3f3b7e7 : SUCCESS in 18m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/552c41053ce848778582685d214486c5 : SUCCESS in 45m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8d6426fd448e4ca4b2bf42998c073717 : SUCCESS in 29m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b187b4b5ac44ebed24433a8810fd97865cff558a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 09:01:37.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29fce3d6982247ac92148605ccd2ce43\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/23ab90f921824bc7a6653e4d8b2988f2 : SUCCESS in 1h 43m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c28158f6c8c4171b45b7e832da182eb : SUCCESS in 38m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b277f8e38fc9483cbe29e41a5b6605da : SUCCESS in 11m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3753c7b82d444ad9059b4c3d99ebf2e : SUCCESS in 20m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f45b753431d24525bf0cd64be381e1c3 : SUCCESS in 19m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af82af069431450fb0963d7242c9eedd : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4769ed4dba914a10a8d215bccbd5dee9 : SUCCESS in 12m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7bb3eff1af0d4d3ab16a5dc410dbe2b2 : SUCCESS in 1h 18m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b5a2443b5d2c4d31aba5324c9e972552 : SUCCESS in 2h 27m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd1eb944868946ab9fe4f199565fcdbb : SUCCESS in 1h 02m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/29baf00cf5b248f0a7b95f668dcc9fed : SUCCESS in 1h 38m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d3bf867cb2e4eb4b79d7a2eb077fa82 : SUCCESS in 1h 38m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/25789b5887b248f599f51d137f203ce1 : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":11},{"id":"b7c9a93b09cec9f4f18fe926789fca5742a52e2e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-17 10:21:17.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"394b0ea472a8d098b826cad432a4022abec18c05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 10:21:40.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"47a9332e186ee02fc56f67b0ed2e3aa365ab4413","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 12:01:58.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/444c6f88fa8d41359dff3d732ca42877\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2cacf758463741db937b9061de107a72 : SUCCESS in 1h 39m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d84fb50f084fb5a11eb6b655ea0359 : FAILURE in 11m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54555a513079466cac3bfa0fe09f9076 : SUCCESS in 19m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca36c884793c46e99bcae08adaf62e26 : SUCCESS in 19m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cabb01878fae4bfebaf44a7cc3a1dd39 : SUCCESS in 12m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cfd4ad8e63464d25a95f6a845bfa0415 : SUCCESS in 1h 34m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/eb52ead8dd1b40bab551eb229c3cd691 : SUCCESS in 1h 24m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/da3f60f47df8432c922de1cdfcdc2f00 : SUCCESS in 1h 28m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/49abecfdf55a4f08ab584a210c392b27 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":11},{"id":"ad3c265688eddfd1d7c51d3fe654968417bb359a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"real_author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-19 06:55:58.000000000","message":"Patch Set 12: Patch Set 11 was rebased on behalf of \u003cGERRIT_ACCOUNT_7730\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"}],"_revision_number":12},{"id":"fcf2eebe0f57d9b0b5ec3947476e5946481befbb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-19 06:56:01.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"95bdfe52329349c5063a3ec3f10c87eba8e80ecf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 07:31:11.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c043d2f93db4e39b049914e75ff308f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b84da24e79e34d78b41b2b0970c946b5 : SUCCESS in 20m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ceb327b7a4ab4f63abc22315b6dcd9f0 : SUCCESS in 22m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5a7fd7f2e05f40ad9fc7968ec98e70ea : SUCCESS in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8e6f3618ae9a4bef0ce815a25e0933e3b57832c9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-19 07:45:53.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/82/940982/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"af6572bb5592686ab791bb235ab400d76875d2f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 08:42:44.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f7a876bfd0f41aaadd41acdfc80b47a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b2cb0831b6c044b5a017ff7b470c5d8f : SUCCESS in 1h 24m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c54e7ca44f64e74b13b9eabf14bff88 : SUCCESS in 49m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/423991e282f94ac58e604867e730866e : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d04b3d5d2fa34e93bdf2eb65e0dd89ca : SUCCESS in 24m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9830ed69dca84f8f998115398bec8d5c : SUCCESS in 25m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/552e8025464440deb8a8a9b0b2cfb083 : SUCCESS in 23m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9afb434702c14f0fa07adc905d492e4f : SUCCESS in 12m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/372d1dcadd264cb18d849d867206e584 : SUCCESS in 1h 39m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4bb3b52dfe404c7c9895393b4bf3462b : SUCCESS in 1h 29m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/924f7e97055f41ecb79b2d6a6e053aa9 : SUCCESS in 56m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7bf658df46004c5080dab054675dd703 : SUCCESS in 1h 42m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9edb16948b0d4b4882f306ffa9a2a435 : SUCCESS in 1h 18m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/813609f22dcc409197719f2313f3776c : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":12},{"id":"0cca4e5250dea93d69bcaf1534fe2473abb6e471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 08:43:10.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"8e106a3598c5ae3341f791fa38731c0267ab1e72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 11:33:12.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2547c50030ff4c3e9abe5e90a2c09a6a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/50373da822684028b3ddfe49e3f06688 : SUCCESS in 1h 47m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01f86562c20426186dfc04b1104cd13 : SUCCESS in 10m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24dafecfc95a42e88aa9a4d9c9bfa551 : SUCCESS in 24m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/214453d8480a436583f754e295cfbbbc : SUCCESS in 15m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fae3d3994f74bb681aa366cd279caca : SUCCESS in 13m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/90d94ebe1ea3433c915dc346eedf476c : SUCCESS in 1h 42m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9bb704f0ceed495cb5f77ef16ac72e22 : SUCCESS in 1h 16m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/df7c3d21c4754b0492668d1e075701af : SUCCESS in 1h 47m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/daeb3d3d9a2342f79f6cd4fc30ad7a24 : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":12},{"id":"d8d3b7eeeb342a021fd92eb99675dd1bba2ee68c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 11:33:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"9e6768236acf24534591977358bfec9894af9de4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 11:34:32.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f984807db04996b12268db7dae8c6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5dd9883b902464facbeebbf1cb86ce1 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"429663845640fac9e3682f70941d9f80644aea1c","revisions":{"c34fa05a19c2ccea4955b43b66b1e667895cc6f9":{"kind":"REWORK","_number":1,"created":"2025-02-07 16:03:51.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/1"}}},"commit":{"parents":[{"commit":"f4559a7220402e4b23cc87c3ad51c47f888def79","subject":"ovs: reimplement signals handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4559a7220402e4b23cc87c3ad51c47f888def79"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:57:21.000000000","tz":60},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c34fa05a19c2ccea4955b43b66b1e667895cc6f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c34fa05a19c2ccea4955b43b66b1e667895cc6f9"}]},"branch":"refs/heads/master"},"137799ec1ee7bf41847c2c5f91b24354bf035067":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-10 09:17:56.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/2"}}},"commit":{"parents":[{"commit":"d3dda8fdeebb62c7cbdfe65ee81165e5a21be1a4","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3dda8fdeebb62c7cbdfe65ee81165e5a21be1a4"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-10 09:17:45.000000000","tz":60},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/137799ec1ee7bf41847c2c5f91b24354bf035067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/137799ec1ee7bf41847c2c5f91b24354bf035067"}]},"branch":"refs/heads/master"},"764f865e25fbd4d4572c942705e53fa2a915286e":{"kind":"REWORK","_number":3,"created":"2025-02-10 13:00:27.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/3"}}},"commit":{"parents":[{"commit":"d3dda8fdeebb62c7cbdfe65ee81165e5a21be1a4","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3dda8fdeebb62c7cbdfe65ee81165e5a21be1a4"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-10 13:00:16.000000000","tz":60},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/764f865e25fbd4d4572c942705e53fa2a915286e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/764f865e25fbd4d4572c942705e53fa2a915286e"}]},"branch":"refs/heads/master"},"3b90d0b9a2028ed5483a106876bd69eeaf5b7c89":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-20 08:03:45.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/4"}}},"commit":{"parents":[{"commit":"3f68b1a0be6ae6dd024584da9aff690a6743ed5c","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f68b1a0be6ae6dd024584da9aff690a6743ed5c"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-20 08:03:30.000000000","tz":60},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b90d0b9a2028ed5483a106876bd69eeaf5b7c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b90d0b9a2028ed5483a106876bd69eeaf5b7c89"}]},"branch":"refs/heads/master"},"66f3b96f95c76c4e4be6b6ea6ea494f24c998c97":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-03-03 18:15:40.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/5"}}},"commit":{"parents":[{"commit":"5ac359c84f8807fabc2495cce66129f125461361","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ac359c84f8807fabc2495cce66129f125461361"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-03-03 18:15:28.000000000","tz":60},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66f3b96f95c76c4e4be6b6ea6ea494f24c998c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66f3b96f95c76c4e4be6b6ea6ea494f24c998c97"}]},"branch":"refs/heads/master"},"427c5fa9d22615cf2bc487758401eba449af3e68":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-03-06 07:14:46.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/6"}}},"commit":{"parents":[{"commit":"d33c9554d4f9594e553056ebdddc61ce257cc183","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d33c9554d4f9594e553056ebdddc61ce257cc183"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-03-06 07:14:37.000000000","tz":60},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/427c5fa9d22615cf2bc487758401eba449af3e68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/427c5fa9d22615cf2bc487758401eba449af3e68"}]},"branch":"refs/heads/master"},"1bc896d2def4dd858e592aefc46a5c061e0ad1f1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-03-27 09:23:28.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/7"}}},"commit":{"parents":[{"commit":"6a2e3594a7e2c0d695a9fc5d5765bb02345289da","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a2e3594a7e2c0d695a9fc5d5765bb02345289da"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-03-27 09:23:18.000000000","tz":60},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bc896d2def4dd858e592aefc46a5c061e0ad1f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bc896d2def4dd858e592aefc46a5c061e0ad1f1"}]},"branch":"refs/heads/master"},"b2b2385fb874d7e0bf36167fd1bea187fe53da7a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-04-15 09:49:25.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/8"}}},"commit":{"parents":[{"commit":"5794b1c68b635ee3313f2b4d69d92683e5847ab4","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5794b1c68b635ee3313f2b4d69d92683e5847ab4"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-04-15 09:49:14.000000000","tz":120},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2b2385fb874d7e0bf36167fd1bea187fe53da7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2b2385fb874d7e0bf36167fd1bea187fe53da7a"}]},"branch":"refs/heads/master"},"2e76d5fa32b4a343cca8cd50c4664ec399ab78ac":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-04-23 15:30:09.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/9"}}},"commit":{"parents":[{"commit":"9835af0661bf4dd1baeab33ba9abc2a9ff39663a","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9835af0661bf4dd1baeab33ba9abc2a9ff39663a"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-04-23 15:30:00.000000000","tz":120},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e76d5fa32b4a343cca8cd50c4664ec399ab78ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e76d5fa32b4a343cca8cd50c4664ec399ab78ac"}]},"branch":"refs/heads/master"},"9bfc23bc913ab0a3172facb413c4dbb0c2009888":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-06-02 13:04:56.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/10"}}},"commit":{"parents":[{"commit":"50e2fe8c3f7b16fcc56e796926cbe5d43d5a5824","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50e2fe8c3f7b16fcc56e796926cbe5d43d5a5824"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-06-02 13:04:48.000000000","tz":120},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9bfc23bc913ab0a3172facb413c4dbb0c2009888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9bfc23bc913ab0a3172facb413c4dbb0c2009888"}]},"branch":"refs/heads/master"},"b744ca8dcfeeba949b73b98b36a29148b689a5be":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-06-04 08:17:57.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/82/940982/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/11"}}},"commit":{"parents":[{"commit":"092412cb612963e7dcf2178a1493d83337586890","subject":"osdb: remove usage of eventlet in dbhandler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/092412cb612963e7dcf2178a1493d83337586890"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-06-04 08:17:44.000000000","tz":120},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b744ca8dcfeeba949b73b98b36a29148b689a5be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b744ca8dcfeeba949b73b98b36a29148b689a5be"}]},"branch":"refs/heads/master"},"429663845640fac9e3682f70941d9f80644aea1c":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-06-19 06:55:58.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"real_uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/82/940982/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/940982/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/940982/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/940982/12"}}},"commit":{"parents":[{"commit":"74954ccc0b378f947befae62e15b7ca97355f245","subject":"Merge \"Fix bad i18n and fstrings interaction with flake8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74954ccc0b378f947befae62e15b7ca97355f245"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-02-07 15:56:06.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2025-06-19 06:55:58.000000000","tz":0},"subject":"ofswitch: remove usage of eventlet timeout","message":"ofswitch: remove usage of eventlet timeout\n\nThis commit provides an implementation that avoid using\neventlet.Timeout.\n\nRelated-bug: #2087939\nChange-Id: Ide6720bde733d75352feb6ee160d23318c7406e3\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/429663845640fac9e3682f70941d9f80644aea1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/429663845640fac9e3682f70941d9f80644aea1c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
