)]}'
{"id":"openstack%2Fneutron~941018","triplet_id":"openstack%2Fneutron~master~If44ce37d072cf55baed8d8064b93865c9f026157","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-11 08:18:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"If44ce37d072cf55baed8d8064b93865c9f026157","subject":"Use centralized constants for [ovn] neutron_sync_mode","status":"MERGED","created":"2025-02-08 07:29:17.000000000","updated":"2025-02-11 13:21:34.000000000","submitted":"2025-02-11 13:20:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941018","meta_rev_id":"efb49872bdd8b5c762980c6882361bd92bef8265","_number":941018,"virtual_id_number":941018,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 13:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-02-10 13:20:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-02-10 15:25:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-02-10 15:25:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-08 08:07:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-08 08:10:22.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-08 11:32:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-10 13:20:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-02-10 15:25:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"b308f30231012d55f9664f560433bb05d139d813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-08 07:29:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e33b28ff758ea5fd089949edb470489cffee8a31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 08:07:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab382f7849d45cbba3b3214a85689c8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe3c10c25ce94d5e83113d1e96a81055 : FAILURE in 21m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ded65bb951e4d4da316dee3665011c6 : FAILURE in 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4585b756191e3e6f03738e7ac553dd96ed9048e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-08 08:10:22.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/18/941018/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b553e3224ca14cf9b7220714b7050651e330a81b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-08 09:30:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1bb7bacceec98570ca47219d904fdd251bddef55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 10:00:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ed178903eeb49ba810bf5949450f7d9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c66b4c94dbbc4e37ae4eebface89dfb6 : SUCCESS in 25m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c1b41ea55c247bb8a6a78de75aa1ca8 : SUCCESS in 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2ccdc8c2b85358d1798178b9fea96ca52b844a7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-08 10:00:32.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/18/941018/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"d5faaa5a2b13c300a273890e77ea1179c573968c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 11:32:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/472ae12a1f1d4c749560765113e5b40e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2aca8fb6f30246de91e7dd1a0e67a778 : SUCCESS in 1h 26m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7c5d6c91770544be83e3816c90cbae4f : SUCCESS in 1h 58m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1e6efccf628e4908a48990a45663e1dc : SUCCESS in 1h 58m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be911981486c4c0a88b7b7044cbef0a1 : SUCCESS in 32m 39s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/633f083cfb4e4a61ae4a2107d432bf23 : SUCCESS in 1h 51m 06s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f42afa460b448809562cfe68685d3a6 : SUCCESS in 1h 54m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b12924bd2ce94f9491f1af99569bb708 : SUCCESS in 28m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa2fb630d5014d3290610df174f3f95a : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d72dc5303b74066876ce501e2998e67 : SUCCESS in 22m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7e85fdcd39949e092bde74d3c009d84 : SUCCESS in 20m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4438cdd419a041ebb72716d651316968 : SUCCESS in 9m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/58162cf346e1491293c62cc6d62a8f43 : SUCCESS in 37m 49s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f226579deaf6411d832c811f7da6bdee : SUCCESS in 54m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3bf66be529df48cb89a9684ca1381ebb : SUCCESS in 49m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3af052877d864439951e8b09a42e8e56 : SUCCESS in 1h 35m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ef5aab688c549928c53117ea4fa5d33 : SUCCESS in 52m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6741628837b844cfaabac2e648d3c576 : SUCCESS in 47m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/78450193331c4b0db07011eea0f84bbe : SUCCESS in 1h 54m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/11a6d0466b6f412283c612d8ffdd95b7 : SUCCESS in 1h 14m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8d90a341ad034a0491e5d293f2bd8138 : SUCCESS in 56m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bf3b7625d4784e588fd5ea01307541d4 : SUCCESS in 1h 04m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43a9f65e95f84d16be24a34c59b5ab6b : SUCCESS in 59m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/10089a6964844c23bba197174b47cc0e : SUCCESS in 1h 19m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9b1b05f2419ba07512de99d24dabb5420f598023","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-10 13:20:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"254a1e5690b19db4f0ab23a19ddc41f2c1defbba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-10 15:25:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b5c4aa3ab8bf9cc24de000ca457a9ed8160b87ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 15:45:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d681f75b6c848cb1efb45612edfb658bfee656d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 19:05:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29cc9f7062e44702ab72085db5b5602e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/10dedbb4305f455ab56c450136011796 : FAILURE in 1h 42m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3c3ddd3ecc9b409694dfdb49e05e459f : SUCCESS in 2h 07m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/83fabf1c6c6a4e4f83306e06d04ae629 : SUCCESS in 1h 44m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de355927c73478a81255a3424712028 : SUCCESS in 12m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3adf7b6790694247952c9f2293e075a5 : SUCCESS in 31m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca6bf9f15e1b4086aec7ee54c26d2a88 : SUCCESS in 18m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63b55f1ea9c245dd9a83d1b55c5409fe : SUCCESS in 12m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/22d0ceda7b8b47c7937c2e07999d64fa : SUCCESS in 1h 09m 10s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4d1c7bedd62a4a2a98e92eff7ae13e31 : SUCCESS in 2h 06m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e8136eaaf7c64754bf43a2a7774e00a6 : SUCCESS in 1h 51m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/658c7c91378846a7a7a7f28d2c1ab659 : SUCCESS in 1h 17m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5464d6ce7825409881d0a98f6224d525 : SUCCESS in 58m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/611bdeb2e93b42b8abb3699c6b22134b : SUCCESS in 1h 10m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d1d1220b1f314ca8a3de32af13b9dd14 : SUCCESS in 1h 25m 40s","accounts_in_message":[],"_revision_number":2},{"id":"116d7140f51de6b5e34631575ba88e34c5db24ed","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-10 19:34:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3541ff029133b82508af86e520862283b1d4cede","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-10 20:30:28.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/18/941018/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"f5bbb301a193a515b84464a5f1fa21d6091c2f81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 21:29:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c7bc4cc1ac4490ab51dbc6f0c86e315\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/776b4a65ac984a9baa4c495ee0b6ec9f : SUCCESS in 37m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f30b9685789b48119cbd7894e5e384a9 : SUCCESS in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab66b68c27623f2b3ff7ad347d572984f074e321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 21:58:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcfb502988744fcc8b942e1eb431e64a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d8867ce258cc432598e3a4cbd5d22bba : SUCCESS in 1h 42m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c179a25bd88f4b4f96c7e17a845436db : SUCCESS in 2h 11m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/17e4998ca0664983abba3e1ae4b53127 : SUCCESS in 1h 43m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5bfddbe16c044a70beedad1c8402f257 : FAILURE in 25m 26s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a8ecd45b39e4d6ca12f6f01f840ede7 : SUCCESS in 1h 27m 54s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b12b0ec29a7b4aeea0495d31e822e229 : SUCCESS in 1h 50m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc31b4656bfe4961ac839a5a54f377c6 : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389952346499432cb5c7bf68dbb8d4e9 : SUCCESS in 17m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91515116553a46c8aea6eb0f4100a846 : SUCCESS in 23m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28d8644d7e16497585f8aaf615446235 : SUCCESS in 18m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e62ed1bd4d481494b6030424ea91db : SUCCESS in 11m 02s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2cca91abba624123aaf71ff1c33d9c22 : SUCCESS in 1h 40m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/22ed0d53526748518b7b57fc76ac373e : SUCCESS in 1h 02m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/646c68422ae342ea9cc4619b05790d7d : SUCCESS in 1h 20m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ef54c1a572934c1cbd6a5f678c29a9d5 : SUCCESS in 1h 56m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/382e4e431f3b450ca5852a86baa85597 : SUCCESS in 54m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8e22d0a969446ae8b138c269634c960 : SUCCESS in 49m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3abea1d7755f4bf48bf0427071dad327 : SUCCESS in 1h 31m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/50742f0c44e74bd692c86b9f9bc3bd1e : FAILURE in 1h 05m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a72df73bf4c845c78c464016ed8da9d2 : SUCCESS in 50m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/92dcc43fe5944a51a3b2b649ec19d1a2 : SUCCESS in 54m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9938d0131124214a5b1eefd895e3f41 : SUCCESS in 1h 06m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8cf1298b839c4e2cb44a839f020d641e : SUCCESS in 1h 34m 55s","accounts_in_message":[],"_revision_number":2},{"id":"ca67cb9ee2787dd3424559549c20ddf9f03e3be8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-10 22:43:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"042af331dab7cb4f0fe9939c113e1b5d38d64e3b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-10 23:04:29.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/18/941018/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b80140a6cadf1a8db6898c0b837b683ba0f333ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 23:16:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d65387cdb82498380c6d15ce0fc3347\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8299f8c9ac5485380febc7bf8a04537 : SUCCESS in 26m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/184a4743f1884f2cbb382bd32a72ec1c : SUCCESS in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8319c76913b4fbfeafb35e19f8f9ad8fa5ef5ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 00:55:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/323e78605a6b41e7a0addc6f31e83d8d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/75835b63249042b7b86c170eed8c8f43 : SUCCESS in 1h 33m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/35e11003a4944d648a5ebda9ea6dfd5e : SUCCESS in 1h 29m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6eee95803d9147e992694125f918f896 : SUCCESS in 1h 25m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04d2e7b4c4cf41fb96be80cb78bbfd10 : SUCCESS in 39m 11s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b202e97353241fea2b9b1be681ad3da : SUCCESS in 1h 40m 01s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e553d06d4e7847f2b08bad16a1940e23 : FAILURE in 2h 10m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9379bf19d9146b9a905ca0232468162 : SUCCESS in 53m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9000bb6e87cb4e2b9c5eec758ece46f8 : SUCCESS in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d97fad7a94704a6d8152e73b7e6ddc18 : SUCCESS in 17m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45a3879db193442d9aae39ca6da57165 : SUCCESS in 15m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b0fc5ff82644f7a11ed333dbe01663 : SUCCESS in 9m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bdb64502d4064f62bfc879004d300059 : SUCCESS in 57m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/08dfd821fb1d4a33822b9863db392ab8 : SUCCESS in 42m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6557376168534ac8a7353bbe29b47549 : SUCCESS in 59m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/92f0b44a6e1a450a9f1f3a1b913dcc74 : SUCCESS in 2h 10m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/026603072bd34ebe86c44bf40666e2bd : SUCCESS in 49m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/570ad09a05434dfa9693e5f636202ee4 : SUCCESS in 48m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0b151171eb974f058c0ae366184319cf : SUCCESS in 1h 33m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d0b21a50acd34dcbbd8de15747eedc1b : SUCCESS in 1h 24m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d6e21edb1aff4b40a2281256e3a2f530 : SUCCESS in 1h 08m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a544394c41d64c678aa548a7954e5613 : SUCCESS in 57m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c3f44343f0e841039285f8d66e6f8115 : SUCCESS in 53m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c8dc8c3e3fd94f77bf8638099968eb8d : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c4d565b6e33310255bc50cf93c49b4205f27380b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-11 01:24:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c42dc16104ced99ef32adcd3f67995f3cad2d06e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 02:00:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61cc2d8b733d4708aab486c8525cb7dc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b976eab32b9443cc8fac844b7631b3d5 : SUCCESS in 23m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6eed5fdb2b6a486f95d8b3c2f694e079 : SUCCESS in 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19cfd64e77a69423acc886db4d924be6bfd5b30b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-11 02:03:03.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/18/941018/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"131b671a3235de3b974b4a9296119b2d3ea3931c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 03:51:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e617125986a41c592687e6e1d888412\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d52a970b4e0c4605a5d9ca1003c2ff60 : SUCCESS in 1h 54m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4b7a32b24b3646a3b74edc0fbf83a29e : SUCCESS in 1h 55m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e7e7d17118b14fab9ea1635ca494289a : SUCCESS in 2h 17m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6017266c36ca490ca229d30ca0864014 : SUCCESS in 33m 17s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b77919c55b243ca94679d7b4ab02ba6 : SUCCESS in 2h 01m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26f3e800b7dd4740ad48fd57a8cfeaae : SUCCESS in 1h 32m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8264da3c652a47e99254109666b9f83f : SUCCESS in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71aff785f214c0ca7d0249e783adc7e : SUCCESS in 10m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66c249067b8443ff8b161a2104602dea : SUCCESS in 23m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27820339f783486c821d70cbfbd24fea : SUCCESS in 17m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1789a4da5aff416b8ea3d22a7a7b9cd3 : SUCCESS in 10m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/59d7c8d0b4334286b1a11f0bb2d57710 : SUCCESS in 34m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1b67b1a5ad0b4fc38e3f1999d4941cf4 : SUCCESS in 50m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d72f45b847f24582a84d16ef3140955d : SUCCESS in 1h 11m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4f5e7197fd324e20b1c50a225e3f0bcc : SUCCESS in 2h 03m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f072f54c1fd4cb5a0dfec2e58a58320 : SUCCESS in 56m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4736ab66ae37483c9260e7f15ad002be : SUCCESS in 1h 26m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f119b76d7c2f497d98651f47a03fe839 : SUCCESS in 1h 49m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/26de8e108e8a4904a47820869e3e1eaa : SUCCESS in 56m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5446abd768bc44df9ab7c9ef8ba1e879 : SUCCESS in 1h 03m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/59b8e10aa1884f13b530a195bc1b3cc5 : SUCCESS in 1h 09m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8367fbd6808e484e89879ec035a0b299 : SUCCESS in 59m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a5a13e365c4840a4b847ba989d692a40 : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c68ce5541cc04f4676816c1b8662b8fc1ff63f38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 03:51:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0bb667547c2b47374fd2c02d8fed1f4dfb717a66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 05:54:29.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8593bd6581545ce90724e5cd26a81ff\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7125313478994c08a5c2551cf5a60ac0 : SUCCESS in 1h 17m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e4ed7716f8024334a4aeb7014d0a574e : SUCCESS in 1h 50m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6ba0efbefed74b30913bef0bf6177d09 : SUCCESS in 1h 38m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/958f413ab303494ba3591b3d4f0b8e11 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67a3b52ed4ac49829e2172c7debc493a : SUCCESS in 20m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a99265d7436462f9d9544698c6c2f03 : SUCCESS in 17m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a872a7cd104bd593d8de4148402936 : SUCCESS in 8m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5ecdd15da1eb4356b2e24d87d5d36faf : FAILURE in 46m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/585508076ee84254ac5b3bd64e46f60c : SUCCESS in 2h 00m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/11653b897e294edc8f9f3aae0c565aad : SUCCESS in 1h 26m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca186509de654513aaa6bb7b4e3d3d68 : SUCCESS in 1h 06m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8518a3d223c24219950ca87963aa3d62 : SUCCESS in 1h 09m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2ac9788313fd4c00bd94334246ec2035 : SUCCESS in 1h 18m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eb73470b187a44dd9b3af65164f0a039 : SUCCESS in 1h 41m 09s","accounts_in_message":[],"_revision_number":2},{"id":"4ce2234f4989693cef633d2ae41e8a11bc40cd4b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-11 08:18:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f937914b10d1b569c68822e3a22bb58a4ed08d83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 08:56:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e31fa559e2e4bc3bec11875b81aa1c0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04cb1dc5e82c45248b4b1f1616be77f1 : SUCCESS in 25m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31182d101bb04602abb39e94cbd32f51 : SUCCESS in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1be31bc831c419911d5c214b672366d2cb2a1e52","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-11 09:35:25.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/18/941018/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0d20127ec06254553c54ca863d2242e06c528548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:00:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e55fcd474bc546e08c4c6515dae0c4aa\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f9cd95ab9194988976c668da12f11ea : SUCCESS in 1h 15m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c96ecb0bc8a46b1bad2f5c4ccc60aa0 : SUCCESS in 2h 23m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dc2afc185b244f449ec158ff09659345 : SUCCESS in 2h 04m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cde55b8716e548f8937374f8541815d4 : SUCCESS in 49m 56s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b653846f8ac343378d90102e7d334855 : SUCCESS in 1h 42m 45s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a157c241a52c4ae7ada69dd5796c1193 : SUCCESS in 2h 40m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50d0a3b0651e4cb197a6fb3290cdcf91 : SUCCESS in 29m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4317910e64b461caf199f5d248f2b6b : SUCCESS in 11m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ef2d4a24e9f4d19a509061e1802428c : SUCCESS in 21m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe5d1c238b924b968b69ab697a24786f : SUCCESS in 18m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6120492f930f4f04bfd1ab70eb7d1755 : SUCCESS in 10m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/36d07fa6dd5046d28c811cdee2e0d0c4 : SUCCESS in 58m 04s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/533ac286114a4830aa344119f0a31094 : SUCCESS in 1h 29m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f3ba410cc1434e5596e497b9eb47d46d : SUCCESS in 1h 12m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/371d1168fbcb44da989661023ecf655a : SUCCESS in 1h 27m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fcb6de0cd344b2f94d534491d9a8cb2 : SUCCESS in 1h 02m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb70547e05a2439fa412439f7046a48a : SUCCESS in 1h 25m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cfea186887064b5dbcbf8c386a034b83 : SUCCESS in 1h 51m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/90057c2708d446e4bd17fac709059988 : SUCCESS in 1h 26m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6b27b5bacc844b9e99cb9facfc76a1a5 : SUCCESS in 1h 13m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3831f9e72bba4d759fb413e6e7a07655 : SUCCESS in 1h 06m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b0e457ffb73e462aaa98200c3072b0b1 : SUCCESS in 1h 25m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ccd136b169324cbc8101d94c0ec43e23 : SUCCESS in 1h 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3548e68f75c7512c631ceae27db07aec864b4399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:00:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7eed36e6df248c93a10f994ac8b6e1803249722a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 13:20:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36cbe4637fc74439a02dbc32fd33fcaa\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/92eea73b15b64f1c9e95c64f8fa7b96c : SUCCESS in 1h 31m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9221b3512ebb4d48955f02fda667623f : SUCCESS in 1h 44m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8874b8b652b4da080c167cd41036b08 : SUCCESS in 1h 46m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a103d22c014f8287da59f56a559d52 : SUCCESS in 11m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c222b7c1a9df402c94bf51ffbb5db541 : SUCCESS in 18m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c499efe7a8e243599b34a4af421f584b : SUCCESS in 19m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c96677f0fa540b8a3ac7ad1d94a128c : SUCCESS in 9m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/46ce1a9a9b044a1e95101e9d09d5dcdb : SUCCESS in 59m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b39259e8979542ef9dfa844c94c7fb8e : SUCCESS in 1h 39m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e77b43ecc92c414abfeda1acd0fa2d06 : SUCCESS in 1h 48m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a1ef8415d4f4d96bf8ba0ecc5ccb04d : SUCCESS in 1h 01m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8dfa96d9e2347c99330440822d53d76 : SUCCESS in 1h 00m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/89ae5eb9cf494af595c39115edf206a1 : SUCCESS in 1h 19m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/deffb5501d99404e81bc10d1f662f4f8 : SUCCESS in 1h 42m 31s","accounts_in_message":[],"_revision_number":2},{"id":"3054e84d2d6d3e9e3dcc8d248c6b8da0c65d3df1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 13:20:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"efb49872bdd8b5c762980c6882361bd92bef8265","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 13:21:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7557421da7d5492b83e6a537a156c7a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e2d2cf1c4d645cc9c08754b8a93dd76 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8f72fa40c06ffdcbcda91107acd8b4165ff5ae8a","revisions":{"7ae537830ef88774bd5db9fbb03a78cf1fee4073":{"kind":"REWORK","_number":1,"created":"2025-02-08 07:29:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/941018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/941018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/941018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/941018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/941018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/941018/1"}}},"commit":{"parents":[{"commit":"8a74d953c2030ab65d2599c3f50cc1f5a8c92560","subject":"Use native interface to explain choices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a74d953c2030ab65d2599c3f50cc1f5a8c92560"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-08 07:27:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-08 07:27:53.000000000","tz":540},"subject":"Use centralized constants for [ovn] neutron_sync_mode","message":"Use centralized constants for [ovn] neutron_sync_mode\n\nMove all distributed definitions/hard-codes to the centralized OVN\nconstants for easy maintenance.\n\nChange-Id: If44ce37d072cf55baed8d8064b93865c9f026157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ae537830ef88774bd5db9fbb03a78cf1fee4073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ae537830ef88774bd5db9fbb03a78cf1fee4073"}]},"branch":"refs/heads/master"},"8f72fa40c06ffdcbcda91107acd8b4165ff5ae8a":{"kind":"REWORK","_number":2,"created":"2025-02-08 09:30:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/941018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/941018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/941018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/941018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/941018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/941018/2"}}},"commit":{"parents":[{"commit":"8a74d953c2030ab65d2599c3f50cc1f5a8c92560","subject":"Use native interface to explain choices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a74d953c2030ab65d2599c3f50cc1f5a8c92560"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-08 07:27:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-08 09:29:53.000000000","tz":540},"subject":"Use centralized constants for [ovn] neutron_sync_mode","message":"Use centralized constants for [ovn] neutron_sync_mode\n\nMove all distributed definitions/hard-codes to the centralized OVN\nconstants for easy maintenance.\n\nChange-Id: If44ce37d072cf55baed8d8064b93865c9f026157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f72fa40c06ffdcbcda91107acd8b4165ff5ae8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f72fa40c06ffdcbcda91107acd8b4165ff5ae8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
