)]}'
{"id":"openstack%2Fneutron~942393","triplet_id":"openstack%2Fneutron~master~I22a275d84584f8e0c9b1adb90ee5b5f22fe908c5","project":"openstack/neutron","branch":"master","topic":"bug_2087944","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-02-27 09:21:45.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-02-28 16:48:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I22a275d84584f8e0c9b1adb90ee5b5f22fe908c5","subject":"[eventlet-removal] Remove eventlet from DHCP agent","status":"MERGED","created":"2025-02-20 18:49:25.000000000","updated":"2025-03-18 06:55:54.000000000","submitted":"2025-02-28 16:48:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"942393-bug_2087944","meta_rev_id":"81c2d4609335d4004ea02db9eca2c7df0bfad0b2","_number":942393,"virtual_id_number":942393,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-02-27 11:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-28 16:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-02-27 09:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-27 15:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-02-28 12:15:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-02-28 12:15:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 19:29:20.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 19:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-20 21:05:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-24 07:28:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-02-24 17:36:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-02-26 09:48:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-02-27 15:11:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"09dfbe4224e7b316714be19ca6730756a92064d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-20 18:49:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07e45ecd22f166591a5d5ce4958477bbefae6906","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-20 19:29:20.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/93/942393/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"f92fd4cce548099a94553abf198783c8b3b3ee2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 19:47:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d5de6955be4b4ebf57596da47d4aa9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75b2c6f3ecfd45c6a90bfbe562e2b1b4 : FAILURE in 53m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f7cc31db84b43e58645ad41299e7f87 : FAILURE in 53m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acb881f27047ac16c910e1138631b566e12931da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:05:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19ceee8808644992be17b91b4360260f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/02b8ead8c4d140918108e8c57b6e9c16 : SUCCESS in 2h 02m 28s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49e790bf0ba4492691bd6080fad35d62 : SUCCESS in 1h 32m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e24c5933529646a496cc881585285d67 : FAILURE in 26m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb366c5f01d3411e852a3411911e8856 : FAILURE in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/741efed8d97d4a76876cb8a9f0fc2058 : FAILURE in 20m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e90be56d082b4226a5a94a10baa763bf : FAILURE in 18m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec1822d963a46728d69c77d301f0a0c : SUCCESS in 9m 55s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0090ba20cab240038ee9df4080d25dc7 : SUCCESS in 59m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dac83ceac6ca4ac093014c1547a11247 : SUCCESS in 1h 16m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6aa14a553f764786a4acccba715dde2d : SUCCESS in 2h 09m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/12c97948cdbc43c58777c05af920ae27 : SUCCESS in 51m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e50830f95ab14ba1a2ac152267e8f2d8 : SUCCESS in 2h 05m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e811cce170d4ad6a90c229786121b79 : FAILURE in 52m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cfdc81d0e1184bd59c13c6b47fd4399c : SUCCESS in 49m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9ca3ebbcdbc58cb49c2371c65ef8d5138db6c15e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-21 15:06:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b46a9b620a6a082475ccc4b8c03b641906a05a86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 15:50:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0839f1316ed74a06af8574a96009f750\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76955de925724db9bbbc02ca61c2d9db : SUCCESS in 39m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce9d3ca9fa8847ec96b2613f4bcb654e : SUCCESS in 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7aab33d614f2dbd1d40988d118ed7081e8f2387","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-21 16:16:18.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/93/942393/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7612e79c178b88efe3a122500fd4a62fc0928603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 17:41:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9844a7b8b3a94f52992400592d2fe062\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/05fa0f598af847829289a013ba64aec6 : SUCCESS in 1h 52m 10s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d216a806a7c40a392da2504da280c01 : SUCCESS in 1h 44m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bbd2789aa3341d688f7d0c4e41ae3f2 : SUCCESS in 32m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e863fbcf1cb4242872d0071efc28270 : SUCCESS in 10m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1665bb919f74f3ca40d9550d0606c6c : SUCCESS in 23m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88e9efca788b40dab336df65f58056db : SUCCESS in 21m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c07ae96cba94014a6f5f2ebb462484e : SUCCESS in 10m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e5aedeea407946b393a9b6a64933cc82 : SUCCESS in 1h 05m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c4ac0ac6ed02434593b662f7123bd75e : SUCCESS in 1h 00m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/029f1ad3c0f74791901d9b5fd112df2d : SUCCESS in 2h 02m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/42bff16e55a647469b9c811e92cc4362 : SUCCESS in 53m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/938f12c207bf435f8b15fe52eee268cb : SUCCESS in 1h 52m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0fce089397584e9f84f9e7fba4af3726 : SUCCESS in 1h 40m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a2ac0a418b3f4c34a917e537d15c22bc : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f4711d2a26b129a542fddd5e0cc3382b467d2079","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-24 07:28:52.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"73babe2dc8e351215c0ad83926d61e0cbc941c0c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-24 07:33:00.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c0a7b9007fa7c687dbafb7161f0b68a441ec413d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-26 09:48:50.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"df31c3169043779803c491b481985d55c3b09925","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-26 09:49:20.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"a924c5b928f1411615ebcaed13f0efcaf07cb7b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-27 09:20:24.000000000","message":"Patch Set 3: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a15c573af087a31398affb01ad0bfc7074cdfd48","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-27 09:20:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"06390f99af7445f712fd1fe5eb2f3b87b7e70540","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-27 09:21:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"93b1112fc7fd83cfdf13e0735b3bb727ca0ae7af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 10:18:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0e7ea83f4b0424482d335f939ee1165\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b2445045548422ebcca2ac9ec61db27 : SUCCESS in 30m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d1d04f0ecf14d1da71de23a1e813b3a : SUCCESS in 44m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dcf7c21807de956b347c688701e8b5e302908d81","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-27 11:35:45.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/93/942393/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"6f9c743628d94eb645bddc97ed6213e2a7426e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 12:03:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/198aa1792c4c488ab5677b04a003dc6a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9f2ba011e0e840c89b25bbacea830773 : SUCCESS in 1h 29m 29s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8cb37e7fadc406fab8031bb99e6060d : SUCCESS in 1h 37m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f72cd6488c3840eeb4db277d39611129 : SUCCESS in 48m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca14760cff5d45f0a7cf1228dbe1c074 : SUCCESS in 13m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5992e20560f4b9395657dcdbe462af0 : SUCCESS in 31m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8d4444d1a704fdea7091f44da689c53 : SUCCESS in 20m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cefe574d1794fa0803522a8f17ad7c5 : SUCCESS in 10m 00s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a49611dd05564f7c9b42b9a16374720c : SUCCESS in 1h 02m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/eb16c4611edf4aaba2ba92936ebb7a2a : FAILURE in 37m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7aead6a0f7b94dc88f232b53bc9230b0 : SUCCESS in 2h 33m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c4261d8a4804550ac9f28738c85dee9 : SUCCESS in 1h 06m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/60ae24a103d54c42b487318f02bda6f6 : FAILURE in 1h 25m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7d0594ef3ae04a7bb1da5a8e02b32423 : SUCCESS in 1h 35m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e4fd7e7024a145188dd4f887d901d318 : SUCCESS in 1h 22m 23s","accounts_in_message":[],"_revision_number":3},{"id":"d95861e63266642a00e398d36cd1db44ae171e14","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-27 15:11:59.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6bb3db91d1875181d9450f3030227ce9ea16dc58","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-27 15:50:13.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/93/942393/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"c186d2c0b81419998d1900cad199b8a6bd89668c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 15:57:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c3ae5a8383e464eb5153664e5dc06d6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31430263592b43b8bb132834078539af : SUCCESS in 34m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6ad312a70a524654be65996bbb989bf8 : SUCCESS in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8c5bb010fb69635d4583ee0f3be235308e82388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 21:46:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6faa16e8ce1a403bbd391199534bec23\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/78b15dca4fe84d16b99cb0734dae409c : SUCCESS in 1h 24m 52s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fded2ad57e0243f48d8776fd67c06629 : SUCCESS in 1h 28m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1537b0a67174f30b5b36901d4b78c4c : SUCCESS in 43m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d793b099244b528c91f9e698cb1376 : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e922e8738b8e4405a18503409242a336 : SUCCESS in 23m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/516ce64eeeb64bfbafe28faad0b43a27 : SUCCESS in 18m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe9574cf2a949609238a906383e44ea : SUCCESS in 9m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b3fac824bb794219bab1ac6c3c9f1151 : SUCCESS in 1h 08m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/84e14ac08f1845d0867eb1173e59fea2 : SUCCESS in 1h 15m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/90ee71e23cd249b685257f1679351f87 : FAILURE in 1h 40m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/944088441f134d13805cdda8fa9138db : SUCCESS in 51m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/44d09cdf3ce1426681633e960df7b566 : SUCCESS in 1h 52m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8e991b474ae4190bfc7450a6016f534 : SUCCESS in 1h 38m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e8f85893d2e945e29ce610833a9d1395 : SUCCESS in 49m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f4a039f85a0f82dc8b19d3196c218f347aef4e45","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-28 00:14:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"07f725b2eee929b7bbe5c6bd30aa0e6c8ceb9dd2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-28 00:47:53.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/93/942393/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"05c6c5f22e1f6b40e53539c4e16264a5213607f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 02:43:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6215d4fb947f49e2a193a224c8527041\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f9a523146627491eae9aea29b85b4286 : SUCCESS in 1h 28m 50s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39fabf57099f47a1a7c989e4d98fd599 : SUCCESS in 1h 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6910d5aba9694cccaaa2535d833da29b : SUCCESS in 32m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25738cc31b30422c87ca73a5292670b2 : SUCCESS in 12m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19198750134b4bceb0fee4ea32df5fd1 : SUCCESS in 22m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97fd7550b6ba440f89774f4ba1862fc2 : SUCCESS in 20m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/815da38b5da04cc89e78407646e8cfcd : SUCCESS in 9m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3815e72b3cae40e4b27094c1efc4de1b : SUCCESS in 1h 26m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/470c0a642ef34f6bbb4620d50b198a3e : SUCCESS in 58m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4d43bf5c41d44933b977f37492215e75 : SUCCESS in 2h 05m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ce179507d9a43f495f5da6c9cdd955b : SUCCESS in 1h 05m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0c85beaedb834736aca0f41ef5c592a0 : FAILURE in 1h 11m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fc555ec8ef3b4e6aaa9f6bb981083aa4 : FAILURE in 1h 34m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e86cb974f82a4da0b14cc702d3569ac6 : SUCCESS in 44m 29s","accounts_in_message":[],"_revision_number":3},{"id":"c40d4225a48f8782b64b33462e813da82c11f4ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 03:26:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c4e9d1e5fd24aa4b96f443d04204ec8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2875a4b0c5c74d75b6816ae38f9af2e7 : FAILURE in 28m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/141e0c46df4d49abad635eb524b4f5a4 : SUCCESS in 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b69283e32b8084dd8f031bd36917b07174634e74","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-28 12:15:34.000000000","message":"Patch Set 3: Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5233fcac0bfb30bff52537de5183f26ddb4b5220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 14:20:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a196ff7d1d7e4e43b1e1af845d5bc240\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/09a47c18faec4896ab7b627198f17e06 : SUCCESS in 2h 03m 03s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c264a9e077de4b4bb3503aa794c50205 : SUCCESS in 1h 39m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e67ad65bcc24f21a09dbc66a9430e27 : SUCCESS in 37m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c53cda7eaf4dd081591a768fefe233 : SUCCESS in 11m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cffde37f059486b8fc56433be8cd250 : SUCCESS in 19m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6471c496f824b02a9bf73432d2f026f : SUCCESS in 17m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44dffc4349854095b6a9c8670da79ed1 : SUCCESS in 10m 19s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a9558d10e26448e7beb6ea9ffa61b1fb : SUCCESS in 1h 04m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9d37b731a3514e08a6d5f3cf6cdb6720 : SUCCESS in 1h 00m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/80592b6f3a404215b6c745768dc3d9d1 : SUCCESS in 2h 01m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4cf2629dcd54ccc938ade93bce39156 : SUCCESS in 1h 06m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5dcf36c82c5c403dbaf98434c03190b5 : SUCCESS in 1h 33m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/918a36d87a634afea9c58afa30e419bf : SUCCESS in 1h 33m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ee6e99d7571144dd8d2a078ff7a15ad4 : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":3},{"id":"43b9315864938ddf9bd59f87c6423a2f6515f63f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 14:21:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aaf39262f568adc4c9c345ecf7689e79276e0029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 16:48:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3400cb1f28714ee6b25a30b56d392d59\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/89dea4720ac44555a39b4367aaef628d : SUCCESS in 1h 56m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003259f99556460a9996d52b4ae93f8f : SUCCESS in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dbc9c274994402e95a0ebfbc8c207dc : SUCCESS in 19m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ac0f7deabf946109c7e2bb944153618 : SUCCESS in 21m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2cf618cbd8747ca935857e323c20b22 : SUCCESS in 8m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e4771bec21eb41ca94f91f375e327ff3 : SUCCESS in 1h 30m 14s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/366ce1a4783048d8abb30eeabfac3213 : SUCCESS in 1h 49m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2d10f74c3718451681828926c820ab9c : SUCCESS in 2h 18m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b077518504a49a7a6ddcd5fe5eb05b5 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":3},{"id":"79bdba0c5e69f16abf770ea3c229de7e1a00eb2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 16:48:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9c860d75b4d491b5b3cd9e4b3ba1ea5d71cae06d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 16:50:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee47028684f94532985f20dd48a459d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bcdf53814f64af8a25f826fde4f9386 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3},{"id":"81c2d4609335d4004ea02db9eca2c7df0bfad0b2","tag":"autogenerated:gerrit:revert","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-18 06:55:54.000000000","message":"Created a revert of this change as I8bd0580bb830e0317ca6663c67b58842c34b87b4","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"534c2dfdaa839ed7b30d1c2ce6d53fb8954b2be2","revisions":{"db591963f0b8e603e8a988162e5364589f651580":{"kind":"REWORK","_number":1,"created":"2025-02-20 18:49:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/93/942393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/942393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/942393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/942393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/942393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/942393/1"}}},"commit":{"parents":[{"commit":"86f94de99aa08b1b4aadca8e90c6e79487171b8e","subject":"[eventlet-removal] Remove logger mechanism from remaining ML2/OVN CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86f94de99aa08b1b4aadca8e90c6e79487171b8e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-20 17:20:47.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-20 18:49:05.000000000","tz":60},"subject":"WIP: [eventlet-deprecation] Remove eventlet from DHCP agent","message":"WIP: [eventlet-deprecation] Remove eventlet from DHCP agent\n\nRelated-Bug: #2087944\nChange-Id: I22a275d84584f8e0c9b1adb90ee5b5f22fe908c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db591963f0b8e603e8a988162e5364589f651580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db591963f0b8e603e8a988162e5364589f651580"}]},"branch":"refs/heads/master"},"853fb4587f2759015865d9a5ec1522eba381dc6a":{"kind":"REWORK","_number":2,"created":"2025-02-21 15:06:17.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/93/942393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/942393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/942393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/942393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/942393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/942393/2"}}},"commit":{"parents":[{"commit":"86f94de99aa08b1b4aadca8e90c6e79487171b8e","subject":"[eventlet-removal] Remove logger mechanism from remaining ML2/OVN CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86f94de99aa08b1b4aadca8e90c6e79487171b8e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-20 17:20:47.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-21 15:04:33.000000000","tz":60},"subject":"[eventlet-deprecation] Remove eventlet from DHCP agent","message":"[eventlet-deprecation] Remove eventlet from DHCP agent\n\nRelated-Bug: #2087944\nChange-Id: I22a275d84584f8e0c9b1adb90ee5b5f22fe908c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/853fb4587f2759015865d9a5ec1522eba381dc6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/853fb4587f2759015865d9a5ec1522eba381dc6a"}]},"branch":"refs/heads/master"},"534c2dfdaa839ed7b30d1c2ce6d53fb8954b2be2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-02-27 09:20:24.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/93/942393/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/942393/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/942393/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/942393/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/942393/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/942393/3"}}},"commit":{"parents":[{"commit":"86f94de99aa08b1b4aadca8e90c6e79487171b8e","subject":"[eventlet-removal] Remove logger mechanism from remaining ML2/OVN CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86f94de99aa08b1b4aadca8e90c6e79487171b8e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-20 17:20:47.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-02-27 09:20:24.000000000","tz":0},"subject":"[eventlet-removal] Remove eventlet from DHCP agent","message":"[eventlet-removal] Remove eventlet from DHCP agent\n\nRelated-Bug: #2087944\nChange-Id: I22a275d84584f8e0c9b1adb90ee5b5f22fe908c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/534c2dfdaa839ed7b30d1c2ce6d53fb8954b2be2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/534c2dfdaa839ed7b30d1c2ce6d53fb8954b2be2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
