)]}'
{"id":"openstack%2Fneutron~942489","triplet_id":"openstack%2Fneutron~master~Ib4ee7e0e8b13141739f5a8600333be856afd4a4e","project":"openstack/neutron","branch":"master","topic":"bug/2053274","attention_set":{},"removed_from_attention_set":{"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2025-04-04 07:40:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib4ee7e0e8b13141739f5a8600333be856afd4a4e","subject":"Add MTU for all non-external ports","status":"ABANDONED","created":"2025-02-21 20:22:53.000000000","updated":"2025-04-04 07:40:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7bbd0fff51fe534548e4cf13afd3dd358bee17ba","_number":942489,"virtual_id_number":942489,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-02-21 20:56:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-21 23:24:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 20:56:48.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 21:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-21 23:24:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-24 22:21:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-04-04 07:40:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"94a20410623fb12e3f0c0928e623e21b41095da4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2025-02-21 20:22:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4825694d0c634fcb4d3ee8ea3c37c4b39be8fd9d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-21 20:56:48.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/89/942489/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b14365c93f388c28e157d1a2e137a81c3650c72c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 21:03:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/444226e603e34c8a934d5dcbe46266a6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70f01cab87f54f2daec39a046a92c8f6 : FAILURE in 28m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32e7e6c5640c42f4adaa2c3df6aaaf97 : FAILURE in 28m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b239fa8a26eac87552b17d603dfdf5d34cb232f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:24:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7739850a0c5c48dd9d85f2085dc63b55\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b6180592832c4824a0ac9d36a9f1e59b : SUCCESS in 2h 59m 33s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/412d49100320466c8b4b76a890388116 : SUCCESS in 2h 02m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46228f4b2ac74b0db723a7c09aefbe81 : FAILURE in 33m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76910c032f7d45e9b3782c50995e90df : SUCCESS in 12m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e58cab4b5f04462ab43b5cb312302e8 : FAILURE in 21m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b33219ada5e34ea7a8a616d0d2915f10 : FAILURE in 23m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e7cb7481e004d05b05de4701bdb875b : SUCCESS in 12m 13s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c15cbc15b27d453cb322f7c46cb26c1f : SUCCESS in 1h 08m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0f7f436d2e8a40629508cc3617cb0be7 : SUCCESS in 1h 11m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/19d320d9ea224be8a3360b1790428113 : SUCCESS in 1h 14m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b495b8afae5a445a99db07988842f757 : SUCCESS in 1h 04m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ae644c07153f4f0e9f472222cb996e94 : SUCCESS in 1h 10m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/432e89750e0d49d8991510b26a9fd5e3 : SUCCESS in 1h 27m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7bbd0fff51fe534548e4cf13afd3dd358bee17ba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-04-04 07:40:29.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cabb48e339a1f646b4a820d41eaef5c34e678708","revisions":{"cabb48e339a1f646b4a820d41eaef5c34e678708":{"kind":"REWORK","_number":1,"created":"2025-02-21 20:22:53.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/89/942489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/942489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/942489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/942489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/942489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/942489/1"}}},"commit":{"parents":[{"commit":"8152b83384c45d48a7256dfb40a2eb06852d5295","subject":"[OVN] Isolate ``test_maintenance.Test*Maintenance`` FTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8152b83384c45d48a7256dfb40a2eb06852d5295"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2025-02-21 20:20:18.000000000","tz":-300},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2025-02-21 20:22:51.000000000","tz":-300},"subject":"Add MTU for all non-external ports","message":"Add MTU for all non-external ports\n\nIn the previous fix, it had brought the MTU lookup within the\nif branch if the vnic_type was VNIC_REMOTELY_MANAGED which\nmeant that it didn\u0027t get added for anything other than that\ntyper of interface.\n\nSince the MTU is used further later, we want to capture it\nbecause it goes into the OvnPortInfo which will go into the\nexternal_ids which will be captured by the metadata agent\nto fix the correct MTU.\n\nCloses-Bug: #2053274\nRelated-Change-Id: I7ff300e9634e5e3fc68d70540392109fd8b9babc\nChange-Id: Ib4ee7e0e8b13141739f5a8600333be856afd4a4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cabb48e339a1f646b4a820d41eaef5c34e678708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cabb48e339a1f646b4a820d41eaef5c34e678708"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
