)]}'
{"id":"openstack%2Fneutron~943372","triplet_id":"openstack%2Fneutron~master~I95b0be8383586a788e1259574c18c14019b73e28","project":"openstack/neutron","branch":"master","topic":"ha_chassis","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-06 08:15:26.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2025-06-04 07:49:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I95b0be8383586a788e1259574c18c14019b73e28","subject":"Introduce noop OVN Router scheduler","status":"ABANDONED","created":"2025-03-05 10:26:15.000000000","updated":"2025-07-09 14:15:30.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a6835718bd64c6e47796e2bcbbb53a9b49279aac","_number":943372,"virtual_id_number":943372,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-03-05 15:29:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-05 16:29:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-03-08 20:27:55.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 10:59:51.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-05 11:17:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-05 11:26:34.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-03-05 12:53:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-05 19:39:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-06-04 07:49:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-07-09 14:15:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"05541e3367a5989d9ad52e86455c6369f7ac21f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-05 10:26:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e00da22307114cbd4227ff3e1631f7b70f32b266","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-05 10:59:51.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/72/943372/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"5b37ced2654e9a341292e06170c732c43268364f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 11:17:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67ff1bb78c774e5a95f1dbab60c54f29\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eeccb7c619af4cc0be2ee97fd8a05b02 : SUCCESS in 46m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32e1898f86574e469ecdfc774c740e30 : SUCCESS in 39m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f99535f8be3a069e86f176010bdec6fbaf538607","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-05 11:26:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78a791de3990f7027582621a64e782ca1ddebf26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 12:53:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db04a4066a4f45f38b7538981e0076bf\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c3380fa264e0410ea88d39ff74a090c6 : SUCCESS in 1h 43m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/17d93b63032e466eb7e0b340451a1a4b : SUCCESS in 1h 31m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fdc3500e3be84ad3bb62d3cecb42fc46 : SUCCESS in 43m 21s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3df7aa822e0a4555a46082bd58656993 : SUCCESS in 1h 39m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c3e306009e2415385a1005dc82dc97b : SUCCESS in 1h 44m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2153252f3ac494c8ed1a26fc596f9af : SUCCESS in 53m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa360ff3d3b84f7cb5b79b9120b20738 : SUCCESS in 15m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c88acb6193e347dab9ba3f601a549afb : SUCCESS in 26m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef296c7d0d9a4b749a61d8821b332662 : SUCCESS in 22m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a968061efd40a9b4449a63228a523d : SUCCESS in 9m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1886f134eaf24ee5ab6db2523462f689 : SUCCESS in 1h 14m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d2e52962f6da4f39a417deeb22f59d10 : SUCCESS in 1h 03m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/34da1c64f93e48af9e7385513e984635 : SUCCESS in 1h 00m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0b8efa680431480981828550a8fa178c : SUCCESS in 2h 25m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f97bc9febfc4c928b8a2c0ef5ff4f1d : SUCCESS in 1h 01m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/75a1cd5405af49cb8afcc6dffb99ed84 : SUCCESS in 1h 11m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/08506905f9a742a292efca2fb05324e8 : SUCCESS in 1h 43m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e9e9f168826946d5a8f02aaf9a4e2506 : SUCCESS in 1h 42m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d380847e27254c14aea98cb83b6f598e : SUCCESS in 1h 04m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8ce08dc80d9841faa66492a9b303c1ee : SUCCESS in 1h 16m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0ecd412cf8fb43ff83348a417a4a9388 : SUCCESS in 1h 17m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d03fe43e28a944f89e430eb30ceace4b : SUCCESS in 1h 29m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2cc9d8ae25375d93936f524d5fac08e31e82f6e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-05 13:48:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"11b25948ef6fce80ecbb8a494003ba72b6517fac","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-05 13:49:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e15f5a2e76f199145d6c208ae1bd127ea3f91780","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-05 13:57:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6aa2a10c3c6e6e97bbf04f08e6baf3e53736ad74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 14:21:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a7a3979f5d14fe6b79f110ed60a7730\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a73297bcf274433addca0d95bb2b5a7 : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb22345959264d0da2aba65cebdea5d2 : SUCCESS in 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50405fe8c4ae90b0e1898ab39cf192758ccca615","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-05 14:27:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"928a63edc114839e5eabe5eb45d210e8b289a9df","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-05 15:29:04.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/72/943372/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b748ab3de0c212c02ac12aa964a3e98008160d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 16:29:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c55cfe1bc504132ae61dc5a3ef603dd\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1dda939ac72f4fab8e8547b3f48dbeed : SUCCESS in 1h 39m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/176e88e1e74041b29a58c6c4ef0e0b1c : SUCCESS in 2h 33m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/654f1391f58f40588e13a03376c93df1 : SUCCESS in 50m 25s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2f2a9389a0448f8848d0a35ba5e7166 : SUCCESS in 1h 44m 12s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f960230c6e374668aa002b5b315c825a : FAILURE in 2h 34m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7baca071450b4f30bbc4864141e3a5f9 : SUCCESS in 30m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df03b3b1021c43f8990f4b468721717f : SUCCESS in 12m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dc548e1c87f4a4ca273674b635d0fca : SUCCESS in 21m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fcb62ebe61b54d92a62fd1fa2e38bc8b : SUCCESS in 21m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9348b86725d54d6687861254ce66f6e1 : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/110f2648dde741e380a0796b8464a3b6 : SUCCESS in 11m 59s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1079f11e970b4e2c9c5ce4a8fafcb87d : SUCCESS in 50m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5bcaf77f84ff4e74a57f5811b1f8b3c3 : SUCCESS in 1h 10m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/796e7864c4394d1bbf5cf7401005ef42 : SUCCESS in 1h 10m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/dcd06ccbdb4a44af84e41f1412b6a79f : FAILURE in 2h 04m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ec708e0d4184e4d91ff0235b30a2816 : SUCCESS in 45m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab483ca0b5b2471794d3feaaa249ac10 : SUCCESS in 1h 23m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2bc5b3c673284e80ba3249596e53231a : SUCCESS in 1h 56m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/968275597af946b2bf3ddc56290ef0b1 : SUCCESS in 1h 43m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/313f033dfb074266b71302c2b39cb730 : SUCCESS in 1h 14m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fbe197e3b6454d4295518ae0261f0651 : SUCCESS in 1h 07m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a9ecd9b2a9944a588308f0fe38c783f9 : SUCCESS in 1h 14m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/205c320e7cd64d5bbd3c2e7f16f8e062 : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":2},{"id":"205ea590f88823f724b4a967673d0e59529a0497","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-06 08:15:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"451204a7a0db5ff1633c97fc8bc09ffc61688db6","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-08 20:27:55.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9e52e6ff2415e152791005f354f30f7519a6f279","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-24 09:31:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"043d96d6247afa8eae7cac2bbf3732af0690c39e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-06-04 07:49:11.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2},{"id":"a6835718bd64c6e47796e2bcbbb53a9b49279aac","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-09 14:15:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"292f865ea751b4dba9c6553a280a8694cbdc35e1","revisions":{"0fb1a16206e7e604fff3996d88ea2eab59d100e9":{"kind":"REWORK","_number":1,"created":"2025-03-05 10:26:15.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/72/943372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/943372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/943372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/943372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/943372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/943372/1"}}},"commit":{"parents":[{"commit":"4cae83bcfd76d60d8eaa3308887e163d4615e5f5","subject":"Merge \"agent/linux: remove eventlet usage from ip_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cae83bcfd76d60d8eaa3308887e163d4615e5f5"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-28 08:48:54.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-03-05 10:25:15.000000000","tz":0},"subject":"Introduce noop OVN Router scheduler","message":"Introduce noop OVN Router scheduler\n\nHistorically neutron implements different schedulers for L3.\nUnfortunetly configuration based on OVN ``gateway_chassis`` is\nnot self-suffucient and requires external process to monitor\nchassis availability (which in OpenStack case is Neutron).\n\nOVN implemented Ha Chassis Groups which is recommended way [0]\nto configure HA for LRP and uses built in mechanism to provide\nHA capabilities.\n\nThis patch introduce ``noop`` sheduler which does nothing\nfrom scheduling standpint. The ``ha_chassis_group`` is\nconfigured for all LRPs in this case.\nAlso ``ha_chassis_group`` sheduling helps to fix various\nbugs related to Baremetal and SRIOV ports.\n\nCo-Authored-By: \u003cAnton Arefiev\u003e aarefiev@mirantis.com\n\n[0] https://man7.org/linux/man-pages/man5/ovn-nb.5.html\n\nRelated-Prod: PRODX-50467\n\nCloses-Bug: #2092271\nCloses-Bug: #2052468\nRelated-Bug: #1875852\n\nChange-Id: I95b0be8383586a788e1259574c18c14019b73e28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fb1a16206e7e604fff3996d88ea2eab59d100e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fb1a16206e7e604fff3996d88ea2eab59d100e9"}]},"branch":"refs/heads/master"},"292f865ea751b4dba9c6553a280a8694cbdc35e1":{"kind":"REWORK","_number":2,"created":"2025-03-05 13:48:46.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/72/943372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/943372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/943372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/943372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/943372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/943372/2"}}},"commit":{"parents":[{"commit":"4cae83bcfd76d60d8eaa3308887e163d4615e5f5","subject":"Merge \"agent/linux: remove eventlet usage from ip_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cae83bcfd76d60d8eaa3308887e163d4615e5f5"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-28 08:48:54.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-03-05 13:48:14.000000000","tz":120},"subject":"Introduce noop OVN Router scheduler","message":"Introduce noop OVN Router scheduler\n\nOVN implemented Ha Chassis Groups which is recommended way [0]\nto configure HA for LRP and uses built in mechanism to provide\nHA capabilities.\n\nThis patch introduce ``noop`` sheduler which does nothing\nfrom scheduling standpint. The ``ha_chassis_group`` is\nconfigured for all LRPs in this case.\nAlso ``ha_chassis_group`` sheduling helps to fix various\nbugs related to Baremetal and SRIOV ports.\n\nThis is first patch in a serie and the end goal is to remove\nscheduling based on ``gateway_chasssis` completely.\nWhich will be implemented in next patches.\n\nCo-Authored-By: \u003cAnton Arefiev\u003e aarefiev@mirantis.com\n\n[0] https://man7.org/linux/man-pages/man5/ovn-nb.5.html\n\nRelated-Prod: PRODX-50467\n\nRelated-Bug: #2092271\nRelated-Bug: #2052468\nRelated-Bug: #1875852\n\nChange-Id: I95b0be8383586a788e1259574c18c14019b73e28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/292f865ea751b4dba9c6553a280a8694cbdc35e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/292f865ea751b4dba9c6553a280a8694cbdc35e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
