)]}'
{"id":"openstack%2Fneutron~943373","triplet_id":"openstack%2Fneutron~master~I21244595f501e35919a97fd048b643f951b18500","project":"openstack/neutron","branch":"master","topic":"bug/2100853","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-07 09:19:06.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-03-07 13:57:46.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-03-06 10:09:46.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I21244595f501e35919a97fd048b643f951b18500","subject":"[QoS] QoS rule check also considers the direction","status":"MERGED","created":"2025-03-05 10:31:44.000000000","updated":"2025-03-07 13:58:55.000000000","submitted":"2025-03-07 13:57:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943373-bug/2100853","meta_rev_id":"246e716286d9c97542e87bd15a2a55a0d7042404","_number":943373,"virtual_id_number":943373,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2025-03-06 10:51:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-07 13:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-03-06 16:21:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-06 12:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2025-03-07 08:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-03-06 15:33:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-06 10:09:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-03-06 15:33:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 11:12:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-05 11:38:09.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-05 13:10:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-05 14:40:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-03-05 17:01:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-03-06 07:36:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-06 07:36:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-03-06 12:13:20.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-03-06 15:33:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-03-06 16:21:26.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-03-07 08:11:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"81e2067f17019276cfca8cac23bfafbd46ac7556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-05 10:31:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a8157c4fdafe658b1a680dde581d0caa26af49e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 11:12:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7b2635008e643ab8185c6445e73f799\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e79f41c425d54a98829ddbfbb0a348ff : FAILURE in 26m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b008a3320b8c4753a4e52985f7b9be63 : FAILURE in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"021e36740e2c9a0823cb1dce0614d281fb4c55f8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-05 11:38:09.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/73/943373/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ac982faf347af8062e76b62f2af09b2b3b716a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 13:10:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1b8915ec5c34bfc8e54b4dbd0d95284\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/59f358cba4e24176a94af315cfbe454d : FAILURE in 1h 37m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1b0ef54fe0a349b6a2b10d4be81ef03c : FAILURE in 2h 20m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7a70dba983d7479b8188e21edbe271b4 : SUCCESS in 32m 09s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1299d8254044f9eb234781a4dd1e828 : FAILURE in 1h 30m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f15128627af24391a35fb5a833253e4b : FAILURE in 1h 43m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cf1cb01b393495e8e128a70ea7a60d0 : FAILURE in 56m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b81b30cc54e4d7c97bbc71e6e644d26 : FAILURE in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1292f28dc94b4e10a51418a35f4bb784 : FAILURE in 24m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5442ae9b8084a1fa7360b8a257d3e86 : FAILURE in 25m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af634f51498f4c05a52d3bd8f1ae432d : SUCCESS in 6m 12s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bfb8a4763a0f4f65bc6fbdd6cc655ee1 : SUCCESS in 59m 31s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e3a94a85537e4294b1ac467bfea374d8 : SUCCESS in 1h 09m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/301c67b038de434884ecb039beda8c7e : FAILURE in 1h 08m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/205767ae66964493a1da8cce8856645b : FAILURE in 1h 46m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/38e74fc233ef4252b29d152c487db3ff : SUCCESS in 49m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/de5fdf0ad08345d98c1025429a14a210 : SUCCESS in 1h 21m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/15715cfe0ba84e388c86cd47d125e040 : FAILURE in 1h 08m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d076a335d9894e20a781076bf6fc4b0c : FAILURE in 1h 12m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6e0c72067fe04e1888634095e088982b : SUCCESS in 1h 12m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/27e21b02c4e14ce995c3d2811f3b780d : SUCCESS in 1h 12m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/79094ac3c1e64a94b3e055730b443003 : SUCCESS in 57m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6a41f9623e914e8f9327c303d09e3934 : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8d0b225826c68e82a110ef250612923f33514b17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-06 06:51:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"46dba36a3332dab585e65cb0527d1198f8034e87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 07:15:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d0b98b99fa14c1081aaacc7059e9eb1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ad74f9121ab494b8e46bb4b70dbe81e : SUCCESS in 22m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a118406545548108e46f5cdd15c95dc : SUCCESS in 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b0d612a223a0f042a232ed391b34d95f2bf0ef3f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-06 07:31:17.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/73/943373/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8a663b75a3be3f04bcc415f2d18c6641d103696d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-06 07:36:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"da165be5781968bac78b79fc54ce6afb377cfb81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 09:27:08.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/669ea9ab16934e02a25deab78b6be450\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e7b48a986a0f4593b807a16020d32275 : SUCCESS in 1h 49m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/86402319c4814fe5b5e8bef90c5cecf7 : SUCCESS in 2h 33m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/363cf3a61d374d65bd7b7212875b6d01 : SUCCESS in 35m 06s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/323d9286f81c4ac18cf7b4df8e72949e : SUCCESS in 1h 46m 46s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fb5a65ee4124673a6672ac530676299 : SUCCESS in 1h 36m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0a63d5a594648dcbd8a8933569d4d59 : SUCCESS in 53m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93060ed7bb3f4814857618918c591e2b : FAILURE in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d06dda81f694bc78de9a98151ff4bb5 : SUCCESS in 18m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c16861698db348bebdd268d27690c65e : SUCCESS in 16m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13491aafa5624a28bd46117b4d942e02 : SUCCESS in 8m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/db8e017268cc42b8b74e9023a064f13b : SUCCESS in 49m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a99c5c6f86ba41038ec5c8eb138ec454 : SUCCESS in 55m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f4715628d59743adbbdbb682c10b073d : SUCCESS in 1h 29m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/39bc8f7ced8f4884928886e482a3a692 : SUCCESS in 1h 50m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/40d97ea6226b48f7b9cad7df3c27b8fc : SUCCESS in 57m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f48a702793034f41bdfe7ee9634213ca : SUCCESS in 1h 04m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6f3af6f5fdeb4ab4938593e27bea789d : SUCCESS in 1h 40m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/06da56dbd0884e4bad72157d946b445c : FAILURE in 1h 07m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/df66da5901a4422b963df6bac597dbcb : SUCCESS in 59m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fee08f0b6fa746faa1e7d2538395f729 : SUCCESS in 1h 09m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/597a4b55f6644f809735abdf3096816f : SUCCESS in 1h 09m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb1f4373a50041e986264c028e6389a6 : SUCCESS in 1h 35m 45s","accounts_in_message":[],"_revision_number":2},{"id":"224a93073e847240e27728383dbbdfe12742ffea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-06 09:50:05.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"819680ff575655f35053343c35de462dc27b4758","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-06 10:09:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"85a632c11db71d38dc0e90a25294f1a3456962db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 10:26:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d17ed6158f7144e791594caf4f9910a8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51807253656647b2a44f01590119cbfb : SUCCESS in 28m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e665260fd7549a197dd23f8cde76b46 : SUCCESS in 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9276ca2b53bbf2fbef3e3ce270969d8b52d270f4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-06 10:51:49.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/73/943373/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"654f380e8c05f78f3398a76294965d1ce5d85caa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:07:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a74da03802504187b948a235fd20dd40\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0f1a1177802b4dbfbc2598708aa9bc6b : SUCCESS in 1h 53m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95ff2fd3af83402e969bad463b3be8a6 : SUCCESS in 2h 14m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8cc21e87464d40998e5deab805664e9d : SUCCESS in 40m 48s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f651c131004440aa044e776f0d060d0 : SUCCESS in 1h 48m 50s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7fe4b653c74400493fae9608eaefc1a : SUCCESS in 2h 03m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5016907e29504f5e94c1b706637e2116 : SUCCESS in 57m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a62460eff1d342188aaa3c06bae53e8b : SUCCESS in 10m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa66816a7add45888f1322744031f5dc : SUCCESS in 20m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41decd703c464091a2fa8ba68ef8020f : SUCCESS in 26m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9af5d7e135d42b2819a518a9081fdee : SUCCESS in 11m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/592d3ffa8234403f853425cc77899584 : SUCCESS in 1h 12m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/77bf32d07a5f4e6ca9b9c512b70ab4b1 : SUCCESS in 1h 15m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/342cd1ce1dfb4ecfa8deea484fb81dae : SUCCESS in 1h 13m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/785c6c384e5b4b769e469129a6478b47 : SUCCESS in 2h 03m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d08cc4a54a6b4f389919f4dabb99dd19 : SUCCESS in 1h 11m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5fa70e95611d4a83997171acfa5c88ad : SUCCESS in 1h 16m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a059061499d24d65938c2c27058f4b90 : SUCCESS in 1h 52m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9dade6ce5ec24dca997fd5856a792299 : SUCCESS in 1h 19m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a9aedd18dce5438394f38bb95a94a0ec : SUCCESS in 52m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8cd640ee89b84b979ceab5ba47859350 : SUCCESS in 1h 02m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/db6e5ca3019b4ae980fd1328095bf0ff : SUCCESS in 54m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a448dcabc9784cc99cc0b5bfafd8262b : SUCCESS in 1h 15m 10s","accounts_in_message":[],"_revision_number":3},{"id":"67e6c783798059ebd4b5f8a1d2058769b65a42cd","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-03-06 12:13:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2e43d77931af07e318b83e34d122648c7c756fd2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-06 15:33:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bc2c0ebd7e315202fed4e3843c27369aba909cc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 15:33:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"446019c8bbadea58292b54074ab1f081257bb909","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-03-06 16:21:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6a31aae4d6f4363369a6b0bef107ed4ce8872bb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 17:41:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eacec19065b54df4be7a7cff63f55b73\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2e5b1b43cf684b11be92d211985c6747 : SUCCESS in 1h 31m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f730f4b29c384d13b4384ebd53bb811b : SUCCESS in 2h 01m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62901e953424416793562dd1574ad402 : SUCCESS in 11m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a99d69d6186445cf8ac959e3ee8d413c : SUCCESS in 21m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd688b39ae91485fb67f09b0d184729a : SUCCESS in 20m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68bd14eeeb748388a08530d95525180 : SUCCESS in 9m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/70f7c27ac35142be85f2d0051543a189 : SUCCESS in 1h 21m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d4120096d78749aa8c6d53b37591d8e4 : SUCCESS in 2h 02m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fb07cecac7804d60bb8d66e26b524c1a : FAILURE in 1h 14m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/873cef28717d45f38f1ef4ff01383815 : SUCCESS in 1h 15m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c3db9120eb54bbeb346ae35bd6e0936 : SUCCESS in 57m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fbdc301e80e8402e9c46da9bd6d39e5e : SUCCESS in 1h 08m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/032c9490e4824afbbcaad7d5dc447668 : SUCCESS in 1h 44m 20s","accounts_in_message":[],"_revision_number":3},{"id":"78d192e1d1eea1e07fa65123b467605320b7b7a5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-06 21:21:14.000000000","message":"Patch Set 3:\n\nrecheck functional","accounts_in_message":[],"_revision_number":3},{"id":"c2cbeca87d526af13520f22de595fd99d2466f42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 21:45:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4865f0a0187d403d9f4962f681090126\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15888e0f4c914ac7a4cf112df6d25707 : SUCCESS in 22m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1528b8ed6a0c4476853bd0a035260bd3 : SUCCESS in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8e892edbf4dc9af37093be7eb5b4ba9b41956a6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-06 22:01:29.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/73/943373/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"861e5c0057db9f1731abcaae27248acbe1e34292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 23:44:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/acdf247810e3432ab663d80350b19de4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f4430aacbd1843dbaf569a1bf17a0eca : SUCCESS in 1h 50m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/657fedc9fb8b4ffd9e17e75eb501285d : SUCCESS in 1h 54m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e4bc29f96877457ea4cff8227d1d0bb5 : SUCCESS in 35m 59s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eadb61efa6e444809648aa931af5db91 : FAILURE in 26m 09s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a90cde6cd6e458eb53a5d5dd2a56d65 : FAILURE in 1h 55m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8f87a07861c41c882aedca78f7d282d : SUCCESS in 49m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05acbaac00f4cf2b2cadd4b9c2e00ac : SUCCESS in 11m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f883151ce84ab083f838f0b5a79be9 : SUCCESS in 20m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba041c0409904e1695d955e408877f37 : SUCCESS in 23m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b85a490774db46388764a7fe5240dbc4 : SUCCESS in 12m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1e347176c79a4790aa88bfc7d95aa26a : SUCCESS in 1h 02m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7c75cdff0dbe485c82662dc467200b17 : SUCCESS in 1h 27m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e99beee82d5c4069bc68fe5ca0ef8844 : FAILURE in 31m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7f88e529d3e04d649915d8594172cb3f : SUCCESS in 2h 11m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/850ae5a2035340f2bef75ff9a12d696a : SUCCESS in 1h 07m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/78bc2f53559b4c0289cae2e6cdb9a128 : FAILURE in 55m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4b18b838675449cfaaff17577c985607 : SUCCESS in 1h 42m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2dde3522cab54354a9dc06209825ec84 : SUCCESS in 1h 33m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/50fe090e58de418da62a9e6936e3d18b : SUCCESS in 1h 07m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/760b7a66846e40cd81f4716008ee6eac : SUCCESS in 1h 05m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d740280e7fec47ebabc14523d4d2737d : SUCCESS in 1h 13m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/69e2634539714ee8b1d7481f16978f10 : SUCCESS in 1h 29m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8145c9b0e42e0606ef4f7b2f0c040f63a16cbc3c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-06 23:51:17.000000000","message":"Patch Set 3:\n\nrecheck functional","accounts_in_message":[],"_revision_number":3},{"id":"1da963b5184012154ce50d4f34b8b80164039e39","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-07 00:31:49.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/73/943373/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f5c1c4a94f9a47b911a4d314d2f51bdc3bcc3f9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 00:38:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dc09da7f2b94a50a1b839740ba18951\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7c2ea22597b4217bf74fb57950d7329 : SUCCESS in 32m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6bee31f5f54c40dc9432792c118653d4 : SUCCESS in 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"062259ea41c31c1948d15b23717212bbf19b473e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 02:51:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8707aa9aa7644f2cb81554c9e44c75b1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c29ccb82e2a44971a9b972c8ec040e03 : SUCCESS in 1h 08m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1bbae4e4df0446c79f338cbe3a1de83a : SUCCESS in 2h 17m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/53b594afe85f406db9b8160c421232c7 : SUCCESS in 43m 20s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05d5c188daad4a20b0720e957b32100a : SUCCESS in 1h 52m 08s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eefac627aebe4927a8429c2590118da5 : FAILURE in 2h 57m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1caed6eb8f154c448fa7326804d2f6e2 : SUCCESS in 50m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70202d49d4b74a0285209853df869603 : SUCCESS in 11m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30f443044e944202b6a5b090cddbf5b3 : SUCCESS in 16m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34b61669695d4711aa2865f7de32c8fe : SUCCESS in 14m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/369b7bf730cf4f93bdaf9997d0a27ae7 : SUCCESS in 9m 56s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4e63b3dc743b4a86bfc06189630ae5a2 : SUCCESS in 41m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b5917e3b8db349c4a57424d5a17e1f55 : SUCCESS in 58m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f704e17173c14cf08635c325ce89479e : SUCCESS in 59m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/235c1b8e753c43c3965b1eea767c6601 : FAILURE in 1h 56m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/14790dc181ea4257a809faf37ecfe457 : SUCCESS in 46m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/400c2257de9c49f1b5bf92de5700e5ea : SUCCESS in 51m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e9c8235eafa4f7ca155c1154b650417 : SUCCESS in 1h 26m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3d3549a8bf7242b2b9338246578fba9f : SUCCESS in 1h 38m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6eeac5549cea4629be7b4b4436d0d30e : SUCCESS in 49m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/748f64bb7ea14a9d9298f669b8f63454 : SUCCESS in 1h 04m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/08f922392251407095365b1d731f7016 : SUCCESS in 1h 05m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e5d6dff084674004b6d1df041611a3f1 : SUCCESS in 1h 19m 01s","accounts_in_message":[],"_revision_number":3},{"id":"bbfd6df216043512f54ee1cf6b60b3490d9af344","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-07 03:21:22.000000000","message":"Patch Set 3:\n\nrecheck functional","accounts_in_message":[],"_revision_number":3},{"id":"8d6b2e31aa7b0e4d3ec892446ddfba397f93595a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 03:48:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a47025897ff545f7b2665c92e0dc56e5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/444dd5ad89b245bf829c034c6436c7cd : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6ef2fbd223574c1eb93911bda7f536c5 : SUCCESS in 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"91a2f3fdf4000929436382ac459830ae3d60ed34","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-07 04:01:51.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/73/943373/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"eb06979dc3876a45bbeaa6a8569885d5d115214c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 05:55:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5801d5e47e548f9a8317ced268f4ae4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/293e92a912374cb9b8de066ba629d606 : SUCCESS in 1h 51m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c663eaa09b2f41488d2a43c54908f422 : SUCCESS in 2h 30m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/faef247fe59940c88431feb80db97a2b : SUCCESS in 36m 05s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9fd85b3a8774867a2464cd0f391814b : SUCCESS in 1h 36m 50s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4e7609241014736b33d4122477c2164 : FAILURE in 1h 49m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63e382f5c86b42769b50fd358362a540 : SUCCESS in 49m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57a0ffbfba1f4f97bcbba1cd7d5b835d : SUCCESS in 12m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3005b880a95e407e989830db9d872eb7 : SUCCESS in 22m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4cf0a249a9824e68b92372e55c485ddb : SUCCESS in 22m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f367cc9288224923aba5543f42543f90 : SUCCESS in 9m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a10d436c65254fc99b4f3e723e4498bf : SUCCESS in 1h 14m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bd8b4accd60f43d7bc27c57689f7394e : SUCCESS in 1h 10m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7fcd8cee72da4be6add299abdfe6596a : SUCCESS in 1h 16m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/eba3fb1c7e1748b2b26399dba3fe0301 : SUCCESS in 2h 08m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6cb675d2e5ae42d6a7ef63d6ce989b86 : SUCCESS in 1h 01m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc3c69a5b65641c3b1c4bdfccdf8758d : SUCCESS in 59m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2b08cd081e7e4b84a150318cccec20d9 : SUCCESS in 1h 24m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/550caa2d59484cfba33bf136f036213a : SUCCESS in 1h 15m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/edde7cd4c7bb4f089693061020099bd1 : SUCCESS in 48m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/eee2e17e12db46b2ac28d96878df03e2 : SUCCESS in 52m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/548e79fe58fd46238d8b248d9c0b5a92 : SUCCESS in 54m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7ff6ae4e0cd345c1959a0b917ba98938 : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":3},{"id":"be98b9bd9ba8d0b4024eb88378ed7b5a71d18943","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-07 06:21:26.000000000","message":"Patch Set 3:\n\nrecheck functional","accounts_in_message":[],"_revision_number":3},{"id":"26faf782fb3f70c77a8b0f8d6a2c94919b820e04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 06:58:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29246e58d64b4ee78932dc9286d48751\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d36b2d996894af88bf37b57e35e3a32 : SUCCESS in 35m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40dc315d770340288177b30c9c261dbd : SUCCESS in 19m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ee9de31d0d037e6366f9463794a1ef386b186fe6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-07 07:02:02.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/73/943373/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"af97e9956ece7f930870559a452aa8a04e901515","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-03-07 08:11:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fd82227b2b344264e45965be9c12b5ac43c62ad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 09:18:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d96087d115694059af63a19cc9770a70\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a8480c7e30a142fb8e8f32174a78351b : SUCCESS in 1h 33m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a3b67625e0db41a2b6c5a12120e367a7 : SUCCESS in 2h 35m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5272b6d6ac047d496085b712af32589 : SUCCESS in 42m 50s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e479155edfe9420398baa444d9c0a86e : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1255f6d7eaee478483df94e95e1c4fe1 : SUCCESS in 1h 14m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0b4b51e883c4287a1bfbdcf0b392d2c : SUCCESS in 36m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d85927204f42baad1ac6a230d4cbca : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a289987218c4652b535bccad46c325b : SUCCESS in 14m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/893d94ffb4f948d18e1c8fa675e733d5 : SUCCESS in 13m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45296f121f348f8a0315e9073e0de33 : SUCCESS in 5m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f00361f20f5440708ece6f4cc74c1193 : SUCCESS in 26m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c34580660915493bbdf582406db87377 : SUCCESS in 37m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c82c8bf2b96542a99385fee6010937e4 : SUCCESS in 52m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2821696fad75473b9dab9c67a953b4ca : SUCCESS in 2h 15m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d720741f6da84a1ca54056c05dd7b172 : SUCCESS in 31m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f83cf41d5d2b4f1499e0b2e15a471e29 : SUCCESS in 44m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a8e74e59dcb04bd5bc406eb89c329785 : SUCCESS in 1h 06m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e023ef19af9405ea22ebd18b72c4fd8 : FAILURE in 39m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b32eca05b86145d689a36a1e3cb312b1 : SUCCESS in 48m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/342cbb287d9045f497faf77609dc3cf6 : SUCCESS in 47m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/889e0323955144fea3ebefb7519f3b41 : SUCCESS in 55m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0e14e00afb264edc938a561afd5e124f : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":3},{"id":"71c0cc431cf58c9ac747cf83e92743f036cd7692","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-07 09:19:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d901c3f320fffab57f3723d41701063bd45792e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 09:49:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce13f22f25d4430bbd5edcb84d67972\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4f959a483b0474391a76fae2c46c2a1 : SUCCESS in 28m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c973ffcc371741db935dc7bb83f9d825 : SUCCESS in 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"db5b952cec03119010ac41d05d0cef54c306c25b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-07 10:25:16.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/73/943373/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b46b2974c779b7f8f3d595ef4b0e02b90b100c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 11:23:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841875d391cd442eafc6a9d07b4d9a89\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4892e3184fb3418b82bf8eb9043fdff5 : SUCCESS in 1h 04m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3ff766e3c8034326aeecbd61ef32a22e : SUCCESS in 1h 28m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42aa4a9b00434f4f80bc23dea2b8f468 : SUCCESS in 37m 38s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2697aa1e388c4b70b4789a19a1e7f304 : SUCCESS in 1h 02m 17s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0f6b82cca924e0bb4e22ac6662a188e : SUCCESS in 2h 02m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b7e547bc3554f338347b95a5c614f56 : SUCCESS in 55m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c26d1588244f1c8d885dd8e69b5a4c : SUCCESS in 12m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f91d31f6d864bb2b1ed397943579a6a : SUCCESS in 23m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd1f7b29f37e4491bb1d933e3ee31c5e : SUCCESS in 20m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea62443ac004f9ea1757c42d7c30835 : SUCCESS in 4m 59s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/21dd8d9d683a40df9a089ecaeea1afab : SUCCESS in 33m 18s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d32ac6bcd0c2446d899a43fdb3a70ea8 : SUCCESS in 52m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fdd769c6ea1e41fb9c029634a3bf6631 : SUCCESS in 1h 15m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5221456b3dcb4b2ca56efd73a418209f : SUCCESS in 1h 19m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0bb05c7f7d148ba8c583adfc607c07b : SUCCESS in 1h 02m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a855d37b73946b2baae395789cfb6a1 : SUCCESS in 54m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6748641be04d4e188f3a9eebbc9d55b1 : SUCCESS in 1h 08m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c63bc21fff0421791d60ba3d97c4fec : SUCCESS in 1h 48m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/58ba96e4c04f45db9ccfd066b1e97183 : SUCCESS in 52m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5460062806a44223a3867d671cc82166 : SUCCESS in 49m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/44d291fa2ec548aaa6d9ad5e96eaec6a : SUCCESS in 1h 16m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d3f38e4c22074dc1a9ffb045e1a294c0 : SUCCESS in 1h 28m 00s","accounts_in_message":[],"_revision_number":3},{"id":"97e77089adf1965755e24bb05d599376bd4de4c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 11:24:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dd906c20dffe93f46b3a682c0dd3c4ba5d3d40bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 13:57:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfb7d772c093426a8c141e661388a9be\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/09e2610d97e94a2fbd7207d1dc167d10 : SUCCESS in 1h 17m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/be77a0d3bb9d42e6b58cbe74ae0e5a2a : SUCCESS in 1h 30m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53e7b5992b04975afefca2c0c548e69 : SUCCESS in 13m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/141754c8b5ee4970857d765aa1376ffc : SUCCESS in 22m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/989b2c55bc534c3e97238f6600bb48ea : SUCCESS in 19m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c60e4c698df4a82abf3e4efb11b1360 : SUCCESS in 9m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ac3af40a8ea84c67a90eece23f7a1bf5 : SUCCESS in 1h 20m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3ba884c44dbc43109873ffc0b84770c3 : SUCCESS in 2h 31m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8aca4ea816c54e7eaad3188ca818d3cb : SUCCESS in 1h 33m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/04c188b18219432abf7b5b86f1c3c1e1 : SUCCESS in 37m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4e19a0297a14883a0a097c58afe4792 : SUCCESS in 46m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ceeeb69789194252ac1b54cea654a046 : SUCCESS in 1h 02m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3c51973a6d9d4477a5950e784d510136 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":3},{"id":"eb828a49cf771dfbeb2b86f46254639769cc6074","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 13:57:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"246e716286d9c97542e87bd15a2a55a0d7042404","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 13:58:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79c03bf49cea438da228d3fef32e019e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d2bd91975646bbae356f9f0065294e : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b658c52d83fb144faee97007781fc6a16bfdb2fc","revisions":{"c7b489d1a307cdd2c9acc50a869dc8707472255e":{"kind":"REWORK","_number":1,"created":"2025-03-05 10:31:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/943373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/943373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/943373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/943373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/943373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/943373/1"}}},"commit":{"parents":[{"commit":"4cae83bcfd76d60d8eaa3308887e163d4615e5f5","subject":"Merge \"agent/linux: remove eventlet usage from ip_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cae83bcfd76d60d8eaa3308887e163d4615e5f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-03 11:24:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-03 11:24:51.000000000","tz":0},"subject":"[QoS] QoS rule check also considers the direction","message":"[QoS] QoS rule check also considers the direction\n\nThe method ``check_bandwidth_rule_conflict`` now takes into account the\nrules direction when checking the compatibility between bandwidth limit\nand minimum bandwidth rules.\n\nCloses-Bug: #2100853\nChange-Id: I21244595f501e35919a97fd048b643f951b18500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7b489d1a307cdd2c9acc50a869dc8707472255e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7b489d1a307cdd2c9acc50a869dc8707472255e"}]},"branch":"refs/heads/master"},"bcd6de0a7bf904c2855e1907406827ddf6928114":{"kind":"REWORK","_number":2,"created":"2025-03-06 06:51:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/943373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/943373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/943373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/943373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/943373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/943373/2"}}},"commit":{"parents":[{"commit":"4cae83bcfd76d60d8eaa3308887e163d4615e5f5","subject":"Merge \"agent/linux: remove eventlet usage from ip_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cae83bcfd76d60d8eaa3308887e163d4615e5f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-03 11:24:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-04 07:56:00.000000000","tz":0},"subject":"[QoS] QoS rule check also considers the direction","message":"[QoS] QoS rule check also considers the direction\n\nThe method ``check_bandwidth_rule_conflict`` now takes into account the\nrules direction when checking the compatibility between bandwidth limit\nand minimum bandwidth rules.\n\nCloses-Bug: #2100853\nChange-Id: I21244595f501e35919a97fd048b643f951b18500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcd6de0a7bf904c2855e1907406827ddf6928114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcd6de0a7bf904c2855e1907406827ddf6928114"}]},"branch":"refs/heads/master"},"b658c52d83fb144faee97007781fc6a16bfdb2fc":{"kind":"REWORK","_number":3,"created":"2025-03-06 09:50:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/943373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/943373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/943373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/943373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/943373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/943373/3"}}},"commit":{"parents":[{"commit":"4cae83bcfd76d60d8eaa3308887e163d4615e5f5","subject":"Merge \"agent/linux: remove eventlet usage from ip_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cae83bcfd76d60d8eaa3308887e163d4615e5f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-03 11:24:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-03-06 09:50:04.000000000","tz":0},"subject":"[QoS] QoS rule check also considers the direction","message":"[QoS] QoS rule check also considers the direction\n\nThe method ``check_bandwidth_rule_conflict`` now takes into account the\nrules direction when checking the compatibility between bandwidth limit\nand minimum bandwidth rules.\n\nCloses-Bug: #2100853\nChange-Id: I21244595f501e35919a97fd048b643f951b18500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b658c52d83fb144faee97007781fc6a16bfdb2fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b658c52d83fb144faee97007781fc6a16bfdb2fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
