)]}'
{"id":"openstack%2Fneutron~943418","triplet_id":"openstack%2Fneutron~stable%2F2024.1~I2b13e120b3c5961f78c169f6776e0dfb958fa83e","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2098996","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-10 12:35:17.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"19234":{"account":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"last_update":"2025-03-10 15:02:41.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2025-03-10 15:02:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2b13e120b3c5961f78c169f6776e0dfb958fa83e","subject":"[ovn][trivial] Add \u0027empty_string_filtering\u0027 extension to OVN","status":"MERGED","created":"2025-03-05 17:27:00.000000000","updated":"2025-03-10 15:04:00.000000000","submitted":"2025-03-10 15:02:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943418-bug/2098996","cherry_pick_of_change":943417,"cherry_pick_of_patch_set":1,"meta_rev_id":"18b711d956d5d97700224f48052073960656c81c","_number":943418,"virtual_id_number":943418,"owner":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-10 15:02:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-03-10 12:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-03-10 07:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2025-03-06 08:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-03-10 12:35:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 18:23:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-05 19:32:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-06 08:43:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-03-10 07:54:13.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-03-10 07:54:13.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-03-10 07:54:13.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"b750dd1f5fffdba67330bd5441e4377386bf1e1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2025-03-05 17:27:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"987623813282a606c555c4dedf8a1462d942d6b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 18:23:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bddd7247919141e7a7af22a1ef194b4b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4be38b29c0d4487783f7f3b86ba3d22b : SUCCESS in 51m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c36be03fd02c44a5af914a0e53c5ebe5 : SUCCESS in 54m 02s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a0932e237f04e22bc4376b944966714 : SUCCESS in 45m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0115336a60e3fd38ecb72545dff28f15746bf8fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 19:32:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c77ea847d0604c3eae139914ba101413\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/5b49d70132f342c5900d307ff9ae20d8 : SUCCESS in 1h 25m 47s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/000ce1b527d842de9ce4b30b8c0b26ee : SUCCESS in 31m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/150cb68726a44585b6deb057c0679f0a : SUCCESS in 43m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd68b1c19a4846d48d039a273f70a953 : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76ad0676f2dc4013bbbdd402f1893e4c : SUCCESS in 26m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6c6d87558044b3ba65fb571a11ca274 : SUCCESS in 23m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f973b4248ca485db5429fc6465220cf : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fbedf4588354d10857172e9c493720f : SUCCESS in 49m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec5e4172b7d24d6f81b28f120ac8cc91 : SUCCESS in 1h 42m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b959c638e4594c51b96d2dc4c10e5379 : SUCCESS in 52m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8784268e7bf741d7ab4cc945022890c0 : SUCCESS in 1h 57m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9fcb67cc6f2145d996a50ba4c7c35ffb : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":1},{"id":"900b3381df5b2b9b8b676eabe7045e4b224f8b69","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2025-03-06 08:07:51.000000000","message":"Topic bug/2098996-stable/2024.2-stable/2024.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"ac41eae0ca10e5f9f8d5b459e160b9000c63b703","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2025-03-06 08:08:02.000000000","message":"Topic set to bug/2098996","accounts_in_message":[],"_revision_number":1},{"id":"c64977f87b36e145748d54bddc631c82a6b269b4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-06 08:43:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"74ff96a65b251f2edd9e2dd7371848f97e8cf780","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-03-10 07:54:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a8788d185d3a59549349bce1cd83bc884b29c27f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-10 12:35:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd7e1bbe5323ae067ece1ce198351f445f05f118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 12:35:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e38c4f0d2fb27afd088978438f410b461a687b56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 15:02:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/402884cd8f2c4a41af4a4eff64be1d9d\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/c8c9d5e7b276482bb1349cd54411c39d : SUCCESS in 2h 01m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28dc67417ed04a66843e4420ef5613a4 : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2338471cc444383a2a14f439e1eb6d9 : SUCCESS in 17m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f50131bd980c41f18c15cdb8fb7216b6 : SUCCESS in 23m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782e55e65a4e4e319d6f21e8a60e7855 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3757311e3b724247972e3c05963e8f78 : SUCCESS in 46m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c5bf2e6488c47a48fb78d50575f578d : SUCCESS in 2h 16m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6826a462b66c46e28333b2eefb5f61ec : SUCCESS in 36m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb1aed941f41405aa5043fbc41d943b2 : SUCCESS in 1h 17m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cea7bc9096c04c66c77407cef1f2dbea63713f8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 15:02:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"18b711d956d5d97700224f48052073960656c81c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 15:04:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88a545ce025b41e888766bb652f9dcfb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2709c4c07c6b46b794e35331c284ce83 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78a996d5c1a50514e3160ac6aeaec7162e90217d","revisions":{"78a996d5c1a50514e3160ac6aeaec7162e90217d":{"kind":"REWORK","_number":1,"created":"2025-03-05 17:27:00.000000000","uploader":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"ref":"refs/changes/18/943418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/943418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/943418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/943418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/943418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/943418/1"}}},"commit":{"parents":[{"commit":"a2a5124e5b77ceb256e3c96bd6b4faa81f7ad20f","subject":"Merge \"Support nested SNAT for ml2/ovn\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2a5124e5b77ceb256e3c96bd6b4faa81f7ad20f"}]}],"author":{"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","date":"2025-02-20 17:45:28.000000000","tz":60},"committer":{"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","date":"2025-03-05 17:27:00.000000000","tz":0},"subject":"[ovn][trivial] Add \u0027empty_string_filtering\u0027 extension to OVN","message":"[ovn][trivial] Add \u0027empty_string_filtering\u0027 extension to OVN\n\nCloses-bug: #2098996\nChange-Id: I2b13e120b3c5961f78c169f6776e0dfb958fa83e\n(cherry picked from commit 3fbb88b47218d1fb18664ef72c0400ee796a6608)\n(cherry picked from commit 649de83b16b5c7963166bd1e2cc3a50bb3a18e72)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78a996d5c1a50514e3160ac6aeaec7162e90217d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78a996d5c1a50514e3160ac6aeaec7162e90217d"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
