)]}'
{"id":"openstack%2Fneutron~943632","triplet_id":"openstack%2Fneutron~unmaintained%2Fyoga~I83b53a07362861da98b8361dafd95e94e5048322","project":"openstack/neutron","branch":"unmaintained/yoga","topic":"bug/2083609-unmaintained/2023.1-unmaintained/zed-unmaintained/yoga","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-03-06 19:05:14.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"36474":{"account":{"_account_id":36474,"name":"Mark Loza","email":"mloza@tucows.com"},"last_update":"2025-03-07 18:13:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83b53a07362861da98b8361dafd95e94e5048322","subject":"Revert \"[HA] Do not add initial state change delay in HA router\"","status":"MERGED","created":"2025-03-06 18:36:34.000000000","updated":"2025-03-07 18:16:19.000000000","submitted":"2025-03-07 18:13:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943632-bug/2083609-unmaintained/2023.1-unmaintained/zed-unmaintained/yoga","cherry_pick_of_change":943631,"cherry_pick_of_patch_set":1,"meta_rev_id":"6877369ed4752b704e0c03cbecc3b9631bb92ad7","_number":943632,"virtual_id_number":943632,"owner":{"_account_id":36474,"name":"Mark Loza","email":"mloza@tucows.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-07 18:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-03-07 16:35:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2025-03-06 20:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-03-07 16:35:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-06 18:36:34.000000000","updated_by":{"_account_id":36474,"name":"Mark Loza","email":"mloza@tucows.com"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2025-03-06 18:36:55.000000000","updated_by":{"_account_id":36474,"name":"Mark Loza","email":"mloza@tucows.com"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-03-06 19:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-06 20:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-06 20:52:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"af7d92aa08935df8be797e992b7c50f53c2131cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36474,"name":"Mark Loza","email":"mloza@tucows.com"},"date":"2025-03-06 18:36:34.000000000","message":"Patch Set 1: Cherry Picked from branch unmaintained/zed.","accounts_in_message":[],"_revision_number":1},{"id":"9074ebad2753b769c9e7c9af98c8aaf2b09ebce7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 19:03:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbefc5ef00244e519ca40261faf7c73a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cadd1e08f5544c45a77380eae98aa41a : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/849a0a7079a5437291097af2639c4f8c : SUCCESS in 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1be5029a98268edf393633dc6e30e62742bdc8ee","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-06 19:05:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c96462060eda7452e1d4a9d5e1aca2790d4b50ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 20:25:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c568c78f07ed461aa27d3bc4b10c26dd\n\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/fe0cba1a68644483ba2a113a25d41a5b : SUCCESS in 1h 06m 39s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/ac57f61fd04c42528b22aa3e1f492e22 : SUCCESS in 54m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/873ce0f440864f6bb529760f2f1ff444 : SUCCESS in 1h 13m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de6ef3af62654ae2b1ea85fb18182c63 : SUCCESS in 33m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b4db9185fe34dfc8539ac3459f36ef6 : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c51056ee4be4804b538c9085e7c0a69 : SUCCESS in 28m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbfeb1e1d1ed4bec95f323cf3f0df188 : SUCCESS in 20m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd848704cecc4466971b4ee098a60254 : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa4f411c4aba40b2aadf4641162a37fa : SUCCESS in 52m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/63923239a7ae4331907bb7205b65eeb5 : SUCCESS in 1h 43m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7feb9b766dfe48199c4607787fec0eea : SUCCESS in 1h 13m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0d0b4363042841b3af9636e644e789f2 : SUCCESS in 1h 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5c64ed1fec4bbfb38aa15dcde1433fa5528d6fde","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-06 20:52:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b22f2a154795787ad3b249b551058f32c4c84be6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-07 16:35:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74ff323083ec9ca2786dfd6c93cc7628a4e9b4c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 16:35:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afeba07cee8162aa0c197a50b7524f2cec59001c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 18:13:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cc65e7883ff44bb99c8d2c7f1951d64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5a8a7c67084128b9616e2ff3e37ae7 : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dc4e613aa6e4cdfb0bfc15d8aaa202f : SUCCESS in 32m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3b3353b77124496b336081a4bbbdcca : SUCCESS in 21m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be3c042d9ec45e6b480a199c6fb729b : SUCCESS in 10m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a90f8a3b0d7497c8221f3e3045f6dd2 : SUCCESS in 49m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdfa7911f8154bfb82d678f92a1a75f4 : SUCCESS in 1h 37m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/771dd598e53d49748e0e191a0d358929 : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b8afb16455e46a2d7fb1dd966f3109acb7de8891","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 18:13:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6877369ed4752b704e0c03cbecc3b9631bb92ad7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 18:16:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36e1e0f89e9d4aaa8f3a095ecf75685e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bfa6498908947fb9e530343d9322f9b : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20694907593ad0fd0f28a963d606d31b76670f93","revisions":{"20694907593ad0fd0f28a963d606d31b76670f93":{"kind":"REWORK","_number":1,"created":"2025-03-06 18:36:34.000000000","uploader":{"_account_id":36474,"name":"Mark Loza","email":"mloza@tucows.com"},"ref":"refs/changes/32/943632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/943632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/943632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/943632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/943632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/943632/1"}}},"commit":{"parents":[{"commit":"dcaf6a3ef8ae25f731d8087a2c00f69d6c62206a","subject":"Merge \"Set trunk parent port as access port in ovs to avoid loop\" into unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcaf6a3ef8ae25f731d8087a2c00f69d6c62206a"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-12-16 04:29:52.000000000","tz":330},"committer":{"name":"Mark Loza","email":"mloza@tucows.com","date":"2025-03-06 18:36:34.000000000","tz":0},"subject":"Revert \"[HA] Do not add initial state change delay in HA router\"","message":"Revert \"[HA] Do not add initial state change delay in HA router\"\n\nThis reverts commit c20f2e5136fd241f4be5c37403ab1ed54cdaefb5.\n\nThe fix of bug #1945512 reintroduced bug #1837635 as after\nthe initial backup state ha router can transition to\n\u0027primary\u0027 state on multiple hosts and due to this\ndelay multiple routers get into \u0027active\u0027 ha_state even\nif one of the host quickly transition to backup after\nthe primary state.\nThe issue got visible since ha router fullstack tests\nare added as part of [1].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/917429\n\nRelated-Bug: #1837635\nRelated-Bug: #1945512\nRelated-Bug: #2083609\nChange-Id: I83b53a07362861da98b8361dafd95e94e5048322\n(cherry picked from commit a3689956dde80b9639a3e805257ac02e1044a4c2)\nConflicts:\n\tneutron/agent/l3/ha.py\n\tneutron/tests/unit/agent/l3/test_agent.py\n(cherry picked from commit f30d7c7093f7900ddb788891c4c3fbbc4037ca1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20694907593ad0fd0f28a963d606d31b76670f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20694907593ad0fd0f28a963d606d31b76670f93"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
