)]}'
{"id":"openstack%2Fneutron~944053","triplet_id":"openstack%2Fneutron~master~I78319cc3c94bf869383ecfaf14cb276b46086b8e","project":"openstack/neutron","branch":"master","topic":"bug/2101998","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-26 15:10:26.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-03-26 20:18:09.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-03-26 15:08:22.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-03-18 17:49:51.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I78319cc3c94bf869383ecfaf14cb276b46086b8e","subject":"[OVN] Delete a resource provider removed from the agent config","status":"MERGED","created":"2025-03-11 14:57:24.000000000","updated":"2025-03-26 20:21:04.000000000","submitted":"2025-03-26 20:18:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944053-bug/2101998","meta_rev_id":"471e2d2ff7ce521fd401647530924d69fcc2d798","_number":944053,"virtual_id_number":944053,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-03-18 09:09:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-26 20:18:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-24 10:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-03-26 15:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-18 17:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-03-26 15:08:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-11 15:36:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-11 16:08:59.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-11 17:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-12 12:38:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-03-15 17:10:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-15 18:01:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-03-15 18:01:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-03-24 10:22:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"4ffdc78178f90224ab05b9d664aed2d79d7fe232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-11 14:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bff710f6bbb68490aa0c21e9d79e6fec49cbe818","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 15:36:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/923ac14c63c7408293da408b1dee463c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aba4a563abd74edabfe6cf892dfda53d : FAILURE in 36m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/260fa98513214320918dc00dddf3e7f3 : FAILURE in 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7030b24d742b89d8d87cf1ab885909c7a40952e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-11 16:08:59.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"96084d6212e1d4fe92dc0a6c21137c6b2e983b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 17:39:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66d945222ee5413aaa9437270ea6bd51\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1ed5c5411094427b9a458eb2c8c8e62b : SUCCESS in 1h 45m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3e7f23aad86b492196ee8cb26191010b : SUCCESS in 2h 24m 19s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2233b2e3b12c418485cddc91769f2c95 : SUCCESS in 1h 56m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b32acab25fd44658b635dd0fe3d464c0 : SUCCESS in 2h 38m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29b2e993d2b24dca8bf8b8cd054bbfe3 : FAILURE in 38m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7936bdc817a44d9d91f1022c665703eb : SUCCESS in 11m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f48a87fadfa42e98afaa463f919a747 : FAILURE in 21m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ab57644057a4c48aec3a903155bac10 : FAILURE in 22m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f5feaf434674e52adbd0ad8591980ce : SUCCESS in 10m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4361c6a6383b45238ad86449dfaa8c70 : SUCCESS in 56m 24s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6c9c6b6832bc4f24ae6026e71a32bb10 : SUCCESS in 1h 07m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8ca1ce29a4d14af8aeafc914da160306 : SUCCESS in 1h 13m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/02016e6e94f94db08928c3a99b155add : FAILURE in 1h 54m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5f8f10a532241bd9d7984c13febcce5 : SUCCESS in 1h 00m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5119a51432d44517a89a6e6dbe706050 : SUCCESS in 1h 18m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/def9b7c7f01b4c9681dd1145201b09fa : FAILURE in 1h 35m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/16bfbe1c5857407b8d42e312d46776bd : SUCCESS in 51m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3efc09f28e7d402981ae330df80578b4 : SUCCESS in 1h 10m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6e4a338e8498428da6d88eb28661dbc2 : SUCCESS in 1h 14m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0c58fda245ad4858bdfd4e7acf46c9f0 : SUCCESS in 1h 27m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ce16598b14134e4ab5d956e2ce9dc1e7 : SUCCESS in 1h 30m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0e27c7eda898c6beca25cd3961c17b9e661583b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 08:07:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f4605356af4c1693b37da0d225f2edd4069f0ad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 08:30:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2a5b96639a74dbba248e6ce98032785\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12d2db9100ed4890a6f44723bd92f221 : SUCCESS in 21m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/604b4cf31e89455996b04a283d937901 : SUCCESS in 19m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3190841f54f7b6735276f1b500e977adbe008d35","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-12 08:55:48.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"eced2613d62597a7fad7c8f314631f66b4d0f1a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 10:28:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b80a1ab3c5fa1eb874ade7bda35470b21497b61c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 10:44:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80d97671c1284b6fb9f7b1b396a89381\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de7a32907a08478caf9930026416d093 : FAILURE in 8m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/adfeb84d504a4a36b27a180abd74911c : FAILURE in 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cbec3f81431286491d94283a5ce88e0ce76653fb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-12 11:00:22.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"3eb9c784a4454df722e78106ef8c3c75e2bba9cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 12:07:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3bac5a18ed8d2faae2366d17899f5763ee0d22c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 13:05:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea3c1b9f8dd148d485f6375cc20b3c3c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d04535cfa4174c6596d2f8c61cf06dab : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e67dc0e011f4c4c86540ab6d38710b5 : SUCCESS in 32m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"71115dff4f03aec2b11d227109daa64471878853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 14:42:07.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2360ab9762143f8a9af5409d6003c83\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4114c6598758419ab85f20225ed0f5d9 : SUCCESS in 2h 05m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1706bb717eb643dc89ed4d8ca51091d3 : SUCCESS in 1h 55m 17s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2baaafc0f1a4831808ae63dd8127c8e : SUCCESS in 1h 53m 03s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/088e27ea471c4303a1d3491e524ed7a6 : FAILURE in 2h 31m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93ffe2f6dc0c4551be33b856a2b45bc8 : SUCCESS in 30m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d083346547354b5e894415a508825ef1 : FAILURE in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c80a8ca685f040a586f4bcdfe73db7ed : SUCCESS in 18m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c80a54938ade4dc6913961a695269295 : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd22a2d85cb48178aff4bd4f465cc48 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/828d0bc5f76b42ac96b1d12889b4297e : SUCCESS in 10m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6329a14b390e4e4088ea57d95ba5129d : SUCCESS in 28m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/987716b2dea943f5aebde425be7aa4bc : SUCCESS in 1h 07m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3e788b6c1cc54d60a624acb4fc743e69 : SUCCESS in 1h 06m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c60c127af7db4b829375f8293bd7a5e6 : SUCCESS in 2h 11m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dfdc9f557df44945a5a9d4a9eea7f93f : SUCCESS in 58m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d54866f563404b0aac093086b23fe3b6 : SUCCESS in 44m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4df1331174584d5991a0d7746ddea1aa : SUCCESS in 1h 46m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/43e2fef1c50f4d86a16cb823813a8c12 : SUCCESS in 1h 40m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/93c59ed7f11446f39ab6d13aea988647 : SUCCESS in 56m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c1794e05ab6942fd8186a490327b8c9a : SUCCESS in 1h 00m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5035058446d74ff88b76baa2b67c0fd2 : SUCCESS in 1h 00m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e192080d3f184c558d2469f31707b6dc : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":4},{"id":"3fbc5e51efc6be50b12ec09a3aa679a3a58d45de","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-12 14:43:38.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f9f411f537ea0bede0685db113e234c8ee15bccc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 22:30:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7aa91d1942ac8f1dd42e34a0da793b7ee6510c03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 22:41:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d2141d4cc7944adae09e02cd523b482\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ae70c64de8e4b2fb7acbbba0671e767 : FAILURE in 7m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/58492016a7cf45b49a1a32326cce32c0 : FAILURE in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ef3a5c1ee9a4841cbe3ede3173b794bfb6053d4f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-12 23:32:25.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"4a0abbbe590583b7583d7f14cea0332c30f9c5f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 23:54:39.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c7fd7ba663e4eafabb3faad1dc582dc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f5bc5819681641dcbd5083d6ebb80ee2 : FAILURE in 13m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/381773b6c61f463cab376c4d3375a33e : FAILURE in 15m 16s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96f986e7d9ac4130b64fa81a7b03ef7d : FAILURE in 12m 48s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51334888789a46089421c318d2ebf53b : FAILURE in 14m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3f906d92da04b4f9979e693cbb1579f : FAILURE in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a065758fbfd14dabb2e920afc4fc15d4 : FAILURE in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ca228cfa0dc433f90afdee763ca53e5 : FAILURE in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a004aac79aa14f49a6fbe7e56827c9a8 : FAILURE in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e873de9748ad4dadb64fa30bbfbe3dc7 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/820ffb3fbd594f949b080d10ffff0d57 : SUCCESS in 11m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f3769f60a3814316949f85fc1564ef3a : FAILURE in 1h 02m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6bebf6c6466044b5a98ba20ea596f171 : FAILURE in 1h 06m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4903cbb425d448af943761fa415be49e : FAILURE in 13m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2aa0b8ded573445abca5fa9c7550a848 : FAILURE in 13m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/80e20e23182644dab3fde6a0b53dbd22 : FAILURE in 40m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/62d531ec95d246639d4ef39cbfb3a38f : FAILURE in 1h 13m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/968b35c8f97c4fc58359148a6ce087c8 : FAILURE in 10m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d9178427e408443092ed259727caea10 : FAILURE in 26m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2262288ad3e04dd49570f3fad9ed5add : FAILURE in 46m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c262f6aa5bf34680a51fa5646422a8da : FAILURE in 10m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/acd1f9ea0fbe4b2db19c04092045e26b : FAILURE in 1h 17m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1aa1e2f6287e44769fbffbb77062f89e : FAILURE in 37m 53s","accounts_in_message":[],"_revision_number":5},{"id":"ba7c8c35f0a40c974096c9f13398a0e68f304da1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-13 21:00:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8a661e6c34665f27ab39a7086143fed0c11710e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 21:34:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d37a05632b23416f804112316af5a228\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76ce5624c3d74c64b7e2ac574428a03b : FAILURE in 32m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8b45b29db221494ca40cab5badcfe659 : FAILURE in 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e0b04557c587142d536ab4f71bee7db9fa156f3a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-13 23:27:34.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"a4c428fdeb11359e2700fdcf8502265d9d318be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 23:29:04.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eba427a189a547f0a8b17f2bf134bc71\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/20498036972b4c1cbc322a32aab64e8b : SUCCESS in 1h 28m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/33bcd1c8fbe541ddb6b7e073caf1042a : SUCCESS in 2h 26m 23s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b0cab6f793a48758963c2a669c69ff1 : SUCCESS in 1h 08m 41s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/776da735f8244c9fa27934736a1d3439 : SUCCESS in 2h 01m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/440ca62d21904a1cb469c152e5f0834c : FAILURE in 53m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/402517878d0a45c48efc020b8f655ea1 : FAILURE in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/843415a4940d47a7bea30d7f63d5d1d6 : FAILURE in 19m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/783f3db45d264f9997fb56d29507b69a : FAILURE in 18m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7451b628a1c44fba9fbca76b1d55121 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6934b4617431445c85e474e07fc6b57d : SUCCESS in 10m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a13e6def6d1246c1a69a8e3e931463ba : SUCCESS in 59m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/77bb7bd27e3146328dc6e9aeb44c3414 : SUCCESS in 38m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6c3e3e94d97c48f0a6b518647a073eb9 : SUCCESS in 1h 41m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f60c86932ba445a689fc4549a8aa8654 : SUCCESS in 2h 11m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/03e7c319b9ae4ac98bd9dd10977375f0 : SUCCESS in 48m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/227f196ba5634ecebeccbb8c5e6ab65e : SUCCESS in 1h 19m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0b0724d3911c46af95419616b068fa68 : SUCCESS in 1h 33m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0e300fdef6474143a75b65b9e14ba99f : SUCCESS in 1h 30m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1394978513af4d9e9aee7e331f7ab37a : SUCCESS in 58m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b42bc23df90e4b6faf66ad42f0b1a532 : SUCCESS in 52m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e573daab76f7422b9f61eacda6a89734 : SUCCESS in 1h 08m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/df5aa94dc2084990884dc230fdfc973c : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":6},{"id":"d3b2c6cf95cd20300015e62f679983ed6f7f621a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-14 10:37:11.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"512b0ec62ef075ff58912c3dbc02d786ea4f296e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 11:10:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a26170a28142fb8a4fc1d3cd00564c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/626d21811ffe427b9218ec1a45a5fd69 : SUCCESS in 29m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9b7a25837b9403dbebbe4859efc5e10 : SUCCESS in 31m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b9d46d645dd023b4e74e6ed9a124409ee9d5a81e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-14 11:15:59.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"57d3967c9f0a737ebc57b96853a2536d18e62cf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 12:38:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a1cd1a0f6614681891f0f4bcce33b94\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/74ed10ffeee24bd686c5616134983205 : SUCCESS in 1h 25m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f64f593bffb844bb8cd5f394de35e9d7 : SUCCESS in 1h 44m 09s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14f0c2394d3a45bd892eb63037522989 : SUCCESS in 1h 46m 15s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f382dbd0825443f8d02f70f036367fc : SUCCESS in 1h 59m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8b82051de55419283ebf3dd346bef67 : FAILURE in 49m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3af1d9e59244bc86c24762536e66eb : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/813080f60ebe4242be532741949aa567 : SUCCESS in 21m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b3616abe4564d5ba10bd6860517b3d6 : SUCCESS in 18m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863c95adb4fb4851b70588ecc9fca842 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ed231cedf534bdf800d2c638b9035a0 : SUCCESS in 10m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fb80e5600c044e85916e08cd58c5e90f : SUCCESS in 46m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/82ec18b9521b4e398d875432e53886a4 : SUCCESS in 55m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a0f22c98416f44b6a3d6b29f7210682c : SUCCESS in 1h 21m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bf9c94b449a244da88415a5632a55d36 : SUCCESS in 1h 19m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f836dc8c7a564c99af9f5e59e947fc63 : SUCCESS in 41m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/142b4d59604d4d6a9e58df0e2c636b72 : SUCCESS in 1h 16m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e6d5419d54c24a4e8001b1e0dabe2861 : FAILURE in 1h 04m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4521193338984c5fa41a41f004972656 : SUCCESS in 55m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aa9f336f32a84320befdc6ac7e739e24 : SUCCESS in 1h 15m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/40e5e3de0b384282955117856e557899 : SUCCESS in 1h 19m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd9d214313fb42729f6cfcb91655578e : SUCCESS in 58m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/826cd664c2c54c06999fc86d68deef35 : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":7},{"id":"14b4e0a26e7a58ee61eff31838e9b365aac64717","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-14 15:59:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dd32f5643b09f15b6b7f457cbe92aa1ad42ee777","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 16:24:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e3476d1e3664967bf4c60684af7ca87\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb1cd9bc06c6429f9bb8614a96443d61 : SUCCESS in 18m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a437b42735484423a1a8c870b13bf38d : SUCCESS in 17m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8067d1387bb696ab6b687e76d674aaea6ebf6ee5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-14 17:48:40.000000000","message":"Patch Set 7:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d43f06d008733a709dcfe37dff8f527dc69e26eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 17:55:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a1df126207249d2aea562d84c7a0be0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/04862c1b917d42248dec01f15e8e6f11 : SUCCESS in 1h 49m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fde70f93f2a64639b1604b7680279c18 : SUCCESS in 1h 28m 06s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adb4218a123542f4a032af9b9ee42882 : SUCCESS in 1h 26m 31s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/032f75d221e446fb90826471480a0caa : SUCCESS in 1h 37m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e1ad6f89e0d4e6a8487c82a211bcccb : SUCCESS in 52m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3991d2505f804592850a44459633a91f : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0500f398eabc40c7bdcad13b8fc0c0f2 : SUCCESS in 22m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/808e51c046e342db8d463201a076c229 : SUCCESS in 18m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b50f53a6dc4d4ebfc2287108e98f55 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73ddc41bcd1d4318bc661d8128ca21b9 : SUCCESS in 13m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2fcf952885144b6fb213f7b52bcb6df8 : SUCCESS in 37m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/af04a5d39a9e4d6b89ab49d64f17c0e4 : SUCCESS in 46m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fdc67d33f6474fa9acbe8566f291e1a6 : SUCCESS in 1h 01m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/52c5054ab94243e096da34bc2b677b77 : SUCCESS in 1h 51m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e944e37be50f4c7bb6cd0f2d95896091 : SUCCESS in 54m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb04f909e7c74beb92d44626b347bcdc : SUCCESS in 1h 02m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6f744ac15740419883a97babf2f5db64 : SUCCESS in 1h 32m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/349662eb20614dfa9817b9dbf97587f3 : SUCCESS in 1h 37m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/29b037bc82034da19cb22505341fdc49 : SUCCESS in 1h 10m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6f61d354e25d4e05b7bb1bf3284e9261 : SUCCESS in 1h 04m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a7a1e0cc06894628a2c1540a519e7ac7 : SUCCESS in 1h 23m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/922bb094b57344be9e9a56926fd1d390 : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":7},{"id":"938bf71b4a9f65ee856b973fe03359b529b78d55","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-15 17:10:11.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"6d123d5466349976c1fd91b3adc4e01e4ad2b738","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-17 09:40:23.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"53f82e4faf55e8fd3bbea6a4eeea7369ed3cae38","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-18 06:59:48.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e7b6d3fd840546e25753f1dd93248a3f5208ae79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-18 07:01:05.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"9bf7fc81ce10e297e4a4da8b4268ae1952396aef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 07:27:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbc8841622fc492bb2c5f2b7a0136dd3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc77cca792b04f948cc3aa8efdacd9a2 : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7fb57dacc66940d9b67d3be1733365d5 : SUCCESS in 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b7147f25ce9e0d6deb65469839e094308683d361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 09:03:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3168df538c84ce7afe44d94ea76996f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/79d23e77a75a4e91bd3a1af374f8475c : SUCCESS in 1h 03m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/949d020dfb8b482b97bb25ce87a77cb8 : SUCCESS in 1h 29m 14s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03a8812cc8364588b80aa0d5cdd7df77 : FAILURE in 1h 04m 48s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b6ec7da402b49c1aae2b8bdac0e8dd9 : FAILURE in 2h 00m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/003258a3b4374f11aad6e9b4546ac72f : SUCCESS in 31m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b6994b482b4c6e83a1d9f774f62573 : SUCCESS in 9m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcf2e9fc4c7e43e387b27419aa569688 : SUCCESS in 17m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13d98dec1f67462b8863d06f0cf3ad07 : SUCCESS in 13m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99de60aa90c474b8f80dde38663da7b : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8854f88cba6748f1967a8cf507a0b4b3 : SUCCESS in 5m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b3081e4f720f46db9588ae90c2396986 : SUCCESS in 46m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/60b32e8b3fea4e8eb747b650267d62a3 : SUCCESS in 39m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b973e50838dd4cee951db602aa744605 : SUCCESS in 1h 15m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/00fa3594454941629876f43f1c5daeb9 : FAILURE in 1h 06m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc817e892fb4477bb404aeb0f85c6cd8 : SUCCESS in 49m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0539bf5408740c284f6d982e516707c : SUCCESS in 58m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1770741abeb54510ab2d45ec71b94351 : SUCCESS in 1h 44m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3134c04403584ce983a7c6906469a750 : FAILURE in 45m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b55fb4dc2111446fa7c8afe8f17c0300 : SUCCESS in 1h 01m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9a1b4382672d45cd8e9beec3f1f2eb3a : SUCCESS in 26m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b88ec2caef614bf382605f5f0bbf299d : SUCCESS in 1h 08m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/31162d3735004321b5170c8417414ae4 : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":8},{"id":"f1a2a13b611adb267db3e9d406aebc11d77fb451","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-18 09:09:16.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"4ba323d664a81a958befee30f8d5a8ce55db39cf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-18 14:16:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"306ad5887d61b7e45c9669288b460a5be1c9503e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 14:47:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d336a1ba7ecc4bb293e457f8e4a0a622\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28862042cd854d389ff85943246b10b8 : SUCCESS in 20m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c384147c5d064923a743484676b20516 : SUCCESS in 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7ebbaed29a3f60f30b0b80a787d79439d33e8c84","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-18 15:16:48.000000000","message":"Patch Set 8:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/53/944053/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"6f4be91972d0cc1829fffe87b283ba28942346f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 16:52:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/850a230f3a214c889a113800f4ce32a0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/db366155f4a14a0e8a1f6212488d5e87 : SUCCESS in 1h 53m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ed3b9a86872d4dfc963f71f90838ebf6 : SUCCESS in 2h 34m 00s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f09c4495dce448b29c6fc712a59d6c17 : SUCCESS in 2h 04m 25s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/362979c02aeb417fa08f7b9314035d14 : SUCCESS in 1h 32m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d9e31894c234b48b115b8f53b7b701c : SUCCESS in 48m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/697b1bff5e2e462dabe6a6c2bb8e72c2 : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfd9668608854f23a71dfdcda9af1b50 : SUCCESS in 18m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e31d3ab5d35a4374b71ecea5b010f264 : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87342cb81c6249a28ab31941c6055063 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adc545d4748a43b8a75f3ff3e3e9bbeb : SUCCESS in 12m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8c427b4f21cf4caeb1051142ba15b5f2 : SUCCESS in 51m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/83f330afd7b14f2e8a091db867c5173c : SUCCESS in 1h 11m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9b020af313d5422c8388a00e0eeb1144 : SUCCESS in 55m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6f64cf45c04d449b86b25e74cf2aa166 : SUCCESS in 1h 51m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/128ccda1c7714b09b60ca77f52472653 : SUCCESS in 57m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d73f5f55725a448db1e699d00ed5f3c7 : SUCCESS in 1h 17m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/965a57cf07384072977f9667ce5e5100 : SUCCESS in 58m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a386f8c0ba994f3c92a56f8665a7ac1b : SUCCESS in 1h 18m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bd7b4da0e448436799ed0417d96edb2e : SUCCESS in 49m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/128157b0fbfe437db83275afae7f3e56 : SUCCESS in 57m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe6dbb86329f44669a049469b2fd8984 : SUCCESS in 1h 09m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a31764464334404ea40a63f65e91da04 : SUCCESS in 1h 30m 27s","accounts_in_message":[],"_revision_number":8},{"id":"2bc26095c2f97b51a47d17a037a68c11efee8725","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-18 17:49:51.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"8fd252f914acb99b945bb11e00694621b358ed9d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-03-24 10:22:17.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"673b2b9e67f1f2543f0be68be16e4eda0eb07196","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-26 15:08:22.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"260cb0cc779486bb40fb9f784cbd2f46433624f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 15:09:07.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5da056fea3653dcaade9533ecebdf833f9bfd0ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 20:18:09.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4156c9291de344b5a650c0723939b4b6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b0bf87cbf6274c81af1bc79098e059f3 : SUCCESS in 1h 10m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/79ac27374dd143ebaa6944928985a141 : SUCCESS in 2h 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd465bd77d24ba8bb892994e470a385 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf9b629475934b1fb43693d5e071eddf : SUCCESS in 22m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d52c1b4a2e6a464f88d94bb410323f5c : SUCCESS in 19m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1e4f3faa504220b1b89cd25e2826e5 : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a7180ca40dd40f680815e0cd263d5da : SUCCESS in 15m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/efe6fa64467147cbaf30406fc8ae8b5b : SUCCESS in 1h 16m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fe2392ce7a374bedab2635111f7f09b6 : SUCCESS in 1h 58m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dec501822b5347e79d7302428d78c062 : SUCCESS in 1h 33m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a43b5e9e75e4ee1a3b9f6daeb474918 : SUCCESS in 1h 04m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4d8395791dc415a983dad43557fd965 : SUCCESS in 49m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1d5e5c08753c410a8b6b1a5844464d85 : SUCCESS in 1h 00m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/371e1418d0c248b3a73aea362048c7c9 : SUCCESS in 1h 09m 45s","accounts_in_message":[],"_revision_number":8},{"id":"b714d60f585802291d457a3391d42e32764442bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 20:18:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"471e2d2ff7ce521fd401647530924d69fcc2d798","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 20:21:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc70cfd202cf487db139f279bf0e11cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ef693fef0f452b93fde3aca4b0ce74 : SUCCESS in 1m 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/978a867bc9c3489c8f055c8e490740fd : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"de237bc92f56bd99c7021ebd79cdbbd8dfe5f450","revisions":{"ee85b164bd8cde0eb7ffe6ffd6839ce617dc1443":{"kind":"REWORK","_number":1,"created":"2025-03-11 14:57:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/944053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/944053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/944053/1"}}},"commit":{"parents":[{"commit":"927d5ff3b092cba4112adb8a648d0e7404d7de48","subject":"Merge \"[eventlet-removal] Remove logger mechanism from functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/927d5ff3b092cba4112adb8a648d0e7404d7de48"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-11 14:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-11 14:52:08.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] Delete a resource provider removed from the agent config","message":"WIP \u003d\u003d [OVN] Delete a resource provider removed from the agent config\n\nIn ML2/OVN, The Placement information per compute node is stored in the\nlocal OVS database, in the Open vSwitch register. Each time this\ninformation changes, ML2/OVS receives this event and updates the OVN\nagent child registers assigned to devices.\n\nSince this patch, if a resource provider is removed, the corresponding\nPlacement resource provider register (and its inventory) is removed too.\n\nCloses-Bug: #2101998\nChange-Id: I78319cc3c94bf869383ecfaf14cb276b46086b8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee85b164bd8cde0eb7ffe6ffd6839ce617dc1443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee85b164bd8cde0eb7ffe6ffd6839ce617dc1443"}]},"branch":"refs/heads/master"},"b7cefcc2f2ef15431a328279ed5ed4dfa7ee65e8":{"kind":"REWORK","_number":2,"created":"2025-03-12 08:07:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/944053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/944053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/944053/2"}}},"commit":{"parents":[{"commit":"927d5ff3b092cba4112adb8a648d0e7404d7de48","subject":"Merge \"[eventlet-removal] Remove logger mechanism from functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/927d5ff3b092cba4112adb8a648d0e7404d7de48"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-11 14:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 08:06:54.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] Delete a resource provider removed from the agent config","message":"WIP \u003d\u003d [OVN] Delete a resource provider removed from the agent config\n\nIn ML2/OVN, The Placement information per compute node is stored in the\nlocal OVS database, in the Open vSwitch register. Each time this\ninformation changes, ML2/OVS receives this event and updates the OVN\nagent child registers assigned to devices.\n\nSince this patch, if a resource provider is removed, the corresponding\nPlacement resource provider register (and its inventory) is removed too.\n\nCloses-Bug: #2101998\nChange-Id: I78319cc3c94bf869383ecfaf14cb276b46086b8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7cefcc2f2ef15431a328279ed5ed4dfa7ee65e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7cefcc2f2ef15431a328279ed5ed4dfa7ee65e8"}]},"branch":"refs/heads/master"},"67bc5d596e0352444c3d55316e629d963fc9daf5":{"kind":"REWORK","_number":3,"created":"2025-03-12 10:28:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/944053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/944053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/944053/3"}}},"commit":{"parents":[{"commit":"faeea0ef6b5ce94f66d69206e3ee84e269927183","subject":"[OVN][Placement] Reduce log level to error when Placement exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/faeea0ef6b5ce94f66d69206e3ee84e269927183"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-11 14:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 10:28:01.000000000","tz":0},"subject":"[OVN] Delete a resource provider removed from the agent config","message":"[OVN] Delete a resource provider removed from the agent config\n\nIn ML2/OVN, The Placement information per compute node is stored in the\nlocal OVS database, in the Open vSwitch register. Each time this\ninformation changes, ML2/OVS receives this event and updates the OVN\nagent child registers assigned to devices.\n\nSince this patch, if a resource provider is removed, the corresponding\nPlacement resource provider register (and its inventory) is removed too.\n\nNOTE: if a resource provider has allocations, the delete operation will\nfail with HTTP 409 (Conflict) and the Neutron API will log this error.\nThe error message will be:\n  Unable to delete resource provider \u003cUUID\u003e: Resource provider has\n  allocations.\n\nCloses-Bug: #2101998\nChange-Id: I78319cc3c94bf869383ecfaf14cb276b46086b8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67bc5d596e0352444c3d55316e629d963fc9daf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67bc5d596e0352444c3d55316e629d963fc9daf5"}]},"branch":"refs/heads/master"},"9a4d43d488d22bf8ac9150c1c3f9973f9236e233":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-03-12 12:07:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/944053/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/944053/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/944053/4"}}},"commit":{"parents":[{"commit":"a3a25fbeb21823b413350f3484e469c62bb99a9f","subject":"[OVN][Placement] Reduce log level to error when Placement exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3a25fbeb21823b413350f3484e469c62bb99a9f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-11 14:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-03-12 12:07:16.000000000","tz":0},"subject":"[OVN] Delete a resource provider removed from the agent config","message":"[OVN] Delete a resource provider removed from the agent config\n\nIn ML2/OVN, The Placement information per compute node is stored in the\nlocal OVS database, in the Open vSwitch register. Each time this\ninformation changes, ML2/OVS receives this event and updates the OVN\nagent child registers assigned to devices.\n\nSince this patch, if a resource provider is removed, the corresponding\nPlacement resource provider register (and its inventory) is removed too.\n\nNOTE: if a resource provider has allocations, the delete operation will\nfail with HTTP 409 (Conflict) and the Neutron API will log this error.\nThe error message will be:\n  Unable to delete resource provider \u003cUUID\u003e: Resource provider has\n  allocations.\n\nCloses-Bug: #2101998\nChange-Id: I78319cc3c94bf869383ecfaf14cb276b46086b8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a4d43d488d22bf8ac9150c1c3f9973f9236e233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a4d43d488d22bf8ac9150c1c3f9973f9236e233"}]},"branch":"refs/heads/master","description":"Rebase"},"02c79b7e5e133f48d539e7ee6320c62d84102d58":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-03-12 22:30:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/944053/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/944053/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/944053/5"}}},"commit":{"parents":[{"commit":"94f0dcf9067678ff71fccdf770e8e999db48e017","subject":"[OVN][Placement] Reduce log level to error when Placement exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94f0dcf9067678ff71fccdf770e8e999db48e017"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-11 14:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-03-12 22:30:59.000000000","tz":0},"subject":"[OVN] Delete a resource provider removed from the agent config","message":"[OVN] Delete a resource provider removed from the agent config\n\nIn ML2/OVN, The Placement information per compute node is stored in the\nlocal OVS database, in the Open vSwitch register. Each time this\ninformation changes, ML2/OVS receives this event and updates the OVN\nagent child registers assigned to devices.\n\nSince this patch, if a resource provider is removed, the corresponding\nPlacement resource provider register (and its inventory) is removed too.\n\nNOTE: if a resource provider has allocations, the delete operation will\nfail with HTTP 409 (Conflict) and the Neutron API will log this error.\nThe error message will be:\n  Unable to delete resource provider \u003cUUID\u003e: Resource provider has\n  allocations.\n\nCloses-Bug: #2101998\nChange-Id: I78319cc3c94bf869383ecfaf14cb276b46086b8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c79b7e5e133f48d539e7ee6320c62d84102d58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c79b7e5e133f48d539e7ee6320c62d84102d58"}]},"branch":"refs/heads/master","description":"Rebase"},"afccbe8c1831b5cfb4cd56437c5202703004a739":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-03-13 21:00:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/944053/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/944053/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/944053/6"}}},"commit":{"parents":[{"commit":"979385d8f992093dc7e0b31631ba9adb89e42e90","subject":"[OVN][Placement] Reduce log level to error when Placement exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/979385d8f992093dc7e0b31631ba9adb89e42e90"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-11 14:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-03-13 21:00:05.000000000","tz":0},"subject":"[OVN] Delete a resource provider removed from the agent config","message":"[OVN] Delete a resource provider removed from the agent config\n\nIn ML2/OVN, The Placement information per compute node is stored in the\nlocal OVS database, in the Open vSwitch register. Each time this\ninformation changes, ML2/OVS receives this event and updates the OVN\nagent child registers assigned to devices.\n\nSince this patch, if a resource provider is removed, the corresponding\nPlacement resource provider register (and its inventory) is removed too.\n\nNOTE: if a resource provider has allocations, the delete operation will\nfail with HTTP 409 (Conflict) and the Neutron API will log this error.\nThe error message will be:\n  Unable to delete resource provider \u003cUUID\u003e: Resource provider has\n  allocations.\n\nCloses-Bug: #2101998\nChange-Id: I78319cc3c94bf869383ecfaf14cb276b46086b8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afccbe8c1831b5cfb4cd56437c5202703004a739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afccbe8c1831b5cfb4cd56437c5202703004a739"}]},"branch":"refs/heads/master","description":"Rebase"},"eedc2ec181431565d6703361ac9393627ab7deaf":{"kind":"REWORK","_number":7,"created":"2025-03-14 10:37:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/944053/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/944053/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/944053/7"}}},"commit":{"parents":[{"commit":"979385d8f992093dc7e0b31631ba9adb89e42e90","subject":"[OVN][Placement] Reduce log level to error when Placement exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/979385d8f992093dc7e0b31631ba9adb89e42e90"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-11 14:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-14 01:13:45.000000000","tz":0},"subject":"[OVN] Delete a resource provider removed from the agent config","message":"[OVN] Delete a resource provider removed from the agent config\n\nIn ML2/OVN, The Placement information per compute node is stored in the\nlocal OVS database, in the Open vSwitch register. Each time this\ninformation changes, ML2/OVS receives this event and updates the OVN\nagent child registers assigned to devices.\n\nSince this patch, if a resource provider is removed, the corresponding\nPlacement resource provider register (and its inventory) is removed too.\n\nNOTE: if a resource provider has allocations, the delete operation will\nfail with HTTP 409 (Conflict) and the Neutron API will log this error.\nThe error message will be:\n  Unable to delete resource provider \u003cUUID\u003e: Resource provider has\n  allocations.\n\nCloses-Bug: #2101998\nChange-Id: I78319cc3c94bf869383ecfaf14cb276b46086b8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eedc2ec181431565d6703361ac9393627ab7deaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eedc2ec181431565d6703361ac9393627ab7deaf"}]},"branch":"refs/heads/master"},"de237bc92f56bd99c7021ebd79cdbbd8dfe5f450":{"kind":"REWORK","_number":8,"created":"2025-03-18 07:01:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/944053/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/944053/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/944053/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/944053/8"}}},"commit":{"parents":[{"commit":"979385d8f992093dc7e0b31631ba9adb89e42e90","subject":"[OVN][Placement] Reduce log level to error when Placement exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/979385d8f992093dc7e0b31631ba9adb89e42e90"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-11 14:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-03-18 07:01:04.000000000","tz":0},"subject":"[OVN] Delete a resource provider removed from the agent config","message":"[OVN] Delete a resource provider removed from the agent config\n\nIn ML2/OVN, The Placement information per compute node is stored in the\nlocal OVS database, in the Open vSwitch register. Each time this\ninformation changes, ML2/OVN receives this event and updates the OVN\nagent child registers assigned to devices.\n\nSince this patch, if a resource provider is removed, the corresponding\nPlacement resource provider register (and its inventory) is removed too.\n\nNOTE: if a resource provider has allocations, the delete operation will\nfail with HTTP 409 (Conflict) and the Neutron API will log this error.\nThe error message will be:\n  Unable to delete resource provider \u003cUUID\u003e: Resource provider has\n  allocations.\n\nCloses-Bug: #2101998\nChange-Id: I78319cc3c94bf869383ecfaf14cb276b46086b8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de237bc92f56bd99c7021ebd79cdbbd8dfe5f450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de237bc92f56bd99c7021ebd79cdbbd8dfe5f450"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
