)]}'
{"id":"openstack%2Fneutron~944099","triplet_id":"openstack%2Fneutron~master~Ib237c7c1e1ba29626a03765a0521180fc2718c32","project":"openstack/neutron","branch":"master","topic":"bug/2102083","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-24 12:31:23.000000000","reason":"Change was submitted"},"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2025-03-24 12:31:23.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2025-03-24 12:31:23.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-03-24 08:31:07.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-03-24 12:31:23.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2025-03-24 12:31:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib237c7c1e1ba29626a03765a0521180fc2718c32","subject":"Allow to delete singleton instances","status":"MERGED","created":"2025-03-12 09:06:19.000000000","updated":"2025-03-24 12:32:56.000000000","submitted":"2025-03-24 12:31:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944099-bug/2102083","meta_rev_id":"8a2bf46819006dda9a14aca7c90065a4df752e11","_number":944099,"virtual_id_number":944099,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2025-03-13 21:31:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 12:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2025-03-24 10:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-03-24 08:31:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2025-03-15 18:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2025-03-24 10:20:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-12 09:16:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-12 09:31:37.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-12 11:03:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-12 12:36:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-03-12 13:27:24.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-12 13:38:43.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2025-03-12 13:56:22.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2025-03-12 19:05:49.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2025-03-15 18:05:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-15 18:05:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-03-15 18:05:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-03-24 10:20:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a626dc331594a9b1451d44d6686f5c0aa7271e5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 09:06:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aef94463adf4a96be1ad206dbcecd32a082c07da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 09:16:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b12a36bda5fe46a8a39a4240a3a61013\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7049fc23ec7c4416a99390a5f78d0563 : FAILURE in 6m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a4887ab46034a9ba1299b5678720b0a : FAILURE in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"568202926c4d52d5b6118d8b23e948d360bf5821","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-12 09:31:37.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a1741223703be68e74964e828d490543eca16a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 11:03:39.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93c7bbc6c86349999a8589b05c19cd21\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d171f40316434c23bc502686e80ac2c9 : FAILURE in 9m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/44b3fd69926c4fe783c2ba3eee013e15 : SUCCESS in 1h 29m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee6e3cda4e21441d9e48930331e30e4a : SUCCESS in 20m 11s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4b1c5adadb149308f0e9f99774d8c83 : FAILURE in 13m 24s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6186baeb47a44434b8702bd783c6dd6b : SUCCESS in 1h 55m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/688331bc29c94b5d87b770b32bbc5583 : FAILURE in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8b4fe3189f45fca121a5006bcf6691 : FAILURE in 11m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c969a2cf97c4128a18af8906895981b : FAILURE in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7aefb0fdedfb491da40adc290992d974 : FAILURE in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a448097d8140aa8f2bfea6dc32d04a : SUCCESS in 9m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3b4b2c7b87c64376b4d080d8882790f2 : FAILURE in 52m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b3d71ec48f604a88b48071eaefaa2eeb : SUCCESS in 49m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d26ac614fc69468db67dfa7dad972afb : FAILURE in 17m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/293f03672d2f4dd79c5db952db648dd1 : FAILURE in 20m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d0c87594c2f45afa111d5900755a4cb : FAILURE in 44m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/66a9a9f72bac421a998506d20d9a0da4 : SUCCESS in 1h 11m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2e1a537ab0634a4a903386496d8f1f10 : FAILURE in 25m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f12db12421b24c2ca22aca7e6ff0e2f4 : FAILURE in 12m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5c7940fea5748588f8750a1d171d184 : SUCCESS in 1h 00m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bcd9912bf5c54b50bd0ba282d024dfc1 : FAILURE in 25m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/587efff9d8c946c8b990fea0eaa97ab1 : SUCCESS in 1h 29m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/186d6c32669843289d18c338cb87cdba : SUCCESS in 1h 25m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4a227b65aff58274bd34b06411332fa2d7868533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 12:06:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0c152c328fa0ace12d20f76f95fdc15279217863","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-12 12:36:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf7a71c22a450e8841731e3b0684f47fd7ef9ac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 12:42:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bb0a20bc2ea45679bd26d095512743e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7694bfac61d4ea7acf154138c63d0db : SUCCESS in 25m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3f64f5c825ba45cc8adca5e18a4240fa : POST_FAILURE in 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d804594b6976c6a4f772a309709465c244963294","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-12 13:27:24.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"12ed4c2ca390d386795cbfe34558a5cdc2a465dc","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-03-12 13:38:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ad7e154506b87c08f9e5e49bbcfdef989fd558db","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-12 13:56:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22344e23b3b9c00e56581e0caefd07a684995265","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-12 14:00:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c623b7d65a820554ce4d08a760a95698225680c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 14:28:35.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dc8782d4b3d4e4aa4118531b9e67ec1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c4b114cafd0a44a68ac7814bb5dd0537 : SUCCESS in 1h 39m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e297e9046b1f4a5ea3170444bccd2ebc : SUCCESS in 2h 19m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/203c916708b7478b995ed669b32be61d : FAILURE in 34m 56s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/caf02ae6e5844e45a833882bfd1c480f : SUCCESS in 1h 08m 01s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65cf51ed3bfc4a7291cc4e861b6686b7 : FAILURE in 1h 37m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8016518b05b24e0a83239b2dda8d02e0 : SUCCESS in 47m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc478c3f3314ef38e53ff3b088dbc3a : FAILURE in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a44aa657a19e494d9948f57f25a98e66 : SUCCESS in 18m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e369edc774e4667bb10acde775fb78d : SUCCESS in 21m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1aac12d1f14fa9a5b855ea04d6c4d7 : SUCCESS in 9m 19s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7ba0bf22cfdc46f4b30984908f421ceb : SUCCESS in 50m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7e714a142a9a4784b4ede687ca53333b : SUCCESS in 44m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f5c9723cfb524a92b3ee62e7d1d285c9 : SUCCESS in 52m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1a2bdc9d3e8849debc1351fce491bee0 : SUCCESS in 2h 10m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c95c3dcad5140d49e7e83de83e5d254 : SUCCESS in 1h 02m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebb1fb45421a470a9d296db4c82ef445 : SUCCESS in 40m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5f3cf89cce0046eb93d3f4b1d921d5f7 : SUCCESS in 1h 33m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3195a3f8fe994f5caffcccfc9adcb49f : SUCCESS in 1h 39m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/05235b6ee5bd498895fba1f2b442d553 : SUCCESS in 1h 04m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2e4bd87663ad4f54bec67fac974029a7 : SUCCESS in 1h 01m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/67eb22b290ab439293cb2a6003c03ee3 : SUCCESS in 41m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/be7108ced433495b89675eae7b141564 : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":2},{"id":"b8f785ba70dfafc1a714084522e24f128d204a73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 16:08:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d8a8d376fef40edc4ef135c74820b50af72f683c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 16:18:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bfba04fd12d41e1ab9495f07135d61a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d479b0c812b408895920f15bcbc5238 : FAILURE in 5m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/02bff9597eec49d79e76e6c5ff5af91b : FAILURE in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d9ed387bbce00c680427dca97d57701ca4ed8c45","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-12 17:04:23.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"4561bcb56e67b8c43c3aa3bda74c0902ec994510","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-12 17:05:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"403017c48784a520ffe9d8359a5b3d346c6ea773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 17:31:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92ec62ee273b48bcb35306382b21001d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0e6f7aeb8e5a412f93dce076517a55ff : FAILURE in 12m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1c4b1fb809e043d0a5ea2ec4d6145286 : FAILURE in 18m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f253f7b3c2f4e9aa2cda09a5ced5a5f : FAILURE in 14m 52s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76152dd89f5b4f9ba787f416c2df7dce : FAILURE in 15m 06s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8953358a9dfe4537a0f88c88c2e314e0 : FAILURE in 18m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7a6a5751ec54c5088040772dd15d0f0 : FAILURE in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b7458e6b454a6499c25f4d15bfa3d9 : SUCCESS in 13m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e43895296ec4f768b97fe6986354813 : FAILURE in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15abead9f6944cc898accd3ad656b9a4 : FAILURE in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b632cdb7b9455c9a23c9fc731326aa : SUCCESS in 10m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/653ca77ac1b1421c9329331a897d936e : FAILURE in 24m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a7876dbaebf34882b981817096a0af70 : FAILURE in 29m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6278badf8b2543368751070eca9bdcf7 : FAILURE in 25m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/61eb9cb41dbc453c82e1a1abac6a0b77 : FAILURE in 20m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2096ec0669e64cb8a99867ba31b391ed : FAILURE in 56m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc70bb0a27e34858a5cb29cf90874c61 : FAILURE in 1h 11m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e5a73af8feb42daa1647d381326649a : FAILURE in 26m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3b57e566f6734a20a5b1d814adaef1f4 : FAILURE in 24m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b186751262934d539026bf4b972d7936 : FAILURE in 43m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c297e97af70245abae759b3ceaafe2b4 : FAILURE in 24m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c2aa1bb1f8d4b0aaf5ea6f3c4189d18 : FAILURE in 1h 17m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/df3a6272b2a3416ea4b7dfb91cab3e96 : FAILURE in 50m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b1a583ff105ad0a19978633c6650109cb17e2bdc","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-03-12 19:05:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a4b1996a4872f0e2cc36c0de243f750f8ef97b1a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-03-12 21:13:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"861329c603e344ac1fb261ea18c5383a0cab7f78","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 22:16:46.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"551b1020c7881fe298014989d7b4e8a9e07e5662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 22:30:47.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"11c98b3d545fef74c57dd4a9724b5fe6a00b0b45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 22:39:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdada8e600094394804028f13aefb9d2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/158a2910b327461288713a8d4d6a8321 : FAILURE in 7m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff864cdbc17a4d95a28209899051b570 : FAILURE in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fd06b52e1bd48ddc6c90bbf45499adc9143fdff6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-12 22:56:29.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"fb576a99f2bd1bb7dbea71896386a51e677f8b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 00:06:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/798d980b91e349f28752064476d03f63\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fa3a511cbc924092bf186d47d71b4395 : FAILURE in 15m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/440e7f8bed454484bfad437512b72e46 : FAILURE in 19m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ef29c702aff04972b87768aa9f5e23cd : FAILURE in 18m 19s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b12e022eb56b433c9fb5869e4e618122 : FAILURE in 16m 46s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83583a248c134f44ade1078dae72ca2b : FAILURE in 13m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb40b4b7134a4acb8727984600bfa18e : FAILURE in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ec87210ce364680a24bdb32123457cb : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e7c6cfd79ed40b8b41f9f5edd5bd0c9 : FAILURE in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df2e1f35b7de44628dc716c95f4c296e : FAILURE in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6aa31d2166745cfa43d8b1bce8d501f : SUCCESS in 11m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/179834541e0a4c0699ce4201124238c5 : FAILURE in 41m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bdb239ee5eb64b4396239e341c90a877 : FAILURE in 1h 03m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e459eb3982954750b6d16a1d10edc828 : FAILURE in 16m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d3c065267fa447c4b4b553f47cbe8e47 : FAILURE in 12m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dd6fea949344cbbb47b2a093f92983b : FAILURE in 25m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0be709d0dfb541e2ae9c0635b92ff87b : FAILURE in 1h 20m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/66c8a813bc084bccb05fbe9280e3e33d : FAILURE in 36m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e25f768ada8a435384e0e8de8f9bf050 : FAILURE in 15m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/065ef98b98f54b1cb4a831d2187a126c : FAILURE in 27m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/caabbe64d2714ddd8a86d355bae0e41a : FAILURE in 27m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce758b9a500f470c80fc0a1f56648c16 : FAILURE in 1h 14m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0fac05bb5b824489a7476b509f895aa2 : FAILURE in 32m 33s","accounts_in_message":[],"_revision_number":4},{"id":"535ae274fc02e1a5260fe832a221b1db78f61d96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-13 09:45:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b1a4b91d22f203d635109b04c7fe4d28d5faf116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-13 09:52:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"27611660e8e8f314698dd8739ea4879b5c227b36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 10:14:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/977e1edb67d749d3bdbd9708837246f9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df48eb5c3bb340a0917bf25e10b4e147 : SUCCESS in 19m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0994958e438443ca5409b3c19e137bf : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6b642d82c663bef71f93ed1571f007975772d00c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-13 10:25:05.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"e5aa855cc497ded233a93c452061a8259fe668a7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-13 11:03:16.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"527904c1e35bedceb45aba56d25c03d60c33ab6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-13 11:21:03.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"68e34adc96f85b57612a1e1023baa5c7cdbbcc70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 11:45:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc00cb1e9b46435dbfc6130c50f6ceeb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8fd04939a36416d89f4b7dbe3e9240c : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d4eff0b0c554e0b997223518de689ae : SUCCESS in 21m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"17062755f37f7f832900b197358b6c28fe88a39e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-13 12:00:12.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"02006748d2699601db40f676f11ecc178b305c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 13:56:23.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c1e51d4ae46450f9838c8ec938a3df1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/82ce5cc69526482c9a7dab4d149a97a2 : FAILURE in 2h 02m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/13c4ceaacd504077ba9575e647e4b78e : SUCCESS in 2h 33m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7317bca06a134dbb9842a3791e09a512 : SUCCESS in 48m 42s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa6d2f40f2dd49478de0b9d3c063e1c3 : SUCCESS in 1h 21m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b22284bd66b84539afd6f89192ee08bd : FAILURE in 2h 19m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1534011b9c32408a95af4aa318c3085c : SUCCESS in 55m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b594b09cb62441aad20414f65c41f92 : FAILURE in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b80f5752c12645e1ad45c0b5eeec2b7c : SUCCESS in 19m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2b5194b3ae54615ae6c000f79ca6213 : SUCCESS in 19m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60be462b9dd843749d4e11dd5d35b0c8 : SUCCESS in 9m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f4143a57409d4623847f322a0ca5f090 : SUCCESS in 52m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5acbfbb34b554a769182ed714bf05993 : SUCCESS in 1h 18m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8eb6d265b27c451584c8a0a0a7b49e67 : FAILURE in 24m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d205c5a049bf45c9906ddc4f64f19bf4 : SUCCESS in 1h 40m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c21bf0158d2141fd895e1956d04a2f92 : SUCCESS in 1h 09m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/789af5b1149e4cf28fec33fdb0b48084 : SUCCESS in 1h 00m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8443eef39d904aef8a98eb0442d74448 : SUCCESS in 1h 28m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/34187825d6d64fbbab458f33ad5d083c : SUCCESS in 1h 55m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/50edfb93062143a1979a4eb079d01ea4 : SUCCESS in 1h 10m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bc9244e4ae0a4400a91211d4e06091bb : SUCCESS in 54m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8f2ea7d19f1e4ca18728b46f86dcf3d3 : SUCCESS in 1h 00m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d17b01dbb123453da22a0f2f4c33c9ce : SUCCESS in 1h 10m 53s","accounts_in_message":[],"_revision_number":7},{"id":"7c3de94d46ae20f37c4db720d8eb9555d711c986","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-13 14:02:41.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2b64815ea2c7e078859bfbff659b47d9a7952a96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-13 14:37:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"573c2faec4c64f000ba158036cdaa069552f58b0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-13 14:41:57.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"5eb577e1c86f791138933765f1a99198cb58b152","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 15:14:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8b6a3e5b3514ad68b77447287f66339\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/946ef2e8f40e462f9dc6cbdfcc4b1464 : SUCCESS in 33m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/55881d3c8f7147588266f596ce2670a3 : SUCCESS in 27m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b6fd22bbc1772d69b9a68b6a08b916c5ac19690e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-13 16:38:41.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"5c992f61db1f8afe708400e72b0fddd3230145e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 17:12:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06ba9a4c825a4dd3b746cf738eb24467\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/107799d6a2774d20adccba476f8c53b5 : SUCCESS in 1h 53m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d49a0232bbf94e28941c67c2084004dd : SUCCESS in 2h 26m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a109a7fcedf6444085ed02f2ddda21ff : SUCCESS in 20m 16s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39652eaa8e414847944aecb6bf49a5ca : SUCCESS in 1h 13m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53cf3d204f854964b2a0eab1afff5662 : FAILURE in 1h 34m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5a87270c78c46598af0a8022a361854 : SUCCESS in 38m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06da8a06f98e47f3b87c4dd477207172 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31e5daafe468458b84096559b4bcb5d0 : FAILURE in 20m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22257472d96345f488378a8ef09daadd : SUCCESS in 23m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58a91e1bf584f328c772f15435c9666 : SUCCESS in 6m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/12a85635fdda427383f7cc12ab8dfe2a : SUCCESS in 1h 11m 52s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aeaae0a0ff234e589e8663728e04521d : SUCCESS in 1h 10m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1456685dae004437baac38f503960da8 : FAILURE in 40m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a03182524a204d01be6d2995ad18cb85 : SUCCESS in 1h 30m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/500cc823f1f84f44a448e187ce12bb01 : SUCCESS in 1h 09m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2744d0e1538b43dd90fbdfee75b2479e : SUCCESS in 1h 15m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2f62433184be4a45a75b4a46b81380e2 : FAILURE in 1h 01m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d9beca33727047088194005005d83490 : SUCCESS in 1h 34m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6e1f716a40344753ad2934d06d69fd4b : SUCCESS in 59m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/16fd4a25093d4ecb93a7dd82af467ef6 : SUCCESS in 56m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/75f709fa426c4d18a4e1827d686d0763 : SUCCESS in 1h 11m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/063fc1bc171a497385bcfb97b47b10c3 : SUCCESS in 1h 33m 47s","accounts_in_message":[],"_revision_number":9},{"id":"aea51bfc4c13fd0aeb14fc0c734b1c7af63ee68c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-13 20:52:01.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"5663c510a4fe714ccda4932d2c09687959d4b92e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 21:12:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676dd3cc208d466e897ff9ed6416cf2e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5adfdd7021954e14875851d323fb805e : SUCCESS in 19m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d23c3d3250d145c19c9919243a6d020e : SUCCESS in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4cda69c419a7adafadc2862f74efcfb4f339bd20","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-13 21:31:53.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"b947713534e78bdeab0e48f40afad4fa3c365e12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 23:05:16.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c24bc88613b24cfea23df6750acfa134\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/473ce24d3189438eb16b9f50eea580c7 : SUCCESS in 2h 01m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/94d564d89d1c426cae192e15033d52e0 : SUCCESS in 2h 08m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51e5955a7e7040c6aab5932fc1d53c0e : SUCCESS in 37m 57s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ea52a6fad5741268a4f7afbafeed21d : SUCCESS in 1h 21m 39s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d10f66eb95f4e9c8a197b46af9d1620 : FAILURE in 1h 30m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0e92602b03f45b691800d48eaf1e0d4 : SUCCESS in 50m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96360fe8cca842128b824d29142151e3 : SUCCESS in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91dc2459f45f461b856821851facaacf : SUCCESS in 21m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85b1621cb96b45b7b205cbb19a200773 : SUCCESS in 26m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19204aa2028e43c6b47b29a337b79950 : SUCCESS in 11m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8f1bc08de3394f19ba30c9e07c5ff85a : SUCCESS in 44m 32s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/00b9dc67ab784f648da419bc45109f90 : SUCCESS in 1h 05m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9da4d4e0e194403d95743c57b7b55f26 : SUCCESS in 1h 18m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/662990a2f31e4bb8bf6888c95c398a0d : SUCCESS in 1h 43m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2531ca399c441028f14f07a7dd6648b : SUCCESS in 1h 06m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f108ded1d4649a88dfec85dd508de13 : SUCCESS in 1h 24m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c76f7c6326cc40598efd326b1ae032c2 : SUCCESS in 1h 17m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3b3b32d249754bbe912b0648cb307ce1 : SUCCESS in 54m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2ad9f68efd1940b6bc43be60fe61595a : SUCCESS in 1h 07m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/721c79570d684261999965e5ca0d68d3 : SUCCESS in 1h 08m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e09cffed79a1470e8ccd09961484028d : SUCCESS in 1h 11m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c5ab7b296f7f4dce8373314b211a66a1 : SUCCESS in 1h 23m 04s","accounts_in_message":[],"_revision_number":10},{"id":"23fe84f960ce1653cd3ddb2ee689339c738e55d1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-14 07:01:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"50724f77b10543834287a4bf8db5fb5e97159c3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 07:31:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4dec3c083c4eca998ade9bdbc3f92a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce6523d729374cbc9896698cb1dfa3d9 : SUCCESS in 29m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d5f784d2dec4d37b0e96a7184b35c5a : SUCCESS in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a1fae18f638dfbe9bba73004b72a1e5a5aa260de","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-14 07:40:49.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"f6ffd7ab1581726a3089b27cf30aeac68236a42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 09:12:09.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39535536d3fb46169fa55fe29b6e1729\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ad785b9028fa4e62adbf6084b3e9d57d : SUCCESS in 1h 07m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/54d430b97ade44d9998c5eeb442ca394 : SUCCESS in 1h 49m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ba9512a85584063b02db6e5ffbdf022 : SUCCESS in 21m 13s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab5394e23a664d5098ec9f0db82e28cd : FAILURE in 1h 10m 52s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b2cb427f03c47b79e19981f82a7dd05 : FAILURE in 2h 02m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c6af1e71bd1439fb5bc6482078bbceb : SUCCESS in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/496ee209930c4d98b28bc442cdd8cc7b : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b97c48de37e2465bb766add00570f048 : SUCCESS in 16m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8bda363372614b55add0db7ad89ae9eb : SUCCESS in 20m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b498fbe76654e95b172ec71380a6901 : SUCCESS in 9m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/12da46a35bcf4aafa296a4f91295348b : SUCCESS in 57m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/77e42120e4b04d27953ec258da526af6 : SUCCESS in 1h 11m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d76ce3d6accb427191887c4f40f35154 : SUCCESS in 1h 14m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/062a0706e49744cf89f9f6c8fd892198 : SUCCESS in 2h 04m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/980157f42e6d461984c38f99b603c21c : SUCCESS in 1h 02m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e312177d698d48b39226e206e088f58d : SUCCESS in 1h 18m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/36f63bf799624a9f856f3df2f64364dd : SUCCESS in 1h 53m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/65495533037d41009b70d59754868d41 : SUCCESS in 54m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/78815fbc309445bcbbca2a0fd93f5913 : SUCCESS in 1h 03m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/850e9621c2214b6580509df9d2975c78 : SUCCESS in 1h 03m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf4beb81a6e94d4b851404e21e9180f7 : SUCCESS in 1h 24m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/11c059837e494914a0771d011ea7a1c4 : SUCCESS in 1h 15m 02s","accounts_in_message":[],"_revision_number":10},{"id":"afc75b2b3a84a53675ea84221769f79569bf1008","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-14 10:37:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"77a3c0a7e9dfb87d79148b9d38f23ca2b5373c70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 11:17:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74ff9b1cca034a79930561009390f344\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4e7f8f1a08c4773bf4f6226e0ca2c84 : SUCCESS in 37m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81ad44b4d80d47d1ab37c8695f4ff3eb : SUCCESS in 35m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"119d8dca96a5552980ed144baec5d95d733322a6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-14 11:55:59.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/99/944099/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"58b20c84211e7044de6818498bf88fb8901bfc3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 13:31:37.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8031c4f638140b48f6cfb3239673e2e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/72445856afee496780ead12236146574 : SUCCESS in 1h 16m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e8083f8b2dc9412cb7cca5cdb50314ec : SUCCESS in 2h 39m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba432ca15e78483cb581d8d09153df3a : SUCCESS in 48m 01s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4e6ac0c47d0457dbd6c50cc4d4d662a : SUCCESS in 1h 44m 37s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/827a41e189d8480ebd4c5c99bb62eccf : SUCCESS in 2h 20m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6381998b63ef45bca443eebf64637671 : SUCCESS in 40m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe4144dfb1e4639a4b914b705ccebe5 : SUCCESS in 13m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcfb968f2859421caf36094a4d23fb89 : SUCCESS in 20m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7bc5f3e946fc4ee6bb688b262595e042 : SUCCESS in 21m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9ea98bb837432bb1f3f84a24e7759b : SUCCESS in 12m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/97f3ae07dace47b3b196fb8964ade978 : SUCCESS in 48m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fc00e2f6c65447b39ea466bb36b0ce2a : SUCCESS in 1h 20m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/43699ba5d94e4c01859e7d2748409379 : SUCCESS in 57m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/98f2e8f810764436b8475550f881fadb : SUCCESS in 2h 06m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/658885bb596f474da5b3b081fbf6c629 : SUCCESS in 1h 04m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9cced0c60a04bb49bf938a36faaf018 : SUCCESS in 58m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cc59d61bd172415da0101b4c5f92f1ef : SUCCESS in 1h 40m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e61fd0981e054a379446ff9246631fbf : SUCCESS in 1h 45m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a7e49eae71b84c6a905d2125eeab9ae8 : SUCCESS in 1h 05m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a4172a8dc70049889b13963736dae2e5 : SUCCESS in 28m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b4c1b793cc7b4090b7b387386155b087 : SUCCESS in 1h 33m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6797a65575ba4940ba865994cdfb7ff8 : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":10},{"id":"0078e721a8c1ea49670654991a2e1e0c189b8d49","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-15 18:05:01.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"66ce260e7bcdb2334e0aed1ba522ada8d46fbc4d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-24 08:31:07.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b7bd67c50cd141c510722485e0ebf118f36d82f5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-24 08:42:48.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"99838a4ccfaef8e4350459bbfd0ffc9f33920412","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-03-24 10:20:53.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"ca12476448a1f6cf3de7dc7a58f844bce3e16c07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 10:21:23.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"d2f40f17816acaa5ee5f165f082125d99355198f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:31:23.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccda9f9aea3749aa93a58e65662331f1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b63884b0a80043b7b6de9b5d8807be81 : SUCCESS in 1h 44m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/18fd21fa943a40988153c674952bc7bb : SUCCESS in 2h 04m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007526b1b71d410ab564b0b60f7c807a : SUCCESS in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0213736626b84b31b488258965be5405 : SUCCESS in 23m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45365c5e25624bb0af81f1b5680c884c : SUCCESS in 19m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94d8dc73ee64f83b519011c35d85f03 : SUCCESS in 12m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/62e57e49c37a40149cf20e1297b167fc : SUCCESS in 1h 25m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c8ab530fd05e427dbd8d8ae6512a8d08 : SUCCESS in 1h 16m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/60c4dfd6233d4b319699aeb3e5275b81 : SUCCESS in 1h 48m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ab14bff42a14d63820e1ca0c863e82e : SUCCESS in 1h 00m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/054f19230c27439aaacfdb3856e5c516 : SUCCESS in 26m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/716e786a5f1246c8bac42169a9fa1a53 : SUCCESS in 58m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f393c224ae94a3bb3673a2a7e9fea40 : SUCCESS in 1h 14m 35s","accounts_in_message":[],"_revision_number":10},{"id":"cd6e5e0c2319b314d1b0c192a8aca1e7d39398fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:31:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"8a2bf46819006dda9a14aca7c90065a4df752e11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:32:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04c3b105105d4b1eb9f688d6cf8f7b84\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5b31174828740a591028dcbbd0e877e : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"1e560ecd8a2a620b7b6d677bae962764302136a0","revisions":{"84dbe4528a2cd8f7965fa4ffcbc4cd9643957f35":{"kind":"REWORK","_number":1,"created":"2025-03-12 09:06:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/1"}}},"commit":{"parents":[{"commit":"2f6640d6dc886314b11b0dd53eb203f27f874047","subject":"Merge \"Add devstack plugin to enable ovs metadata_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f6640d6dc886314b11b0dd53eb203f27f874047"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 08:58:03.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 08:58:03.000000000","tz":0},"subject":"Reimplement the singleton class","message":"Reimplement the singleton class\n\nNow the singleton class used in Neutron is a metaclass. It also adds\nthe subclass method ``destroy`` that removes the created instance\nfrom the singleton, allowing the user to create another instance. This\ndestroy method is idempotent.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84dbe4528a2cd8f7965fa4ffcbc4cd9643957f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84dbe4528a2cd8f7965fa4ffcbc4cd9643957f35"}]},"branch":"refs/heads/master"},"21995ce8d270bf9e8ee76047b1cb104e663c30b6":{"kind":"REWORK","_number":2,"created":"2025-03-12 12:06:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/2"}}},"commit":{"parents":[{"commit":"2f6640d6dc886314b11b0dd53eb203f27f874047","subject":"Merge \"Add devstack plugin to enable ovs metadata_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f6640d6dc886314b11b0dd53eb203f27f874047"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 08:58:03.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 11:53:12.000000000","tz":0},"subject":"Reimplement the singleton class","message":"Reimplement the singleton class\n\nNow the singleton class used in Neutron is a metaclass. It also adds\nthe subclass method ``destroy`` that removes the created instance\nfrom the singleton, allowing the user to create another instance. This\ndestroy method is idempotent.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21995ce8d270bf9e8ee76047b1cb104e663c30b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21995ce8d270bf9e8ee76047b1cb104e663c30b6"}]},"branch":"refs/heads/master"},"9600af01d8f2d7d57641bcbd418a1c374b66313c":{"kind":"REWORK","_number":3,"created":"2025-03-12 16:08:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/3"}}},"commit":{"parents":[{"commit":"2f6640d6dc886314b11b0dd53eb203f27f874047","subject":"Merge \"Add devstack plugin to enable ovs metadata_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f6640d6dc886314b11b0dd53eb203f27f874047"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 08:58:03.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 16:08:35.000000000","tz":0},"subject":"Reimplement the singleton class","message":"Reimplement the singleton class\n\nNow the singleton class used in Neutron is a metaclass. It also adds\nthe subclass method ``destroy`` that removes the created instance\nfrom the singleton, allowing the user to create another instance. This\ndestroy method is idempotent.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9600af01d8f2d7d57641bcbd418a1c374b66313c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9600af01d8f2d7d57641bcbd418a1c374b66313c"}]},"branch":"refs/heads/master"},"ad738ec3486573cd7e8e23427402ef6043ab0d51":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-03-12 22:30:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/4"}}},"commit":{"parents":[{"commit":"2f6640d6dc886314b11b0dd53eb203f27f874047","subject":"Merge \"Add devstack plugin to enable ovs metadata_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f6640d6dc886314b11b0dd53eb203f27f874047"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 08:58:03.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 22:17:50.000000000","tz":0},"subject":"Reimplement the singleton class","message":"Reimplement the singleton class\n\nNow the singleton class used in Neutron is a metaclass. The instances\nare stored in a ``weakref.WeakValueDictionary``. If the subclass object\nis deleted, it will be removed from this dictionary; the next time that\nthis subclass is instantiated, a new object will be created and stored\nagain in the metaclass.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad738ec3486573cd7e8e23427402ef6043ab0d51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad738ec3486573cd7e8e23427402ef6043ab0d51"}]},"branch":"refs/heads/master"},"1a101770bd342e02d1919213b2d63d7892db9a74":{"kind":"REWORK","_number":5,"created":"2025-03-13 09:45:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/5"}}},"commit":{"parents":[{"commit":"09ba4f99b8b48828546005ba2ebb2255d06bead9","subject":"Merge \"[OVN] Retrieve the HA_Chassis_Group in ``_sync_ha_chassis_group``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09ba4f99b8b48828546005ba2ebb2255d06bead9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 08:39:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 09:45:39.000000000","tz":0},"subject":"Allow to delete singleton instances","message":"Allow to delete singleton instances\n\nNow the decorator ``SingletonDecorator`` forgets the instantiated object\nif all references to this object have been deleted. It is needed to\nkeep at least one reference to the decorated class in order to make\nany new instantiation idempotent.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a101770bd342e02d1919213b2d63d7892db9a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a101770bd342e02d1919213b2d63d7892db9a74"}]},"branch":"refs/heads/master"},"24c025d68e3594badf621c4f545c01c377c6b799":{"kind":"REWORK","_number":6,"created":"2025-03-13 09:52:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/6"}}},"commit":{"parents":[{"commit":"09ba4f99b8b48828546005ba2ebb2255d06bead9","subject":"Merge \"[OVN] Retrieve the HA_Chassis_Group in ``_sync_ha_chassis_group``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09ba4f99b8b48828546005ba2ebb2255d06bead9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 08:39:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 09:52:32.000000000","tz":0},"subject":"Allow to delete singleton instances","message":"Allow to delete singleton instances\n\nNow the decorator ``SingletonDecorator`` forgets the instantiated object\nif all references to this object have been deleted. It is needed to\nkeep at least one reference to the decorated class in order to make\nany new instantiation idempotent.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c025d68e3594badf621c4f545c01c377c6b799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c025d68e3594badf621c4f545c01c377c6b799"}]},"branch":"refs/heads/master"},"c6b65c3cd05862ab816d0fa500f997b57518b926":{"kind":"REWORK","_number":7,"created":"2025-03-13 11:21:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/7"}}},"commit":{"parents":[{"commit":"09ba4f99b8b48828546005ba2ebb2255d06bead9","subject":"Merge \"[OVN] Retrieve the HA_Chassis_Group in ``_sync_ha_chassis_group``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09ba4f99b8b48828546005ba2ebb2255d06bead9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 08:39:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 11:20:47.000000000","tz":0},"subject":"Allow to delete singleton instances","message":"Allow to delete singleton instances\n\nNow the decorator ``SingletonDecorator`` forgets the instantiated object\nif all references to this object have been deleted. It is needed to\nkeep at least one reference to the decorated class in order to make\nany new instantiation idempotent.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6b65c3cd05862ab816d0fa500f997b57518b926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6b65c3cd05862ab816d0fa500f997b57518b926"}]},"branch":"refs/heads/master"},"172ace36f424121cf68132e357a0d9cca45c707e":{"kind":"REWORK","_number":8,"created":"2025-03-13 14:02:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/8"}}},"commit":{"parents":[{"commit":"09ba4f99b8b48828546005ba2ebb2255d06bead9","subject":"Merge \"[OVN] Retrieve the HA_Chassis_Group in ``_sync_ha_chassis_group``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09ba4f99b8b48828546005ba2ebb2255d06bead9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 08:39:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 14:02:12.000000000","tz":0},"subject":"Allow to delete singleton instances","message":"Allow to delete singleton instances\n\nNow the decorator ``SingletonDecorator`` forgets the instantiated object\nif all references to this object have been deleted. It is needed to\nkeep at least one reference to the decorated class in order to make\nany new instantiation idempotent.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/172ace36f424121cf68132e357a0d9cca45c707e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/172ace36f424121cf68132e357a0d9cca45c707e"}]},"branch":"refs/heads/master"},"e0841662b4738a5c7f4ee43635de9d481c24f294":{"kind":"REWORK","_number":9,"created":"2025-03-13 14:37:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/9"}}},"commit":{"parents":[{"commit":"09ba4f99b8b48828546005ba2ebb2255d06bead9","subject":"Merge \"[OVN] Retrieve the HA_Chassis_Group in ``_sync_ha_chassis_group``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09ba4f99b8b48828546005ba2ebb2255d06bead9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 08:39:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 14:37:23.000000000","tz":0},"subject":"Allow to delete singleton instances","message":"Allow to delete singleton instances\n\nNow the decorator ``SingletonDecorator`` forgets the instantiated object\nif all references to this object have been deleted. It is needed to\nkeep at least one reference to the decorated class in order to make\nany new instantiation idempotent.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0841662b4738a5c7f4ee43635de9d481c24f294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0841662b4738a5c7f4ee43635de9d481c24f294"}]},"branch":"refs/heads/master"},"1e560ecd8a2a620b7b6d677bae962764302136a0":{"kind":"REWORK","_number":10,"created":"2025-03-13 20:52:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/944099/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/944099/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/944099/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/944099/10"}}},"commit":{"parents":[{"commit":"09ba4f99b8b48828546005ba2ebb2255d06bead9","subject":"Merge \"[OVN] Retrieve the HA_Chassis_Group in ``_sync_ha_chassis_group``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09ba4f99b8b48828546005ba2ebb2255d06bead9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 08:39:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 20:51:44.000000000","tz":0},"subject":"Allow to delete singleton instances","message":"Allow to delete singleton instances\n\nNow the decorator ``SingletonDecorator`` forgets the instantiated object\nif all references to this object have been deleted. It is needed to\nkeep at least one reference to the decorated class in order to make\nany new instantiation idempotent.\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e560ecd8a2a620b7b6d677bae962764302136a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e560ecd8a2a620b7b6d677bae962764302136a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
