)]}'
{"id":"openstack%2Fneutron~944982","triplet_id":"openstack%2Fneutron~stable%2F2024.1~Ic2d471978ad12ef69d6324bef006aadda0ab8c13","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2103597","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-20 19:54:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic2d471978ad12ef69d6324bef006aadda0ab8c13","subject":"Remove the duplicated YOGA database milestone","status":"MERGED","created":"2025-03-19 10:47:01.000000000","updated":"2025-03-20 19:56:07.000000000","submitted":"2025-03-20 19:54:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944982-bug/2103597","meta_rev_id":"cfecb97999c32b8f3d125a27c58d57fd623d59be","_number":944982,"virtual_id_number":944982,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-20 19:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-03-19 14:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-03-20 13:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-19 19:42:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-03-20 13:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-19 11:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-19 13:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-19 14:05:43.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-03-19 19:42:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-20 13:33:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"3183ad76628bf32738b8133e306a673dc4b40d4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-19 10:47:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f50e9e77e1e81a0b9d7a160e2e4d92afe93415e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 11:35:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba07d30c0814bcdbfe59829b92d75a9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64ff3d30bc60406d964bb05024591aa7 : SUCCESS in 46m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8a95c628a4e48c78a396571c5ac5284 : SUCCESS in 46m 53s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3dfff22c68794ddf8385074c521d9692 : SUCCESS in 42m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7f091b2d044912142f9ffaceecdc2d10fdff352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-19 12:16:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3062cfe548266d3fb5220d8450b093ee2ee79cb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 13:26:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3742e235026418696a348dac03ac666\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38c962f97eb849f5bae9bd553184ea42 : SUCCESS in 24m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f9d2e73b8399469d88794e2dddf820ce : SUCCESS in 41m 38s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e2fe59ba596d4229a635f7b9b6a17cce : SUCCESS in 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb8d758b7c3790f6e3d7be895cee9269dd55fa8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 13:49:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1427ecfe5074c90bbc4761f7b4bedbf\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/0dbc4fec52374502bae5e3b69183ecec : SUCCESS in 59m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/f3b09414d535410490d3fa127c72e0fd : SUCCESS in 1h 29m 51s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/885818a1539b441180c67375a3affee5 : SUCCESS in 1h 25m 06s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/ea787a50d8b74a74a62c3b2a3e4902e6 : SUCCESS in 28m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f372c1da3329439d96d85b24bf6cc694 : SUCCESS in 35m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b21464a3ef44507b0bde5fa2453359a : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fc81d12c0e34fa7baa5a4ff0156361e : SUCCESS in 25m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2afbc584eb1846c8bea91a0059e2a826 : SUCCESS in 23m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b46dc9e4a16c4ca7b6344fd1bbaa0c94 : SUCCESS in 11m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aec7ca0ad3d54251bda239c84c31cd84 : SUCCESS in 54m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/77ea074a346f48d19192971cb14ce74a : SUCCESS in 1h 27m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c796593285647f1b848e82131e6e0e9 : SUCCESS in 57m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2379008f62714bd0b63a26c75062fce3 : SUCCESS in 51m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/44709bb79b2a447f87ddbee1ed3790df : SUCCESS in 52m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4d62037787924e1aa59c03879ba4a63b : SUCCESS in 1h 14m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2bae4e370d644ea4a54b3ac0dbd3da12 : SUCCESS in 1h 19m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3806bb1862a048549f878bd5274b52c6 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b71161dd656d667254de13bacef55d48b0b3037b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-19 14:05:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ab2620e9d126b50a1ea0d6cfc8b19ef9af82d66d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-19 19:42:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"22e5706c5afc363dcb114f99a7c43cf2ff07af98","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-20 13:33:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"618d3b390d292b51813d55528ada172c21d6e966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 13:34:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2bf2fb41025e2e19292036873749ffc513a70f29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 15:33:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e9521ed415b430d815bf3928adc7658\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/75a4da6eb87740f8ab09a0d80b44d1bd : SUCCESS in 1h 57m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e646ceb21f264c6986e1990a2df5a311 : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4380e64bf6c244db841a849e7f3523ca : SUCCESS in 23m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01858592da2e4c1a96f1018ec8b459f8 : SUCCESS in 20m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef60a1755bc04d8788ae72df18dccd75 : SUCCESS in 10m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/addf18a6dd594f218810040374695a0b : FAILURE in 44m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/67f3a081dda44fee96f274603ac44d28 : SUCCESS in 1h 30m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/49bd79593cf143ae9f159b10df42916c : SUCCESS in 48m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f00d52c742454d608dc4e15b3843ba03 : SUCCESS in 51m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f22de32adc4e498fb16391915b20814e : SUCCESS in 33m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/367bf78f8daa4f4f9445fb26975bac9b : SUCCESS in 1h 23m 28s","accounts_in_message":[],"_revision_number":2},{"id":"82b5c095dceb6822e9a1e621c489dd6b574052e5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-20 15:37:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"67507168edb9ea8c830342aa6503b50b005c82e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 16:26:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/899be4a515224735a46c902b99051aa2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff53980588e9447b882026af8c6873de : SUCCESS in 30m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c47e37a6a0b4e8083a1282a42fe93f0 : SUCCESS in 39m 34s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4f4ecbf7f19647f9b0d5bb9a6672610d : SUCCESS in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"75a49b9a7a3ec2e4d42243cc23f03d6f326d2d75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 17:25:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41a8103049464504818572487e1a2808\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/cec77a3b2dec44828b2b13278337073c : SUCCESS in 1h 07m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/4c97f29b5cd042b280ed9e2ac040918d : SUCCESS in 1h 11m 23s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/0e9db2e00f634fc1b23346be16a1505f : SUCCESS in 1h 29m 08s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/30a92db764d04de992dc6fa3899b79e6 : SUCCESS in 27m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71bb3eea45d14d8eaabba14091e98dd0 : SUCCESS in 37m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239b8abc851f4aa19ed400c87868f657 : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/220bceb02f3d415ab213b3234ad002ce : SUCCESS in 27m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34d8c4f5986649139bed9979a929ce78 : SUCCESS in 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9da9c0754204ec48cf649f99210db67 : SUCCESS in 10m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/661c29077a8a445686df9648e5bf3eee : SUCCESS in 1h 02m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4d77dfe69cb4e5dbe08b37e2edd9fcb : SUCCESS in 1h 43m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/54a72155965e4e8bb8830bd263f0966a : SUCCESS in 49m 48s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9474895fd69b41fbafe3a3c53a783cad : SUCCESS in 56m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e5023fc6719f421fbb2fc58704283720 : SUCCESS in 1h 05m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f86d2a2aa62248bfa12f0a45bba0f27c : SUCCESS in 1h 25m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/243b6495be224c6b90bed8e7e24297c6 : SUCCESS in 1h 15m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/14617e8b769f4c6fa3e18875c9a4e33c : SUCCESS in 1h 30m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5d33e3aba8c63acb87a3a10c091c7e9e41d4c77a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 17:26:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a5ba851951a2993acaff32e1ed96b6bceb22f51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:54:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca0e72c8392c4c859c0e8fb4903fc734\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/b1a75df646304d04870bfde26b05f973 : SUCCESS in 1h 37m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e272681d1ce4a5aa075ce0b8703b3e0 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a404b735a99c43f2abac2d965736a359 : SUCCESS in 25m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2215c4c8da2a4a8e96dde28c2540ff37 : SUCCESS in 22m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47c5e38654247168a6616a554b1bc01 : SUCCESS in 12m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7477c1ac1d8643eca1227957fb005191 : SUCCESS in 1h 17m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfe7462690d64386835803da2c654912 : SUCCESS in 1h 35m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e8175ce489d4d968d897d9edfc45b6a : SUCCESS in 53m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9c4e6a33769d43bda94f40f36b72540b : SUCCESS in 27m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3ef75f196e77457d9e0c8bd5a8ee6e0b : SUCCESS in 45m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c72d7702dfdf4c1e8f42af2af8cee852 : SUCCESS in 1h 12m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1813e55696cc397381c7f4077e072857682f9db6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:54:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cfecb97999c32b8f3d125a27c58d57fd623d59be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:56:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d5568fc6b7d4e90b96e4696fe7f5e49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930eb2e789e94a02ad0973e20f535712 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5018cdf988283a892ba581fafcad72105d513b20","revisions":{"92fc3e9f09d00f41b434e2b915f8053a631016e1":{"kind":"REWORK","_number":1,"created":"2025-03-19 10:47:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/82/944982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/944982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/944982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/944982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/944982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/944982/1"}}},"commit":{"parents":[{"commit":"e10d7cf7bdced4f24096657672289df73d460a54","subject":"Merge \"[OVN] Do not delete twice the agent from the cache\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e10d7cf7bdced4f24096657672289df73d460a54"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-19 10:32:10.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-19 10:46:55.000000000","tz":0},"subject":"Remove the duplicated YOGA database milestone","message":"Remove the duplicated YOGA database milestone\n\nCloses-Bug: #2103597\nChange-Id: Ic2d471978ad12ef69d6324bef006aadda0ab8c13\n(cherry picked from commit f6d8f2568ee6f9da23c1332b07f7712565f3b601)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92fc3e9f09d00f41b434e2b915f8053a631016e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92fc3e9f09d00f41b434e2b915f8053a631016e1"}]},"branch":"refs/heads/stable/2024.1"},"5018cdf988283a892ba581fafcad72105d513b20":{"kind":"REWORK","_number":2,"created":"2025-03-19 12:16:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/82/944982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/944982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/944982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/944982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/944982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/944982/2"}}},"commit":{"parents":[{"commit":"e10d7cf7bdced4f24096657672289df73d460a54","subject":"Merge \"[OVN] Do not delete twice the agent from the cache\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e10d7cf7bdced4f24096657672289df73d460a54"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-19 10:32:10.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-19 12:16:45.000000000","tz":0},"subject":"Remove the duplicated YOGA database milestone","message":"Remove the duplicated YOGA database milestone\n\nCloses-Bug: #2103597\nChange-Id: Ic2d471978ad12ef69d6324bef006aadda0ab8c13\n(cherry picked from commit 6c6759c3b4c86f0eaaeb32c19fa672bda7d9d312)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5018cdf988283a892ba581fafcad72105d513b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5018cdf988283a892ba581fafcad72105d513b20"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
