)]}'
{"id":"openstack%2Fneutron~947080","triplet_id":"openstack%2Fneutron~stable%2F2024.2~I960a734f364959c391ecbd6b630f997de4ce49ff","project":"openstack/neutron","branch":"stable/2024.2","topic":"bug/2088423","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-04-14 18:15:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I960a734f364959c391ecbd6b630f997de4ce49ff","subject":"[OVN] Isolate test_ovn_client* Functional Tests","status":"MERGED","created":"2025-04-14 08:40:20.000000000","updated":"2025-04-14 18:17:44.000000000","submitted":"2025-04-14 18:15:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947080-bug/2088423","meta_rev_id":"251000fd2404267e3deecf39007c7ff67502d7a9","_number":947080,"virtual_id_number":947080,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-14 18:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-04-14 08:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-04-14 15:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-04-14 15:39:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-14 08:40:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-04-14 09:24:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-14 10:14:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-14 15:39:02.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"0691df6c5c372079d07cbcc220c564bcb8af197d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-04-14 08:40:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11fb641c3458f258431e910eb464138038254bc0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-14 08:40:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98d04fff5d4dc416dd54cc1f0ffbacc0c1f47ccf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 09:24:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/854f78e038f74cb18742912de3feae3d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a44a3b0d0a8f481db73c666da67fc4f3 : SUCCESS in 41m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b4ff89c4c5744ac192090c02cb559b06 : SUCCESS in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc22d5f183bf4e350c58bc9aea27be2547adea3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 10:14:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8d8eb5ea7624c4c9fc3ac21174292b1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/def1453bb1e84fa09f889d155eb8af57 : SUCCESS in 43m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32b1f705f1b4d72b6de65b850751f5f : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3dc3b4c1fdb467aa9c5ef69baf0c81f : SUCCESS in 16m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc5ae2fea5e94da2828499cd96212290 : SUCCESS in 17m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9336bdd3ee8844fa950d7eece884b14c : SUCCESS in 17m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4f6bc0337e43bb8a4aedf7256896fb : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c4a3ab8f88f4238a48e656e5ccd13e6 : SUCCESS in 11m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c35df46ff6b4bce8ebe969a52e7244a : SUCCESS in 1h 22m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/007c4ad86bb241159bba7130e2b941eb : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":1},{"id":"173c02bae86b2627ef05bf99666f3a588c8694eb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-04-14 15:39:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c57246eaef32c619e186a9d357709c9d2684904","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 15:39:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b74526f3b2636f57b298aad68e5894bdaf737100","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 18:15:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd659854f20f4180a5107fbcb4afae48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abcdc03cb4a94eada2bd6eebeca2f3dd : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67bc8cd29b364589abf1bc6da2dc56a4 : SUCCESS in 14m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ccdd3730f194b9e9cc3ada39666f750 : SUCCESS in 13m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e500b047a424df58eae91aba4815a51 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79991004dfb84e008078dea02bbb8133 : SUCCESS in 8m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b962735306343c8b3d3f4b547ed786a : SUCCESS in 59m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4fb461d3133a4cfdab8b1048b8f556eb : SUCCESS in 1h 46m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0e24b2d9330395013817c6f169c8c1c3192350d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 18:15:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"251000fd2404267e3deecf39007c7ff67502d7a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 18:17:44.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9db4f2d1ba1c43409e81816bbb01bee6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c9ee92db4c4cf98e9fe570f9af7a2d : SUCCESS in 1m 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e755e0a6ce374f64be8853a5b023a096 : FAILURE in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6a81bf181d003a4e6d35445efe332f33a8cc579","revisions":{"d6a81bf181d003a4e6d35445efe332f33a8cc579":{"kind":"REWORK","_number":1,"created":"2025-04-14 08:40:20.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/80/947080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/947080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/947080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/947080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/947080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/947080/1"}}},"commit":{"parents":[{"commit":"821e4f501bb18510841be238159778c5132516ae","subject":"If OVS Manager creation failes retry to set values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/821e4f501bb18510841be238159778c5132516ae"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-04-08 07:01:44.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-04-14 08:38:52.000000000","tz":330},"subject":"[OVN] Isolate test_ovn_client* Functional Tests","message":"[OVN] Isolate test_ovn_client* Functional Tests\n\nThis solution is similar to what was implemented in [1].\n\n[1] https://review.opendev.org/q/topic:%22bug/2088423%22\n\nRelated-Bug: #2088423\nChange-Id: I960a734f364959c391ecbd6b630f997de4ce49ff\n(cherry picked from commit 750ff49f6b36c3f90500de7a1d30723ba1613613)\n(cherry picked from commit 4dcb0394938ec56d60720218398b6cb90ba76a48)\nConflicts:\n\tneutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py\n\ttox.ini\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6a81bf181d003a4e6d35445efe332f33a8cc579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6a81bf181d003a4e6d35445efe332f33a8cc579"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
