)]}'
{"id":"openstack%2Fneutron~948280","triplet_id":"openstack%2Fneutron~master~I20a48ee4a085e786753f40db32000a3b2a3e9dfc","project":"openstack/neutron","branch":"master","topic":"debug-L3HA-fullstack","attention_set":{},"removed_from_attention_set":{"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2025-08-04 07:48:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I20a48ee4a085e786753f40db32000a3b2a3e9dfc","subject":"[DNM] Debug L3 HA fullstack tests","status":"ABANDONED","created":"2025-04-25 23:44:31.000000000","updated":"2026-04-09 16:01:51.000000000","total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32be3cde536d0c216efefa42ef5240dccdc76d53","_number":948280,"virtual_id_number":948280,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-05-15 22:34:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-29 20:08:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-05-15 21:58:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-26 00:18:10.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-26 00:27:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-26 01:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-04 07:48:12.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"e8774b3b6ea22eb45275b19f3d36b188eac5a645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-25 23:44:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c3cb0896c21220495d6e389e296dc88401b57ad","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-25 23:44:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"6ce6afb05c83b45b6b6399cf05f114af56a3345b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-26 00:18:10.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ea1e602b489cff7861cf69cbc3764d15938fc97a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 00:27:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c0513e4d12f47c281bc46663a7d6781\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b79f9d5b54746e7b799b30a2aa87120 : SUCCESS in 21m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0dc6933ce198470a912ed1b17a858e92 : SUCCESS in 33m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/25ea3379f20e445ba4c6bb9be9c15ccd : SUCCESS in 41m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9483e3ea3960a12148af8b5e66b546e472cb66c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 01:40:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c2030de4bda4a2385af7d28d8ba59c7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d2c5eda757e4498bbcd54ee43d79ef01 : SUCCESS in 1h 45m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c8db0e5198e149a69aab15cc56aae37b : SUCCESS in 1h 49m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/261f498cdfa5492fa64b6c36d8058c8e : SUCCESS in 28m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73280c631bee4a34841f0a784e7f0cb5 : SUCCESS in 29m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b9577c82bd40a898dcdc6ca8e6cd5c : SUCCESS in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c34947caff046c89025ca850e6d54cf : SUCCESS in 16m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b50c486d748b410d9017cb8c5ca883a3 : SUCCESS in 13m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8e111124c784015a1fe110afe5badf7 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca966ca0613a43eb9f53e378370fa553 : SUCCESS in 8m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6ed5dd39cf224251b0426c627fad9602 : SUCCESS in 59m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ada7843e284842e28f5389c896ec3140 : SUCCESS in 1h 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1e0d8a155a3c6230492796de94ed3996339d0228","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-26 15:52:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92d09c1521a947c2e224041cd85d6141eb21a914","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-26 16:27:28.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"5c443b6e2fabe45eb1afaa2b8c183bd70aabb56d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 16:37:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59d6695bdbe24651ab24623111e0b256\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c602d7eb118346dca0aacad50f060893 : SUCCESS in 36m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/26f5fbe1b9294da0890eac4cb84fd93a : SUCCESS in 33m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b8c9a52122ad448897ecd68bb3f0fda6 : SUCCESS in 42m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9725e9ee7fd0341b78638ec92f5eb81664bda6d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 18:15:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22681ba044d643d2b6b11038bbc7892c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d77b44f620514bdebfadc45577b3d173 : SUCCESS in 1h 20m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ffb5cd5b235c443bb22704d1e0ff396c : SUCCESS in 2h 20m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b6e9a3221804dffb6a444bd8a011adc : SUCCESS in 31m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27d860b14ae14e7d85f912370dd937c2 : SUCCESS in 44m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eae5c7942634d9db2a035122ec42ec4 : SUCCESS in 8m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fb2db322f1747f89175c1fe502a4dd0 : SUCCESS in 17m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8fb4c59bf1ca4319ae75b7f283b2cfe5 : SUCCESS in 15m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/637952703eb2428e896c79dd4342d0a2 : SUCCESS in 26m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/239447dd740f49b687d34c848ba78338 : SUCCESS in 5m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c4f0b82e666546318b2ec9005788bbf2 : SUCCESS in 1h 01m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ff34458204044af688fa7b132ffd033e : SUCCESS in 1h 57m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fbbe466972e7e340ffcb85de11aa241a714ad552","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-26 23:23:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d51672626cb033b856eb4067ea02a1947060a11","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-26 23:56:53.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0ad7d6b72862d17e058f387f9207841db63679a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 00:04:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09d250da6a134e5c9bec9b720ce01762\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/341e39b08d3445e6970dffc91f4cb1ce : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46283ad445594824a63dd8376302a05a : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cff6d83ab7d64eb7820241b22d7e73f6 : SUCCESS in 38m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f5eceb17de56bdef5c1e39db0e203c16cd9f678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 01:29:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d40981b392424bbd471facb5ffaf1c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b355f03aa7964e63b9d23910f62dd53d : SUCCESS in 1h 56m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/56400ca436734aadbee5be07d612cb21 : SUCCESS in 1h 50m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c042e9fedaa1412098b618d8df207591 : SUCCESS in 42m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db629470726c472a8f17ceb859d6e655 : SUCCESS in 54m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4a324e597b546ea8a956c2db7aa4786 : SUCCESS in 10m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adc77aee8d9649d8950c700891b9bc5a : SUCCESS in 22m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b93edde76b614a41b8b6e81065e9566c : SUCCESS in 15m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0927828ef3a347049a3da91220f59147 : SUCCESS in 27m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbcfc841b73e41638f3b39e1895caf50 : SUCCESS in 12m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/93407b93a8704cf389ed4757da68ae2a : SUCCESS in 58m 14s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/59bc4166c0314163a82f8a3884f27a0c : SUCCESS in 1h 59m 25s","accounts_in_message":[],"_revision_number":1},{"id":"05d5017f859b0654019536fc6ce256d2841f3aa2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-27 17:07:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0ed12faf53c86b872d91d376634fd2068a5f9e5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-27 17:42:04.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"094200f825704d8c06c84a62ea50c87039a5e2f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 17:55:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/792db97b970f41b99ac602988f934634\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1a54ab2b2cf46eba155d30c19c0bd80 : SUCCESS in 41m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6769e7eec23946c1880f67b5670c8406 : SUCCESS in 38m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/564486f8f9794c93acc6f40d1cca8bb6 : SUCCESS in 46m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fccd6d6c6ceff90c513de56ece24106d3ad696fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 19:04:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8f66d28d9de4243bd2952ad596cc41a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f603f6669eb4903821cf101449867ff : SUCCESS in 1h 04m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d2c973dd126b4389b78c5e694cef7c79 : SUCCESS in 1h 47m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b2ba759a4069464f827b3b8602f783ff : SUCCESS in 18m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4192de129c734f6cbe4f9a7db8f468a3 : SUCCESS in 30m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62f87e0c548746969355671278701c9f : SUCCESS in 9m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6aaf8e702bf343bf8f312e9e680f3781 : SUCCESS in 15m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/071e976bad8a44818628f8bfa5ed47d1 : SUCCESS in 18m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba21d0c959f34cd499a05837e9819296 : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee9f7238e0544a34b5de4e9f0e2a85fb : SUCCESS in 9m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4ef7626dd3bd42798b8cfb864636e955 : SUCCESS in 1h 09m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/325527452c2541a9819712f20fce65f3 : SUCCESS in 1h 55m 11s","accounts_in_message":[],"_revision_number":1},{"id":"75b79ed4bb26b0b0ce3095efa4cf1e93c97415ca","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-27 19:20:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09b5ba73f3c58cd2e4b860112cae139b3ffa4084","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-27 19:53:48.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"5a845b9a53d2a7a851a4025059d8b25919ec59c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 19:58:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bba0ff6e0fe046489b076ddff1c21b5b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f562c59a230f4174b9f6c3ede62af524 : SUCCESS in 33m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bfa0fba341f04c0281a3d44529642ede : SUCCESS in 24m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a05c347393d94178af82a6e263d524ee : SUCCESS in 36m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d2f9a480788dd0eddceb80fa7a55d47733a1f5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 21:15:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7934a09f77bc425da1514eb37afa81fe\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cbd6f9e143e3478da1f02303a42cfa85 : SUCCESS in 1h 36m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/685c4e28ac754d8f97f020daf81cfc25 : SUCCESS in 1h 48m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab5563d2011244e4894897c64df7e2be : FAILURE in 19m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b4681788cf64c89898c1b541010e936 : SUCCESS in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a019dc3846c74a8db6b540a67a8c9655 : SUCCESS in 8m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d90ea35da1274e9d980c98bba67b49b7 : SUCCESS in 16m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99051d110c4c4cba8ca33e77b6c6a9b6 : SUCCESS in 15m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/85be7eee38484ce2954d08a7f7e2745e : SUCCESS in 18m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8462afb694724007b51873126bb9df1d : SUCCESS in 10m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c258ce5fd5e440e7a2c2cebe114bea2c : FAILURE in 1h 13m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3d4c89ccdf0a4c4eb61fce1d73694f19 : SUCCESS in 1h 53m 33s","accounts_in_message":[],"_revision_number":1},{"id":"15d47ebbf461626a7d771b6dff4c1cec5e0c4a00","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-27 23:41:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ea9d767d85184a46eee065b5ebdfcac8263691d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-28 00:14:47.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"8559f24e02d81215a5c4b3840e72f39a2e6d77af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 00:18:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f0cab9f9384abc81f51a666409b0a3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f42201b8b7d8423793e8aaac218102c0 : SUCCESS in 32m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/478ce88d60054308a45c0a8a3dfda3f6 : SUCCESS in 29m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e4b510101b474111b02213bfb42402e2 : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"427ec4cc437bca19a089fdfd2709ff65121d96e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 01:44:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b878d72100824e068a6b32e67a2805d1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c59220d6810d4975a3e6564c8fba6974 : SUCCESS in 1h 54m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/13b7d59847bc4031ba5b2f8809ce42d5 : SUCCESS in 1h 47m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b491dd7b99784a0e81940ca39914b9bf : FAILURE in 41m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa5cb5aec1f440488c1c62161e63790b : SUCCESS in 48m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67884487c3c461cb2fe193e3c42d919 : SUCCESS in 11m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ace07a7d7764c24a2eae3bf0a89bb38 : SUCCESS in 27m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a08d4cf93d0455bbf2c143d9a4bcca7 : SUCCESS in 21m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/776cd7d6cfee41ba9b294ec09f2fd86a : SUCCESS in 27m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9dbeb840724cc897652f85cc46ac58 : SUCCESS in 5m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c8aa6c8d90544409bfc8a8c5a01f4e3c : SUCCESS in 1h 30m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c83ed0b6eb10456190621545e5ebb0d5 : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3187b973645727c00c7bb443ea257da5bd3619cc","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-28 13:50:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01ba00d09dfedfe5716832fa00fb7927dd13003b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-28 14:25:53.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ed4aeae60bfe6ded337d3e21a987e739db5b3a01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 14:36:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/606b30bf4a824334886cc3672f165516\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3080c78f877046da8267d6ca2bad557f : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ece05107b8ff4268be359ed023649832 : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af933f4443cc402eb65e4cba484c47fd : SUCCESS in 36m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5273bfe8542676230b410e09f3be5e176722b6b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 16:33:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2022de8a61f14999a4b8107d071fba26\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d45ec6cd77b64e43a5ab4c24bfaf4bb9 : SUCCESS in 1h 22m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/45b67bb9ad3449c1babd59cec5ce4d06 : SUCCESS in 1h 22m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44243eeb543a49439feb8a236635b95e : FAILURE in 16m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aad9cabb8d094a5abcf13686fa96cd0b : SUCCESS in 42m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ef5c698cd84ec0a68f8683d87c0a90 : SUCCESS in 12m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1efdfc49783b4e1395e3e3f79a5d75f5 : SUCCESS in 20m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c4ac0f776d945e08b7871da878e0604 : SUCCESS in 21m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52b1f284fba14743b907f668ad8135ca : SUCCESS in 25m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c86b1c60fca04834acd267a65488bb06 : SUCCESS in 13m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/42506b2411e246478d53366fb97006ee : SUCCESS in 1h 15m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fa2f02fbd7294e3caac495cbdfbba500 : SUCCESS in 1h 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6cd7943acfa185bd364d89d5363037c1921c5f63","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-28 21:26:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41a349aee68bb0c13591e3aa4e404f1942d02c40","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-28 22:02:14.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a85c3fbbd87dd4b9d75b3ebeac2ae4756cf1015a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 22:14:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1f4e372d6b4e3d85f65b2bf7f41d57\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b718fe5b7ad4d30a472ca0e430e03a2 : SUCCESS in 38m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d37056592f042f791d9ab49c58a2242 : SUCCESS in 34m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d7f67eb102e40288b536a3aad84b1e6 : SUCCESS in 46m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"962fe926c172e5a76f8e782039fb51651758136a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 23:20:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a63fe038ad54577a01467b387b07d4d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b14ef94014e8452aa444d415d0d65008 : SUCCESS in 1h 45m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9388361144794d769fe89b28493ea6ac : SUCCESS in 1h 51m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90b0d5a0345443f69eeed4d42352281c : FAILURE in 32m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e33b4ca058d4e1a9cd7814ece2e3ea0 : SUCCESS in 25m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640f0975d2944abca4703aeca009a20d : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cfb8741283e49eebec6d250b3a374ad : SUCCESS in 13m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e78aedf78ca749a69704267588c22912 : SUCCESS in 13m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ee7878d87694182844e7cacb994c132 : SUCCESS in 16m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbbe2691523841238b4a0bbf14d6c224 : SUCCESS in 13m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c3bb09eb87654435b9cc645314974092 : SUCCESS in 1h 21m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2efbbbc6cb0846a7ade97d0f7c4517ad : SUCCESS in 1h 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6933974fcc8ed36095ef2e21739e47280e920347","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-29 00:02:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dea6ed9f1049cd0d0067074517a74e6eb0609215","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-29 00:36:19.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"21666973ac99e55768b8fcd81bde62ca09c19f3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 00:50:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1548f106e50846daac13839bc1353b5d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f69d151b921a485ab01c5c3e7d84f61b : SUCCESS in 46m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9305c74196fb41eb8a039f90208270b9 : SUCCESS in 41m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c19018c24a36427e86626c6deee4645e : SUCCESS in 24m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e39e49ed7c27c888543bf17668f22dcc1b4d7af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 02:30:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c731cd7bfec6470f8aa519a5b4506c33\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/93c6d82ee7c646b7b0a03fb480a74229 : SUCCESS in 1h 48m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5f367b5f9a4f4073aff09e7a9069f983 : SUCCESS in 2h 27m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c7f89faec5024cc9a5fee21fcf775717 : FAILURE in 35m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/505209e92948488c8b1c545af424a65a : SUCCESS in 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba31a325a0f4116bd38e39c0a5fbd3b : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/639975c864b74c09819462a34d911a32 : SUCCESS in 17m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/744cc261cfef4b5e96b7f0df803e47f9 : SUCCESS in 13m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e65a19f91924aaea83776cb520c53a6 : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43f9980a96a450b9abec0e9d260aa12 : SUCCESS in 5m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fd5b9883dd004de1a205fb983d4eb893 : SUCCESS in 1h 26m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c8da0abf63634dbc84e61b57a9fa18a1 : FAILURE in 1h 07m 05s","accounts_in_message":[],"_revision_number":1},{"id":"31f3473cf69388dd024e6b02396494a824394800","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-30 17:13:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a5f568e982948d23c843e52ccabba30fd2318d4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-30 17:48:34.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"f7303d45822505e6686cb86da48396a2b132630c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:08:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa3c68fa394407c83075f93dba6bc3c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/594cdd9a67d84600bca42fd36ad69385 : SUCCESS in 50m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4661ca2c47154352a49a723dc4daba4e : SUCCESS in 45m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/73353bc38b2a4e06b6f3f519fedde7e5 : SUCCESS in 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acb67b65c3b8022b35b9a1530f0e47589d4e115d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 19:24:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbfd288db9ae480bba220d2072fd1475\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/86f84d17c59749779e8687324998f4ee : SUCCESS in 1h 23m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0c91a83a4f154bccb955eb8896775b73 : SUCCESS in 2h 02m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68c14c89024b4fc396bc805e0651517d : SUCCESS in 32m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71947d988f904cfda005b6555787cc4a : SUCCESS in 29m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81870fb9edcd4947ae60440fb857ef72 : SUCCESS in 11m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8773aa6097174c6eb2f70c30b25baa6a : SUCCESS in 21m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17b4f60a86ef46c6a99ab66c4c9f2b82 : SUCCESS in 21m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbbd3ce18d884f44a59a570a17b33020 : SUCCESS in 28m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f6f42e642947239ab56827db0b80d9 : SUCCESS in 11m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/38d12e1079ff495f9568a1d0f28db75d : SUCCESS in 1h 35m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8e14562153384c7c96ac4d77030cb801 : SUCCESS in 1h 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"961e1073fe3b92e3736e1569ca34636b391d122d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-04-30 21:37:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11463b51194b14933bc0dd9b2957dd8fa1ef7371","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 22:07:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8cd579c4b846e287335be115b0ce60\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7eecdd1a141b42e090f65d3e8f777861 : SUCCESS in 28m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/da8ffaa11e8440bebdb9ba2628370df7 : SUCCESS in 16m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2bb9fef2a1846a3b47366f890c626c8 : SUCCESS in 19m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"179af8b30fd730c822c871e1e74cc6707c393c24","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-30 22:11:37.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4229a4ac455cd64b4ff15dae2ed5f5e6f0c7c9d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 23:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c913256fc4b4190969a09437e5c480e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/450b1c7d183449d396ec6bc638f8c80b : SUCCESS in 1h 41m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/47ca1fab15644f2f97fbef35c0d4008e : SUCCESS in 2h 16m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c653943f36b4685be8696ab29d225cd : SUCCESS in 40m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e8eed8cc10444329ef9856b000b3082 : SUCCESS in 36m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fce96bc696745579e7d3576cb2bbecc : SUCCESS in 14m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e4e883794f24129b37adbc190a42cc5 : SUCCESS in 25m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24cab7ef661d4057a097e7fa0008ab16 : SUCCESS in 15m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52ae6248d2224ddbae2b8ceaba8b2f76 : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf3ea210d59439f8e19f23619e1c859 : SUCCESS in 9m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/97d0f15066d2405189c5bda7a7e4284f : SUCCESS in 1h 30m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3b1e433600dc4ef796e2a4a7c04c7484 : SUCCESS in 2h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"895dca0fd579a9a22901c1e7dc54c17a55d1d4c1","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-01 02:43:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71457cf6d79ba1fff016d24cdb3673161ba5dcde","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-01 03:18:27.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"43cd4263caefab667df35a91b31dedada1a02dbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 03:56:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04687c9babbd4207a21576c37b1a4279\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81dc2d3cdf5b4f969a1e77e48aaf22ea : SUCCESS in 1h 04m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d961fe9a3e744fbb493bbc8dc6f1f12 : SUCCESS in 1h 01m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9c0f8eb85b884e4f9a9aa0f1bb250ef2 : SUCCESS in 1h 09m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ad4b7f2b9d9753596f5a6b9d520559e794ba060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 04:46:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e593417ceb9844048460cf82181c503d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/05da9a7c9fdb44f0b90a9cdf764ba110 : SUCCESS in 1h 30m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fee7acba19f24e5ba2af9f5964cb182d : SUCCESS in 1h 40m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/28c9fc309e6d4489a46d9536ae29c962 : SUCCESS in 32m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ec39cce8acc4ccdb3ba5cd520b885de : SUCCESS in 45m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cecf7394ff5462eb33009d2e9c8b1a3 : SUCCESS in 11m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98a4b3c2e836474ba612d136a486589a : SUCCESS in 20m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77b2b43bb60b4b9aa28ecca9e2b6ce9d : SUCCESS in 18m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/631c3f026cb34a0695f5fb6074b148c0 : SUCCESS in 32m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da9bc0b523e4892946d99c0475beca3 : SUCCESS in 11m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4b274f11a5e946199111ea3ec4f95097 : SUCCESS in 1h 25m 10s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/36d65346638d476080939ba46f26ec67 : SUCCESS in 1h 57m 48s","accounts_in_message":[],"_revision_number":1},{"id":"aba35238443e524b55d18a26c419e432533c201d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-01 13:06:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23349c10e9434433b669350cc8ed198812ba7ef7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-01 13:40:19.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b6ecb812a25cc6460ea753765b994b3676e533d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 13:46:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/591ea875efe7469786e967498ae40678\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78f3ce40c3c642bea2583d48793d87a1 : SUCCESS in 29m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d82a0576b0594bf6891de356e359a12b : SUCCESS in 32m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4488e6634d2b4f0ca398cf1a9d85d00d : SUCCESS in 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b478ee81fa94c997f04040c55729d3782f4d4ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 15:28:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6b5077cdf44459bb82c426cc32b1c7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/22a5a0ff896d49ca8e166ba461c360b9 : SUCCESS in 1h 25m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3beed7d760c64f28b6ac3a898c6cbddf : SUCCESS in 2h 20m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8f23657465142969de1ec11d9f8b9e7 : SUCCESS in 33m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cab83d4cfb6242dcba174ab1dcee0e1d : SUCCESS in 47m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa9775eee3e4c0f9cd8d57df695262c : SUCCESS in 11m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc3d778d047649f9bf236381cebd59f5 : SUCCESS in 17m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57f5ed503cb44280939e1e63c7afd5cd : SUCCESS in 20m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2720fe7a4b5443e7becdc0835cc985ad : SUCCESS in 26m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e736902ce9c438bbc92e65bf6441797 : SUCCESS in 11m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd0cd706b1434f95ae1d631d5de84bc6 : SUCCESS in 1h 22m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b8a48b2a73e94ce39db2bb1be55e8571 : SUCCESS in 1h 52m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c093d48bc896acedec65809ddfef1c3c49e3ed6f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-02 15:24:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2a1a6621369ed90be4ba3d2d36af762f803260c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-02 15:59:21.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d9df4147d357deefae46e3e78f6e28be52546ebe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 16:07:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88658cfc518a46b4b4899a52c423edec\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4580553dcc804ad2aa38043c7d98064e : SUCCESS in 35m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/48f95bbdcf544700b45ae9af4ea40bde : SUCCESS in 33m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/09b4cd7206d747518b9c08d675d4d664 : SUCCESS in 41m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7946a3300418ad009a3ecb3621bf12613a9ed1ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 17:51:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/464d08d114ff47a78d13098230d9c6d5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f031764c1e754dd184739a5d93b65106 : SUCCESS in 1h 29m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/15debe98221b44bebb1488cb3f9f7cde : SUCCESS in 2h 19m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e9a5f9c604ba49f18bb0bbcbb3acfc2f : SUCCESS in 37m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d9a9273dc35416993b43451e99cee1c : SUCCESS in 32m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af64ada87924b4f8a837b9030a81973 : SUCCESS in 9m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aaf05a12b4f04478938af0484fd336ca : SUCCESS in 17m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a329943021884b2e9c0276cc3ca1cdac : SUCCESS in 20m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2772e2b15dc544648bb7bdd476bd83bc : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131437f5be7f4c438522edf20517c439 : SUCCESS in 10m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6a4cf2e3641949e6a79252defbf44e2c : SUCCESS in 1h 26m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/32719f937f4b43aaa34b60b1ec4806ce : SUCCESS in 2h 02m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a3d9d0db45d414b52958db262ac81a15111c375d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-05 00:17:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a83388303ce4f7b293a8e0f3dcd636c7f3d85eae","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-05 00:51:55.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e1d45abcc87cff90250bd8dbe764712a6cc7d8ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 00:56:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf0c72e3611e48a6ac5f986ac85eb6ae\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e7295d3fc94d4732897ba729ede72d13 : SUCCESS in 18m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ebca3d097c245d18b952c6165cc83b8 : SUCCESS in 29m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad2538c55629423e95d8bf3642047476 : SUCCESS in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b11dc7924e85d57023fec3e74cd6fab3a4a7fae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 02:15:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9489abb09bc3448b932f768ef769b92f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/20956ae56c01453e8b7f231f9c4e6451 : SUCCESS in 1h 22m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d819f8e0d12407da93ae9c2bc070fa4 : SUCCESS in 1h 52m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2d6572b1799e484291506bb4a3f97381 : SUCCESS in 34m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/059921b4f1e84412ad950c46594e585c : SUCCESS in 42m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23379316073a4ee595f3faf66ebc9567 : SUCCESS in 11m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a734785aa714e3f9c042145e21d48f8 : SUCCESS in 16m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/553248e2d7e54966a960f25edd756cb6 : SUCCESS in 21m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad820c2f4f04425f918e6b8286cdf50d : SUCCESS in 19m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a46bf6454834a8a850739f9ca99386a : SUCCESS in 11m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f89a73bfcabb4a1e812f4132326978e8 : SUCCESS in 1h 12m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0c45a128895d49afab4b34e055fff1a0 : SUCCESS in 1h 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8d230f2a9d45312e93a836f4309a386ab63f25f0","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-05 13:56:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a7bd07802928c8302da5a3dc9b6829e3cb725c5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-05 14:31:16.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/80/948280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"feb06723108d19ef70e3dea4844a502355df9ec6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:16:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c7effe6bdb434896b6a17989f30976\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/06e444a327054961ac387028e7eca88b : SUCCESS in 29m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e5887a681354fefa486c164201ce59e : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ef907b93d81640da9d8d46b8f0f825f0 : SUCCESS in 53m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8367189ce699b589b7e192f047e444103030deb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 16:24:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dc70b7ef04a49499914470d7b7b3c60\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2b915c82e9594e86998f8994b5b58ae8 : SUCCESS in 1h 33m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/70209d5724754de39180fa2a3378dfff : SUCCESS in 2h 26m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a3e6772ccd0417099c48f52ec0031f8 : SUCCESS in 34m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a83a47f4e7d4c5ca1988b28b500a166 : SUCCESS in 46m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987d7ccb92be4952baa8204d5d98a64c : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc4be552a738492289a9a5009b7897db : SUCCESS in 23m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f91cb10f3c15402fb26e4e166dc072d3 : SUCCESS in 23m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c2518a17f804c9ebec7fec2791cfb45 : SUCCESS in 23m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a7f7369d0b47cb88c6db49616bb8d5 : SUCCESS in 11m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/08dfed3f18b445ee9722c0fa569d58fe : SUCCESS in 1h 26m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e06c363b399b477f9b3ddc895f479701 : FAILURE in 1h 34m 57s","accounts_in_message":[],"_revision_number":1},{"id":"909798ef77a896f9fce2f91038c3056d47e05184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-08 21:53:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c9053003f35eb71b88e7f19133dcb4c840887d9e","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-08 21:53:42.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"111c8ed5db6c2f180898efc22435e15541520327","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-08 22:28:39.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/80/948280/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"bee1434b8ae0ff631a18e17779e78a6f74594f94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 22:37:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d7a8af571fa4d1482dd6ac5f981742b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c30ec9ed83c466b9f475696cac0ed7d : SUCCESS in 35m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a2db3d707a54678b144647eddc7ba1c : SUCCESS in 34m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/24b6f8afa902493ba2621a08c4db2ae4 : SUCCESS in 42m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"437c9967ad7f81746c4c1ec734889efc01cdedcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-08 22:38:25.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1b789ba417e962784350253dcacfe3d11ef9e3d6","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-08 22:44:44.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"2fc13a8df7abb86e1b4bc4752e9f30d859b80186","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-08 23:13:53.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/80/948280/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b3816742cba1aa294575c41eac564bf6e8bc44a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 23:19:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6452093c7ef4a5cb41221dc6aa10a0c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6994cd39bfbc4224ad6e59ac5e41e21c : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/13e0cb66d3524f66badbf2f973677f39 : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6dbfa96c0854b038f5f07a5e1994a5b : SUCCESS in 39m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb4755e0d3b34fdddaad246b54c606e8424282b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 00:40:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b38ba729c7074657b204f8530b871a39\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/177d846baa9b4f67a3e911f928790fc9 : SUCCESS in 2h 00m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5e45244ac55c40dc9137828a39f4fabb : SUCCESS in 1h 47m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5591e6496ba4441fbd0eb5f0cc824654 : SUCCESS in 32m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23190f684d594ffa8fc5c572d0f9f697 : SUCCESS in 32m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e037aa7e54c468a91b799098aefcc90 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5858ed9c04942f09ea5ab5284ed992d : SUCCESS in 18m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48f3f220d908473dbddeb9f90b764a73 : SUCCESS in 32m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcd630d6ec0b4de38564dce73680ca2e : SUCCESS in 22m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64edffe3aa55484ca0f65b7d020fd121 : SUCCESS in 11m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/123be44248fc4eb2821e273e06248349 : FAILURE in 1h 19m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d638303e32fc4c6389e24055a20ff62a : SUCCESS in 1h 59m 17s","accounts_in_message":[],"_revision_number":3},{"id":"c82b19a6dc3919ba17735e933c9f21e9000184b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-15 21:58:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"063a8668d7e62707946211d7424cb7507a259f1c","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-15 21:58:36.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"f8b88f9cde561ff55e9917ecdfc6852afccae86a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-15 22:34:22.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/80/948280/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"efb4e7f62b7acfbe2d7813c085753cba507ff3fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 22:34:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1202fc61334f4971a02b660cf4ed2ea1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2fcbcbb3f56a4cae90eb091373a27fd6 : SUCCESS in 26m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b672eb94e4064328aa20f202b310bbd8 : SUCCESS in 20m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/235d5c472e8a4ce7920d2cecd2efef43 : SUCCESS in 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dfb616e0d1f75c03f09b5ec58a53a775d8ba2476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 00:10:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eacd15414064452099f6d1681b31a3d0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d19b3cb4be76423f9ff1b6f962ab1c57 : SUCCESS in 1h 00m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c7e9119633164dc78c3e63c70a7c7492 : SUCCESS in 1h 43m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1a232788dfb24f96b581d12a39c339dd : SUCCESS in 21m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9bda50de40f47f2be12565a3d687df6 : SUCCESS in 44m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b7a8d218c94a2f8db366da96e6ed00 : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a6f61c03392428894ce89ab7b0e6c38 : SUCCESS in 17m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/240f969156324a5687186aead3cdecd2 : SUCCESS in 21m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c317f464eb944f74843b351fd81431c2 : SUCCESS in 19m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5eecdd2c743488e85d7329c75e6b015 : SUCCESS in 10m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/58b3517133dc4f33a3bda88306f12dfb : TIMED_OUT in 2h 12m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5e5e0bbc39d9404ca2df818c0553bd73 : SUCCESS in 1h 59m 06s","accounts_in_message":[],"_revision_number":4},{"id":"d5436e938f9adc271683f5d9b84a3cc1a43a5a77","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-04 07:48:13.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":4},{"id":"f3f3b7e14c8cef5f0d97fddab29ea4d4a58f6f3b","tag":"autogenerated:gerrit:restore","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-08-29 17:40:49.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"46461eab7bc68ed0e84c1164c27a567ede801062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 20:08:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aea2ea3f571a49768816c78470ab07b1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f6791decde444124a248929be64a9a78 : SUCCESS in 1h 21m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3d33d0d6b8434461b0eff4895d395082 : SUCCESS in 2h 25m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7a7039b4ed4b4b1889e3b1ec88044573 : SUCCESS in 48m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ac181b13b4e41f8a3feb861d3824375 : SUCCESS in 34m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f31a6ca4e74a6092fb4962e367c8a5 : SUCCESS in 14m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8153dfeb7e504fd08003d65ac3ab807c : SUCCESS in 16m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95f5bea68e024c3297e2027822e2e1de : SUCCESS in 15m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/488220e8f90c48649d6ef244bd37f147 : SUCCESS in 27m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06bf8ad27f0a491ca8af61e98b04d687 : SUCCESS in 13m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d1d0bf4019c84480884a1983f37bfa74 : SUCCESS in 1h 42m 48s","accounts_in_message":[],"_revision_number":4},{"id":"9fe69a5e0086fe91114f2930c1fcbef5f3200e66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 06:33:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407555624fb44aee87f3b8371ac391a5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa797c7d30b44f8db35763a9838edd84 : SUCCESS in 34m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/902ec464d9524fb1adb1fa2e066aa5d9 : SUCCESS in 45m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/38e000f920d74f808234ce370e65ac8c : SUCCESS in 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"32be3cde536d0c216efefa42ef5240dccdc76d53","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-09 16:01:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dcf0079a1df1dc88e98b4eaa6835fd111a1d465d","revisions":{"5155faec3ae815cf8261bda1abc7f4dd9cda18c0":{"kind":"REWORK","_number":1,"created":"2025-04-25 23:44:31.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/80/948280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/948280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/948280/1"}}},"commit":{"parents":[{"commit":"9112672da6a6269e18e4f6a36211b732aed81575","subject":"Merge \"Unify ``MetadataProxyHandlerBaseSocketServer`` class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9112672da6a6269e18e4f6a36211b732aed81575"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-04-25 23:37:57.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-04-25 23:37:57.000000000","tz":-300},"subject":"[DNM] Debug L3 HA fullstack tests","message":"[DNM] Debug L3 HA fullstack tests\n\nDebug L3 HA fullstack tests\n\nChange-Id: I20a48ee4a085e786753f40db32000a3b2a3e9dfc\nPartial-Bug: #2104019\nPartial-Bug: #2091351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5155faec3ae815cf8261bda1abc7f4dd9cda18c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5155faec3ae815cf8261bda1abc7f4dd9cda18c0"}]},"branch":"refs/heads/master"},"4e85d14d7fc4611270faeada004482693f7926de":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-08 21:53:10.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/80/948280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/948280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/948280/2"}}},"commit":{"parents":[{"commit":"0faf89c5af5a44827c3b69c7a331010ceabbec35","subject":"Merge \"[OVN] Add a Logical_Router_Port HA_Chassis retrieval method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0faf89c5af5a44827c3b69c7a331010ceabbec35"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-04-25 23:37:57.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-05-08 21:52:59.000000000","tz":-300},"subject":"[DNM] Debug L3 HA fullstack tests","message":"[DNM] Debug L3 HA fullstack tests\n\nDebug L3 HA fullstack tests\n\nChange-Id: I20a48ee4a085e786753f40db32000a3b2a3e9dfc\nPartial-Bug: #2104019\nPartial-Bug: #2091351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e85d14d7fc4611270faeada004482693f7926de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e85d14d7fc4611270faeada004482693f7926de"}]},"branch":"refs/heads/master"},"e7f12157b60bcb10ddeffdf5c27332edd956c81d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-08 22:38:25.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/80/948280/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/948280/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/948280/3"}}},"commit":{"parents":[{"commit":"aa64bd2651c95e9a88d79f43ff473906ad81b005","subject":"Merge \"Fix DHCP extension if subnet has no gateway_ip\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa64bd2651c95e9a88d79f43ff473906ad81b005"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-04-25 23:37:57.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-05-08 22:37:22.000000000","tz":-300},"subject":"[DNM] Debug L3 HA fullstack tests","message":"[DNM] Debug L3 HA fullstack tests\n\nDebug L3 HA fullstack tests\n\nChange-Id: I20a48ee4a085e786753f40db32000a3b2a3e9dfc\nPartial-Bug: #2104019\nPartial-Bug: #2091351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7f12157b60bcb10ddeffdf5c27332edd956c81d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7f12157b60bcb10ddeffdf5c27332edd956c81d"}]},"branch":"refs/heads/master"},"dcf0079a1df1dc88e98b4eaa6835fd111a1d465d":{"kind":"REWORK","_number":4,"created":"2025-05-15 21:58:12.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/80/948280/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/948280/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/948280/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/948280/4"}}},"commit":{"parents":[{"commit":"1bc4b526e9c743423069ab4cf6ef3883d5e48217","subject":"Merge \"[OVN] Create a HA_Chassis_Group without raising an exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bc4b526e9c743423069ab4cf6ef3883d5e48217"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-04-25 23:37:57.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-05-15 21:57:54.000000000","tz":-300},"subject":"[DNM] Debug L3 HA fullstack tests","message":"[DNM] Debug L3 HA fullstack tests\n\nDebug L3 HA fullstack tests\n\nChange-Id: I20a48ee4a085e786753f40db32000a3b2a3e9dfc\nPartial-Bug: #2104019\nPartial-Bug: #2091351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcf0079a1df1dc88e98b4eaa6835fd111a1d465d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcf0079a1df1dc88e98b4eaa6835fd111a1d465d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
