)]}'
{"id":"openstack%2Fneutron~948313","triplet_id":"openstack%2Fneutron~master~I3374196b7ef0f37057b0ca05e1b8642879138cf5","project":"openstack/neutron","branch":"master","topic":"bug/2106463","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-26 05:49:52.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-06-22 18:55:07.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I3374196b7ef0f37057b0ca05e1b8642879138cf5","subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","status":"MERGED","created":"2025-04-28 07:45:00.000000000","updated":"2025-06-26 10:08:03.000000000","submitted":"2025-06-26 10:06:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948313-bug/2106463","meta_rev_id":"79b5ee2ef47d2d743206fd9dcbcac2dbd54e39e1","_number":948313,"virtual_id_number":948313,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-06-19 10:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-26 10:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-25 13:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-06-22 18:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-25 13:12:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-28 08:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-28 08:55:07.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-28 10:19:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-30 03:02:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-20 01:24:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-06-23 08:45:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-06-23 08:45:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-06-25 13:12:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"8e5e826f47c75311dcfa9cea7257d99a11509ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-28 07:45:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8257bd4bd3187dc6016fdcfe41104201f3150a8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 08:29:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2687e4a15f7647aa8744c9a00d22067f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4276ac3bbf3445d892f5d30acd36f799 : FAILURE in 37m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c73f4c133ab4d38a11020594fc4b1e0 : FAILURE in 33m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0ef5c7cdec6041a8a19a5693c725b09d : FAILURE in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d14308530d26f00749df7cbe9c6a04c0a584b75","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-28 08:55:07.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/13/948313/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ad1813b071645e6d83b6ce5d1f57c21f49e98853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 10:19:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9a28593047d4c2eb59a91ffdfc02e7a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e40e435cf348444ba77434661ddb636c : SUCCESS in 1h 49m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b69fe8420bc44247ab18ca0f89d70697 : SUCCESS in 1h 47m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2c5a2ae9cb440d1b788306a3271b4da : FAILURE in 42m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2cce620ed3949c99567f40691707ec8 : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/473b614d02cf45d8a8d188b5e6d0ac51 : FAILURE in 22m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3f467d759a848cdac22553f228c35d3 : FAILURE in 22m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92598737c4964ee8bb888c34a9183da0 : FAILURE in 16m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc761c9bf20a4156a4809610b5b8d983 : SUCCESS in 12m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b380dda1dc2d44069a4b101cc10ef62c : SUCCESS in 57m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bcc946b289b04e1c83cd38152ecf3fb8 : SUCCESS in 1h 53m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/efed10a4447f4f29abc7ff725d5e97a4 : SUCCESS in 54m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc3082e316594526b6f8d86963a3f313 : SUCCESS in 1h 07m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a8258cc8fd2d4d11b2d297a346d76ffa : SUCCESS in 1h 30m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6ffec7ac89814b3eb05e929a44c36a06 : SUCCESS in 1h 26m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5afc684b37f94330a9ecbf5d2b9adf12 : SUCCESS in 59m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6a79656581d84cbd8be2e153413b50fb : SUCCESS in 26m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/11d83c358ad741ab83e48283d2d14031 : FAILURE in 1h 38m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d0f3597cba094d38b480f9f20a823617 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6a376a32b302033b622e736f23f685d9be36804a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-29 21:22:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1859d9281cba62d127f08cdd302991fc8d3e40e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 23:48:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e47e8aa99d6045cdba5e574cda530377\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23c76d0698d2417f971f336e5c59ecd4 : TIMED_OUT in 1h 20m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/68ebb3ce7b6f4a5594437ccbe0e1d5ba : TIMED_OUT in 1h 20m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/41a4e217e2c44e56af01fcba31be23a3 : TIMED_OUT in 1h 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22029a0f618609a4b6baa4ea4318d14fff365368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 23:52:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21869c7717174f328def8551ff5c4ff7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/da33d048541d429cbda8dc8f7dc7a2c0 : SUCCESS in 1h 45m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/28020c7588ed4fdab748142ae48ec472 : SUCCESS in 1h 51m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c717b8e3563a4e92af941ec876355d20 : SUCCESS in 42m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7d2c1cc58d14415adc0d1b81a7d08be : SUCCESS in 8m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15abcd81b68240d485030d68f338b7e1 : SUCCESS in 22m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed0c198f4b3e4218b9f0de75834e2121 : SUCCESS in 24m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a83d739f1e154eb08e505bb09e5cc32f : SUCCESS in 15m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd6103341b64d6cacdf71af5f54a95b : SUCCESS in 12m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/053109a55c0546ccaf74cbfc1b33978b : SUCCESS in 55m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3b37c04b9b544f15b669adfd3c0c4746 : SUCCESS in 2h 14m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fdfb510e6ba4b9cb8ee6ae3dd4e0327 : SUCCESS in 27m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7947b5fbf2ad44aa8794df2635f6f436 : SUCCESS in 55m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4e8aa64a6ea4437c93874addf4a3dcd9 : SUCCESS in 2h 03m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3d222ad3c43941e79e7cc0530b0b963a : SUCCESS in 1h 31m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d543fe5d64a1457b88ea2d4eefbe1c26 : SUCCESS in 58m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cc24e1c92c2840289277121e30ce4e71 : SUCCESS in 58m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9ac63dc6e38b465aa78e3111823fd7e2 : SUCCESS in 1h 13m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e7a7c2db20e146f191cfc9712b3957f7 : SUCCESS in 1h 12m 28s","accounts_in_message":[],"_revision_number":2},{"id":"6708cb35d0769379610228e9159b333254d14550","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-04-30 03:02:28.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2025/Arista_Tempest_Testrefs/changes/13/948313/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7732ab5bfa2233042c0e804018a9687feff478a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-16 05:50:48.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6f131ac78c8965eded41a44dd42530af2f76ad23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-16 05:51:42.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c6cadc7d65cbb59440f20f2e5cd551492914dd00","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-16 07:44:08.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/13/948313/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"e3ddf6de48f65857a2d8613371eb20b0dd448140","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-16 08:19:19.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/13/948313/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"289c2b5ae33bb11a78509d73fcc065a5a629a967","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-20 01:24:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c4c1666a249e610c6f955c3138821e2818b79900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-11 06:55:29.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8db84c0a3d5c2a9cf2c2f423eb83fb79d43c5d3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-11 07:14:53.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"5f0a1aa3af9edec60095f51aee03f703d9d769e6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-11 08:43:53.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/13/948313/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"eb65f8c8f7e0c2c6ac3ca8892420023ab9e4ac37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 08:55:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf20819b7caa4ab3a7bb57dff6fad748\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b6a1f65518b4033902943e3c7e2413d : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/069ea06f65a041918aad06c4738a6f9e : SUCCESS in 59m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5dc906fe7b454c8cab38ab3e9dbf6fd4 : SUCCESS in 27m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ff75c2a298841ed371d59cf220bba38e6e038394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 09:16:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19e0fa5d42904c0f9265c4eac6c70956\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cc052fe2e0a34d0288199bbdaab6a5ad : SUCCESS in 1h 18m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca5c52fcc04e48eeae61b6a0aebe33a0 : SUCCESS in 1h 59m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09b0914ea12742e5b202fe8692a6c136 : SUCCESS in 30m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc9868eecca4bab95132bd819affa47 : SUCCESS in 11m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0babb93a4ed4589ac06da7bcf0ca3c7 : SUCCESS in 16m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a92f93dc366c403da6f1158911e20c13 : SUCCESS in 18m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1b5906832aa43759977bd6275eee2a7 : SUCCESS in 19m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60901f244d4945dba483f049dffdf64d : SUCCESS in 10m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bcff59b97755406fa18d8a0d86b4ceaf : FAILURE in 21m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/eae65c7e5ea94617961099bf1f354dc6 : SUCCESS in 1h 31m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5f05687cb104dfeb8ac335fb782dc87 : SUCCESS in 57m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/67e59d8616224ca8b0174cf33e1edbbe : SUCCESS in 1h 11m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/df88c113e86347c29d897d6633ade432 : FAILURE in 1h 13m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/29f07bf52da9416b955120d1b3ba652a : SUCCESS in 1h 25m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/178c8836e32f4375829c3a8c5142dcf9 : SUCCESS in 1h 11m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5c6c30250bba4b3b9fe417482684fe4a : SUCCESS in 1h 06m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/055fdc7361ac4fe7913c4fc0ee09410f : SUCCESS in 1h 08m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/68944a4fdde9472796ed594b267058fc : SUCCESS in 1h 23m 19s","accounts_in_message":[],"_revision_number":6},{"id":"022f962c0e315c08e5269f83c858a26035bdbf2d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-11 09:47:05.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/13/948313/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"58f424685905ec3ea489d5dee1831b7844280228","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-13 15:35:02.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"31a82d90b9bda88397986a519f4ef786555a47ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 17:24:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05aeec628a2e4d158e8b06d6d0632dbf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f254ec2d3eb7444eb3abeb76e57bf804 : SUCCESS in 20m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e382a1e0814424297a980a6e9cd8485 : SUCCESS in 28m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6ad386f17b5f439cac1a7beb4670a61d : SUCCESS in 43m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5e94cea038ce26d52bf487bf1d8d45cc47b5c1ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 17:37:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23332f0ac46140b9af366c37fb97f1f7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4a45aa88033045b8a2cc79e84db884f2 : SUCCESS in 1h 41m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d53ab85857834552bd7c7b8351e4a9f9 : SUCCESS in 1h 59m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd88f2f0f1c447628acc037db188d569 : SUCCESS in 47m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eeef87571924ed38adef083c7819ab4 : SUCCESS in 11m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ba4bc0015c6401d829f4f628a5077d6 : SUCCESS in 19m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f36d8c37aad48d7b87eba838b2bc8c3 : SUCCESS in 19m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3685df8fe4c34326b7fc28111fba9c0c : SUCCESS in 27m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c80661fc43d948a487c983a14f7a88b1 : SUCCESS in 12m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8e12f67172e745c6995b5b83c78d8806 : FAILURE in 28m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/17fae3db8f1d4563ba49176cb3f02068 : SUCCESS in 1h 48m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d06b185fe12486b81fec97caaa33be5 : SUCCESS in 53m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4b8fa9de87f4835b235cbb853e6193e : SUCCESS in 1h 16m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/34a586e43e474a67a21fc4b9b8a443f5 : SUCCESS in 1h 39m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1d35d33d1b0c4b13b01fdfee6e02c78f : SUCCESS in 50m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/30efd670103b4cfd95fbfb08e32410cc : SUCCESS in 1h 02m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/85985410c5294b8ab707081bad604242 : SUCCESS in 55m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/23c93a9e847c4dd3a129e8ee944fbb44 : SUCCESS in 1h 15m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f72aea9c5bc743e7a0731d3c8bba54b9 : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":7},{"id":"279dfe5a5093eb9718d81d3c1ce93f6a2ed458cd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-13 17:55:46.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/13/948313/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"e88820d1a15254cc177740bf0df5ee045785e45b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-18 11:04:39.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"eab48b252c7882146bdd966d6af8331f8c2ad197","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-18 11:47:41.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/13/948313/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"6ed883a6726256b1da44f3fea8a69b26dc32668e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 12:28:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/478143b8b3aa4df1ba6fbd3dcc4bc1ee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b2bbb13dd902447cb2f119b1ecdd30fd : SUCCESS in 25m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c5dcad6550c4612be6b095b59610932 : SUCCESS in 30m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/85a0bb3cd5d44868880f23781ac80917 : TIMED_OUT in 1h 20m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4d3e4c25b8c4d19e18f02956972f7851b4f6ac0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 13:48:31.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96a02ea7bac94940bd42e6ff5906d01d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/67b00512d30145ed8a7af3591c00a047 : SUCCESS in 1h 41m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5b12a4972c5e496ea45e22bceedf47db : SUCCESS in 2h 42m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daf24827603141ea8404b88241d1b9af : SUCCESS in 41m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99a98dedfd14f76814b3711f7ae86bf : SUCCESS in 12m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/094106e9d99c4553902012e1ef860afa : SUCCESS in 20m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/928596ce9c1d4ba0adeb5873ad03bc6c : SUCCESS in 19m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/855f32b243f9423cac759832b80266ed : SUCCESS in 24m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac66c7df708d4ebc88f7506433c1698a : SUCCESS in 13m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/47fb614e02ad4f9c970dcb5d437df27c : FAILURE in 28m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a2a40dff608440fc9e5634ed2968c753 : SUCCESS in 1h 35m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c6bfd0c8dd04e819826f2e8838904bb : SUCCESS in 1h 01m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/90d7dff7ae894e529bb3a738230ed949 : SUCCESS in 1h 06m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e3b0182f0c2e4e019e314c14c0f3d0f6 : SUCCESS in 1h 42m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/46ed90f7d8604a6e8a375181da57e1cf : SUCCESS in 1h 29m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/32cd4041c18e4c3f983e7dd2af656180 : SUCCESS in 1h 07m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fcecd24c7b58467184342dd6dc4a403e : SUCCESS in 1h 01m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/58363718800c484e8fb28f715be81ca2 : SUCCESS in 1h 02m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ec1bf0ffc8fd4a77b76c326e47fd642c : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":8},{"id":"ced0a08392b33f6fddc65ba4e03954ca8f75b4ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-19 07:23:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"86fbc03f0f5eb092244e631c9fd842f331c366f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 08:19:19.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1509ee53da5448da9753462bca8511c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/976d8825d6df413a88b05bf9d2fcc886 : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7337bbc78e394a458e0fcd993233696d : SUCCESS in 27m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/82ceffc6bfa0459daea4a1011848222f : SUCCESS in 30m 44s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c8f96529e8952bcee397f8b9170a40b0121eb512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 09:49:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5034984eb6894c4cb53c434c437417f5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/50abf52ee58748689999731de85b75bb : SUCCESS in 1h 55m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/baf440c7fb06481bb682baaab5d7d796 : SUCCESS in 2h 23m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa7d79919e084abaa68731b2f587c613 : SUCCESS in 31m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f747126dad948e9b38f22170d05bb99 : SUCCESS in 13m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/622a7f51c13e4bbebbf6761973258334 : SUCCESS in 18m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08a685ae755246d5aa3ad67529e97b49 : SUCCESS in 17m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0dde828f96ce42129ca54fea1759d925 : SUCCESS in 19m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddf3d5ea8c9e4d3eb9840a722fde8916 : SUCCESS in 12m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9ccfdbb0c1574eee993d8ccd71005a8d : SUCCESS in 1h 36m 10s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f9fed94313e4413988e02c8a6862ab83 : SUCCESS in 1h 50m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/89f328261f4340eab24f96395bbb7bbf : SUCCESS in 51m 08s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a15dd246f324467da3649dac0a88a229 : SUCCESS in 1h 05m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dfdf1c5dc05a4f0b8efad9e7030842bf : SUCCESS in 1h 42m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3a6dd263f5954591920bacd9b72b33be : SUCCESS in 1h 34m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/64ef30924c0544fea88144dc6f3077f8 : SUCCESS in 55m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/74f3875a4edc4ed2a064fe7a48ff9a89 : SUCCESS in 59m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/224bb56cb4624125934e68c7e7b0d477 : SUCCESS in 1h 04m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/39426a1622fb49bebe2b49cad9c346bf : SUCCESS in 1h 12m 19s","accounts_in_message":[],"_revision_number":9},{"id":"ec2950b08daf9c036fa954f206554b5faabbabbd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-19 10:50:04.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/13/948313/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"a286434f3cc3b11563a5703b5d0e72266e55e54f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-22 18:55:07.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"393644242d20ca400d13d18bf8d42afc9d6b2c68","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-25 13:12:06.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6af8c057e60f0f035008a4dccc452f1534d34dba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 13:12:28.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"25cf2931c697a11debbfc25f87aa4a1347e98034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 15:26:42.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3480ec5d24c4651856c33e4e3855ab2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/141f39d4de2d40c5bbf2c76ce8c491a2 : SUCCESS in 1h 39m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fe0b3e6882804552b81b5209816c8430 : SUCCESS in 2h 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1a3e6a603942518e17e33e5caa5962 : SUCCESS in 10m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d368f0b0b01418ca630e939aa5b62e4 : SUCCESS in 21m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6691ef2389744e39e68a2deb3846e57 : SUCCESS in 19m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28d5625c3033497abc4e23f578570516 : SUCCESS in 13m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5a3341333b1a47daa9521c88d85b9566 : FAILURE in 31m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/eebc26b945b94bcca0e0f5907ef710b7 : SUCCESS in 1h 15m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b175122a1ecf452c93513302ef7f12b1 : SUCCESS in 1h 37m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/830a180738c14d6e93d321abedbcabeb : SUCCESS in 1h 13m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/76a2a31f7091411884aaf2c70d905662 : SUCCESS in 32m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dfd51e3f835f4298a7ecd1efcf321256 : SUCCESS in 1h 03m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c8e6868d515a468d80a461488e39a4bb : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":9},{"id":"81179c7ae4448d5846d90da30b1ab45e65c31166","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-26 05:49:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"53e2133db7a3ed0b48a8db5aa4899f83ee1846ef","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-26 06:23:50.000000000","message":"Patch Set 9:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/13/948313/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"ebbb50adfb28d98a8dbcf571e038f8257047dcb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 06:33:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a992a386bd47dd929897695a499b4f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d07f2d0af4324e45a9866318615a94c9 : SUCCESS in 38m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4865246e90e4bab96507bd757e3b498 : SUCCESS in 33m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/decf130513a4470ca4f32351d14cc338 : SUCCESS in 42m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a02becaf348a2cac4662117ef280b627762c9201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 07:49:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f888c2144ee4465da655a952024aac08\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c18f7564a9674a55b6eae0c2cd82722b : SUCCESS in 1h 34m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f23c92c239fd40f8acbd5f6107d49603 : SUCCESS in 1h 58m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5de4f835b5b44e17b6f7fa40ea769d9a : SUCCESS in 48m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568e4f27835c43ab97768bcb453dd4eb : SUCCESS in 8m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e4f12b5218d48569e2dfc3b8a11ed2f : SUCCESS in 26m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc304b6e386a4cd998798c2b575efbc5 : SUCCESS in 25m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0826c5d684d14c199868dddad45c66a5 : SUCCESS in 28m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4e2fd8c8d24548b202b90d53db387f : SUCCESS in 13m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/085ef6ebf0944630869c36ab91e2a873 : SUCCESS in 1h 37m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7137f8b8d13d454390abd8740a24747e : SUCCESS in 1h 30m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e859aee4e82f40338c06f8c1a7d3937e : SUCCESS in 1h 00m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2a40d671bba4978ade6ca08e0fc5afd : SUCCESS in 1h 09m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/83d61af57ae84a9ab245e1433a70661c : SUCCESS in 1h 38m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f25139e806b2410091cf8d67ec12fd78 : SUCCESS in 1h 14m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/44ddd21801e14359bd2b6747f1a854a2 : SUCCESS in 1h 00m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e1375e50c2bc4d8792cff5ee46e69214 : SUCCESS in 1h 03m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a25020743fee4bc58b960d688099e486 : SUCCESS in 1h 14m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/95ab290f9f404613b8186a3e4d15faba : SUCCESS in 1h 15m 28s","accounts_in_message":[],"_revision_number":9},{"id":"7d85332b2f4762807d3cad217bffb25520ab9c87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 07:49:48.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0504d6fd02093b2f854491599be8af6fa3acbad6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 10:06:00.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d513d10ae5c436e93c6fb1e3e8c436f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d0dc95d290b84bde9173a2ca7407e3fc : SUCCESS in 1h 28m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/381e88ecea7849ff85cb275923f3b843 : SUCCESS in 2h 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd6247cc4f44fc98cf3652390ded232 : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/331d6b797223491baa56878ce0c40cd2 : SUCCESS in 18m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce686d5c5e04468ea8081b70dd71490e : SUCCESS in 20m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2902af614cc04b958013c1a3d351f392 : SUCCESS in 16m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a2c2e3014bdb467db4189e4c4e5bf4ca : SUCCESS in 1h 54m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/42cdb407436049619a67a70618f30428 : SUCCESS in 1h 28m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a0c1a2b62cfd4acd8567f1647ae5839b : SUCCESS in 1h 37m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/352f7cfb594c4a36ac57adad5e375440 : SUCCESS in 1h 07m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d844c9aab9cd46bda7f471044bfd1879 : SUCCESS in 51m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/265af36eaed3498a99f2378a36500038 : SUCCESS in 53m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/edb897732caf4b88a62596c775da78f6 : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":9},{"id":"66772e1d29478f46c49b058de3db11f8eb68bf92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 10:06:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"79b5ee2ef47d2d743206fd9dcbcac2dbd54e39e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 10:08:03.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/191994e5f8a7492c851914e8c7b01e8f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/568419dafcf54761a33e54f302f87b16 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3135b35f263ae50a16e7538e1d70fe7a7cbdd512","revisions":{"06736610f50055ad9aa303cc06828b6c52837c1b":{"kind":"REWORK","_number":1,"created":"2025-04-28 07:45:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/948313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/948313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/948313/1"}}},"commit":{"parents":[{"commit":"1c0abb81bb29b0b4c24c1fa550e55a5250d8e12a","subject":"Initialize the network segment ranges only in first WSGI worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c0abb81bb29b0b4c24c1fa550e55a5250d8e12a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 16:58:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 17:18:41.000000000","tz":0},"subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","message":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members\n\nMake ``_TunnelTypeDriverBase.tunnel_ranges`` and\n``VlanTypeDriver.network_vlan_ranges`` members private. These are\ninternal variables that could store the static configuration values or\nthe database stored segment ranges, but the values can be outdated.\nThere is a public method instead: ``get_network_segment_ranges``.\n\nChange-Id: I3374196b7ef0f37057b0ca05e1b8642879138cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06736610f50055ad9aa303cc06828b6c52837c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06736610f50055ad9aa303cc06828b6c52837c1b"}]},"branch":"refs/heads/master"},"f5529ee19401ce689f8be7dfc8b93fbfd1bfd7dd":{"kind":"REWORK","_number":2,"created":"2025-04-29 21:22:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/948313/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/948313/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/948313/2"}}},"commit":{"parents":[{"commit":"1c0abb81bb29b0b4c24c1fa550e55a5250d8e12a","subject":"Initialize the network segment ranges only in first WSGI worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c0abb81bb29b0b4c24c1fa550e55a5250d8e12a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 16:58:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-29 21:06:11.000000000","tz":0},"subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","message":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members\n\nMake ``_TunnelTypeDriverBase.tunnel_ranges`` and\n``VlanTypeDriver.network_vlan_ranges`` members private. These are\ninternal variables that could store the static configuration values or\nthe database stored segment ranges, but the values can be outdated.\nThere is a public method instead: ``get_network_segment_ranges``.\n\nChange-Id: I3374196b7ef0f37057b0ca05e1b8642879138cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5529ee19401ce689f8be7dfc8b93fbfd1bfd7dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5529ee19401ce689f8be7dfc8b93fbfd1bfd7dd"}]},"branch":"refs/heads/master"},"2a460f41d2a874b3208185bbe62e3f2fc1571d67":{"kind":"NO_CHANGE","_number":3,"created":"2025-05-16 05:50:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/948313/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/948313/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/948313/3"}}},"commit":{"parents":[{"commit":"1c0abb81bb29b0b4c24c1fa550e55a5250d8e12a","subject":"Initialize the network segment ranges only in first WSGI worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c0abb81bb29b0b4c24c1fa550e55a5250d8e12a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 16:58:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-16 05:50:29.000000000","tz":0},"subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","message":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members\n\nMake ``_TunnelTypeDriverBase.tunnel_ranges`` and\n``VlanTypeDriver.network_vlan_ranges`` members private. These are\ninternal variables that could store the static configuration values or\nthe database stored segment ranges, but the values can be outdated.\nThere is a public method instead: ``get_network_segment_ranges``.\n\nChange-Id: I3374196b7ef0f37057b0ca05e1b8642879138cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a460f41d2a874b3208185bbe62e3f2fc1571d67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a460f41d2a874b3208185bbe62e3f2fc1571d67"}]},"branch":"refs/heads/master"},"02c0a13fc014735f443b03aacf0636a6e59a002b":{"kind":"NO_CHANGE","_number":4,"created":"2025-05-16 05:51:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/948313/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/948313/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/948313/4"}}},"commit":{"parents":[{"commit":"1c0abb81bb29b0b4c24c1fa550e55a5250d8e12a","subject":"Initialize the network segment ranges only in first WSGI worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c0abb81bb29b0b4c24c1fa550e55a5250d8e12a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 16:58:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-16 05:51:33.000000000","tz":0},"subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","message":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members\n\nMake ``_TunnelTypeDriverBase.tunnel_ranges`` and\n``VlanTypeDriver.network_vlan_ranges`` members private. These are\ninternal variables that could store the static configuration values or\nthe database stored segment ranges, but the values can be outdated.\nThere is a public method instead: ``get_network_segment_ranges``.\n\nChange-Id: I3374196b7ef0f37057b0ca05e1b8642879138cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c0a13fc014735f443b03aacf0636a6e59a002b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c0a13fc014735f443b03aacf0636a6e59a002b"}]},"branch":"refs/heads/master"},"1f963689be2def5744f94635e2975062f7118378":{"kind":"REWORK","_number":5,"created":"2025-06-11 06:55:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/948313/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/948313/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/948313/5"}}},"commit":{"parents":[{"commit":"97c72a58c1d1045a77a16ca36c6cdb9eefd9452a","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97c72a58c1d1045a77a16ca36c6cdb9eefd9452a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 16:58:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-11 06:55:13.000000000","tz":0},"subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","message":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members\n\nMake ``_TunnelTypeDriverBase.tunnel_ranges`` and\n``VlanTypeDriver.network_vlan_ranges`` members private. These are\ninternal variables that could store the static configuration values or\nthe database stored segment ranges, but the values can be outdated.\nThere is a public method instead: ``get_network_segment_ranges``.\n\nChange-Id: I3374196b7ef0f37057b0ca05e1b8642879138cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f963689be2def5744f94635e2975062f7118378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f963689be2def5744f94635e2975062f7118378"}]},"branch":"refs/heads/master"},"2d8fb47d67701676bd880b8ff45f6157eaa5bb32":{"kind":"REWORK","_number":6,"created":"2025-06-11 07:14:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/948313/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/948313/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/948313/6"}}},"commit":{"parents":[{"commit":"97c72a58c1d1045a77a16ca36c6cdb9eefd9452a","subject":"Merge \"Drop explicit dependency on python-subunit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97c72a58c1d1045a77a16ca36c6cdb9eefd9452a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 16:58:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-06-11 07:14:53.000000000","tz":0},"subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","message":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members\n\nMake ``_TunnelTypeDriverBase.tunnel_ranges`` and\n``VlanTypeDriver.network_vlan_ranges`` members private. These are\ninternal variables that could store the static configuration values or\nthe database stored segment ranges, but the values can be outdated.\nThere is a public method instead: ``get_network_segment_ranges``.\n\nChange-Id: I3374196b7ef0f37057b0ca05e1b8642879138cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d8fb47d67701676bd880b8ff45f6157eaa5bb32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d8fb47d67701676bd880b8ff45f6157eaa5bb32"}]},"branch":"refs/heads/master"},"b6f8a0b2a629732b0f0c9dbf4c56c5288e257789":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-06-13 15:35:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/948313/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/948313/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/948313/7"}}},"commit":{"parents":[{"commit":"9190d8988be1326ea44027fd63660dd4bd28c1ce","subject":"Merge \"Consider logging options when using OVNdbsync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9190d8988be1326ea44027fd63660dd4bd28c1ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 16:58:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-06-13 15:35:02.000000000","tz":0},"subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","message":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members\n\nMake ``_TunnelTypeDriverBase.tunnel_ranges`` and\n``VlanTypeDriver.network_vlan_ranges`` members private. These are\ninternal variables that could store the static configuration values or\nthe database stored segment ranges, but the values can be outdated.\nThere is a public method instead: ``get_network_segment_ranges``.\n\nChange-Id: I3374196b7ef0f37057b0ca05e1b8642879138cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6f8a0b2a629732b0f0c9dbf4c56c5288e257789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6f8a0b2a629732b0f0c9dbf4c56c5288e257789"}]},"branch":"refs/heads/master","description":"Rebase"},"46358c0919970baabf1c577926c951a4e79d6d09":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-06-18 11:04:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/948313/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/948313/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/948313/8"}}},"commit":{"parents":[{"commit":"67ad7818216b55746774360675acb979ce49684d","subject":"Bandit: add B311 to skiplist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67ad7818216b55746774360675acb979ce49684d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 16:58:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-06-18 11:04:39.000000000","tz":0},"subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","message":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members\n\nMake ``_TunnelTypeDriverBase.tunnel_ranges`` and\n``VlanTypeDriver.network_vlan_ranges`` members private. These are\ninternal variables that could store the static configuration values or\nthe database stored segment ranges, but the values can be outdated.\nThere is a public method instead: ``get_network_segment_ranges``.\n\nChange-Id: I3374196b7ef0f37057b0ca05e1b8642879138cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46358c0919970baabf1c577926c951a4e79d6d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46358c0919970baabf1c577926c951a4e79d6d09"}]},"branch":"refs/heads/master","description":"Rebase"},"3135b35f263ae50a16e7538e1d70fe7a7cbdd512":{"kind":"REWORK","_number":9,"created":"2025-06-19 07:23:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/948313/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/948313/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/948313/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/948313/9"}}},"commit":{"parents":[{"commit":"74954ccc0b378f947befae62e15b7ca97355f245","subject":"Merge \"Fix bad i18n and fstrings interaction with flake8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74954ccc0b378f947befae62e15b7ca97355f245"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-27 16:58:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-18 22:34:00.000000000","tz":0},"subject":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members","message":"Make ``tunnel_ranges`` and ``network_vlan_ranges`` private members\n\nMake ``_TunnelTypeDriverBase.tunnel_ranges`` and\n``VlanTypeDriver.network_vlan_ranges`` members private. These are\ninternal variables that could store the static configuration values or\nthe database stored segment ranges, but the values can be outdated.\nThere is a public method instead: ``get_network_segment_ranges``.\n\nRelated-Bug: #2106463\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3374196b7ef0f37057b0ca05e1b8642879138cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3135b35f263ae50a16e7538e1d70fe7a7cbdd512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3135b35f263ae50a16e7538e1d70fe7a7cbdd512"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
