)]}'
{"id":"openstack%2Fneutron~948779","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I3728aa3932931320d9ccbc24fb9f8b3533dffff7","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2076328","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-13 07:00:28.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I3728aa3932931320d9ccbc24fb9f8b3533dffff7","subject":"Subnet filter by \"router:external\" needs to be changed to \"external\"","status":"MERGED","created":"2025-05-05 07:09:18.000000000","updated":"2025-05-13 11:19:32.000000000","submitted":"2025-05-13 11:18:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948779-bug/2076328","meta_rev_id":"565a443e9679be62cb5c5cc2282e5e4c0c3730cf","_number":948779,"virtual_id_number":948779,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-13 11:18:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-05-05 14:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-05-06 01:09:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2025-05-06 01:09:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-05 08:20:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-05 10:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-05 14:02:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-05-05 14:04:33.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2025-05-06 01:09:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"f70782c5527d438a41bdd4261a58216912043efe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-05 07:09:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"454e18a9938ef0e9361267661b52a7783a8b9825","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 08:20:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b9aca7ce97b4d41ba89a3bc0dfa2267\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ed402431adb4e06bda85c7630abc3c1 : SUCCESS in 37m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/89dc68ee40a64712a82b082e8942cee1 : SUCCESS in 1h 08m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4bcc2ec6f77d50c202f01480d8494d71510aa2b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 10:22:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bf912c893e94a4db048df09e0dc9173\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/23abe63174024047850de94abb1380ac : SUCCESS in 1h 36m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/68949708806f4e9984e9fd1dae527531 : SUCCESS in 1h 58m 47s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/df06e23036504e52a45501b2498479e9 : SUCCESS in 1h 40m 32s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/4509d7232ebc44d6982e0812c03cebb7 : FAILURE in 33m 49s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d7f4aa2a92014ac2945284d9ae492d7d : SUCCESS in 1h 33m 00s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/1a00d5646d15423eacbbca4d4f60883a : FAILURE in 2h 33m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4814643bdedf4f80bb46b6c05b6b9baa : SUCCESS in 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78c2f765afc64c959ebd25207ed2eaa8 : SUCCESS in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f2cbfc5da6440d3aa1af4bc2479fc55 : SUCCESS in 14m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/655fc76bca1d49999fa75ad9a5eff618 : SUCCESS in 15m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f32626b1714779aa3550fc8c440e85 : SUCCESS in 5m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/19b16ca99c404fe5b5b1601341cf83f6 : SUCCESS in 54m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b5e8ec9070f740059732bb9606026a74 : SUCCESS in 31m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fa873e0b73e2471e9ad6e53de84e9379 : SUCCESS in 1h 34m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/301b9b8d01ca47b38115c5f86935eca1 : TIMED_OUT in 3h 01m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/df4fce698d4a44ada3a084940c490293 : SUCCESS in 50m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/85af316c77e74ed8b301c5e7fa8c7814 : SUCCESS in 33m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/14cf583a68914e50bf9c6a6c9a961f10 : SUCCESS in 1h 46m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0ea911f379874cc7a7f9928524b1b0d3 : FAILURE in 1h 05m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9b7f5f61ed0c47149ef3ee211bd3a987 : SUCCESS in 54m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc7854f67f1f4f4898ae5ce4f9e53b16 : SUCCESS in 1h 11m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d7595592fc7d4852a84f301a9483843a : SUCCESS in 1h 13m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0c5ff15cba254f0b84eeed6e2af663c8 : SUCCESS in 1h 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"54ad913ad85c8a54633dbd93d35f97af1ea2df4b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-05-05 14:02:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9623dde1ad4b435f45ed7035d5d7d7b24594d58c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-05-05 14:04:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a161bce32022dbfea029e3b8df62334ac079f4b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:39:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03aadc42fba046bd83eee466717bfa85\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c84c3bb39ce14cc59d28b4e36b1501c2 : SUCCESS in 49m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/447c78b8c07e4f75af0dc4a3eeaf9d73 : SUCCESS in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62059d3807127253e55894d9f6567faab11a1f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 16:10:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b0791a232e4f3dacadef0715354817\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/997e9ee69e94490f9d5bd407c07a7762 : SUCCESS in 1h 32m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/4c64d2ca586142229faf1ab110945339 : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/08ae290be7334ef9a2039637672ae285 : SUCCESS in 1h 47m 00s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/03b405febfed4bafb7dcbf5502c7186a : FAILURE in 30m 33s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fbe176c27f43490e91c3549432d6c853 : SUCCESS in 1h 20m 01s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d257246c274644e19103b750b7a0546b : SUCCESS in 1h 41m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56036d24511849c0a2358403aacc0e80 : SUCCESS in 43m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c4dcb3afa240bdb36fd02ed14ad56f : SUCCESS in 11m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e72250781a9d47eb846c0da525cc48f3 : SUCCESS in 40m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d872a113bf574fafbe047b3f69694266 : SUCCESS in 19m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71c77ce2377745f1a60ef509cb8f4a86 : SUCCESS in 12m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/51f960d2a1b542cc90ce4b228b305a06 : SUCCESS in 50m 32s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ceac20b4aedf450d9c0049d4041d0501 : SUCCESS in 1h 11m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e3655238807c4bc88032b0e649f997c3 : SUCCESS in 1h 16m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2c6389832a6f4bdcb366b5230a3a0c2d : SUCCESS in 1h 52m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac4b2bca200d4445938c617f5cb9c340 : SUCCESS in 52m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5f30576d3f24f3586ed860a5374bc26 : SUCCESS in 1h 12m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d76882ceb733468fa698540eea569ceb : SUCCESS in 1h 54m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c50144a04e544f86964174e373a4424a : SUCCESS in 1h 23m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b6f274b02df64e29a8395c4a6164700c : SUCCESS in 1h 00m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3af9d667ab5648108045cf8c93eef136 : SUCCESS in 58m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/54cf08ac3e094097aadea2fa66502beb : SUCCESS in 1h 02m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a39daf73b7034b279387420c54bf4ec3 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a894bb77a8b04260fcfe4a89bdc21be192562303","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-06 01:09:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a1629aee4d149cc6a02d1b525a811a2f1713caa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 01:10:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"252549e16c269fca7ec11f355ffb2a180a64b7a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 03:28:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/182198977c6c4b70b62d251c721262a9\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/00fae8f0be3742778ff7b79c869c272d : SUCCESS in 2h 17m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b393d6147b5b4325a069a5b44acc4e85 : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50c0d9dc7df347feaaeaf7e1b2f88749 : SUCCESS in 16m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c827e8a38bf424fb06de8d851ffc837 : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7056edac2ef4f62b415ba4fd65159fe : SUCCESS in 5m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/80a1ebac15bc4950ba773fadc463ea18 : SUCCESS in 1h 18m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3e67e915b9784060ada098fe2c70b923 : FAILURE in 1h 54m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c4461a40adc541169c72c9ce0f4cc55b : SUCCESS in 1h 39m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0411730bd3954d028d94ce4118b509f6 : SUCCESS in 1h 13m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf21ed4090b24e609590353ffb9241ec : SUCCESS in 54m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1769d6a607fa4c5584b6297c259c7a23 : SUCCESS in 43m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1152b92927ad4d6daa73b692d58cf2c5 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"50b76dadf85e410c6e49c6a14c6234b06e32c7a6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-06 03:28:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2d20f0156afc79999937b481f8ec4d2ae8ce9ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 04:05:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cfde1c324b249bda1d4c0592c793c83\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b653184beb6643f7a13be780e91cfe7b : SUCCESS in 34m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56d7075af1eb4d8483073f51f4d855a7 : SUCCESS in 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"839610469d616fe48e31c2f2141f17c0dd725484","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-06 05:44:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2629dccb3748f5c00565679d8cb807a65881fbf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 05:46:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3291c0772e924c68be2a4b97d5c9b521\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/fac799c4a688416589a1c37c0b3ead53 : SUCCESS in 1h 41m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/35a34ed02e0f4e7f816b572acb94212a : SUCCESS in 1h 32m 24s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c447b9e7e0eb40e884feed36a68631e7 : SUCCESS in 1h 41m 45s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/c1620c918fae40969e49bf1d45a4fc08 : FAILURE in 30m 27s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a7abcc89735b419faa0a6f897a10175d : SUCCESS in 1h 24m 31s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ab60d5120aae4b99b5fedc876be0aac2 : FAILURE in 1h 44m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad73f40951c3485dbda60d769fbd8cc4 : SUCCESS in 45m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/510ebf7cb26841cb9299e17b444f23dc : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2799c3fc85b40b3a891c9d7d438ac73 : SUCCESS in 25m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76d0621a7ce442b390342c29da3219a1 : SUCCESS in 19m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a8f3d5f5e944d496eb3f15b92579c6 : SUCCESS in 5m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c22148abae644e05a58b6f295a04d5f6 : SUCCESS in 37m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3b21d68085e3464c8b3727544965b204 : SUCCESS in 57m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0aaa4a61b023480288e03267ff3c1730 : SUCCESS in 57m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3d20b0f3cebd416a83b775c3c2ef17cc : SUCCESS in 2h 16m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/97e1faad22f94ce086ab9e129b863acb : SUCCESS in 44m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a566e63f1084d2c9cae7da2d73c564b : SUCCESS in 36m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/455a9250164c47c48902ad514a49a6b6 : SUCCESS in 1h 37m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9bb4f665d4c24c05bf74fdb692733353 : SUCCESS in 53m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f54e9152b68d4616afa2a009903a1c82 : SUCCESS in 1h 01m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a861224566b44681a344f4711d93f42f : SUCCESS in 58m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/09c13f2a47394d49989cc2a16aba0df4 : SUCCESS in 1h 18m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bf30b5e2a8a248dfb6045913740bc9ef : SUCCESS in 1h 20m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0af4f7dcfd33496631586926b8159019539d0627","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 06:20:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5aa5967006b4adab099eea721ef1477\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac82bc332aa34347a139e7103dc22552 : SUCCESS in 33m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9599be3605ab427e8fc9863f4982b2ff : SUCCESS in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2af791de4d272260e323489981f0ac6ec33c7889","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-05-06 08:03:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a44338e1b01982a3f786df918dfa35458b66067b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 08:54:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9fb462f86874ca7bf9d446244ddfd02\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90c15ae9ac6e42c99a3f07ec183a1bec : SUCCESS in 49m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57501d8e6fe949f0b61e8af0c0cb1aa8 : SUCCESS in 22m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"688c9971a30119cb8249bbf9efa3856e6f1f2d89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 10:10:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fdd6b025369467681d60d53efb737e7\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/0647edbde00e4f4b91e9f7ff8a1b9714 : SUCCESS in 1h 31m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/17d2484f1d6f439db87129b57e83fe64 : SUCCESS in 1h 48m 04s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9516d426807a4bd98d13f7000d648a05 : SUCCESS in 2h 04m 51s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/43ff34afc78840f8a84bc9cc165d91f5 : FAILURE in 36m 42s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e626370d79e547c4900d51d46cebb106 : SUCCESS in 1h 45m 27s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a8705e26cd5d4ddca67803a2fcd398fc : FAILURE in 19m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d6271f8dcd84f9c82156db6b4504f6e : SUCCESS in 38m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70eb3011332e45f188ed3400e812532d : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ab7e812030f4cb5af5ee445624f233a : SUCCESS in 21m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/354388ef4c4e4ce6a99ed67ce6a855b3 : SUCCESS in 22m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0325933321724989b24d28c407f1a10e : SUCCESS in 10m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c9d43da10ff245509b803f0a23e8cc54 : SUCCESS in 50m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1955b2fb2116407e8255128b6e7ca0e2 : SUCCESS in 1h 08m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/780c6bbff17d429fb14545f50f564d7e : SUCCESS in 1h 20m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e1bb7fc6f85b4ede88018e95152cd6a8 : SUCCESS in 1h 55m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a4b2bf84b474d528b7cdc254bdb4e5d : SUCCESS in 55m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/41d9d56d5c0b40bb8f0ae6896ad3af3d : SUCCESS in 1h 05m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a50707cb5ade4ace8b1ba85aa92c410d : SUCCESS in 1h 40m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/646154cf2e32478795f29981141d16ce : SUCCESS in 1h 43m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1eeda3ec48164046b43cc028b1e529bb : SUCCESS in 1h 08m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7684bde380984b38ba5aac77af463b92 : SUCCESS in 1h 03m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a4c1471153284e98989b44291f3f9696 : SUCCESS in 1h 07m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/03029dcf05734d86b75686ad0e8330b9 : SUCCESS in 1h 26m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a83faeaaff07b6b2ec53d972fc9bdc4aa5994535","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-06 14:14:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"140bcdaebe1b48216835043711a1abd3cde7af77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 14:44:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d7643b321d34e49ba0f2582371d6d48\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1989f79541bf4a57a496a3bfa865553f : NODE_FAILURE Node(set) request 300-0026948126 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1aae06cf1d0747c5b7ddc09e33a7a83e : NODE_FAILURE Node(set) request 300-0026948127 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67ad7a290e9ae7aebeea382a4e8f26ae056afe75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 16:31:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66b7fe0bb8014a6db837df0d8a844166\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/8cc42e8946fa42eca9c5a00568df91e4 : SUCCESS in 1h 19m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/dca303c1f94141628214b1017a821bdd : SUCCESS in 1h 29m 59s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cd40a4eacbf84d318cb0b503b98c8088 : SUCCESS in 2h 01m 10s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/4586590a7dc743ac9ed0e2d3e6c91956 : FAILURE in 29m 59s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b9e14ae5f2c348f9ab39384afecfb537 : SUCCESS in 1h 14m 00s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0e0ca1d6f7474bae8ddafdb89c2b2fa7 : FAILURE in 2h 15m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0414583ce353451a8789a7746a4990ef : SUCCESS in 35m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5cafbea50754a98999b12b0f8b68741 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab54177c7b674e9b9876183162b4e39b : SUCCESS in 23m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32f25b765c8a4f1faf22866d0acd6c07 : SUCCESS in 19m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f15de0214a44978ea7633a83287180 : SUCCESS in 11m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7154d434d8394f09bc75194ead951c8c : SUCCESS in 24m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b5a5cb7076c84d8e92b2747a9dbca231 : SUCCESS in 1h 03m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/65f9996aac614779aba4e1c3920d75e9 : SUCCESS in 55m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2cd20f562dac4c3781b6e499e89d21d8 : SUCCESS in 1h 51m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/26dafa6aaa56450dabd0c9f946cb4d61 : SUCCESS in 55m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc3906a561964611bfb2cf07355f8216 : SUCCESS in 1h 05m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6de853954d1d449cbae1970a502dd6e3 : SUCCESS in 1h 39m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b353fd2c5d084035afa6eae01a95f837 : SUCCESS in 1h 33m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ac0bf6dedf264f539e9997d71bc272c0 : SUCCESS in 53m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d19c5aaec75c4a858701e162dcf4263a : SUCCESS in 52m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6bf79ebc78ca4cb4ab68ba33fbd7100c : SUCCESS in 1h 06m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c828af53c58d4b25b6c72f5dac817404 : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a0605db7dc68fc2a29e7d968125bbf74534739a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-07 11:22:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34bb393571840fc23ac124c3325f6ceec9c1f61a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:57:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cbb6c188991476aa6219c699be45abb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe5016bc14554568824b5204c5e4080d : SUCCESS in 33m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23629a5687704b35b570163bc9236839 : SUCCESS in 30m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"008fb4110e79bf4c7f962467968ccf4ad83e5456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:07:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17000b8575034fe3b57d0dbad6cadbd3\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/fb28c2e293b341f2aab894068db05ec8 : SUCCESS in 1h 39m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/fe18204032d148be936fcef7690a83bc : FAILURE in 18m 11s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/1c74d784b83248daa89191b315f0689b : SUCCESS in 2h 43m 38s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/f2fa7953b6c945db8b6abb9d01a1e73c : FAILURE in 21m 02s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c940292b4b1d4d2ab30b4f32638fb99c : SUCCESS in 1h 42m 08s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/cccb3fb712b540699abd6c3c9f0d543c : FAILURE in 24m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/759c64d3a9914ce8a626c23d77068475 : SUCCESS in 32m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf9f5ac4059458a9951ec420d6e0a2e : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98b46068b0dc4d04a6ca3ebfdca37a1b : SUCCESS in 20m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75acee5720114143bbb8bb276d7e1e1c : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414652f993824771ac71ce74e9feba0a : SUCCESS in 11m 56s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/28bd301eb4da4518a6e36dc229fcf137 : SUCCESS in 53m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c4dc3384bb3343cdae402c2e3d7f64cf : SUCCESS in 58m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bd4444bd937f4e4989b64462bde69a9c : FAILURE in 29m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b077122a693e41fbb25dd606b3720516 : SUCCESS in 2h 00m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8516b524bcb24b37a152840b29ba76b8 : SUCCESS in 57m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d27d9ad0c69409a82681b102e763ab5 : SUCCESS in 1h 09m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/831d9528f8d440788912d6add44de3fa : SUCCESS in 1h 38m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3dd6ab4bc4c34342ba8d10caa0ac00d8 : SUCCESS in 1h 33m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/53aafb2f2b79443ba2945cc8717ef261 : SUCCESS in 52m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/65895065a88f4a5fb1cd24ca29bc28f2 : SUCCESS in 1h 00m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1476a817823b4f2fa5905f63430ff8a1 : SUCCESS in 58m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e433cac9c833432b9cebf4909a5477f2 : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":1},{"id":"102aae7fe676495c31bd6b216c30a878bc352331","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-08 08:33:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"476b48fc79af3db7a5744ddcc4376e3de59a8336","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:59:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbbb5970506b4e779769428095ff710e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0034fd7bae31432a9d01043e96e3b760 : SUCCESS in 24m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79147f39fdbb43bca8a7d813eaa76858 : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e43640722bc56bed466c095bd2934c983e9ecee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 10:43:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dd608efedb44b608dc8941b8f9cfc11\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/5ffe0c1375e349fda27fcf8bd0a27f40 : SUCCESS in 1h 30m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/e4bead8d37964c87a91e2344c085e02b : SUCCESS in 2h 07m 43s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9267a6ba082f4e53890e9a0e184a6cfa : SUCCESS in 2h 02m 48s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/f93e183687294b9a8859a544a2203951 : SUCCESS in 30m 43s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/08395c28953a4c5395938bffb8df393c : FAILURE in 15m 17s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/addce7e448fe4ecdba7071e64316526a : SUCCESS in 1h 42m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/379f9431ba1d4564bd8a6cdd54ebe38b : SUCCESS in 42m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e65ccf753dc4f55b2143e5734853ad7 : SUCCESS in 11m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dafd0b27b81c4a059d701a03b197b8b1 : SUCCESS in 21m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5551803dbf94760bb1372c9c2dac37f : SUCCESS in 19m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7131c130abe43e78489a7247bd3b50c : SUCCESS in 10m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f873210add714769be06a6b916ae3630 : SUCCESS in 52m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d5897a9f128742f6924195540d033a68 : SUCCESS in 1h 02m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/76c4cbf9d2c140fd8922f87556100804 : SUCCESS in 1h 07m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ca72340c6f9446669980cc08c2f50200 : SUCCESS in 1h 59m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/be8e80ad41ed4deaac0c3f3ccfe0f56b : SUCCESS in 1h 01m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e0fd1b3eb064164bc8c2dbc15a55fdc : SUCCESS in 1h 10m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/50cde089c5534abb82d505db41c3439a : SUCCESS in 1h 47m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2288b8798bbe403a85a573e0713230fc : SUCCESS in 1h 23m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3489724d54a2475d9d7104de7a8eb83c : SUCCESS in 48m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3f319a98f0034d5984e72153ce66661d : SUCCESS in 59m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/55433cb1f40d44a59aec64a30df1e9b7 : SUCCESS in 55m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3164eee2618345099f4d02630f903735 : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8043aa27f2c31e04ef0f68f50e9d703e97ff5a54","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-12 10:33:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b1fa8422468c5f6a906d2df1c5b12682c517808","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 11:09:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b277b692f724048a4767642841e957d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c189d937b614af99218cf11480d1f80 : SUCCESS in 25m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/574ed797e0fe4b7aadc77670c61b544e : SUCCESS in 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"326a81a70371e57aa8c4c41ab6a0b57a25bcaf35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 12:31:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c06d32356734019a986200f43a66544\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/0302f6e6274142638950e9979fdbdbb7 : SUCCESS in 1h 25m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/3db50a1647014f6a9f27220539f1b173 : SUCCESS in 1h 42m 02s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/45daf5f178d144df991bb4865a9cae2e : SUCCESS in 1h 57m 56s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/116e5af8813b418b94962c0d471b35e0 : SUCCESS in 24m 56s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/694c0e07b63940db8698f87f739c728f : SUCCESS in 1h 27m 28s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/495131e3467844a7919f7bf964f314a2 : SUCCESS in 1h 33m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22c5405a1eec4c69928b3a74f1dd3c62 : SUCCESS in 42m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bad05b201844194a661f71d92e884f3 : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ea71620b5994366afd4ba36996885dc : SUCCESS in 26m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9321da7ec0e46e38027f806d2b70755 : SUCCESS in 15m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc8d9c59aaa414fbc72844dcd4c2c96 : SUCCESS in 11m 30s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/54b6fb7f47ea4a028f82db829ee0b453 : SUCCESS in 55m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5ab50807416946ba89fc00e24bc2d112 : SUCCESS in 59m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cb8793fbe747493aa413009f9bb3ab2f : SUCCESS in 1h 16m 14s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8527e47b4f5843d9addd864ce02ca60a : SUCCESS in 1h 51m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/417e913107814964bce726f09d052cc2 : SUCCESS in 57m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c059b8a370de4fc0a51b5a3784b89f35 : SUCCESS in 1h 08m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/23c6484946fe435aac6b92651465e68f : SUCCESS in 1h 40m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a1e1178c6f94884b1ff0f2ebcfac96f : SUCCESS in 1h 15m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3de4ac182c98498d8ed607de190f5515 : SUCCESS in 1h 02m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6af6966d9c7946f69c0652b910759f57 : SUCCESS in 51m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8b6b99dea28b47a090112b64da4d28e6 : SUCCESS in 1h 13m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/036bd639be47443b9349fa1ecc7b8e53 : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b21394bcdd815f56c13454cc2be1aa106b0c7917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 12:32:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8692ec3535a583086387b7d12dddea4f43061a36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 14:35:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cc75c78e8c94a1f908821d0de624b3d\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0d2f4b6628da4772ae7acdcb35894e8c : SUCCESS in 2h 02m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712259f2029c4b61afd7c27d0e87f164 : SUCCESS in 10m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adcbd57156e14ba2b41d90b46456c52f : SUCCESS in 26m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b07cbbd79eb3432d86cc4cfe601147d3 : SUCCESS in 16m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9128fdf5b5401ea381313d97ac43f9 : SUCCESS in 10m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3ee62b20f14e4314ae5eef6d8b8182a7 : SUCCESS in 1h 20m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/93954faf0c0644b7886db0567f582398 : SUCCESS in 1h 34m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/66be916d9a024fd0abb9b32312828100 : SUCCESS in 1h 42m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b19b36bb471d438585e4abcd41bb8e5b : SUCCESS in 1h 07m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cff57f4c17bb4c4eb46edcf558f3153a : SUCCESS in 27m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0fbd809c9a1c4a1784b1d837fe1890b9 : SUCCESS in 1h 01m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f6899410f4f146e8a0776d5258c1747b : FAILURE in 1h 09m 32s","accounts_in_message":[],"_revision_number":1},{"id":"018dae251ed0aa32c31fed668779ab53792225e4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-12 14:52:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"400ca6399c0b05eb401479f06909a4960a2feeaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 15:32:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66e44451dd264b4bbfc7a0a11534e75f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f81997db56042829ece721ea0d046a9 : SUCCESS in 33m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ae0f736ce114d09be825521846618ad : SUCCESS in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4511243a9194320915206457df6659adbc9d4fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 17:00:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a765213f3f2c4bb4b7439d720dcb86ee\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/db0c8cfa11b341cf9b740825f15d4d5b : SUCCESS in 1h 44m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/9bf0533c099f495ab8ad276dbd80de07 : SUCCESS in 2h 04m 52s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9c3ac77979f2443094e824059c4f74d6 : SUCCESS in 2h 02m 39s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/c8931cc6b84341e9b8a4c31f1b2bef60 : SUCCESS in 37m 31s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/1401bb40835b448baf385b356b5fd227 : SUCCESS in 1h 37m 33s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a7c0ac7588f74847a0e9e75d61f4b010 : SUCCESS in 2h 05m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5f2487f05594987857380302d0ed7b2 : SUCCESS in 45m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6fc0f94ed2f4082a230095874fa282e : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74fb95124113473aba3ccf6480ea3b0a : SUCCESS in 19m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99413d6d58454131b2aed1ce8ab6a3c7 : SUCCESS in 17m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873d4ebc9a1c4e628a810539f27df714 : SUCCESS in 12m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/99e42da6bb994ef1b0ae065507aeff6e : SUCCESS in 50m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/26e194a1adde471bac7ef4b994591e63 : SUCCESS in 1h 03m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/737abb8b76e74673aa3b4b6125f0f5f2 : SUCCESS in 1h 25m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/abbfe47f0b1147d78636fd30852636ee : SUCCESS in 1h 58m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/86c1c0b79fd4407388b2546f1559b190 : SUCCESS in 59m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4409ba7821f4274870b4bfb364f226e : SUCCESS in 1h 10m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8afc6022fa794a58a221aaff19d46247 : SUCCESS in 1h 51m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aff3cd54052544b99a4ddd46fc508fab : SUCCESS in 1h 38m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/44688b3ee59d41ffbdcb61804dea4b3f : SUCCESS in 54m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ba124fe6f7674f6695645e457d5c6bd5 : SUCCESS in 27m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/baf79190a5e44c39ad67390ccca0fad1 : SUCCESS in 40m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0d29803c4f104c7bb8f1a7d5f4017263 : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"529878f346e3f18334fac4d2c9e297b3f0a2899a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 17:01:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"735e3127de2990557090b1dccb7694907cc80c8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 19:17:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e60dc24aab3646ecb1181cd12e181d21\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0ccbf245909345a7b74b99c1827eb435 : SUCCESS in 1h 55m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752478c399e0427392f603bce9315449 : SUCCESS in 12m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05b9707c2c8245df83ab744de9e54170 : SUCCESS in 20m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bd79dee5c8047e9a2961da0e11dee26 : SUCCESS in 19m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d54007931640ab887b29e964b39fdd : SUCCESS in 6m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/65b243a8fac74507a9afa16b4e3aa0fc : FAILURE in 40m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8175ef13acba478a91b5d507158e1dc1 : SUCCESS in 2h 04m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0c3ae9ac53454adb90a6034274379430 : SUCCESS in 1h 46m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d17921712d77433393682111b5fd1c6d : SUCCESS in 1h 11m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c572683270348fa9a0179d3535bc7fa : SUCCESS in 1h 04m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0c192b888d234252b23b0c728f54cc8b : SUCCESS in 58m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5d115b8f79a74c8da75daf09a0521aa7 : SUCCESS in 1h 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b35dc99d3bcfa6ec5318fba8df78dacd65a77386","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 07:00:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad7d36be0249d9b33a6f2ae185870bb63c70b3a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 07:34:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a295af46f143e88084f249f20ae3b1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8ac5a0778494a3aab9377c182edd47e : SUCCESS in 33m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d70dfb96ee6f44e48ae93c2f1a2fc863 : SUCCESS in 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97b44a9af226b5b7ead055146fdfe06a551b4783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 09:27:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa1b575054bc41abbcd607361138a746\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/40a856a7022c47bb934bb54b6475e8b9 : SUCCESS in 1h 23m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/5622284d8af348509093cf6b7e4d7b51 : SUCCESS in 1h 30m 53s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ac577ea831f74ea090bf75629de1450e : SUCCESS in 2h 08m 31s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/903d7eba593e4137844a26621357667e : SUCCESS in 49m 03s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e507bb30b44c4592abd2aa9c8ff463bb : SUCCESS in 1h 20m 00s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/85513283cb964a8f8fee350ad61472ed : SUCCESS in 2h 25m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d0eb456406b40529b2960384eb71cef : SUCCESS in 42m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75871152614d4b15b9c646581159b045 : SUCCESS in 11m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89f277ac6dbd424290096d79f4744646 : SUCCESS in 20m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05c998383f154b499d0be374b02dd529 : SUCCESS in 20m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051bfdd0ed374c2fbdcdbf55b4ada484 : SUCCESS in 12m 59s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ff305c559ffd4f4abccfcb5059514736 : SUCCESS in 52m 23s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6c12de02cd46431fa801748de444c4e5 : SUCCESS in 1h 03m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/58a54e79dbc54b5c9f801d084fb2bdb7 : SUCCESS in 1h 36m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c6987ba6680b405b9fdca132ace4da0b : SUCCESS in 1h 38m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d192934d0484410ba93dcc8610d25baf : SUCCESS in 1h 01m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/afe6d8a420124155bfb8baa2511e53e7 : SUCCESS in 37m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/613d40d768fd4047bcba9554d5d66664 : SUCCESS in 1h 45m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/819c452b3f904224979ccf8ae4423dc8 : SUCCESS in 1h 15m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aefd1ba1c99d4763ac212548c50e01ba : SUCCESS in 58m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/87a7495709844be099539fb2bec362d5 : SUCCESS in 1h 42m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b6456aec546a458295f80170da88d280 : SUCCESS in 32m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fcd53eb45b8e454eaa514446305062d2 : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f9c464d28e5afba83b2d4bb5fc0bc1d289b34fa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 09:27:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"903f798a011221c0e00deb784029110c26b90fd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 11:18:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a212e1c071d4e18a64518a91c9ac81a\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0a004c3cd95c467b88b6127a4b9b3f0c : SUCCESS in 1h 21m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/365e364d762540ae99228eee4334b5ba : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57fe562c9fbd475e9ab41563f54449e5 : SUCCESS in 22m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d0ec88a38324b88afdc639eca1f2e2f : SUCCESS in 19m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c521d3300c6a4c838e1975e87763de9e : SUCCESS in 5m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b26bd593e72d468ab2059c35eec55fe7 : SUCCESS in 1h 20m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7c58d458d4764a259bf077caa2525f59 : SUCCESS in 1h 49m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f005021b68fc43919620c02df278eb0e : SUCCESS in 1h 46m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/69958280f78040d5af0f062b2edb45a3 : SUCCESS in 35m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/302ad3cc4057412da5b31436822e4713 : SUCCESS in 55m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2952e0073e094ec0b7023d5dea65dbf9 : SUCCESS in 1h 02m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1372f102b4cb45ed94e5e776eb8b79c4 : SUCCESS in 1h 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"cb8ee51254b6778e0bce7a4eccf6e90ad9e7b316","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 11:18:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"565a443e9679be62cb5c5cc2282e5e4c0c3730cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 11:19:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e032ac4e00e44a9b6dd2cfce0d458aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83fa926b17b42a48a734972c2f5b933 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05a719a3fb6bbb73986767ae0457e55ba6533fe0","revisions":{"05a719a3fb6bbb73986767ae0457e55ba6533fe0":{"kind":"REWORK","_number":1,"created":"2025-05-05 07:09:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/948779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/948779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/948779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/948779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/948779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/948779/1"}}},"commit":{"parents":[{"commit":"b204a0656039ad65b12f16dd4a13d0f847319a95","subject":"Return a HTTP404 message when the Metadata instance is not present","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b204a0656039ad65b12f16dd4a13d0f847319a95"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-23 22:27:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-05 03:41:45.000000000","tz":0},"subject":"Subnet filter by \"router:external\" needs to be changed to \"external\"","message":"Subnet filter by \"router:external\" needs to be changed to \"external\"\n\nThe extension \"subnet-external-network\" added a new API field to the\nsubnet resource: \"router:external\". This name was used to maintain a\ncoherence with the network resource field, that has the same name.\n\nHowever, the Neutron subnet OVO defines this synthetic field as\n\"external\". This patch changes how the ``Subnet.get_objects`` method\nfilters by \"router:external\"; the field name is converted to \"external\".\n\nThis patch is being tested in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/926508\n\nPartial-Bug: #2076328\nRelated-Bug: #2051831\nRelated-Bug: #2082883\nChange-Id: I3728aa3932931320d9ccbc24fb9f8b3533dffff7\n(cherry picked from commit 7a490c855b7316f6e844ddc8a5b77d604a7e2da3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05a719a3fb6bbb73986767ae0457e55ba6533fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05a719a3fb6bbb73986767ae0457e55ba6533fe0"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
