)]}'
{"id":"openstack%2Fneutron~949136","triplet_id":"openstack%2Fneutron~stable%2F2025.1~Ic7807a440b479c0fa08c6d13813cd2a81eff9bb1","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2109388","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-20 15:22:27.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-05-14 09:13:03.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic7807a440b479c0fa08c6d13813cd2a81eff9bb1","subject":"[OVN] Only update the MTU of the router GW LRPs","status":"MERGED","created":"2025-05-08 09:26:24.000000000","updated":"2025-05-20 15:23:49.000000000","submitted":"2025-05-20 15:22:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949136-bug/2109388","meta_rev_id":"1f87093d767fa8b06613ddde313780b293f43194","_number":949136,"virtual_id_number":949136,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-20 15:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-05-20 13:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-05-14 09:13:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2025-05-20 13:37:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 09:52:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-08 11:09:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-08 14:44:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-05-13 20:01:17.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-05-20 13:37:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"b535ec108c06bfe4eca5241b1c4e63e7c7bbe97a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-08 09:26:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ed198aa66065c1f4e4e3e87508ca26f87811945","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-08 09:26:35.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"913c731007bf3e34e37b814f70a0ea84ed5ef341","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 09:52:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4365b2a7318845009e7c340c2a85682d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/406a63488d1442959be2dc758d856a07 : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/26c192e8d45c4bf88ff32bbd23b52cef : SUCCESS in 24m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5ab3b88745b30664a7ccc55d7ece001ddfad808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 11:09:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a94babb7648a4da9b0aeff2d2b598836\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/494288c3a489498ca7a1f6e038aaacd9 : FAILURE in 10m 38s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d2ded747c7ff49c9a805f8f1725bd180 : SUCCESS in 1h 41m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f5cceba1fc24c4680dff9d782aae622 : SUCCESS in 28m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d4ec236ab842bf882408ed28267cbc : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d4347d68cf74024b325b1382de6d130 : SUCCESS in 22m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff817f2cf46f41d5acb285d437974d1f : SUCCESS in 14m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc8561489d247918e6888db8b638763 : SUCCESS in 10m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f441255045fd44f183a1af64ded04706 : SUCCESS in 48m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/41066a8e1b8f45b4a3c62ad6651a0517 : SUCCESS in 1h 28m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae2c346eac40463b917d0fa8c3d9657f : SUCCESS in 1h 04m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9448a224a43e4dffa5cc3e38aca50e8e : SUCCESS in 45m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c711db8df39547c785d24aab820be582 : SUCCESS in 1h 14m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/17880e01d91545d0b5bdcad1bb5caca4 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b0a8b99e30e8169ac922af7504bef830ac478180","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-08 11:31:30.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"50582a42972863e38ec79d76d9db6c8df77422e9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-08 14:44:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96c25c0a27d384e6d20fe43f1e218f58628a522c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 15:12:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edeee64a09ff426383f5b58a44a56e89\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/038d2911e3eb4e2ebb2b4ea8589ed3a6 : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/872b260325624aadb0bc062882cd21f7 : SUCCESS in 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"babe6c7262e488d380bbf40711e6bf3b8667c4f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 16:19:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf18b09d47dc49edb812c824a5e43575\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d9a34330bc0d48c8955a1bff7ccc540c : FAILURE in 21m 28s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fd317e9142674ea1ac800fdf799cd752 : FAILURE in 24m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8983d4028a14fddb347eb78c095874a : SUCCESS in 35m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efcb21a1be174ed9a4e281988acdf4ad : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93ef54697a2e4660a195e3f68b2678bf : SUCCESS in 23m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e774aa24c7a4265a42ade7d59f67f20 : SUCCESS in 20m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7105dd71e7a740b79cd1cc32f2b4c96d : SUCCESS in 18m 52s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/55e7e1131efe40018d0e593da0b6653e : SUCCESS in 52m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/faca85544a6e4650b677022564872e68 : SUCCESS in 1h 15m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c52c19749fd24f59aaff0361d4b44455 : SUCCESS in 1h 10m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9e78fff777494529a308b36394bef293 : SUCCESS in 58m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/31be518dcd624898bdcb64b88161d1fd : SUCCESS in 1h 00m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2844d71e4be14eb2bda1f79facc1564f : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4bb0b2a85078a5129d6f391e379499524481db9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 08:07:14.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"71642361509a1674685f0a00302238cba1b31638","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 08:40:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f72ab7c415b74d649315dba045f6b265\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/416adbfe9ad2423289ec8c7614909668 : SUCCESS in 31m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f14617f7b2024a119f07f7f5d1364403 : SUCCESS in 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5408aca235a16ff9ee6b17236a469f9da94bfd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 10:51:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a2f159595ca45549448a5b9a2d6c676\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0772d5caf17e4835b9ce8fb8f2c9d96f : SUCCESS in 2h 19m 32s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/6219b56a786846bf956b9dfb80045d43 : FAILURE in 2h 43m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cffa56f0df4f460ca6bc1efd6fd9340a : SUCCESS in 36m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf40cdf572234413ac71e479bdcb8f9c : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8503f072ad4d4ce187dd80cf0a117cae : SUCCESS in 22m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d57f7919465644009db1b839e21a0152 : SUCCESS in 15m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6fb6ce8942416bb9eee8afcfe1c3de : SUCCESS in 13m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b1cced396b9249f4a79905ee49df527e : SUCCESS in 1h 07m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3eb6f16f59d943428f9e1ad830b5fc26 : SUCCESS in 1h 42m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/24661814e3854e59aca00b5d144092c6 : SUCCESS in 37m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ed45a0e13a154074a83b62175de305dd : SUCCESS in 52m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/20924865633d4ed397bc43df0157f87f : SUCCESS in 37m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5335b44c0ac7495dbfc2d08b4e9c81bb : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f331c6ea706102a18ae8ba91c2132dce4135e4f8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-05-13 20:01:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65292fbeb2fdc324230a01e7908f91df46eac601","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 20:39:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e33f77aa16d64ee1926292b28f8b3394\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4548f5d2fbb45678ffbeaeb27f2798c : SUCCESS in 36m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bba80676acd544a79fad3bdecb2a375a : SUCCESS in 32m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d57b60b5b1243031ab310d518be1c2ccb4e25b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 21:43:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7ade25d32e441a486cd811ad0701898\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/a265b901adf64758924c1500a83b2d9b : SUCCESS in 1h 36m 06s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/25126bc060d34cfc8be912a81e0e34a3 : SUCCESS in 1h 41m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a8f32172b804d77b438a7e62f33dae5 : SUCCESS in 45m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871b7239bcd5453786b85dc663838bfc : SUCCESS in 10m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31e37e11956342cdb9dc97c68a62226f : SUCCESS in 14m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7281a9ad9322410e9315b1c1b0343ddc : SUCCESS in 20m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ada96b24a0548eba38b0a210dedf7df : SUCCESS in 7m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3e3007debbe1453a9f0c17da0a8e4f93 : SUCCESS in 1h 03m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/32541a24a7ad4c9a9c05a44ffc822951 : SUCCESS in 56m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/50a7e76d76124174832e9b72f9f9e3a2 : SUCCESS in 42m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/47302dfb98f9412eb74fc99a671c610b : SUCCESS in 1h 00m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f7c372ac6834421cbe43297cb9275a09 : SUCCESS in 1h 09m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6a48ec8b98ce4e7ea66afab638e34635 : SUCCESS in 1h 13m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e37869bb2ea091bb09ade21ffa51f1182f42dd69","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-14 09:13:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f4492ca962072dd746b2c627deb07ce1c5a9ec42","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-05-20 13:37:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53cfbf3622625e24e2950ab035e8a0f698f4916d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 13:37:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68575fc20030a7444fcf1aaf01cf137a5f77395c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:22:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e82e3fbed2a541d2874de0639bb00a48\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/f8f6c1c1ddcb4d03aa87229243e6595e : SUCCESS in 1h 44m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be98604d3404658b8c965f6a4304437 : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d28e47693df45948bf984fd7ad1322f : SUCCESS in 20m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54ceac0d5da04b1cb73a5ac32da0ab6d : SUCCESS in 20m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b09c426e9b4409185c122943b251fc1 : SUCCESS in 6m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a969cda4c240474a90aed3d81104a7b8 : SUCCESS in 1h 03m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5bfe920d3554507b5ebf743c4571d0b : SUCCESS in 1h 10m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bab648f970674960840d1468067eae09 : SUCCESS in 58m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4aca875468fb45c4bfd734d43751b812 : SUCCESS in 1h 16m 51s","accounts_in_message":[],"_revision_number":2},{"id":"067c3445c9cf153bd1e87bea6a42176a816434f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:22:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1f87093d767fa8b06613ddde313780b293f43194","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:23:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/894084236efd46f3bdeaafa9c110f782\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2eda1769f3449bb8560558bafdfeba : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b689e57899aac0933e75f34c6cea3685de18e54a","revisions":{"f9c5384e6f5ef1eefec89139e646543e6a88705f":{"kind":"REWORK","_number":1,"created":"2025-05-08 09:26:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/949136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/949136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/949136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/949136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/949136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/949136/1"}}},"commit":{"parents":[{"commit":"de8c3bf5cf1468deaf53bf62a7bd15e519d80de1","subject":"[Stable Only][CI][fips jobs] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de8c3bf5cf1468deaf53bf62a7bd15e519d80de1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-29 15:55:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-08 09:26:12.000000000","tz":0},"subject":"[OVN] Only update the MTU of the router GW LRPs","message":"[OVN] Only update the MTU of the router GW LRPs\n\nWhen a new internal network is added to a router, the GW LRP (or LRPs)\nMTU value is updated, matching the lowest MTU of all networks connected\nto this router.\n\nThis patch is updating the method ``set_gateway_mtu`` to be called only\nfor a specific router and update only the router GW LRPs.\n\nCloses-Bug: #2109388\nChange-Id: Ic7807a440b479c0fa08c6d13813cd2a81eff9bb1\n(cherry picked from commit 2c35dd53f209f60ca60b8a2c077d97bb0c4c7d9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9c5384e6f5ef1eefec89139e646543e6a88705f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9c5384e6f5ef1eefec89139e646543e6a88705f"}]},"branch":"refs/heads/stable/2025.1"},"b689e57899aac0933e75f34c6cea3685de18e54a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-13 08:07:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/949136/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/949136/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/949136/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/949136/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/949136/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/949136/2"}}},"commit":{"parents":[{"commit":"de8c3bf5cf1468deaf53bf62a7bd15e519d80de1","subject":"[Stable Only][CI][fips jobs] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de8c3bf5cf1468deaf53bf62a7bd15e519d80de1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-04-29 15:55:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-10 08:46:13.000000000","tz":0},"subject":"[OVN] Only update the MTU of the router GW LRPs","message":"[OVN] Only update the MTU of the router GW LRPs\n\nWhen a new internal network is added to a router, the GW LRP (or LRPs)\nMTU value is updated, matching the lowest MTU of all networks connected\nto this router.\n\nThis patch is updating the method ``set_gateway_mtu`` to be called only\nfor a specific router and update only the router GW LRPs.\n\nConflicts:\n  neutron/tests/unit/services/ovn_l3/test_plugin.py\n\nCloses-Bug: #2109388\nChange-Id: Ic7807a440b479c0fa08c6d13813cd2a81eff9bb1\n(cherry picked from commit 2c35dd53f209f60ca60b8a2c077d97bb0c4c7d9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b689e57899aac0933e75f34c6cea3685de18e54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b689e57899aac0933e75f34c6cea3685de18e54a"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
