)]}'
{"id":"openstack%2Fneutron~949558","triplet_id":"openstack%2Fneutron~unmaintained%2F2023.1~I0590788179e8b3e05422bc942cb517423b555878","project":"openstack/neutron","branch":"unmaintained/2023.1","topic":"remove-devstack-tobiko-neutron","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-13 12:46:02.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2025-05-13 15:55:23.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-05-13 15:55:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0590788179e8b3e05422bc942cb517423b555878","subject":"[stable-only] Remove \"devstack-tobiko-neutron\" job","status":"MERGED","created":"2025-05-13 05:35:13.000000000","updated":"2025-05-13 15:56:53.000000000","submitted":"2025-05-13 15:55:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949558-remove-devstack-tobiko-neutron","meta_rev_id":"80b2df582b89836c08498eaacce93c8aec5a82e6","_number":949558,"virtual_id_number":949558,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-13 15:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-05-13 07:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2025-05-13 07:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-05-13 07:31:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 05:37:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2025-05-13 05:37:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-05-13 06:07:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-13 07:14:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-05-13 07:31:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-05-13 08:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e86b47129fdac49c811864058558c1b49f576314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 05:35:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d90853008fb005ea622efdd9570a49d499cd114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 05:36:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"84b4d956721d1cf410ae9053b857b98ebe1acee1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 06:07:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef7e2b0095fd4d149f78d0fd62b82df9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ca1b5006da740dd839e9c40bf23129e : SUCCESS in 19m 18s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81abba2340b843cca8c7c9dcf590c006 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":2},{"id":"964b09693a7ac99f9ecd2965633c2d8789d3962b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-13 07:14:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e70daba0f5c8c75cca1dcdedb054d90bd7e82592","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-13 07:31:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"94e8f343dad3f23346a6376c24fd045e952ba9b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 08:36:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/502e639e03dd40539211bd22e9bd392d\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/56e44b18d5554b27b439032ce009ed60 : SUCCESS in 1h 27m 07s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/044d3fb563f749ed9d7b08be3502af3f : TIMED_OUT in 2h 50m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/ffc04737839b40cd8bc3d0d2fa9f5b70 : SUCCESS in 1h 01m 23s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/27a3ba5011134b38b70a478ceff6bcf5 : SUCCESS in 1h 01m 15s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/fa22d620a80f4e3ab9283ff27af2814d : SUCCESS in 20m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/004277f8724b40098c28087fd03d6b2f : SUCCESS in 34m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14e69570db3b409bbf7b4d93a6e56e23 : SUCCESS in 12m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf74dd09a5934d3081f5624cdd25914f : SUCCESS in 15m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b4dfdb59c6f43f489e87c341f741304 : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53d27a1738948dd9bdfe2c91468b30c : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f583a91be547ad1ab8f04d08be4858a6a2358ef0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 12:46:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4adb6a5e68a516ca7e18dfb1b975a83d22edb66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 14:28:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e141eee9a281486dbe360972c22f505e\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/58ac531b2ac74b2f99d0db81013df503 : SUCCESS in 1h 31m 26s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/4213b133a92d41a29383a667e21e7229 : SUCCESS in 1h 02m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/88e4e3fce4ca45cdaf0013f748286179 : SUCCESS in 1h 06m 19s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/5811e8ac570447b0aaefd5a4ff1a9cf3 : SUCCESS in 1h 19m 40s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/eeb4f74168b742b4bd2235a69c3f9692 : SUCCESS in 26m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46f8c6dcc83a4019a62a5789c3086f8f : SUCCESS in 33m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0980241df92448b3ab63954b4aee4319 : SUCCESS in 12m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e48d24222254468ba8f877a4fb1a11a1 : SUCCESS in 22m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5978dea9b074e00b23f184f9f7ee1d9 : SUCCESS in 23m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82bc3bc184e84fecb54eba3f1803c94b : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":2},{"id":"db309657210e3fde28fbbf80514fd29839b3e3d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 14:29:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b9dec48add7c37c52040b0e53a3164178cc8cfc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 14:38:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c2fa48b514b4fa4b64ff990d115be14\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17f62a4f19714972aa53644192fbf3e5 : SUCCESS in 35m 38s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56c1b5b2c276441cbe8631a482832df4 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b7b1e9e9709284ca57d90030cac77a210f9ff705","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:55:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b357b099efc04be1b6cd87329f0d9270\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/901843ae354541999ff43298638831ad : SUCCESS in 1h 25m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99744f27418a4eceaa6723abb6a7af4b : SUCCESS in 10m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51b271ae20af496ab0bd72d42dc8cd39 : SUCCESS in 18m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d02e2782e3d546df80c0ff30136af3a9 : SUCCESS in 19m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31433f77378249f89e0143bb3f9d82d3 : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f07c083f15fa4fd2c2956ea69f1853680f5ad1ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:55:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"80b2df582b89836c08498eaacce93c8aec5a82e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:56:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d7bac7937274ca9abb75ba4890906f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b6875b6ff84a46a29ce64198398a05 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59c9ad47d876bb589d116c496d6a893c502d0e7e","revisions":{"1d049f64a63222e7c2924eef2237bf5289079785":{"kind":"REWORK","_number":1,"created":"2025-05-13 05:35:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/58/949558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/949558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/949558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/949558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/949558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/949558/1"}}},"commit":{"parents":[{"commit":"bbb09c7852384a0972bd3461bbce5f3216c447e1","subject":"Merge \"Use the declarative attribute ``standard_attr_id``\" into unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbb09c7852384a0972bd3461bbce5f3216c447e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-10 06:11:10.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-10 06:14:20.000000000","tz":0},"subject":"[stable-only] Remove \"devstack-tobiko-neutron\" job","message":"[stable-only] Remove \"devstack-tobiko-neutron\" job\n\nThe job is failing in the periodic queue. As decided in the Neutron CI\nmeeting [1], this job is removed from the CI.\n\n[1]https://meetings.opendev.org/meetings/networking/2025/networking.2025-05-06-13.00.log.html\n\nChange-Id: I0590788179e8b3e05422bc942cb517423b555878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d049f64a63222e7c2924eef2237bf5289079785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d049f64a63222e7c2924eef2237bf5289079785"}]},"branch":"refs/heads/unmaintained/2023.1"},"59c9ad47d876bb589d116c496d6a893c502d0e7e":{"kind":"REWORK","_number":2,"created":"2025-05-13 05:36:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/58/949558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/949558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/949558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/949558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/949558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/949558/2"}}},"commit":{"parents":[{"commit":"bbb09c7852384a0972bd3461bbce5f3216c447e1","subject":"Merge \"Use the declarative attribute ``standard_attr_id``\" into unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbb09c7852384a0972bd3461bbce5f3216c447e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-10 06:11:10.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-10 06:15:29.000000000","tz":0},"subject":"[stable-only] Remove \"devstack-tobiko-neutron\" job","message":"[stable-only] Remove \"devstack-tobiko-neutron\" job\n\nThe job is failing in the periodic queue. As decided in the Neutron CI\nmeeting [1], this job is removed from the CI.\n\n[1]https://meetings.opendev.org/meetings/networking/2025/networking.2025-05-06-13.00.log.html\n\nChange-Id: I0590788179e8b3e05422bc942cb517423b555878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59c9ad47d876bb589d116c496d6a893c502d0e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59c9ad47d876bb589d116c496d6a893c502d0e7e"}]},"branch":"refs/heads/unmaintained/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
