)]}'
{"id":"openstack%2Fneutron~949972","triplet_id":"openstack%2Fneutron~stable%2F2024.1~Ib951d1bcb39eedd2100adad33b2bb8e106883ba4","project":"openstack/neutron","branch":"stable/2024.1","topic":"address_group_api_collections-stable/2024.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-05-20 16:03:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib951d1bcb39eedd2100adad33b2bb8e106883ba4","subject":"AddressGroup API collection should be the resource name in plural","status":"MERGED","created":"2025-05-16 08:45:09.000000000","updated":"2025-05-20 16:05:13.000000000","submitted":"2025-05-20 16:03:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949972-address_group_api_collections-stable/2024.1","cherry_pick_of_change":932380,"cherry_pick_of_patch_set":1,"meta_rev_id":"600d4f1f55e30acfdb0a55d76b073cb6cc99abb6","_number":949972,"virtual_id_number":949972,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-20 16:03:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-05-20 13:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-05-19 07:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-05-20 13:38:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 08:45:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-05-16 09:44:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-16 10:47:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-19 07:25:55.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-05-20 13:38:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"4dde647d1318f5e2f2b572333a440272aab5f33b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-16 08:45:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"92f64c6ceb742f1c7e3d538d4a9d3d8ea1662396","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:44:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c96a86ed073741dda9d72ffb38ca530d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71b995ec46ef4e3b8531096c0d9d83f1 : SUCCESS in 51m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/98264b866d0b4015b3b11fd008aa6bb5 : SUCCESS in 53m 26s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9d84d5df51f54cd78d52cec3b8bf292f : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71aaa380f7c17ac7cc2ca4bbef524719b0ea8be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 10:47:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/083fcad0a51d42fe9666d712d5b2c2f0\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/b57a3df5740b4558a8f4fbb16a306985 : SUCCESS in 1h 00m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/e4778b361b084f759382b0d5f495f7d2 : SUCCESS in 1h 59m 59s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/7ef172931a99446baff7bd2cccb828b7 : SUCCESS in 1h 52m 18s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/13113b0fe6bb40578c23d1e8503d67c5 : FAILURE in 17m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66ee8c03835b4d47bd99d04cd9896c2f : SUCCESS in 47m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f86246e573204ce7bd86a70379e37b63 : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec31680ad48e4b44b700ec388d3800a3 : SUCCESS in 26m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4845e18623184be386738d59f8b9c00d : SUCCESS in 26m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/189f22626560433289878566c5de372b : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b9a06cd2c484bed93adaddd15dbbac7 : SUCCESS in 1h 47m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a376d7d3c02f4ef3a7c2477904d37787 : SUCCESS in 1h 09m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/72b37215d1834cc6a64bea4a225fc54d : SUCCESS in 1h 00m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b15247e257fe40cda837a4eb7fad7d17 : SUCCESS in 1h 00m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/55104b7b627c4e1199362ff5db73f02f : SUCCESS in 1h 17m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/031914ff28e44dfa8fb9bcbf92ad7ee6 : SUCCESS in 1h 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3a75f0947b8776510e38527e05859d863ea6b07c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-19 07:25:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2be5236342d28ca089f38f31526424b81045a4f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 07:59:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da9b218cc1ca4399be4275e93d7215f8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eee26154a6b4cf8a5b30ec60ff34231 : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6da691ec2bcb48bdb0c2e44a1a8b8fda : SUCCESS in 25m 23s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6433a44dfd8f4a2a8e514e77e309b517 : SUCCESS in 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71248859db631546a1762181fd1fc22c77f09ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 09:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce0af4a59d5f4c7ba61d9b599fa42305\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/d7909007d05e4856a5b1302bba094aba : SUCCESS in 1h 31m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/de626addcb2e44d7b8959f00253e4997 : SUCCESS in 1h 27m 21s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/9a903e25f29e4182bbb348bab841d5f6 : SUCCESS in 2h 00m 05s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/31a458bcecbe48e08425887f6ea16897 : SUCCESS in 33m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5da9dd2b4813481a8d422854d49e6dba : SUCCESS in 43m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4d12a25f43f44959fb72fe5212edc3a : SUCCESS in 10m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f2ffadcc21848e9b2bd4cb6e4220a72 : SUCCESS in 26m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9dc72fde41254765a5f36065b5d3bfc2 : SUCCESS in 23m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adac73272dd2468589f6ac048df30733 : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3b68234573b4a148ad72033cd8a9674 : SUCCESS in 1h 23m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f78abe178d5491881fff5855d120203 : SUCCESS in 1h 35m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2f0ba859ce634b2bb6e726eeb12c80a8 : SUCCESS in 58m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c5e46e57a74b441cb1c9526cb2970210 : SUCCESS in 1h 09m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d4369c1805ed464c81d9e3ce9ea21cd4 : SUCCESS in 1h 13m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8d1c49897aa046ceac77c262b971e90d : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3041fff1cb0404a618284d0d46a8b6b3ff57eb70","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-20 13:38:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61c7497525ac8f4e210a9f5bed64b5a4175003e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 13:38:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49faca5e2cb5fe9e0298aae20cbde963d9b69ae3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 16:03:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/918a38e171da40b1a1aef577e3378614\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/e391e681bf1142388e2a9d4f442781a0 : SUCCESS in 1h 43m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff3635db9c44afa8efff9191bf1fd1f : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e9d88dc9544b51b4013e66db5c79de : SUCCESS in 24m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65fc709fb0a34a4580e499f88b17c946 : SUCCESS in 17m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/733bcb1e147a49088377fe7156b60c05 : SUCCESS in 12m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3e996a110aa43618bfd3da44788e02a : SUCCESS in 1h 18m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c6682bd1aa24d6aa98c06b5ff817cff : SUCCESS in 1h 44m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ea5f707e7a9445b4a99e0df3f3fdab44 : SUCCESS in 51m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f2be1c0ab6d944578724d92cb5578cb4 : SUCCESS in 1h 23m 51s","accounts_in_message":[],"_revision_number":1},{"id":"013964d9f83e50c3edb91ce16f86e9b7e31d171e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 16:03:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"600d4f1f55e30acfdb0a55d76b073cb6cc99abb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 16:05:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/741ce42e284a4cb19eca4e5de1bbe284\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad2fc86633f54a1ca0bf78db6fd197bf : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79a92aa60f7538546e0391a9c6241bcb9bee2c20","revisions":{"79a92aa60f7538546e0391a9c6241bcb9bee2c20":{"kind":"REWORK","_number":1,"created":"2025-05-16 08:45:09.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/72/949972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/949972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/949972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/949972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/949972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/949972/1"}}},"commit":{"parents":[{"commit":"0d4ef860549a552db3a9d04934154ad256662bbb","subject":"Merge \"[Stable Only][CI][fips jobs] Use stable constraints for tempest\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d4ef860549a552db3a9d04934154ad256662bbb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-15 07:08:37.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-05-16 08:45:09.000000000","tz":0},"subject":"AddressGroup API collection should be the resource name in plural","message":"AddressGroup API collection should be the resource name in plural\n\nTrivial-Fix\n\nChange-Id: Ib951d1bcb39eedd2100adad33b2bb8e106883ba4\n(cherry picked from commit 7919126e812b39efa9d8ec5562d635608e500f4f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79a92aa60f7538546e0391a9c6241bcb9bee2c20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79a92aa60f7538546e0391a9c6241bcb9bee2c20"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
