)]}'
{"id":"openstack%2Fneutron~950139","triplet_id":"openstack%2Fneutron~master~Ia14f3748245b59850bc21cbc87e04ffbdbb5850f","project":"openstack/neutron","branch":"master","topic":"bug/2100770","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-20 10:34:41.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-05-28 11:11:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia14f3748245b59850bc21cbc87e04ffbdbb5850f","subject":"Require alembic db migration scripts to be idempotent","status":"MERGED","created":"2025-05-16 10:15:45.000000000","updated":"2025-06-07 01:52:11.000000000","submitted":"2025-05-28 11:11:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950139-bug/2100770","meta_rev_id":"7ca0a7212bfe4f71b1bb36b14978e21c41889d27","_number":950139,"virtual_id_number":950139,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-06-06 22:13:00.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-28 11:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-05-20 10:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-26 09:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-26 09:25:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 10:19:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-05-16 10:20:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-05-16 10:26:41.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-05-16 10:39:05.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-16 11:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-16 11:17:34.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 13:13:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-20 07:27:56.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-05-26 09:25:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"497ae4313a8e62dfeb95d65638fa1823601673f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-16 10:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4486bc6adbe8459f060b37ab71d857400716b9c7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-16 10:39:05.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/39/950139/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a3531b02e0e18a426217ee58e136fad82c70bd11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-16 10:42:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5461d930e4c3276eae7f6c97d38d19039e66439","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 11:14:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/487dafb3912045dda0f391f751240d74\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b3c2c8f46bae464895b89c7ba8715f38 : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/af0ba0ab54464306a0bcf8c3fd70c796 : SUCCESS in 30m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6a8df39e47404e66be27089eb27123f2 : SUCCESS in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0187ac906aa5558444b3d708a3959836d3582c48","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-16 11:17:34.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/39/950139/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8222f82e004bbc385e2c9d4cdfd46a53dc32e09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 13:13:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93899c02e03a40cf959a4b7eac411191\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e026aa6d629d4ddb837e44190ea0f730 : SUCCESS in 1h 36m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c85a0ff5834841b5b3d3174bbb4abb64 : SUCCESS in 2h 29m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1a3412c3270549ec99bc0e04a98435de : SUCCESS in 36m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21ebc559ab084a8bbf3cad183868f762 : SUCCESS in 50m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969ee61c28744ac28396ddc3b98fb85a : SUCCESS in 9m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2edb1cea988c422c99b67fbe50dd1cdf : SUCCESS in 22m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5e9a18f8ed9412d9c4d6a29229ca9ee : SUCCESS in 13m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c973962b1094b3fb2d90c30223ab5b5 : SUCCESS in 16m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984562e9d96e41ce98069cd5c4dd7dc4 : SUCCESS in 7m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c73e875ae34a412b976dd6a7abb470ce : SUCCESS in 57m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d6844cfd832f4052aff987a17609b63d : SUCCESS in 1h 13m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d5c433cc9c9475f91bb2c3ac7359648 : SUCCESS in 35m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/54a39d1ad58d4b06933c405e6ecad235 : SUCCESS in 1h 03m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6259818a784e4bbdb1015066c0f005c5 : SUCCESS in 1h 33m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c7b2f15a49da408ca7d8f65458b34a89 : FAILURE in 42m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1f2790abde63422dbd79c27b61cea06b : FAILURE in 55m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d52fbf9af0dd45a99ddb906d0ae2bb46 : SUCCESS in 31m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b337a4b4b01745e3abc07aa6361ad407 : SUCCESS in 1h 14m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d628865add31406d802d3b985eaf05ba : SUCCESS in 1h 13m 31s","accounts_in_message":[],"_revision_number":2},{"id":"883ee820cbcbe0bb8cc9f78e2ef1146bbc69758a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-16 14:46:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"42e9e842425ed622fa4a2402d4407476bf21a98e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-16 15:22:13.000000000","message":"Patch Set 2:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/39/950139/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3ed0c74b48501fa667859c0f65ad05e99b45fc5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:45:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3208d690bdd4adc8c871b80e6b0cff2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0a04ff92437d47a8a47dc4cdabe4b6c9 : SUCCESS in 40m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/25992ffe24724b53b85eb3fb5d73ed35 : SUCCESS in 29m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/355381e6d89349138eeb3a433937d3b5 : SUCCESS in 20m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e9b1dcf16bbea9500dadafbdc382cb5e7120cb90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:40:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d04dd5c20b44b16b62fa108dad16681\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e7a288b2fe244661af5f80c00d1de611 : SUCCESS in 1h 37m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/df79bd948b8446b2a2382267e986182a : SUCCESS in 1h 50m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df201fe84cb841c79b1f27997c7c4b84 : FAILURE in 19m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f74e787e9bbc44d192b64a14a2d80a69 : SUCCESS in 33m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62092611aa94f9db1ddd62a62152b0d : SUCCESS in 11m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e374589b134d4454a08e6c018908369a : SUCCESS in 19m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68ece95f7f60478396544166b2b2e6dc : SUCCESS in 21m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9bb442501f04071af383b1fe22f1646 : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ea3f64e8344ad4bad4a0660f26dbc6 : SUCCESS in 11m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c4ca5c42fdf446a4bde37322501f58e2 : SUCCESS in 1h 30m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b4a07884b5274486811ad967e6835b25 : FAILURE in 1h 29m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4bd836c08ba49458339c1cab1d9b679 : SUCCESS in 51m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec7af58f26b544a092c8f9d904dc49c1 : SUCCESS in 1h 02m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e745118ca961471ba9f9c977fc95772a : SUCCESS in 1h 40m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6c7232064c7d4ec0a0d2f0c252dbd6ee : SUCCESS in 1h 38m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9528672bec694921b2db6859ee729fc5 : SUCCESS in 1h 09m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7efd75ecb8b44a04933cddca58d5cc4e : SUCCESS in 56m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d8511a9e0abd4bf2aae51bfe38093fdf : SUCCESS in 1h 18m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6530cdfc324443fbb548017232b9ce17 : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":2},{"id":"0d6872c9415717dc6f7477aa09f935be18ada35a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-19 07:25:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30d65692fe8763c7b6ad0ddb31ddffaab7c39ed6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 08:07:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02253212a4004e2582403b69da2a7179\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/647eb1375df14fe58fe0337711789988 : SUCCESS in 20m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a880199737d4b02b0c9959ae9f35479 : SUCCESS in 26m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ff5d4fc463c54829843e3865e442fb03 : SUCCESS in 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"716c13ac97d5766b523b84fe4189888a7cbb435c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-19 08:30:56.000000000","message":"Patch Set 2:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/39/950139/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0815b74598bf9caf7d91314ce6cbce1fad485e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 09:53:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33a162da8c40423f8641d7ab14e7e4da\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/213cc9ffe6f34a7193e33b3b198b2f4f : SUCCESS in 1h 40m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2fe07272642041588bb041cae46c0fcf : SUCCESS in 1h 52m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c3bab24f0914b2fa576618e378930f6 : SUCCESS in 45m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10772d4de251412e9003619edb76e3e1 : SUCCESS in 44m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594c3ef718e1478cb6c925bcfd91c5e2 : SUCCESS in 10m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/046cc8efa6b4495b859a4a0c46f40171 : SUCCESS in 21m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f8273d4c38745fcb1b0a0bf5baadc44 : SUCCESS in 18m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89e420efabdf4fdc96ee9bc0332d5900 : SUCCESS in 26m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a41f447fa0426ebed9179803d42a64 : SUCCESS in 13m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/302582c8192c499ab020f1640089df3c : SUCCESS in 1h 37m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7ae9090c3ad44619b5899b8fafefda44 : SUCCESS in 2h 26m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d9c81a7c52844cda7bc10b150451f00 : SUCCESS in 54m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/123bbe18add7412bb608cadf1712ae4e : SUCCESS in 1h 12m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4ff3e916b7004d3d9e8823e34cdacdd5 : SUCCESS in 1h 38m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e1672f70980f4ed5bb406bfeca81ed3b : SUCCESS in 1h 28m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7ff4e56655104ec5ae5e54fb5caaa4b5 : SUCCESS in 49m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb9a86f420344b7381d40b832b7a1e6b : SUCCESS in 1h 03m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/68d5e2b06c444f3c86e03c7e2502e469 : SUCCESS in 1h 12m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/39e2a77b2515473fa2c18a7338827233 : SUCCESS in 1h 16m 48s","accounts_in_message":[],"_revision_number":2},{"id":"f6b82ccc1885580f10b92f6d4e6e1ea005d2ccca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-20 07:27:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5ad67ae4e7df458c2a11fbe37a138a3ece6f51be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-20 09:17:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"14bcbf88e4d80bef38f99ca1dee08d9459c3346f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-20 09:17:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db42a30717129cb59fb34aa8bff6f6fe0064efa1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-20 09:39:38.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/39/950139/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"c39cd1555b865cca299955f30f1e1af95790cffe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 10:33:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/919b46626b36477a861e9546c11b7170\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/403894c68a6d41b88b591d0d42e65ed0 : SUCCESS in 23m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5ed05d2023349d1a14ef0c5f9ff32ca : SUCCESS in 30m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/83b1f357e1cc4799ac685152e690de0e : SUCCESS in 39m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"35a2d62a10c512eb3810e5c0f4b9b81a22f6b033","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-20 10:34:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eaae3bad6334e2aa087e40ab327f7b28c1f0c87a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 11:24:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc73552020eb46c887a9b7373cb76001\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a2f02f3ecd7b409b860ccbff9311e129 : SUCCESS in 1h 31m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ce8aa65b62204156bff77dac84dced9e : SUCCESS in 2h 05m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6889b37b94e54c6288d8d95beb32e151 : SUCCESS in 50m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40b6e4cc9221477598c8e60888ccd567 : SUCCESS in 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba1215632a6a49c29039548173b6c7b5 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6215b19b901c41139a2425d1fd88e3e7 : SUCCESS in 23m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4932c29b117422a9db14ec645782304 : SUCCESS in 27m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5824bdebd53432ca26418092ae6902f : SUCCESS in 33m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01128fa5b2a43a6a7b6f2933fdb5bac : SUCCESS in 10m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6fe6c3861ff64e39b85927d60225d3d4 : SUCCESS in 1h 21m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/28301a30c77948039897c4011d183eb7 : SUCCESS in 1h 54m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a0d86894e894207aef6819620f340b5 : SUCCESS in 58m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/89bcca434da34b5d8b8014a93873d906 : SUCCESS in 1h 10m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b2864b7ad1034f17adf41ae581bd86d7 : SUCCESS in 1h 00m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c17ddeb914c641de8bc7987b1788a481 : FAILURE in 1h 33m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/214ca4d930db49a098a65d06d808e67d : SUCCESS in 56m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d2e74660e50f4923ae8d3e91ff429d20 : SUCCESS in 59m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91596dfd161846e798f840ab1bd423f5 : SUCCESS in 1h 10m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6e46231dfaf54219903c98eacc604370 : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":3},{"id":"60d27102b603214be6e5e70e8158e1d79674b547","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-22 12:59:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce3f2e76b495d012579dd2f4a7a7d5b2b9c79993","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-22 13:22:05.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/39/950139/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"a4253fbae2c84750172edd3ad6020d68c007a501","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:56:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e51c10f3e24489f965276cc5870ce89\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/050d7cbc00a94ab184e3ef2958713b6a : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/376a2964eab94584a9f1a9db93139b0e : SUCCESS in 39m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6003d0f86159410ca9a5fac8fe6187a6 : SUCCESS in 25m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"04bbeea52ffb4b28bdc148c9a0b90f521c25c83e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 14:59:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/433729d9fe5a43cd810713dcede53438\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8c891427c387489d9b761ef2ab3218d7 : SUCCESS in 1h 22m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/21c364553e364f74a38346a0f72b425c : SUCCESS in 1h 57m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb32077060a944bb8ea42f41e8f25be2 : SUCCESS in 31m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a213704596a443eab42489ccefafc8fd : SUCCESS in 55m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b729926f84cf4da28a0052011694064c : SUCCESS in 10m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce027dad447b4ec3a41018a51a4cc6bc : SUCCESS in 19m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6134ea3ae84473c9c32f0e3212e6c26 : SUCCESS in 19m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d31b727db714908af9217719ac6a004 : SUCCESS in 29m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c215006ff970454f914683c5dbfa240a : SUCCESS in 10m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3ff6f72f0224426aae8af673251ae75a : SUCCESS in 1h 29m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2f3acb745eb74878bbd88548f1a00526 : SUCCESS in 1h 47m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d9cb4c58c4748e49ddfc24f3cad4e61 : SUCCESS in 1h 04m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ead6cd04c5b54959a2d6d00cf8069612 : SUCCESS in 1h 06m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/be887bfd636d42dea6058714cb90f554 : SUCCESS in 1h 33m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a3fc03696164e459958c339bb4b1a20 : SUCCESS in 1h 39m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cbff69ea9ea74ba7bc946b7b64443b54 : SUCCESS in 59m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/45f70dc7334c4fed81ab69cba069fcbe : SUCCESS in 1h 00m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0347898e74e0424bb0da93548ae4325e : SUCCESS in 1h 15m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/26bfc7917d72447b86bb6fb76b197df1 : SUCCESS in 1h 14m 01s","accounts_in_message":[],"_revision_number":3},{"id":"d3963917ce57f0389d7b24553c433fbbe495ebe8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-26 09:25:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1dd5145fcdb3549b5719bb1e805cefa7907aa287","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 09:26:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9c9eb17d2ef4f78a3308d03b69e9f0e1d98d2c89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 11:59:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/255e059786d4473083e2b541d43ef5f2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/67f7b2d6f72b4f718271661b014e3d58 : SUCCESS in 1h 51m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aee7cc7f919c4ac2897c5487a30d99b3 : SUCCESS in 2h 26m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e59a71b36af6479e86b64539a853fc31 : SUCCESS in 11m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bf219d6956a45ceaeb2702641b26cd5 : SUCCESS in 22m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff98cc71c99d4a75949585318f74299d : SUCCESS in 17m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c781219fd46f456eadc20cb83b3fefad : SUCCESS in 14m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/334cb078254f4a08a6b44b934e6b68c9 : FAILURE in 1h 31m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/610e1203163f4ab39e7a97f73e10f348 : SUCCESS in 2h 32m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2ce1ea9547e54c0c87986ca85267cb4d : SUCCESS in 1h 43m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/009e125939f24f1c848691d0da5a8c63 : SUCCESS in 1h 09m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f871e942bb5a4f3da155ffcb040c2710 : SUCCESS in 49m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0249c82134ac4ac2b6f23fcf65f33c08 : SUCCESS in 52m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b40d0060273641438344651e72d3c050 : SUCCESS in 1h 07m 41s","accounts_in_message":[],"_revision_number":3},{"id":"2a5d2f1112ffea1fc31daee491759449725d9f63","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-26 12:42:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb0a744f6f105145c97801cd2f3fcc30dd7516bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 13:28:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad3bb3c8fe641338dc985a3792ff845\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a0739e82cd24f5faf95a87be2be617e : SUCCESS in 21m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96ef0a390b4a4f9c8f3ea98ddc6fa619 : SUCCESS in 20m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/396940c929af44999bc5cf1b6c779ab0 : SUCCESS in 43m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c31e6901f38802840cc5e8c334ba5ca07222491","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-26 13:52:37.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/39/950139/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"07e05ddb49b287f0a92fc24cd78ffb7d4af6456c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 15:40:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/859f7234d2a740c285dc940776dedc33\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/099610b7a7d54117b4fd78fbcbcda56c : SUCCESS in 1h 40m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5b46628a4c1c457f8427d1e927defb2c : SUCCESS in 2h 28m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/edf5244548194ae68c90d787626ea2e0 : FAILURE in 20m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f6000a990df4b5b92e38b61d4cddb9d : SUCCESS in 49m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56265134775407a83f1c0431c6fc8ca : SUCCESS in 12m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e09889eba8114d1cabea483dde1adb99 : SUCCESS in 24m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a791afec21643229cd0233b40d0ce3d : SUCCESS in 19m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/513f3202d7dd456089547dd4ed976e32 : SUCCESS in 29m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ccf8fa07a74401be84f1b9f0ed1710 : SUCCESS in 11m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/310b15e56a7a49d69318e12565629286 : FAILURE in 1h 19m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/62bcf180d4f24a8f902ba0216fcc88c4 : SUCCESS in 2h 54m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b150bdc550b460fb1c8d259750f5d16 : SUCCESS in 1h 00m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/21facd51df0c4fc2a793a3d771178fa5 : SUCCESS in 1h 31m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e16bdbaab74140cb825b7b35c0976cee : SUCCESS in 1h 48m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a7f4e0443a44dd4a6d089d4b53079b8 : SUCCESS in 1h 32m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/90ae300f9773490cb60f4b4cb16a98a9 : SUCCESS in 59m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c592f3e1a23049fa8702e042ac2324df : SUCCESS in 58m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e392c1481329478fb526012990a213af : SUCCESS in 1h 14m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f6853e222450422eb9c0235865690476 : SUCCESS in 1h 22m 09s","accounts_in_message":[],"_revision_number":3},{"id":"8355a60286b0ad8dd23b18a45a41804bcb386da0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-27 07:09:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5777a6b2812fcccd0393ae3051485f77d4cbfa10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 08:17:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a43f236a4a2498b933dfc888e866550\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/862d0b0435d24d6a93113bd3ce8cab7c : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7be18bb40c7a470da5bfb4a318283157 : SUCCESS in 27m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e57d6fccea8e41f2bec0c1c7a067b4ef : SUCCESS in 1h 04m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09b3f887a11ff44a82e5553c815df0c9e7ea0b70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 09:57:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2267aed453f74895b2f387b0c73bb2b1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eda4c8ee7ebb440299b26eee70b1eb9f : SUCCESS in 1h 59m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7cb04f107d4c44339fe48a33c28e973a : SUCCESS in 2h 04m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea7b7eee9d174449b6da249630a7a26c : SUCCESS in 48m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f5e37e54ac54e859ef9b1087058e44f : SUCCESS in 1h 00m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c0d04ba74634478bf80d7751be0e8cf : SUCCESS in 8m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ff4083305bd4f65bc2b7056f1e06c41 : SUCCESS in 20m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25fb4f3231d14ad9bef5f732fa53a993 : SUCCESS in 17m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e3a0f161981499d9c4a3b5a5ec2f6be : SUCCESS in 26m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa6d0f43aae44989fae61b1ea649851 : SUCCESS in 14m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5cdc50aa39d5418c9a5a4ac7c86acf59 : FAILURE in 31m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0780a4b772da4451b66589f44f29d967 : SUCCESS in 1h 55m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/edf4efa7f71a4688b176a02ffd13b98c : SUCCESS in 54m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b03f5cdf07b74fa294972a313f293b82 : SUCCESS in 1h 12m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ecdf5867aa9a49d4b6ed5d509b982bd4 : SUCCESS in 2h 42m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/923ee6c2b1154d3fb75d6ecfa21b1e5c : SUCCESS in 2h 27m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/308236e5cffe478894a4543128313f9a : SUCCESS in 2h 08m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/150eafaa31884a04bbc53c4673170470 : SUCCESS in 1h 01m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/deee611969414a46aa73b208c1f1f0df : SUCCESS in 1h 15m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/48d54a63628e47eca017252528e34da2 : TIMED_OUT in 2h 04m 35s","accounts_in_message":[],"_revision_number":3},{"id":"cc3e2a90cbd3807f45ae4b906d0935569f8bd853","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-28 06:44:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c4ad8c799a4de50f54e9b69bdcf6e28dc11470cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 07:31:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c66e6768bfd4451f8f2c101e7c966729\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ab0134647694a65bf5be35d832a768f : SUCCESS in 28m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8953847a40e6471f9e6e13f1bfeeae06 : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/708708027d1347daae3c2499b6d1c84c : SUCCESS in 44m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"670475611125a3973094c8c9084bbd081e07e51b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 08:58:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7238158e3105423a85f0de5b3c55a52f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6c9bbff266fc4dbc89cb60d0c8fcfc2f : SUCCESS in 1h 30m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e3c2e26ae9a941cc81b083924a032a71 : SUCCESS in 1h 52m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e7ce871f180a4aff800b0e6812a93bdf : SUCCESS in 32m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b15aa7699bb64909888fa558e29756d9 : SUCCESS in 45m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b8b8663197f4d99866701a6bc2759d1 : SUCCESS in 10m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/415076a6efdd47ba86ced7ac80be8c10 : SUCCESS in 21m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c56b0820f0044cf492ec4b3eba4ecc48 : SUCCESS in 19m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e605ca0bf9f4340a698954677ba5efc : SUCCESS in 28m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4857eef5e544ca847f6eef15669dcb : SUCCESS in 10m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/031721ed210649149936fafe0e8fafba : SUCCESS in 1h 04m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8b5c79639ef147cfa2d9b08db96b877f : SUCCESS in 1h 59m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e00f414987c4b98a019595ec5afcf7f : SUCCESS in 1h 00m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff782d3fdd8548768e17f767b1442e5d : SUCCESS in 56m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c38111b930b2436fb454ea89831181d8 : SUCCESS in 1h 24m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/984b8c478802459a8ceda6f5ebaffbd5 : SUCCESS in 1h 17m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dd0dcd49ca2b420cac8ca8e9bbdf5175 : SUCCESS in 1h 02m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ee44a01ba8814feabcb2e9b4a9746a25 : SUCCESS in 47m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b5131bf7ae5340fcaf6a5c1668b8ada4 : SUCCESS in 1h 14m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c5751c7d27e14d49b91deb681e07ba68 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":3},{"id":"df62beaf767d576249c74dbbe49c6ff99d2a2466","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 08:59:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"800af33eb62ff5727dc97ab8813c36f11397953a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 11:11:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1702bb662f794c2b9c2c693d892490fa\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9fd48b91177341e58e8a6323f652532b : SUCCESS in 1h 41m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/be53dcafaf3c455e840b654878586aac : SUCCESS in 2h 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be516aba29e0477782077015a0fb4746 : SUCCESS in 10m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af493067b28343bf8d77a1b6d818be66 : SUCCESS in 18m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03c7f2f0eea249c9bf242ce40602260d : SUCCESS in 20m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/074f3640d14f40eeae76032fd09ffcdb : SUCCESS in 11m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d87af3178f9e4f37baa9fa47d9692cdf : SUCCESS in 1h 51m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9cfe3d30192545c69c5c9bbbdd96f188 : SUCCESS in 1h 49m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b720ef2608a4a42b265e4c52e75b6c4 : SUCCESS in 1h 40m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/73af2cc674cf42cb843f5671ba138b8a : SUCCESS in 1h 05m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6165bbd3b8541469c55f23a74a84b10 : SUCCESS in 55m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d1a3727cd09947689e3a474b7270cce6 : SUCCESS in 54m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5b984b7a829a483f8c70b7a3bfa17205 : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":3},{"id":"257b78dcc754fecdf1c9d7865939630317db40c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 11:11:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"79299b259eaef4e271e8133a304a9387b14ca129","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 11:13:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/881db94a88a345e09f731fb330b86f71\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef139efa04447f69502409e6bfb4179 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3},{"id":"d606146587a3c880f85b9c16314370c5d38c0cfc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-06 22:13:00.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/39/950139/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7ca0a7212bfe4f71b1bb36b14978e21c41889d27","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-07 01:52:11.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/39/950139/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6af3801ac822d7825ddfbc62a5c1893605999b3e","revisions":{"2fe40f8bbfe8cffdc86138cebde606767672f9e7":{"kind":"REWORK","_number":1,"created":"2025-05-16 10:15:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/950139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/950139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/950139/1"}}},"commit":{"parents":[{"commit":"1bc4b526e9c743423069ab4cf6ef3883d5e48217","subject":"Merge \"[OVN] Create a HA_Chassis_Group without raising an exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bc4b526e9c743423069ab4cf6ef3883d5e48217"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-05-16 10:13:14.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-05-16 10:13:14.000000000","tz":120},"subject":"Require alembic db migration scripts to be idempotent","message":"Require alembic db migration scripts to be idempotent\n\nWith this patch alembic migration script should always be idempotent.\nThis is forced by new functional test which makes sure that migrations\nstarting from first alembic migration script from the stable/2024.1\nrelease is idempotent.\n\nCloses-bug: #2100770\nChange-Id: Ia14f3748245b59850bc21cbc87e04ffbdbb5850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2fe40f8bbfe8cffdc86138cebde606767672f9e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2fe40f8bbfe8cffdc86138cebde606767672f9e7"}]},"branch":"refs/heads/master"},"2644421487e9955867e2c1da063f74bfa597e610":{"kind":"REWORK","_number":2,"created":"2025-05-16 10:42:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/950139/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/950139/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950139/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950139/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950139/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/950139/2"}}},"commit":{"parents":[{"commit":"1bc4b526e9c743423069ab4cf6ef3883d5e48217","subject":"Merge \"[OVN] Create a HA_Chassis_Group without raising an exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bc4b526e9c743423069ab4cf6ef3883d5e48217"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-05-16 10:13:14.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-05-16 10:42:02.000000000","tz":120},"subject":"Require alembic db migration scripts to be idempotent","message":"Require alembic db migration scripts to be idempotent\n\nWith this patch alembic migration script should always be idempotent.\nThis is forced by new functional test which makes sure that migrations\nstarting from first alembic migration script from the stable/2024.1\nrelease is idempotent.\n\nCloses-bug: #2100770\nChange-Id: Ia14f3748245b59850bc21cbc87e04ffbdbb5850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2644421487e9955867e2c1da063f74bfa597e610"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2644421487e9955867e2c1da063f74bfa597e610"}]},"branch":"refs/heads/master"},"6af3801ac822d7825ddfbc62a5c1893605999b3e":{"kind":"REWORK","_number":3,"created":"2025-05-20 09:17:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/950139/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/950139/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950139/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950139/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950139/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/950139/3"}}},"commit":{"parents":[{"commit":"1bc4b526e9c743423069ab4cf6ef3883d5e48217","subject":"Merge \"[OVN] Create a HA_Chassis_Group without raising an exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bc4b526e9c743423069ab4cf6ef3883d5e48217"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-05-16 10:13:14.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-05-20 09:13:47.000000000","tz":120},"subject":"Require alembic db migration scripts to be idempotent","message":"Require alembic db migration scripts to be idempotent\n\nWith this patch alembic migration script should always be idempotent.\nThis is forced by new functional test which makes sure that migrations\nstarting from first alembic migration script from the stable/2024.1\nrelease is idempotent.\n\nCloses-bug: #2100770\nChange-Id: Ia14f3748245b59850bc21cbc87e04ffbdbb5850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6af3801ac822d7825ddfbc62a5c1893605999b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6af3801ac822d7825ddfbc62a5c1893605999b3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
