)]}'
{"id":"openstack%2Fneutron~950239","triplet_id":"openstack%2Fneutron~stable%2F2025.1~Iacd78020ee41399c3a8ce6cfdf9e37c4f865c90e","project":"openstack/neutron","branch":"stable/2025.1","topic":"fix-2110060-stable/2025.1","attention_set":{},"removed_from_attention_set":{"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2025-05-19 22:07:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iacd78020ee41399c3a8ce6cfdf9e37c4f865c90e","subject":"Fix ovn_db_sync for user defined flavor routers","status":"MERGED","created":"2025-05-16 21:17:56.000000000","updated":"2025-05-19 22:08:01.000000000","submitted":"2025-05-19 22:07:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950239-fix-2110060-stable/2025.1","cherry_pick_of_change":949816,"cherry_pick_of_patch_set":1,"meta_rev_id":"f16315603da852e3368d49a60b59a1ead2d16a77","_number":950239,"virtual_id_number":950239,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-19 22:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-05-19 07:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-05-19 20:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-05-19 20:14:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 21:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-16 22:47:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-19 07:28:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-05-19 20:14:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8a4f906915de3c3bad74173b018f640aa97012d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-16 21:17:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5732c117c50827763522f60ca17bbace375c2095","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 21:41:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f97485e36ded445aa23ceb3daff869b1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e932bbdfd634534a63c96363d71d80b : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5cf80b1e6fc44aa6b58151c659de8524 : SUCCESS in 18m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"228a06fafca36ec39dbe1d925d6870b264e25bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 22:47:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1ac7b14aa10484a9ad51852a875a78c\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/2eb27e4a36bf46358fdbffccc356e1ae : SUCCESS in 1h 18m 15s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/cbbd2c5e965e4bfcb55a73bd4388a3a7 : FAILURE in 1h 28m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c0c18cd4ee9460dbffed03ec17553d9 : SUCCESS in 51m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53144adf235a4798974e39b7ebc788e4 : SUCCESS in 11m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67c8608c3ba94571b17aa25c870bab84 : SUCCESS in 23m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d52559d453844e5813f787ca69ba695 : SUCCESS in 17m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d12a69e8ae420c9b8b52cc2ce3a3f0 : SUCCESS in 12m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1e077cfb23a8416ba9871f7b20d43b5f : SUCCESS in 1h 06m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ccb700d964bf4d73bcdfd005f191b7d6 : SUCCESS in 1h 17m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/266bfa4580c7498aa8e43a4fe3eeb777 : SUCCESS in 34m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9ac7cada3aa7401087c65ac3ce4a7557 : SUCCESS in 1h 04m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d327fa00ec8a4462b872238d6ec6b4ae : SUCCESS in 39m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/923b61b41b874871995c4e1213e97e95 : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f242af5f0f53eef3bf10e0036af94c1f33d9e94f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-16 23:59:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b731ea9285980ebeb732ebd5c824818c394f0259","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 00:30:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c09794e94be74b5c906382f42cc2bcf9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b04a37ff4cbb4b8db872d44955190478 : SUCCESS in 29m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab7d0c6ac030425cb65421579d61d539 : SUCCESS in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b83e342663f94713fbc0b3ce76251fa32fe626d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 02:08:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/472ecb2b0a7e473c8236573733d9e4bb\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/32732a0ad0004c379ca45df91c34bdac : SUCCESS in 2h 08m 24s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5cc157740ed2471bab811b4027058603 : SUCCESS in 2h 02m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbbccf82080740d4b0834451e18b7f97 : SUCCESS in 27m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff4d9f03d1434edb8384adbaff065d61 : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31b9b59e16644692b7863bd32b663605 : SUCCESS in 22m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9e7f84fc85845a3b6846dc0818b36df : SUCCESS in 21m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e9f4ecae2e46eaa0fccda3af73b38a : SUCCESS in 11m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/79b3428e0488405ea3611a7a65443e5d : SUCCESS in 1h 02m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b77cc3440ba74662ab0f1a9ac45986f0 : SUCCESS in 2h 03m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/af2f7b924a7740aca012d0e416699fba : SUCCESS in 1h 04m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d085c1df1aeb47f6850cd9e9c49b18fa : SUCCESS in 54m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3fe4c7b8380c485a84814c484662f7fa : SUCCESS in 31m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a81443812c8c4ceaa26ee525c779cc83 : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":1},{"id":"42f23a84691b98fbe05b526629ec6bf7e6f3641c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-19 07:28:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc4379098cab03d9f397589586cfc9b6cf14acc5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-19 20:14:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8095671d7ebf431c272b718fe304a74ecf30be7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 20:14:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27553a0120c85360fca1e93e13f21c4f48b61e0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 22:07:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f07c37463f54772bef60c4a0aa5044b\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/39b28094182b49ae9c3ea1a672b0e8c8 : SUCCESS in 1h 51m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2907f20555234ed3ae9f6faddb575fd3 : SUCCESS in 11m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/103d3dab832d42eabf7828a37695ec3a : SUCCESS in 18m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7e2ec92b99e43edbbd9d5ddbe5ef696 : SUCCESS in 18m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae2482b00d954a4e819f5f50b4faf174 : SUCCESS in 12m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2f028c2e9ed9454f952d952990663bed : SUCCESS in 1h 16m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c3c63d89b6e4658b976229109f776a8 : SUCCESS in 58m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/61d1e348ca7342da92bb766e860b092a : SUCCESS in 1h 04m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/46d96600f6444555b156f486e6e40143 : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"bda059ccfad5fe385aa23eedd70b4689b836e3e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 22:07:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f16315603da852e3368d49a60b59a1ead2d16a77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 22:08:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a49be53ba81042fb9aaa800e14d24f4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c148397b6f7499e92d4fb1001b66e59 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00130dc806accd77b0510d332e539dc2fc086781","revisions":{"00130dc806accd77b0510d332e539dc2fc086781":{"kind":"REWORK","_number":1,"created":"2025-05-16 21:17:56.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/39/950239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/950239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/950239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/950239/1"}}},"commit":{"parents":[{"commit":"3cd0491291a83dd157d2a2323088f475c4660ecf","subject":"Merge \"[OVN] Change the OVN QoS rule priority for floating IPs\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cd0491291a83dd157d2a2323088f475c4660ecf"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-05-15 00:13:42.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-05-16 21:17:56.000000000","tz":0},"subject":"Fix ovn_db_sync for user defined flavor routers","message":"Fix ovn_db_sync for user defined flavor routers\n\nWhen ovn_db_sync is executed in repair mode after a Neutron server\nrestart, it erronously adds user defined flavor routers to the OVN NBDB.\nThose routers are not managed by OVN. Hereby the issue if fixed.\n\nChange-Id: Iacd78020ee41399c3a8ce6cfdf9e37c4f865c90e\nCloses-Bug: #2110060\n(cherry picked from commit a37c2d2a1e27266f51259b33c5b2720f41f00b71)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00130dc806accd77b0510d332e539dc2fc086781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00130dc806accd77b0510d332e539dc2fc086781"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
