)]}'
{"id":"openstack%2Fneutron~950241","triplet_id":"openstack%2Fneutron~stable%2F2024.1~Iacd78020ee41399c3a8ce6cfdf9e37c4f865c90e","project":"openstack/neutron","branch":"stable/2024.1","topic":"fix-2110060-stable/2024.1","attention_set":{},"removed_from_attention_set":{"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2025-05-19 22:07:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iacd78020ee41399c3a8ce6cfdf9e37c4f865c90e","subject":"Fix ovn_db_sync for user defined flavor routers","status":"MERGED","created":"2025-05-16 21:18:58.000000000","updated":"2025-05-19 22:08:22.000000000","submitted":"2025-05-19 22:07:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950241-fix-2110060-stable/2024.1","cherry_pick_of_change":949816,"cherry_pick_of_patch_set":1,"meta_rev_id":"a33c036f8a5c9f85c34926a02b074982da421a10","_number":950241,"virtual_id_number":950241,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-19 22:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-05-19 07:28:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-05-19 20:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-05-19 20:14:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 22:24:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-16 22:43:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-19 07:28:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-05-19 20:14:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"ef092f98b6a522206160867716f151a3b9c354b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-05-16 21:18:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c2d2a6e2dbe8faa460eb798dcf8bad332cf423da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 22:24:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/378401b76f044b26b9dff2fd0e8d0368\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01e8f8ea7a1a43a286b15a2c50d1cda1 : SUCCESS in 1h 01m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a5205df4e8541efa334da78def6de0a : SUCCESS in 1h 02m 55s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bebf06d63e874d8e9dbe706b93519565 : SUCCESS in 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b66c42dce048fd52c6e2007b261bcba9c9936ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 22:43:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfff7c15a13240f7aa418514147d0aa7\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/83e87c3b8dda4da6982ebcdf38fca242 : SUCCESS in 1h 23m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28f2dc66971847c49195aef39804dc34 : SUCCESS in 30m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2429952a88640a59c09ea984c15ff3d : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51537fd2deca4420a363581dc45c82f9 : SUCCESS in 21m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0cf620f8b9e406dab1ebe4f7d72fe1f : SUCCESS in 23m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82aecf165f5443e78d26198147d98c49 : SUCCESS in 10m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dac73182a2eb4c52ac63b4afa2ca5fd1 : SUCCESS in 1h 21m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b286b7d8b6b24be7859f23a27b77ea30 : SUCCESS in 46m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/11c17ccf8e1b472d84f2098ca7522718 : SUCCESS in 1h 13m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/37ce0174eb6d44ab91f59d45f1853e94 : SUCCESS in 51m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d3284410ef5fd198c411a642c4f96febb32ee864","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-19 07:28:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ad08c1076aa2a94af1f464a2e9216e399b8db31","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-19 20:14:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"03ca6ac92473dc2d561f171cd033dc4c12c1ec5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 20:14:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d68de0a47fdeab07577278ab6d9f83df10e4527","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 22:07:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77f12c8ef8d1497a99b01288aa4de91e\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/e2669bead6c347ea9fd9c2265051fba4 : SUCCESS in 1h 49m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3373bbb7b76447f8ceb106a8c7b2491 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f341dc29e2c4b028d0a062b4d7da042 : SUCCESS in 21m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc155d6a6e634a40b6e70c3a6d390211 : SUCCESS in 19m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb69a75974e4b7291708ea1bb26248c : SUCCESS in 6m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6b899323f9949a59f3de6f73b625ceb : SUCCESS in 1h 02m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eecc6e6ca7fd4c01926aa818924225f9 : SUCCESS in 38m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8d2e6a62c16340ceb700a60d114cb794 : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"84d53151b0c7b4a5de4bdc39f90749aa7dfd772b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 22:07:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a33c036f8a5c9f85c34926a02b074982da421a10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 22:08:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45dc3ef9c916479d869312a1335db8d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9be7e5439f24920b37e53095ef2d173 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b20dcbfcfc0836e756a244bded55984402df4674","revisions":{"b20dcbfcfc0836e756a244bded55984402df4674":{"kind":"REWORK","_number":1,"created":"2025-05-16 21:18:58.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/41/950241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/950241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/950241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/950241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/950241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/950241/1"}}},"commit":{"parents":[{"commit":"0d4ef860549a552db3a9d04934154ad256662bbb","subject":"Merge \"[Stable Only][CI][fips jobs] Use stable constraints for tempest\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d4ef860549a552db3a9d04934154ad256662bbb"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-05-15 00:13:42.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2025-05-16 21:18:58.000000000","tz":0},"subject":"Fix ovn_db_sync for user defined flavor routers","message":"Fix ovn_db_sync for user defined flavor routers\n\nWhen ovn_db_sync is executed in repair mode after a Neutron server\nrestart, it erronously adds user defined flavor routers to the OVN NBDB.\nThose routers are not managed by OVN. Hereby the issue if fixed.\n\nChange-Id: Iacd78020ee41399c3a8ce6cfdf9e37c4f865c90e\nCloses-Bug: #2110060\n(cherry picked from commit a37c2d2a1e27266f51259b33c5b2720f41f00b71)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b20dcbfcfc0836e756a244bded55984402df4674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b20dcbfcfc0836e756a244bded55984402df4674"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
