)]}'
{"id":"openstack%2Fneutron~950303","triplet_id":"openstack%2Fneutron~master~Ifa79de5098cdf191346717bcb8e779a5b91c78a7","project":"openstack/neutron","branch":"master","topic":"bug/2110004","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-08-04 07:46:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifa79de5098cdf191346717bcb8e779a5b91c78a7","subject":"[DNM] Check functional timeouts","status":"ABANDONED","created":"2025-05-19 12:38:57.000000000","updated":"2025-08-04 07:46:48.000000000","total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b291b2efdfad6400e941b6a4116939899d1df333","_number":950303,"virtual_id_number":950303,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-29 17:25:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 14:22:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-04 07:46:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"7b5515d50a097440f888262562950bc73f0bbb10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-19 12:38:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab51d3a1524611e476241f2517219fdd7e671e3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 14:22:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a922bbd65da4d4f95378118a3c37fae\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/37f30e058feb43d4b46158f58a96524a : SUCCESS in 1h 20m 45s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/5ac60796d846439ea5b8fa1c7678a764 : SUCCESS in 1h 15m 28s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/bb355b45589f4703a79954ce8d0164ff : SUCCESS in 1h 19m 40s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/990b419d1efd4c72a4b835da3bdccc7c : SUCCESS in 1h 25m 43s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/352dc75bb52041e5a15cbc3a2f201980 : SUCCESS in 1h 30m 43s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/2896e99abed745369e046ddaa7e93747 : SUCCESS in 1h 18m 25s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/931bdc6537f94242bbcd3e3485f06eb0 : SUCCESS in 1h 25m 46s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/1c1e97442ed44b338bc7b32ad6f69cbb : SUCCESS in 1h 22m 27s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/772dd93e29e74b29895079f90c8861ca : SUCCESS in 1h 16m 52s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/e127b65205e948d189b6bddf648ea8c1 : SUCCESS in 1h 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9d1f8b2aff0935065493bf35bd3e85d15da6ccc4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-19 14:27:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9210bec8e0ede32a324c59c43415045a3f50415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 16:50:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68606d1825cf454a86f828b10a93505a\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/87aa13c20805477b99238189223bdcc9 : SUCCESS in 1h 27m 32s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/f08e82c68712440f83bf6f4be8f4cab1 : SUCCESS in 1h 08m 14s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/eb3322ea7a1642ddb75837edcf1b21e0 : SUCCESS in 1h 32m 12s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/2f4f13ddc88f442ca912b0b2448639fe : SUCCESS in 1h 31m 42s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/77a99d45d503432987c8549904eacb41 : SUCCESS in 1h 22m 51s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/364dbf0a475d4996820590157191586f : FAILURE in 1h 56m 13s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/9befb41a1c3042578af232ec5a18efa5 : SUCCESS in 2h 01m 19s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/3c89f6a4cc034a4fb7e462a27e77962b : SUCCESS in 1h 25m 43s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/b24e19d868514a538126fa5fcf2e7f31 : SUCCESS in 1h 37m 20s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/2248eeaa176b4f6eb7ccf5a0ac36dbd6 : SUCCESS in 1h 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e3c4b6259c641a1e5c472d673252a6c149750e3f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-19 16:59:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7893ce8bdf34596448b40d4e39a0276588ff2801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 18:53:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/183e6528052347e9b39b84824a772a1c\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/7bb3f812b7484910a150d009bbda3e4f : FAILURE in 26m 52s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/ad290d9a56324977b8cd10252f1122e3 : FAILURE in 1h 39m 29s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/9ad660287dfa477aa0f3809d012788fd : SUCCESS in 1h 52m 38s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/abfc4a4b294245af8e2600fa296e3260 : SUCCESS in 1h 35m 27s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/29f99ccfffe54d5fb60443c78d6a2d6d : SUCCESS in 1h 27m 44s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/9cb1d7cc4bd3461a96eb9fd20f90cc0a : SUCCESS in 1h 48m 54s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/cbb89af8056a4761ac82c73035b054db : SUCCESS in 1h 20m 04s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/89a6e27212204eef9a06cd1b6eb9e360 : SUCCESS in 1h 06m 22s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/96acb0cb0a2a49afb9215d2bfbbcf334 : SUCCESS in 1h 22m 33s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/0384c0ffa97e4490b70304c05135f189 : SUCCESS in 1h 50m 54s","accounts_in_message":[],"_revision_number":1},{"id":"033ff0ae2799978b7fd27aa5706263aa2b273e0a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-20 01:48:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6418e26b4e294a4c7a8f34695e1f87c31ac86de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 03:31:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c81b72f5643542a3832c6403416a49d4\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/2d215c37285f40ff98f5e34cad31c5a1 : SUCCESS in 1h 26m 09s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/540acf89b5b74d429e12d6aa2edf9e0a : SUCCESS in 1h 26m 46s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/40b3d07f8c2f465db7a07366e2041774 : SUCCESS in 1h 10m 56s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/b02535aa2b2f4505a2ef30438ed25750 : SUCCESS in 1h 13m 54s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/d4df745c04fc4cbaa172995262a48c7b : SUCCESS in 1h 12m 35s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/1f303f8fb1d04ce9b1b2ef3230e15e6c : SUCCESS in 1h 13m 35s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/d48d14209a1f47f79c81d1819a6d317d : SUCCESS in 1h 05m 28s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/a5c193301bf14619aa3168d6e0eed728 : SUCCESS in 1h 40m 40s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/da73c35e2e2f4dce83d3b83d751c9141 : SUCCESS in 1h 17m 54s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/e104203f09894de4b7e2fca5a9a9ce07 : SUCCESS in 1h 37m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4fce84a37f6552c547a1b82e070945606f3f58f2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-20 04:10:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a4c25a1ad03989806785ba8d4ff99503ad93994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 05:47:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e390c078e34586b528e807f732c39e\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/83bca99d11ee4e1180fc2614de8866c5 : SUCCESS in 1h 00m 04s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/72a2eb8e5fea485d823516ace439caa0 : SUCCESS in 1h 21m 45s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/3c0fd926eb2e4faeb08c9a9c92c5de38 : SUCCESS in 1h 23m 41s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/45e792443d74430ca90ce02a265400ea : SUCCESS in 57m 17s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/09c317e410fe4306aafd1bac5c645927 : SUCCESS in 1h 35m 49s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/d22587cd6e98471a9c33346f556906c0 : SUCCESS in 1h 15m 51s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/08f8fb4ea9d24ebdb0457483c5178591 : SUCCESS in 58m 10s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/92eb8ac9cb004da085287c10db4dc05e : SUCCESS in 1h 22m 42s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/d56e842484c34fc4b1874f8c7e332ec1 : SUCCESS in 1h 10m 20s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/e0bc19fe29ea43f68c4969189add56fa : SUCCESS in 1h 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f25084d033724cb73a19dd89c35e31a18701a46b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-20 06:46:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8142a317fe50bc459afeeb2fae99c370c5c3a46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 08:27:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3592c16aaf4d400292a7a2a318ef74f1\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/d6e9646c0f414b13ae114a248b6eec06 : SUCCESS in 1h 10m 54s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/b3cb9a997da647d58bcd235d25df291d : SUCCESS in 1h 03m 08s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/fb0d69aa6bf04b46885466f4df9401ee : SUCCESS in 1h 09m 25s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/822b0f41bc7c405d949e7411262becdb : SUCCESS in 1h 20m 35s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/ff80bfcc227a48c28a1c430d345f3ebe : SUCCESS in 57m 10s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/129e16263e1144e2bbd5b5a6315ec269 : SUCCESS in 1h 20m 08s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/89158695e5ad4707aa690a26f8985e86 : SUCCESS in 1h 39m 58s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/ebd73bea9f144384bec78b3241fd3e04 : SUCCESS in 1h 28m 50s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/4e9bf4fb26864f199714ff7ddc562f2f : SUCCESS in 1h 19m 13s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/dd8277082bbb467cba68c31989fa5e61 : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":1},{"id":"89aecfdeec3f24bed80f777d47f7cfd2e0e1b11f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-20 08:32:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a51b65b028b1e52a6f5601f82fe952720fcd95cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 10:47:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0eb3a555321d4494a7546f47344f63a6\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/bfbc27f07df64da280504cad5763e120 : SUCCESS in 1h 08m 36s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/13f7c0d637004d30935cabfbfb04388b : FAILURE in 1h 02m 16s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/c8be419ae2e14d3fa0465b9c2512ea52 : SUCCESS in 1h 15m 26s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/c9a14225f3e242d995133cf65c35b60a : SUCCESS in 1h 22m 11s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/50cc0852af7144f8993ba388d00bff26 : SUCCESS in 1h 20m 46s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/8a4a289678094ba194732811686ef026 : SUCCESS in 57m 02s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/fd41d1c01e22430d978920d06c0c1848 : SUCCESS in 1h 21m 15s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/613ab893be0947b7ba9495cfc72747ea : SUCCESS in 1h 24m 17s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/ff60b02178ba44d7a7b6a5dd12ef852e : SUCCESS in 1h 20m 35s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/f91939ed1cf740609a66fca85f25f3ed : TIMED_OUT in 2h 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"47ddcc4bdc070ba2fb7d7931d5a363814438b852","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-20 10:49:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82e48d3092a5aa62df0ff6c858a2298d99c898d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 13:03:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc35b0046ead413f9b88268beab6127a\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/8b465b7e4a5e42f48d4b46833e61c0cb : SUCCESS in 1h 00m 49s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/26f60b4ba3b44f91bd028b73a3165161 : FAILURE in 34m 07s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/4ccfb29dcbad42f4924ef2fcfe41d798 : SUCCESS in 57m 46s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/0316699a0d69451e8cd68d55a23e1f8e : SUCCESS in 2h 11m 35s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/eacac102ec6f445aab3b8fecb028167a : SUCCESS in 1h 04m 06s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/d04a7d675fc044398b318591b3eb7d56 : SUCCESS in 2h 06m 51s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/9f83817c71514bbabc11a8514346820b : SUCCESS in 59m 47s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/ba4ba44ee8dd42619338109fa1b41393 : FAILURE in 36m 52s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/4f4ea1d3021641db8db2dad237d99271 : SUCCESS in 1h 01m 53s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/63b84e43156a485a958765e9c13f443c : SUCCESS in 2h 13m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b22d5373f679c84802561d683be848f1adc7fa70","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-20 13:09:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c246bf1a286b49cb2354720d47665b93bb99a0d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 16:53:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d20f5a391ecb478386b675a98d86f818\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/591d186fe531435aba1c51198b7e0b33 : FAILURE in 1h 43m 57s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/5970686e0ece4bcf8baa43d5924661e3 : SUCCESS in 1h 30m 02s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/ecc58df81e9b412ab60aa6c26f231cd6 : SUCCESS in 1h 27m 20s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/84f7005024ba4b8d9ba006ec103fd77b : FAILURE in 1h 44m 44s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/3cfaae64e7804f3cb2180dd38ece94c7 : SUCCESS in 1h 03m 19s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/ed4ef793f2e8407b94230c5a22aef0f0 : SUCCESS in 1h 19m 54s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/8ac4ecd381bf47788ff94ce011e1fc20 : SUCCESS in 1h 16m 04s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/5d6066f1398042be9f58eba6f195f71b : SUCCESS in 1h 28m 37s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/53f17bfc491d4eaabe422a24cafb41c0 : SUCCESS in 1h 23m 30s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/f9bd4918bc2449bfbe6f44c6e88ed40b : SUCCESS in 1h 14m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4716cc8a088be947a5dd3034b23049ad20d3340f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-21 02:29:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbf95a1ea5e098122bc2621a5c4ffe708e1707e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 04:07:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb3bcc2faee8445fa29c77a09849e463\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/610cee45ae9c4851846009b9934d32da : SUCCESS in 1h 13m 35s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/d4d72a75f6444c0597c7d86b0f6177bd : SUCCESS in 1h 18m 07s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/2595e1bc2f0b4742a37cb083cfb9850e : SUCCESS in 1h 06m 29s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/173068033d1a45bbb5ecd0124762dcd0 : SUCCESS in 1h 10m 54s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/265f04d71d1548a7ac3c3fc17c71b15c : SUCCESS in 1h 21m 54s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/fb0130879fd1468691b6e86f5f8fe98a : SUCCESS in 1h 19m 31s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/f9dd44c7b7844fae93a235c9c5585bee : SUCCESS in 1h 12m 56s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/603999929ed74388b2081a86a5e1fb72 : SUCCESS in 1h 09m 42s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/d1b7cc8825f14d5ab8db782f196564c0 : SUCCESS in 1h 14m 15s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/4626388067524cd884037c4ebb01fa5a : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"53e35ef1d8d5821d8d047e4a1d7e758be719924f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-21 04:31:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6b83e2d19038d3b8bc783c55bd37c5a58e6a4b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 06:09:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58251f033afa4901b09499b42ef75c21\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/cf049ef2134243de8f587779b2488bc8 : SUCCESS in 1h 25m 40s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/b45c158e2d684cc597b3a021fbd6f361 : SUCCESS in 1h 19m 35s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/b6703fb808c04379a9ebde87724125c5 : SUCCESS in 1h 19m 03s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/704609bbaa5d435b8974028551050900 : SUCCESS in 1h 20m 43s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/3eb2a4ae7f2f4457b288258d73293540 : SUCCESS in 1h 22m 10s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/a908a80a7e924a8f8642de778878824a : SUCCESS in 1h 24m 56s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/b03b26fb385f4733b6fddb966c7c32eb : SUCCESS in 1h 22m 57s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/bc9897d897be4b7f94f5b78c1c23dd06 : SUCCESS in 1h 22m 07s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/1f69d11abeae4ae884cfbd1c9fec33ee : SUCCESS in 1h 19m 17s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/4e3bf20a6fd643c79e646755ee59ecc0 : SUCCESS in 1h 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d21af5a02e4a6930015f07a051fe9463e666c840","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-21 06:09:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b288fe1db2a08b02da271bc9adac4470bfec8e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 07:44:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f646081e24d1423783d2c15fcb76761d\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/e74a7875691748e8b5659a8b4d78352e : SUCCESS in 1h 23m 24s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/c6a5a1f9e2104d14b9be775db544259a : SUCCESS in 1h 27m 03s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/27cec64029a74a86b652cd3bf160e559 : SUCCESS in 1h 23m 08s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/71252910e98c4ef9b731856ef155729f : SUCCESS in 1h 32m 21s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/f6717f39f40a45348edfa04716cae523 : SUCCESS in 1h 22m 25s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/beae031344b443cc8745d30c3a0d6136 : SUCCESS in 1h 25m 14s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/573089e4580b458382ee5b6918308dc2 : SUCCESS in 1h 26m 07s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/0cee8c7a78ba4438a3b0ea2235c799ac : SUCCESS in 1h 21m 23s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/24142344b8c142068d65aa61b6c8f182 : SUCCESS in 1h 29m 34s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/44176823216749e890fd7538eab73f50 : SUCCESS in 1h 29m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bd842f25fb98db9b8bc1d94bc37ee59ad8287802","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-21 07:50:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"593d46d17ecd9cb56c7c7d2fbcfadc52dc5fc8f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 09:34:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b4ad72125f48fab008ac7cbe941819\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/c578aa1a7ccf4d58bdfb8ddefa197389 : SUCCESS in 1h 43m 53s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/cbb0ee4243914085b93b0f129e14fb1e : SUCCESS in 1h 37m 27s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/366fa7e2d4904ed2ba576433417fe1a1 : SUCCESS in 58m 24s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/eb8917a9f1da4932a6bc0e43ba6d6d29 : SUCCESS in 59m 05s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/e17cc10e296040fb99a22ef6f0170943 : SUCCESS in 57m 32s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/833cab5e9d4d4d5988477fa1253fe0a0 : SUCCESS in 1h 22m 33s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/5f3bd7c976d84d1588fa68921a471fcc : SUCCESS in 1h 31m 52s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/cb785ba3f4fd4edcb4c99f78d6e0a05f : SUCCESS in 1h 27m 43s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/d787662202b64983b626150acebd3d19 : SUCCESS in 1h 30m 43s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/9abb36ed62fa44ffa497eeaf49883855 : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a8f3d4cc508b7b6c133ea8ffa8a744132377ab67","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-21 09:42:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ca8d3c46d1df31c9c0250fbcdb12367246c1fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 11:35:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9497d81ce6454be89ea9258a636d0caf\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/ef29340a1d32404bad80b9c83b2c3234 : FAILURE in 1h 05m 58s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/6b02d6af26ac489a8642d68770894f9d : SUCCESS in 1h 22m 52s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/d88b752cc430498791264d860e5130b0 : SUCCESS in 1h 17m 00s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/a45f3829c2f64e558a03fbf447acfefc : SUCCESS in 1h 25m 38s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/418b9b9db9924d15bbfe8b2cf5b83c2d : FAILURE in 1h 08m 01s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/26027213d0c94754af7e7fac51b93212 : SUCCESS in 1h 30m 09s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/810c17ee428e49bbb38a810ace0d05d3 : SUCCESS in 1h 09m 58s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/60c0023904f348af8cb28aefff354673 : SUCCESS in 1h 48m 43s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/fe8bb8dd7d684dab9d9b779c9f5e80ab : SUCCESS in 1h 35m 23s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/45f2ea3476774518a8ddd739e6e475f5 : SUCCESS in 1h 35m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f438ba69b4a64d8fb7a17dfeffcfe0b92a2597af","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-21 11:37:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79eca5527b1539e6e1e6d9f1bc0faa3463c97be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 13:32:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e4fd2b829a44ffebc2cc8e5cbb3212a\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/b1a31836da5b47528699df1383a5f6f9 : SUCCESS in 1h 26m 44s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/78297a79261b47dcb423bb3fe3727875 : FAILURE in 1h 07m 46s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/faac4a33c3194f10b73cfd61b5d51448 : SUCCESS in 1h 23m 36s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/3d9754da4a4740dcac9a216d08676eb1 : SUCCESS in 1h 14m 12s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/d2617608a0624019a15a6c67b049a52c : SUCCESS in 1h 24m 28s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/9fb7fb2b73f04c709c512053f5e793e8 : FAILURE in 21m 03s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/d8980d2177b441ed9b8b34b153475630 : SUCCESS in 1h 22m 03s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/d17027d18dd94d9788aafdf0ab2475ce : SUCCESS in 1h 54m 02s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/efb133d6bfe0456cab7a414e956360ac : SUCCESS in 1h 51m 50s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/2b9276574a8e4d54ac31bae603666bd5 : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e79614effe2fa4fb0bd93dc139c030880e3eff75","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-21 14:00:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18b83c10ba65b1161ba998decad052378b18fa02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 16:13:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b2790bcda61477a84fc4a796b2692b2\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/cae669b27f59470aaf742ed4b6e62573 : SUCCESS in 1h 08m 54s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/e490902ca37b46a18f924fb03100eee4 : FAILURE in 29m 08s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/71150976a9b14267b4018ac5e88c6f2e : SUCCESS in 1h 12m 03s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/c98e2a18e829442eb1cc82650f055f5e : FAILURE in 55m 47s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/6ef594b7c4f24b158899d05c163429e1 : SUCCESS in 1h 13m 25s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/e94673a02b27443a8ecc1a94f74b8bc5 : SUCCESS in 2h 01m 43s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/056237a722384a3a9cad6d8a1509dc4b : SUCCESS in 1h 30m 13s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/89a1a21eabfa40f2ba583b9e5e71b7b5 : SUCCESS in 1h 35m 42s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/f3d81aaa45a74e29bc02c3301ae9d9ca : SUCCESS in 2h 12m 18s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/4807293e16a6467788c5dcbde33caca4 : SUCCESS in 1h 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"385a8a1b37d8acf135db5f7574fb78dc9801fba5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 01:25:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af118691ef7ee5168bc3d3669c2951b3ab5a8037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 03:22:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4986e7a8efc94a70aef44f65fd391049\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/3387c27db772467aaf2991d840b8db44 : SUCCESS in 1h 27m 28s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/4d5948885f0d42f6989a201079aca338 : SUCCESS in 1h 38m 05s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/179875483da1488f8adb8726dd04ca9f : SUCCESS in 1h 18m 57s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/2181bc171a0e48abb8f5be324a306490 : SUCCESS in 1h 37m 50s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/cad94b9402894ba38426a5cec3e0816d : SUCCESS in 1h 25m 45s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/e194200552614a6cad204d16ccc10739 : SUCCESS in 1h 20m 27s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/d66038a4ce424ad2a305fe84b9fec397 : SUCCESS in 1h 23m 27s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/9297cd1b0f9c4563b0e2669cc7ffe21e : SUCCESS in 1h 21m 21s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/d50b108ba7bd4757a7e17ff86d7a0971 : SUCCESS in 1h 42m 35s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/21d13741c4ad4c0aa6f7bc4d9a63e07e : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7eea3db1705cd95bb75324120af7ea63996b12a0","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 03:24:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a003ea447c4fef6a8a620eb49e9ab2747fc72b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 04:49:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a134c30258f49c7a6f0efbff974850f\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/2c21f23ae7ec44cbbd1b4a9f3b928d52 : SUCCESS in 1h 20m 40s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/7dbd8f4c0df54c70a7eddeed68385552 : SUCCESS in 1h 19m 52s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/0dc3c5a264b14922b5b0b64dad9fb026 : SUCCESS in 1h 22m 20s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/2d2f041d3fb0469a85f943ea7febb25b : SUCCESS in 1h 19m 06s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/a0183442018343f8b8ba7a13f735ba2c : SUCCESS in 1h 22m 47s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/06c1cc87d98d4f229db18588f14cb0b7 : SUCCESS in 1h 18m 00s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/50be0fc60f5349f0a02d7b7f1f2cb421 : SUCCESS in 1h 17m 01s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/74dbc5c420834c1ba3ca8d0b1bfe5262 : SUCCESS in 1h 19m 54s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/d719c034b7a941b5a23df53a2b30d579 : SUCCESS in 1h 04m 15s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/1996b6c2e0314e0e8224beea6b525b77 : SUCCESS in 1h 12m 57s","accounts_in_message":[],"_revision_number":1},{"id":"42d28127466d0084626a5116975391a1b873592e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 04:56:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61a5624c52c74b30f081f7e2e51932af2cf82bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 07:47:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd2fadf4f2594111a4c7589399c53cae\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/1a1db0db33714f1d98498c8288e9e9f5 : SUCCESS in 1h 20m 49s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/68ffb5fc720c41a0bf6270cd766675ff : SUCCESS in 1h 50m 38s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/c9dd342ed9704acf862b8a0dc503f211 : SUCCESS in 1h 48m 39s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/bef7e6a1337d4839a7a366c3ffefd369 : FAILURE in 44m 06s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/b2f44135c94f435d93ab5e84eb83c687 : SUCCESS in 2h 00m 14s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/ce5f6af635e84ff0a05ee5db1814bedf : SUCCESS in 1h 18m 36s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/968075c099164e9fadc94a97fe0ab022 : SUCCESS in 1h 50m 14s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/24a68ade24674e17a8c67167d1bc4409 : SUCCESS in 1h 09m 02s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/526c6c133093424eb9f67df8295dd960 : SUCCESS in 1h 18m 33s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/943d1a05465c42109d7f5ea78e881045 : SUCCESS in 1h 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ca5aa1814300538df5c4f8dd84d3c68911696a35","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 07:50:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88d5ac4698116c9ee05d87a56e35908c90b56c0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 09:43:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57ff714e04b646b7b0a55be9159ae13a\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/6770643b97b44c1db9f6154d17a646f2 : SUCCESS in 1h 14m 19s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/02836f8e3eef4bac9c00e4f772258e92 : SUCCESS in 1h 26m 28s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/09ca114656ba49deb8d67dea6418cb39 : SUCCESS in 1h 28m 53s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/90f0b57ddb0740fd82f858ab88bcb3ec : SUCCESS in 1h 10m 10s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/ebc5a90540694a60905c81ce7f638e4c : SUCCESS in 1h 20m 05s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/43e4399c9482408e914c0893b99fdb65 : FAILURE in 1h 07m 40s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/37e31617b51d4b22af7fc379dcb19bb0 : SUCCESS in 1h 28m 09s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/75476474188f4ccaacb77627d7c091ce : SUCCESS in 1h 24m 40s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/05a0ac3b76b1418498fd58a18bf32fbf : SUCCESS in 1h 52m 12s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/4c10d7fbc33a4375893085bcb9061145 : SUCCESS in 1h 52m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b6bbaa06d59baf2bbc52370db8f5cc716ad49fa6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 10:46:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eecb06f7c5484d1243c221dc79aecd89c97f1c1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 12:36:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b5b42457d4747e38f1e789a10808d90\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/d20a9cdb5a4d40c48c9bd805e23bb3ca : SUCCESS in 1h 20m 28s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/7913476722ed428bbfb33805c1d95d44 : SUCCESS in 1h 24m 53s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/e1c9cc9f24a8488cb75cab9a6a1cb97c : SUCCESS in 1h 21m 20s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/0c3554d9e01e42c29c7b0401d3306ea8 : SUCCESS in 1h 28m 22s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/8247d2e99e294f8593f04e0ca06a6e8b : SUCCESS in 1h 24m 14s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/8d047f351660497a9de66539b2904361 : SUCCESS in 1h 20m 29s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/4572fcf937c649d5bbb2e6cd10ef977e : SUCCESS in 1h 49m 20s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/b54dfc5e5d104ad09da850469a6533c8 : SUCCESS in 1h 22m 45s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/91f40d8507ee42b4b2491ea96472a739 : SUCCESS in 1h 47m 44s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/cb1eaaa8d4e043d2aff53a9fc52e5155 : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b7f8d012216b8e69b4edaa473c5a7e535168e97d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 12:45:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f70c00952fbe2f244a3e92cf4e69530ac0666e9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 14:39:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/687a0fb89b8b4cfeb437e0de46298a32\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/b569df47f891435eb70245aa537eb940 : SUCCESS in 1h 53m 34s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/6615b1fb390f4a48a9e6ef342034afcd : FAILURE in 34m 19s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/3c15fcfeabef43aaa8280b868e8b5574 : SUCCESS in 1h 19m 58s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/711d71cc196348a7a53975c928f61e19 : SUCCESS in 1h 07m 39s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/62c88e3687ad4103a4b826498c37bdda : SUCCESS in 1h 19m 50s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/e18834ef8b23488b94d2db23cc5a3ac2 : SUCCESS in 1h 14m 10s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/97512489335d48529599ba56c896ff85 : SUCCESS in 1h 17m 45s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/72ae415d277640c0a2a7c6df9a760344 : SUCCESS in 1h 19m 50s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/64b3fb184e284121a4e677caea49bfa9 : SUCCESS in 1h 24m 16s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/a7d377c46ec34850b3479e570508c787 : SUCCESS in 1h 15m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4802c913821218c8bd86d76930811d2164646bc5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-27 07:58:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0863ada5f3c34c8bb5a089fca36f888a13673c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 09:55:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6022ce7da3c14935b526c73944d1e4d9\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/4f766a506e7143e1b0bc98e80c8cb07a : SUCCESS in 1h 28m 51s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/58bb1784c6634faeabe2360575cce4f5 : SUCCESS in 1h 15m 37s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/a5d515aaf7d04381a2197ea346dd583d : FAILURE in 1h 11m 00s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/20bdea044f0b4b39a056d433502a45ff : SUCCESS in 1h 43m 49s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/c4b16c1de2434a9e8eb450cee156273c : SUCCESS in 1h 19m 41s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/ee575800b5654f84807c8c894bff1299 : SUCCESS in 1h 23m 58s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/762482a974544909825f0d8518936f26 : SUCCESS in 1h 06m 05s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/c2f3e39b34af4f6c9cf7bc474cc80fdb : SUCCESS in 1h 08m 38s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/84ec5bd290874a84bee05c1935af165f : SUCCESS in 1h 02m 48s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/1f4fb90d491945b6b54db3f468517003 : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":1},{"id":"906adc53cfc94184a417355aaebc917e050135f7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-27 09:57:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57aabdc1fe4478081de2067e167d6de116d04c42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 12:12:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d32e0c054cf41dc8080c00019ee22d6\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/df23c4d1b56d4340be85edeabaef5980 : SUCCESS in 1h 21m 47s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/0f79b38bcb4a44d79ad8313bfcf5144c : SUCCESS in 1h 24m 29s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/36c1fbff20744cccb294244d3d87701b : SUCCESS in 1h 25m 26s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/ec0c1fff3848490aa7ca45674581839d : SUCCESS in 1h 16m 12s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/e667d6feb2294dcfa23d0f55f2ae6cb7 : SUCCESS in 1h 18m 39s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/0925e636dd744413ae817ef6dbba94d9 : TIMED_OUT in 2h 14m 15s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/a90695ad05e4458283b959adb0857c7d : SUCCESS in 1h 16m 39s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/01484ee087fc467f8bde4984294d133a : SUCCESS in 1h 20m 09s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/f6d4981bd56c42e6bb9939082205abb9 : FAILURE in 30m 04s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/0883294c10704faa8f3cb58c509b666c : SUCCESS in 1h 22m 37s","accounts_in_message":[],"_revision_number":1},{"id":"34d32dc35be40538465260d5d5d8a1eeca735e28","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-29 02:27:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b108fae2248ca69d3d6774c88db748dc019cac9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 04:46:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4260ac5505f4657afda95d02237ee50\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/5214a1ee239748ab8b17aa141af66692 : SUCCESS in 1h 20m 34s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/4b0c8906e04a49499ef8e7de98aea9e0 : SUCCESS in 1h 22m 24s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/f746a5b50a28410797d673f68e2f7291 : SUCCESS in 1h 19m 50s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/dc6b596553be4028b5410619c6c18351 : SUCCESS in 1h 17m 30s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/8f5a79cb1e244aae8b77e4076ad5eff6 : SUCCESS in 1h 23m 03s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/6fbed5374d614716bc1e764022ebf3b3 : SUCCESS in 1h 05m 06s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/34cb16d853db47e891a14e71f9b697fe : SUCCESS in 1h 18m 29s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/c0b8413396a44f1594466ef806678170 : SUCCESS in 1h 04m 52s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/a84d28a4c8f14c6783df9485d7c03b5c : SUCCESS in 1h 15m 28s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/3e0e5582704c491fa8a11a7a27b70ff6 : FAILURE in 1h 05m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fefb14a4d0ade5f6682bcbb57d6698e45877031d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-29 05:53:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8613db42ab762cd8827e6f931635ea3be1fab84e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 07:54:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fae6a9a152cd41d89d0845b12d3fa7f3\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/393822fc143d4d51a3c0fe592c90184a : SUCCESS in 1h 06m 01s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/81d821f25bcf4cffb5642bbf97c585e9 : SUCCESS in 1h 02m 15s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/1be48cdaeb5a4945a6d4c748a1f918c3 : SUCCESS in 1h 00m 33s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/82dbd202fe6d4de1bd3acffa638dbd71 : SUCCESS in 1h 30m 32s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/6766e463820940bdac888870a0f0733b : SUCCESS in 1h 20m 43s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/5146da107fc344d194d3f37d9e6b5f50 : SUCCESS in 1h 27m 23s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/fb1ea5d64ba94d6e8b6391fab94de18a : SUCCESS in 1h 28m 13s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/fb8928802e6246939857cab0e45a1b29 : SUCCESS in 57m 48s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/92e8557eec4d4946acf9bf329bfd230c : POST_FAILURE in 1h 44m 26s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/ae27bdbda1894216a52e31f1b1c69fd0 : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":2},{"id":"b41e1c59555052b8bfb732be13c1f838d3f046ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-29 08:14:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a887d8582fac399d98ecd2fb5ef70ce1c42f1a6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 10:27:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bb23c79721a478d9228c91005480406\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/89a5c87bec9e42199472bebeccaf5828 : SUCCESS in 1h 23m 22s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/a13b0b702c0a421b8c540525b3f91444 : SUCCESS in 1h 29m 05s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/f614680de8874faeaf77530c63937911 : SUCCESS in 1h 29m 55s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/a921b24bb49445088e9d04704484b5d2 : SUCCESS in 1h 22m 01s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/67148804304c4d9db5df340b6ba87d9d : FAILURE in 1h 57m 53s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/f47b41676364456a99e390729b41be4d : FAILURE in 1h 48m 26s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/549e5878ca2b4fae884f49ecdd8fe043 : SUCCESS in 1h 20m 43s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/c983d227657140c8a4b36f6ce7c39f5e : SUCCESS in 1h 18m 10s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/dc5e0a73553f4628928333d54dcef10e : SUCCESS in 1h 20m 53s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/dbdd8216e4b4497187f5b7f4bbac7ad0 : SUCCESS in 1h 30m 02s","accounts_in_message":[],"_revision_number":3},{"id":"7c923e5ce24be91744ff6a5a4eee0e2ad88a7d83","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-29 10:41:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d0451a41dc60e9dde55c69e20a00a811b4205f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 13:15:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e16d6a24cd8d4438b2851f3f58187631\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/3ba1f1bdac154f7d838bbfeca6804cb0 : SUCCESS in 1h 48m 54s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/d3c6fc0c73184fbfb56880f9eff8aa14 : SUCCESS in 1h 24m 00s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/7df9d467066a478eb8b9e0650f361968 : SUCCESS in 1h 48m 36s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/036e1c4654e14ad9953b8bd3a74a5017 : SUCCESS in 1h 31m 35s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/ee5b2ff0024e48cba21ea4db210f131e : SUCCESS in 1h 23m 33s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/f2bae028ff4c4f31b441555dfac86871 : SUCCESS in 1h 09m 52s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/9a5d36131d8b4645ad01bbe7fc722130 : SUCCESS in 1h 32m 09s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/183ea6aab5c44f738c50f41260a6c976 : SUCCESS in 1h 23m 27s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/c841be9272de4397857f4f6f873f2f0b : SUCCESS in 1h 22m 14s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/808de14be4114e809837a0b88e9c46c0 : SUCCESS in 1h 22m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ac0ac719ca6e1f7465a2f530714c86d5e05f0c50","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-29 14:19:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d42057b2c9c9d9009be04a8e2333f96ed83144d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 17:25:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32841e79739647ecb8d941d1d0d6140e\n\n- neutron-functional-1 https://zuul.opendev.org/t/openstack/build/ef3463899f304d38b711e65f5f9a8017 : TIMED_OUT in 2h 53m 40s\n- neutron-functional-2 https://zuul.opendev.org/t/openstack/build/4b4949b9f6764c6198b70e6fa5288c18 : FAILURE in 1h 39m 47s\n- neutron-functional-3 https://zuul.opendev.org/t/openstack/build/1c79cd1067374b3aab60335b48537d65 : SUCCESS in 1h 12m 46s\n- neutron-functional-4 https://zuul.opendev.org/t/openstack/build/77f7cda0fc1c47d7a7f2126644a8258b : FAILURE in 1h 07m 49s\n- neutron-functional-5 https://zuul.opendev.org/t/openstack/build/3fb02eac025843f586ba657186652314 : SUCCESS in 1h 08m 49s\n- neutron-functional-6 https://zuul.opendev.org/t/openstack/build/21002b8cf73143c3ba1c5e028262fb3d : SUCCESS in 1h 05m 39s\n- neutron-functional-7 https://zuul.opendev.org/t/openstack/build/0258ea9335b94e438cdc3fb41dca31f5 : SUCCESS in 1h 31m 13s\n- neutron-functional-8 https://zuul.opendev.org/t/openstack/build/88036a2ea3814a6aa7beafa0f99a1ad7 : SUCCESS in 1h 36m 00s\n- neutron-functional-9 https://zuul.opendev.org/t/openstack/build/8bff87dca6ed4979b70b5a37b0758b8f : SUCCESS in 1h 10m 49s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/9084f883af8640ef9a7830813629c507 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":3},{"id":"b291b2efdfad6400e941b6a4116939899d1df333","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-04 07:46:48.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"caf38fae3f56c45ce754f82092b419e4d142b8ba","revisions":{"32641484a733f41e891abb7132421cba4b159795":{"kind":"REWORK","_number":1,"created":"2025-05-19 12:38:57.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/03/950303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/950303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/950303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/950303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/950303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/950303/1"}}},"commit":{"parents":[{"commit":"a37c2d2a1e27266f51259b33c5b2720f41f00b71","subject":"Fix ovn_db_sync for user defined flavor routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37c2d2a1e27266f51259b33c5b2720f41f00b71"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-19 12:37:37.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-19 12:38:25.000000000","tz":330},"subject":"[DNM] Check functional timeouts","message":"[DNM] Check functional timeouts\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/950298\nRelated-Bug: #2110004\nChange-Id: Ifa79de5098cdf191346717bcb8e779a5b91c78a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32641484a733f41e891abb7132421cba4b159795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32641484a733f41e891abb7132421cba4b159795"}]},"branch":"refs/heads/master"},"1d3f143b7465a2c161bd4f8949f4035f5a7390dd":{"kind":"REWORK","_number":2,"created":"2025-05-29 05:53:30.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/03/950303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/950303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/950303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/950303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/950303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/950303/2"}}},"commit":{"parents":[{"commit":"a37c2d2a1e27266f51259b33c5b2720f41f00b71","subject":"Fix ovn_db_sync for user defined flavor routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37c2d2a1e27266f51259b33c5b2720f41f00b71"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-19 12:37:37.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-29 05:53:15.000000000","tz":330},"subject":"[DNM] Check functional timeouts","message":"[DNM] Check functional timeouts\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/950298\nRelated-Bug: #2110004\nChange-Id: Ifa79de5098cdf191346717bcb8e779a5b91c78a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d3f143b7465a2c161bd4f8949f4035f5a7390dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d3f143b7465a2c161bd4f8949f4035f5a7390dd"}]},"branch":"refs/heads/master"},"caf38fae3f56c45ce754f82092b419e4d142b8ba":{"kind":"REWORK","_number":3,"created":"2025-05-29 08:14:49.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/03/950303/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/950303/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/950303/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/950303/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/950303/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/950303/3"}}},"commit":{"parents":[{"commit":"a37c2d2a1e27266f51259b33c5b2720f41f00b71","subject":"Fix ovn_db_sync for user defined flavor routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37c2d2a1e27266f51259b33c5b2720f41f00b71"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-19 12:37:37.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-29 07:49:10.000000000","tz":330},"subject":"[DNM] Check functional timeouts","message":"[DNM] Check functional timeouts\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/950298\nRelated-Bug: #2110004\nChange-Id: Ifa79de5098cdf191346717bcb8e779a5b91c78a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/caf38fae3f56c45ce754f82092b419e4d142b8ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/caf38fae3f56c45ce754f82092b419e4d142b8ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
