)]}'
{"id":"openstack%2Fneutron~950327","triplet_id":"openstack%2Fneutron~stable%2F2024.2~I96ad1c224455c1835b5648c8eec64f2a865f9904","project":"openstack/neutron","branch":"stable/2024.2","topic":"bug/2109354","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-09 13:20:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96ad1c224455c1835b5648c8eec64f2a865f9904","subject":"Add provider network resources filter for \"network\"","status":"MERGED","created":"2025-05-19 14:49:12.000000000","updated":"2025-06-09 13:22:44.000000000","submitted":"2025-06-09 13:20:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950327-bug/2109354","meta_rev_id":"c7245dba01bf6f1fce870ad1eedcd03995825a8d","_number":950327,"virtual_id_number":950327,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-09 13:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-06-05 07:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-06-09 11:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2025-06-09 11:02:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 16:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-19 17:03:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-05 07:56:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2025-06-09 11:02:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"972fe3c1940539842689937f309053671656b196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-19 14:49:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"430b61b5e794f1ce4921ba2eafbeaf7c51df3ac8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 16:47:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d69947118bb420d826357bd4e92290b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e1e13e6d70546b69da00a2f2ae1a6dd : SUCCESS in 36m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3e799793a5bb47339fbe2da885488773 : SUCCESS in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78762af538b9f7d14c0a8455344790b908752a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 17:03:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efcf7122d14b4d558b6aaac49e0bf817\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/4bd6444736914094a76494ec9e8b0005 : SUCCESS in 1h 49m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/8b7f1f97691e4062a58f18dc00c0d73f : FAILURE in 21m 17s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/c0a506296d7b44a2b7d1b811751751c6 : SUCCESS in 2h 05m 50s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/d19a419b2b7d4b67b2899f9cb78812d7 : SUCCESS in 29m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6280e0acd60a4bc094f0273d63ea753e : SUCCESS in 34m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2bb04b07c7f4de1ae1a00a30f96511f : SUCCESS in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89d089141b80460591fb0e1ae882529c : SUCCESS in 23m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed62bb93d013400bb71eb276407d2fb5 : SUCCESS in 20m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cbafb6bdbfdd4d2f958389b81b5eb546 : SUCCESS in 21m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2bed8bcd7345d4b21bb156bcd07306 : SUCCESS in 11m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/44836a3a90894756b5d331f6e3141147 : SUCCESS in 1h 10m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e614fa4b40eb42ca83f0bbfd55ea2c0f : SUCCESS in 1h 46m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2aaf549bd00744f0a1c8e628049eb585 : SUCCESS in 49m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5359c4ec504f4d59bc28f60a18ee27fc : SUCCESS in 1h 01m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/abcea6b386a046b19454d3d56c2291f6 : SUCCESS in 52m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2972e29283e941f48183eb12de529d5c : SUCCESS in 1h 04m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e4f4d8237894ab591e30fa20094efff : SUCCESS in 58m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1d34bdf228514895b4e210f78894b12e : SUCCESS in 1h 12m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7b53468030eff795e0f9b2df2424a6bf6cf47520","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-20 05:48:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"108af565a5ae2e22c1665d10e64a74cc38f0e8a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 06:24:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e79824dde974665b3d2dfeff5855715\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65de139d88bb4ab99619056279240d7c : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5954f3731257469d88ddf9976f64ded4 : SUCCESS in 33m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c35724c6681149a6a702f3a3072312fe0da9ed47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 08:03:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7902e88e4fae4acbb395dae14d3736f5\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/e13f10e4f0ed4aeaa42da7e981e13ded : SUCCESS in 1h 52m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/83158a5873924168aed8bb339b14101b : SUCCESS in 2h 14m 17s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/b14bdf82883a4bf49fa50ad29968fa64 : SUCCESS in 1h 36m 28s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/7f81de7031ae42698ceceafbb2e02e8b : SUCCESS in 27m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6e16d22573449e8a6c2b0ee88e0d300 : SUCCESS in 40m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a879d8ef1a246c8a573fd0a9a5b4381 : SUCCESS in 9m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f54647ec7774a38815ce564bd9150c9 : SUCCESS in 20m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/addc83a8208b40ed9ac4abd4fabd0d83 : SUCCESS in 20m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a17ef258c43d455ca014eee357487303 : SUCCESS in 24m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f29a2d806c4a0f9bba74771fe02c79 : SUCCESS in 11m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c462437b5754ed791f1ff0ac65d2422 : SUCCESS in 1h 27m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a9e39e81a0142928e7694b5c2f408db : SUCCESS in 1h 19m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb51fd6d9bd643d0b9478d02543ffa61 : SUCCESS in 51m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdd644555c3f4772b527f5af535719b2 : SUCCESS in 30m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e90b80a3a2f54782a26e3db86e20db78 : SUCCESS in 1h 01m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f613da3012384a66ad49b955475ee5dc : SUCCESS in 26m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7045c6442ec4360ada76efebbbff62c : SUCCESS in 58m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e162db6db99e48898d15811d6934d812 : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c4ac56b317367e95927bdcef709da915fa782c19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-03 13:49:36.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"552fed3fe654fdac46bbf3df283c5261e638d4fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-03 13:52:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25dcbfb6705450eec006a2462c8f3b4be201dfcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:57:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f83c1447e08445b3a35a86cb5b7a1549\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/5ccac778e89f48bc8aae75d776ea8f1f : SUCCESS in 1h 07m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/b62a2bf6b36649489b76c621043ba888 : SUCCESS in 1h 31m 34s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/76c92d098a3a4c9c94bdfcf2740caad9 : SUCCESS in 1h 57m 46s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/cb230dc7dfef4b568a68d23865160447 : SUCCESS in 29m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58f01092da734baea0c3475c39d32a15 : FAILURE in 33m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f54f0b53b44213bf559a0d0f60d8dd : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e644757fa024632a9d24813d6b61356 : FAILURE in 24m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6f8ffe8facb40ec9bc8eb789a133e19 : FAILURE in 21m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62d2272fb95647d092511a173318f5ee : FAILURE in 20m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82dca504db64013afc76d5c2c2b5253 : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/370036cfb12542929f62187b0b028cc8 : SUCCESS in 1h 28m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbe6b6ba4aba48518a12ed2a56ed474e : SUCCESS in 1h 52m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e847c02bc7f4720a2a3b3c29823334f : SUCCESS in 53m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/48123c65dc5645b68551fa67266fd8de : SUCCESS in 55m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cc3422297dcc4e058cd734120f416f76 : SUCCESS in 56m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7fae3c829f044d748fabfc8424fac232 : SUCCESS in 1h 04m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/61feeed7ff2e40ce95b943b3ee2eaeb3 : SUCCESS in 1h 07m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e05299cdd1c247e0b2c97a2582e1c949 : SUCCESS in 1h 58m 01s","accounts_in_message":[],"_revision_number":3},{"id":"75d1cd179e7e8b4f23233f2444f409885c077f2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 20:03:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af85bfea2f1e44c4ba42059b1ffa7adf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/491cd44b4c1b4f5091e5472713b94e99 : SUCCESS in 48m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8bec69293bcd44ddaf74f997cdf65a32 : SUCCESS in 31m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ce0fbeb6de53ad0d8d32f42cea8c44de253c0357","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-04 05:50:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8e48611c9312d5143a75301d86ab5eef95b4cca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 08:40:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04bf426966f4498c816db52e1cd73cc4\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/ae175a62639348caa8a3adcbc6eae665 : SUCCESS in 1h 19m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/b8a9b71df20d4bd9820c9c74c5c0f42d : SUCCESS in 1h 44m 41s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/ae4ca4d8c2d74bdbbffce5d62904608f : SUCCESS in 1h 40m 11s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/81ba224f337f47de9b15d1b47b65b88c : SUCCESS in 27m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d25fe88a3a2b440088c415fdfeb569fd : SUCCESS in 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a1ca600d014f7589db9aeeb6a25ff3 : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cc7879492e3443cabbc5ec98656bf66 : SUCCESS in 21m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5aa1dd1d590e478ba41ac743c43fdea9 : SUCCESS in 21m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ce451c9108d4b8fa6090c73243913e5 : SUCCESS in 24m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad4af8d18264c65b3bdafa77cf25134 : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dbc0d1c15a0f45e88b8eb933e1996c17 : SUCCESS in 1h 27m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd6d130e84f64a79a885e8488cf6de81 : SUCCESS in 2h 46m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b90ccfdcb48d430cad0269a46f878f74 : SUCCESS in 52m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c34b8f49f9a46b3bbb3afd010fed417 : SUCCESS in 1h 03m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/903035f1727446f483037ffab8397bb1 : SUCCESS in 53m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fe887ea7d51d46c495c92a0e224f2667 : SUCCESS in 1h 06m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/980fa140486a49cbaa1072a7199fd39d : SUCCESS in 53m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/33d06cec439a4969a787360e6a5f2f68 : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":4},{"id":"0bcc220272c908d81c41f2d529040eab8794886e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 09:12:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/106581d82be041f5acf8e882d87b58f6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c94f7cb45b4442719ae188006215577f : SUCCESS in 1h 02m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/674cf48bb3904a8d8b180fcbc31f97b8 : SUCCESS in 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6083d694096b288657eda26c3e4016ef9029057d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-06-05 07:56:16.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e382ac8b652a95a37db88eb812719828743bd48d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-06-09 11:02:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f359abdd9b4d02d7c9ea247802d02991cd5e42f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 11:03:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9734682439f1493f763abd719cf68c9de6a79f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:20:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5ebd15491324c25b7f0ce00dbadbb3c\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/d7d583088b07463bae65caf75d783914 : SUCCESS in 1h 44m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb4138ddaf34c2b825d4342551674e6 : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7457be9177040c8add6bd1f3b95857e : SUCCESS in 24m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82cbaa7e591a4085a4d79ad8e838fd3b : SUCCESS in 18m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86be427c3a304c5a9b1ddb272444ea61 : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c425f7ce15d40ecac5300b2d1791e85 : SUCCESS in 1h 22m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f158f15c90da486babc56e5d2d7bad6f : SUCCESS in 2h 15m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/26813989a8024eb2b5be1b1f0b38c816 : SUCCESS in 1h 00m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/621a3b2ac76445aebcbff321ca41f1f3 : SUCCESS in 57m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bb3842b5afeb4f57a48c9609ed9a643c : SUCCESS in 59m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/70950435fcb14e85b42b0557e5c110e8 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":4},{"id":"4b8e8bcca6ea64936c499238587d7cc6707daf6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:20:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c7245dba01bf6f1fce870ad1eedcd03995825a8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:22:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b23576247ec54436a5bbe37eb2231b08\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e0d2c06e674eff95b7fb7534a25c8f : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c12d12b6706825bcdbd33b33e7dd0dcaaa194a3e","revisions":{"1f0e5f1ae03c2d5e5c033ab6b3effc84891cea0d":{"kind":"REWORK","_number":1,"created":"2025-05-19 14:49:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/27/950327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/950327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/950327/1"}}},"commit":{"parents":[{"commit":"df3688b1d8ac97ced97703318703614150e3edba","subject":"AddressGroup API collection should be the resource name in plural","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df3688b1d8ac97ced97703318703614150e3edba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-14 15:10:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-19 14:48:51.000000000","tz":0},"subject":"Add provider network resources filter for \"network\"","message":"Add provider network resources filter for \"network\"\n\nA new result filter hook is added to the resource \"network\". This result\nfilter will filter by the provider network attributes\n(\u0027provider:network_type\u0027, \u0027provider:physical_network\u0027 and\n\u0027provider:segmentation_id\u0027).\n\nThis patch is also being tested in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/949975\n\nConflicts:\n  neutron/common/utils.py\n  neutron/db/db_base_plugin_v2.py\n\nCloses-Bug: #2109354\nChange-Id: I96ad1c224455c1835b5648c8eec64f2a865f9904\n(cherry picked from commit 9d4a79459aa3d42432e1b7d14cbdb7b4b5ca1f2c)\n(cherry picked from commit 3cffe86c0ea91ceab7c5b8995b1fd730e3875a30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f0e5f1ae03c2d5e5c033ab6b3effc84891cea0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f0e5f1ae03c2d5e5c033ab6b3effc84891cea0d"}]},"branch":"refs/heads/stable/2024.2"},"f7ebf0b169a4b58eea6fe3aaa633a53ad55aca7d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-03 13:49:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/27/950327/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/950327/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/950327/2"}}},"commit":{"parents":[{"commit":"b264bb9fd16fecbaa845e9622c0838a65a659bd9","subject":"[OVN] OVN agent retrieval filter matching improvement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b264bb9fd16fecbaa845e9622c0838a65a659bd9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-14 15:10:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-03 13:49:19.000000000","tz":0},"subject":"Add provider network resources filter for \"network\"","message":"Add provider network resources filter for \"network\"\n\nA new result filter hook is added to the resource \"network\". This result\nfilter will filter by the provider network attributes\n(\u0027provider:network_type\u0027, \u0027provider:physical_network\u0027 and\n\u0027provider:segmentation_id\u0027).\n\nThis patch is also being tested in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/949975\n\nConflicts:\n  neutron/common/utils.py\n  neutron/db/db_base_plugin_v2.py\n\nCloses-Bug: #2109354\nChange-Id: I96ad1c224455c1835b5648c8eec64f2a865f9904\n(cherry picked from commit 9d4a79459aa3d42432e1b7d14cbdb7b4b5ca1f2c)\n(cherry picked from commit 3cffe86c0ea91ceab7c5b8995b1fd730e3875a30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7ebf0b169a4b58eea6fe3aaa633a53ad55aca7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7ebf0b169a4b58eea6fe3aaa633a53ad55aca7d"}]},"branch":"refs/heads/stable/2024.2"},"621339542a4a9459229b95fe9626b7e33473c7a1":{"kind":"REWORK","_number":3,"created":"2025-06-03 13:52:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/27/950327/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/950327/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/950327/3"}}},"commit":{"parents":[{"commit":"b264bb9fd16fecbaa845e9622c0838a65a659bd9","subject":"[OVN] OVN agent retrieval filter matching improvement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b264bb9fd16fecbaa845e9622c0838a65a659bd9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-14 15:10:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-03 13:52:05.000000000","tz":0},"subject":"Add provider network resources filter for \"network\"","message":"Add provider network resources filter for \"network\"\n\nA new result filter hook is added to the resource \"network\". This result\nfilter will filter by the provider network attributes\n(\u0027provider:network_type\u0027, \u0027provider:physical_network\u0027 and\n\u0027provider:segmentation_id\u0027).\n\nThis patch is also being tested in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/949975\n\nConflicts:\n  neutron/db/db_base_plugin_v2.py\n\nCloses-Bug: #2109354\nChange-Id: I96ad1c224455c1835b5648c8eec64f2a865f9904\n(cherry picked from commit 9d4a79459aa3d42432e1b7d14cbdb7b4b5ca1f2c)\n(cherry picked from commit bceb866190696e8609f78d6516493fdbc351e108)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/621339542a4a9459229b95fe9626b7e33473c7a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/621339542a4a9459229b95fe9626b7e33473c7a1"}]},"branch":"refs/heads/stable/2024.2"},"c12d12b6706825bcdbd33b33e7dd0dcaaa194a3e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-06-04 05:50:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/27/950327/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/950327/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/950327/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/950327/4"}}},"commit":{"parents":[{"commit":"8463c6ec84b6321f91a496cea2850d8ebf8c9f8c","subject":"[OVN] OVN agent retrieval filter matching improvement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8463c6ec84b6321f91a496cea2850d8ebf8c9f8c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-14 15:10:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-06-04 05:50:29.000000000","tz":0},"subject":"Add provider network resources filter for \"network\"","message":"Add provider network resources filter for \"network\"\n\nA new result filter hook is added to the resource \"network\". This result\nfilter will filter by the provider network attributes\n(\u0027provider:network_type\u0027, \u0027provider:physical_network\u0027 and\n\u0027provider:segmentation_id\u0027).\n\nThis patch is also being tested in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/949975\n\nConflicts:\n  neutron/db/db_base_plugin_v2.py\n\nCloses-Bug: #2109354\nChange-Id: I96ad1c224455c1835b5648c8eec64f2a865f9904\n(cherry picked from commit 9d4a79459aa3d42432e1b7d14cbdb7b4b5ca1f2c)\n(cherry picked from commit bceb866190696e8609f78d6516493fdbc351e108)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12d12b6706825bcdbd33b33e7dd0dcaaa194a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12d12b6706825bcdbd33b33e7dd0dcaaa194a3e"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
