)]}'
{"id":"openstack%2Fneutron~950328","triplet_id":"openstack%2Fneutron~stable%2F2024.1~I96ad1c224455c1835b5648c8eec64f2a865f9904","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2109354","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-09 13:20:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96ad1c224455c1835b5648c8eec64f2a865f9904","subject":"Add provider network resources filter for \"network\"","status":"MERGED","created":"2025-05-19 14:50:14.000000000","updated":"2025-06-09 13:21:51.000000000","submitted":"2025-06-09 13:20:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950328-bug/2109354","meta_rev_id":"a088ac17f7dabf1332fc8974c1bbe1bc7665310a","_number":950328,"virtual_id_number":950328,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-09 13:20:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-06-05 07:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-06-09 11:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2025-06-09 11:02:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 16:58:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-05 07:56:27.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2025-06-09 11:02:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"aa8e8f497a5d27042fa6ccd357b08d3894125b94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-19 14:50:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6c4ac7be89e5c91c6737305667bcfca757ee3b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 16:58:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0afafb928cae47b5866d4a5e12764e11\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/e76864373a8149d197ae0604c0e7fa28 : SUCCESS in 1h 19m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/0d69f8327615475b97ded03ac615aca3 : SUCCESS in 1h 23m 17s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/680ba9c1e0224e8f801319344eda1241 : SUCCESS in 1h 54m 03s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/602c0a0f64ce482fa46b92c6d6e9830f : FAILURE in 17m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b9baf95bd06493191af95d2c31acf87 : SUCCESS in 33m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd748b9bade3401c99a2c2c229e24eea : SUCCESS in 10m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36a6f158b8df402aa0b07079ddcc97bb : SUCCESS in 27m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c74cff795df74ab28a1094c1669c381d : SUCCESS in 23m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1f2d982fca4e228d642c046b8492ff : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a3fbd585f6724ce9989880880230b056 : SUCCESS in 1h 26m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/13b591b0e1114d51a6d28c21620c9729 : SUCCESS in 1h 34m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aa99f697e48f4695a177f10737b49740 : SUCCESS in 55m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2139cbb8efcd45068e9f762893b069c0 : SUCCESS in 1h 11m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/67857dc1ffca4f94b4c58f5506be4572 : SUCCESS in 1h 10m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/71f838f72f6d43109e94ebb2bc583946 : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":1},{"id":"464712f9bca1fcc9589a3910fcafff2fcd9a4caf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 17:28:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae2c73ccaf514ca386ce5a1ef911a382\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4f24293db344f5981e2597ea0623d29 : SUCCESS in 21m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3159353f68a45a0a4f4d2d4c9de395d : SUCCESS in 1h 09m 25s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f1f54d92839f411782e1c7b8c5be8d89 : SUCCESS in 1h 05m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4f86f0f4c6baa5f362dbba285421eb8bb5c930c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-20 05:48:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"811b9c1c384edcc7e3d8a447a2199bea66dcdaec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 06:32:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/229ca7ba073b499abe6f9e0fe3828ff7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5aad5332b7545bf8c9c38e8835965bc : SUCCESS in 38m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/265868c3fffe4562b84076111beaddf1 : SUCCESS in 23m 17s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7866683b554e4f12bc06b9fc48d17bd7 : SUCCESS in 42m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e913c1ba6a8f655a432449aa481fe9f4bca38334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 07:19:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d8e8c62b1fe48b09f043966732ed383\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/069189f8eb6c4a41811e2cb573d2e6ba : SUCCESS in 1h 04m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/ef6261a9516b4227850e44e56cc73515 : SUCCESS in 1h 12m 01s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/a534ac868a624e1cba4c3664e3495835 : SUCCESS in 1h 25m 22s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/6a903f3b523345e6b41b7d4d66326657 : SUCCESS in 27m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af932b5fe9be4d0c9c29b2889dec8483 : SUCCESS in 41m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee4c226eccd4b1da3b48b3e66394524 : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f904c152b9c2424e9cc4b9b74f5b8f99 : SUCCESS in 23m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2b28b5d6c1b4846808a8e309ff7faeb : SUCCESS in 26m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374ecdacbd5d41c9a574e1fc7f497158 : SUCCESS in 10m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/04b6fc4819f048c09832b761a9dc8cab : SUCCESS in 1h 20m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3771f11b065c480a9b34b78baca70a73 : SUCCESS in 1h 03m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f1eb47a33193402595d519094801cfbb : SUCCESS in 42m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/339f082d50c64ea8869fefde18dde5b5 : SUCCESS in 1h 15m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd6d59af6c0e499c9e4a21d14f9b5e8d : SUCCESS in 35m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e2b6c1e98ad4b17892f0d741ea9c47f : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a61df672e42dd0017da1f0019618391fb865e271","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-03 13:54:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"187c13e43ee8827f7a2c24bbc5341002e9f2e69a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:05:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/340b915d0366428ba7936d1016dc418d\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/9b995f92414d46e79ec5312a9d5431c3 : SUCCESS in 1h 23m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/d9e3c24766a94df895113402adfe5638 : SUCCESS in 1h 21m 26s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/6bafa6d2b9944734bce7289ca7e21924 : SUCCESS in 1h 34m 30s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/d9614a46fc4241ffa244612989832ad1 : SUCCESS in 39m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6725111257a44c82b9f9a729f10a58a5 : FAILURE in 38m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec0d89d8fad94844a7c1a826e9cf2235 : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/880b362eeb9541e8900fc0ffb90b4615 : FAILURE in 24m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5beee1d0f44349faa7977003b11abd09 : FAILURE in 25m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/000bdda68830451c80326ec1ab46cf71 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e06f67e5b74434bb2a0a3e30fc024ec : SUCCESS in 1h 29m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d304555bba6c41d4bd9ec1bc0d1a300c : SUCCESS in 1h 27m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fc01b148c68347edabbf05403313410e : SUCCESS in 54m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ecdfe0e0e41648b98f6821b43d8899b8 : SUCCESS in 1h 22m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/39ba208180834dde883a082ecb92d9ce : SUCCESS in 1h 28m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6a45237fad3c4630be7988c58ffd73bb : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":2},{"id":"27816940557d7253460c1ecac7407c643e6bf1d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 20:23:18.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6570f4522335486698f03025c467dbe9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8eb6dc49fa0461b9cd42ba6670c0e9c : FAILURE in 1h 01m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ccec800e739640979693d416dfc6cb5f : FAILURE in 38m 24s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b4b26f48f1e44ce9888a4cf1ca0aca4 : RETRY_LIMIT Ansible setup timeout in 1m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"904307ace8c81fcdd9b7b1aa907cbacd6693318e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-04 05:50:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0857d77a6a2fb56f31fbf3ab4b93194d0ab5df68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 08:20:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/807a325b109d4591a94573e7155c0e80\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/6c5ee9a53ddf4fb388c0f5d7bea8a9cb : SUCCESS in 1h 00m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/1009b9daf4634dac9321dd82ac03f361 : SUCCESS in 1h 47m 35s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/496c21d803aa4961b5be9942c63d4709 : SUCCESS in 2h 19m 55s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/73ada57a1e2041cda971794e40cf8895 : SUCCESS in 36m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1046104ffc494405af674493489781a4 : SUCCESS in 34m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2e4a2526aba45a3834e5053cbba6e3f : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3366d61654b542c19d181923de86a7e0 : SUCCESS in 26m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04851047a19b44019e8d1024760c52b1 : SUCCESS in 29m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a59eb0a2b6429e8409765d3945753f : SUCCESS in 12m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c34d97e59804416b64e01e5bfd496b6 : SUCCESS in 1h 51m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/259482bd6735469fabebe98a5cd46a1d : SUCCESS in 2h 28m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b906613b0170435188692e4a7ab9fecc : SUCCESS in 55m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/549230ed56844bf585162e523b816218 : SUCCESS in 1h 16m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/12357513853c444b8c1391a616fa502a : SUCCESS in 1h 10m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1e1f432c97be48919f1cfecad27585e2 : SUCCESS in 1h 19m 42s","accounts_in_message":[],"_revision_number":3},{"id":"c4699e646e4e694e288fcab2bec3e66bb5620670","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 12:19:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c816a59e97e49ed971bfa367802478d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4ea24cc43cd40d287145e1cd8e6f714 : SUCCESS in 1h 16m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9d89916d9e448929b5efc2bb070880f : SUCCESS in 59m 52s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6152ead847264a269736ecf98146ecc6 : SUCCESS in 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a62b4b575589033c15499c2339be5216f3d450df","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-06-05 07:56:27.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa561edfa6af5170cfd2cd84e365f878fc6c05c7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-06-09 11:02:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"081955193a1804a494c75828399ce3a8748ff8f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 11:03:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"89e2ca9f923bc954da5534b197f7b776dca35ca5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:20:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4632d3bedff4d4f887177f9b6e5dbe6\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/8f9dbe89012a4a59af83caad3ab0dfe3 : SUCCESS in 1h 28m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79b52c3ba5624b9990fd2bd9e0b7ba3f : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b48df8fba4214c1694060d9ef47c329b : SUCCESS in 33m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5909aca236d2422a8f4e237d9e003c02 : SUCCESS in 26m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbafe47f44d414e87503db4331aa2ac : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d44d04c6471f432aa683d60d8ff47213 : SUCCESS in 1h 37m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/be77bfa9dfc441399274eb1ac930465e : SUCCESS in 1h 47m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/516d24c4fd374bb7bc2d973bf101d642 : SUCCESS in 1h 03m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3b39ba1dab104feb8626a22a2959dc99 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":3},{"id":"a3ce5ff276f2aa78e2820d8c394007b95429bef2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:20:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a088ac17f7dabf1332fc8974c1bbe1bc7665310a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:21:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f5aa6040f344ce8c7c408e2dfaa1e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9eff0ee0b1d4be5855940ff67058eb9 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bcc380b39e7e2aea32337c45dae4c3cd898a7bbb","revisions":{"c7f3416ec2c17d4787231f96699d8ef8466388bf":{"kind":"REWORK","_number":1,"created":"2025-05-19 14:50:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/950328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/950328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/950328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/950328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/950328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/950328/1"}}},"commit":{"parents":[{"commit":"0d4ef860549a552db3a9d04934154ad256662bbb","subject":"Merge \"[Stable Only][CI][fips jobs] Use stable constraints for tempest\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d4ef860549a552db3a9d04934154ad256662bbb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-14 15:10:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-19 14:50:06.000000000","tz":0},"subject":"Add provider network resources filter for \"network\"","message":"Add provider network resources filter for \"network\"\n\nA new result filter hook is added to the resource \"network\". This result\nfilter will filter by the provider network attributes\n(\u0027provider:network_type\u0027, \u0027provider:physical_network\u0027 and\n\u0027provider:segmentation_id\u0027).\n\nThis patch is also being tested in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/949975\n\nConflicts:\n  neutron/common/utils.py\n  neutron/db/db_base_plugin_v2.py\n\nCloses-Bug: #2109354\nChange-Id: I96ad1c224455c1835b5648c8eec64f2a865f9904\n(cherry picked from commit 9d4a79459aa3d42432e1b7d14cbdb7b4b5ca1f2c)\n(cherry picked from commit 3cffe86c0ea91ceab7c5b8995b1fd730e3875a30)\n(cherry picked from commit 1f0e5f1ae03c2d5e5c033ab6b3effc84891cea0d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7f3416ec2c17d4787231f96699d8ef8466388bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7f3416ec2c17d4787231f96699d8ef8466388bf"}]},"branch":"refs/heads/stable/2024.1"},"78395c2f5fe367db6c6ff3e6407308709bd4fb92":{"kind":"REWORK","_number":2,"created":"2025-06-03 13:54:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/950328/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/950328/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/950328/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/950328/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/950328/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/950328/2"}}},"commit":{"parents":[{"commit":"0e29296cc56da1feee7d566ed99c23bc25af0a96","subject":"[OVN] OVN agent retrieval filter matching improvement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e29296cc56da1feee7d566ed99c23bc25af0a96"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-14 15:10:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-03 13:53:27.000000000","tz":0},"subject":"Add provider network resources filter for \"network\"","message":"Add provider network resources filter for \"network\"\n\nA new result filter hook is added to the resource \"network\". This result\nfilter will filter by the provider network attributes\n(\u0027provider:network_type\u0027, \u0027provider:physical_network\u0027 and\n\u0027provider:segmentation_id\u0027).\n\nThis patch is also being tested in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/949975\n\nConflicts:\n  neutron/db/db_base_plugin_v2.py\n\nCloses-Bug: #2109354\nChange-Id: I96ad1c224455c1835b5648c8eec64f2a865f9904\n(cherry picked from commit 9d4a79459aa3d42432e1b7d14cbdb7b4b5ca1f2c)\n(cherry picked from commit bceb866190696e8609f78d6516493fdbc351e108)\n(cherry picked from commit 621339542a4a9459229b95fe9626b7e33473c7a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78395c2f5fe367db6c6ff3e6407308709bd4fb92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78395c2f5fe367db6c6ff3e6407308709bd4fb92"}]},"branch":"refs/heads/stable/2024.1"},"bcc380b39e7e2aea32337c45dae4c3cd898a7bbb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-06-04 05:50:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/950328/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/950328/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/950328/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/950328/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/950328/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/950328/3"}}},"commit":{"parents":[{"commit":"1814fbb4c85c68873bbe0037101f46169ccd617e","subject":"[OVN] OVN agent retrieval filter matching improvement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1814fbb4c85c68873bbe0037101f46169ccd617e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-14 15:10:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-06-04 05:50:53.000000000","tz":0},"subject":"Add provider network resources filter for \"network\"","message":"Add provider network resources filter for \"network\"\n\nA new result filter hook is added to the resource \"network\". This result\nfilter will filter by the provider network attributes\n(\u0027provider:network_type\u0027, \u0027provider:physical_network\u0027 and\n\u0027provider:segmentation_id\u0027).\n\nThis patch is also being tested in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/949975\n\nConflicts:\n  neutron/db/db_base_plugin_v2.py\n\nCloses-Bug: #2109354\nChange-Id: I96ad1c224455c1835b5648c8eec64f2a865f9904\n(cherry picked from commit 9d4a79459aa3d42432e1b7d14cbdb7b4b5ca1f2c)\n(cherry picked from commit bceb866190696e8609f78d6516493fdbc351e108)\n(cherry picked from commit 621339542a4a9459229b95fe9626b7e33473c7a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcc380b39e7e2aea32337c45dae4c3cd898a7bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcc380b39e7e2aea32337c45dae4c3cd898a7bbb"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
