)]}'
{"id":"openstack%2Fneutron~950384","triplet_id":"openstack%2Fneutron~master~If200300b043a283e0dc79294fa5263dc4f2e2a27","project":"openstack/neutron","branch":"master","topic":"bug/2099896","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-26 09:30:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If200300b043a283e0dc79294fa5263dc4f2e2a27","subject":"Delete the metadata agent file from the eventlet directory","status":"MERGED","created":"2025-05-20 08:33:21.000000000","updated":"2025-05-26 09:31:19.000000000","submitted":"2025-05-26 09:30:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950384-bug/2099896","meta_rev_id":"e3de57cd69b717b4ee1ee3d95440c23e9a642e53","_number":950384,"virtual_id_number":950384,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-26 09:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2025-05-26 07:13:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-05-20 17:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2025-05-26 07:13:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-20 09:17:33.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-20 09:31:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-20 10:33:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-20 17:14:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-05-26 07:13:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"853b0443cb4f23d3a2ac4562db722b3eda8045a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-20 08:33:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8b153ea3ac8aa724c0e841659a24669a56bc0a6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-20 09:17:33.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/84/950384/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a99b4b0f78bb521feaa915d6521514a98e764074","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 09:31:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0894e0ffe48439e812bf13b3d3ba4e8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a63202d5bb0441be97ffac44a969a31d : SUCCESS in 47m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/489fb000c59b40deb7720cca815617cf : SUCCESS in 43m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/822e1be3cd814b0d9730231c01e1b33b : SUCCESS in 55m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fca83e457b74a0cbb139e644d3204cf1bb61a6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 10:33:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/473fa278e8af45789ac8d2585cedca88\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/903010870f634be598dcfa3250752f26 : SUCCESS in 1h 21m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8c8f82e9ff1b48faa44a630c2bd5e8b1 : SUCCESS in 1h 45m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e23b0d81e064adeb5cafacdf6f1b262 : SUCCESS in 35m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889601d572ff401c9a4dc86d4503fb7b : SUCCESS in 9m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75919ec3d8e64d2bb137d5b7e9d0e3cd : SUCCESS in 17m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c59036f29cf9417286645ac20df5d57f : SUCCESS in 19m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/730eb506cd384c899dd285b494619a53 : SUCCESS in 26m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2871f398146c4e6e84a50534655cf741 : SUCCESS in 13m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/45b23685700d443f80fa665085c96413 : SUCCESS in 1h 08m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5c99560af9af4641b1a34dbbf9544308 : SUCCESS in 1h 53m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8e5fd1ecb344df4b0634d16b41d6928 : SUCCESS in 58m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6e4bb526f994dcd9c9b7fb3aa943d8c : SUCCESS in 1h 13m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/02f347e2de1c4fadb900789dfe825561 : SUCCESS in 1h 47m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3742e3781a7a4ca79cc04db922d8a7a5 : SUCCESS in 1h 31m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/689e2af6389e461d861e59cf60cf2e63 : SUCCESS in 1h 03m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4ed47674adbb4735ac42cd62389560db : SUCCESS in 1h 03m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f984ba9d891349f693131aa3c09c7889 : SUCCESS in 1h 17m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/27be765d1fb746359f2e6dd4696f243c : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":1},{"id":"906dbd7e093759f1f922b633712e910096705076","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-20 17:14:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a8a9fbbec4c3c067d6bc136eaa0c651622e8eea","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-26 07:13:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dbf9d08f260bfa01e32035c2ae8a1392e399133","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 07:13:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ef9b391c86d072fb05e24ae06bd8ff84913fe16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 09:30:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a4dd7524d14971a33cb30d97e670ce\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f6bde1c4452e4917b9c38013e8c91b50 : SUCCESS in 1h 27m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1667be8f684c4f219da6bda8faeb47a9 : SUCCESS in 2h 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e13d28f33d9b4b13a10c25f8bf151208 : SUCCESS in 11m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/811e32da58ed43ce9b87094f43889e06 : SUCCESS in 21m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53f05046aefd400eab5bcddc1595e4e3 : SUCCESS in 21m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5d99531d114ce3bdf34de8df825b2c : SUCCESS in 12m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a13e2fbc055c4edcb1e99bba626eb613 : SUCCESS in 1h 25m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5387bf4c3ecb4f97b3c0e1552c3cb849 : SUCCESS in 1h 53m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3bcc275e11bd4d71ac5420170f212a68 : SUCCESS in 1h 49m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/085c6802b08741fca9af2fbfa646b405 : SUCCESS in 1h 03m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e4ca9c829394f35a94cbee5edbac636 : SUCCESS in 52m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3d090b3d21fc4059a13c91ef04fb82b7 : SUCCESS in 51m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/53f2f86c44f54d9caf69bbbf0fb16e65 : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4d1b14bebdd72a5867d9bdb48aac42ccf3c2d157","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 09:30:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e3de57cd69b717b4ee1ee3d95440c23e9a642e53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 09:31:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5228eb2975db43fdb959acf4e45d24a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3ee5ac386b43f9a8c1b4e7e0f8328f : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4c58a341752d48ba79dcb30b495e7ad02932fa3","revisions":{"e4c58a341752d48ba79dcb30b495e7ad02932fa3":{"kind":"REWORK","_number":1,"created":"2025-05-20 08:33:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/950384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/950384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/950384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/950384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/950384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/950384/1"}}},"commit":{"parents":[{"commit":"87392ea6df002ffd00488b9984996ca7e588cf0e","subject":"Merge \"[OVN] Do not supply gateway_port if it\u0027s not bound to chassis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87392ea6df002ffd00488b9984996ca7e588cf0e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-20 08:31:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-20 08:32:56.000000000","tz":0},"subject":"Delete the metadata agent file from the eventlet directory","message":"Delete the metadata agent file from the eventlet directory\n\nThis removal was missing in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/942916\n\nRelated-Bug: #2099896\nChange-Id: If200300b043a283e0dc79294fa5263dc4f2e2a27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4c58a341752d48ba79dcb30b495e7ad02932fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4c58a341752d48ba79dcb30b495e7ad02932fa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
