)]}'
{"id":"openstack%2Fneutron~950385","triplet_id":"openstack%2Fneutron~stable%2F2025.1~If200300b043a283e0dc79294fa5263dc4f2e2a27","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2099896","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-02 14:22:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If200300b043a283e0dc79294fa5263dc4f2e2a27","subject":"Delete the metadata agent file from the eventlet directory","status":"MERGED","created":"2025-05-20 08:33:39.000000000","updated":"2025-06-02 14:23:27.000000000","submitted":"2025-06-02 14:22:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950385-bug/2099896","meta_rev_id":"58f3c1e2423640d17e335dfd03cc0efafac1cdcc","_number":950385,"virtual_id_number":950385,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-02 14:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-06-02 11:41:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-05-28 18:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2025-06-02 11:41:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-20 09:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-20 11:02:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-28 18:52:42.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-06-02 06:02:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-06-02 11:41:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"39767c34b0dad852ab075dffee78f835f5ac7f08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-20 08:33:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42a67bad9533ff18f90d51a34eec1d63200a228b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 09:25:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/165934eacc9a48db922dc6c3f641b170\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ed6782c1d8b42eb9cade9105ef23971 : SUCCESS in 49m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/992fec6a80f34dc78cf725cb97a80f6c : SUCCESS in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4868d14ddaf4f1c9a6ec03420142e671a2544d11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 11:02:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ad777346b6744dbadf3fbb51cb1cfc7\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/3e816f9b97a0410382c72195ef854945 : SUCCESS in 1h 57m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/d3471e9817c54fcd98540056150638b0 : SUCCESS in 2h 26m 43s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/236f1aeaa69743e0b71e13a9d2158248 : SUCCESS in 2h 07m 45s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a868e9cb206c444b987197ea47a01857 : SUCCESS in 1h 44m 04s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/eb07276a7f0d4221944a0fd8d5bcc5d7 : SUCCESS in 2h 18m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bbc19a1de22454cb159815d7c5ef9b3 : SUCCESS in 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68f872412104ee2918a46c3e68ee758 : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f73d0b71db4646ca9c44ca16843b6609 : SUCCESS in 23m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45dd1012a04c4fe291e52751e0ede71f : SUCCESS in 17m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d184b133f14406091ef3c162c1dc1b2 : SUCCESS in 12m 44s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/28a9750399444ab99d06aa1988eca1e3 : SUCCESS in 54m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6c611a3f63704a9d8d732aa49be2f442 : SUCCESS in 1h 12m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/db7468d548cf4306a7f1a291d2b17597 : SUCCESS in 1h 58m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b4f47ac8c6bd4607802279ab9e9c06b6 : SUCCESS in 1h 43m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5d48b81a1f5456487db3ce7b3eace61 : SUCCESS in 48m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f98db35d4a544e3b5f4969e43c26256 : SUCCESS in 1h 09m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8025c6fe85b5475c936982de453268d2 : SUCCESS in 1h 39m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2f128a2c341b4c5896858490bf802dd1 : SUCCESS in 1h 30m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/43c0eb159b454c74bad5d08d582d4fc9 : SUCCESS in 52m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c9fe2c080fa94ae0b616341aaff5dfbd : SUCCESS in 27m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/409aa89c5ebc4c34b2c54cd40672e9f8 : SUCCESS in 1h 12m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a374a3390824048bf277a89acc9f9de : SUCCESS in 1h 22m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e5473f420db959a1a5ae813dac22052fd7893b34","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-28 18:52:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ba0b69abee026832dfab28e93bb8fadefc40380","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-06-02 11:41:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07f87dfb0bca30db478b958c70ef7423485016c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 11:41:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"389121487ab21c1832420b8a8755336732c5ba72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 14:22:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a24e0c2c6834f8cacf79db966c57d00\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e5cdb1252c8e4f1b8f85547e0988d36e : SUCCESS in 2h 08m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/391564f368fa48c7832ad96d30de9a45 : SUCCESS in 9m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/149f0205937849709c82168a5a93e5a3 : SUCCESS in 23m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22435ca9dd2a4a82acf02c75e5629bb3 : SUCCESS in 17m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6fdc458cffe4fa4881e5225cb353a44 : SUCCESS in 10m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ec9a4997c9f445dd8ea65f989806a31f : SUCCESS in 1h 18m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/71ca09ba232544e2bf3103e7f33e76b0 : SUCCESS in 1h 59m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/85ed5990d4934996aefc069ac51c6933 : SUCCESS in 1h 41m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b59d3ffb08ce496b8e1ac98826433c2c : SUCCESS in 1h 24m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb729d54869e4e06a9af2a06c7793182 : SUCCESS in 52m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/90bcb05331bf42d49d880c95787f1808 : SUCCESS in 1h 13m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e2055d958d634c95b4cc3c37021d5685 : SUCCESS in 1h 23m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3ca0abca652b098d4eae9fb2cd921bb4d8d25e97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 14:22:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"58f3c1e2423640d17e335dfd03cc0efafac1cdcc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 14:23:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5a96b6ea82474e8ecd927b84c2e27d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2969ff58124f4e8f8a560a084205489c : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4815390cdc890e8533686454f8a2f031be487c42","revisions":{"4815390cdc890e8533686454f8a2f031be487c42":{"kind":"REWORK","_number":1,"created":"2025-05-20 08:33:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/950385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/950385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/950385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/950385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/950385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/950385/1"}}},"commit":{"parents":[{"commit":"eea7ee1c76dcda361b740c7073571ad14806e5d9","subject":"Merge \"[OVN] Create a HA_Chassis_Group without raising an exception\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eea7ee1c76dcda361b740c7073571ad14806e5d9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-20 08:31:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-20 08:33:32.000000000","tz":0},"subject":"Delete the metadata agent file from the eventlet directory","message":"Delete the metadata agent file from the eventlet directory\n\nThis removal was missing in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/942916\n\nRelated-Bug: #2099896\nChange-Id: If200300b043a283e0dc79294fa5263dc4f2e2a27\n(cherry picked from commit e4c58a341752d48ba79dcb30b495e7ad02932fa3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4815390cdc890e8533686454f8a2f031be487c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4815390cdc890e8533686454f8a2f031be487c42"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
