)]}'
{"id":"openstack%2Fneutron~950479","triplet_id":"openstack%2Fneutron~stable%2F2024.2~I2319535fd6ee4527f0476e8abcf96106be5edd84","project":"openstack/neutron","branch":"stable/2024.2","topic":"support-ovn-log-sync-2024.2","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2025-08-04 07:47:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2319535fd6ee4527f0476e8abcf96106be5edd84","subject":"Fix ovn db sync with log resources","status":"ABANDONED","created":"2025-05-21 06:38:07.000000000","updated":"2025-08-04 07:47:52.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc35bcd53705ae781e67af61d192fcaaf9b14245","_number":950479,"virtual_id_number":950479,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-21 10:45:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2025-05-21 16:44:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-21 07:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-21 10:45:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 16:44:19.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-08-04 07:47:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"6c78648728f9b1decad82e813bb939cd413343d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-21 06:38:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"131d4e78aaa91b6306091cab62a0fd145d4df0a9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-21 06:40:38.000000000","message":"Topic support-ovn-log-sync-2024.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"ceb408c13e334579ca5755e2565b33d4469d706e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-21 06:40:41.000000000","message":"Topic set to support-ovn-log-sync","accounts_in_message":[],"_revision_number":1},{"id":"4bfe3c64589e51af1a1170751292346b970699ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 07:06:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd56ae2f67b6481584a32d635dd887cf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdcb507467b0441d87bd64f882caeba7 : SUCCESS in 26m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f46a06ff7ed8453bba462974a1a6a5d8 : SUCCESS in 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4571ced4c1366112bd62afc00650ec82b2e58207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-21 08:00:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"efd84b37afe584615c96e00def378bd755824a1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 08:22:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de918d391f8a4e2ea018c9dba0076cea\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4426568fcb44536a96f6f9a32bb0b71 : FAILURE in 20m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9c3c663ad6b94576be145786c38c4175 : FAILURE in 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db7109e83b2aece822d2f36f5bad786d1bc21018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 10:45:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdbe879bf7da44c2a682194980ebc746\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/bfb1c86550ff4ffba66ccb7cf00d8d4c : SUCCESS in 2h 13m 16s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/373f402490254bf8bfc108e6cf8a559e : SUCCESS in 34m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dae42c432d1649b9b99cc6fdf6c77954 : FAILURE in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6106b073f74324b064f28654720b63 : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ed933dbd2e64371a6b64e19dbeeb8da : FAILURE in 22m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e96ce41a4064d62ac4d1abec25c11a4 : FAILURE in 19m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4de9a3aab3424f1ca92c456cd2b33044 : FAILURE in 21m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f7d19b0458448caa8b55a04c15591c : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76a4d9c1627f46248aae90ba7674000f : SUCCESS in 18m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77f609b8b2e74579ac59df9f8dea58a3 : FAILURE in 1h 01m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4ebfa22fffd45b488b24a7aa645883f : SUCCESS in 1h 48m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/90ad2fd953ed4f498c7791e785da68e3 : SUCCESS in 1h 02m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c6350e52541f4a2582a0da51dcbc8baf : SUCCESS in 52m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/735ce994b00a461c858d9edc0f05cc09 : SUCCESS in 59m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/08e07679ad904445b78a7b60b5e4d539 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6db5e3229d4199d777e760f8a936eb2b92519f4f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-21 16:44:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc35bcd53705ae781e67af61d192fcaaf9b14245","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-04 07:47:52.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81310cce12b5c148d0add8ee684fb0973247779b","revisions":{"2af1ecc83dae3fba06ebfd34a5bd15ebd1675ee4":{"kind":"REWORK","_number":1,"created":"2025-05-21 06:38:07.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/79/950479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/950479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/950479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/950479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/950479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/950479/1"}}},"commit":{"parents":[{"commit":"aeb7c6a09791e2bd62c27bdd6c0bb20529619afe","subject":"Merge \"Fix ovn_db_sync for user defined flavor routers\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeb7c6a09791e2bd62c27bdd6c0bb20529619afe"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-21 06:36:57.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-21 06:36:57.000000000","tz":480},"subject":"Fix ovn db sync with log resources","message":"Fix ovn db sync with log resources\n\nMake sure neutron ovn db sync counts log resource changes.\nThis help to avoid important acls (like acls for `neutron_pg_drop`)\ngot deleted due to miss match case.\n\nAdditionally, switch to update instead of recreate acl in ovn sync.\nTo update optional columns, `pg-acl-add` could in some cases,\nnot able to properly update label and options columns.\nWith change to using db_set, we will sure we not face that same issue\nin any cases.\n\nChange-Id: I2319535fd6ee4527f0476e8abcf96106be5edd84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2af1ecc83dae3fba06ebfd34a5bd15ebd1675ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2af1ecc83dae3fba06ebfd34a5bd15ebd1675ee4"}]},"branch":"refs/heads/stable/2024.2"},"81310cce12b5c148d0add8ee684fb0973247779b":{"kind":"REWORK","_number":2,"created":"2025-05-21 08:00:02.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/79/950479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/950479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/950479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/950479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/950479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/950479/2"}}},"commit":{"parents":[{"commit":"aeb7c6a09791e2bd62c27bdd6c0bb20529619afe","subject":"Merge \"Fix ovn_db_sync for user defined flavor routers\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeb7c6a09791e2bd62c27bdd6c0bb20529619afe"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-21 06:36:57.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-21 07:59:50.000000000","tz":480},"subject":"Fix ovn db sync with log resources","message":"Fix ovn db sync with log resources\n\nMake sure neutron ovn db sync counts log resource changes.\nThis help to avoid important acls (like acls for `neutron_pg_drop`)\ngot deleted due to miss match case.\n\nAdditionally, switch to update instead of recreate acl in ovn sync.\nTo update optional columns, `pg-acl-add` could in some cases,\nnot able to properly update label and options columns.\nWith change to using db_set, we will sure we not face that same issue\nin any cases.\n\nChange-Id: I2319535fd6ee4527f0476e8abcf96106be5edd84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81310cce12b5c148d0add8ee684fb0973247779b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81310cce12b5c148d0add8ee684fb0973247779b"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
