)]}'
{"id":"openstack%2Fneutron~950480","triplet_id":"openstack%2Fneutron~stable%2F2025.1~Iaec0fce556f6e6635beb554e03e6720e0e04b2b1","project":"openstack/neutron","branch":"stable/2025.1","topic":"support-ovn-log-sync-2025.1","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2025-08-04 07:47:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaec0fce556f6e6635beb554e03e6720e0e04b2b1","subject":"Fix ovn db sync with log resources","status":"ABANDONED","created":"2025-05-21 06:43:09.000000000","updated":"2025-08-04 07:47:40.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"57a9f7ef7b0cb7a8cf6f8ab6beeb1e2190f582d3","_number":950480,"virtual_id_number":950480,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-21 10:40:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2025-05-21 16:43:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-21 07:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-21 10:40:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 16:43:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-08-04 07:47:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"e66c87548b9d31e779b90ce6213accd34b0fd610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-21 06:43:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e46c58563d87514ebe1b52898124777c6c05e17a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-21 07:09:22.000000000","message":"Topic support-ovn-log-sync-2025.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"11ae30870ed50f22d9978a0a8679eec36906af78","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-21 07:09:30.000000000","message":"Topic set to support-ovn-log-sync","accounts_in_message":[],"_revision_number":1},{"id":"6e996219783c2f2a622431512b5f6736dadac07a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 07:24:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf94baa556ad4e338fce30cf8bc39a55\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbc010b5602a4c00b9cc66fadf659000 : SUCCESS in 39m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/940383a3b9ee434781459efac84f8959 : SUCCESS in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4d3d7c6f0501abb68d3f61b31f11c3cff8ea9c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-21 08:08:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72ed9bd13d1325615a962fc0a2294b9cdfbf6877","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 08:54:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/881c70c1d3454d3f863ed9c53a674984\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52eede4240104c1aade599ac596c076e : FAILURE in 44m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/041ec131c3904fcd85859a412d9d8926 : FAILURE in 39m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e23d23f1237371367c3d720f115492b793c89983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 10:40:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c22696e65124463a0d666d452e25a4d\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/365f9454e03744c8838226995d9452bc : SUCCESS in 2h 04m 50s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/ed0e430a952245c0954b846936eee87e : SUCCESS in 33m 42s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/053858d557cf430a8a6afd47ae5f0b69 : SUCCESS in 2h 29m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68b0f4d45bb34124822747737809ab90 : FAILURE in 36m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752fcd1ae8d0484d8bf47390228bef29 : SUCCESS in 11m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd6eee560fcb4b4cb221a5cb44c4e516 : FAILURE in 21m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb19e93aca354f10b5e76527470ecaf0 : FAILURE in 18m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b50908b6ad994b17a36dc8460c5e810a : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68890383323446a798feb9963358d22c : SUCCESS in 19m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0ffab0f57a1849c38ff95bc4beef7b66 : SUCCESS in 56m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fa399d8543a24c88a731e3d07eb914ae : FAILURE in 1h 05m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3eb5df3dccb14ad3b322cb99dd9469bc : SUCCESS in 1h 56m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb1d74ba776b414789122e9cb29727f5 : SUCCESS in 1h 08m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1998238f599c492bb352df42db784104 : SUCCESS in 1h 00m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0653941568ea43ddae25c92622db7690 : SUCCESS in 1h 12m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a69b861b54ff45b5acc3ab5d0089645a : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":2},{"id":"eb48ede610a6f58d5b3431e096f5d4f3f6b9cf45","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-21 16:43:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4b8b42b1e0a7785e9bb9e413308034b535c11d6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-21 16:45:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57a9f7ef7b0cb7a8cf6f8ab6beeb1e2190f582d3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-04 07:47:40.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"99812079538a2549810710bdfe663be0447b1c75","revisions":{"27440c5d20cbea0ac70332204056183296b6aa35":{"kind":"REWORK","_number":1,"created":"2025-05-21 06:43:09.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/950480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/950480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/950480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/950480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/950480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/950480/1"}}},"commit":{"parents":[{"commit":"30837fce0286860ce05b483b064b03cb2c06c480","subject":"Merge \"[OVN] Only update the MTU of the router GW LRPs\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30837fce0286860ce05b483b064b03cb2c06c480"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-04-24 06:19:47.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-21 06:42:27.000000000","tz":480},"subject":"Fix ovn db sync with log resources","message":"Fix ovn db sync with log resources\n\nMake sure neutron ovn db sync counts log resource changes.\nThis help to avoid important acls (like acls for `neutron_pg_drop`)\ngot deleted due to miss match case.\n\nAdditionally, switch to update instead of recreate acl in ovn sync.\nTo update optional columns, `pg-acl-add` could in some cases,\nnot able to properly update label and options columns.\nWith change to using db_set, we will sure we not face that same issue\nin any cases.\n\nCloses-Bug: #2107925\nCloses-Bug: #2087909\n\nChange-Id: Iaec0fce556f6e6635beb554e03e6720e0e04b2b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27440c5d20cbea0ac70332204056183296b6aa35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27440c5d20cbea0ac70332204056183296b6aa35"}]},"branch":"refs/heads/stable/2025.1"},"99812079538a2549810710bdfe663be0447b1c75":{"kind":"REWORK","_number":2,"created":"2025-05-21 08:08:28.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/950480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/950480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/950480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/950480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/950480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/950480/2"}}},"commit":{"parents":[{"commit":"30837fce0286860ce05b483b064b03cb2c06c480","subject":"Merge \"[OVN] Only update the MTU of the router GW LRPs\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30837fce0286860ce05b483b064b03cb2c06c480"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-04-24 06:19:47.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-21 08:05:24.000000000","tz":480},"subject":"Fix ovn db sync with log resources","message":"Fix ovn db sync with log resources\n\nMake sure neutron ovn db sync counts log resource changes.\nThis help to avoid important acls (like acls for `neutron_pg_drop`)\ngot deleted due to miss match case.\n\nAdditionally, switch to update instead of recreate acl in ovn sync.\nTo update optional columns, `pg-acl-add` could in some cases,\nnot able to properly update label and options columns.\nWith change to using db_set, we will sure we not face that same issue\nin any cases.\n\nCloses-Bug: #2107925\nCloses-Bug: #2087909\n\nChange-Id: Iaec0fce556f6e6635beb554e03e6720e0e04b2b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99812079538a2549810710bdfe663be0447b1c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99812079538a2549810710bdfe663be0447b1c75"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
