)]}'
{"id":"openstack%2Fneutron~950502","triplet_id":"openstack%2Fneutron~master~I14735f0c11aa5aa0a66b6761607cf6765f1b6215","project":"openstack/neutron","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-27 22:58:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I14735f0c11aa5aa0a66b6761607cf6765f1b6215","subject":"[UT] Removal of eventlet from the unit tests","status":"MERGED","created":"2025-05-21 09:29:38.000000000","updated":"2025-05-27 23:00:38.000000000","submitted":"2025-05-27 22:58:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950502-eventlet-removal","meta_rev_id":"a15c8f0786b2b3062905d72982f9bea101b6007d","_number":950502,"virtual_id_number":950502,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 22:58:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-05-26 08:49:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-05-27 20:55:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-05-27 20:55:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-21 10:09:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 10:44:07.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-26 07:13:34.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-05-26 07:13:34.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-05-26 07:13:34.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-05-26 08:49:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-05-27 20:55:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"a46ddfcfb2d3d349ec5bfc7f352cfafa27d10946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-21 09:29:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5563326838a255c75c2be1fc5d7c4d3903073b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 10:09:56.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36520ed0a7b043c28ebd9d7b37f229e8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/181a48e396674521868229050fbb2087 : SUCCESS in 39m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c071f30cbe64779bc0e99ae62326661 : FAILURE in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/393b6b6d98044b25a312fc402dd0715b : SUCCESS in 20m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41c1e05cb2f94b5f81909eeee86fba1b : SUCCESS in 21m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3afc91a4498d4bd4ab0f80fa962af221 : SUCCESS in 21m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da4a5e2184b450ab9edfece8064e4bf : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":1},{"id":"aeadb9a796e07a6bb4016742f5690450a912d618","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 10:19:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/924f49ac843c41b69410d6bbf5dabdbb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0aab7f4aac724e9eb3ec12cf3fed754a : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a9cae346bcef40988e39ea44949dc7b8 : SUCCESS in 34m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/392faaa62cfe45b2b471a547de10e3c5 : SUCCESS in 26m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"307e34fec1b7bc8c56ce9b1fe96976a24057edae","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-21 10:44:07.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/02/950502/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"14aa417c29fc4c8fa836bb128095cd53656a8c59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-21 13:13:20.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9b4a69183a6b0d99bb30a2ff7802783632907e2d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-21 13:56:34.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/02/950502/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"2eeb03aafe6d29c5aca026639094179f740c483f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 14:21:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0129d6a2abfb4980906dfa9507adac14\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/496ad38ffd7a4ec4a6d218796f8d548e : SUCCESS in 55m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a48bcc3b39d948a389fb2bb4420a7ef5 : SUCCESS in 12m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02bf0251715f49478e2ff99f29ec1673 : SUCCESS in 14m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0c7adc555bd4dc4a73bfaaa82709050 : SUCCESS in 22m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c65e13c38b44b8f908868bd6b96b097 : SUCCESS in 22m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382baf1ba6544ba58170faff23d89644 : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5d5376d49cdea159b33845d956012680c46b569a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 14:28:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c35b8075ab4240fcb638aa848e52a706\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/728573ebd2b14c098bde970aa8e4ecb2 : SUCCESS in 32m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b24142dd75c46f6b6c2ef8d27093a50 : SUCCESS in 44m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cdb78498a6ed48f2b4865eed0edc55d9 : SUCCESS in 57m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d165fcc76c68a33fbc53d264d217fe8604a540d4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-26 08:49:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"43ac469f6f1c6e8c0903d95822234b9ff872a61d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-27 20:55:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b72a5c00fa8884602b672a4fc2593e75b25514ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 20:56:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e44fe89498962db425c6073871dd5f6a42a02356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 22:58:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80fae0be24764fce938a1d5375169f46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562d85da899f40e48b73c09f16ba5dc4 : SUCCESS in 11m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca87a4236ada487abeb7c62d9bc57253 : SUCCESS in 20m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b80aa478f25f44778e73a2b240bf392d : SUCCESS in 21m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e101353ad54cddb5fb0fe0fb61f1a4 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":2},{"id":"a2e71cadbaa1c8799ae11e3532f2c60f033e9af9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 22:58:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a15c8f0786b2b3062905d72982f9bea101b6007d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 23:00:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c2691214f8c416381a25be8e2a67952\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72e88fa32a148259ff9f34862d348a1 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad065fdcfe23b309428e30611394be35b32a8e14","revisions":{"176f3d39955a5a340d4b3ba8131fc2535fb04529":{"kind":"REWORK","_number":1,"created":"2025-05-21 09:29:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/950502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/950502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/950502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/950502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/950502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/950502/1"}}},"commit":{"parents":[{"commit":"87392ea6df002ffd00488b9984996ca7e588cf0e","subject":"Merge \"[OVN] Do not supply gateway_port if it\u0027s not bound to chassis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87392ea6df002ffd00488b9984996ca7e588cf0e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-20 17:57:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-20 17:57:27.000000000","tz":0},"subject":"[UT] Removal of eventlet from the unit tests","message":"[UT] Removal of eventlet from the unit tests\n\nFiles:\n* neutron/tests/unit/common/test_utils.py\n\nRelated-Bug: #2111414\nChange-Id: I14735f0c11aa5aa0a66b6761607cf6765f1b6215\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176f3d39955a5a340d4b3ba8131fc2535fb04529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176f3d39955a5a340d4b3ba8131fc2535fb04529"}]},"branch":"refs/heads/master"},"ad065fdcfe23b309428e30611394be35b32a8e14":{"kind":"REWORK","_number":2,"created":"2025-05-21 13:13:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/950502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/950502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/950502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/950502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/950502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/950502/2"}}},"commit":{"parents":[{"commit":"87392ea6df002ffd00488b9984996ca7e588cf0e","subject":"Merge \"[OVN] Do not supply gateway_port if it\u0027s not bound to chassis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87392ea6df002ffd00488b9984996ca7e588cf0e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-20 17:57:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-05-21 13:13:20.000000000","tz":0},"subject":"[UT] Removal of eventlet from the unit tests","message":"[UT] Removal of eventlet from the unit tests\n\nFiles:\n* neutron/tests/unit/common/test_utils.py\n\nRelated-Bug: #2111414\nChange-Id: I14735f0c11aa5aa0a66b6761607cf6765f1b6215\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad065fdcfe23b309428e30611394be35b32a8e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad065fdcfe23b309428e30611394be35b32a8e14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
