)]}'
{"id":"openstack%2Fneutron~950581","triplet_id":"openstack%2Fneutron~master~I87110682c280ccdad940dd9420c8b37774dddd19","project":"openstack/neutron","branch":"master","topic":"pre-commit","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-14 06:41:54.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I87110682c280ccdad940dd9420c8b37774dddd19","subject":"tox: Fix indentation","status":"MERGED","created":"2025-05-21 17:17:31.000000000","updated":"2025-07-14 06:41:54.000000000","submitted":"2025-07-09 16:22:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"950581-pre-commit","meta_rev_id":"875078205a7c00ff6145cbafa3d511f08f89da6d","_number":950581,"virtual_id_number":950581,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-09 16:22:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-07-04 08:50:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-07-01 18:31:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-07-04 08:50:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-21 19:03:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 19:58:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-07-01 18:31:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-04 08:50:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-07-14 06:41:54.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"dacd9cc8b94b0eddc5c08d36ba388b6cdf1f09de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-21 17:17:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f75612061266188b4de14eb49cfa86e12a8035ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 19:03:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/488eed5f4ea94df294e2b0eb663abd3d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3cb1a0910dc4a73a41b690ece77fe06 : SUCCESS in 42m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1bca3f25954d45bbc046425785e472 : FAILURE in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59e5cd8f58c645f2813b81d4b251b70d : SUCCESS in 22m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/002641776f3d43dd89bd1784d5f03c97 : SUCCESS in 21m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cfa10c358194626a30327bfdeda8458 : SUCCESS in 23m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862b4ca4a61f492c8e8b9843df707fc8 : SUCCESS in 14m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dad67d62f5d54f9880873498c8c4ff7d : SUCCESS in 15m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/34f88cf02a0d4f1fb49190032e33c783 : SUCCESS in 1h 26m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f6b58b4e04d84383ad1d3b2b05df2403 : FAILURE in 1h 39m 55s","accounts_in_message":[],"_revision_number":1},{"id":"131c638d5926144371a147d6b97510a0a7431120","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 19:30:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa53e4bd63e24c939ac87f883360a62c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/effa6e73ec2f4926ace5b2e08296edce : SUCCESS in 20m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94b696906084486db929cc5a457e23e2 : SUCCESS in 32m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7329676d5c28453eb8cf66e7e8cb1d35 : SUCCESS in 36m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87ef2b39116edb1589c61b84749729d2303f9505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-22 10:26:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"962be0194feeffd21e261b774aee944dc6652944","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:44:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf53b0ddc2434f7aaebc6e95d3582b75\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca0848c1b7ee4f0f92dc771fd2151d3a : FAILURE in 45m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d379d933a5445df91949dff1228499f : SUCCESS in 42m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/58c956d83c01415d814318667c02dfd8 : SUCCESS in 21m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d71a11da9c402cd611e33c645d33049d1cb17da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:01:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35feabd47d9c4738854cc83451ff2f09\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/863c6aaaf7ae4d52bec5ebe418a8d28b : SUCCESS in 44m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e9a3bca29247fb85451acea10edd98 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e917573e618346f7a722a4545b3afc6b : SUCCESS in 27m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7db6bcc04cca423c975f579f3793d524 : SUCCESS in 20m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7957957816b4abbb27641d376989279 : SUCCESS in 21m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2397536eb67e40da9c36044bdeced411 : SUCCESS in 13m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02dabaea6b094187ac4ce761bb78cc00 : SUCCESS in 8m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/199a52bac6ef47cc8d596f7f89a0f1fa : SUCCESS in 2h 09m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bbc73719ca274665b0aa2c5ec9829e0d : FAILURE in 2h 33m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6ceb1b9d08dab86c5e0bc307c7afdace2b0f206a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-23 16:09:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc4664faf4586455f080e716c770644ae5cee553","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 17:22:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06532a7c3f8349cc98e5072fb8381e64\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23087c0ab19f4573b82bdd08392e2d5f : SUCCESS in 35m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2babf0f7fb06465287a438abbc005a43 : SUCCESS in 16m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c66ee90dc705455c961a66576b24fb36 : SUCCESS in 46m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a2fd49ded4270d6c916f77a846b1def9616a6433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 17:38:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eedba16bfa1748a495548c1c8506aa8d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a8e1dd26e8941cbbc0f782dbea02f46 : SUCCESS in 32m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e9764b7ffd42c2aaddaa3d3502d3f8 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f604951f0c80467fae4dcc16b0d0e2d2 : SUCCESS in 24m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/399d2ebe143e4c69b4437b14e4681bbb : SUCCESS in 18m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0adb2bc1c1a46209bd834db3562093d : SUCCESS in 26m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3e86eccf574745a6ce66fa90344711 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d60e18b248e5483faec953c961549e5a : SUCCESS in 18m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/30e549d37ef64f49abb8c163c13d1c87 : SUCCESS in 1h 23m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ec67bf98a4d04e25a0cf9c7bb9c6eaf5 : FAILURE in 1h 26m 55s","accounts_in_message":[],"_revision_number":2},{"id":"07525d4e7d3f2abebca4eb4e747b896185dc1e37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-26 11:28:22.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"151808d364c0929ba74adbe18720cd393b846ac7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 12:44:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3552ab27baba4816a48e49acb9ad5ebd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7aafc12a798d4293b219ca53686236a9 : SUCCESS in 1h 14m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9d8a08a8a1541c999670b010b68d75c : SUCCESS in 1h 08m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5c5f816bdc554a739419891837ded15f : SUCCESS in 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b512926800c6e963b8b0b86668225a6a708544c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 12:56:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/293d1bb0f0cd4c26904f8bcb1d240022\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d190327a37b5417d912513efb8c29716 : SUCCESS in 41m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1608675a885419b9573462c69d736f2 : SUCCESS in 11m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96f92712aa574189a0fcec7c55fd78c0 : SUCCESS in 18m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a841c246db5c4eb1918ea00f7807e818 : SUCCESS in 21m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b85ca16042514faf93731bbc323fc253 : SUCCESS in 26m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97814333e01c4d2ea0546488b6e397d7 : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79e174247ad648839c9c0e84a5434524 : SUCCESS in 16m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a92dc5a045d94a17979450d26a60c051 : SUCCESS in 1h 19m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/527c95d3b0ed4753b21951f6697bad3e : FAILURE in 1h 28m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a2b78b5aa289a2494ce1d76cdbbabde76097f33b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-26 15:09:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"941dbd89fa178fb0acc77be5ae572c482a7f805d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 15:54:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/852aca615a694fd19f21a5c2ae54ffc8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ba8f05b6fc340ae964ff3d4de3ae4fe : SUCCESS in 19m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/459e8840aa1441e4973ee850ad35dfec : SUCCESS in 20m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e64c2709f2524384a14086c9bba0edee : SUCCESS in 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5380b297f34f92c6f0d1d04a38d8820b662806a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 16:42:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ee06051bd11475390c312b5cec94639\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02ee6b511daa40bb84eddd975ed4708f : SUCCESS in 44m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f123e75eba4431ab0196c5487ba8e4 : SUCCESS in 11m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d28b49bfa8940c284cb70e5a064f1f6 : SUCCESS in 22m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/221c6530a4464ab38c692bebf3f3e1c8 : SUCCESS in 20m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e77c73ba671f451381fd550fda6ae119 : SUCCESS in 21m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be86b12749643cf9c47368b0de6f3f4 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c122d5ec1e94283aba57a600bd28b40 : SUCCESS in 17m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5093bee85b1e4a5e9faf14ef8969d69a : FAILURE in 32m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2bf3e45ee35545b094c4685153efc17d : FAILURE in 1h 31m 33s","accounts_in_message":[],"_revision_number":3},{"id":"97e53628b0df933afe7246beda0166464d1175ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 09:48:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2ed2f3cd6f719e4325fd4df973f2bca12e66f8d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 12:03:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55cec70ea7d54d659dc876e71adc316a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4adfc7c8c4e40a18cf585d7b8c2f72b : SUCCESS in 34m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5669af4703c24e1a81049ee17d42ad92 : SUCCESS in 13m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85eeb21a079a49fb8c0d29d3509906e4 : SUCCESS in 30m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c45767f671e4f2ba0440c518fde9bb9 : SUCCESS in 24m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/823031cee45348998cf64c7f514440c0 : SUCCESS in 28m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c9dd4703da428d8362be22134d342c : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e704f83dfe8474f924484096281fadb : SUCCESS in 19m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/784ff22741c14c0783f7c56a480118ee : SUCCESS in 1h 42m 59s","accounts_in_message":[],"_revision_number":4},{"id":"c168b869b2582aaa60429721ad9716a6ad38cf98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 15:52:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/638d644d621f40caae5739f6665fad51\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59f90575e78d4d3c854ee20a92628700 : SUCCESS in 38m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46cf11602f434c328d4a9e0cd1021034 : SUCCESS in 29m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0659f7598a654ef8a5d1014328ffa85b : SUCCESS in 46m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d2cf58d1c548f05dd97c5d0b1e176b774fa85cea","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-01 18:31:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f686d618cf02adb962d818340d256163d8f7f68c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-04 08:50:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6387ce32a1da2686f26d564fdb73f01d052397b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 11:06:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0a90ebeb01208cc2e17d1fdb329c3a9161abd24b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 14:31:22.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/950580 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"612de1f499cd805548866686e6061b0dcdf1d69e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-05 17:39:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9911fa3a2cd2a47fb5141a8fc376ac5fbfddd4f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 18:59:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cd843dd364f41a2a9b3a4f8aafc39ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e88fb5395a2d4470b88eee72773df289 : SUCCESS in 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/104687d850bc452c9d3a191dccea5d95 : SUCCESS in 12m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa25623cf3c1456f8854cca37a61748c : SUCCESS in 16m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac04f352de03436e98d7856c7a6f08c7 : SUCCESS in 21m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9e79bcccf7c4e59a48ca9e61e87db4e : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dad2042d45b542089c0ab7c22e942256 : SUCCESS in 15m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f866c49207d42d88f375dfe87f4a824 : SUCCESS in 21m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2787db2a864f4361b244393089a319a6 : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":4},{"id":"8264ca47cdb326a18bc50e73a98190cb4e21c08f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 19:01:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4d909e27f84d15b0b84d8193fbb692\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04dbdac0af5b42a7870df76538ec0318 : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/edf7ae836b7f4c3f80262311c7e16daa : SUCCESS in 1h 14m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f7266c74a3646778410f92cc7055c15 : SUCCESS in 1h 20m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"47ee08bf9b1a5e4ac824c26b1e6d95a9746e7d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 19:34:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fdf0cfc0476ead35417651f42460a93fe698de61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 21:24:21.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/950580 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"f0f27f544be5ed6bba028a00700fd28ff84bb2fe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-07 08:39:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bcb34fb78818fb0a06c6dd307513de5b31f90aa8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 09:53:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d07be02b204d4aaf8c908b82a82bdb94\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42710da6af4c41fd9625b03414a20ace : SUCCESS in 1h 04m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f6d57868e28b4a80b1962135c3520673 : SUCCESS in 1h 02m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cc202bdc9a3b4272b72daec1ccadd8ab : SUCCESS in 1h 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"152532c1495712c564634ba91d0d71d08304979d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 10:31:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a9a0585579427ea9e150f0faf90279\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e90c088dd16546fca1be2112b514afe9 : SUCCESS in 48m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05450b1ad2a14258a4415a9356922772 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87dd527ea61d40458ba731f50ef7df7f : SUCCESS in 21m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6101dca8c3064ce9b67d347e3206d90d : SUCCESS in 18m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f860bc192bca48b68b12610d05e633ee : SUCCESS in 22m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a6ac00f49a4ecdae746eaad08b7017 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c56dd8afcc244ffa93d16a9539a7fd22 : SUCCESS in 20m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9bf63e1e460f4dd4a2d236b21aff1387 : SUCCESS in 1h 45m 24s","accounts_in_message":[],"_revision_number":4},{"id":"fb41d62119cc4d37e2803117bab9a67dadc04665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 23:48:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e6d06c133511df9333420dc1c75d38998af608e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 02:05:08.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/950580 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"b4f2351e05d69e1ee8f02540e966a69f929e024d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-09 12:40:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"47a07cf5a73ca7ccb5d1b0979d9c8ec2920dd074","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 13:14:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0575979e394a4981ac25ebba39d1a64b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43d1031350e14f9682922e6b5caced5c : SUCCESS in 31m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32ccd10083934b96a2b15628a2f7e8a8 : SUCCESS in 25m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3dc8c6b7c9534f22806220c9263b5b32 : SUCCESS in 21m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7eae407bcb5e2114166c121390e3a529d6c93f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 14:07:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb5585f610c4c6fbe3942f7466dd38f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cedc252ed8c94893837ea541712c7d8b : SUCCESS in 26m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9fcde16274b496eb3d9232822b45987 : SUCCESS in 11m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84befef389d44603a5042d0e64f7ec99 : SUCCESS in 14m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/720358aa6bf84ed1b78f50201f0dadb2 : SUCCESS in 15m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42e709f9ba714534b721848c2f9877e3 : SUCCESS in 24m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf59cfcd92354776b9f96cc7ae10c272 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57fd4a84406a43a3891756dd11091bdb : SUCCESS in 10m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ad8f86443d7c4e2ebc63ac5d442ef770 : SUCCESS in 1h 25m 19s","accounts_in_message":[],"_revision_number":4},{"id":"48ded2a5d44d2006612f60fa45be4b18e7888148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 14:07:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"53c6e3fe22aac8e2242ef161c626ea8bfaf5efc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:22:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b4441c8cb57457c9faba1054231be55\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9439654513f648d1b39b21c7d8196105 : SUCCESS in 9m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e768656c8faa41ffa76eaf9d0aa93d7f : SUCCESS in 23m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f47c00b47634dc7bbebc60b3f8159fd : SUCCESS in 20m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/742aa16e2c5c41f19e82372fa83d9a43 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/718e4fef5c6f47f1859bba7743447259 : SUCCESS in 20m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b121bbfc8bda429b981ea5ab9e66d34e : SUCCESS in 1h 39m 01s","accounts_in_message":[],"_revision_number":4},{"id":"b3584e2f8131bd67cd7915a0140d5abb73cb8e83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:22:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fee30f2b29a97efd0461dd873f99315de0db05ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:23:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38621112076d4222b0b627851bbabbf3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f482a38ce34e0791b5b2c23ee94e1c : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99b877ddae5a4ee49ec4e60bb6b2cba9 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4},{"id":"875078205a7c00ff6145cbafa3d511f08f89da6d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-14 06:41:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d72adaa8d64291e9f7ca05de4c681c71b894af93","revisions":{"bd68df5ad26f7c0a4c7af404315c9f91e5248cba":{"kind":"REWORK","_number":1,"created":"2025-05-21 17:17:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/950581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/950581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/950581/1"}}},"commit":{"parents":[{"commit":"77a98cf30fd60bc247c234594c213117bbba7121","subject":"pre-commit: Migrate to ruff, enable autopep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77a98cf30fd60bc247c234594c213117bbba7121"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:10:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:15:35.000000000","tz":60},"subject":"tox: Fix indentation","message":"tox: Fix indentation\n\nChange-Id: I87110682c280ccdad940dd9420c8b37774dddd19\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd68df5ad26f7c0a4c7af404315c9f91e5248cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd68df5ad26f7c0a4c7af404315c9f91e5248cba"}]},"branch":"refs/heads/master"},"10748007fc88b1d4b797482dc9f807050a99db73":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-22 10:26:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/950581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/950581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/950581/2"}}},"commit":{"parents":[{"commit":"ec12c30606a6d62607023d9d0e54718cbbbf4139","subject":"pre-commit: Migrate to ruff, enable autopep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec12c30606a6d62607023d9d0e54718cbbbf4139"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:10:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-22 10:26:33.000000000","tz":60},"subject":"tox: Fix indentation","message":"tox: Fix indentation\n\nChange-Id: I87110682c280ccdad940dd9420c8b37774dddd19\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10748007fc88b1d4b797482dc9f807050a99db73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10748007fc88b1d4b797482dc9f807050a99db73"}]},"branch":"refs/heads/master"},"498eaa10dfe1f83fe88c042c9af5aa40407f2e5e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-26 11:28:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/950581/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/950581/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/950581/3"}}},"commit":{"parents":[{"commit":"2b392b1243b0195da2e1683ad36ed1056743318b","subject":"pre-commit: Migrate to ruff, enable autopep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b392b1243b0195da2e1683ad36ed1056743318b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:10:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-26 11:00:31.000000000","tz":60},"subject":"tox: Fix indentation","message":"tox: Fix indentation\n\nChange-Id: I87110682c280ccdad940dd9420c8b37774dddd19\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498eaa10dfe1f83fe88c042c9af5aa40407f2e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498eaa10dfe1f83fe88c042c9af5aa40407f2e5e"}]},"branch":"refs/heads/master"},"d72adaa8d64291e9f7ca05de4c681c71b894af93":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-01 09:48:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/950581/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/950581/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/950581/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/950581/4"}}},"commit":{"parents":[{"commit":"be77ed3a813709eebb2659b5473b890b5978c743","subject":"pre-commit: Migrate to ruff, enable autopep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be77ed3a813709eebb2659b5473b890b5978c743"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:10:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-07-01 09:47:55.000000000","tz":60},"subject":"tox: Fix indentation","message":"tox: Fix indentation\n\nChange-Id: I87110682c280ccdad940dd9420c8b37774dddd19\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d72adaa8d64291e9f7ca05de4c681c71b894af93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d72adaa8d64291e9f7ca05de4c681c71b894af93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
