)]}'
{"id":"openstack%2Fneutron~950582","triplet_id":"openstack%2Fneutron~master~I474d98cee375f8a7af5feb07a6988dfcb635bad4","project":"openstack/neutron","branch":"master","topic":"pre-commit","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-15 18:41:37.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-07-07 20:07:47.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I474d98cee375f8a7af5feb07a6988dfcb635bad4","subject":"trivial: Autofix low-hanging linter failures","status":"MERGED","created":"2025-05-21 17:17:31.000000000","updated":"2025-07-24 10:34:08.000000000","submitted":"2025-07-15 18:41:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950582-pre-commit","meta_rev_id":"7e17ad4c48d5c21e4657a10ce888c55790008de1","_number":950582,"virtual_id_number":950582,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-07-24 00:20:18.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-15 18:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-07-07 09:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-07 20:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-07 20:07:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-21 18:47:30.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-21 19:34:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 19:59:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-07-01 11:50:31.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-01 18:43:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-07 09:43:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"ec931779b8d3f73741b2d79bda14d21a6d9fdd0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-21 17:17:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8336bf7219f14bf87f69bd85a2f30ce4e2ba6e7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-21 18:47:30.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/82/950582/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a454ae61f1d0dd912307a21121bda661556d634d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 19:34:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e70b8e2b5cfa43dc8e80017790a2ec8e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/175b3d452a714485a84ec46c12abf424 : SUCCESS in 1h 43m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1c7d8a524d57448b9f73b001f75a4754 : SUCCESS in 1h 52m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/619fcf9f745d48bba79f465cf54e68ee : SUCCESS in 38m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88e07f49d632442e82abeaab75f1ae0a : SUCCESS in 51m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3782e757d2042f8a973cc0b3223b9fc : FAILURE in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f562cadb4695447da0b0d4428842ce5e : SUCCESS in 19m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa04ff5cb9f8489db0f18fdd9817fc63 : SUCCESS in 17m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5b78b0fe3ee4137897c7ba8554e1511 : SUCCESS in 27m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c9031b857344f58bdff35663b6c8ae : SUCCESS in 12m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5d1dc427c67c4ace8c7b5082c8a07fac : SUCCESS in 1h 24m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cf047f5421264bda9c08c6563378ab92 : FAILURE in 1h 48m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7781e1d559324c84958cd08d7cf635e9 : SUCCESS in 55m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8841e42ee2e7404b937282030f3991ff : SUCCESS in 1h 01m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/67f9b72b9c1f4640ba63426f7e355a17 : SUCCESS in 1h 41m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2111780546314307828ca9f357da8525 : SUCCESS in 1h 27m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ddc83ba553d14334b589cfa88afddf80 : SUCCESS in 1h 05m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7d39da07af9e421c8d97341d36e1fbb0 : SUCCESS in 58m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65a1331251e545279b2c39083f286e9f : SUCCESS in 1h 08m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6764e838962d445b9519b9d2b818a76d : SUCCESS in 1h 27m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1298319966ea38b59c922cdc9ccc3a38a040ddbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 20:18:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78b6a9b1799444208135f0e4728a20a8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/290fd71cc68c4374bf8ca2fd2fd82a00 : SUCCESS in 1h 05m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e633c97a800444c8b59c525eb460e09 : SUCCESS in 58m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c56a86289beb4ee1a420300e265131c2 : SUCCESS in 1h 09m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1a43614d4286406de9eabcd121990e0d4b8bab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-22 10:26:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8eeedc8666bf8d94732832a405416e35553ae97f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-22 11:11:07.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/82/950582/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"af331946cd130345a543fb8da90aa8daed374c47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 12:22:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d1cb9545d2243f08d111e7e260f9360\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68fcda4f2d884f569463b9ad5daf3cd4 : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/468685e34e034e97988066cd737d38a1 : SUCCESS in 37m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/886338217a4843cfa970083fb44ba3f3 : SUCCESS in 48m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d120d1ba9a21a68087f1708fc5e6c5a05007035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:09:14.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ac5ba97abc44047bdba90acb401d187\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dee01f72156e47a896d9579fd8bae8c0 : SUCCESS in 1h 51m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/531b0aaf5d2d4d52a5a469cee9f6e4fd : SUCCESS in 2h 24m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ca2063aea38345aab963514ae17d6e99 : SUCCESS in 38m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9639397882d648d3942e97679c4203a1 : SUCCESS in 39m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c87659c6f2744308ef6ea8c878801a3 : FAILURE in 13m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6ef820589474fd89113a92a7d369f3b : SUCCESS in 27m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cea6f957b1ca44c1ad842e7bde5b3129 : SUCCESS in 19m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8fda577cf654de682b0bf621df371d1 : SUCCESS in 24m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de7b3838a08140da8ed06584447b26ee : SUCCESS in 12m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ac1eba56295f488788817d270b546d74 : SUCCESS in 1h 21m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/44beb27cd7c94b14b1bffc06020609bb : FAILURE in 2h 40m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b601f6eb7a5c4ab480e21af7f6680bcf : SUCCESS in 59m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e490ce628ed349818f16ad3edc1aff39 : SUCCESS in 1h 18m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c0d9b8b1082b4139aff89585ffeccfa1 : SUCCESS in 1h 40m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1617255fa3f4010a132539fc2939270 : SUCCESS in 1h 38m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/680ea9c2bc6e43d39aae3a2641138e52 : SUCCESS in 1h 02m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5d9d6c2a9635449c9a73d67ec90e6a43 : SUCCESS in 1h 00m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/087b7140db9645c59c51f95cff4e4945 : SUCCESS in 1h 13m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0f386c5351ba49a48fca8c47073717ad : SUCCESS in 1h 18m 54s","accounts_in_message":[],"_revision_number":2},{"id":"3ec4bdabb622fdadfd4134c2c366a73f0ee22ce5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-26 11:28:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1c91f24da8d38ee34325f0c3b42e1971c948eacf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 12:49:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f5846fa15d4b0d853bc98d43cb9024\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fbb2cb2b3c204b33a55e71161dcb12c8 : SUCCESS in 1h 14m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7cf4e82e94e54e7ba7869df406644cf9 : SUCCESS in 1h 07m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0601c5e9f0e24d818bc2d2db99967099 : SUCCESS in 1h 18m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cf5e1ff49d5e19f30eb67a4b4b46183c8ed72852","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-26 13:07:50.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/82/950582/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"351f30f36134866a8ee8a70450441d9728182257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 13:53:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b2e6f4ce02f406a8ddd48fbf2afc256\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b9b0f23866f54bcb954f13fdce6cffc3 : FAILURE in 22m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/92cd340b93484a47b7600bd395e6a5e0 : SUCCESS in 2h 23m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c229f58d16bb436597fa76ff4a21eb39 : SUCCESS in 41m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7328c1ed74024dd8b720f0a5c51bab44 : SUCCESS in 44m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b61f51bff2b94307b6a02c159f423fd8 : SUCCESS in 12m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b055a5d1d9e4b44b403883e02d6f478 : SUCCESS in 21m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59000d9808fb4845a48c0f4402482efb : SUCCESS in 20m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc75f98b5248451ca2a1025b8beb304d : SUCCESS in 27m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e0db2fcf7f4dc0a077eb9fbe5b86cd : SUCCESS in 11m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/83a4812d417443b38bd487cb32c66e20 : SUCCESS in 1h 21m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6ba811dc32c84fb0b4daa2c4b8503358 : FAILURE in 1h 43m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/145b39c17d964235a749c285ae717f1b : SUCCESS in 51m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a071aafcaca404aa490971e9313eba6 : SUCCESS in 1h 00m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ab0bebfb09434c34ab3cf5a0b184ad3e : SUCCESS in 1h 50m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2df950f24e284fcc836a5d7c14b7cede : SUCCESS in 1h 14m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/859e0fc262eb40fba68e0481a1b7cca1 : SUCCESS in 57m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/09f3a92858084a47888221505241a6c0 : SUCCESS in 56m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b005ab38fb3495e877f93c7b7113208 : SUCCESS in 1h 11m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3ab87b58c2d44a8b8756e41905a28db3 : SUCCESS in 1h 26m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d241a75b6cd866b458813ebd07f8ee8bb75bf091","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 09:48:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"61fe8ce7f0876bb3653b7d9c1d50c59784b39f1a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-01 11:50:31.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/82/950582/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"eacd78f0ed73177cde8e9eebda9e25d2bb8ef743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 12:12:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/320decae746e4fdf9888cc5961122f23\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fecbf6dd1f3e41cdba687693c71ea875 : SUCCESS in 1h 44m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/61b0cca8d6734e2e9ffad936b8ffe65d : SUCCESS in 1h 15m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cbcb5890e745453686644813dd012ed4 : SUCCESS in 33m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b5f38fb27d143708298618f5eb93464 : SUCCESS in 33m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfbee05dbbb24451b94effdb75977328 : SUCCESS in 12m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cec3d3744b774f6880e3361589e3f984 : SUCCESS in 19m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a730ebd0220840569bbacd5c0532406f : SUCCESS in 22m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df87d0c5d8e94d8881c5dbef0193794f : SUCCESS in 26m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57212301b3e14b718a8107e2b503d94b : SUCCESS in 13m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a2a848c7a0014a908c838fd3eecf3d83 : SUCCESS in 1h 31m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e6c72607ee348b8b7c860b45064f069 : SUCCESS in 1h 00m 08s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e53c3475cc24495843c93775ee33f63 : SUCCESS in 1h 03m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e954f1a3a1d343b0b9358bc88336e4b9 : FAILURE in 1h 16m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e11bcf112dae4027a8a30ff52639ac8b : SUCCESS in 1h 25m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5b2699c3773548f99fd2275914a1134a : SUCCESS in 1h 05m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/011265cb7f5e4754a9fc91b71ecda912 : SUCCESS in 51m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/664a74f5505e4c34803924057d516c07 : SUCCESS in 1h 06m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c9fbfac6605940368b201a42484de819 : SUCCESS in 1h 22m 48s","accounts_in_message":[],"_revision_number":4},{"id":"2f8be0704298af9962719ced0ea2fa0c5a246303","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 16:06:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f2a17b8a78c475f9775275e81ac1b33\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9e7ab87108b94c739d87f8faa4d16f3d : SUCCESS in 38m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf9894be700649eba33c37fc52b222e5 : SUCCESS in 36m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c157c5149274850b3071b72867f3935 : SUCCESS in 46m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"da1a92d059b27c6b5369e2b824021593aef5463c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 16:24:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3d1bc1d00dce7ec9400b99e69dd72489de31bbc4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-01 16:58:02.000000000","message":"Patch Set 4:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/82/950582/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"0bd2a3dbdee1b9955ecf43359ac80de8bc12acc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 17:15:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca41eb712c6b49dea597dada8dd0c1c3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9f189a5024994781a11f4b5fb4aaf21d : SUCCESS in 33m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/61e4f0ba7bb7400ebb57bf96c2c5d0c0 : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4841faf284b1463b8dc6ad393053962a : SUCCESS in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"815e84dd2e247ab1044c84e0a604e258885ad70c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 18:20:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/657c51264f154293a99aaf6272497645\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c21ff1736f904688b227bf39fef9de4e : SUCCESS in 1h 06m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/946f73be2bc749e7befbe4f4e0d203e4 : SUCCESS in 1h 52m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4f7c6e38dce6493da57f9634cae149f4 : SUCCESS in 31m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9a4aa81879047baabe87f5fd0486f8b : SUCCESS in 33m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae4508c3e99949ac8871728c73d18a96 : SUCCESS in 12m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fdfffaece33c4da581952bd76554446e : SUCCESS in 19m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c18b538ff09e480880ea70ae0db546f2 : SUCCESS in 17m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3cc44e14d94149fab82c2a2cc8b814b2 : SUCCESS in 29m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d976a2824043699c6cca68e3899e54 : SUCCESS in 12m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/936e002558584b6a87ad44ad497162e0 : SUCCESS in 1h 29m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/175620f4a32e4fe29ebd62e0cbf37947 : SUCCESS in 1h 01m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/df5fe3ad460645aa8d69db05e1157c83 : SUCCESS in 1h 06m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0e4cc15483e344ad91f3ecec1015d969 : FAILURE in 1h 11m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9f80c1e196184b0aa21ed6cc42cadb5b : SUCCESS in 1h 30m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/80ba135a1c3c46c18df413cfe06f2fa4 : SUCCESS in 1h 13m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/abed82f041694a769c85e678f29629ef : SUCCESS in 58m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/02e213835a924918afcbb53b37abd90d : FAILURE in 1h 01m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0ab5c939153b4f57bfcf509ecd74606d : SUCCESS in 1h 20m 01s","accounts_in_message":[],"_revision_number":4},{"id":"d9918d1c4359c70641e4d732fe2b8cbd42095c60","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-01 18:43:23.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8c4cfe7f014b463ea4f6ea4625844b891436f35b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-01 18:46:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"80e2924f3477062b733c47404de681b42d3a75c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-01 19:30:35.000000000","message":"Patch Set 4:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/82/950582/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"95cceecea24b3ce5572924fce22c8466591ffca0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 20:10:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/687327fb8d1c4261837dc94fd4cc0c3d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54a907abd42540c38e0ea0b3c83cc2ce : SUCCESS in 36m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4b9b212b7c14163a892dd01bae4e284 : SUCCESS in 30m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e234659c5d87480a8c23b6d4ff77a86d : SUCCESS in 41m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e45f58eed18aa18b816601bb7c5be76bbe36c0d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 20:42:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ca1485e2bf54f68b9b2a13616deb6bb\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/824b35c1b97e47df9f14b7047dbdd416 : SUCCESS in 1h 19m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0b764410457047a48491a2b8f48279d6 : SUCCESS in 1h 28m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8d63881654d4923a655bec299aa6393 : SUCCESS in 19m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ddf4356276c416bab1010a1eb9e29d9 : SUCCESS in 49m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c92fff5f95b842eb9dffd3b8991db4f9 : SUCCESS in 13m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85fae555ead142c6bcd5c0dd100cf866 : SUCCESS in 22m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dbdc4c981605439ea3ff6009f24bb3a2 : SUCCESS in 22m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b894801139e4412ab6d7352a618cacfe : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9033c19272e4856be68ee8e5e284343 : SUCCESS in 12m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/db020ddd81334e3fbd2efb24c2dc69e1 : SUCCESS in 1h 43m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5fb75e0c2514dc5894b2ea1c4ad5497 : SUCCESS in 57m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a21c3192d31747e6bd43d6c65ff81c48 : SUCCESS in 1h 11m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4f83b798df044fd1b6bc130bcf887759 : SUCCESS in 1h 44m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e8d90726aefe4e4d8dfff8ea6ca1cd25 : SUCCESS in 1h 39m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/be4dcb104bd34210aeb93129c7ff29ce : SUCCESS in 56m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e05bc70c78924a62bbda894096018f9d : SUCCESS in 1h 09m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f54b0879f8e8463eb2e0338c6287aa01 : SUCCESS in 1h 11m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/92af7f8ddb5f4ec9b64ee1563c1df184 : SUCCESS in 1h 15m 48s","accounts_in_message":[],"_revision_number":4},{"id":"4a125a575275c291115726cc783cad84c6c4017e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 09:34:44.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cf04fe45139b42ba224406811ee4c604f355a3a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 09:34:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1bdc8b1a26a147143b3d9f8c3eaa606d44d3ec6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 11:30:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6eabb3a0eb424dac848d7c5ae48fe6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fad1c950ad4a4f7daa92c0500f094ee5 : SUCCESS in 1h 09m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1218e2b3b0064724b26d66fb771c1e7c : SUCCESS in 1h 31m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/354691dd71a8454399e53b5a8d78f146 : SUCCESS in 23m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc9bfc51304b481d9684eac99fd7d5d6 : SUCCESS in 46m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216d14dc1c504a4aa3251a2c11f270c4 : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5068e5c37214b19bf12f40ee8acffb2 : SUCCESS in 21m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e51f876e7ee49978c83a85ea458f656 : SUCCESS in 19m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5e8bc2f62904b41b454152f30c67407 : SUCCESS in 35m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c18a0d7ce23d4a7fb56b28230c6ff148 : SUCCESS in 13m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9ec8ba6d69f04f35b7f545ef334faf14 : SUCCESS in 1h 40m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf1fa222529e4345a602012ff001bfc7 : SUCCESS in 45m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/abf9cc9f80474357a068fc0dd8b3e474 : SUCCESS in 1h 09m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ab06bfe3c624e03929c22c0d24447c1 : SUCCESS in 1h 49m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2c89eff9e20e408ca09534c9156daef8 : SUCCESS in 1h 21m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4f81a40f23e6420bb63e36e5f2e5b53f : SUCCESS in 1h 00m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c0caeed415f3487d8ae13398ced7308f : SUCCESS in 46m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3431e7d9a93f407f92bc0cedacd2c636 : SUCCESS in 38m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fb88dc2f136b43b0946f8156140fc199 : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":5},{"id":"66d63bb196cb8d471acd7d71e98c0fb6b86c500e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 11:43:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6063fa752b84207b35147904a553b4e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be8537f821514c53a59170a1f7af8f73 : SUCCESS in 35m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/688d5b232b3942fba218d1afc24d0604 : SUCCESS in 21m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30c2890303fc4eeca6152ef161a910bd : SUCCESS in 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9a54f0d2f0c35a9970dd8d5f98ea79e52de503bc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-07 09:43:25.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"11a5a0058730ca6a2adaf223de327cc7babc8094","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-07 20:07:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"11043f18b68bc1e24ed5cda568a1e0ab59871f63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 23:48:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6543c0c3de4de41518e16f0c54f53b06d6e2e917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 02:05:09.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/953865 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"523bc72f8058c493d7e80e1f69dc0d8d7e9a8cd7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-09 12:40:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"36afb27e36acd011238053054b9d10d0707e6c80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 13:22:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ba54e8052de4f79a90f6fdd7f37f9c5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9a3ece89a8964c1cb95654f5f8193f82 : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/973a13f2ccc440978d76f8f6b00a22ba : SUCCESS in 19m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea97cdf1172a4645a63a6c4be782712f : SUCCESS in 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"06b3bf64bfdb2aa9f1548885e66815d3e3cf4503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 14:47:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/996d00cfadde41bbbb6291ab47ed5466\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/92711a82c9094b56a84d3be4a90983ea : SUCCESS in 1h 28m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/582d41b59be946eb8688cce2dae1b2ed : SUCCESS in 2h 04m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30921289f0594167ab7327594c00b1ec : SUCCESS in 33m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0145b3fe0734d0182a98c89006e7286 : SUCCESS in 35m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae311877cc74426be4ee3d94f73305f : SUCCESS in 12m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a2f144bd9864421842c6e9078551158 : SUCCESS in 19m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c6b7716aeec4e409292acfe89213b57 : SUCCESS in 13m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1bb6eee26ff4dd88cd13c33b7d28132 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7ddd02cb20433f9bc839f33f666ede : SUCCESS in 13m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/687f049d2e3d46c29f98f189d61af3c4 : FAILURE in 22m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c78d356bcbe4433a46da9eb5bc61692 : SUCCESS in 56m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a9ee2ef0b104ee79254394c44678a16 : SUCCESS in 1h 11m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1ce77ae8e3404457bb327f931ca8472d : SUCCESS in 1h 26m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b364145acb34c54955a25132e5772af : SUCCESS in 1h 17m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/25fcd07e9498412ea50352781679b826 : SUCCESS in 1h 01m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ba38c3c0170d49d888d0348e96640a63 : SUCCESS in 57m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ec60af0e8cc4522b7b2dc8ea01d0515 : SUCCESS in 51m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ab2b5e8ebd8b407bb718ea9d58282879 : SUCCESS in 1h 17m 48s","accounts_in_message":[],"_revision_number":5},{"id":"60ebe8303e17df9c5524488e08f4c7e0c59ec8e8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-09 15:14:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6de4de92eac1d3a478136556210c07955432b2c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:09:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/624bfa8506ea4489a961e13e8255495d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1316b5cc5f2e4181a0d8b1a8af574078 : SUCCESS in 29m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1414222e91cd410fbd0176abba0cf9df : NODE_FAILURE Node(set) request d4fb7ca31d364f0c8ae27112ed6fbec6 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/537741bc62da408bbf1ef4ad5bfa15ec : NODE_FAILURE Node(set) request 9fa5e40dfe54475ead4f46ef9a766e83 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5c5c94557764d9f1ab47fb064a79f14bb1b1f7d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 17:51:18.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d11f6dace3e4c1ba57604ee40e81706\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5b7cc5afa7934d9c955b89cf37611286 : SUCCESS in 1h 21m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ab47b53f18c843e880a407c1127c858d : SUCCESS in 1h 39m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/744f5a9132ae4828b1c3ab26392857f7 : SUCCESS in 33m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e059cd78ffd4c8bad5619e0f3595171 : SUCCESS in 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86c4da3e3f014d7ead1f031367631b58 : SUCCESS in 7m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72ffb56b06d54833977caccb4d0dc18f : SUCCESS in 19m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8fb5cf0ff40847ea96e051b98559c820 : SUCCESS in 21m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70c6f37e1cf14f19ad2e3d9f8e9819e1 : SUCCESS in 26m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1235503af04c7c828d7b18ea86dfdf : SUCCESS in 6m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e5220eaa4e2d4de7973366580b12d8be : SUCCESS in 1h 26m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff430092ce4741b79c238d542c74c77a : SUCCESS in 38m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eeb61b0422cf4f278910d100f15f2ddc : SUCCESS in 43m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f8d3b2212ca9488d8ae5909a32154117 : SUCCESS in 1h 18m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bd5ea25d79c349798ea1d4176ad90f0b : FAILURE in 1h 32m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ab97d08fc3df41ad87d2f2f6b766aef1 : SUCCESS in 1h 06m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1a0b109a97aa410e99934dd67dca42fe : SUCCESS in 30m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e02a4170a52045ca9a646b7d5fc7254f : SUCCESS in 39m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c0b3f581a5e4eba99268cb04c5baa5a : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":5},{"id":"01e64977e1b098786cb0862a8ef7d6824d0e5348","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-09 19:36:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7fc219d8f9966dcd8b8da9c65fc630cc69d556a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 20:13:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74cce643976b41d9916c9a51a15dbcdb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e6ca65e1099c4ddaaea14dbe9d3ab53a : SUCCESS in 35m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2697e98a0b214f2c9ad5782fe7840502 : SUCCESS in 32m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ada37c5bd0fd4409b55808e5d8c8291b : SUCCESS in 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4d22b09aea73de6c2687d0050fbb32190b9a09f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 21:35:37.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0aef7d480b754d6dbd75b5bfc335238a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/41bc994e07c94030839cd9a301506080 : SUCCESS in 1h 11m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9e716964a0544e198d7a023b2d7890aa : SUCCESS in 1h 57m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/890051cfa8b74a6484c0c02964fe9809 : SUCCESS in 31m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc82df8defa6493abcbf34c21100c3f1 : SUCCESS in 48m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a180dfaef24f9a9721efb7ba0fd7f8 : SUCCESS in 13m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f9b688dcd594965ac08535ff76fad40 : SUCCESS in 24m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81adbade61c54e88bd4eda48b2f9b575 : SUCCESS in 25m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a14cea5bd564c59a819f2270f06d21a : SUCCESS in 22m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8bb4debd6bb484cb7f547ac9f5fb7b3 : SUCCESS in 11m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c831e58c398f4d5a9fe36b78f2dd0ac9 : SUCCESS in 1h 37m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/89af489a89e04e8d909a87753e117629 : SUCCESS in 59m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4714c4be45645c0a306e344823acb2e : SUCCESS in 1h 17m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7c088bb9110a49cc9b2926c710ec43ea : SUCCESS in 1h 50m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d9c20bbaf0fc4656bfd78b7633bd00f5 : FAILURE in 1h 21m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7b5b74c4d0014169a36fb47e1051b8ff : SUCCESS in 1h 00m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4aec3889486f4c28afe7145056ea9f54 : SUCCESS in 1h 06m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a040b42a31b346688945901855bc2302 : SUCCESS in 1h 13m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1ae914e99ee84a10abf9e510dcff94b1 : SUCCESS in 1h 20m 07s","accounts_in_message":[],"_revision_number":5},{"id":"05c9bc5cfc8164c2a824ec5ebd071fbc8e70e8eb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-10 08:45:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8080cf64ffad8246e11b8d10d6e36bb76ab37ecd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 09:21:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f351e8609e274c2fadaf84b521b4bc65\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e3ea53de7b444e2b55640febc86a785 : SUCCESS in 34m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/550a4a880e454534bfb8c838c5d9545d : SUCCESS in 17m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6b0e723a81e84426a17587dda2c8b02a : SUCCESS in 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5e5d12e765d2811aeb77b8545f901dd2494cf4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 10:48:45.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85bae152a0bf4c9aa957a4f21806904e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e310f6d0cc114cc9a40edb87e514dfda : FAILURE in 1h 28m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/74a90e74f6b84d1b99f18e44ab3d3416 : SUCCESS in 2h 00m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/92582dcc40b54762ad83ad0e454fa530 : SUCCESS in 34m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc458396fb0f44bf91be527ffd799c36 : SUCCESS in 35m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d547e8f6cd45a2830e6dd0faf6fe0c : SUCCESS in 14m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3194efd0d050415d9c5fd27f0bae771a : SUCCESS in 20m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86447a36705848c1abac15bfd79689f5 : SUCCESS in 17m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bcf5dd7dd5a454e80108822e60477b9 : SUCCESS in 22m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38c286ee145495096c91655ddb56136 : SUCCESS in 12m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/206f62e3a97e4259bd24b140c49b6d60 : FAILURE in 28m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c5b39c3ee594b3ab78f3ccf063f3a88 : SUCCESS in 1h 03m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/64a6795ec6f24305bc4c01d27d0e9c6f : SUCCESS in 1h 14m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d3a580946c23469daa451df5a3e3baa5 : FAILURE in 1h 44m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2f7e1c633d154554bbc50fd98e273121 : SUCCESS in 1h 35m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/30a3c677051c4137811babe4890e181c : SUCCESS in 1h 02m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7e2441b01b104156aa55230fc38a3b30 : SUCCESS in 1h 00m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6e3cb6b9a4c04063913fb9de860d7aee : SUCCESS in 1h 14m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c3ab81c6929649499b51bce156b8c092 : SUCCESS in 1h 26m 34s","accounts_in_message":[],"_revision_number":5},{"id":"31d445199a375a1db6b2d75e3389008a2e1eae82","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 12:55:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3d838e5a99cc2505b03bc4e7b1f8e5dda855c818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:11:57.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a1d7ab3dde64623aaa2c99c2bd8ec13\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e6977d1f1a38416da9d295008cdfe3b4 : SUCCESS in 1h 22m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/90b1558eb6094dc9a65604cd6163fc6b : SUCCESS in 1h 43m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/964f2304d02f483f80dbb7d765b7a330 : SUCCESS in 30m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/480e1ace208b4619837019e1265a2ba9 : SUCCESS in 52m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90668704602b4fc9b8f89e28b0dcdc96 : SUCCESS in 15m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c08683e6150840a7ac83456552aeac99 : SUCCESS in 20m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ee1079bef494a328055e20b24bb6498 : SUCCESS in 21m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df7bd4342882451a9532ccf925c63e4b : SUCCESS in 29m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8028339613d540dd858d0b912d765f0f : SUCCESS in 13m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2d1ef9db4ad147899989f3b22772523c : FAILURE in 24m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/722d69aab9fd436f94297663af46f84f : FAILURE in 35m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef477a89b4154995bd75db9216e19b21 : SUCCESS in 33m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a6eb84d63f994300bc6f5be8834c3cc2 : FAILURE in 1h 57m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e61011fd64e54f3f96ba5037e8e90910 : FAILURE in 1h 31m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8db1e4248ff344e7abd5b2b7e0cacee4 : SUCCESS in 1h 10m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/53f64d5168294f7f9552267966afa440 : SUCCESS in 59m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2aec2190c36d48d3b8b2c55d1c65911a : SUCCESS in 1h 24m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ac18cdf445f54ca4a4c6b0ac7cfe8e31 : SUCCESS in 1h 24m 40s","accounts_in_message":[],"_revision_number":5},{"id":"5e2c532ebf2bb82b56245e2b56f469cdeae1a7ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:12:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19dd7d284c894bc0b1fab90d860f4bfc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ef6c96f5b564a37bbd102451b457438 : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d0efdefed2344fc98ae6f42e527b245e : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9fa4f16213da42fe9fa52823e621dbc6 : SUCCESS in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"64622f664f43a2f04686adb94689fce94ccceb26","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-10 15:22:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9e432254932ca2db4de9cc5a3aea33d3d065e45a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 17:35:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fd3e437aa054213b9bc7f6a1121b2e9\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a58afa344454325b8d2c6bfcabfc16b : SUCCESS in 1h 27m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6fa37fe2f1c24f5cbe23cb76f254792a : SUCCESS in 1h 51m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c7cf3836f2054ddcb022e714566c4c49 : SUCCESS in 32m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a82b5cd3388a433186ce0ac60a60930a : SUCCESS in 46m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd342c54ec14bef8ef54dce1ba17126 : SUCCESS in 13m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f60ec036f0bc47c3b7c089740aaf44dc : SUCCESS in 20m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de694049b650484b921922d688b0aed4 : SUCCESS in 18m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13fe89ba655d4a0d8a970de079b0f475 : SUCCESS in 16m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d0df0e1f1f4b26a93aaed17aa6b5a5 : SUCCESS in 12m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/781cc3bc01034c13b4e05ef85bdd4d4f : FAILURE in 1h 25m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/03540abe9020477b98820972084999b9 : SUCCESS in 53m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c9db36ff86147c28bc33b98ea03b3cd : SUCCESS in 1h 03m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/899f952669a64ad9913febc3a8045041 : SUCCESS in 1h 59m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1c0aedb8ee64ef4ad9110435f788ce7 : SUCCESS in 57m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/16728fe8023246ec86b29129a621b1db : SUCCESS in 1h 02m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f35f2da1235243e0afd2ef4a2ee1d6aa : SUCCESS in 52m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62170ce6a17c48aab1ce2364a56c1be7 : SUCCESS in 1h 24m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d0916180c98b49128b3006d9d9f7e0bc : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":5},{"id":"6c2d03353fff164152925d114fab6265c66a628a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:39:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6af21e3b19a4b15bcc03832cba516c4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12b7afdfb9d643cdb1be758acfff783c : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f4c6fdb3159d4d9a89cd5589394bca02 : SUCCESS in 36m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2a690f5663724f80abf63d51d3562b37 : SUCCESS in 21m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"01691c71daf24dece9c8d4d9d3c9456254c08712","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-15 14:44:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8b9555c70c8dd60be6962b67ccdf028ae8120211","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 16:37:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5b4afdfad7f4111812aaa100c93c373\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b45241b4538743ac8f404551f0d0dcc4 : SUCCESS in 43m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/98a176c3e2c04557b7f8d3cbb562f192 : SUCCESS in 39m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a59dc812d51448ff95aabe9e35e18c32 : SUCCESS in 45m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"31bc4c830b53d0851e7b95ca431cd6332bc4e08e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 16:40:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc6e8fd5ed584dae9658cbc4ff2730fd\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ca28ab1f195f41d5b654fa708e284778 : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e35ad8217f084ad08c3cac62c3d0db61 : SUCCESS in 1h 51m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e69f8b0063e144cb95f86c2c74aba065 : SUCCESS in 32m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/797e2b4771e24dceadb0dc63ac78e6eb : SUCCESS in 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddafbec8759a4f9d9c7a1576382ee63b : SUCCESS in 11m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc595451b6f344058a7d932ddf9c751a : SUCCESS in 18m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2c7a3e7fd734426aefb2138025bc6f1 : SUCCESS in 19m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e1d7b64921c4e6fbace470bdbe39a97 : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93a7319b1c24a08bcb9e98f4ee48612 : SUCCESS in 12m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1c290265e3b34d88b90e371efdc6baaf : SUCCESS in 1h 42m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8925843e8bb14c4e8a1f631e657f7cc4 : SUCCESS in 48m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce9d4b6515dd47a5b9fc73334f545f89 : SUCCESS in 1h 06m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/543b8788fdb74f0da12cd4211808e256 : SUCCESS in 1h 49m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e56e5ad188e4443aea3d735653b3de6 : SUCCESS in 1h 35m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/289540efdf654ea587def04a195e1818 : SUCCESS in 53m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f60e82cc06954ff1bd320ad6b7c74f88 : SUCCESS in 1h 09m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3d157438ea534ac181d58400a5e06c57 : SUCCESS in 1h 11m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9d563f8dfdb945f9bc78b6fe6a360099 : SUCCESS in 1h 18m 44s","accounts_in_message":[],"_revision_number":5},{"id":"4bcdd6444adbefdb8a22418a87620e833054f3d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 16:40:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"44e2cd01f147bb7e0061c847d7a7c48ecdfa2c1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 18:41:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/119420841a8e49afa7bb9a52b84b92ae\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a9d374fd92b043bb85b82be244eb615b : SUCCESS in 1h 25m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9258199db57f4d6dac0af011929b0af7 : SUCCESS in 2h 00m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/575dfb55d8664dd59442d84b3b1a21dd : SUCCESS in 14m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71652197c36741cea6b010473b5ba63d : SUCCESS in 23m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/080acca1ee0a4df2bfc3a457829d5e8c : SUCCESS in 15m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49ae7913a7bd4680bfd2a60043939366 : SUCCESS in 12m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/714471f2943048718f997ba4a5124551 : SUCCESS in 1h 41m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7e7392bf957a4175bd3f84e13953f582 : SUCCESS in 1h 53m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c728cce227ee42b48027fd50ffb5535f : SUCCESS in 1h 08m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3acb76e153a4b1a89ec31643a5eba89 : SUCCESS in 58m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6dbb70be471e4a53861547856a1267e0 : SUCCESS in 50m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5e4c16a2b1e74d9eaacb00096f8e5404 : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":5},{"id":"2dd33df2891416c6a2bcdfd5df55fa68e61f9032","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 18:41:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ecf37e5f4af86926d20b20980e8c76b1a2550cd5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 18:42:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42d559faf20a4e7480d52cef73a15ef3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d73b8c472d4b7fb5fdfac842e6f18c : SUCCESS in 53s","accounts_in_message":[],"_revision_number":5},{"id":"e6858b02a03ccd183c1f330efa83d97d0d82ab49","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 00:20:18.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/82/950582/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"b01debda855935ef479cf4bce506006de4cce826","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 03:04:24.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/82/950582/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"61d6d5e6ca750c24b14f13fdc7c7eafba627f3e1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 05:48:15.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/82/950582/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"068c2979627d5a0d4ca8f47e39d233ed0d5dd338","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 07:49:27.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/82/950582/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"41f7e99ecc6dc9a9f6091eac9c4b3c45dbac263e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 09:11:56.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/82/950582/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"7e17ad4c48d5c21e4657a10ce888c55790008de1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 10:34:08.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/82/950582/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a8999d0c038b0b30ec700c27547380240e6d247a","revisions":{"488888a5e9051b661d0ce907c8c87f2d11ce1509":{"kind":"REWORK","_number":1,"created":"2025-05-21 17:17:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/950582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/950582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/950582/1"}}},"commit":{"parents":[{"commit":"bd68df5ad26f7c0a4c7af404315c9f91e5248cba","subject":"tox: Fix indentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd68df5ad26f7c0a4c7af404315c9f91e5248cba"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:13:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:16:30.000000000","tz":60},"subject":"trivial: Autofix low-hanging linter failures","message":"trivial: Autofix low-hanging linter failures\n\nChange-Id: I474d98cee375f8a7af5feb07a6988dfcb635bad4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/488888a5e9051b661d0ce907c8c87f2d11ce1509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/488888a5e9051b661d0ce907c8c87f2d11ce1509"}]},"branch":"refs/heads/master"},"17b6e6e9723c63025e5909188cfb34d92049f8f2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-22 10:26:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/950582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/950582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/950582/2"}}},"commit":{"parents":[{"commit":"10748007fc88b1d4b797482dc9f807050a99db73","subject":"tox: Fix indentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10748007fc88b1d4b797482dc9f807050a99db73"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:13:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-22 10:26:33.000000000","tz":60},"subject":"trivial: Autofix low-hanging linter failures","message":"trivial: Autofix low-hanging linter failures\n\nChange-Id: I474d98cee375f8a7af5feb07a6988dfcb635bad4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17b6e6e9723c63025e5909188cfb34d92049f8f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17b6e6e9723c63025e5909188cfb34d92049f8f2"}]},"branch":"refs/heads/master"},"e5783bcaf86d398c398f22f5dfe5ccf78147f131":{"kind":"REWORK","_number":3,"created":"2025-05-26 11:28:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/950582/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/950582/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/950582/3"}}},"commit":{"parents":[{"commit":"498eaa10dfe1f83fe88c042c9af5aa40407f2e5e","subject":"tox: Fix indentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498eaa10dfe1f83fe88c042c9af5aa40407f2e5e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:13:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-26 11:16:45.000000000","tz":60},"subject":"trivial: Autofix low-hanging linter failures","message":"trivial: Autofix low-hanging linter failures\n\nChange-Id: I474d98cee375f8a7af5feb07a6988dfcb635bad4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5783bcaf86d398c398f22f5dfe5ccf78147f131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5783bcaf86d398c398f22f5dfe5ccf78147f131"}]},"branch":"refs/heads/master"},"7ceb12385109b53b6166e82dc05a79f08171f1f0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-01 09:48:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/950582/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/950582/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/950582/4"}}},"commit":{"parents":[{"commit":"cb913a8c440ffb0ba4300a42115e7e0476f23a87","subject":"docs: Remove crud from conf.py files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb913a8c440ffb0ba4300a42115e7e0476f23a87"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:13:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-07-01 09:47:55.000000000","tz":60},"subject":"trivial: Autofix low-hanging linter failures","message":"trivial: Autofix low-hanging linter failures\n\nChange-Id: I474d98cee375f8a7af5feb07a6988dfcb635bad4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ceb12385109b53b6166e82dc05a79f08171f1f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ceb12385109b53b6166e82dc05a79f08171f1f0"}]},"branch":"refs/heads/master"},"a8999d0c038b0b30ec700c27547380240e6d247a":{"kind":"REWORK","_number":5,"created":"2025-07-04 09:34:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/950582/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/950582/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/950582/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/950582/5"}}},"commit":{"parents":[{"commit":"cb913a8c440ffb0ba4300a42115e7e0476f23a87","subject":"docs: Remove crud from conf.py files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb913a8c440ffb0ba4300a42115e7e0476f23a87"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 17:13:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-07-04 09:34:26.000000000","tz":60},"subject":"trivial: Autofix low-hanging linter failures","message":"trivial: Autofix low-hanging linter failures\n\nChange-Id: I474d98cee375f8a7af5feb07a6988dfcb635bad4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8999d0c038b0b30ec700c27547380240e6d247a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8999d0c038b0b30ec700c27547380240e6d247a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
